)]}'
{"id":"openstack%2Fglance~1003625","triplet_id":"openstack%2Fglance~master~I399ad8d345341791111b11239889a7f423c074bf","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-03 15:11:54.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I399ad8d345341791111b11239889a7f423c074bf","subject":"DNM testing tempest change","status":"ABANDONED","created":"2026-09-03 03:15:40.000000000","updated":"2026-09-03 17:50:39.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f98c2c4dd87825c9b95ebc29e31135469b5cbc15","_number":1003625,"virtual_id_number":1003625,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-03 17:12:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 03:28:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-03 05:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47c91ba9a647d896da11ee4ea28cead4ac585a62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-03 03:15:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d52ff4d813ad30d8d7c38d8af50275a0bd76a3bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 03:28:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e888f05b9ba438c8ec85c91c2263868\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/69af409b5c2f491c89f6d9027363c874 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e13e42cd7689466886a9c2f2a22e35e5 : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ff1d5924099b47929fa75c6b02f0ae2b : SUCCESS in 8m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d967432a9f4d7bab2fd2c8c4ed740e54c606d29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 05:06:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cebde5a9f2443d0949d8fbf9d5c8e15\n\n- grenade https://zuul.opendev.org/t/openstack/build/4b84f62d77814ec5aac3f9499bd7f52d : FAILURE in 16m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/585d47501163490fbaf2b8a3a8ed539f : SUCCESS in 1h 05m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a3dc145d37a9436b99d83085c3fed71a : SUCCESS in 1h 21m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1210d24137f4fe987c484c55bb81349 : SUCCESS in 50m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33cb0e3906c844e8a1e3f20ca42c7b04 : SUCCESS in 6m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/71d8162a4d9c4ca3962be75769ec97bb : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14b73318d8f04d6caf01eecd2b127d5e : SUCCESS in 3m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ca3835ffd36d43b9b259f49976a0aecf : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff68e604007c482893707fd7b0187bdf : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8df0a7afe650456281a44ebf1e907732 : SUCCESS in 6m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6611739f2c5c41119aba7894ab794acb : SUCCESS in 5m 56s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b5ecbe7a8d6346d0a89fa4c97d50273d : SUCCESS in 6m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7fe4852e79414bbe9f0ad8275fe23951 : SUCCESS in 1h 28m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/93cbeaaa95324ac795380f956c5506dd : SUCCESS in 1h 38m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/06197e125e7c4e018834258cdda40c47 : SUCCESS in 1h 47m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41032a8b4d024017904d973be2207f0d : SUCCESS in 32m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/67597ec1c90c4ef5abb0fd43880ec0ea : SUCCESS in 1h 25m 27s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/dd956504219a480596dedd849cc14211 : FAILURE in 15m 25s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/6438c8afe4384663b7a844aa041543ce : FAILURE in 25m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/965b46349d194b419bf35e2eb27c45a8 : SUCCESS in 20m 29s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/b82a36c55987478e9fea6ef8160658ea : SUCCESS in 27m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdd2950d96538009d3de20d4855fb1dc42a33ada","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-03 15:11:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebdf8e8014d52930a47260ab0a0f77a130b19dd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:25:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f3efc9617cd42a59774bb40f8a58bc2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/278728b7afcb457382f49d488afc1208 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a0f583c2e69f4e2d8c0ca0f7bd10cff8 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5d1d470b05c14509a026065910d27b75 : SUCCESS in 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2759cc597e46b2984cf5ae50ddc3755670d6bcb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-03 16:29:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81d6337c1fd030ec5991c9dd6f1b7232d1076b20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 17:12:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/544894c2cee74ebf845aac8685fa5184\n\n- grenade https://zuul.opendev.org/t/openstack/build/7c1baa4fbbdc4af0af8f02454ad12cf5 : SUCCESS in 38m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e7900beef30a4e61aa4ffcd620ea8fe2 : SUCCESS in 1h 04m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2699d7e838324862ba1023f20a07eacd : SUCCESS in 1h 29m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37f9673a769449a79c9d203ffc066736 : SUCCESS in 1h 03m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06cad62a28c545ff9856f0cfdbcd7df2 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d1e615c80e642d3a8bdff340086ff74 : SUCCESS in 5m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cbc5a2ba13434b1e8a8fb2abff333811 : SUCCESS in 5m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5fa478aca3a04d6f94212b334bd94b87 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ddf296ad04f44b497bad8c9e3403290 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eea36fa4dce04f96af70263bf08cbbd0 : SUCCESS in 7m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7a27d2a86e144d85b910c2f4d56a1a62 : SUCCESS in 6m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/96b29bc1de2d489ab6ef77591c209ea5 : SUCCESS in 5m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8291723679314584b5a29325e2950c81 : SUCCESS in 1h 35m 23s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9f2bc13261b7421e8b55861640cfed3b : SUCCESS in 52m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/de1afebb675c47abbc7b0129a810edf9 : SUCCESS in 1h 56m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b14edf8a0c5c403e9824a4e6e6531236 : SUCCESS in 53m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f22a8e626a634b618c11c80c36633c3f : SUCCESS in 46m 27s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/08bfd33eba5443ff8faafe4ede11eb08 : FAILURE in 27m 09s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/453d750836ea434dbde245e1de28ee8c : FAILURE in 28m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/35cd0aa2a3ea458da02b4315cd0af31c : SUCCESS in 36m 14s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/05d9b583fe20486483e6c49b73031f97 : SUCCESS in 24m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f98c2c4dd87825c9b95ebc29e31135469b5cbc15","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-03 17:50:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f282e6ef2add28ec8023591f8b88800c0f9b4bb1","revisions":{"f282e6ef2add28ec8023591f8b88800c0f9b4bb1":{"kind":"REWORK","_number":1,"created":"2026-09-03 03:15:40.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/25/1003625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/25/1003625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/1003625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/1003625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/1003625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/25/1003625/1"}}},"commit":{"parents":[{"commit":"7f16823bf52b802b985ebbfcc5f42704360c3990","subject":"Merge \"Drop unused management_url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7f16823bf52b802b985ebbfcc5f42704360c3990"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-03 03:13:42.000000000","tz":-240},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-03 03:15:39.000000000","tz":-240},"subject":"DNM testing tempest change","message":"DNM testing tempest change\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/1003560\n\nChange-Id: I399ad8d345341791111b11239889a7f423c074bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f282e6ef2add28ec8023591f8b88800c0f9b4bb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f282e6ef2add28ec8023591f8b88800c0f9b4bb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
