)]}'
{"id":"openstack%2Fglance~1003865","triplet_id":"openstack%2Fglance~stable%2F2026.1~I684ed3953d17025e801d8bea2bb3be180c3bce8a","project":"openstack/glance","branch":"stable/2026.1","topic":"cephx-cipher-compat","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-09-04 07:44:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I684ed3953d17025e801d8bea2bb3be180c3bce8a","subject":"DNM: testing nova-ceph-multistore","status":"ABANDONED","created":"2026-09-03 18:18:26.000000000","updated":"2026-09-04 07:44:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"035e2458d1b03aa5988578a803dbc3cbcbf2cc61","_number":1003865,"virtual_id_number":1003865,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-03 20:56:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2026-09-03 21:59:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 18:19:47.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-09-03 18:20:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-03 18:20:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 21:59:02.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"}],"messages":[{"id":"5cefe0a4933a92996faf82a553d2e36513937900","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-03 18:18:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ddb623e6e3ac70ae1e2edae50013aa26c2253f3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-03 18:19:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0b026dd7c8ca7d2067952aea50b4b12eddd6ae6b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-03 18:19:43.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"bf56dc102444b347e01b4d236c80e46ef15dee55","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-03 18:19:47.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch stable/2026.1.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/glance/.zuul.yaml@stable/2026.1\", line 335, column 31","accounts_in_message":[],"_revision_number":2},{"id":"03762ccc8470ed164228842d2714fc5b12b61c73","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-03 18:19:47.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"edd9e87cab7fe5fdc58242e3abe25990f9c99947","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 18:20:00.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch stable/2026.1.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/glance/.zuul.yaml@stable/2026.1\", line 335, column 31","accounts_in_message":[],"_revision_number":2},{"id":"3fb4a3d14c4efca1180e258ac9750ebc6d0b4173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 18:20:16.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch stable/2026.1.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/glance/.zuul.yaml@stable/2026.1\", line 335, column 31","accounts_in_message":[],"_revision_number":2},{"id":"cdd8eb13f7a7bb1820a5e81b17ba0267a523f8d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-03 19:13:24.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d29a74003ab66129c349f43f81cfce916b8ff611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 20:56:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14572de3d0ca4a1e93367ff0ca3dde49\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/06aa8c7b91584087bc94d0baf8648a59 : SUCCESS in 1h 39m 49s","accounts_in_message":[],"_revision_number":3},{"id":"8cef0c3c7bd6541851dbd3b0216d2b284f0cb100","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-09-03 21:59:02.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"035e2458d1b03aa5988578a803dbc3cbcbf2cc61","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-04 07:44:59.000000000","message":"Abandoned\n\nIt has served it\u0027s purpose, safe to abandon.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9f0e3efdf95cfbd4b2c32a254ef777cbe2aa0f8c","revisions":{"b915a5df88f298b6b8091eede017c4510edbdb44":{"kind":"REWORK","_number":1,"created":"2026-09-03 18:18:26.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/65/1003865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/1003865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/1003865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/1003865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/1003865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/1003865/1"}}},"commit":{"parents":[{"commit":"ddeb19c445f8f8786a5e581dab065b4281f9d35f","subject":"Fix flaky import task waits in functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ddeb19c445f8f8786a5e581dab065b4281f9d35f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-03 18:18:26.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-03 18:18:26.000000000","tz":0},"subject":"DNM: testing nova-ceph-multistore","message":"DNM: testing nova-ceph-multistore\n\n\nDepends-On: I7e62696c9e334bf9ca6df38e654ae2de0fd35456\nChange-Id: I684ed3953d17025e801d8bea2bb3be180c3bce8a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b915a5df88f298b6b8091eede017c4510edbdb44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b915a5df88f298b6b8091eede017c4510edbdb44"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"4d0a3560ebe27b0d555422ad0d042309df270f35":{"kind":"REWORK","_number":2,"created":"2026-09-03 18:19:37.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/65/1003865/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/1003865/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/1003865/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/1003865/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/1003865/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/1003865/2"}}},"commit":{"parents":[{"commit":"ddeb19c445f8f8786a5e581dab065b4281f9d35f","subject":"Fix flaky import task waits in functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ddeb19c445f8f8786a5e581dab065b4281f9d35f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-03 18:18:26.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-03 18:19:37.000000000","tz":0},"subject":"DNM: testing nova-ceph-multistore","message":"DNM: testing nova-ceph-multistore\n\n\nDepends-On: I7e62696c9e334bf9ca6df38e654ae2de0fd35456\nChange-Id: I684ed3953d17025e801d8bea2bb3be180c3bce8a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4d0a3560ebe27b0d555422ad0d042309df270f35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4d0a3560ebe27b0d555422ad0d042309df270f35"}]},"branch":"refs/heads/stable/2026.1"},"9f0e3efdf95cfbd4b2c32a254ef777cbe2aa0f8c":{"kind":"REWORK","_number":3,"created":"2026-09-03 19:13:24.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/65/1003865/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/1003865/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/1003865/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/1003865/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/1003865/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/1003865/3"}}},"commit":{"parents":[{"commit":"ddeb19c445f8f8786a5e581dab065b4281f9d35f","subject":"Fix flaky import task waits in functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ddeb19c445f8f8786a5e581dab065b4281f9d35f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-03 18:18:26.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-03 19:13:24.000000000","tz":0},"subject":"DNM: testing nova-ceph-multistore","message":"DNM: testing nova-ceph-multistore\n\n\nDepends-On: I7e62696c9e334bf9ca6df38e654ae2de0fd35456\nChange-Id: I684ed3953d17025e801d8bea2bb3be180c3bce8a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f0e3efdf95cfbd4b2c32a254ef777cbe2aa0f8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f0e3efdf95cfbd4b2c32a254ef777cbe2aa0f8c"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
