)]}'
{"id":"openstack%2Fglance~1003956","triplet_id":"openstack%2Fglance~master~Ia479c736dc25af87156b3c2f4ff9fd9225e8e890","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-09-07 12:26:15.000000000","reason":"Change was abandoned"},"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2026-09-07 12:26:15.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-09-07 12:26:15.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-09-07 12:26:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia479c736dc25af87156b3c2f4ff9fd9225e8e890","subject":"Allow to limit DNS resolution by IP version","status":"ABANDONED","created":"2026-09-04 13:22:26.000000000","updated":"2026-09-07 12:26:15.000000000","total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"13d0d29945223c642453d3ef20994897d234d1e8","_number":1003956,"virtual_id_number":1003956,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-04 17:01:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":-1,"date":"2026-09-04 16:51:40.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 13:47:22.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2026-09-04 13:47:22.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-09-04 13:47:22.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-09-04 13:59:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-04 17:01:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5cfd02e3cf26766e8f14a6511647a0eba65a0f9d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 13:22:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0c7e54aab4e7b3329a59d8bd561dda21f5711f4","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 13:23:23.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"3b6c602c9be7bce3a9d239dc1b39c65a3191ad3b","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 13:23:28.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"735cf04486e2103cea10a00c321d3c23bca4ed3c","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 13:23:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ccb9f97218ffa219d3b35d658c567048d005fe3","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 13:32:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6441ede433d39105d2e21e1377192f442cba5afa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 13:59:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a9ba1c357304a2eb94c32b59e933a74\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/13fda07e69cb47e38e69ca5b73ef5a0f : FAILURE in 10m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/931bd64e3da1472cbd65f0e468b4a687 : FAILURE in 9m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b9f596b3fee847d9a79c81449909c1f6 : FAILURE in 17m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"764ccd1853e38dccae50991ff4afb55d6bb0eabc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 14:40:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"0051b2a0ad272a410cf7ebbdcfbadf9d066b3d5f","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 14:45:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0af0ebbcdda80e5fe160730798079762dfbf62a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 14:54:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d6a7d1fa493474bbd12af562472aafe\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/544401fa052041e7a363f832164407cc : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/30353d38d7d346a4b2744eb904671998 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e8a2c6e5bdd64304b41d2e0f87937b2e : SUCCESS in 11m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7c288afae3044c4ceaf086d0f758ebb0c204af28","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-04 14:56:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"144d4de935d5d52a7629af3d20b2db569bc6c0b1","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 15:24:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3de7dfe30575f44521268bbb3b52074bbfccdcbb","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 16:12:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22b16e11cf6f2b2c306a7d251e80ed9131a409cd","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 16:51:40.000000000","message":"Patch Set 2: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f62bb185830090709adcde797eec792a0874b3de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 17:01:03.000000000","message":"Patch Set 2: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/753453abcd3b43f49d5f349f0fa5bed4\n\n- grenade https://zuul.opendev.org/t/openstack/build/2e02c895cf9e459b8e1a765c1a070864 : SUCCESS in 37m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d1607bc5f0c4b1981cb1f8a3dda03ea : SUCCESS in 1h 18m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b4e171aff1f8423494e45d86edf7a09c : SUCCESS in 1h 09m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5503c36c2ca486db6ea644963c8e6d1 : SUCCESS in 1h 03m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de1a35aa88f24122babea074aa81159a : FAILURE in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a8bb346470d41f4861db3d2453a8680 : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/030954d71b994b3f8999acec1f4621ae : SUCCESS in 6m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c559bedd5f946e7ae0110e426ebd54b : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f43a25f7e97143c8814f3209e0782fca : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8147fafd7de2409e83881bde7e03544c : SUCCESS in 8m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2f5e1031b7d3458d8b5de007a460ec60 : SUCCESS in 6m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/20ae17ee726c48d7a3e2271f539df4f6 : SUCCESS in 5m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/700650dad67849d7912ad03241e3968f : SUCCESS in 1h 41m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/620cd1c758fc4d2a8efaf9e1060bdefd : SUCCESS in 57m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1fa510b11c794c5ba70321f2b5728cac : SUCCESS in 2h 14m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e8b071a1f94447a9316427650d21a4c : SUCCESS in 1h 01m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8ffc82e3ff6e4b19b46fb2afb01e6a54 : SUCCESS in 43m 41s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a3b3a7eac9c44e83b23fdbff7d352368 : FAILURE in 15m 03s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/84b4bc6a8f354afa8788fb6ce4d7b3fb : FAILURE in 32m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e1c8ab7d271e467995d956f0a3a66d26 : SUCCESS in 16m 56s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/474dbcb07b2946a087f49701bbbe8be0 : SUCCESS in 22m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13d0d29945223c642453d3ef20994897d234d1e8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-07 12:26:15.000000000","message":"Abandoned\n\nabandoned in favor of https://review.opendev.org/c/openstack/glance/+/1004004","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1677b44551f3f8a77337e483e1f3da3704da503e","revisions":{"9ea7121ab09f047ed8bca64c0a19ad7cc32a6ff0":{"kind":"REWORK","_number":1,"created":"2026-09-04 13:22:26.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/56/1003956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/1003956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/1003956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/1003956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/1003956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/1003956/1"}}},"commit":{"parents":[{"commit":"d1c3ffc71db15970b547a9ff65ccd034b138ff62","subject":"Merge \"Block restricted hosts when adding HTTP image locations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1c3ffc71db15970b547a9ff65ccd034b138ff62"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-09-03 13:10:34.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-09-04 13:22:18.000000000","tz":0},"subject":"Allow to limit DNS resolution by IP version","message":"Allow to limit DNS resolution by IP version\n\nin some deployment scenarios, Glance runs in a single-stack network.\nWhen during image import DNS resolution returns both IPv4 and IPv6\nnetworks, we must be able to select a proper one if such limitation\nis in place.\n\nThis patch adds new config option `[image_filtering]allowed_ip_version`,\nwith possible values of `IPv4`, `IPv6` and `all` (default).\nWhen set to non-default value, only addresses of this IP version will\nbe allowed.\n\nCloses-Bug: #2166488\nChange-Id: Ia479c736dc25af87156b3c2f4ff9fd9225e8e890\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ea7121ab09f047ed8bca64c0a19ad7cc32a6ff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ea7121ab09f047ed8bca64c0a19ad7cc32a6ff0"}]},"branch":"refs/heads/master"},"1677b44551f3f8a77337e483e1f3da3704da503e":{"kind":"REWORK","_number":2,"created":"2026-09-04 14:40:47.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/56/1003956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/1003956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/1003956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/1003956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/1003956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/1003956/2"}}},"commit":{"parents":[{"commit":"d1c3ffc71db15970b547a9ff65ccd034b138ff62","subject":"Merge \"Block restricted hosts when adding HTTP image locations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1c3ffc71db15970b547a9ff65ccd034b138ff62"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-09-03 13:10:34.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-09-04 14:40:40.000000000","tz":0},"subject":"Allow to limit DNS resolution by IP version","message":"Allow to limit DNS resolution by IP version\n\nin some deployment scenarios, Glance runs in a single-stack network.\nWhen during image import DNS resolution returns both IPv4 and IPv6\nnetworks, we must be able to select a proper one if such limitation\nis in place.\n\nThis patch adds new config option `[image_filtering]allowed_ip_version`,\nwith possible values of `IPv4`, `IPv6` and `all` (default).\nWhen set to non-default value, only addresses of this IP version will\nbe allowed.\n\nCloses-Bug: #2166488\nChange-Id: Ia479c736dc25af87156b3c2f4ff9fd9225e8e890\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1677b44551f3f8a77337e483e1f3da3704da503e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1677b44551f3f8a77337e483e1f3da3704da503e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
