)]}'
{"id":"openstack%2Fglance~1004004","triplet_id":"openstack%2Fglance~master~I335820470f4474bbec95efc05ca3b45a6a370d2b","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2026-09-07 13:50:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I335820470f4474bbec95efc05ca3b45a6a370d2b","subject":"Try all allowed DNS addresses when importing image","status":"MERGED","created":"2026-09-04 16:10:52.000000000","updated":"2026-09-07 13:52:35.000000000","submitted":"2026-09-07 13:50:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004004","meta_rev_id":"d69254380eb39813f8bf4dfd40f910083d4cc1cf","_number":1004004,"virtual_id_number":1004004,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 13:50:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":2,"date":"2026-09-07 11:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-09-07 12:16:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-09-07 12:16:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 16:26:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-04 18:38:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-07 06:20:10.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2026-09-07 11:30:13.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-09-07 12:16:09.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"59a5496caad788880e7d998c47bc5dafa3b36a03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 16:10:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3f00cdfe5a573fc626c08e903f1c3ec58605f0f","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 16:11:22.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b8833cfa2a1f4c8114867242486889c38c0d89a","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 16:18:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"55fa20ccac55c543266400a263082da0e80cc32c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 16:26:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4731de4940540bc86bf2471329f66a1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ff4f7b0fea3445dabd590ddbf2955035 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a9f3fa7b6616432fa64063e876ce593f : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5b3e94f35ab24118b3aea3a7eb33d878 : SUCCESS in 13m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0eac822e1975aacedaf9da4d629b67fad194fb9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 16:49:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"73f01ec206eb2ea181561da1b6f120aacef6e18d","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-04 16:51:28.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"21f6eb92b69a4114b178412f7a75755aa1869067","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 17:05:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1063f16fd99b45d99eb43ce7d8d66567\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/adafe30b28b94138bf36971e561a25c0 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3649b967543e4e3ca36cd4ceda61ba4a : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f9a887040eda4a048ed63db6c4e885e8 : SUCCESS in 12m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa381dc92fc19250d797a84ad79c23eab5c99720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 18:38:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad811b8881e64b5995f9b9fb0ab5fb7a\n\n- grenade https://zuul.opendev.org/t/openstack/build/1cc2e0a0f4524c20b921ef447ceafd72 : SUCCESS in 1h 05m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8c6659acade947d6a6b2b1fbb97cd775 : SUCCESS in 42m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/50718e5462354f669ecfedeab04862d6 : SUCCESS in 1h 30m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/439b9308e7444026853a0e8bc096af65 : SUCCESS in 58m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ae615d3916d40e7abfc3abc56c4be17 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63ffd822a3eb479db4c8f1dec26d1b31 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6aef2c54d21546a78c830df2e2b5f6d3 : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2f7fbe842550475c8b330d9858a469f7 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d551a2866181482c8fd2890a462dcc98 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6a6624f0fba492db997516423617120 : SUCCESS in 8m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/285930133d394171926961a5f5004b25 : SUCCESS in 4m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e92aceba90424e9097e84eab5973f820 : SUCCESS in 7m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/434abb1d64ec44669928dfdc8fbbd96c : SUCCESS in 1h 42m 50s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/440e3636cead45fab3ff15e226902b81 : SUCCESS in 1h 35m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4e68606312ab446681ff6f14d9e96336 : SUCCESS in 55m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ccb1981db1d4a019233174b6a6f5979 : SUCCESS in 38m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8372de813a8b41fdaf9b04043368bc04 : FAILURE in 1h 39m 46s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2eb7b53fb0b344a191f1f24a9b1c2996 : FAILURE in 28m 04s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/a3d3a0203a8f4da0ac3ebe227101cda4 : FAILURE in 25m 54s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/97ed8373374148679b6c1eddc60c1dec : SUCCESS in 34m 46s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/9fc9bf3f1b6d49e99e1355574cbc8803 : SUCCESS in 51m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"43b08bd4a1e4c34031a5ac496f3992cc7f4d5188","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-07 06:20:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a27eb9b25fa7bf29b9ce1ea7023e53a22c5e570","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-07 09:18:42.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0d0b85b240463480e352e26bad6281506218e025","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 10:45:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70f2cf3bfe4248c8aa34489b1860e1de\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1821f5a2126a44e3b0de3641d2f5e5f6 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f4fb4b13e1b247f99618eb30954c6f05 : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7b28053d85dd4de7a2bad0664c438df1 : SUCCESS in 12m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"38e628a539f9347236b0aa15aff24384c75667b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 11:20:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f3f9fbad2a48efbf21d61bc57ecffd\n\n- grenade https://zuul.opendev.org/t/openstack/build/278545716afb41e6ad7700319dac380c : SUCCESS in 1h 00m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7a1ee790097849138cb29b5195f5b820 : SUCCESS in 1h 03m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/94779589d3784f86a59adfdb5f7aeebf : SUCCESS in 1h 22m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ea3a0cff632460d8a89ec88433f274b : SUCCESS in 1h 09m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c70e6ce738da40afa10bb5fd9e98adb7 : SUCCESS in 6m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/301509e317544d15ad3fa798d3e07332 : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/54e0209008264b9084164ddc3ff3018c : SUCCESS in 5m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/04606b378cd2448a9b9ffaf78d8e2b13 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ae5fb0c3fd0440f91c25d64d13f1ea1 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/effe9f0ee0a24b8cb0b5315273c880c5 : SUCCESS in 8m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7176b865aa8f4d74bc707bfe6524003b : SUCCESS in 7m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bcfde19c985d4c5c820f47b40a0a189b : SUCCESS in 7m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/da4290e88fe741809da67e0830af6f4e : SUCCESS in 1h 39m 30s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/478b2ac2f46b453680db1be85b4e0039 : SUCCESS in 1h 45m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5c6247262db34c87be31a9dacce8a3a3 : SUCCESS in 1h 44m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/869b791bb45d496dbf1b014b4151af7e : SUCCESS in 56m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/926c85bb5ca84c358456261290ac6fa1 : SUCCESS in 1h 44m 55s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/440d9a78824043f0a7793f59aa91d030 : FAILURE in 29m 02s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/fe16d648766e49878f0eda6ad57e37a9 : FAILURE in 27m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0196564d50474ca29d1ec0deff1bbf94 : SUCCESS in 37m 33s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/29765bac77e04e7c87bd3eed7a0dfe35 : SUCCESS in 51m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"264a8d7afb3bf0ec0c37834006ddab5366a7817e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-07 11:30:13.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"090d1c40bdb60e576d32394680c7deada9222f69","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-09-07 12:16:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"16f5e185172ea02af878ec0f65b237f36f59140e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 12:16:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"86659e9a072d9f53e9aa526e8bee7d25d729336d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 13:50:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0ffb61f7361481b968c42ee5b0f1206\n\n- grenade https://zuul.opendev.org/t/openstack/build/239b4bd95f2f49c2974c13fb7ceb2847 : SUCCESS in 1h 10m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5a37bf2de81545d5bd4c57ac78dcae3f : SUCCESS in 1h 00m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0ab167c9e1f34e088a1000ffdba5808e : SUCCESS in 1h 21m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0aeba9d32f0e470e97e2aa9965fd76ef : SUCCESS in 59m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5edc5b31194526a01ad4d056f348e8 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a345f56ac3d3457280cd4015ec2fa22c : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d8614a92dd74766982b9077e4865ea8 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e5414be8be24c5bb76de86826432193 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ab8373a3f374ee8b372da8a630595df : SUCCESS in 8m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/85911b252bb846099deee75eec792f6f : SUCCESS in 5m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a414d069ae68423bbedeb291555592d9 : SUCCESS in 4m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5adc2c5afcbd478187642ea0cd979017 : SUCCESS in 1h 32m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4e9ab9728c34877a971fdf51f9a982a : SUCCESS in 27m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e9c5b60cf8144c8bb197e4334b37daee : SUCCESS in 55m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e068094e647849d89708bce10227a987 : SUCCESS in 33m 09s","accounts_in_message":[],"_revision_number":2},{"id":"4e6aa4eae65c32d7910c283a2a5d6882077360ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 13:50:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d69254380eb39813f8bf4dfd40f910083d4cc1cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 13:52:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32ce885576494008a5c023b8212cedfc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d26e312d4843c9aa4e875530e1ce4e : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb5d912a5f084f6f931e9c0f770d6d90 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"abc1077f16e8d0818825618f64cdd7c0fb6edcf1","revisions":{"dd13a194b69628f4c42012cc8b522bc06cac6625":{"kind":"REWORK","_number":1,"created":"2026-09-04 16:10:52.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/04/1004004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/04/1004004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/1004004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/1004004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/1004004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/04/1004004/1"}}},"commit":{"parents":[{"commit":"d1c3ffc71db15970b547a9ff65ccd034b138ff62","subject":"Merge \"Block restricted hosts when adding HTTP image locations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1c3ffc71db15970b547a9ff65ccd034b138ff62"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-09-04 15:41:22.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-09-04 16:09:06.000000000","tz":0},"subject":"Try all allowed DNS addresses when importing image","message":"Try all allowed DNS addresses when importing image\n\ninstead of picking the first of returned DNS addresses of the\nhost from where to web-download the image, consider all addresses\nreturned by DNS that pass filtering, and try them one by one.\n\nCloses-Bug: #2166488\nChange-Id: I335820470f4474bbec95efc05ca3b45a6a370d2b\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd13a194b69628f4c42012cc8b522bc06cac6625"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd13a194b69628f4c42012cc8b522bc06cac6625"}]},"branch":"refs/heads/master"},"abc1077f16e8d0818825618f64cdd7c0fb6edcf1":{"kind":"REWORK","_number":2,"created":"2026-09-04 16:49:57.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/04/1004004/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/04/1004004/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/1004004/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/1004004/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/1004004/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/04/1004004/2"}}},"commit":{"parents":[{"commit":"d1c3ffc71db15970b547a9ff65ccd034b138ff62","subject":"Merge \"Block restricted hosts when adding HTTP image locations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1c3ffc71db15970b547a9ff65ccd034b138ff62"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-09-04 15:41:22.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-09-04 16:19:27.000000000","tz":0},"subject":"Try all allowed DNS addresses when importing image","message":"Try all allowed DNS addresses when importing image\n\ninstead of picking the first of returned DNS addresses of the\nhost from where to web-download the image, consider all addresses\nreturned by DNS that pass filtering, and try them one by one.\n\nCloses-Bug: #2166488\nChange-Id: I335820470f4474bbec95efc05ca3b45a6a370d2b\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/abc1077f16e8d0818825618f64cdd7c0fb6edcf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/abc1077f16e8d0818825618f64cdd7c0fb6edcf1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
