)]}'
{"id":"openstack%2Fglance~1004175","triplet_id":"openstack%2Fglance~master~I868bfaf15500c349f4fbb111e29285336a4d48bd","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-17 16:40:05.000000000","reason":"Change was submitted"},"34510":{"account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"last_update":"2026-09-09 08:05:41.000000000","reason":"\u003cGERRIT_ACCOUNT_34510\u003e replied on the change","reason_account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}}},"hashtags":[],"change_id":"I868bfaf15500c349f4fbb111e29285336a4d48bd","subject":"Use common environment classifier","status":"MERGED","created":"2026-09-05 13:26:49.000000000","updated":"2026-09-17 16:41:51.000000000","submitted":"2026-09-17 16:40:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004175","meta_rev_id":"199987376b0226e856eb737c7f744952daacf0c9","_number":1004175,"virtual_id_number":1004175,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-17 16:40:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-09-17 13:56:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-09 08:05:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-09-17 13:56:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 13:40:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-05 15:17:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-09 07:51:16.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"REVIEWER"},{"updated":"2026-09-17 13:56:07.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"86fc5fb89e92c7a7a63f18da58b8948fd89e5dd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 13:26:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42742b23257c974cc840d7be8a9de44f98fbfba6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 13:40:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3b3611c16c6405bbdba299d955de37a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c6c5afd76fa444e39acc229043e9e1c8 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5fa863aa524049758522f1e045fdd9df : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fb3fd817c7a24b46b7596abae3c8ca59 : SUCCESS in 11m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b057a3ba9f16714010e66958138f0c1229d78626","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 15:17:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be1016af4c8c4f1a906b3f1cb56746e2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a02f46ef87cc4ac88b987cba5741307d : SUCCESS in 2m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/82cdb51cb8cd4778b9ded649f6d4e286 : SUCCESS in 2m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/646d46f8adcf4f499780282982d095c3 : SUCCESS in 36m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d10ee509205e4a17a13e989af64df2d7 : SUCCESS in 1h 06m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b08f3b650df452b9a25808a0a068154 : SUCCESS in 1h 40m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75d07b6054e04508bec6651c97dbb92d : SUCCESS in 1h 01m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b1c5bbb0ca74e7094fcc632f4d6f883 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3d40766cf0d4d31b715393de0392153 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38d335572b9a4e5cae39e7287c45d2dd : SUCCESS in 5m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7ab2c9bf58ff4954a4a72b28a31afc59 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae51ed483414bd0a84e03662fc16420 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d406cab2bf24928a87ac6150266f3ce : SUCCESS in 7m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ae92c49a118e482bbe762eecfd736774 : SUCCESS in 7m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cc96a99c5142422e84942a2f55b5a7b4 : SUCCESS in 7m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/32f6dbcba5d24efeb4b7ab44c6a8442e : SUCCESS in 1h 28m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6ec12a26fd9f4e10a2c31027c0d99277 : SUCCESS in 1h 23m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c086efbff21e4aad93759f36f584774c : SUCCESS in 1h 48m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0565c810896a4c83ac57eee26707afae : SUCCESS in 53m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9dc6638d63964a1e9d2e882e69cef219 : SUCCESS in 1h 28m 16s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/63568d26f73e4bc7b471410860d63898 : FAILURE in 19m 41s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/368bd76970ca4dd1abca647db325c73b : FAILURE in 27m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a0f3c0ac9f6a4676842b22387d6b4a24 : SUCCESS in 33m 06s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/717a26a16c3a42cabd2090cbc4ade34a : SUCCESS in 38m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7309ce4dd7685157580a1ecf304bb07bcb4e5266","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-09-09 08:05:41.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07456c39246464867349bd24659fe59db36d9dbe","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-09-17 13:56:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c69d1e242c1cda64b31f379b5e260a10c4f9293","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 13:56:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"634b0eaee5629a111daddcd52c1c006e77b0f548","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:40:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3c6ac7463ba4549aa28f19f768b0941\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8454b73646224e6b847b6160f4a1c9c1 : SUCCESS in 2m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0fed6d551d784154b14860eae8dd7293 : SUCCESS in 3m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/96ca962471c04ac389e31596b0e9c946 : SUCCESS in 1h 08m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/626cda61791146149f165a5ba29f45da : SUCCESS in 1h 09m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd213e0d774942edb1af23a8d8a83c90 : SUCCESS in 1h 25m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cc7749b82e9409e993f2c10f45bf4f0 : SUCCESS in 45m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc0a1a30d25e40ef96f76c8fc84f2ead : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/556ab702699e44b3898912bcbee4c439 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/25261fa03ac44756a34e2b1c102fc944 : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a9254035ae9437ab16776f6ae79b789 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51f7a72c4ea747458570fd07c351bfce : SUCCESS in 6m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/22d60dde2b6e4a3395754171cac9a475 : SUCCESS in 7m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5659cdf3d8144cd187fb7b83d1a2b5d8 : SUCCESS in 7m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b9823c9f067845beb61985cb465eb774 : SUCCESS in 1h 44m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7a4c0a12ffb4314aa50c453b704e8f0 : SUCCESS in 59m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b6d13f29a09f4478a892112fcbc1eab4 : SUCCESS in 1h 38m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d35207f4c17541ffa53c2f634a22df20 : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d45fb7d4edee51277eb5c6e25c0c4ac118f894d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:40:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"199987376b0226e856eb737c7f744952daacf0c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:41:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3695bcacce4b4867b3b133341e7db26a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b5620792fe42058d373c9fd0a65842 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e987346b78cb5b46fa8a3d820ba6906214bcc576","revisions":{"e987346b78cb5b46fa8a3d820ba6906214bcc576":{"kind":"REWORK","_number":1,"created":"2026-09-05 13:26:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/1004175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/1004175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/1004175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/1004175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/1004175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/1004175/1"}}},"commit":{"parents":[{"commit":"d1c3ffc71db15970b547a9ff65ccd034b138ff62","subject":"Merge \"Block restricted hosts when adding HTTP image locations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1c3ffc71db15970b547a9ff65ccd034b138ff62"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 13:26:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 13:26:20.000000000","tz":540},"subject":"Use common environment classifier","message":"Use common environment classifier\n\nThe \"Environment :: OpenStack\" classifier is commonly used in most of\nthe OpenStack projects.\n\nChange-Id: I868bfaf15500c349f4fbb111e29285336a4d48bd\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e987346b78cb5b46fa8a3d820ba6906214bcc576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e987346b78cb5b46fa8a3d820ba6906214bcc576"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
