)]}'
{"id":"openstack%2Fglance~1004983","triplet_id":"openstack%2Fglance~master~I12976626b02a3833a1404b10342c52a9f1b4d1b5","project":"openstack/glance","branch":"master","attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-09-10 09:40:37.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I12976626b02a3833a1404b10342c52a9f1b4d1b5","subject":"Update glance-image-import.conf.sample","status":"NEW","created":"2026-09-10 09:40:13.000000000","updated":"2026-09-10 11:37:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"151cf2f2907f46711358efbf4880907b332ae7f4","_number":1004983,"virtual_id_number":1004983,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-10 11:37:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 09:40:37.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-09-10 10:12:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 11:37:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4c1412c76abeb6207bbd33b533bbe1a6d16a0ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-09-10 09:40:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff8a792dae8a1442ecc8009df5e939bbeba35f0e","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-09-10 09:40:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6be8115781dc52554870469733be926c9eaf2cdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 10:12:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c88e25cd8dd3409eb5c81162f09dd11b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe217f5c44b846b4bddb82980ab9b260 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c648a36f4fda480faee3b3cf60d60c96 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f790e5f2b8e340be9091b047eaa6ab22 : SUCCESS in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"151cf2f2907f46711358efbf4880907b332ae7f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 11:37:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bace311025a44c168373e9bffe2ba29e\n\n- grenade https://zuul.opendev.org/t/openstack/build/597b6abb6f004aaaa1d2d1c1a88c7af8 : SUCCESS in 35m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c937895051b4912b16d2ce3c4554c33 : SUCCESS in 35m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3bb2817c21484a7db15de737ffef6767 : SUCCESS in 1h 01m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/036cd21faddf4e9085b3c7cc599ab237 : SUCCESS in 1h 03m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d94552fd21eb457fa57715a71e7e4e53 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a6c9fa10a9e412bb48bcaf65c6b95f2 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8c78a9207004c9bba73c0a0b82f6216 : SUCCESS in 5m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3bb8b71f69574689943b264b6094d69e : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f55da910cf48d0b244f9964d403005 : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a5ce71b9df849b0a9ef1719cd04dcb1 : SUCCESS in 9m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3405d3ecc9e240a2a967b85289f22e65 : SUCCESS in 4m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/64512219567944cdb1aaffd5116d71ef : SUCCESS in 4m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cfca1bcc422843fcb108f895388b840e : SUCCESS in 1h 39m 20s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f379df70d37649d6b07c4f9219bde56d : SUCCESS in 1h 38m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/86433f9938cd40239b06007666b98031 : SUCCESS in 1h 54m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30e377b36a1e4d119e66187919c88b1e : SUCCESS in 58m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a03c317612514117aa831b4c765d45db : SUCCESS in 1h 37m 51s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/0d058fa21bd644769d64d89dd998bf99 : FAILURE in 27m 46s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/9862e526fcd84af5ae4bea4645ff4697 : FAILURE in 14m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/57b405ad047c471389402757e5dae155 : SUCCESS in 16m 39s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/c614147fa3e44fbb847b3e861901d9a6 : SUCCESS in 25m 42s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43ae788c332949755dbeee7113b110f34752303c","revisions":{"43ae788c332949755dbeee7113b110f34752303c":{"kind":"REWORK","_number":1,"created":"2026-09-10 09:40:13.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/83/1004983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/1004983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/1004983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/1004983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/1004983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/1004983/1"}}},"commit":{"parents":[{"commit":"abc1077f16e8d0818825618f64cdd7c0fb6edcf1","subject":"Try all allowed DNS addresses when importing image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/abc1077f16e8d0818825618f64cdd7c0fb6edcf1"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2026-09-10 09:38:10.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2026-09-10 09:40:03.000000000","tz":120},"subject":"Update glance-image-import.conf.sample","message":"Update glance-image-import.conf.sample\n\nFew month ago, a new param option has been introduced in code, but the\nsample file generated and commited in code has never been updated\naccordingly.\n\nChange-Id: I12976626b02a3833a1404b10342c52a9f1b4d1b5\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43ae788c332949755dbeee7113b110f34752303c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43ae788c332949755dbeee7113b110f34752303c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
