)]}'
{"id":"openstack%2Fglance~1005016","triplet_id":"openstack%2Fglance~master~I6aafe6c1f32915299eb1b5152de9e3bac6038726","project":"openstack/glance","branch":"master","attention_set":{"37800":{"account":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"last_update":"2026-09-10 18:13:45.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2026-09-10 21:56:38.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I6aafe6c1f32915299eb1b5152de9e3bac6038726","subject":"trust_auth: Register missing session conf options","status":"NEW","created":"2026-09-10 13:56:55.000000000","updated":"2026-09-10 21:56:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fde3afe0cfd18c3fe8f7641fd79f2804fbfe838d","_number":1005016,"virtual_id_number":1005016,"owner":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-10 20:40:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-10 21:56:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 15:19:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 15:56:54.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2026-09-10 16:03:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f44c72998692ba5479678145cff161a73de7617a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"date":"2026-09-10 13:56:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46667bfbb4255aa965c06777163d451944a9ace8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:19:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/372baaa47243432caba9e92f84677809\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3b8425420920421ca8ddec7e4b742819 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/67e21a754fb8429d9c4f1ac746bd061d : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0c78b7c6d67e4cffb67f99b8d8575d2b : SUCCESS in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"408f01f5107e6efabd4ec5c53e4c8e8e7464dc88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:03:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df076aa2e1f74e13b7a9c22af32a0e69\n\n- grenade https://zuul.opendev.org/t/openstack/build/37497ccdfc1f4b16b62eed01cfa404ed : SUCCESS in 1h 05m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b97dd4243a3a4b479e3b214166caa13a : SUCCESS in 58m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4f4d50642724782adf754c5b7f824bb : SUCCESS in 1h 29m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc121831944441a7ad39e736742997f0 : SUCCESS in 1h 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c07e6e8097904837836a79a36fd66450 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65a7ddbff97c45668902f2c3ea14dbc3 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e8249c851554e8f98ac3bc78e09a219 : SUCCESS in 5m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/757c3ac27a8141bdbef419875d2442f8 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29b545c9476b44abb846e902b7ffa3fe : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dedb6d49f5a4351b8a1e27ba6e1e58c : SUCCESS in 9m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/518ed4c3d48d4136ba0666a8b1bc17ad : SUCCESS in 6m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/43d5881d20d84ae795d67c6c2a7886be : SUCCESS in 8m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/42160f72d21f4305a1b9a26190c9e71b : SUCCESS in 1h 45m 48s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d1f7095ce3334506935f178e9baf0ff7 : SUCCESS in 1h 53m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4c7f746718c745aba462b5e2d256604f : SUCCESS in 56m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d1be8ac222c4680a9c55718823e72e1 : SUCCESS in 1h 03m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b5ed85675ad344b384bbe7ad836eba62 : SUCCESS in 43m 06s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f2238ab164204d7ba563b4e0190518a6 : FAILURE in 34m 37s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/37acc7acac324110a854a8e441cff53b : FAILURE in 13m 44s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/17a098071e86404a98c2aafeefbd5a02 : SUCCESS in 17m 55s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/384832c5281747bbb16afd2db638bdd1 : SUCCESS in 50m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72d1ce0d4b2f3294ff256bb0724e2b4013df602b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"date":"2026-09-10 16:23:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"96634d9ca7a1edb48ee515ade87220a8da23bf93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:35:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcd6367341394e959ff1a7b53deec3e2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1e84b5c1903e48289d40b9f8f4730d3c : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9803399d23d94dc1b2449124b804b24f : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/42f40a17359e40009d6716cd611245d6 : SUCCESS in 9m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7768c81c5c4380d8870ef4ee21fec42a62406ad1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 18:13:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0bfacc06ae547948b54f0a35d0d36f9\n\n- grenade https://zuul.opendev.org/t/openstack/build/ffe968183c5b4702b3c2504d440cc3de : SUCCESS in 29m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/14e6160e67f04b16998cc57ad20260cb : SUCCESS in 31m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ebdd10303b4f412786b101cd520702fe : SUCCESS in 59m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0372f1790446469f84f17a29f51ebb1e : SUCCESS in 28m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7caa481e911441d8df569e4e1a90b90 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4588f3e663c84dada4122ab2fe385dc1 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6153ed1ae5d744049932214d84b0f80b : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b6d4a9b80dd44a17acc0ee3d6227366f : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae8f60d0a9da42d1b35ba279c326835c : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4648c39b5b1548ddbed6e0ce7f2e328b : SUCCESS in 9m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cd41a8e05498490da5606cc73e495f01 : FAILURE in 8m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7f6debbd2d3341308ea61c0b2e013c50 : FAILURE in 8m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f4e2dd55819446fd833ca8b19e29e6af : SUCCESS in 1h 35m 44s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a30b1263039543b0a92f002c8ca4e9eb : SUCCESS in 1h 05m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/683faa4e26644ae98453d7f92b789d57 : SUCCESS in 1h 12m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a6582a1060242dc8276ba765e4f9ba3 : SUCCESS in 29m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f6c182b2c25843c894370ac222565289 : SUCCESS in 1h 43m 25s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/32e3c8c34149419f810fce385bed7f34 : FAILURE in 17m 45s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/72f399dc94564a71beb4a2a9972efb6a : FAILURE in 32m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/db8f0d012b0344719ba64eb7b7001fc9 : SUCCESS in 40m 34s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/1bceb928923445138b4d2eec2c03545e : SUCCESS in 52m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c425d37b4819eb0a43b45bf01af688aec71a2be4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"date":"2026-09-10 18:49:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5b23c55efdb39fe63a9edf90fe7b9da128a29f6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 19:05:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af179826dece45d2a9f78c707ce50784\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e32bbae5e70542e9b4cbebb32c26c609 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/91c577150f194d81a0ff564e138a5e4a : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6531a2d4c9c54ac19c809858f1f13f87 : SUCCESS in 12m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"622b30ca4e3acada612dcf7c84c5c65a82a46d9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 20:40:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c4f0d92251740ea91c5ca15b3e20fc6\n\n- grenade https://zuul.opendev.org/t/openstack/build/65270f881f534ffda944d1a394446b47 : SUCCESS in 1h 04m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/65978cc698d84b098ec944f1bac959c4 : SUCCESS in 1h 01m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0625eaa0e97948f185b60bfa9241c625 : SUCCESS in 1h 01m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4d0f3ef2e3e4ea9a35d52a727aabc76 : SUCCESS in 45m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3d32d2db9f7425ab9ba96a263b03f28 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/548e7f3ec7d64dad843540a23024eff2 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d62c91551124ddc90cd299e4c804c6b : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c6f831f7d234ac3bb6563d60601d08c : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/240acc3949604bcd82dceb4ec24b06aa : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d30e0e1f6b61452391d872d502dff432 : SUCCESS in 8m 07s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c640054f8a8a4c1ca386b7a3e352efe9 : SUCCESS in 8m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/46505862cf354143a4fdac0d8ba6a8e7 : SUCCESS in 7m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0263f0c552024c9e856965774654dd15 : SUCCESS in 1h 30m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d85a67b1675c474eb695255d3210febe : SUCCESS in 1h 48m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5bb0c22d0a4141ecad7455d4efa7c05e : SUCCESS in 1h 45m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/075cfe16593d4e9faf62e675948bbf26 : SUCCESS in 56m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dda18d95358341cbab857f5ff7e897c3 : SUCCESS in 1h 31m 25s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/13cd3bac93a34ce6aff481efbb89d5c9 : FAILURE in 28m 08s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/61e15831e87940df972e997def0d0270 : FAILURE in 12m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/37a863e91ce84e6aa633e8725628cca0 : SUCCESS in 16m 46s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/09bcdef90d73424ea777dcbf83f78e6a : SUCCESS in 52m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fde3afe0cfd18c3fe8f7641fd79f2804fbfe838d","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-09-10 21:56:38.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c27a241a8c631a87514697e20b460005063fc018","revisions":{"1fe8b01c0f1d4a5d6f47bd1a131deff6b16248de":{"kind":"REWORK","_number":1,"created":"2026-09-10 13:56:55.000000000","uploader":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"ref":"refs/changes/16/1005016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/1005016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/1005016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/1005016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/1005016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/1005016/1"}}},"commit":{"parents":[{"commit":"abc1077f16e8d0818825618f64cdd7c0fb6edcf1","subject":"Try all allowed DNS addresses when importing image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/abc1077f16e8d0818825618f64cdd7c0fb6edcf1"}]}],"author":{"name":"Sebastian Krott","email":"sebastian.krott@sap.com","date":"2026-08-27 10:16:47.000000000","tz":120},"committer":{"name":"Sebastian Krott","email":"sebastian.krott@sap.com","date":"2026-09-10 13:52:31.000000000","tz":120},"subject":"trust_auth: Register missing session conf options","message":"trust_auth: Register missing session conf options\n\nThe `TokenRefresher` creates Keystone sessions via keystoneauth\u0027s\n`load_session_from_conf_options` to reauthenticate the trustee client.\nThis requires  `collect_timing` and `split_loggers` to be registered as\nadditional config options.\n\nRegistering these options explicitly requires manual maintenance\nwhenever the options required by keystoneauth change. To avoid running\ninto this issue again, we instead register all missing options\ndynamically.\n\nSince Glance reuses keystonemiddleware\u0027s `keystone_authtoken` group,\ndynamic registration at module-level would raise an exception if\n`keystonemiddleware.auth_token` was ever imported after `trust_auth`,\ne.g. for testing. We therefore register the missing options during\n`TokenRefresher` initialization.\n\nThe presumably cleanest fix would be the introduction of a separate\nconfig group for the Glance trustee/service user. Other OpenStack\nservices seem to take this approach. But this is a breaking change\nrequiring operators to adjust their configuration.\n\nNOTE:\n\nOther OpenStack services use keystoneauth\u0027s\n`register_session_conf_options` to register all the relevant config\noptions. Due to the reuse of the `keystone_authtoken` config group this\nwould raise `DuplicateOptError` for Glance because some session options\nare already registered by keystonemiddleware.\n\nIn more detail, Glance API initialization runs\n`common.config.load_paste_app`. If using a `glance-api-paste.ini`\npipeline with `authtoken` filter, the import of keystonemiddleware\u0027s\n`auth_token._opts` registers numerous options including `insecure`,\n`cafile`, `certfile` and `keyfile` for the `keystone_authtoken` config\ngroup. Any subsequent call to `register_session_conf_options` will fail.\n\nCloses-Bug: #1916052\nChange-Id: I6aafe6c1f32915299eb1b5152de9e3bac6038726\nSigned-off-by: Sebastian Krott \u003csebastian.krott@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fe8b01c0f1d4a5d6f47bd1a131deff6b16248de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fe8b01c0f1d4a5d6f47bd1a131deff6b16248de"}]},"branch":"refs/heads/master"},"671b4540771bb74b4adaf19ef056c2592cb23f55":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-10 16:23:07.000000000","uploader":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"ref":"refs/changes/16/1005016/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/1005016/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/1005016/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/1005016/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/1005016/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/1005016/2"}}},"commit":{"parents":[{"commit":"e3bbfef7d08b733b51b78a1384f7ba0042f9df2c","subject":"Add reproducer for upload failing when the user token expires","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3bbfef7d08b733b51b78a1384f7ba0042f9df2c"}]}],"author":{"name":"Sebastian Krott","email":"sebastian.krott@sap.com","date":"2026-08-27 10:16:47.000000000","tz":120},"committer":{"name":"Sebastian Krott","email":"sebastian.krott@sap.com","date":"2026-09-10 16:22:34.000000000","tz":120},"subject":"trust_auth: Register missing session conf options","message":"trust_auth: Register missing session conf options\n\nThe `TokenRefresher` creates Keystone sessions via keystoneauth\u0027s\n`load_session_from_conf_options` to reauthenticate the trustee client.\nThis requires  `collect_timing` and `split_loggers` to be registered as\nadditional config options.\n\nRegistering these options explicitly requires manual maintenance\nwhenever the options required by keystoneauth change. To avoid running\ninto this issue again, we instead register all missing options\ndynamically.\n\nSince Glance reuses keystonemiddleware\u0027s `keystone_authtoken` group,\ndynamic registration at module-level would raise an exception if\n`keystonemiddleware.auth_token` was ever imported after `trust_auth`,\ne.g. for testing. We therefore register the missing options during\n`TokenRefresher` initialization.\n\nThe presumably cleanest fix would be the introduction of a separate\nconfig group for the Glance trustee/service user. Other OpenStack\nservices seem to take this approach. But this is a breaking change\nrequiring operators to adjust their configuration.\n\nNOTE:\n\nOther OpenStack services use keystoneauth\u0027s\n`register_session_conf_options` to register all the relevant config\noptions. Due to the reuse of the `keystone_authtoken` config group this\nwould raise `DuplicateOptError` for Glance because some session options\nare already registered by keystonemiddleware.\n\nIn more detail, Glance API initialization runs\n`common.config.load_paste_app`. If using a `glance-api-paste.ini`\npipeline with `authtoken` filter, the import of keystonemiddleware\u0027s\n`auth_token._opts` registers numerous options including `insecure`,\n`cafile`, `certfile` and `keyfile` for the `keystone_authtoken` config\ngroup. Any subsequent call to `register_session_conf_options` will fail.\n\nCloses-Bug: #1916052\nChange-Id: I6aafe6c1f32915299eb1b5152de9e3bac6038726\nSigned-off-by: Sebastian Krott \u003csebastian.krott@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/671b4540771bb74b4adaf19ef056c2592cb23f55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/671b4540771bb74b4adaf19ef056c2592cb23f55"}]},"branch":"refs/heads/master"},"c27a241a8c631a87514697e20b460005063fc018":{"kind":"REWORK","_number":3,"created":"2026-09-10 18:49:56.000000000","uploader":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"ref":"refs/changes/16/1005016/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/1005016/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/1005016/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/1005016/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/1005016/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/1005016/3"}}},"commit":{"parents":[{"commit":"e3bbfef7d08b733b51b78a1384f7ba0042f9df2c","subject":"Add reproducer for upload failing when the user token expires","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3bbfef7d08b733b51b78a1384f7ba0042f9df2c"}]}],"author":{"name":"Sebastian Krott","email":"sebastian.krott@sap.com","date":"2026-08-27 10:16:47.000000000","tz":120},"committer":{"name":"Sebastian Krott","email":"sebastian.krott@sap.com","date":"2026-09-10 18:49:27.000000000","tz":120},"subject":"trust_auth: Register missing session conf options","message":"trust_auth: Register missing session conf options\n\nThe `TokenRefresher` creates Keystone sessions via keystoneauth\u0027s\n`load_session_from_conf_options` to reauthenticate the trustee client.\nThis requires  `collect_timing` and `split_loggers` to be registered as\nadditional config options.\n\nRegistering these options explicitly requires manual maintenance\nwhenever the options required by keystoneauth change. To avoid running\ninto this issue again, we instead register all missing options\ndynamically.\n\nSince Glance reuses keystonemiddleware\u0027s `keystone_authtoken` group,\ndynamic registration at module-level would raise an exception if\n`keystonemiddleware.auth_token` was ever imported after `trust_auth`,\ne.g. for testing. We therefore register the missing options during\n`TokenRefresher` initialization.\n\nThe presumably cleanest fix would be the introduction of a separate\nconfig group for the Glance trustee/service user. Other OpenStack\nservices seem to take this approach. But this is a breaking change\nrequiring operators to adjust their configuration.\n\nNOTE:\n\nOther OpenStack services use keystoneauth\u0027s\n`register_session_conf_options` to register all the relevant config\noptions. Due to the reuse of the `keystone_authtoken` config group this\nwould raise `DuplicateOptError` for Glance because some session options\nare already registered by keystonemiddleware.\n\nIn more detail, Glance API initialization runs\n`common.config.load_paste_app`. If using a `glance-api-paste.ini`\npipeline with `authtoken` filter, the import of keystonemiddleware\u0027s\n`auth_token._opts` registers numerous options including `insecure`,\n`cafile`, `certfile` and `keyfile` for the `keystone_authtoken` config\ngroup. Any subsequent call to `register_session_conf_options` will fail.\n\nCloses-Bug: #1916052\nChange-Id: I6aafe6c1f32915299eb1b5152de9e3bac6038726\nSigned-off-by: Sebastian Krott \u003csebastian.krott@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c27a241a8c631a87514697e20b460005063fc018"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c27a241a8c631a87514697e20b460005063fc018"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
