)]}'
{"id":"openstack%2Fglance~1005371","triplet_id":"openstack%2Fglance~stable%2F2025.1~I335820470f4474bbec95efc05ca3b45a6a370d2b","project":"openstack/glance","branch":"stable/2025.1","hashtags":[],"change_id":"I335820470f4474bbec95efc05ca3b45a6a370d2b","subject":"Try all allowed DNS addresses when importing image","status":"MERGED","created":"2026-09-12 05:33:35.000000000","updated":"2026-09-12 13:11:14.000000000","submitted":"2026-09-12 13:09:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005371","cherry_pick_of_change":1005006,"cherry_pick_of_patch_set":1,"meta_rev_id":"480796603bed127ce9fcaf844c9a988469a66a16","_number":1005371,"virtual_id_number":1005371,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-12 13:09:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-09-12 11:09:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-09-12 11:09:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-12 05:33:35.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"CC"},{"updated":"2026-09-12 07:59:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b69c6b54d784aa6e48214cd5ff0e96d6fbc41ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-12 05:33:35.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"4a6dd5595f4870885cfef463a93f9eeda3a075db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 07:59:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2209bda9e124548883d10e7d2850413\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9bcf202cb5974854aa697d157fb7998c : FAILURE in 45m 18s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/00793a42f59b40ed9691ab7fffdd7b53 : SUCCESS in 1h 45m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f8ba441ff09454798a5ffbee2af94c4 : SUCCESS in 1h 01m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c48ae1fd5f0437a98fe65d3de707e91 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1208688508646b9a392f5210ea15073 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e765703d3a014185934a5fee4b0eec3a : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c01d68129f34f3297b7de1e35b73204 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73d504892eeb402299ef73d697ece320 : SUCCESS in 8m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/74943defcef24bdd97f4e8924f5d838b : SUCCESS in 12m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62eae67e38cf4c6e9abf73b29f7eff66 : SUCCESS in 8m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cf79aba2894746c8be3668555fd4b953 : SUCCESS in 9m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2ac1f2ce077b4f38bd70ee8eb8bfdc28 : SUCCESS in 1h 28m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d658d4cc3c2b428f8336971e11ab5570 : SUCCESS in 2h 12m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4e558c9fe587421397c6e48e5cd30139 : SUCCESS in 2h 08m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5ded79e426ac46e3b19cf832487a0cce : SUCCESS in 2h 20m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5ecaf646321c4daba1cadbb438b915ce : SUCCESS in 2h 11m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fb6b47ea28242fa881cf337c782d97f : SUCCESS in 46m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9d98cee2134d4637b9b715ad0581b937 : SUCCESS in 1h 56m 11s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/768bb05600ed4387a13d2dec426ae38c : SUCCESS in 1h 53m 20s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/311cfa70bdcc44549d84df707ec97331 : FAILURE in 35m 00s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/619056e534ae46eb8c95d634fed09b47 : FAILURE in 25m 35s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ed6aecf7fee04c3ab38c7a347d200499 : SUCCESS in 38m 28s","accounts_in_message":[],"_revision_number":1},{"id":"93cf4a2873d492131559b23eedb2c082916562de","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-12 11:09:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3ac92b82de0b00d62d444a501cd2fa2b3db2370","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 11:09:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f662f0e3ff5687279697628f73b7f078c368ce6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 13:09:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4395bdce97f459d852c6a34bf811b4b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/89a8e918088f4f19b26d237901858074 : FAILURE in 48m 26s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dca5bb11b5b543ea889b53c14063787a : SUCCESS in 1h 04m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62c12798093c4d84b205e842276b2245 : SUCCESS in 43m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfd3adb2323a43a0aff2cae1fa46a3ff : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f185b78a53844285b425d9ad215898ca : SUCCESS in 9m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0f6c21a415140abaf35b9b248f08e00 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a22ac58a88364b628482442a3096debf : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55fd7f4f7d4b4434bab5f9c271378736 : SUCCESS in 8m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d1829d6407d94561a0c7c1c5f6750c7b : SUCCESS in 9m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aabb948717124001a901639b831ea674 : SUCCESS in 9m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9829dc47e2d64adea8b4c3eec0fbc60d : SUCCESS in 10m 06s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1603f54b49c640f9ac3f378d902eb466 : SUCCESS in 1h 47m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/025dc413cc6a4e718ebd6bec66ed172f : SUCCESS in 1h 50m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c86674ca57f14765b38cc867ec048df7 : SUCCESS in 1h 27m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/850a9e78f87c4b57bdcdb315d873a683 : SUCCESS in 47m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7dae8f1e8868462fa0c877479e17a11b : SUCCESS in 1h 28m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/84859cb913394b3dbd4421d26b8d40a1 : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":1},{"id":"bf5c52dca6e44fcfc872fa2042c47d4634493730","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 13:09:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"480796603bed127ce9fcaf844c9a988469a66a16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 13:11:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a3764745266416f9014139ab9d8341a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873c0165a7a04487bc22ae30a2f975b8 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0558ab58e1414ce994e32d444c447533 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0341c8222da563d57b495d04dda1141a5b02c1f","revisions":{"e0341c8222da563d57b495d04dda1141a5b02c1f":{"kind":"REWORK","_number":1,"created":"2026-09-12 05:33:35.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/71/1005371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/71/1005371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/1005371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/1005371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/1005371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/71/1005371/1"}}},"commit":{"parents":[{"commit":"d964f48c5c780ab4bd727b78e38246bbed2530f1","subject":"Merge \"Block restricted hosts when adding HTTP image locations\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d964f48c5c780ab4bd727b78e38246bbed2530f1"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-09-04 15:41:22.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-09-12 05:33:35.000000000","tz":0},"subject":"Try all allowed DNS addresses when importing image","message":"Try all allowed DNS addresses when importing image\n\ninstead of picking the first of returned DNS addresses of the\nhost from where to web-download the image, consider all addresses\nreturned by DNS that pass filtering, and try them one by one.\n\nCloses-Bug: #2166488\nChange-Id: I335820470f4474bbec95efc05ca3b45a6a370d2b\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n(cherry picked from commit abc1077f16e8d0818825618f64cdd7c0fb6edcf1)\n(cherry picked from commit 0eccf0e972d3606ab005d2b5bf7ccf6a955d5491)\n(cherry picked from commit 2b36783ab6b06d8e8b1491ce1b34fa9e8d438c76)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0341c8222da563d57b495d04dda1141a5b02c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0341c8222da563d57b495d04dda1141a5b02c1f"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"base":"10ccbcdbd786cb73074704e900c47ba61c8abd63","ours":"d964f48c5c780ab4bd727b78e38246bbed2530f1","theirs":"2b36783ab6b06d8e8b1491ce1b34fa9e8d438c76","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
