)]}'
{"id":"openstack%2Fglance~1005535","triplet_id":"openstack%2Fglance~master~I09f976480a1b5e5111e544a3b8030e4cc403e6fe","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"I09f976480a1b5e5111e544a3b8030e4cc403e6fe","subject":"Deprecate glance-control","status":"NEW","created":"2026-09-14 12:29:35.000000000","updated":"2026-09-14 14:29:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f90c028890fad03ce5d791fd18bfea389fd9c0d4","_number":1005535,"virtual_id_number":1005535,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-14 14:29:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-14 12:43:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-14 14:29:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd672bdbf8d8a041e70f74ba2f431287d37dd2cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 12:29:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8e1116a7838a8d7ecaa14de356ec3419605eec9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 12:43:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cda6f2c93c844306b12cc8255a735662\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bca8c8c5fc9645f39a2dcb4b1760b679 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/75211fa54fb646da9b4fb086c57ee8af : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fc4bf09e394e4f7cae8e42bf4210579b : SUCCESS in 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f90c028890fad03ce5d791fd18bfea389fd9c0d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 14:29:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/309f76f8d24241e6b25da926efe19d5d\n\n- grenade https://zuul.opendev.org/t/openstack/build/22db54f2df324d62b1967d3da72f177e : SUCCESS in 1h 05m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/96018177de3747938084344d4766172d : SUCCESS in 1h 11m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/db7176616af34b1b82e84fad6b3c693d : SUCCESS in 1h 24m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61435f43a3d848b2b8787cb8a37a7ff1 : SUCCESS in 1h 06m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/268c46e079df4409b6a24562838633c5 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/486770ec7bd546c79c4e725d086448d0 : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2c65a55698c4c1baa776a0c1a0e8390 : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/56df2d802001473d82046f405b04b317 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2099e5afbfbe45bca1edb025ac4a9979 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd4693b95a9f41c4be5045544275f9af : SUCCESS in 9m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d9cc30641b494b4380410895f0b5c192 : SUCCESS in 8m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ce0f76708bde40009b4fddacba45142c : SUCCESS in 7m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/91d65efe73db43d892beb9434df7ac22 : SUCCESS in 1h 27m 52s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/185b1643659f4c41ac45ed6bc2af9a7e : SUCCESS in 1h 35m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6b4e0182329b4e5fa91629e87177b5ef : SUCCESS in 1h 54m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/400a6b5c70984475a42e9647e775f5cd : SUCCESS in 59m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/82af8000fd2e41108a665deaf4fd4b88 : SUCCESS in 1h 37m 25s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/349c0737d8a84f32b53603a6daa388eb : FAILURE in 13m 16s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/0b038dca210e43b896fed7e1a08d3ead : FAILURE in 28m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/05a6e1b19e5f48a6a2166e7296dd17c7 : SUCCESS in 37m 25s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/07fef6beb7ea4c2f9104dd0cd90c0efc : SUCCESS in 37m 04s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"498ea9f85a0a57854d3042e011454018419d259f","revisions":{"498ea9f85a0a57854d3042e011454018419d259f":{"kind":"REWORK","_number":1,"created":"2026-09-14 12:29:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/1005535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/1005535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/1005535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/1005535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/1005535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/1005535/1"}}},"commit":{"parents":[{"commit":"7b87a481f919eb7cee0637a21c40990283f8d354","subject":"Remove glance-api support by glance-control command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b87a481f919eb7cee0637a21c40990283f8d354"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 12:28:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 12:28:06.000000000","tz":540},"subject":"Deprecate glance-control","message":"Deprecate glance-control\n\nIt currently supports glance-scrubber, because the standalone\nglance-api service was removed, in favor of the wsgi server.\n\nIn major distributions and deployment tools we use the script directly\ninstead of using this command.\n\nChange-Id: I09f976480a1b5e5111e544a3b8030e4cc403e6fe\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/498ea9f85a0a57854d3042e011454018419d259f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/498ea9f85a0a57854d3042e011454018419d259f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
