)]}'
{"id":"openstack%2Fglance~1005674","triplet_id":"openstack%2Fglance~stable%2F2025.2~I50909e6bdb3227ca99b7eba642546da791f9552a","project":"openstack/glance","branch":"stable/2025.2","topic":"glance-service-api-stable/2025.2","attention_set":{},"removed_from_attention_set":{"34034":{"account":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"last_update":"2026-09-17 18:14:53.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-09-17 15:52:32.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-09-17 13:52:27.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I50909e6bdb3227ca99b7eba642546da791f9552a","subject":"Fix glance service policy rule","status":"MERGED","created":"2026-09-15 08:38:46.000000000","updated":"2026-09-17 18:16:05.000000000","submitted":"2026-09-17 18:14:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005674-glance-service-api-stable/2025.2","cherry_pick_of_change":958715,"cherry_pick_of_patch_set":5,"meta_rev_id":"019165fb8122ac52d349327f2266945d1702c087","_number":1005674,"virtual_id_number":1005674,"owner":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-17 18:14:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2026-09-17 13:54:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2026-09-17 15:52:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-09-17 13:52:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2026-09-17 15:52:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 08:38:46.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2026-09-15 08:53:44.000000000","updated_by":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"reviewer":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"state":"CC"},{"updated":"2026-09-15 11:03:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 05:39:54.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-09-16 05:39:54.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-09-17 13:54:33.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"}],"messages":[{"id":"934cb64b1bee43bb134546a5e588d02850b71171","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-09-15 08:38:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* glance/tests/functional/v2/test_images.py","accounts_in_message":[],"_revision_number":1},{"id":"8691b02ea0d6b64ec169ef9d0c6157503275f573","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-09-15 08:45:33.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bef695c54c407ad92ad2dd65d9976d82b5032bca","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-09-15 08:45:44.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"32536acee25bfe5d0ed7df24448c773207f6c899","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-09-15 09:38:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e2696fe313812171db1961d1a19d3cd65251ed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 11:03:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d84acd34bfc544b287b005b86e11a42f\n\n- grenade https://zuul.opendev.org/t/openstack/build/8d977f95cdcf45d78b0d5e8d8d9be141 : SUCCESS in 1h 06m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c6dabc0ed1754f5e8792250c8390c82b : SUCCESS in 1h 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74c86f8a83d14a8e9c8b8f498d7a6a79 : SUCCESS in 1h 03m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b9f3b2bb83841aeb91e0d28cd0e292d : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6005fa3e17044eb4927ee8d6ae3d505f : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43c8b7e557fa4735afb9de374a34e593 : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b3972b9b97a4895984fabb8a9fa3b16 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae5109145134f43882ef72b51a5724a : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91f007c966c7466795b76bcf0fad121a : SUCCESS in 7m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/331436c8365e425e9041451c6a80b654 : FAILURE in 8m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/27bf13bf77da410295256bc9a3b95f64 : FAILURE in 5m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d816a61c2183478db3dcc5374a131c76 : SUCCESS in 1h 37m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/15b04ea8e50443ddbaf365b6df115ed5 : SUCCESS in 2h 05m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e07e22992ee14f4a95c29c0e08f9a3b5 : FAILURE in 1h 51m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d6f1f1baae4e4ba8894292f6e48d3d7a : SUCCESS in 2h 11m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc004da0c16d41e8ae048cda0ed88590 : SUCCESS in 59m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0c2fc318f6ae44cc91cf8850a55538a9 : SUCCESS in 2h 07m 57s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9d50e2b41bfe42a2ae3344d1ef287636 : FAILURE in 8m 12s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/85ed313f917a4c8f90b2a772ca36db8d : FAILURE in 24m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e6e4ffaca7f745d8a3bc9db325bc0441 : SUCCESS in 38m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e09a837277384fa4ae6d89beb236e79f : SUCCESS in 1h 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36131d7cea712dc4166b501af6196d60c34df522","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-09-15 11:05:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2fbf38c3a3f84d1a63a8b30868f5a28b9ca2a63e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 13:29:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfe38622de3a4ec6a50803a2ae43ead6\n\n- grenade https://zuul.opendev.org/t/openstack/build/94c9259807a348e4b37c0ba6a1aea033 : SUCCESS in 59m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32c48a87f6464bac84fa7dde4cda59e9 : SUCCESS in 1h 41m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e0a70b6b3af496281392bd62d86e687 : SUCCESS in 1h 02m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78785f1146cc44efa6329233202779eb : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7558c176f1774e3aafe61921d6d97de8 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f91c8e01aada4c2398fbd68c15aa829d : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1cbd3c669f14c5585f1ea646b277e58 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b5ef1ac48a4eb08b6d1b20004228e3 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f36387459e74284a407866ac34fe88a : SUCCESS in 3m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cac29f30422243fc8a606650ef7f4bf4 : FAILURE in 5m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d8df92c6af2f45af90e75f4195c745a9 : FAILURE in 7m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1a43fc35ebf74b8a8a44fb1481504045 : SUCCESS in 1h 57m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9457ed53783f459080ff233f866addde : SUCCESS in 2h 17m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/16580e2aa7fd43b8b30b7c7a3aeb44b7 : SUCCESS in 1h 34m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1d9352be04f349848310ed8ced72e5e2 : SUCCESS in 2h 03m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f9cb5c2349f475dbc964e4ca4861cd1 : SUCCESS in 56m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5e50f8e7ac304c64b353769d1d8e04a7 : SUCCESS in 1h 01m 20s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6622e4aea90149e288f908b12a6b44b2 : FAILURE in 12m 22s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/831ac3f5ccba4b0fb9c614674f716d0f : FAILURE in 27m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/de2af46a6c5c46d4b88815c644821a30 : SUCCESS in 32m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/023e6ffa77ea44bca5102db5f041ef04 : SUCCESS in 1h 09m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1f7cff38f45a1e3c2c05e743d6b27ce5df6ca21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-09-15 18:08:37.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d6f4f0f203a74dc4ffe6c9d13b74969d5455cbc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-09-15 18:09:28.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"f44f96a9c84b002fb455938113d9dbfb26d7b81f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 20:47:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c10c9b5ee2aa45b3b875896ff1e727bf\n\n- grenade https://zuul.opendev.org/t/openstack/build/74ef7059685e4eeda1d06659ca908eef : SUCCESS in 32m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56188e81de084269acea7cf20aca511b : SUCCESS in 1h 39m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ae2504ae56d4363aede8683968417fa : SUCCESS in 34m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fb47bbb263d44b7811ff458a851cae5 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4994dd54dde3470e9beb4b7d4c9aaa2a : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2cc5845a860426f9f3e95c60f1f7f98 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f315a91853f4ce19452e7e9e8f3e603 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2e5f651fb6945678b77d035c1e4e4ed : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22d21dc75bf84acba356f3e5d84b9345 : SUCCESS in 5m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1427f90dd6de4a91af0097e5c1b70775 : SUCCESS in 9m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/21a99965eb71494bb39c4c5959e53819 : SUCCESS in 8m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c9985bd7d73040e8a45a8d12f6e251c7 : SUCCESS in 1h 38m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a4b82e1a02b948d0ab4e7718701ee6f5 : SUCCESS in 2h 02m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4ecdb23704d741848d6648e92aa36b7f : SUCCESS in 1h 40m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3a143594f8fc440098e3b33f453b2ce8 : SUCCESS in 2h 22m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5b6b2b1436941cf8c15c081df156611 : SUCCESS in 31m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2bd3386c2e3747f4bae8293e2137565e : SUCCESS in 1h 46m 46s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d32c8a867a8145839576bf85464f095b : FAILURE in 15m 22s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b3592837cf1742de96b4b46c6c42bff5 : FAILURE in 14m 50s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a7f9251d7d85427ea6834050eda77676 : SUCCESS in 16m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/101322bee1c141e18f57208828966f37 : SUCCESS in 1h 02m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8661461ece0ad9623fe2204c4e606bd7cb39aaee","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-09-17 13:52:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"005325c9e8b0dc91c2df4c650a0f19d1b7d27865","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-09-17 13:54:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8bf0ac8e5577ae70ee25820163cfa6b501e32cbe","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-17 15:52:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d6b9a5bc99100277ab24a78225b534c9679bcf60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 15:53:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"90186fcee3a6a7dc3d2017e74ec2a042f5091d34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 18:14:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9d71fb817a94721a6ae6d20be571a95\n\n- grenade https://zuul.opendev.org/t/openstack/build/2a76fe6bf52b423d8bf1201644c1a218 : SUCCESS in 1h 02m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f0bd002a9c9a47c28c22914fff810892 : SUCCESS in 1h 44m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e86712c71294d3e9c5f0989121c541b : SUCCESS in 35m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da5a32fab3e44c3e9af18ccc05cb5d43 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4b824c6badf46d79673e8e0d184b235 : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/690074c0c18e49e081c77ccc5d26aa53 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4090f3179d9e46d2af016cad57b28c8c : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8047262256f44d4bb7cf99c822b98b95 : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4acc1a67b7fa496aa3b466498f142340 : SUCCESS in 8m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/914cca800c894f47b219a3b775aaf15b : SUCCESS in 8m 57s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d2b5203c9ebd408d8142282ad472b159 : SUCCESS in 1h 59m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4527afd7a1544836ba00665ef4c09539 : SUCCESS in 2h 07m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fbf2a922b8e44318b8e2bd0824eddaf : SUCCESS in 56m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3c212c6a5c74481795f6f664e2609987 : SUCCESS in 1h 48m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bf81e57bbf0c4bd2b75c27a880da6ff8 : SUCCESS in 33m 11s","accounts_in_message":[],"_revision_number":4},{"id":"64a76532dc265ce7a991040d7c8d94a25df01a87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 18:14:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"019165fb8122ac52d349327f2266945d1702c087","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 18:16:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0797121f3c0943358c3ae8974453fd09\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a565904ab8d4101ac082034cf43cc86 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bfaf159d825490e946d3fd2dc6e2e57 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"890770df675e5427f36235ada7ec42c94b517865","revisions":{"66d58108b865cc684d8ca8691bbf25a7f229381c":{"kind":"REWORK","_number":1,"created":"2026-09-15 08:38:46.000000000","uploader":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"ref":"refs/changes/74/1005674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/1005674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/1005674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/1005674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/1005674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/1005674/1"}}},"commit":{"parents":[{"commit":"2b36783ab6b06d8e8b1491ce1b34fa9e8d438c76","subject":"Try all allowed DNS addresses when importing image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b36783ab6b06d8e8b1491ce1b34fa9e8d438c76"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-28 03:19:27.000000000","tz":0},"committer":{"name":"Franciszek Przewoźny","email":"przewozny.franciszek@gmail.com","date":"2026-09-15 08:38:46.000000000","tz":0},"subject":"Fix glance service policy rule","message":"Fix glance service policy rule\n\nGlance service APIs are default to \u0027service_roles: service\u0027\n\n- https://github.com/openstack/glance/blob/6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17/glance/policies/base.py#L116\n\nThe issue here is the service token, which is sent from the\nservice for the user token expiry case but glance uses that\nservice token (keystonemiddleware sets the service token roles\nin Requestcontext in \u0027service_roles\u0027 field) for RBAC, which\nis not correct.\n\nThe OpenStack services communicate with each other by\npassing the user token and service token wrapped in\nkeystoneauth\u0027s ServiceTokenAuthWrapper. The only purpose\nof passing the service token is for long-running\noperations and in case the user token gets expired.\n\nFor RBAC, we need to check if a user token has the \u0027service\u0027\nrole or not. Service needs to load the configured user auth\nplugin (where the user should have the \u0027service\u0027 role) from\nkeystoneauth and pass that to the other services (for example,\ncinder change depends-on) and glance will use that user role\nto verify the policy permission. To fix that, we need to make\nthe service APIs default to ``role:service`` and not\n`service_role`:`service`.\n\nThis commit does one more change. Cinder does not have the\nway to configure the glance service user, we are adding the\nnew config in this release. For backward compatibility,\nwe need to allow admin access in service policy rule. In\nfuture release (after one SLURP release), we cna remove\nthe admin access.\n\nCloses-Bug: #2121622\n\nCo-Authored-By: : Sean Mooney \u003cwork@seanmooney.info\u003e\n\nChange-Id: I50909e6bdb3227ca99b7eba642546da791f9552a\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n(cherry picked from commit 301d8f580f4ce6a4f6089f877c0f200ff4ff633b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66d58108b865cc684d8ca8691bbf25a7f229381c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66d58108b865cc684d8ca8691bbf25a7f229381c"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17","ours":"2b36783ab6b06d8e8b1491ce1b34fa9e8d438c76","theirs":"301d8f580f4ce6a4f6089f877c0f200ff4ff633b","merge_strategy":"recursive","contains_conflicts":true}},"79e1681161053b08f0a5e1d06b25afbbe3994920":{"kind":"REWORK","_number":2,"created":"2026-09-15 08:45:33.000000000","uploader":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"ref":"refs/changes/74/1005674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/1005674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/1005674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/1005674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/1005674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/1005674/2"}}},"commit":{"parents":[{"commit":"2b36783ab6b06d8e8b1491ce1b34fa9e8d438c76","subject":"Try all allowed DNS addresses when importing image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b36783ab6b06d8e8b1491ce1b34fa9e8d438c76"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-28 03:19:27.000000000","tz":0},"committer":{"name":"Franciszek Przewoźny","email":"przewozny.franciszek@gmail.com","date":"2026-09-15 08:45:06.000000000","tz":0},"subject":"Fix glance service policy rule","message":"Fix glance service policy rule\n\nGlance service APIs are default to \u0027service_roles: service\u0027\n\n- https://github.com/openstack/glance/blob/6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17/glance/policies/base.py#L116\n\nThe issue here is the service token, which is sent from the\nservice for the user token expiry case but glance uses that\nservice token (keystonemiddleware sets the service token roles\nin Requestcontext in \u0027service_roles\u0027 field) for RBAC, which\nis not correct.\n\nThe OpenStack services communicate with each other by\npassing the user token and service token wrapped in\nkeystoneauth\u0027s ServiceTokenAuthWrapper. The only purpose\nof passing the service token is for long-running\noperations and in case the user token gets expired.\n\nFor RBAC, we need to check if a user token has the \u0027service\u0027\nrole or not. Service needs to load the configured user auth\nplugin (where the user should have the \u0027service\u0027 role) from\nkeystoneauth and pass that to the other services (for example,\ncinder change depends-on) and glance will use that user role\nto verify the policy permission. To fix that, we need to make\nthe service APIs default to ``role:service`` and not\n`service_role`:`service`.\n\nThis commit does one more change. Cinder does not have the\nway to configure the glance service user, we are adding the\nnew config in this release. For backward compatibility,\nwe need to allow admin access in service policy rule. In\nfuture release (after one SLURP release), we cna remove\nthe admin access.\n\nCloses-Bug: #2121622\n\nCo-Authored-By: : Sean Mooney \u003cwork@seanmooney.info\u003e\n\nChange-Id: I50909e6bdb3227ca99b7eba642546da791f9552a\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n(cherry picked from commit 301d8f580f4ce6a4f6089f877c0f200ff4ff633b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/79e1681161053b08f0a5e1d06b25afbbe3994920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/79e1681161053b08f0a5e1d06b25afbbe3994920"}]},"branch":"refs/heads/stable/2025.2"},"180b271f4359297134934d4ccdb014f0ceea5915":{"kind":"REWORK","_number":3,"created":"2026-09-15 18:08:37.000000000","uploader":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"ref":"refs/changes/74/1005674/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/1005674/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/1005674/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/1005674/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/1005674/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/1005674/3"}}},"commit":{"parents":[{"commit":"2b36783ab6b06d8e8b1491ce1b34fa9e8d438c76","subject":"Try all allowed DNS addresses when importing image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b36783ab6b06d8e8b1491ce1b34fa9e8d438c76"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-28 03:19:27.000000000","tz":0},"committer":{"name":"Franciszek Przewoźny","email":"przewozny.franciszek@gmail.com","date":"2026-09-15 18:08:37.000000000","tz":0},"subject":"Fix glance service policy rule","message":"Fix glance service policy rule\n\nGlance service APIs are default to \u0027service_roles: service\u0027\n\n- https://github.com/openstack/glance/blob/6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17/glance/policies/base.py#L116\n\nThe issue here is the service token, which is sent from the\nservice for the user token expiry case but glance uses that\nservice token (keystonemiddleware sets the service token roles\nin Requestcontext in \u0027service_roles\u0027 field) for RBAC, which\nis not correct.\n\nThe OpenStack services communicate with each other by\npassing the user token and service token wrapped in\nkeystoneauth\u0027s ServiceTokenAuthWrapper. The only purpose\nof passing the service token is for long-running\noperations and in case the user token gets expired.\n\nFor RBAC, we need to check if a user token has the \u0027service\u0027\nrole or not. Service needs to load the configured user auth\nplugin (where the user should have the \u0027service\u0027 role) from\nkeystoneauth and pass that to the other services (for example,\ncinder change depends-on) and glance will use that user role\nto verify the policy permission. To fix that, we need to make\nthe service APIs default to ``role:service`` and not\n`service_role`:`service`.\n\nThis commit does one more change. Cinder does not have the\nway to configure the glance service user, we are adding the\nnew config in this release. For backward compatibility,\nwe need to allow admin access in service policy rule. In\nfuture release (after one SLURP release), we cna remove\nthe admin access.\n\nCloses-Bug: #2121622\n\nCo-Authored-By: : Sean Mooney \u003cwork@seanmooney.info\u003e\n\nChange-Id: I50909e6bdb3227ca99b7eba642546da791f9552a\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n(cherry picked from commit 301d8f580f4ce6a4f6089f877c0f200ff4ff633b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/180b271f4359297134934d4ccdb014f0ceea5915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/180b271f4359297134934d4ccdb014f0ceea5915"}]},"branch":"refs/heads/stable/2025.2"},"890770df675e5427f36235ada7ec42c94b517865":{"kind":"REWORK","_number":4,"created":"2026-09-15 18:09:28.000000000","uploader":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"ref":"refs/changes/74/1005674/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/1005674/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/1005674/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/1005674/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/1005674/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/1005674/4"}}},"commit":{"parents":[{"commit":"2b36783ab6b06d8e8b1491ce1b34fa9e8d438c76","subject":"Try all allowed DNS addresses when importing image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b36783ab6b06d8e8b1491ce1b34fa9e8d438c76"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-28 03:19:27.000000000","tz":0},"committer":{"name":"Franciszek Przewoźny","email":"przewozny.franciszek@gmail.com","date":"2026-09-15 18:09:28.000000000","tz":0},"subject":"Fix glance service policy rule","message":"Fix glance service policy rule\n\nGlance service APIs are default to \u0027service_roles: service\u0027\n\n- https://github.com/openstack/glance/blob/6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17/glance/policies/base.py#L116\n\nThe issue here is the service token, which is sent from the\nservice for the user token expiry case but glance uses that\nservice token (keystonemiddleware sets the service token roles\nin Requestcontext in \u0027service_roles\u0027 field) for RBAC, which\nis not correct.\n\nThe OpenStack services communicate with each other by\npassing the user token and service token wrapped in\nkeystoneauth\u0027s ServiceTokenAuthWrapper. The only purpose\nof passing the service token is for long-running\noperations and in case the user token gets expired.\n\nFor RBAC, we need to check if a user token has the \u0027service\u0027\nrole or not. Service needs to load the configured user auth\nplugin (where the user should have the \u0027service\u0027 role) from\nkeystoneauth and pass that to the other services (for example,\ncinder change depends-on) and glance will use that user role\nto verify the policy permission. To fix that, we need to make\nthe service APIs default to ``role:service`` and not\n`service_role`:`service`.\n\nThis commit does one more change. Cinder does not have the\nway to configure the glance service user, we are adding the\nnew config in this release. For backward compatibility,\nwe need to allow admin access in service policy rule. In\nfuture release (after one SLURP release), we cna remove\nthe admin access.\n\nCloses-Bug: #2121622\n\nCo-Authored-By: : Sean Mooney \u003cwork@seanmooney.info\u003e\n\nChange-Id: I50909e6bdb3227ca99b7eba642546da791f9552a\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n(cherry picked from commit 301d8f580f4ce6a4f6089f877c0f200ff4ff633b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/890770df675e5427f36235ada7ec42c94b517865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/890770df675e5427f36235ada7ec42c94b517865"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
