)]}'
{"id":"openstack%2Fglance~1007271","triplet_id":"openstack%2Fglance~master~I5c22a7a4af30a6d99657cc8a45ff9aef4f24287e","project":"openstack/glance","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-10-08 10:02:03.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5c22a7a4af30a6d99657cc8a45ff9aef4f24287e","subject":"Use oslo.utils to format IPv6 address","status":"NEW","created":"2026-09-24 16:23:38.000000000","updated":"2026-10-08 12:02:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4e290a83e09a995eeab0fac3ff4522370eaba9d3","_number":1007271,"virtual_id_number":1007271,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-08 12:02:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2026-09-29 07:40:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2026-09-25 15:13:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-24 17:03:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-24 18:27:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-25 15:13:44.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-09-29 07:40:03.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"},{"updated":"2026-09-29 07:40:13.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"}],"messages":[{"id":"74335964c09854a4e4e7734c976ab2fdd97c10f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-24 16:23:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfcd226f6ac2575ba00905f91f859871751e7892","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 17:03:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99cd2c90653842ddb1252d8ba13b8a14\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/acdf5e8ce9fc44e19375083d6f08f5ef : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4b4f6f1aec584dd6819b8a261f77db77 : SUCCESS in 9m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86de3ab05987a5fb1b8ce5183521ec010951669f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 18:27:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f776d1a8f6f14e99a96b6b4a25094923\n\n- grenade https://zuul.opendev.org/t/openstack/build/f56f78685ecc4cb5a657d46a31f30cfb : SUCCESS in 47m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0cd063dc80634e8e979ccf716c6bf90a : SUCCESS in 1h 09m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69d2692a07af4478bd79cab12412181c : SUCCESS in 1h 22m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8116225db4fb474ea8e21225464a94fe : SUCCESS in 28m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e51744882f04d2bbbb369e6716ca62c : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27da10ab7e9c459c81f61e175de2b37b : SUCCESS in 5m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/932d3596880e436c83614c0e8aab0b8a : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5195d3cd8aeb4035b9d8e4dd310dbca7 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f195acf74b94df0ba7a9e14c4a2fb66 : SUCCESS in 9m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6195129305764f778634e41082a44891 : SUCCESS in 8m 18s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7b49d988dd8f4e8a876a80db115bb1dc : SUCCESS in 8m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c9bf135768af41f0a2184e5efe54590b : SUCCESS in 1h 37m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/89337dbabb9348eca85c571ecd86363b : SUCCESS in 1h 11m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/74839f33d7b04d07923bf76eab38e27f : FAILURE in 1h 23m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/307e9674871947d38f4fc1f7a3971303 : SUCCESS in 55m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/58d52b1a36394fa389670d9a5344e605 : SUCCESS in 1h 44m 22s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/8f1b1537e6a8403ea98a87d88c405383 : FAILURE in 28m 15s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/825a364a00b94cfca57d15aa33a0ed37 : FAILURE in 16m 49s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6ea7f32d501f4e64a77584aa5dce7934 : SUCCESS in 37m 34s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/b7df5ac1b4da457580472245644d970f : SUCCESS in 58m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b924aaa027ed68150bfc10a9cbe45bae5981dc49","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-09-25 15:13:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"73fb840bd94c8857d0139475348e33f8321d2898","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-09-25 15:13:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b175acfd7abaac55b614dc0c1e40018dedb00e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 15:26:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8879db1c4af41fdb9676189675dc5f7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/09d7a7069850496e9ce0a7212ed0b9a3 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/915752a0fba64de5b86ebd9973a817ee : SUCCESS in 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb91461f3cd0ed3499b06b5e3fbe2ec732ee05aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 17:30:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83f0705fdf064fa5b4a129f9f71aee81\n\n- grenade https://zuul.opendev.org/t/openstack/build/defaad0a12324d7294be9cc19255c77f : SUCCESS in 1h 07m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5c5f51f79ce942df8b0278e44da64761 : SUCCESS in 38m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b6698edc7d34485b88afb42fb953325a : FAILURE in 1h 02m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/366ee3d2faad4bf3ab29aeec7a7a5d3d : SUCCESS in 58m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17633147bc2c4815b9b54f042db8eddb : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e3c829aa8344f58beb674ff1d46ff64 : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/67f31765a1f64a318bdbf426b530160a : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb393ffb9f9404fa21548ca43283a20 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/303531dbb68947a792461303b931908e : SUCCESS in 7m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/143f2580f70849e6a2f00287cba12c88 : SUCCESS in 7m 03s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/67bba4ab7b884265ba4c0bdcf8cf0803 : SUCCESS in 5m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a6208e138f0940feb2554e56d561acbf : SUCCESS in 52m 42s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8bcf320c41d740d58b9557a06cfc04b9 : FAILURE in 45m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/af394f080f3049859d53c745e5807b4c : SUCCESS in 2h 10m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c4963ab368f4a3ea1d40214f6b942e0 : SUCCESS in 58m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d90c26c14ecf45ec89942d0739336b49 : SUCCESS in 1h 36m 02s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/8840cc705abe4a9cb87a84f7acda3c44 : FAILURE in 31m 28s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/c6e6f96bd62d4c8390500e5922e6a8b3 : FAILURE in 13m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/839057fcb2644f5eb126a56c5fb5767a : SUCCESS in 46m 06s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/61324959288c4c5d90931b4316256c7d : SUCCESS in 31m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b67e248840006979b146f10f90de6e6ae7167fe","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-09-29 07:40:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d585479e62e2423ed29aa8d368fde4b0de78a0b9","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-09-29 07:40:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"33229cd3e7bb8308f45d7c68924522b814620929","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 07:54:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62b1e49439d04139b144997e734c85b0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/519eab50ea5447f48905100474b5d8d0 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/026e429bf8014a2c8b1d2dda822f294b : SUCCESS in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a51af0155eeac942146ac41d86671337d898030e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 10:02:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d06e300c86974ae7b6b744d92e61fdb7\n\n- grenade https://zuul.opendev.org/t/openstack/build/529e9a85c8194e79b955ecf706c4abd6 : SUCCESS in 1h 07m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/da7ac5513cf44b81b3ec60d914562ccc : SUCCESS in 31m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9db67c61f56a42d38f1358e5071972a7 : SUCCESS in 1h 26m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ed9edbccfad4b33969dbb8640341fc1 : SUCCESS in 28m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b0071b4ab624250a5c9ca02ec492efa : SUCCESS in 6m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c41ddc96721147608408f213d477b189 : SUCCESS in 5m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/129003861f1845e1b1a3b28fe4724a06 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b995c76722914c09b57eb0fd2d994cce : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/144dfdf39d8d4455a2ff5cc29b1181d3 : SUCCESS in 9m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/efa28b330fc348b186e665db16ef2743 : SUCCESS in 7m 26s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/acd10098b29c4938b018d1270bc5b008 : SUCCESS in 4m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/427a73928a434ae6b359be210d42e980 : SUCCESS in 1h 19m 54s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/25b72cad3088465abd2dde89060f673c : FAILURE in 1h 35m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/196ae6afda324ef5a1bd78233037a388 : SUCCESS in 2h 04m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c25b60ff6e6f49d4bf05402c85ea752f : SUCCESS in 56m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ce2e731efeba4d0abf6b9501a6e881c5 : SUCCESS in 1h 38m 01s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/fdf3a28ae5da45a783c0922691efc077 : FAILURE in 28m 26s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/88ed39e41aea4ef59f7018a8a63ae2a0 : FAILURE in 28m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0c1f2956ef5a4cdfb40959efc3ff1101 : SUCCESS in 39m 09s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/5a5b981a81ed4d70af1de716430f2fcd : SUCCESS in 50m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fae3375e155a1abf255af14718adfda5c63c2a6a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-10-08 10:02:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"72855e3f9f6ab81a249852d9a6f0eba6bb0e3441","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-08 10:14:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8c55af886e642d89ff3ea009342c858\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/40a8e8d7cec64511a7fee82518d0ffde : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6e95be6381034558b2af9360ce4193a5 : SUCCESS in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e290a83e09a995eeab0fac3ff4522370eaba9d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-08 12:02:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b2934cee84a41d4a0d7c767070ff3fd\n\n- grenade https://zuul.opendev.org/t/openstack/build/cf0a29d24a46417ea3cd2c7b67e1a0bd : SUCCESS in 35m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5f28e51d61a147b2ad0280f9e96914e4 : SUCCESS in 37m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/479cb74a29794420896b0708018ab44a : SUCCESS in 1h 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21642f14e7534cd38447763075b19d1f : SUCCESS in 59m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277c2694872b4f2ea1d4502c78fcc202 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b93a6410ad6f4a3eaa4bbd70ed802a16 : SUCCESS in 6m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ed472fa988464a61a937382260c1e948 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47cfbb6d192c4d0d85f30db72512aea7 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8cf2404856045ecb4e9cbae32700663 : SUCCESS in 8m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d75cccc5f3d94f4e8fbdfb9cf67ee688 : SUCCESS in 5m 46s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/12bbb79d67c941d6ae37aa2263ee4e68 : SUCCESS in 10m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/60dea6b23df342c097caf065983a09c7 : SUCCESS in 1h 35m 51s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1f7174746dd049839d1f6c11bbec3ac3 : SUCCESS in 1h 06m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/416a2650dce145afa6e58cbbe4450bb6 : SUCCESS in 1h 55m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ab84a952a4846a0847e3611ab0b74e1 : SUCCESS in 1h 04m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/65fa50279c204e9397de9f504d8fece5 : SUCCESS in 1h 29m 44s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/e8918491a90d48338db90426b4187578 : FAILURE in 28m 37s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/08b29c21882041988c9a188640ff6fa7 : FAILURE in 12m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0e8ecf78f81f497fae1aa93b15336190 : SUCCESS in 22m 13s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/8cd843c5a3e6466aa0a1eeb42b84d9e3 : SUCCESS in 32m 13s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a27c78edd3c7c7a2efb3fa201fb19f4340f0218","revisions":{"5a27c78edd3c7c7a2efb3fa201fb19f4340f0218":{"kind":"REWORK","_number":1,"created":"2026-09-24 16:23:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/1007271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/71/1007271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/1007271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/1007271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/1007271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/71/1007271/1"}}},"commit":{"parents":[{"commit":"355014c46c7d4b0da466aa51f78aca3c0c4c6940","subject":"Merge \"Conditionally import defusedxml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/355014c46c7d4b0da466aa51f78aca3c0c4c6940"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-24 16:22:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-24 16:22:58.000000000","tz":540},"subject":"Use oslo.utils to format IPv6 address","message":"Use oslo.utils to format IPv6 address\n\nThe escape_ipv6 helper has been available since 3.29.0 .\n\nChange-Id: I5c22a7a4af30a6d99657cc8a45ff9aef4f24287e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5a27c78edd3c7c7a2efb3fa201fb19f4340f0218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5a27c78edd3c7c7a2efb3fa201fb19f4340f0218"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
