)]}'
{"id":"openstack%2Fglance~148595","triplet_id":"openstack%2Fglance~master~I45f6a8c59ec3040aaf06f8bab46d8001c44dac7a","project":"openstack/glance","branch":"master","topic":"bp/healthcheck-middleware","hashtags":[],"change_id":"I45f6a8c59ec3040aaf06f8bab46d8001c44dac7a","subject":"Healthcheck Middleware","status":"MERGED","created":"2015-01-20 15:28:37.000000000","updated":"2015-08-06 01:16:43.000000000","submitted":"2015-07-14 13:55:55.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c0904d89579dc65dc55391706dd9b2c68c80c4cf","_number":148595,"virtual_id_number":148595,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2015-07-14 13:55:55.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-08-06 01:16:43.000000000","post_submit":true,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":9751,"name":"IBM DB2 CI","email":"scecd@cn.ibm.com","username":"ibmdb2","tags":["SERVICE_USER"]},{"value":0,"date":"2015-07-07 09:53:35.000000000","_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},{"value":0,"_account_id":11356,"name":"Louis Taylor","email":"louis@kragniz.eu","username":"kragniz"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"value":0,"_account_id":7487,"name":"hzrandd","email":"82433422@qq.com","username":"hzrandd"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"all":[{"value":2,"date":"2015-07-09 20:39:12.000000000","_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":9751,"name":"IBM DB2 CI","email":"scecd@cn.ibm.com","username":"ibmdb2","tags":["SERVICE_USER"]},{"value":0,"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},{"value":2,"date":"2015-07-10 11:52:14.000000000","_account_id":11356,"name":"Louis Taylor","email":"louis@kragniz.eu","username":"kragniz"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"value":2,"date":"2015-07-07 09:56:36.000000000","_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"value":0,"_account_id":7487,"name":"hzrandd","email":"82433422@qq.com","username":"hzrandd"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11356,"name":"Louis Taylor","email":"louis@kragniz.eu","username":"kragniz"},"all":[{"value":0,"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":9751,"name":"IBM DB2 CI","email":"scecd@cn.ibm.com","username":"ibmdb2","tags":["SERVICE_USER"]},{"value":0,"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},{"value":1,"date":"2015-07-10 11:52:14.000000000","_account_id":11356,"name":"Louis Taylor","email":"louis@kragniz.eu","username":"kragniz"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"value":0,"_account_id":7487,"name":"hzrandd","email":"82433422@qq.com","username":"hzrandd"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":9751,"name":"IBM DB2 CI","email":"scecd@cn.ibm.com","username":"ibmdb2","tags":["SERVICE_USER"]},{"value":0,"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},{"value":0,"_account_id":11356,"name":"Louis Taylor","email":"louis@kragniz.eu","username":"kragniz"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"value":0,"_account_id":7487,"name":"hzrandd","email":"82433422@qq.com","username":"hzrandd"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"_account_id":7487,"name":"hzrandd","email":"82433422@qq.com","username":"hzrandd"},{"_account_id":9751,"name":"IBM DB2 CI","email":"scecd@cn.ibm.com","username":"ibmdb2","tags":["SERVICE_USER"]},{"_account_id":11356,"name":"Louis Taylor","email":"louis@kragniz.eu","username":"kragniz"},{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-01-22 02:13:17.000000000","updated_by":{"_account_id":7487,"name":"hzrandd","email":"82433422@qq.com","username":"hzrandd"},"reviewer":{"_account_id":7487,"name":"hzrandd","email":"82433422@qq.com","username":"hzrandd"},"state":"REVIEWER"},{"updated":"2015-01-22 02:18:16.000000000","updated_by":{"_account_id":9751,"name":"IBM DB2 CI","email":"scecd@cn.ibm.com","username":"ibmdb2","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9751,"name":"IBM DB2 CI","email":"scecd@cn.ibm.com","username":"ibmdb2","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-06-16 16:12:18.000000000","updated_by":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"reviewer":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"state":"REVIEWER"},{"updated":"2015-07-07 09:53:35.000000000","updated_by":{"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},"reviewer":{"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},"state":"REVIEWER"},{"updated":"2015-07-07 09:56:36.000000000","updated_by":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"reviewer":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"state":"REVIEWER"},{"updated":"2015-07-09 20:39:12.000000000","updated_by":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"reviewer":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"state":"REVIEWER"},{"updated":"2015-07-10 11:52:14.000000000","updated_by":{"_account_id":11356,"name":"Louis Taylor","email":"louis@kragniz.eu","username":"kragniz"},"reviewer":{"_account_id":11356,"name":"Louis Taylor","email":"louis@kragniz.eu","username":"kragniz"},"state":"REVIEWER"},{"updated":"2015-07-14 13:55:55.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2015-08-06 01:16:43.000000000","updated_by":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"reviewer":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"state":"REVIEWER"}],"messages":[{"id":"811403eedf02311676033ef1f3f5800ba9361d03","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2015-01-20 15:28:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47fdc73cf828842fe3aee5ee61be5d5300bb260b","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2015-01-20 15:32:12.000000000","message":"Patch Set 2: Commit message was updated","accounts_in_message":[],"_revision_number":2},{"id":"72721c7e9f053cd42b9109d4ef5352d43ee9d961","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2015-01-20 15:36:12.000000000","message":"Patch Set 2:\n\nNoticed the extra whitespaces on glance-api-paste.ini will fix those on next when fixing the failing tests as well.","accounts_in_message":[],"_revision_number":2},{"id":"eecd088ab43ea2c2f879c77d77ce66fc9503a567","author":{"_account_id":11356,"name":"Louis Taylor","email":"louis@kragniz.eu","username":"kragniz"},"date":"2015-01-20 15:53:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nCan you add the new option to glance/common/config.py?","accounts_in_message":[],"_revision_number":2},{"id":"76e4d3ab877f4b30daa70186866a7f9cf9dd4d2c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-01-20 16:28:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-glance-pep8 http://logs.openstack.org/95/148595/2/check/gate-glance-pep8/ac87fef/ : FAILURE in 2m 45s\n- gate-glance-docs http://docs-draft.openstack.org/95/148595/2/check/gate-glance-docs/c546623//doc/build/html/ : SUCCESS in 3m 05s\n- gate-glance-python27 http://logs.openstack.org/95/148595/2/check/gate-glance-python27/f53b288/ : FAILURE in 6m 45s\n- check-tempest-dsvm-full http://logs.openstack.org/95/148595/2/check/check-tempest-dsvm-full/5c054d4/ : FAILURE in 13m 09s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/95/148595/2/check/check-tempest-dsvm-postgres-full/45a9279/ : FAILURE in 12m 04s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/95/148595/2/check/check-tempest-dsvm-neutron-full/0e712d4/ : FAILURE in 12m 15s\n- check-grenade-dsvm http://logs.openstack.org/95/148595/2/check/check-grenade-dsvm/9bff6da/ : SUCCESS in 34m 45s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/95/148595/2/check/gate-tempest-dsvm-large-ops/3359155/ : FAILURE in 9m 05s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/95/148595/2/check/gate-tempest-dsvm-neutron-large-ops/dc4341a/ : FAILURE in 10m 09s\n- check-devstack-dsvm-cells http://logs.openstack.org/95/148595/2/check/check-devstack-dsvm-cells/9d2d0c0/ : FAILURE in 13m 39s\n- gate-rally-dsvm-glance http://logs.openstack.org/95/148595/2/check/gate-rally-dsvm-glance/b589751/ : FAILURE in 13m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e39e61c6772424380124c8774412d3858915b03","author":{"_account_id":9751,"name":"IBM DB2 CI","email":"scecd@cn.ibm.com","username":"ibmdb2","tags":["SERVICE_USER"]},"date":"2015-01-21 00:24:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- ibm-db2-ci-glance http://dal05.objectstorage.softlayer.net/v1/AUTH_58396f85-2c60-47b9-aaf8-e03bc24a1a6f/cilog/95/148595/2/check/ibm-db2-ci-glance/663b9c2 : FAILURE in 1m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19ef57f4aae02aa23d836b4c67ab2a7a31bf2d1e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2015-01-21 12:43:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f5da0e3d8149aa45fbfde6eaab6f60e14647b77a","author":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"date":"2015-01-21 12:54:43.000000000","message":"Patch Set 3: Code-Review-1\n\nThis already exists in oslo.middleware, only paste configuration need to be updated to use the oslo.middleware healthcheck: \n\nhttps://github.com/openstack/oslo-specs/blob/master/specs/kilo/oslo-middleware-healthcheck.rst\n\nhttps://review.openstack.org/#/c/147805/","accounts_in_message":[],"_revision_number":3},{"id":"9f798115a299fcf4208688e6a9b30135c6bf10a4","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2015-01-21 13:27:39.000000000","message":"Patch Set 3:\n\nThanks Mehdi,\n\nDoes not seem to be that simple as I can\u0027t find the middleware from the repo. Also by the spec what you\u0027re doing with Julien to oslo seems to be quite overcomplicated for the needs, vs. the proposed to Glance and the existing in Swift (to which this is based on).","accounts_in_message":[],"_revision_number":3},{"id":"4bd5c895ddb4f94bce8f3b32add6e21f30ddaaa0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-01-21 13:39:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-glance-pep8 http://logs.openstack.org/95/148595/3/check/gate-glance-pep8/b2b9985/ : FAILURE in 4m 07s\n- gate-glance-docs http://docs-draft.openstack.org/95/148595/3/check/gate-glance-docs/ad957eb//doc/build/html/ : SUCCESS in 1m 37s\n- gate-glance-python27 http://logs.openstack.org/95/148595/3/check/gate-glance-python27/a967d45/ : SUCCESS in 6m 22s\n- check-tempest-dsvm-full http://logs.openstack.org/95/148595/3/check/check-tempest-dsvm-full/fde9b8d/ : SUCCESS in 44m 44s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/95/148595/3/check/check-tempest-dsvm-postgres-full/f3ba48b/ : SUCCESS in 33m 47s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/95/148595/3/check/check-tempest-dsvm-neutron-full/b1f1601/ : SUCCESS in 52m 35s\n- check-grenade-dsvm http://logs.openstack.org/95/148595/3/check/check-grenade-dsvm/501a31f/ : SUCCESS in 35m 54s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/95/148595/3/check/gate-tempest-dsvm-large-ops/480f754/ : SUCCESS in 23m 14s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/95/148595/3/check/gate-tempest-dsvm-neutron-large-ops/622d49a/ : SUCCESS in 20m 07s\n- check-devstack-dsvm-cells http://logs.openstack.org/95/148595/3/check/check-devstack-dsvm-cells/cb70eae/ : SUCCESS in 17m 17s\n- gate-rally-dsvm-glance http://logs.openstack.org/95/148595/3/check/gate-rally-dsvm-glance/ecc53df/ : SUCCESS in 26m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1380ac910124e1be81572e8d82483942c1ed304e","author":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"date":"2015-01-21 13:50:23.000000000","message":"Patch Set 3:\n\nThis is exactly the same thing, a wsgi middleware that just check the present of a file on a filesystem and return a 503 if so, It\u0027s generic this is why this middleware will leave in oslo program, instead of duplicate the code in each project.","accounts_in_message":[],"_revision_number":3},{"id":"6a5a44668eaae152cada4960001ed45ae79a1608","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2015-01-21 14:19:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ac547b5cfa971bc2e9158318e4b2c6972de51022","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-01-21 15:35:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-glance-pep8 http://logs.openstack.org/95/148595/4/check/gate-glance-pep8/3379e9e/ : SUCCESS in 2m 37s\n- gate-glance-docs http://docs-draft.openstack.org/95/148595/4/check/gate-glance-docs/fea84d5//doc/build/html/ : SUCCESS in 2m 26s\n- gate-glance-python27 http://logs.openstack.org/95/148595/4/check/gate-glance-python27/6b4f2b1/ : SUCCESS in 6m 41s\n- check-tempest-dsvm-full http://logs.openstack.org/95/148595/4/check/check-tempest-dsvm-full/9763fab/ : SUCCESS in 48m 38s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/95/148595/4/check/check-tempest-dsvm-postgres-full/bffc5a8/ : SUCCESS in 48m 28s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/95/148595/4/check/check-tempest-dsvm-neutron-full/3d38a8b/ : SUCCESS in 1h 06m 55s\n- check-grenade-dsvm http://logs.openstack.org/95/148595/4/check/check-grenade-dsvm/4c5d82b/ : SUCCESS in 39m 25s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/95/148595/4/check/gate-tempest-dsvm-large-ops/52c3bff/ : SUCCESS in 22m 51s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/95/148595/4/check/gate-tempest-dsvm-neutron-large-ops/905140c/ : SUCCESS in 20m 10s\n- check-devstack-dsvm-cells http://logs.openstack.org/95/148595/4/check/check-devstack-dsvm-cells/541a482/ : SUCCESS in 19m 50s\n- gate-rally-dsvm-glance http://logs.openstack.org/95/148595/4/check/gate-rally-dsvm-glance/3c706dd/ : SUCCESS in 26m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1427eda7f336964a527790b71c95bba69111c3de","author":{"_account_id":7487,"name":"hzrandd","email":"82433422@qq.com","username":"hzrandd"},"date":"2015-01-22 02:13:17.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1b160eb63148a03387a073147cab71d85664a167","author":{"_account_id":9751,"name":"IBM DB2 CI","email":"scecd@cn.ibm.com","username":"ibmdb2","tags":["SERVICE_USER"]},"date":"2015-01-22 02:18:16.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"3046cf7fd3399ca04dd9a62739db540f68f42a47","author":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"date":"2015-01-22 08:53:14.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"a942acb5afb6f21a1b62594ee595d79e8b6202fe","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2015-01-22 12:27:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a1f874dd458e87e1f2b37ff112575c29b19c15ba","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2015-01-22 15:22:46.000000000","message":"Patch Set 4:\n\n@Mehdi: Would you mind providing more info as to how the middleware in oslo.middleware should be configured? There doesn\u0027t seem to be documentation about this anywhere and it\u0027d be really helpful.\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"140b9f92788e414dfdd37806636e80b0bdeedbf7","author":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"date":"2015-01-26 12:36:39.000000000","message":"Patch Set 4:\n\n@Flavio Percoco: I have just pushed the documentation:\n\nhttps://review.openstack.org/#/c/150019/","accounts_in_message":[],"_revision_number":4},{"id":"35b72a14458d793063e964f8095a35f71d25ccbc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-01-27 15:05:21.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"87cc1042a0345eabb425bf2c1fa4a647e91d177f","author":{"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},"date":"2015-06-16 13:13:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e922459f787ef18017f3720974041b5fce0e4b2b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-16 14:21:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-glance-pep8 http://logs.openstack.org/95/148595/5/check/gate-glance-pep8/5d9e314/ : SUCCESS in 1m 42s\n- gate-glance-docs http://docs-draft.openstack.org/95/148595/5/check/gate-glance-docs/36d1417//doc/build/html/ : SUCCESS in 2m 32s\n- gate-glance-python27 http://logs.openstack.org/95/148595/5/check/gate-glance-python27/1309147/ : SUCCESS in 9m 07s\n- gate-glance-requirements http://logs.openstack.org/95/148595/5/check/gate-glance-requirements/18a953f/ : SUCCESS in 21s\n- check-tempest-dsvm-full http://logs.openstack.org/95/148595/5/check/check-tempest-dsvm-full/689158e/ : SUCCESS in 45m 17s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/95/148595/5/check/check-tempest-dsvm-postgres-full/c896f09/ : SUCCESS in 41m 26s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/95/148595/5/check/check-tempest-dsvm-neutron-full/71bf789/ : SUCCESS in 57m 16s\n- check-grenade-dsvm http://logs.openstack.org/95/148595/5/check/check-grenade-dsvm/e2ce294/ : SUCCESS in 41m 49s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/95/148595/5/check/gate-tempest-dsvm-large-ops/19abc50/ : SUCCESS in 26m 49s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/95/148595/5/check/gate-tempest-dsvm-neutron-large-ops/02af59c/ : SUCCESS in 23m 01s\n- check-devstack-dsvm-cells http://logs.openstack.org/95/148595/5/check/check-devstack-dsvm-cells/75f7a39/ : SUCCESS in 19m 34s\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/95/148595/5/check/check-tempest-dsvm-full-ceph/d087ea0/ : SUCCESS in 1h 05m 30s\n- gate-rally-dsvm-glance http://logs.openstack.org/95/148595/5/check/gate-rally-dsvm-glance/ce178f4/ : SUCCESS in 29m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2d0fe4d98ce3c7caf1956442c95003405d7277fa","author":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"date":"2015-06-16 16:12:18.000000000","message":"Patch Set 5: Code-Review+1\n\nawesome","accounts_in_message":[],"_revision_number":5},{"id":"b66c280d8bcf6408c96c850887d5d2f5f34f0ab7","author":{"_account_id":11356,"name":"Louis Taylor","email":"louis@kragniz.eu","username":"kragniz"},"date":"2015-06-29 13:24:56.000000000","message":"Patch Set 5: Code-Review-1\n\nHi Kamil, this looks great! Could you add a brief section to the docs about configuring and using this?","accounts_in_message":[],"_revision_number":5},{"id":"1309a1ddea3d926f8826f418a3ad58b2682875b4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-02 20:01:32.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"511260cfcd7069616c46a1fd7f095143ebf9c7b3","author":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"date":"2015-07-03 00:14:04.000000000","message":"Patch Set 5: Code-Review+1\n\n+1 to Louis suggestion.","accounts_in_message":[],"_revision_number":5},{"id":"92695e0190f5a77b341f80cab63e4cf263b1c79a","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2015-07-07 09:02:40.000000000","message":"Patch Set 5: Code-Review+1\n\nWaiting for the docs updates","accounts_in_message":[],"_revision_number":5},{"id":"412e0c86996f9d99f6c0eb69d14dcc7290204999","author":{"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},"date":"2015-07-07 09:52:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0e0496e0433aa1bb993cef0cb23ca6c79fc17268","author":{"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},"date":"2015-07-07 09:53:35.000000000","message":"Patch Set 6:\n\nI\u0027ve uploaded early docs about configuring healthcheck middleware.","accounts_in_message":[],"_revision_number":6},{"id":"3155d4a113b901a33a7dfec29367af81e1a33169","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2015-07-07 09:56:36.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"de1f2e3f312facca228273be5974c076f931c7b7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-07 10:57:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-glance-python34 http://logs.openstack.org/95/148595/6/check/gate-glance-python34/41270b6/ : SUCCESS in 4m 43s\n- gate-glance-pep8 http://logs.openstack.org/95/148595/6/check/gate-glance-pep8/aaf5267/ : SUCCESS in 2m 22s\n- gate-glance-docs http://docs-draft.openstack.org/95/148595/6/check/gate-glance-docs/7cb4320//doc/build/html/ : SUCCESS in 2m 12s\n- gate-glance-python27 http://logs.openstack.org/95/148595/6/check/gate-glance-python27/70f9416/ : SUCCESS in 9m 09s\n- gate-glance-requirements http://logs.openstack.org/95/148595/6/check/gate-glance-requirements/c8d7e03/ : SUCCESS in 1m 17s\n- check-tempest-dsvm-full http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-full/a067d99/ : SUCCESS in 52m 39s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-postgres-full/c9cc543/ : SUCCESS in 47m 16s\n- check-tempest-dsvm-neutron http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-neutron/c99eaa4/ : SUCCESS in 44m 23s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-neutron-full/2d8266d/ : SUCCESS in 1h 02m 49s\n- check-grenade-dsvm http://logs.openstack.org/95/148595/6/check/check-grenade-dsvm/a0f59bb/ : SUCCESS in 46m 02s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/95/148595/6/check/gate-tempest-dsvm-large-ops/b1e93bb/ : SUCCESS in 34m 03s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/95/148595/6/check/gate-tempest-dsvm-neutron-large-ops/effb311/ : SUCCESS in 30m 48s\n- check-devstack-dsvm-cells http://logs.openstack.org/95/148595/6/check/check-devstack-dsvm-cells/ba45f27/ : SUCCESS in 20m 12s\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-full-ceph/62ca73d/ : SUCCESS in 50m 00s\n- gate-rally-dsvm-glance http://logs.openstack.org/95/148595/6/check/gate-rally-dsvm-glance/91111e9/ : FAILURE in 27m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"19124ee55c731f334eda0ef98b69c424f56a29ef","author":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"date":"2015-07-09 20:39:12.000000000","message":"Patch Set 6: Code-Review+2\n\nI\u0027m tempted to also approve this and set the task of updating the spec myself. Does anyone have objections?","accounts_in_message":[],"_revision_number":6},{"id":"6e8ad73b5cf3067bfa6118ec24bc24c27d1fbb98","author":{"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},"date":"2015-07-10 05:27:56.000000000","message":"Patch Set 6:\n\nThanks Ian, sounds great, no objection in here.","accounts_in_message":[],"_revision_number":6},{"id":"ed7a12311900dad1141778e7e4c3fd228d283e5e","author":{"_account_id":11356,"name":"Louis Taylor","email":"louis@kragniz.eu","username":"kragniz"},"date":"2015-07-10 11:52:14.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nThe spec is all yours!","accounts_in_message":[],"_revision_number":6},{"id":"f26536ad8d2c9d156d77a85734c51aa98efdf246","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-10 11:52:22.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"76f3006cf7f4701c1c90fb7f76dced79be6f0198","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-10 12:59:16.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-glance-python34 http://logs.openstack.org/95/148595/6/gate/gate-glance-python34/bf8c838/ : SUCCESS in 3m 15s\n- gate-glance-docs http://docs-draft.openstack.org/95/148595/6/gate/gate-glance-docs/98705bf//doc/build/html/ : SUCCESS in 2m 04s\n- gate-glance-pep8 http://logs.openstack.org/95/148595/6/gate/gate-glance-pep8/8136b19/ : SUCCESS in 1m 58s\n- gate-glance-python27 http://logs.openstack.org/95/148595/6/gate/gate-glance-python27/3c951d5/ : FAILURE in 11m 08s\n- gate-glance-requirements http://logs.openstack.org/95/148595/6/gate/gate-glance-requirements/6ad8e54/ : SUCCESS in 19s\n- gate-tempest-dsvm-full http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-full/34fbfb3/ : SUCCESS in 42m 27s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-postgres-full/ee8780e/ : SUCCESS in 50m 36s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-neutron/aca4f1b/ : SUCCESS in 49m 05s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-neutron-full/1bd8dc6/ : FAILURE in 1h 05m 21s\n- gate-grenade-dsvm http://logs.openstack.org/95/148595/6/gate/gate-grenade-dsvm/93099bf/ : SUCCESS in 52m 13s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-large-ops/45c6b9f/ : SUCCESS in 36m 41s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-neutron-large-ops/c0f7adf/ : SUCCESS in 30m 58s\n- gate-devstack-dsvm-cells http://logs.openstack.org/95/148595/6/gate/gate-devstack-dsvm-cells/9a0dce5/ : SUCCESS in 28m 55s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-full-ceph/41c559e/ : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":6},{"id":"a3c043a48fa167c5434e6256b59143e33226380f","author":{"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},"date":"2015-07-11 17:08:22.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b918c759dc1876c416b1684e374f543dd9a6bc87","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-11 18:09:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-glance-python34 http://logs.openstack.org/95/148595/6/check/gate-glance-python34/def4c11/ : SUCCESS in 3m 21s\n- gate-glance-pep8 http://logs.openstack.org/95/148595/6/check/gate-glance-pep8/745a53a/ : SUCCESS in 1m 53s\n- gate-glance-docs http://docs-draft.openstack.org/95/148595/6/check/gate-glance-docs/6db68f6//doc/build/html/ : SUCCESS in 2m 05s\n- gate-glance-python27 http://logs.openstack.org/95/148595/6/check/gate-glance-python27/8030c6d/ : FAILURE in 7m 55s\n- gate-glance-requirements http://logs.openstack.org/95/148595/6/check/gate-glance-requirements/72964fc/ : SUCCESS in 31s\n- check-tempest-dsvm-full http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-full/983f035/ : SUCCESS in 58m 41s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-postgres-full/755c5bd/ : SUCCESS in 59m 06s\n- check-tempest-dsvm-neutron http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-neutron/de2e070/ : SUCCESS in 40m 30s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-neutron-full/3742309/ : SUCCESS in 1h 01m 16s\n- check-grenade-dsvm http://logs.openstack.org/95/148595/6/check/check-grenade-dsvm/e1ff352/ : SUCCESS in 45m 25s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/95/148595/6/check/gate-tempest-dsvm-large-ops/b5d1897/ : SUCCESS in 35m 06s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/95/148595/6/check/gate-tempest-dsvm-neutron-large-ops/ee34ce7/ : SUCCESS in 31m 11s\n- check-devstack-dsvm-cells http://logs.openstack.org/95/148595/6/check/check-devstack-dsvm-cells/27458ca/ : SUCCESS in 23m 35s\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-full-ceph/18ca36e/ : SUCCESS in 57m 03s\n- gate-rally-dsvm-glance http://logs.openstack.org/95/148595/6/check/gate-rally-dsvm-glance/ecfda52/ : FAILURE in 33m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f73123991fe173ad228b0d20a71b44be65387714","author":{"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},"date":"2015-07-13 07:07:28.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"06b09b626f434fecb6e8c869d5a46c4309450c0b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-13 08:27:11.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-glance-python34 http://logs.openstack.org/95/148595/6/check/gate-glance-python34/2b65d93/ : SUCCESS in 3m 32s\n- gate-glance-pep8 http://logs.openstack.org/95/148595/6/check/gate-glance-pep8/11d1a43/ : SUCCESS in 2m 38s\n- gate-glance-docs http://docs-draft.openstack.org/95/148595/6/check/gate-glance-docs/eabee55//doc/build/html/ : SUCCESS in 2m 11s\n- gate-glance-python27 http://logs.openstack.org/95/148595/6/check/gate-glance-python27/903b467/ : FAILURE in 8m 10s\n- gate-glance-requirements http://logs.openstack.org/95/148595/6/check/gate-glance-requirements/2d372d7/ : SUCCESS in 22s\n- check-tempest-dsvm-full http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-full/eb7a879/ : SUCCESS in 42m 56s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-postgres-full/f289592/ : SUCCESS in 40m 49s\n- check-tempest-dsvm-neutron http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-neutron/4ca544f/ : SUCCESS in 40m 37s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-neutron-full/7cbd9c0/ : SUCCESS in 58m 28s\n- check-grenade-dsvm http://logs.openstack.org/95/148595/6/check/check-grenade-dsvm/5f8ccd0/ : SUCCESS in 46m 18s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/95/148595/6/check/gate-tempest-dsvm-large-ops/c99a910/ : SUCCESS in 36m 48s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/95/148595/6/check/gate-tempest-dsvm-neutron-large-ops/686d36f/ : SUCCESS in 24m 37s\n- check-devstack-dsvm-cells http://logs.openstack.org/95/148595/6/check/check-devstack-dsvm-cells/9e7d415/ : SUCCESS in 20m 50s\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-full-ceph/938d78a/ : SUCCESS in 1h 14m 15s\n- gate-rally-dsvm-glance http://logs.openstack.org/95/148595/6/check/gate-rally-dsvm-glance/ce26f12/ : FAILURE in 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e1d56880a499b02c07363e2704ee68ac7726edc2","author":{"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},"date":"2015-07-14 10:37:37.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"55ab4aa70e2d928e101d535a3b07c3ed32cafa76","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-14 11:46:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-glance-python34 http://logs.openstack.org/95/148595/6/check/gate-glance-python34/17caca8/ : SUCCESS in 4m 30s\n- gate-glance-pep8 http://logs.openstack.org/95/148595/6/check/gate-glance-pep8/347ec0e/ : SUCCESS in 2m 11s\n- gate-glance-docs http://docs-draft.openstack.org/95/148595/6/check/gate-glance-docs/b65f1f8//doc/build/html/ : SUCCESS in 2m 28s\n- gate-glance-python27 http://logs.openstack.org/95/148595/6/check/gate-glance-python27/5d1d99e/ : SUCCESS in 9m 50s\n- gate-glance-requirements http://logs.openstack.org/95/148595/6/check/gate-glance-requirements/fd2f87b/ : SUCCESS in 33s\n- check-tempest-dsvm-full http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-full/f819a01/ : SUCCESS in 41m 21s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-postgres-full/25f409c/ : SUCCESS in 48m 45s\n- check-tempest-dsvm-neutron http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-neutron/b88bd1c/ : SUCCESS in 41m 08s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-neutron-full/642a867/ : SUCCESS in 59m 46s\n- check-grenade-dsvm http://logs.openstack.org/95/148595/6/check/check-grenade-dsvm/2d8c0ff/ : SUCCESS in 45m 59s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/95/148595/6/check/gate-tempest-dsvm-large-ops/8a8582e/ : SUCCESS in 34m 15s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/95/148595/6/check/gate-tempest-dsvm-neutron-large-ops/5609149/ : SUCCESS in 32m 47s\n- check-devstack-dsvm-cells http://logs.openstack.org/95/148595/6/check/check-devstack-dsvm-cells/9b65e39/ : SUCCESS in 27m 46s\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/95/148595/6/check/check-tempest-dsvm-full-ceph/523e44d/ : SUCCESS in 58m 06s\n- gate-rally-dsvm-glance http://logs.openstack.org/95/148595/6/check/gate-rally-dsvm-glance/32fb4aa/ : FAILURE in 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1a8c299486f4aa372f0ee44b3d6fd384dde93e3b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-14 12:01:15.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"0466460767acb17fdedf3cc37a3d69030cc0712e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-14 13:55:55.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-glance-python34 http://logs.openstack.org/95/148595/6/gate/gate-glance-python34/bda54bc/ : SUCCESS in 3m 59s\n- gate-glance-docs http://docs-draft.openstack.org/95/148595/6/gate/gate-glance-docs/b96c17b//doc/build/html/ : SUCCESS in 2m 53s\n- gate-glance-pep8 http://logs.openstack.org/95/148595/6/gate/gate-glance-pep8/4ae3702/ : SUCCESS in 2m 27s\n- gate-glance-python27 http://logs.openstack.org/95/148595/6/gate/gate-glance-python27/c038df1/ : SUCCESS in 7m 57s\n- gate-glance-requirements http://logs.openstack.org/95/148595/6/gate/gate-glance-requirements/c2dfa66/ : SUCCESS in 23s\n- gate-tempest-dsvm-full http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-full/1376f27/ : SUCCESS in 52m 50s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-postgres-full/691cdbc/ : SUCCESS in 47m 17s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-neutron/a47a641/ : SUCCESS in 41m 41s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-neutron-full/cb87d32/ : SUCCESS in 1h 02m 13s\n- gate-grenade-dsvm http://logs.openstack.org/95/148595/6/gate/gate-grenade-dsvm/76f4ff2/ : SUCCESS in 45m 14s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-large-ops/1a23058/ : SUCCESS in 34m 41s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-neutron-large-ops/acf79a9/ : SUCCESS in 32m 05s\n- gate-devstack-dsvm-cells http://logs.openstack.org/95/148595/6/gate/gate-devstack-dsvm-cells/2eed4a4/ : SUCCESS in 22m 04s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/95/148595/6/gate/gate-tempest-dsvm-full-ceph/a38cc48/ : SUCCESS in 53m 29s","accounts_in_message":[],"_revision_number":6},{"id":"a73596cc25fae0a669ab0ac4e248d06c0deb5d2e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-14 13:55:57.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":6},{"id":"fffe13f88d3ee84b3fae2297ebb117df3bbc20aa","author":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"date":"2015-08-06 01:16:43.000000000","message":"Patch Set 6:\n\nWhy is this not a deployer decision? I don\u0027t see any reason for this to be in Glance. The tests included here prove that point: there\u0027s absolutely nothing to test beyond \"was it deployed correctly?\"","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"562cb0429f9e2657dbcf108c534ef6c4779eb177","revisions":{"e54203a410fc66cd00987f6c141a5f1d7de59813":{"kind":"REWORK","_number":1,"created":"2015-01-20 15:28:37.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/95/148595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/148595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/148595/1"}}},"commit":{"parents":[{"commit":"3445866403b2076a576838dbaabd1541a66cb426","subject":"Merge \"Cleanup chunks for deleted image that was \u0027saving\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3445866403b2076a576838dbaabd1541a66cb426"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@hp.com","date":"2015-01-20 15:24:56.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@hp.com","date":"2015-01-20 15:24:56.000000000","tz":0},"subject":"HealthCheck Middleware","message":"HealthCheck Middleware\n\nMiddleware providing /healthcheck endpoint returning 200 OK.\n\nIf config option disable_path is set and exists will return\n503 DISABLED BY FILE\n\nSimilar behavior as Swift has for the healthcheck.\n\nimplements bp: healthcheck-middleware\n\nChange-Id: I45f6a8c59ec3040aaf06f8bab46d8001c44dac7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e54203a410fc66cd00987f6c141a5f1d7de59813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e54203a410fc66cd00987f6c141a5f1d7de59813"}]},"branch":"refs/heads/master"},"964c7fd588e786b37c1f9599aa368fd7b9df43ce":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2015-01-20 15:32:12.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/95/148595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/148595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/148595/2"}}},"commit":{"parents":[{"commit":"3445866403b2076a576838dbaabd1541a66cb426","subject":"Merge \"Cleanup chunks for deleted image that was \u0027saving\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3445866403b2076a576838dbaabd1541a66cb426"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@hp.com","date":"2015-01-20 15:24:56.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2015-01-20 15:32:12.000000000","tz":0},"subject":"HealthCheck Middleware","message":"HealthCheck Middleware\n\nMiddleware providing /healthcheck endpoint returning 200 OK.\n\nIf config option disable_path is set and exists will return\n503 DISABLED BY FILE\n\nSimilar behavior as Swift has for the healthcheck.\n\nimplements bp: healthcheck-middleware\n\nDocImpact\n\nChange-Id: I45f6a8c59ec3040aaf06f8bab46d8001c44dac7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/964c7fd588e786b37c1f9599aa368fd7b9df43ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/964c7fd588e786b37c1f9599aa368fd7b9df43ce"}]},"branch":"refs/heads/master"},"6d3847070ae61c11f58d1f07d6276ce920b7ffeb":{"kind":"REWORK","_number":3,"created":"2015-01-21 12:43:25.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/95/148595/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/148595/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/148595/3"}}},"commit":{"parents":[{"commit":"3445866403b2076a576838dbaabd1541a66cb426","subject":"Merge \"Cleanup chunks for deleted image that was \u0027saving\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3445866403b2076a576838dbaabd1541a66cb426"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@hp.com","date":"2015-01-20 15:24:56.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@hp.com","date":"2015-01-21 12:43:03.000000000","tz":0},"subject":"HealthCheck Middleware","message":"HealthCheck Middleware\n\nMiddleware providing /healthcheck endpoint returning 200 OK.\n\nIf config option disable_path is set and exists will return\n503 DISABLED BY FILE\n\nSimilar behavior as Swift has for the healthcheck.\n\nimplements bp: healthcheck-middleware\n\nDocImpact\n\nChange-Id: I45f6a8c59ec3040aaf06f8bab46d8001c44dac7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d3847070ae61c11f58d1f07d6276ce920b7ffeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d3847070ae61c11f58d1f07d6276ce920b7ffeb"}]},"branch":"refs/heads/master"},"353455a1d478af739ba9f570e11d93522d7fc4f4":{"kind":"REWORK","_number":4,"created":"2015-01-21 14:19:25.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/95/148595/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/148595/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/148595/4"}}},"commit":{"parents":[{"commit":"3445866403b2076a576838dbaabd1541a66cb426","subject":"Merge \"Cleanup chunks for deleted image that was \u0027saving\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3445866403b2076a576838dbaabd1541a66cb426"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@hp.com","date":"2015-01-20 15:24:56.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@hp.com","date":"2015-01-21 14:19:18.000000000","tz":0},"subject":"HealthCheck Middleware","message":"HealthCheck Middleware\n\nMiddleware providing /healthcheck endpoint returning 200 OK.\n\nIf config option disable_path is set and exists will return\n503 DISABLED BY FILE\n\nSimilar behavior as Swift has for the healthcheck.\n\nimplements bp: healthcheck-middleware\n\nDocImpact\n\nChange-Id: I45f6a8c59ec3040aaf06f8bab46d8001c44dac7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/353455a1d478af739ba9f570e11d93522d7fc4f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/353455a1d478af739ba9f570e11d93522d7fc4f4"}]},"branch":"refs/heads/master"},"11d8fe2fabe21028c0ca0274717e1ce24eb289a9":{"kind":"REWORK","_number":5,"created":"2015-06-16 13:13:11.000000000","uploader":{"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},"ref":"refs/changes/95/148595/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/148595/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/148595/5"}}},"commit":{"parents":[{"commit":"8c3a1e210e3d9c84e28b1d6c136a666167a392c9","subject":"Merge \"Fix Python 3 issues\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c3a1e210e3d9c84e28b1d6c136a666167a392c9"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@hp.com","date":"2015-01-20 15:24:56.000000000","tz":0},"committer":{"name":"Kamil Rykowski","email":"kamil.rykowski@intel.com","date":"2015-06-16 12:59:04.000000000","tz":120},"subject":"Healthcheck Middleware","message":"Healthcheck Middleware\n\nProvide healthcheck middleware from oslo_middleware to be able to\ndisable given nodes from loadbalancer. It\u0027s achieved by adding a new\npipeline which depending on existing of the\n/etc/glance/healthcheck_disable file can return one of the following\nresults:\n\n- 200 OK (if file does not exist)\n- 503 DISABLED BY FILE (if file exist)\n\nThe healthcheck is available under /healthcheck URL, and the whole\nmechanism behave similar to the Swift healthcheck system.\n\nimplements bp: healthcheck-middleware\n\nCo-Authored-By: Erno Kuvaja \u003cjokke@usr.fi\u003e\nCo-Authored-By: Kamil Rykowski \u003ckamil.rykowski@intel.com\u003e\n\nDocImpact\n\nChange-Id: I45f6a8c59ec3040aaf06f8bab46d8001c44dac7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11d8fe2fabe21028c0ca0274717e1ce24eb289a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11d8fe2fabe21028c0ca0274717e1ce24eb289a9"}]},"branch":"refs/heads/master"},"562cb0429f9e2657dbcf108c534ef6c4779eb177":{"kind":"REWORK","_number":6,"created":"2015-07-07 09:52:55.000000000","uploader":{"_account_id":13161,"name":"Kamil Rykowski","email":"kamil.rykowski@gmail.com","username":"krykowski"},"ref":"refs/changes/95/148595/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/148595/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/148595/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/148595/6"}}},"commit":{"parents":[{"commit":"0262ac58353da97f4fc9c651e4649f5d40d5ac47","subject":"Merge \"Functional of the HTTPclient was put in own method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0262ac58353da97f4fc9c651e4649f5d40d5ac47"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@hp.com","date":"2015-01-20 15:24:56.000000000","tz":0},"committer":{"name":"Kamil Rykowski","email":"kamil.rykowski@intel.com","date":"2015-07-07 09:55:21.000000000","tz":120},"subject":"Healthcheck Middleware","message":"Healthcheck Middleware\n\nProvide healthcheck middleware from oslo_middleware to be able to\ndisable given nodes from loadbalancer. It\u0027s achieved by adding a new\npipeline which depending on existing of the\n/etc/glance/healthcheck_disable file can return one of the following\nresults:\n\n- 200 OK (if file does not exist)\n- 503 DISABLED BY FILE (if file exist)\n\nThe healthcheck is available under /healthcheck URL, and the whole\nmechanism behave similar to the Swift healthcheck system.\n\nimplements bp: healthcheck-middleware\n\nCo-Authored-By: Erno Kuvaja \u003cjokke@usr.fi\u003e\nCo-Authored-By: Kamil Rykowski \u003ckamil.rykowski@intel.com\u003e\n\nDocImpact\n\nChange-Id: I45f6a8c59ec3040aaf06f8bab46d8001c44dac7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/562cb0429f9e2657dbcf108c534ef6c4779eb177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/562cb0429f9e2657dbcf108c534ef6c4779eb177"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
