)]}'
{"id":"openstack%2Fglance~313936","triplet_id":"openstack%2Fglance~master~I1c5cb7834c2cf3295e10bad7fd07cfacb8e4ac50","project":"openstack/glance","branch":"master","topic":"show-multiple-location","hashtags":[],"change_id":"I1c5cb7834c2cf3295e10bad7fd07cfacb8e4ac50","subject":"Deprecate `show_multiple_locations` option","status":"MERGED","created":"2016-05-08 15:15:57.000000000","updated":"2016-08-30 18:22:04.000000000","submitted":"2016-08-30 18:22:04.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dc1a431f09d95954428d46b07114f5598a0325f5","_number":313936,"virtual_id_number":313936,"owner":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2016-08-30 18:22:04.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":455,"name":"Stuart McLaren","email":"stuart.mclaren@hpe.com","username":"stuart-mclaren"},{"value":0,"_account_id":8040,"name":"Timur Sufiev","email":"tsufiev@gmail.com","username":"tsufiev"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":7179,"name":"Brad Pokorny","email":"brad_pokorny@symantec.com","username":"bpokorny"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"value":0,"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"all":[{"value":2,"date":"2016-08-30 16:54:19.000000000","_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":2,"date":"2016-08-29 18:39:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":455,"name":"Stuart McLaren","email":"stuart.mclaren@hpe.com","username":"stuart-mclaren"},{"value":0,"_account_id":8040,"name":"Timur Sufiev","email":"tsufiev@gmail.com","username":"tsufiev"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":1,"date":"2016-08-29 18:13:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7179,"name":"Brad Pokorny","email":"brad_pokorny@symantec.com","username":"bpokorny"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"value":0,"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"all":[{"value":1,"date":"2016-08-30 16:54:19.000000000","_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":455,"name":"Stuart McLaren","email":"stuart.mclaren@hpe.com","username":"stuart-mclaren"},{"value":0,"_account_id":8040,"name":"Timur Sufiev","email":"tsufiev@gmail.com","username":"tsufiev"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":7179,"name":"Brad Pokorny","email":"brad_pokorny@symantec.com","username":"bpokorny"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"value":0,"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":455,"name":"Stuart McLaren","email":"stuart.mclaren@hpe.com","username":"stuart-mclaren"},{"value":0,"_account_id":8040,"name":"Timur Sufiev","email":"tsufiev@gmail.com","username":"tsufiev"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":7179,"name":"Brad Pokorny","email":"brad_pokorny@symantec.com","username":"bpokorny"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"value":0,"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":455,"name":"Stuart McLaren","email":"stuart.mclaren@hpe.com","username":"stuart-mclaren"},{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"_account_id":7179,"name":"Brad Pokorny","email":"brad_pokorny@symantec.com","username":"bpokorny"},{"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},{"_account_id":8040,"name":"Timur Sufiev","email":"tsufiev@gmail.com","username":"tsufiev"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"},{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-05-09 09:07:32.000000000","updated_by":{"_account_id":455,"name":"Stuart McLaren","email":"stuart.mclaren@hpe.com","username":"stuart-mclaren"},"reviewer":{"_account_id":455,"name":"Stuart McLaren","email":"stuart.mclaren@hpe.com","username":"stuart-mclaren"},"state":"REVIEWER"},{"updated":"2016-05-09 15:22:09.000000000","updated_by":{"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"},"reviewer":{"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"},"state":"REVIEWER"},{"updated":"2016-05-11 19:33:20.000000000","updated_by":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"reviewer":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"state":"REVIEWER"},{"updated":"2016-06-08 09:13:58.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2016-06-08 15:30:15.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2016-06-09 11:08:33.000000000","updated_by":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"reviewer":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"state":"REVIEWER"},{"updated":"2016-08-12 21:32:35.000000000","updated_by":{"_account_id":8040,"name":"Timur Sufiev","email":"tsufiev@gmail.com","username":"tsufiev"},"reviewer":{"_account_id":8040,"name":"Timur Sufiev","email":"tsufiev@gmail.com","username":"tsufiev"},"state":"REVIEWER"},{"updated":"2016-08-12 21:32:49.000000000","updated_by":{"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},"reviewer":{"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},"state":"REVIEWER"},{"updated":"2016-08-29 14:03:04.000000000","updated_by":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"reviewer":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"state":"REVIEWER"},{"updated":"2016-08-29 18:13:37.000000000","updated_by":{"_account_id":7179,"name":"Brad Pokorny","email":"brad_pokorny@symantec.com","username":"bpokorny"},"reviewer":{"_account_id":7179,"name":"Brad Pokorny","email":"brad_pokorny@symantec.com","username":"bpokorny"},"state":"REVIEWER"},{"updated":"2016-08-29 18:39:27.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2016-08-30 16:54:19.000000000","updated_by":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"reviewer":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"state":"REVIEWER"},{"updated":"2016-08-30 18:22:04.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"ee094a05830b9fdd381df14c75a1513c95d7408c","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2016-05-08 15:15:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8727fe8f6dff0fb290f3d3646b5871576c3cbba8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-08 16:32:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-glance-python34-db http://logs.openstack.org/36/313936/1/check/gate-glance-python34-db/fb351cb/ : SUCCESS in 5m 13s\n- gate-glance-docs http://docs-draft.openstack.org/36/313936/1/check/gate-glance-docs/1b71ebe//doc/build/html/ : SUCCESS in 6m 51s\n- gate-glance-pep8 http://logs.openstack.org/36/313936/1/check/gate-glance-pep8/d94168f/ : SUCCESS in 2m 23s\n- gate-glance-python27-db http://logs.openstack.org/36/313936/1/check/gate-glance-python27-db/2350e8f/ : SUCCESS in 7m 17s\n- gate-tempest-dsvm-full http://logs.openstack.org/36/313936/1/check/gate-tempest-dsvm-full/d4749e4/ : FAILURE in 37m 35s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/36/313936/1/check/gate-tempest-dsvm-postgres-full/086be34/ : FAILURE in 44m 32s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/36/313936/1/check/gate-tempest-dsvm-neutron-full/b582c71/ : FAILURE in 48m 27s\n- gate-grenade-dsvm http://logs.openstack.org/36/313936/1/check/gate-grenade-dsvm/a08c7b9/ : SUCCESS in 37m 25s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/36/313936/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph/803cbb9/ : FAILURE in 35m 38s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/36/313936/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/a792afb/ : FAILURE in 1h 16m 52s (non-voting)\n- gate-rally-dsvm-glance http://logs.openstack.org/36/313936/1/check/gate-rally-dsvm-glance/818f1d5/ : SUCCESS in 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8611609534a723859b295be31f8aa3a49dd6b2c","author":{"_account_id":455,"name":"Stuart McLaren","email":"stuart.mclaren@hpe.com","username":"stuart-mclaren"},"date":"2016-05-09 09:07:32.000000000","message":"Patch Set 1: Code-Review-1\n\nPeople know I\u0027m not a big fan of exposing the image locations stuff, but I think we need to be careful here in terms of trying to deprecate the multiple locations stuff.\n\nThere\u0027s no detail here on why this is being done, or its affect on the location\u0027s strategy stuff for example.","accounts_in_message":[],"_revision_number":1},{"id":"c7a31c30dd7c4f74be3cc358ba0b51e6f01986d7","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2016-05-09 17:27:44.000000000","message":"Patch Set 1:\n\nStuart, yup! This draft is just to kickstart the discussion. FWIW, this has been discussed in the past already as part of other patches and bugs.\n\nI\u0027ll provide more details but I wanted to get some sort of consensus before I even go and try to fix all the CI jobs and whatnot.","accounts_in_message":[],"_revision_number":1},{"id":"3afcf4edfaf9bd10df4a25dc6ed74fa5efdb81ca","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-05-09 17:44:12.000000000","message":"Patch Set 1:\n\nCan you please mark this as -W? Just to avoid scaring people.\n\n---\n\n\n1:41:14 PM \u003cnikhil\u003e flaper87: I was thinking -- this is a good idea. but we should do it in newton preferably after nova stuff is merged though\n1:41:31 PM \u003cnikhil\u003e flaper87: mostly as this can result into disrupting that workflow\n1:41:38 PM \u003cnikhil\u003e flaper87: those are my initial thoughts\n1:41:39 PM \u003cflaper87\u003e nikhil: sure, not super urgent. TBH. I just want to start poking folks heads before I go and fix gfate issues and what not\n1:41:45 PM \u003cflaper87\u003e the change per-se is quite small\n1:42:10 PM \u003cnikhil\u003e flaper87: yeah, it\u0027s good idea overall (at least I think so)","accounts_in_message":[],"_revision_number":1},{"id":"38ea7cb9edc5a3173f4d1e4c365c361f09eb6bd2","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2016-05-11 16:35:25.000000000","message":"Patch Set 1: Workflow-1\n\nDone :)","accounts_in_message":[],"_revision_number":1},{"id":"f68e6696c8a3af9a61fbf09e8f637bcc51787e60","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2016-06-07 06:20:08.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"15fb4e71c2edc53c074dd0337ec974d385429585","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2016-06-07 06:20:38.000000000","message":"Patch Set 2:\n\nKeeping this as WIP won\u0027t help moving the discussion forward. I\u0027ve added more content to the commit message and I\u0027ll work on fixing the gate issues.","accounts_in_message":[],"_revision_number":2},{"id":"7a7876160965474e2aa42f2e66f8fa8cc2cfe6b0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-07 07:17:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-glance-python34-db http://logs.openstack.org/36/313936/2/check/gate-glance-python34-db/76cfdbb/ : SUCCESS in 5m 24s\n- gate-glance-docs http://docs-draft.openstack.org/36/313936/2/check/gate-glance-docs/f2256a9//doc/build/html/ : SUCCESS in 7m 43s\n- gate-glance-pep8 http://logs.openstack.org/36/313936/2/check/gate-glance-pep8/0970c1e/ : SUCCESS in 3m 03s\n- gate-glance-python27-db http://logs.openstack.org/36/313936/2/check/gate-glance-python27-db/dfda6d4/ : SUCCESS in 7m 49s\n- gate-tempest-dsvm-full http://logs.openstack.org/36/313936/2/check/gate-tempest-dsvm-full/70ca644/ : FAILURE in 50m 57s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/36/313936/2/check/gate-tempest-dsvm-postgres-full/58ae011/ : FAILURE in 39m 28s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/36/313936/2/check/gate-tempest-dsvm-neutron-full/00ebc5b/ : FAILURE in 51m 40s\n- gate-grenade-dsvm http://logs.openstack.org/36/313936/2/check/gate-grenade-dsvm/ed6d08c/ : SUCCESS in 39m 32s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/36/313936/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph/3b19e23/ : FAILURE in 46m 07s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/36/313936/2/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/9fa1d9a/ : FAILURE in 52m 10s (non-voting)\n- gate-rally-dsvm-glance http://logs.openstack.org/36/313936/2/check/gate-rally-dsvm-glance/6e6f9b2/ : SUCCESS in 47m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"76c472e8f549faf8ffa18ef4d591bf69641ee925","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2016-06-07 13:22:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"96c3fc2c7cbf89191385bd0d9fdb31ff73055e1d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-07 14:42:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-glance-python34-db http://logs.openstack.org/36/313936/3/check/gate-glance-python34-db/2c44e0c/ : SUCCESS in 5m 01s\n- gate-glance-docs http://docs-draft.openstack.org/36/313936/3/check/gate-glance-docs/ae7220e//doc/build/html/ : SUCCESS in 7m 52s\n- gate-glance-pep8 http://logs.openstack.org/36/313936/3/check/gate-glance-pep8/2e60f4d/ : SUCCESS in 2m 16s\n- gate-glance-python27-db http://logs.openstack.org/36/313936/3/check/gate-glance-python27-db/a4e5d7d/ : SUCCESS in 7m 48s\n- gate-tempest-dsvm-full http://logs.openstack.org/36/313936/3/check/gate-tempest-dsvm-full/18dbebc/ : SUCCESS in 38m 58s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/36/313936/3/check/gate-tempest-dsvm-postgres-full/b8979d2/ : SUCCESS in 42m 11s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/36/313936/3/check/gate-tempest-dsvm-neutron-full/bc4419a/ : SUCCESS in 55m 31s\n- gate-grenade-dsvm http://logs.openstack.org/36/313936/3/check/gate-grenade-dsvm/aeeb717/ : SUCCESS in 26m 06s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/36/313936/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph/9281ed0/ : SUCCESS in 36m 21s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/36/313936/3/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/4296c5b/ : SUCCESS in 56m 58s (non-voting)\n- gate-rally-dsvm-glance http://logs.openstack.org/36/313936/3/check/gate-rally-dsvm-glance/d69d302/ : SUCCESS in 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d66205058ce1c4b3fe24333357d31e2230df8fc8","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2016-06-08 09:13:58.000000000","message":"Patch Set 3: Code-Review+1\n\nI perfer the RBAC way. Currently, users can\u0027t add/replace/remove the image\u0027s locations by default(show_multiple_locations\u003dfalse), even for admin users. the locations can be updated only when show_multiple_locations\u003dtrue. Use a config option to limit a request seems heavy, this kind of work should be done by RBAC mechanism IMO.","accounts_in_message":[],"_revision_number":3},{"id":"6699c688d8fb60d0ddd447baf90e56b9505a878a","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-06-08 13:58:50.000000000","message":"Patch Set 3: Code-Review-1\n\nIs the debtcollector being used by oslo.config? The config option merely marks removal for deprecation.\n\nWe need UpgradeImpact, DocImpact, (not Security though). Also, we need good release notes -- I am happy to write those if you are short of bandwidth, just let me know.","accounts_in_message":[],"_revision_number":3},{"id":"dc299e270803ccc16462066ec7bded270cc5f8de","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2016-06-08 15:30:15.000000000","message":"Patch Set 3: Code-Review-1\n\nNikhil is right, sysadmins should somehow be notified of the removal of this option, and the doc should be up to date.","accounts_in_message":[],"_revision_number":3},{"id":"2a9a5f1c9b73f8c062fa55f0e4ebf3e4303b84e5","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2016-06-09 11:08:33.000000000","message":"Patch Set 3: Code-Review-1\n\nFrankly speaking I think that marking it as deprecated is not enough. First, we need to set \u0027role:admin\u0027 to related policies like it was done in previous PS; second, change this value to True by default; and the last - completely disable showing locations to users and create new API admin-only request like GET /v2/images/{image_id}/locations, that returns json representation of image locations. We may discuss it on IRC meeting.","accounts_in_message":[],"_revision_number":3},{"id":"ad7d12888a58d8d9019018a9d47d40795739600d","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2016-06-14 15:52:26.000000000","message":"Patch Set 3:\n\nNikhil, thanks I\u0027ll update accordingly. +1 to all the tags and release notes.\n\nMike, I\u0027m afraid changing the policy file will change the default behavior and it\u0027ll break the API. I believe we shouldn\u0027t be changing any of the rules now and delay that until the next cycle. Not changing the rules now is part of the deprecation path.","accounts_in_message":[],"_revision_number":3},{"id":"11efdaaed4ae2c1b24bcb1872f1e309d6e83af5b","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2016-06-14 15:59:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7c372503361a82dc30fdd28a81c579ebcee23820","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-14 17:54:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-glance-python34-db http://logs.openstack.org/36/313936/4/check/gate-glance-python34-db/b70cfe3/ : SUCCESS in 6m 58s\n- gate-glance-docs http://docs-draft.openstack.org/36/313936/4/check/gate-glance-docs/525561b//doc/build/html/ : SUCCESS in 8m 37s\n- gate-glance-pep8 http://logs.openstack.org/36/313936/4/check/gate-glance-pep8/4367ffb/ : SUCCESS in 4m 09s\n- gate-glance-python27-db http://logs.openstack.org/36/313936/4/check/gate-glance-python27-db/9b6d2c3/ : SUCCESS in 9m 08s\n- gate-tempest-dsvm-full http://logs.openstack.org/36/313936/4/check/gate-tempest-dsvm-full/76d0a47/ : SUCCESS in 50m 02s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/36/313936/4/check/gate-tempest-dsvm-postgres-full/8914979/ : SUCCESS in 51m 03s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/36/313936/4/check/gate-tempest-dsvm-neutron-full/9300534/ : SUCCESS in 58m 07s\n- gate-grenade-dsvm http://logs.openstack.org/36/313936/4/check/gate-grenade-dsvm/1f93ced/ : SUCCESS in 45m 45s\n- gate-glance-releasenotes http://docs-draft.openstack.org/36/313936/4/check/gate-glance-releasenotes/7290c96//releasenotes/build/html/ : SUCCESS in 3m 09s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/36/313936/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph/efe491a/ : SUCCESS in 40m 46s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/36/313936/4/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/cb16de6/ : SUCCESS in 1h 06m 06s (non-voting)\n- gate-rally-dsvm-glance http://logs.openstack.org/36/313936/4/check/gate-rally-dsvm-glance/ba20cd4/ : SUCCESS in 43m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"be0a652ab6dd496fccfaa06d551b481aae4f8025","author":{"_account_id":7179,"name":"Brad Pokorny","email":"brad_pokorny@symantec.com","username":"bpokorny"},"date":"2016-08-12 21:32:02.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nA question inline.","accounts_in_message":[],"_revision_number":4},{"id":"92f334f730f9d1242dc6282fa5b072494f8a5cf2","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-12 22:20:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1691bd7bd289a7bde5d71a4767594b0439be29a1","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-25 04:40:21.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nThanks for proposing this. Few minor suggestions inline otherwise looks good.\n\nLemme know if you want me to make the changes and move ahead -- either way works.","accounts_in_message":[],"_revision_number":4},{"id":"38d23099df4cea77a15b36a5229f1b2fdefe3e8f","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-25 04:40:25.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"029f7eae6547d0db5cf182665d25820b990ff126","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2016-08-26 14:54:48.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nI think we need \u0027deprecated_reason\u0027 and \u0027deprecated_since\u0027 added to the option definition; suggested text inline.","accounts_in_message":[],"_revision_number":4},{"id":"69a8ee6a4b4f3df4074525332c19b55a3ed271ec","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-27 03:37:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"116e0b2a9d8a85e9b778a0ced14a022891a90b46","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-27 03:56:30.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3198b498d2297633fe5ecc54ccc8dd5d2e19940f","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-27 03:56:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d35c6b58ec491aef425b763202be2112ed09617c","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-27 03:58:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"037b11a72ff297df2905b211d4f747f2d3032ee1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-27 05:00:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-glance-python34-db http://logs.openstack.org/36/313936/6/check/gate-glance-python34-db/4805727/ : SUCCESS in 3m 47s\n- gate-glance-python35-db-nv http://logs.openstack.org/36/313936/6/check/gate-glance-python35-db-nv/5384cfe/ : FAILURE in 41m 31s (non-voting)\n- gate-glance-docs-ubuntu-xenial http://docs-draft.openstack.org/36/313936/6/check/gate-glance-docs-ubuntu-xenial/8807194//doc/build/html/ : SUCCESS in 4m 59s\n- gate-glance-pep8-ubuntu-xenial http://logs.openstack.org/36/313936/6/check/gate-glance-pep8-ubuntu-xenial/a3bcc2d/ : SUCCESS in 3m 25s\n- gate-glance-python27-db-ubuntu-xenial http://logs.openstack.org/36/313936/6/check/gate-glance-python27-db-ubuntu-xenial/4c3e58e/ : SUCCESS in 12m 09s\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/36/313936/6/check/gate-grenade-dsvm-neutron-ubuntu-trusty/0a1dcfb/ : SUCCESS in 43m 12s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/36/313936/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/5ff6536/ : SUCCESS in 1h 02m 02s\n- gate-glance-releasenotes http://docs-draft.openstack.org/36/313936/6/check/gate-glance-releasenotes/030f7f1//releasenotes/build/html/ : SUCCESS in 4m 21s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/36/313936/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph/ab0e126/ : SUCCESS in 56m 23s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/36/313936/6/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/7823f48/ : FAILURE in 21m 13s (non-voting)\n- gate-rally-dsvm-glance http://logs.openstack.org/36/313936/6/check/gate-rally-dsvm-glance/cf8ee26/ : SUCCESS in 35m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4a22fb85dc0394add0364fd25f0733e2e3c8b5f4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2016-08-28 03:41:56.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nSuggestion inline for making the release note more clear.","accounts_in_message":[],"_revision_number":6},{"id":"082b4de5053bbed7f08af092c666f483f5acdb96","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-29 13:59:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"394514f9c3b74129976ea2553bea9233b10e47cd","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-29 14:00:52.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"569481596e0522870e5b77472d5286ed6a8274d8","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-29 14:01:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f3b32cc4c7eece2b97819fcae0c6555933f0f53e","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-29 14:03:04.000000000","message":"Patch Set 8:\n\n@Flavio: I thought to have made considerable amount of changes to add myself to the co-author list. Please point me out if you think otherwise.","accounts_in_message":[],"_revision_number":8},{"id":"afd21891e129265f6e226304f3bd7ff610fd8018","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-29 15:12:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-glance-python34-db http://logs.openstack.org/36/313936/8/check/gate-glance-python34-db/ca0a85b/ : SUCCESS in 8m 20s\n- gate-glance-python35-db-nv http://logs.openstack.org/36/313936/8/check/gate-glance-python35-db-nv/3633029/ : FAILURE in 41m 43s (non-voting)\n- gate-glance-docs-ubuntu-xenial http://docs-draft.openstack.org/36/313936/8/check/gate-glance-docs-ubuntu-xenial/7d5689f//doc/build/html/ : SUCCESS in 5m 13s\n- gate-glance-pep8-ubuntu-xenial http://logs.openstack.org/36/313936/8/check/gate-glance-pep8-ubuntu-xenial/08c4389/ : SUCCESS in 2m 34s\n- gate-glance-python27-db-ubuntu-xenial http://logs.openstack.org/36/313936/8/check/gate-glance-python27-db-ubuntu-xenial/49f1860/ : SUCCESS in 9m 18s\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/36/313936/8/check/gate-grenade-dsvm-neutron-ubuntu-trusty/fe637a4/ : SUCCESS in 1h 02m 37s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/36/313936/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/0d1b942/ : SUCCESS in 1h 09m 40s\n- gate-glance-releasenotes http://docs-draft.openstack.org/36/313936/8/check/gate-glance-releasenotes/5931ab2//releasenotes/build/html/ : SUCCESS in 3m 04s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/36/313936/8/check/gate-tempest-dsvm-full-devstack-plugin-ceph/4656939/ : SUCCESS in 58m 38s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/36/313936/8/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/dae6f53/ : FAILURE in 22m 29s (non-voting)\n- gate-rally-dsvm-glance http://logs.openstack.org/36/313936/8/check/gate-rally-dsvm-glance/1360730/ : SUCCESS in 39m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3ce7c803933db67e3bc7fb9e99da663b18ebb6ed","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2016-08-29 16:16:32.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)\n\nTwo nits that you can take or leave.","accounts_in_message":[],"_revision_number":8},{"id":"a55b408c3b4eae3bc6b328efe56e586c44093986","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-29 17:05:44.000000000","message":"Patch Set 9: Published edit on patch set 8","accounts_in_message":[],"_revision_number":9},{"id":"bac86f085266c95206cdb9cf52603f76a2557633","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-29 17:06:38.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e1283cbef38aadff84d96c6cfd6cab1aff87ad42","author":{"_account_id":7179,"name":"Brad Pokorny","email":"brad_pokorny@symantec.com","username":"bpokorny"},"date":"2016-08-29 18:13:37.000000000","message":"Patch Set 9: Code-Review+1\n\nLooks good to me. Thanks for providing the release notes.\n\nI don\u0027t know if this should be part of this same commit, but the documentation for the option should be updated here also (for the newton release):\n\nhttp://docs.openstack.org/mitaka/config-reference/image-service/api.html","accounts_in_message":[],"_revision_number":9},{"id":"b28639536892a82b988d28b1b505b521eae9dc46","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-29 18:19:29.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-glance-python34-db http://logs.openstack.org/36/313936/9/check/gate-glance-python34-db/498ca7d/ : SUCCESS in 4m 21s\n- gate-glance-python35-db-nv http://logs.openstack.org/36/313936/9/check/gate-glance-python35-db-nv/31ad196/ : FAILURE in 41m 56s (non-voting)\n- gate-glance-docs-ubuntu-xenial http://docs-draft.openstack.org/36/313936/9/check/gate-glance-docs-ubuntu-xenial/b038194//doc/build/html/ : SUCCESS in 5m 42s\n- gate-glance-pep8-ubuntu-xenial http://logs.openstack.org/36/313936/9/check/gate-glance-pep8-ubuntu-xenial/3b99846/ : SUCCESS in 2m 40s\n- gate-glance-python27-db-ubuntu-xenial http://logs.openstack.org/36/313936/9/check/gate-glance-python27-db-ubuntu-xenial/29f941e/ : SUCCESS in 11m 12s\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/36/313936/9/check/gate-grenade-dsvm-neutron-ubuntu-trusty/56b40c4/ : SUCCESS in 34m 15s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/36/313936/9/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/d6f3dcb/ : SUCCESS in 1h 07m 29s\n- gate-glance-releasenotes http://docs-draft.openstack.org/36/313936/9/check/gate-glance-releasenotes/c5fd306//releasenotes/build/html/ : SUCCESS in 2m 29s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/36/313936/9/check/gate-tempest-dsvm-full-devstack-plugin-ceph/541f6f3/ : SUCCESS in 49m 12s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/36/313936/9/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/e4731e4/ : FAILURE in 12m 20s (non-voting)\n- gate-rally-dsvm-glance http://logs.openstack.org/36/313936/9/check/gate-rally-dsvm-glance/12f3713/ : SUCCESS in 29m 41s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e7f1b5cb9f8c34a2c65faffae53611eec34a1ece","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2016-08-29 18:39:27.000000000","message":"Patch Set 9: Code-Review+2\n\n@Brad: good point about the docs.  Let\u0027s rely on the DocImpact flag and make doc update/enhancement a different patch.","accounts_in_message":[],"_revision_number":9},{"id":"ac4a60a0d832a0af80e36d3c734819eb9286b1d2","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-29 20:03:33.000000000","message":"Patch Set 9:\n\n\u003e Looks good to me. Thanks for providing the release notes.\n \u003e \n \u003e I don\u0027t know if this should be part of this same commit, but the\n \u003e documentation for the option should be updated here also (for the\n \u003e newton release):\n \u003e \n \u003e http://docs.openstack.org/mitaka/config-reference/image-service/api.html\n\nThanks for your review Brad. We do intend to make the doc changes and those are usually included as a consequence of the DocImpact flag (like Brian rightly pointed out).","accounts_in_message":[],"_revision_number":9},{"id":"57a2f553f82bef89cce62d02d49dd2201de3d2b0","author":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"date":"2016-08-30 16:54:19.000000000","message":"Patch Set 9: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"ac6bf21212d13fe47e9b9c7046e23a95055020fe","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-30 16:54:31.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":9},{"id":"f6acd514c55eeb4393021091dbeb6fb69cba2c53","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-30 17:22:02.000000000","message":"Patch Set 9:\n\n\u003e Patch Set 9: Workflow+1 Code-Review+2\n\nThanks Ian!!","accounts_in_message":[],"_revision_number":9},{"id":"fdbc270f2e0258c160fcd03bc9a84e34aef2ed6e","date":"2016-08-30 18:22:04.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":9},{"id":"781a5ceb4fe2c2a7b5e3cbb57814ec425323f23e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-30 18:22:04.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-glance-python34-db http://logs.openstack.org/36/313936/9/gate/gate-glance-python34-db/d4d548c/ : SUCCESS in 4m 20s\n- gate-glance-docs-ubuntu-xenial http://docs-draft.openstack.org/36/313936/9/gate/gate-glance-docs-ubuntu-xenial/9d74758//doc/build/html/ : SUCCESS in 8m 08s\n- gate-glance-pep8-ubuntu-xenial http://logs.openstack.org/36/313936/9/gate/gate-glance-pep8-ubuntu-xenial/3fbf31e/ : SUCCESS in 3m 24s\n- gate-glance-python27-db-ubuntu-xenial http://logs.openstack.org/36/313936/9/gate/gate-glance-python27-db-ubuntu-xenial/9bea2ab/ : SUCCESS in 10m 56s\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/36/313936/9/gate/gate-grenade-dsvm-neutron-ubuntu-trusty/5231e1c/ : SUCCESS in 33m 23s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/36/313936/9/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/cb03375/ : SUCCESS in 1h 02m 34s\n- gate-glance-releasenotes http://docs-draft.openstack.org/36/313936/9/gate/gate-glance-releasenotes/033866b//releasenotes/build/html/ : SUCCESS in 2m 44s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/36/313936/9/gate/gate-tempest-dsvm-full-devstack-plugin-ceph/407faec/ : SUCCESS in 49m 19s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"dbfc121072dff4981532b068bb210f9c73c7b98c","revisions":{"169984baf76138a7b6407b090a1d0f5b8c677ff1":{"kind":"REWORK","_number":1,"created":"2016-05-08 15:15:57.000000000","uploader":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"ref":"refs/changes/36/313936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/313936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/313936/1"}}},"commit":{"parents":[{"commit":"13ea7f77ecffc485acb91c9f36018dd8d106c1fe","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13ea7f77ecffc485acb91c9f36018dd8d106c1fe"}]}],"author":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-05-08 15:06:33.000000000","tz":-240},"committer":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-05-08 15:09:51.000000000","tz":-240},"subject":"WIP: Deprecate `show_multiple_locations` option","message":"WIP: Deprecate `show_multiple_locations` option\n\nThis patch marks the aforementioned option as deprecated and sets the\npolicies for managing multiple locationsas admin-only by default. This\nmakes managing locations based on policies at the cost of not having a\nturn off/on switch.\n\nChange-Id: I1c5cb7834c2cf3295e10bad7fd07cfacb8e4ac50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/169984baf76138a7b6407b090a1d0f5b8c677ff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/169984baf76138a7b6407b090a1d0f5b8c677ff1"}]},"branch":"refs/heads/master"},"2317721f347e7690b017e7c116f67cab82620b6e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2016-06-07 06:20:08.000000000","uploader":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"ref":"refs/changes/36/313936/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/313936/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/313936/2"}}},"commit":{"parents":[{"commit":"13ea7f77ecffc485acb91c9f36018dd8d106c1fe","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13ea7f77ecffc485acb91c9f36018dd8d106c1fe"}]}],"author":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-05-08 15:06:33.000000000","tz":-240},"committer":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-06-07 06:16:01.000000000","tz":120},"subject":"Deprecate `show_multiple_locations` option","message":"Deprecate `show_multiple_locations` option\n\nThere are currently several ways to disable/enable multiple_locations.\nOne is through a global config option (show_multiple_locations) and the\nother one is through a more granular RBAC using Glance\u0027s policies.\n\nMaintaining two different ways to configure, enable and/or disable a\nfeature is painful for developers and operators. Given that we have a\nmore granular way to manage this feature and that it\u0027s the preferred one\ngiven the provided ability to chose what each role can/cannot do, this\npatch proposes to remove the global config option.\n\nThere have been concerns about this proposal mostly on the fact that\nit\u0027ll remove the ability to \"turn off\" the entire feature with a single\noption. This doesn\u0027t seem to be a strong enough motivation to warrant\nthe effort of maintaining these 2 options.\n\nThis patch marks the aforementioned option as deprecated and sets the\npolicies for managing multiple locationsas admin-only by default. This\nmakes managing locations based on policies at the cost of not having a\nturn off/on switch.\n\nChange-Id: I1c5cb7834c2cf3295e10bad7fd07cfacb8e4ac50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2317721f347e7690b017e7c116f67cab82620b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2317721f347e7690b017e7c116f67cab82620b6e"}]},"branch":"refs/heads/master"},"e3fe2e7df1766331518f2ac067df9ac48511fe2b":{"kind":"REWORK","_number":3,"created":"2016-06-07 13:22:58.000000000","uploader":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"ref":"refs/changes/36/313936/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/313936/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/313936/3"}}},"commit":{"parents":[{"commit":"13ea7f77ecffc485acb91c9f36018dd8d106c1fe","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13ea7f77ecffc485acb91c9f36018dd8d106c1fe"}]}],"author":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-05-08 15:06:33.000000000","tz":-240},"committer":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-06-07 13:21:52.000000000","tz":120},"subject":"Deprecate `show_multiple_locations` option","message":"Deprecate `show_multiple_locations` option\n\nThere are currently several ways to disable/enable multiple_locations.\nOne is through a global config option (show_multiple_locations) and the\nother one is through a more granular RBAC using Glance\u0027s policies.\n\nMaintaining two different ways to configure, enable and/or disable a\nfeature is painful for developers and operators. Given that we have a\nmore granular way to manage this feature and that it\u0027s the preferred one\ngiven the provided ability to chose what each role can/cannot do, this\npatch proposes to remove the global config option.\n\nThere have been concerns about this proposal mostly on the fact that\nit\u0027ll remove the ability to \"turn off\" the entire feature with a single\noption. This doesn\u0027t seem to be a strong enough motivation to warrant\nthe effort of maintaining these 2 options.\n\nThis patch marks the aforementioned option as deprecated.\n\nChange-Id: I1c5cb7834c2cf3295e10bad7fd07cfacb8e4ac50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3fe2e7df1766331518f2ac067df9ac48511fe2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3fe2e7df1766331518f2ac067df9ac48511fe2b"}]},"branch":"refs/heads/master"},"22007969f89e74a84ad58e2c9a2a8c9d2401714b":{"kind":"REWORK","_number":4,"created":"2016-06-14 15:59:02.000000000","uploader":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"ref":"refs/changes/36/313936/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/313936/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/313936/4"}}},"commit":{"parents":[{"commit":"13ea7f77ecffc485acb91c9f36018dd8d106c1fe","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13ea7f77ecffc485acb91c9f36018dd8d106c1fe"}]}],"author":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-05-08 15:06:33.000000000","tz":-240},"committer":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-06-14 15:58:33.000000000","tz":120},"subject":"Deprecate `show_multiple_locations` option","message":"Deprecate `show_multiple_locations` option\n\nThere are currently several ways to disable/enable multiple_locations.\nOne is through a global config option (show_multiple_locations) and the\nother one is through a more granular RBAC using Glance\u0027s policies.\n\nMaintaining two different ways to configure, enable and/or disable a\nfeature is painful for developers and operators. Given that we have a\nmore granular way to manage this feature and that it\u0027s the preferred one\ngiven the provided ability to chose what each role can/cannot do, this\npatch proposes to remove the global config option.\n\nThere have been concerns about this proposal mostly on the fact that\nit\u0027ll remove the ability to \"turn off\" the entire feature with a single\noption. This doesn\u0027t seem to be a strong enough motivation to warrant\nthe effort of maintaining these 2 options.\n\nThis patch marks the aforementioned option as deprecated.\n\nUpgradeImpact\nDocImpact\n\nChange-Id: I1c5cb7834c2cf3295e10bad7fd07cfacb8e4ac50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/22007969f89e74a84ad58e2c9a2a8c9d2401714b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/22007969f89e74a84ad58e2c9a2a8c9d2401714b"}]},"branch":"refs/heads/master"},"8e84ac86b9616b09275eb59229b886c50f494126":{"kind":"REWORK","_number":5,"created":"2016-08-27 03:56:40.000000000","uploader":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"ref":"refs/changes/36/313936/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/313936/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/313936/5"}}},"commit":{"parents":[{"commit":"13ea7f77ecffc485acb91c9f36018dd8d106c1fe","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13ea7f77ecffc485acb91c9f36018dd8d106c1fe"}]}],"author":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-05-08 15:06:33.000000000","tz":-240},"committer":{"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","date":"2016-08-27 03:56:08.000000000","tz":-240},"subject":"Deprecate `show_multiple_locations` option","message":"Deprecate `show_multiple_locations` option\n\nThere are currently several ways to disable/enable multiple_locations.\nOne is through a global config option (show_multiple_locations) and the\nother one is through a more granular RBAC using Glance\u0027s policies.\n\nMaintaining two different ways to configure, enable and/or disable a\nfeature is painful for developers and operators. Given that we have a\nmore granular way to manage this feature and that it\u0027s the preferred one\ngiven the provided ability to chose what each role can/cannot do, this\npatch proposes to remove the global config option.\n\nThere have been concerns about this proposal mostly on the fact that\nit\u0027ll remove the ability to \"turn off\" the entire feature with a single\noption. This doesn\u0027t seem to be a strong enough motivation to warrant\nthe effort of maintaining these 2 options.\n\nThis patch marks the aforementioned option as deprecated.\n\nUpgradeImpact\nDocImpact\n\nLite-spec: https://review.openstack.org/360220\nChange-Id: I1c5cb7834c2cf3295e10bad7fd07cfacb8e4ac50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e84ac86b9616b09275eb59229b886c50f494126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e84ac86b9616b09275eb59229b886c50f494126"}]},"branch":"refs/heads/master"},"bb0545b18fdee0ab13507d482fa32a78ff834ff8":{"kind":"REWORK","_number":6,"created":"2016-08-27 03:58:25.000000000","uploader":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"ref":"refs/changes/36/313936/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/313936/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/313936/6"}}},"commit":{"parents":[{"commit":"13ea7f77ecffc485acb91c9f36018dd8d106c1fe","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13ea7f77ecffc485acb91c9f36018dd8d106c1fe"}]}],"author":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-05-08 15:06:33.000000000","tz":-240},"committer":{"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","date":"2016-08-27 03:58:16.000000000","tz":-240},"subject":"Deprecate `show_multiple_locations` option","message":"Deprecate `show_multiple_locations` option\n\nThere are currently several ways to disable/enable multiple_locations.\nOne is through a global config option (show_multiple_locations) and the\nother one is through a more granular RBAC using Glance\u0027s policies.\n\nMaintaining two different ways to configure, enable and/or disable a\nfeature is painful for developers and operators. Given that we have a\nmore granular way to manage this feature and that it\u0027s the preferred one\ngiven the provided ability to chose what each role can/cannot do, this\npatch proposes to remove the global config option.\n\nThere have been concerns about this proposal mostly on the fact that\nit\u0027ll remove the ability to \"turn off\" the entire feature with a single\noption. This doesn\u0027t seem to be a strong enough motivation to warrant\nthe effort of maintaining these 2 options.\n\nThis patch marks the aforementioned option as deprecated.\n\nUpgradeImpact\nDocImpact\n\nLite-spec: https://review.openstack.org/360220\nChange-Id: I1c5cb7834c2cf3295e10bad7fd07cfacb8e4ac50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb0545b18fdee0ab13507d482fa32a78ff834ff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb0545b18fdee0ab13507d482fa32a78ff834ff8"}]},"branch":"refs/heads/master"},"9f8620ac2d73bcce97bf2c5dd7e9332318cfe40b":{"kind":"REWORK","_number":7,"created":"2016-08-29 13:59:45.000000000","uploader":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"ref":"refs/changes/36/313936/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/313936/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/313936/7"}}},"commit":{"parents":[{"commit":"13ea7f77ecffc485acb91c9f36018dd8d106c1fe","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13ea7f77ecffc485acb91c9f36018dd8d106c1fe"}]}],"author":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-05-08 15:06:33.000000000","tz":-240},"committer":{"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","date":"2016-08-29 13:59:36.000000000","tz":-240},"subject":"Deprecate `show_multiple_locations` option","message":"Deprecate `show_multiple_locations` option\n\nThere are currently several ways to disable/enable multiple_locations.\nOne is through a global config option (show_multiple_locations) and the\nother one is through a more granular RBAC using Glance\u0027s policies.\n\nMaintaining two different ways to configure, enable and/or disable a\nfeature is painful for developers and operators. Given that we have a\nmore granular way to manage this feature and that it\u0027s the preferred one\ngiven the provided ability to chose what each role can/cannot do, this\npatch proposes to remove the global config option.\n\nThere have been concerns about this proposal mostly on the fact that\nit\u0027ll remove the ability to \"turn off\" the entire feature with a single\noption. This doesn\u0027t seem to be a strong enough motivation to warrant\nthe effort of maintaining these 2 options.\n\nThis patch marks the aforementioned option as deprecated.\n\nUpgradeImpact\nDocImpact\n\nLite-spec: https://review.openstack.org/360220\nChange-Id: I1c5cb7834c2cf3295e10bad7fd07cfacb8e4ac50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f8620ac2d73bcce97bf2c5dd7e9332318cfe40b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f8620ac2d73bcce97bf2c5dd7e9332318cfe40b"}]},"branch":"refs/heads/master"},"c53fcf8d87f8a32264d292e5899dd1d2a7f8b41e":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2016-08-29 14:00:52.000000000","uploader":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"ref":"refs/changes/36/313936/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/313936/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/313936/8"}}},"commit":{"parents":[{"commit":"13ea7f77ecffc485acb91c9f36018dd8d106c1fe","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13ea7f77ecffc485acb91c9f36018dd8d106c1fe"}]}],"author":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-05-08 15:06:33.000000000","tz":-240},"committer":{"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","date":"2016-08-29 13:59:52.000000000","tz":-240},"subject":"Deprecate `show_multiple_locations` option","message":"Deprecate `show_multiple_locations` option\n\nThere are currently several ways to disable/enable multiple_locations.\nOne is through a global config option (show_multiple_locations) and the\nother one is through a more granular RBAC using Glance\u0027s policies.\n\nMaintaining two different ways to configure, enable and/or disable a\nfeature is painful for developers and operators. Given that we have a\nmore granular way to manage this feature and that it\u0027s the preferred one\ngiven the provided ability to chose what each role can/cannot do, this\npatch proposes to remove the global config option.\n\nThere have been concerns about this proposal mostly on the fact that\nit\u0027ll remove the ability to \"turn off\" the entire feature with a single\noption. This doesn\u0027t seem to be a strong enough motivation to warrant\nthe effort of maintaining these 2 options.\n\nThis patch marks the aforementioned option as deprecated.\n\nUpgradeImpact\nDocImpact\n\nCo-Authored-By: Flavio Percoco \u003cflaper87@gmail.com\u003e\nCo-Authored-By: Nikhil Komawar \u003cnik.komawar@gmail.com\u003e\n\nLite-spec: https://review.openstack.org/360220\nChange-Id: I1c5cb7834c2cf3295e10bad7fd07cfacb8e4ac50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c53fcf8d87f8a32264d292e5899dd1d2a7f8b41e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c53fcf8d87f8a32264d292e5899dd1d2a7f8b41e"}]},"branch":"refs/heads/master"},"dbfc121072dff4981532b068bb210f9c73c7b98c":{"kind":"REWORK","_number":9,"created":"2016-08-29 17:05:44.000000000","uploader":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"ref":"refs/changes/36/313936/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/313936/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/313936/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/313936/9"}}},"commit":{"parents":[{"commit":"13ea7f77ecffc485acb91c9f36018dd8d106c1fe","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13ea7f77ecffc485acb91c9f36018dd8d106c1fe"}]}],"author":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-05-08 15:06:33.000000000","tz":-240},"committer":{"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","date":"2016-08-29 17:05:35.000000000","tz":0},"subject":"Deprecate `show_multiple_locations` option","message":"Deprecate `show_multiple_locations` option\n\nThere are currently several ways to disable/enable multiple_locations.\nOne is through a global config option (show_multiple_locations) and the\nother one is through a more granular RBAC using Glance\u0027s policies.\n\nMaintaining two different ways to configure, enable and/or disable a\nfeature is painful for developers and operators. Given that we have a\nmore granular way to manage this feature and that it\u0027s the preferred one\ngiven the provided ability to chose what each role can/cannot do, this\npatch proposes to remove the global config option.\n\nThere have been concerns about this proposal mostly on the fact that\nit\u0027ll remove the ability to \"turn off\" the entire feature with a single\noption. This doesn\u0027t seem to be a strong enough motivation to warrant\nthe effort of maintaining these 2 options.\n\nThis patch marks the aforementioned option as deprecated.\n\nUpgradeImpact\nDocImpact\n\nCo-Authored-By: Flavio Percoco \u003cflaper87@gmail.com\u003e\nCo-Authored-By: Nikhil Komawar \u003cnik.komawar@gmail.com\u003e\n\nLite-spec: https://review.openstack.org/360220\nChange-Id: I1c5cb7834c2cf3295e10bad7fd07cfacb8e4ac50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dbfc121072dff4981532b068bb210f9c73c7b98c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dbfc121072dff4981532b068bb210f9c73c7b98c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
