)]}'
{"id":"openstack%2Fglance~351848","triplet_id":"openstack%2Fglance~master~Iee695870aefc9cb6142dbc7400672b335588fcc9","project":"openstack/glance","branch":"master","topic":"remove-conf-tree","attention_set":{},"removed_from_attention_set":{"2537":{"account":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"last_update":"2022-07-08 15:25:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iee695870aefc9cb6142dbc7400672b335588fcc9","subject":"Remove the configs shipped in source tree","status":"ABANDONED","created":"2016-08-05 17:04:23.000000000","updated":"2022-07-08 15:25:40.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0cfb91de61b89cac0d8e7adfc410190d4ad10e93","_number":351848,"virtual_id_number":351848,"owner":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"date":"2016-08-05 18:51:06.000000000","_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":-2,"date":"2016-08-08 08:19:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":-1,"date":"2016-08-05 17:07:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"REVIEWER":[{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-08-05 17:07:40.000000000","updated_by":{"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"},"reviewer":{"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"},"state":"REVIEWER"},{"updated":"2016-08-08 08:19:27.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-07-08 15:25:40.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"}],"messages":[{"id":"cd29d51e857a9433aca1c518aa9a57e4b71ac62d","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-05 17:04:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97e0768e52a4d5baabf3817ec1de81b14ddd9108","author":{"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"},"date":"2016-08-05 17:07:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa9394ea8bf5b6f080420730f6713e796b658483","author":{"_account_id":8158,"name":"Hemanth Makkapati","email":"makstacker+osgerrit@gmail.com","username":"hemanth.makkapati"},"date":"2016-08-05 17:09:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c270e1fa165fe5d5fb2c587ce6ca216b47ba086","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-08-05 18:51:06.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n\nCan you explain what\u0027s blocking from generating this?","accounts_in_message":[],"_revision_number":1},{"id":"2ad218801c69c554b68447a13562e37cf73341c9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-05 20:20:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-glance-python34-db http://logs.openstack.org/48/351848/1/check/gate-glance-python34-db/71af418/ : SUCCESS in 6m 31s\n- gate-glance-python35-db-nv http://logs.openstack.org/48/351848/1/check/gate-glance-python35-db-nv/e4a4116/ : FAILURE in 45m 20s (non-voting)\n- gate-glance-docs-ubuntu-xenial http://docs-draft.openstack.org/48/351848/1/check/gate-glance-docs-ubuntu-xenial/2675297//doc/build/html/ : SUCCESS in 7m 46s\n- gate-glance-pep8-ubuntu-xenial http://logs.openstack.org/48/351848/1/check/gate-glance-pep8-ubuntu-xenial/34b79fc/ : SUCCESS in 2m 27s\n- gate-glance-python27-db-ubuntu-xenial http://logs.openstack.org/48/351848/1/check/gate-glance-python27-db-ubuntu-xenial/cb02f5f/ : SUCCESS in 9m 03s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/48/351848/1/check/gate-tempest-dsvm-full-ubuntu-xenial/f576adc/ : FAILURE in 10m 22s\n- gate-tempest-dsvm-postgres-full-ubuntu-xenial http://logs.openstack.org/48/351848/1/check/gate-tempest-dsvm-postgres-full-ubuntu-xenial/f2f3823/ : FAILURE in 17m 18s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/48/351848/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/d7b276a/ : FAILURE in 14m 12s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/48/351848/1/check/gate-grenade-dsvm-ubuntu-trusty/9d096ba/ : SUCCESS in 50m 17s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/48/351848/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph/ae18a32/ : FAILURE in 22m 35s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/48/351848/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/0e5d49d/ : FAILURE in 21m 20s (non-voting)\n- gate-rally-dsvm-glance http://logs.openstack.org/48/351848/1/check/gate-rally-dsvm-glance/5a71486/ : FAILURE in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"571f14712ae5669f64d55a34f5406e2d43699dad","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2016-08-08 08:19:27.000000000","message":"Patch Set 1: Code-Review-2\n\nFew things before we should go this far:\n\n1) As Hemanth pointed out, lets refer to the correct link\n2) Let\u0027s make sure we have all files covered before removing them\n3) Let\u0027s document this for any possible consumer first, remove later. So lets take care of the TODOs\n4) I do not want to let this merge before we have released Newton and see that those doc pages are available for stable and reflect only stable (not master). Also any docs linking to those pointing towards correct release.\n\nWill be happy to reconsider this in Ocata if all the preconditions above can be met by the release of Newton.","accounts_in_message":[],"_revision_number":1},{"id":"bcb9bc27492589031168583d576d35c4ed63279c","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2016-08-08 08:22:05.000000000","message":"Patch Set 1:\n\nAlso based on the test failures it seems that these configs are depended more than you wanted to believe. We Might need to give time to migrate away from the in-tree configs.","accounts_in_message":[],"_revision_number":1},{"id":"811fede899551f1437b51406c30697d7fae964e7","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2016-09-16 13:26:06.000000000","message":"Patch Set 1:\n\nBy quick search I could not find stable branch docs (and so config examples from the stable branch). Thus I\u0027m still against removing the examples.","accounts_in_message":[],"_revision_number":1},{"id":"7a0f81b24f2a020d3ee9d83250c56bbe879cd5f9","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2016-09-16 22:36:00.000000000","message":"Patch Set 1:\n\n\u003e By quick search I could not find stable branch docs (and so config\n \u003e examples from the stable branch). Thus I\u0027m still against removing\n \u003e the examples.\n\nthank you. yeah we need to work on the stable branch docs. will update when ready.","accounts_in_message":[],"_revision_number":1},{"id":"39d16c79bcec79534b69d4d485d9a0b35663a5a2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2017-12-18 04:51:34.000000000","message":"Removed reviewer Jenkins with the following votes:\n\n* Verified-1 by Jenkins (3)\n","accounts_in_message":[],"_revision_number":1},{"id":"0cfb91de61b89cac0d8e7adfc410190d4ad10e93","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-08 15:25:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d00986090d6891dd93c798498779c1077c4945a2","revisions":{"d00986090d6891dd93c798498779c1077c4945a2":{"kind":"REWORK","_number":1,"created":"2016-08-05 17:04:23.000000000","uploader":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"ref":"refs/changes/48/351848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/351848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/351848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/351848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/351848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/351848/1"}}},"commit":{"parents":[{"commit":"53b8909e41707d9a9aad9ed4156ffaf6a01bea4b","subject":"Merge \"Add a requirements guidelines to docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53b8909e41707d9a9aad9ed4156ffaf6a01bea4b"}]}],"author":{"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","date":"2016-08-05 17:01:02.000000000","tz":-240},"committer":{"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","date":"2016-08-05 17:01:16.000000000","tz":-240},"subject":"Remove the configs shipped in source tree","message":"Remove the configs shipped in source tree\n\nAs the configs are now being published real time on the docs page [1],\nit\u0027s okay to remove these configs from the source tree that need to be\nmaintained manually everytime a change occurs.\n\nThe publishing on the docs page is good for non-developer, non-operator\naudience too.\n\nTODO:\n\n* Add a releasenote\n* Add intree documentation for how to use tox to generate sample configs\n\n[1] http://docs.openstack.org/developer/glance/opts/index.html\n\nUpgradeImpact\n\nChange-Id: Iee695870aefc9cb6142dbc7400672b335588fcc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d00986090d6891dd93c798498779c1077c4945a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d00986090d6891dd93c798498779c1077c4945a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
