)]}'
{"id":"openstack%2Fglance~395441","triplet_id":"openstack%2Fglance~master~I785a402c48ed41ce14f95f02f82c1c70fc38fd9e","project":"openstack/glance","branch":"master","topic":"remove_pylintrc","attention_set":{},"removed_from_attention_set":{"23733":{"account":{"_account_id":23733,"name":"pawnesh kumar","email":"pawnesh.kumar@nectechnologies.in","username":"pawnesh.kumar"},"last_update":"2022-08-02 07:25:01.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-08-02 07:25:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I785a402c48ed41ce14f95f02f82c1c70fc38fd9e","subject":"Remove unused pylintrc","status":"MERGED","created":"2016-11-09 07:41:28.000000000","updated":"2022-08-02 07:26:09.000000000","submitted":"2022-08-02 07:25:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"395441-remove_pylintrc","meta_rev_id":"9da7e62ed991d52a126a7c6b06016df5f9388ef4","_number":395441,"virtual_id_number":395441,"owner":{"_account_id":23733,"name":"pawnesh kumar","email":"pawnesh.kumar@nectechnologies.in","username":"pawnesh.kumar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":23733,"name":"pawnesh kumar","email":"pawnesh.kumar@nectechnologies.in","username":"pawnesh.kumar"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-02 07:25:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":23733,"name":"pawnesh kumar","email":"pawnesh.kumar@nectechnologies.in","username":"pawnesh.kumar"},{"value":2,"date":"2022-08-02 05:43:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-08-01 18:18:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":23733,"name":"pawnesh kumar","email":"pawnesh.kumar@nectechnologies.in","username":"pawnesh.kumar"},{"value":1,"date":"2022-08-02 05:43:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":23733,"name":"pawnesh kumar","email":"pawnesh.kumar@nectechnologies.in","username":"pawnesh.kumar"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23733,"name":"pawnesh kumar","email":"pawnesh.kumar@nectechnologies.in","username":"pawnesh.kumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-02-03 13:53:33.000000000","updated_by":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"reviewer":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"state":"REVIEWER"},{"updated":"2022-08-01 18:18:55.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-08-01 20:04:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-02 05:43:36.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"29166fc79663e6bde51be4bee4c3908fdcf2a7db","author":{"_account_id":23733,"name":"pawnesh kumar","email":"pawnesh.kumar@nectechnologies.in","username":"pawnesh.kumar"},"date":"2016-11-09 07:41:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52bd38994bac30fb1b520a8037db872be511dee3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-09 08:48:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-glance-python34-db http://logs.openstack.org/41/395441/1/check/gate-glance-python34-db/3d16252/ : SUCCESS in 4m 33s\n- gate-glance-python35-db-nv http://logs.openstack.org/41/395441/1/check/gate-glance-python35-db-nv/20a1766/ : FAILURE in 45m 15s (non-voting)\n- gate-glance-docs-ubuntu-xenial http://docs-draft.openstack.org/41/395441/1/check/gate-glance-docs-ubuntu-xenial/005687e//doc/build/html/ : SUCCESS in 4m 58s\n- gate-glance-pep8-ubuntu-xenial http://logs.openstack.org/41/395441/1/check/gate-glance-pep8-ubuntu-xenial/a21b36c/ : SUCCESS in 3m 03s\n- gate-glance-python27-db-ubuntu-xenial http://logs.openstack.org/41/395441/1/check/gate-glance-python27-db-ubuntu-xenial/5992e62/ : SUCCESS in 7m 04s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/41/395441/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/15b859f/ : SUCCESS in 47m 22s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/41/395441/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/e76dbe8/ : FAILURE in 49m 40s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/41/395441/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/9b363be/ : SUCCESS in 1h 05m 52s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/41/395441/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/6266c7b/ : FAILURE in 49m 50s (non-voting)\n- gate-rally-dsvm-glance http://logs.openstack.org/41/395441/1/check/gate-rally-dsvm-glance/dfc8356/ : SUCCESS in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e10fef2aa2dd87e3f9df665d12f00cb83035ba14","author":{"_account_id":23733,"name":"pawnesh kumar","email":"pawnesh.kumar@nectechnologies.in","username":"pawnesh.kumar"},"date":"2016-11-10 11:21:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0d487ea1f327a96392757ef97da7b31b0bbc0e5d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-10 12:13:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-glance-python34-db http://logs.openstack.org/41/395441/1/check/gate-glance-python34-db/5286702/ : SUCCESS in 3m 13s\n- gate-glance-python35-db-nv http://logs.openstack.org/41/395441/1/check/gate-glance-python35-db-nv/dd6d733/ : FAILURE in 45m 25s (non-voting)\n- gate-glance-docs-ubuntu-xenial http://docs-draft.openstack.org/41/395441/1/check/gate-glance-docs-ubuntu-xenial/e3eeea6//doc/build/html/ : SUCCESS in 9m 13s\n- gate-glance-pep8-ubuntu-xenial http://logs.openstack.org/41/395441/1/check/gate-glance-pep8-ubuntu-xenial/9c22b03/ : SUCCESS in 3m 48s\n- gate-glance-python27-db-ubuntu-xenial http://logs.openstack.org/41/395441/1/check/gate-glance-python27-db-ubuntu-xenial/43afb96/ : SUCCESS in 7m 21s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/41/395441/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/2ca25f7/ : SUCCESS in 47m 00s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/41/395441/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/1c4b4cf/ : SUCCESS in 32m 04s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/41/395441/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/8412337/ : SUCCESS in 45m 15s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/41/395441/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/ef584f1/ : SUCCESS in 48m 42s (non-voting)\n- gate-rally-dsvm-glance http://logs.openstack.org/41/395441/1/check/gate-rally-dsvm-glance/90a1383/ : FAILURE in 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d61e7649e9356b89a93033007ec8c3d758ee599","author":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"date":"2017-01-25 14:53:53.000000000","message":"Patch Set 1: Code-Review-2\n\nThere is at least one other change that is attempting to drop the MANIFEST.in file and *two* other changes that are attempting to drop pylintrc. Please coordinate with those patch authors.\n\nThis -2 will be lifted after RC-1 is released.","accounts_in_message":[],"_revision_number":1},{"id":"05f037de02c74b92052a4b3bb097bfb2acf11538","author":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"date":"2017-02-03 13:53:33.000000000","message":"Patch Set 1: -Code-Review\n\nGlance RC-1 is released, this can proceed.","accounts_in_message":[],"_revision_number":1},{"id":"8e3b105534cda1957b7d85d72476cd81e26ad214","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-17 02:12:40.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"1681ad86ed35ab26f494f45d4e73474f20835b07","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2017-12-18 04:05:12.000000000","message":"Removed reviewer Jenkins with the following votes:\n\n* Verified-1 by Jenkins (3)\n","accounts_in_message":[],"_revision_number":1},{"id":"40f86f312fdc4f74008fcbbd7a89ec8d7cbb6b5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-08-01 18:17:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9e94130baeed575808fc17a94dc16ccbbf1adbfb","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-08-01 18:18:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78ccbee1693409ff1a43e11fbd4b2d083904c083","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 20:04:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c294ad9d47a8489cb1423649f9956304\n\n- grenade https://zuul.opendev.org/t/openstack/build/efcd5678ce3a4efbb44d7875fc67a175 : SUCCESS in 54m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/61015ce76f7048afa5a094bd66507932 : SUCCESS in 1h 05m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ba95c04b9744782af754c5dcece3063 : SUCCESS in 43m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff92a038a6fa4a7b89c48e3b34080057 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/476f4a468f7c4004a672e2bcb3db65e0 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15b802e1121d44d09004b02e74e3e500 : SUCCESS in 7m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61abdfcabd424eac8fcfa02413e9db29 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aac5853531f4a29939cd0c59f623d87 : SUCCESS in 6m 51s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a78f5db6a65e4af49d7103248736383a : SUCCESS in 11m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a0a8d742e10444a98028e31a984936b7 : SUCCESS in 10m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d4936132c93e499e966018ecb5ccec34 : SUCCESS in 10m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4e3793b28b53432c843e7d1da43c3970 : SUCCESS in 5m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7c68568249da410da186e560e9d5bf59 : SUCCESS in 1h 08m 04s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/945e17f6a4eb46f1ad7aedf85198fbe8 : SUCCESS in 1h 12m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7d93d0bd6ade410a8da14ff0fd4fb91d : SUCCESS in 1h 10m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/96106e5d1cd34d0d946afe0d6c8beaa2 : SUCCESS in 1h 13m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2f5433e1fa34d0fb091612f424e1aa8 : SUCCESS in 50m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b914173a5e884eb0a3b06c74db0025d3 : SUCCESS in 1h 05m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4c93f545df914307aa47cb38b7f2b8b9 : SUCCESS in 34m 39s","accounts_in_message":[],"_revision_number":2},{"id":"944d8abb9014cc827da7233c7606b79da1d20439","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-02 05:43:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c88d3871761d23aaa87cae9674960b12180c9c8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 05:44:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"93889192692690b293feb088a3783d5cfa381909","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 07:25:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55130330c5d0495ea4088353044a381d\n\n- grenade https://zuul.opendev.org/t/openstack/build/ae953e6fc2454e0ca99926a583ea89a7 : SUCCESS in 1h 08m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c5aaec798c94757a9d7306e7eb490e9 : SUCCESS in 1h 25m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/100c2be68df049e4b9de4cf39ecb1f76 : SUCCESS in 1h 02m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10d791ca40fd4e62a511abf8effc6d37 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab77d758bcc54372bc0d2ff91b9c5d35 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33ba58c314d2418086a78cb0d20f059c : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69977e23e31842cf98cbab35dfe86fbf : SUCCESS in 6m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f3fc4af2daf846738789c275f2958f04 : SUCCESS in 7m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e4a9258056648139ecfa2a9dfb3f307 : SUCCESS in 9m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/097ef28a485841da94fb8868f9a8b42c : SUCCESS in 1h 36m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4d311931c8234511b1425d11ccc92b2d : SUCCESS in 1h 26m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0b186c55db04c8e96f7dfb73ce22d02 : SUCCESS in 51m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/911ffe52a1eb414dbfe5ae33dff3d8f4 : SUCCESS in 1h 10m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/12c7df40640749908209b5f8641f34ce : SUCCESS in 32m 31s","accounts_in_message":[],"_revision_number":2},{"id":"05a8cbc368cbd3f98aac69cf98194de1cc4630f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 07:25:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9da7e62ed991d52a126a7c6b06016df5f9388ef4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 07:26:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b20526ee3b524a4aade1679437d3c096\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fca46940b144e6c97b8587b8af0f19a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"db916163381b7c2470fea280916a70829a80547f","revisions":{"49c4ed108ed27ac6756f74709b372c1efc42772c":{"kind":"REWORK","_number":1,"created":"2016-11-09 07:41:28.000000000","uploader":{"_account_id":23733,"name":"pawnesh kumar","email":"pawnesh.kumar@nectechnologies.in","username":"pawnesh.kumar"},"ref":"refs/changes/41/395441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/41/395441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/395441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/395441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/395441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/41/395441/1"}}},"commit":{"parents":[{"commit":"09ca4452224897753074b78a8d729e4ab500d271","subject":"Merge \"Added overwrite warning for db_export_metadefs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/09ca4452224897753074b78a8d729e4ab500d271"}]}],"author":{"name":"pawnesh.kumar","email":"pawnesh.kumar@nectechnologies.in","date":"2016-11-09 07:51:19.000000000","tz":330},"committer":{"name":"pawnesh.kumar","email":"pawnesh.kumar@nectechnologies.in","date":"2016-11-09 07:51:19.000000000","tz":330},"subject":"Remove unused pylintrc","message":"Remove unused pylintrc\n\nWe didn\u0027t use it in the gate and it\u0027s unmaintained. Clean that.\n\nChange-Id: I785a402c48ed41ce14f95f02f82c1c70fc38fd9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/49c4ed108ed27ac6756f74709b372c1efc42772c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/49c4ed108ed27ac6756f74709b372c1efc42772c"}]},"branch":"refs/heads/master"},"db916163381b7c2470fea280916a70829a80547f":{"kind":"REWORK","_number":2,"created":"2022-08-01 18:17:50.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/41/395441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/41/395441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/395441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/395441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/395441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/41/395441/2"}}},"commit":{"parents":[{"commit":"81fee546aeb822671170a243d4f4639ce2bd5991","subject":"Merge \"glance-manage: fix crash under Python3.11\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/81fee546aeb822671170a243d4f4639ce2bd5991"}]}],"author":{"name":"pawnesh.kumar","email":"pawnesh.kumar@nectechnologies.in","date":"2016-11-09 07:51:19.000000000","tz":330},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-08-01 18:17:00.000000000","tz":120},"subject":"Remove unused pylintrc","message":"Remove unused pylintrc\n\nWe didn\u0027t use it in the gate and it\u0027s unmaintained. Clean that.\n\nChange-Id: I785a402c48ed41ce14f95f02f82c1c70fc38fd9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/db916163381b7c2470fea280916a70829a80547f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/db916163381b7c2470fea280916a70829a80547f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
