)]}'
{"id":"openstack%2Fglance~486674","triplet_id":"openstack%2Fglance~master~I69c84ba7e090a5eb83f412d8ac14819522658e70","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-06-24 20:43:45.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-06-24 20:43:55.000000000","reason":"Change was abandoned"},"5196":{"account":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"last_update":"2026-06-24 20:43:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I69c84ba7e090a5eb83f412d8ac14819522658e70","subject":"Add steps on checking the stored image to verification guide","status":"ABANDONED","created":"2017-07-24 15:54:36.000000000","updated":"2026-06-24 20:43:55.000000000","total_comment_count":11,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"4df6b66b442a106adc2c51cff0e7017caf8ac34d","_number":486674,"virtual_id_number":486674,"owner":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-05 22:22:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":-1,"date":"2022-07-07 22:33:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":-1,"date":"2026-06-24 20:22:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-07-24 17:10:55.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-07-25 01:57:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2017-09-05 14:44:35.000000000","updated_by":{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"},"reviewer":{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"},"state":"REVIEWER"},{"updated":"2018-01-12 12:45:12.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-07-05 22:10:21.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2022-07-05 22:22:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-24 20:22:53.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"92181c5e7ec8d5ddcd528024ef8dc848263bd00a","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2017-07-24 15:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe020db29b9f3e8df14869ad5e4642ae6d443ea2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-24 17:10:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-glance-docs-ubuntu-xenial http://docs-draft.openstack.org/74/486674/1/check/gate-glance-docs-ubuntu-xenial/a4931b8//doc/build/html/ : SUCCESS in 5m 53s\n- gate-glance-pep8-ubuntu-xenial http://logs.openstack.org/74/486674/1/check/gate-glance-pep8-ubuntu-xenial/009c88f/ : SUCCESS in 6m 21s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/74/486674/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/e236e8c/ : SUCCESS in 1h 15m 21s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/74/486674/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/8b0a444/ : SUCCESS in 1h 11m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d718d21113bff7904105786b6ae387b33e744639","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2017-07-25 01:57:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nJust a thought inline. Definitely agree this is a step that should be added to the guide.","accounts_in_message":[],"_revision_number":1},{"id":"b2cf5edd8b99bae872b1dc839497925f44189107","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2017-07-25 15:40:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90facbcee5bc5a5b0c32a0d7e70c83d7aa3da92c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2017-08-02 18:55:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nLike Sean, I think this is a good addition to the docs, but I also have a question inline.","accounts_in_message":[],"_revision_number":1},{"id":"a80e070dc483f7735f25a45ac64d48c5459df1a6","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2017-08-02 19:03:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6a2ed1634b70bbad2fc07eb08fce7169998a787","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2017-08-03 11:58:30.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nSuggestion inline.","accounts_in_message":[],"_revision_number":1},{"id":"31d90c63860ce0f20ec00e0b1c741a2f20b6440b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2017-08-03 12:06:13.000000000","message":"Patch Set 1:\n\nThat\u0027s quite a blog post, by the way!  I think it may be worth mentioning your experience at the end of the validation section, something about if you\u0027re doing a non-standard install, to check that the requirements are being satisfied; glance is only tested with particular versions of its dependencies, and if other versions are used, there could be some anomalous behavior.","accounts_in_message":[],"_revision_number":1},{"id":"1b3abba0d8c93474c83d617626e91b484d92763a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2017-08-10 12:09:07.000000000","message":"Patch Set 1: Code-Review-2\n\nDo not merge until after stable/pike is cut.","accounts_in_message":[],"_revision_number":1},{"id":"99c672a7cf44fb5ed74f06875cac21e15e39526e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2017-08-24 21:54:09.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"2c2c5be279f18631cfdda9829f4269958cbcde9d","author":{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"},"date":"2017-09-05 14:44:35.000000000","message":"Patch Set 1: Code-Review+1\n\nI agree with Brian, his proposed or recommended flow would allow greater granularity and understanding of how the image ended up not matching. On the other hand, I do agree that diff is more direct so I think it\u0027s a solid middle ground that addresses both concerns. Otherwise looks good.","accounts_in_message":[],"_revision_number":1},{"id":"39ff65fa0de9ea690343d50db39be39a62cab936","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2018-01-12 12:45:12.000000000","message":"Patch Set 1: Code-Review+1\n\nJust want to nudge this.  It\u0027s OK as is, but I think it would be better with the suggested text added.","accounts_in_message":[],"_revision_number":1},{"id":"8b59cb58bb619e04e72c480ddb42239982783da4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-05 22:09:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ead7633059329e8bef630edc350152220ce759a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-05 22:10:21.000000000","message":"Patch Set 2:\n\nI rebased this patch on top of master, included Brian\u0027s suggestion and used the glanceclient rather than OSC.","accounts_in_message":[],"_revision_number":2},{"id":"6b0da8fbc62d4a2d81386631e5ca23e83746ef9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 22:22:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b3e05c601044aed83c8fd4482ab4cb9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd2d2ceaa5954fc9bc7d0b6d090ae932 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1025dddb7ef944b387e700261480e3aa : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6e5988a9c6f855fd614539492a6f99a14ef67ff4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-07 22:33:47.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7042b822f4cb26a89aa4f635569bcbef6d3a1661","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-24 20:22:53.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e220717a8e78f716471f383286d8122bbb5072f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-24 20:43:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4df6b66b442a106adc2c51cff0e7017caf8ac34d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-24 20:43:55.000000000","message":"Abandoned\n\nNo longer relevant.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b8903cc6f846f6b8c774a38b4c50c2a297c21182","revisions":{"a84b01fd96b5932c26ae749ec9a8f95d7ebc705d":{"kind":"REWORK","_number":1,"created":"2017-07-24 15:54:36.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/74/486674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/486674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/486674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/486674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/486674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/486674/1"}}},"commit":{"parents":[{"commit":"adb219cf4570dc5ff36b49b84988d4773ed71d6b","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/adb219cf4570dc5ff36b49b84988d4773ed71d6b"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2017-07-24 15:47:00.000000000","tz":-240},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2017-07-24 15:47:00.000000000","tz":-240},"subject":"Add steps on checking the stored image to verification guide","message":"Add steps on checking the stored image to verification guide\n\nThis commit adds a final step the verification page in the Glance\ninstall guide to actually check the stored image in glance matches the\nupload. This is an important step of the verification process to make\nsure that everything in the install is 100% operational. Personally it\nwas something I had trouble with in my local cloud install (which I\ndocumented here: http://blog.kortar.org/?p\u003d380 ) and ensuring the stored\ndata matches the uploaded data is just a good sanity check to verify the\ninstall.\n\nChange-Id: I69c84ba7e090a5eb83f412d8ac14819522658e70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a84b01fd96b5932c26ae749ec9a8f95d7ebc705d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a84b01fd96b5932c26ae749ec9a8f95d7ebc705d"}]},"branch":"refs/heads/master"},"b8903cc6f846f6b8c774a38b4c50c2a297c21182":{"kind":"REWORK","_number":2,"created":"2022-07-05 22:09:20.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/74/486674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/486674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/486674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/486674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/486674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/486674/2"}}},"commit":{"parents":[{"commit":"2866713392de2a122145a186a7ed2391e208b4ba","subject":"Merge \"glance-cache-prefetcher: Set up access to multistore backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2866713392de2a122145a186a7ed2391e208b4ba"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2017-07-24 15:47:00.000000000","tz":-240},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-05 22:08:57.000000000","tz":120},"subject":"Add steps on checking the stored image to verification guide","message":"Add steps on checking the stored image to verification guide\n\nThis commit adds a final step the verification page in the Glance\ninstall guide to actually check the stored image in glance matches the\nupload. This is an important step of the verification process to make\nsure that everything in the install is 100% operational. Personally it\nwas something I had trouble with in my local cloud install (which I\ndocumented here: http://blog.kortar.org/?p\u003d380 ) and ensuring the stored\ndata matches the uploaded data is just a good sanity check to verify the\ninstall.\n\nChange-Id: I69c84ba7e090a5eb83f412d8ac14819522658e70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8903cc6f846f6b8c774a38b4c50c2a297c21182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8903cc6f846f6b8c774a38b4c50c2a297c21182"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
