)]}'
{"id":"openstack%2Fglance~510382","triplet_id":"openstack%2Fglance~master~I59ebbceed0616fa166d5cdfcde26a6594564dfa9","project":"openstack/glance","branch":"master","topic":"bug/1722128","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2022-06-02 13:12:13.000000000","reason":"Change was abandoned"},"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2022-06-02 13:12:13.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-06-02 13:12:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I59ebbceed0616fa166d5cdfcde26a6594564dfa9","subject":"Fix DB login step in Ubuntu install guide","status":"ABANDONED","created":"2017-10-08 23:26:16.000000000","updated":"2022-06-02 13:12:13.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"868e416b06135b255c7f5dd144accaf5d873722e","_number":510382,"virtual_id_number":510382,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-01 22:57:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":-1,"date":"2022-06-02 13:12:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2022-06-02 13:12:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2022-06-01 21:03:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-06-02 13:12:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2022-06-01 21:03:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-10-08 23:32:42.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-10-09 00:09:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2017-10-09 00:54:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-09 18:26:55.000000000","updated_by":{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"},"reviewer":{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"},"state":"REVIEWER"},{"updated":"2022-05-31 17:35:20.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-06-01 15:14:21.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"c9e3b81c03c2a16d1a149a4fb55bb7fbfaa4169d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2017-10-08 23:26:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a09cc6366b51cb95d61066f7277f9caf871cb7e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2017-10-08 23:27:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bb92125508ea1bc27e917b1944ef4e8179116ad0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-08 23:32:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-glance-docs-ubuntu-xenial http://docs-draft.openstack.org/82/510382/2/check/gate-glance-docs-ubuntu-xenial/3edc2c3//doc/build/html/ : SUCCESS in 4m 57s\n- gate-glance-pep8-ubuntu-xenial http://logs.openstack.org/82/510382/2/check/gate-glance-pep8-ubuntu-xenial/ccbc69e/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"129770d61349d458a8c008d04c419b75fd875588","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2017-10-09 00:09:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ef303b8c7a16f16a836fe364bfa033f102d7e6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-09 00:54:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/82/510382/2/check/openstack-tox-pep8/ac604ae/ : SUCCESS in 5m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/82/510382/2/check/build-openstack-sphinx-docs/d46166b/html/ : SUCCESS in 5m 42s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/82/510382/2/check/legacy-tempest-dsvm-py35/4cc9e35/ : SUCCESS in 1h 26m 35s","accounts_in_message":[],"_revision_number":2},{"id":"bd73eec711e5938c76c6b73ede8b36c49643c1dd","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2017-10-09 17:18:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"585cc8d5a81901e078fc9a962edb3df82d8a316b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2017-10-09 18:09:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"957e3b7546f2a3506766a3669054130c6800054a","author":{"_account_id":26734,"name":"Scott McClymont","email":"scott.mcclymont@verizonwireless.com","username":"smcclymont"},"date":"2017-10-09 18:26:55.000000000","message":"Patch Set 2: Code-Review-1\n\nI am on the same page as Sean on this one I think.. I say leave the command as it was but if we absolutely must change it make sure the su step is added.","accounts_in_message":[],"_revision_number":2},{"id":"2664582c010ee069ad5480dc2146f153981770b4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2017-10-10 04:31:31.000000000","message":"Patch Set 2:\n\n\u003e I am on the same page as Sean on this one I think.. I say leave the\n \u003e command as it was but if we absolutely must change it make sure the\n \u003e su step is added.\n\nI think we must change it :) because the installation steps is broken right now. I am not sure what is the \"su step\" but it seems it is already documented in https://docs.openstack.org/install-guide/common/conventions.html#command-prompts .\n\nNote: it looks most services is using \"# mysql\". For example: https://docs.openstack.org/keystone/latest/install/keystone-install-ubuntu.html#prerequisites","accounts_in_message":[],"_revision_number":2},{"id":"053e6215f1ddae1171904096ba26f620f130a7fb","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2018-01-12 05:17:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0638bba0862501add90112c9c7710dc1ffe83087","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-31 17:35:20.000000000","message":"Patch Set 2: Code-Review+2\n\nCinder also uses \u0027# mysql\u0027 in their documentation, so I think this is fine.","accounts_in_message":[],"_revision_number":2},{"id":"c50f2c7e7d0144cfe0726c7e52d93589d0d84ec0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-06-01 15:14:21.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8abff98a20dd994a383f22fbaeaac918bcd31c92","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-06-01 15:15:00.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"445ef67561dc81c17aa146a35a258320786b42bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 16:00:11.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 510382,2","accounts_in_message":[],"_revision_number":2},{"id":"76f54ac934e8e4a089898e17bcced4549aaaed9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-06-01 21:03:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2b6138541c5941003a63a45f59385c5009a20930","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-06-01 21:03:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e2ff7133e12218228d3fb5cfedd0361033e87d45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 22:57:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/983c4dd9850d421caa0b17957e6be894 : SUCCESS in 3m 00s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/422ce7f0cb0b47f09f03cb126cbf89de : SUCCESS in 4m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/43012075e29a483b8cad6ad86638ab53 : SUCCESS in 2m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/168c3ed538184492948713f961934b05 : SUCCESS in 45m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/899c49ac94004212a63bf950485fbffe : SUCCESS in 1h 24m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d382685350e14d04a8bfa1deccfc2317 : SUCCESS in 49m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eaea0696fb94bf687e0eb9133ba9c6d : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5e11801529c4e70aec08ae65c414a59 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43c423f8e45f4e3c9e2bf2cff2f73643 : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2853fb2fa31e4663a2dadba8c4d05805 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bfde5d000a747e084403b2005e5097b : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02e97fa8c2074de4986231bfdb374f89 : SUCCESS in 7m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/aab6b22ac8cd46fb9261977bb9db1b8d : SUCCESS in 11m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/52edb731564447df86d334d0f90362da : SUCCESS in 9m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3472762b78494bb18b9b504141d53335 : SUCCESS in 7m 58s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8e3b8ea9a30743cb9921a12de83ab5d7 : SUCCESS in 4m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ab50c60adac045ab94a7a4857ed92a28 : SUCCESS in 1h 24m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7d01339072614c32b60547c75797260e : SUCCESS in 1h 13m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/592d8a30726e4d98a4ea83cd21401acf : SUCCESS in 1h 07m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7713e9695c984b799eb925c5fc838b69 : SUCCESS in 1h 47m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68ce6c83a53a41e491d496411eb43145 : SUCCESS in 50m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9dc1c59af2fa49ecadee56425b6f8e41 : FAILURE in 46m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/510b8b02b7e1468d943e6464c76d26db : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":3},{"id":"b165cc9344d43e25533747e6ca002ec26b92a6fa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-06-02 13:12:13.000000000","message":"Abandoned\n\nOK this was already fixed in 7191c7ac055404aebe43a3e6279cafe353f83762","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"168ffb3594a6d3b3f572c5a8bc66d9294e3d7ddf","revisions":{"62967b29c6d2a36aeb29225320ba112df77bb50c":{"kind":"REWORK","_number":1,"created":"2017-10-08 23:26:16.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/82/510382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/510382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/510382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/510382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/510382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/510382/1"}}},"commit":{"parents":[{"commit":"2562da28c6171cc5372d3a695af00e22134a360b","subject":"Fix 500 if user passes name with more than 80 characters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2562da28c6171cc5372d3a695af00e22134a360b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2017-10-08 23:22:44.000000000","tz":-240},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2017-10-08 23:26:15.000000000","tz":-240},"subject":"Fix DB login step in Ubuntu install guide","message":"Fix DB login step in Ubuntu install guide\n\nAccording to the latest install guide [1], root user no longer uses\npassword for local access to MariaDB server.\n\n[1] https://docs.openstack.org/install-guide/environment-sql-database-ubuntu.html\n\nChange-Id: I59ebbceed0616fa166d5cdfcde26a6594564dfa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62967b29c6d2a36aeb29225320ba112df77bb50c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62967b29c6d2a36aeb29225320ba112df77bb50c"}]},"branch":"refs/heads/master"},"84d1716c80e30cb30e52ffdf1bd78020b065795c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2017-10-08 23:27:32.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/82/510382/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/510382/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/510382/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/510382/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/510382/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/510382/2"}}},"commit":{"parents":[{"commit":"2562da28c6171cc5372d3a695af00e22134a360b","subject":"Fix 500 if user passes name with more than 80 characters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2562da28c6171cc5372d3a695af00e22134a360b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2017-10-08 23:22:44.000000000","tz":-240},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2017-10-08 23:26:55.000000000","tz":-240},"subject":"Fix DB login step in Ubuntu install guide","message":"Fix DB login step in Ubuntu install guide\n\nAccording to the latest install guide [1], root user no longer uses\npassword for local access to MariaDB server.\n\n[1] https://docs.openstack.org/install-guide/environment-sql-database-ubuntu.html\n\nCloses-Bug: #1722128\nChange-Id: I59ebbceed0616fa166d5cdfcde26a6594564dfa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/84d1716c80e30cb30e52ffdf1bd78020b065795c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/84d1716c80e30cb30e52ffdf1bd78020b065795c"}]},"branch":"refs/heads/master"},"168ffb3594a6d3b3f572c5a8bc66d9294e3d7ddf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-06-01 21:03:26.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/82/510382/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/510382/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/510382/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/510382/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/510382/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/510382/3"}}},"commit":{"parents":[{"commit":"6bd1d14a528e8454d83434e4d7e9bd8fc28a4390","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bd1d14a528e8454d83434e4d7e9bd8fc28a4390"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2017-10-08 23:22:44.000000000","tz":-240},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-06-01 21:03:26.000000000","tz":0},"subject":"Fix DB login step in Ubuntu install guide","message":"Fix DB login step in Ubuntu install guide\n\nAccording to the latest install guide [1], root user no longer uses\npassword for local access to MariaDB server.\n\n[1] https://docs.openstack.org/install-guide/environment-sql-database-ubuntu.html\n\nCloses-Bug: #1722128\nChange-Id: I59ebbceed0616fa166d5cdfcde26a6594564dfa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/168ffb3594a6d3b3f572c5a8bc66d9294e3d7ddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/168ffb3594a6d3b3f572c5a8bc66d9294e3d7ddf"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
