)]}'
{"id":"openstack%2Fglance~523666","triplet_id":"openstack%2Fglance~master~I2724e2cc0f6225ac2bec02061ced9d2e1cb71dd8","project":"openstack/glance","branch":"master","topic":"bug/1735051","attention_set":{"27248":{"account":{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},"last_update":"2024-06-25 18:30:14.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"23140":{"account":{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},"last_update":"2024-06-25 18:30:14.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"9414":{"account":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"last_update":"2023-06-27 17:06:45.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-06-28 14:15:16.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I2724e2cc0f6225ac2bec02061ced9d2e1cb71dd8","subject":"Doc: use {project,user}_domain_id instead of {project,user}_domain_name","status":"NEW","created":"2017-11-29 03:15:55.000000000","updated":"2024-06-25 18:37:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4fbbdbcc0907cefcd216d2e7d083f059c4f0db75","_number":523666,"virtual_id_number":523666,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-25 18:37:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-11-29 04:40:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-29 19:35:30.000000000","updated_by":{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},"reviewer":{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},"state":"REVIEWER"},{"updated":"2017-12-11 12:03:38.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2018-02-10 17:52:08.000000000","updated_by":{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},"reviewer":{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},"state":"REVIEWER"},{"updated":"2023-06-27 17:06:45.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2023-06-28 14:15:16.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"59b7ede9b84174343274c834e5047885f3557638","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-11-29 03:15:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"630e6c850b73d69fb4656d04d3c9fe4e00517fbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-29 04:40:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/66/523666/1/check/openstack-tox-pep8/d4f1b8c/ : SUCCESS in 4m 04s\n- build-openstack-sphinx-docs http://logs.openstack.org/66/523666/1/check/build-openstack-sphinx-docs/aa7c8a7/html/ : SUCCESS in 4m 34s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/66/523666/1/check/legacy-tempest-dsvm-py35/0220686/ : SUCCESS in 1h 23m 10s","accounts_in_message":[],"_revision_number":1},{"id":"994d8d3452672d4fd056fc229e2f949c7812991c","author":{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},"date":"2017-11-29 19:35:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ff2b8262d02d8e40b79bee22f4fb3052bd601e66","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2017-12-11 12:03:38.000000000","message":"Patch Set 1:\n\nNot sure about the validity of these instructions. Just looked through the configs we generate and I wasn\u0027t able to find these options at all.","accounts_in_message":[],"_revision_number":1},{"id":"62edc5ec6dc2951be6b0e072bad80108fa6ebe55","author":{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},"date":"2018-02-10 17:52:08.000000000","message":"Patch Set 1: Code-Review-1\n\nI have a devstack pike setup running with glance config set as (project/user)_domain_name. I looked at the bug, but did not find details to claim these attributes were the real cause of this. It could even be keystone version setting issue. It would be very helpful if, you can share more details.","accounts_in_message":[],"_revision_number":1},{"id":"94cf4de59ca2ffebd7db994c613a06c82ba1698f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-27 17:06:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f561188da25dcf9b81b159d266c013f934dd2c8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-25 18:30:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dac4dc57a023d197afa3f59fb22cabff261dc9c8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-25 18:33:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4fbbdbcc0907cefcd216d2e7d083f059c4f0db75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 18:37:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04c2f6d227154abc9b53253f98c833fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71390394366446f3813c2020171f4508 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14cf692512544191916d4c532c77bdc4 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"77ed805fd338e6126b74fe94f3a126819bfef15d","revisions":{"a4aa56f0eb474a45a3d83e53a53295b1d56a80b0":{"kind":"REWORK","_number":1,"created":"2017-11-29 03:15:55.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/66/523666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/523666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/523666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/523666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/523666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/523666/1"}}},"commit":{"parents":[{"commit":"c462d803ecf3a984706333d06a134eb6e96f50f4","subject":"Merge \"Use new oslo.context arg names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c462d803ecf3a984706333d06a134eb6e96f50f4"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2017-11-29 03:14:51.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2017-11-29 03:14:51.000000000","tz":480},"subject":"Fix project_domain_name and user_domain_name in doc","message":"Fix project_domain_name and user_domain_name in doc\n\nChange-Id: I2724e2cc0f6225ac2bec02061ced9d2e1cb71dd8\nCloses-Bug: #1735051\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a4aa56f0eb474a45a3d83e53a53295b1d56a80b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a4aa56f0eb474a45a3d83e53a53295b1d56a80b0"}]},"branch":"refs/heads/master"},"77ed805fd338e6126b74fe94f3a126819bfef15d":{"kind":"REWORK","_number":2,"created":"2024-06-25 18:30:14.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/66/523666/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/523666/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/523666/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/523666/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/523666/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/523666/2"}}},"commit":{"parents":[{"commit":"cfc9c0eb1a41f8f49343efacae4a9667ada2c791","subject":"Merge \"Make functional-py312 voting and add it to the gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfc9c0eb1a41f8f49343efacae4a9667ada2c791"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-06-25 15:16:55.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-06-25 18:24:15.000000000","tz":120},"subject":"Doc: use {project,user}_domain_id instead of {project,user}_domain_name","message":"Doc: use {project,user}_domain_id instead of {project,user}_domain_name\n\nIn devstack:\n\n$ openstack domain list\n+---------+---------+---------+--------------------+\n| ID      | Name    | Enabled | Description        |\n+---------+---------+---------+--------------------+\n| default | Default | True    | The default domain |\n+---------+---------+---------+--------------------+\n\nThe \"openstack image list\" command works with the following config\noptions:\n\n\tproject_domain_name \u003d Default\n\tuser_domain_name \u003d Default\n\nand:\n\n\tproject_domain_name \u003d default\n\tuser_domain_name \u003d default\n\nand:\n\n\tproject_domain_id \u003d default\n\tuser_domain_id \u003d default\n\nBut not with:\n\n\tproject_domain_id \u003d Default\n\tuser_domain_id \u003d Default\n\n$ openstack image list\nHttpException: 401: Client Error for url:\nhttp://192.168.121.98/image/v2/images, This server could not verify that\nyou are authorized to access the document you requested. Either you\nsupplied the wrong credentials (e.g., bad password), or your browser\ndoes not understand how to supply the credentials required.\u003cbr /\u003e\u003cbr /\u003e\n\nNames are not case-sensitive, but ids are.\n\nCinder merged a similar patch[1].\n\n[1] https://review.opendev.org/c/openstack/cinder/+/523673\n\nChange-Id: I2724e2cc0f6225ac2bec02061ced9d2e1cb71dd8\nCloses-Bug: #1735051\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77ed805fd338e6126b74fe94f3a126819bfef15d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77ed805fd338e6126b74fe94f3a126819bfef15d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
