)]}'
{"id":"openstack%2Fglance~527635","triplet_id":"openstack%2Fglance~master~I98be97c42f23b60a72d520aad5f6078a96372c59","project":"openstack/glance","branch":"master","topic":"bp/inject-automatic-metadata","hashtags":[],"change_id":"I98be97c42f23b60a72d520aad5f6078a96372c59","subject":"Implementation of Inject metadata properties","status":"MERGED","created":"2017-12-13 08:59:32.000000000","updated":"2018-01-24 09:40:03.000000000","submitted":"2018-01-24 09:40:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":37,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"527635-1516786803723-c1faceed","meta_rev_id":"8ddc27d7d6940fe8cb80a933875eeac819b4e21b","_number":527635,"virtual_id_number":527635,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-01-23 22:02:12.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"date":"2018-01-23 01:57:57.000000000","_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2018-01-23 17:58:39.000000000","_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":2,"date":"2018-01-24 09:39:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2018-01-22 04:35:55.000000000","_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2018-01-20 18:01:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":2,"date":"2018-01-19 17:40:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2018-01-20 18:01:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-14 12:52:43.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-12-21 12:19:33.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2018-01-18 19:51:40.000000000","updated_by":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"reviewer":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"state":"REVIEWER"},{"updated":"2018-01-19 17:40:06.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2018-01-20 18:01:10.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2018-01-22 04:35:55.000000000","updated_by":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"reviewer":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"state":"REVIEWER"},{"updated":"2018-01-23 01:57:57.000000000","updated_by":{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},"reviewer":{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},"state":"REVIEWER"},{"updated":"2018-01-23 22:02:12.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-24 09:39:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a254c8b4513a47b24e14e99c0cbf60fa9ef88cf","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2017-12-13 08:59:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"558c3c3fae853cd1556549a90b4791e91534eac5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-13 09:50:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/1/check/openstack-tox-py35/4c13520/ : SUCCESS in 6m 29s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/1/check/openstack-tox-pep8/d217b95/ : SUCCESS in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/1/check/openstack-tox-py27/cee475a/ : SUCCESS in 9m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/1/check/build-openstack-sphinx-docs/5e016e8/html/ : SUCCESS in 4m 46s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/1/check/legacy-tempest-dsvm-py35/b6a2d85/ : FAILURE in 18m 55s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/1/check/build-openstack-releasenotes/e35b881/html/ : SUCCESS in 4m 59s\n- build-openstack-api-ref http://logs.openstack.org/35/527635/1/check/build-openstack-api-ref/94a75af/html/ : SUCCESS in 5m 21s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/4b8177e/ : FAILURE in 22m 51s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/1/check/legacy-rally-dsvm-glance/be49c0c/ : FAILURE in 23m 38s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/1/check/legacy-rally-dsvm-py35-glance/0e48d97/ : FAILURE in 10m 16s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/35/527635/1/check/legacy-tempest-dsvm-neutron-full/6af2b75/ : FAILURE in 20m 31s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/35/527635/1/check/legacy-grenade-dsvm-neutron/a616585/ : FAILURE in 48m 19s","accounts_in_message":[],"_revision_number":1},{"id":"c6d62a89f828d26421d596bf8804406fab858e84","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2017-12-13 10:01:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6e5480900011cbff25f6b182166865fba130ec4d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-13 10:23:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/1/check-not-vote/ext-glance-zuul/91926ee : FAILURE in 21m 00s","accounts_in_message":[],"_revision_number":1},{"id":"201992ee0699e1afe69bcba1a2466ae671df959f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2017-12-13 10:30:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nIMO, To avoid zuul check failure you need to break this patch in 3 patches:\nPatch 1: create glance-image-import.conf file (all changes regarding new file will go in this patch, also you need to add sample file under etc directory as glance-image-import.conf and not as glance-image-import.conf.sample)\n\nPatch 2: this should be submitted in devstack to copy glance-image-import.conf file to /etc/glance directory (similar to https://github.com/openstack-dev/devstack/blob/master/lib/glance#L114)\n\nPatch 3: Code to inject metadata properties.\n\nAdded minor nits here","accounts_in_message":[],"_revision_number":1},{"id":"5df4d111704f59ad92533869b0aa26184f5902be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-13 11:05:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/1/check/openstack-tox-py35/e27c51b/ : SUCCESS in 7m 08s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/1/check/openstack-tox-pep8/3aea9b8/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/1/check/openstack-tox-py27/6fb5171/ : SUCCESS in 7m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/1/check/build-openstack-sphinx-docs/65e62ba/html/ : SUCCESS in 5m 03s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/1/check/legacy-tempest-dsvm-py35/8ff59a9/ : FAILURE in 19m 35s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/1/check/build-openstack-releasenotes/e1d6fa7/html/ : SUCCESS in 6m 12s\n- build-openstack-api-ref http://logs.openstack.org/35/527635/1/check/build-openstack-api-ref/c01efe0/html/ : SUCCESS in 3m 47s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/0a244ff/ : FAILURE in 23m 26s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/1/check/legacy-rally-dsvm-glance/9e6de80/ : FAILURE in 22m 05s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/1/check/legacy-rally-dsvm-py35-glance/7eb74f3/ : FAILURE in 9m 32s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/35/527635/1/check/legacy-tempest-dsvm-neutron-full/9db822d/ : FAILURE in 24m 04s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/35/527635/1/check/legacy-grenade-dsvm-neutron/d60c0f6/ : FAILURE in 1h 00m 55s","accounts_in_message":[],"_revision_number":1},{"id":"dee351d9652a83b8413e68fc4bc7607951e0330f","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2017-12-14 12:39:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2350375da352957ed0373c7324d140c27a8ae2e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-14 12:45:04.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"84772a2732207f4a9d8c131b764257f23268c05f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-12-14 12:45:05.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"b249a0b50ea3dec2381e0313213974109de2d632","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2017-12-14 12:49:10.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"8fdcc789d882f57d8a2aea5a8fa1f2d2fcc8a948","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-12-14 12:52:43.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"2baebdfa72d1046d6b5e9e5bb19dc96d653b79d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-14 12:52:46.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"2da02d591238beda370f5e9756a92ee32177252f","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2017-12-14 16:36:31.000000000","message":"Patch Set 3: Code-Review-1\n\nHi Bhagyashri,\n\nThe functionality looks like a good start. There is couple of things we need to address about the implementation \u0027though before this meets the design points that were discussed when we agreed on the feature.\n\nFirst this is tightly coupled to the api_image_impor taskflow and will execute on every import call regardless the configuration. it does not implement the hooks where these tasks can be plugged/unplugged to the flow as discussed during the design phase. What we really need to come up with is entrypoint for these new tasks to be loaded dynamically into the flow, not hardcoded into one specific taskflow. We need to have something that deployer can literally go and delete (or drop a new one and enable it) the file containing the task and glance will still run. There is no way with this implementation that the operator could prevent the property injection task being executed as part of the flow. This also means that the module needs to be self sufficient cotaining the config options within etc. instead of having all that tightly integrated to the rest of the glance codebase.\n\nWe need to be able to do this on all of the image import methods (that will be their own taskflows) from single source. I will try to find time by next week to come up with some kind of boiler plate how to do this (or if this really makes sense to you and you have an idea in mind how to do it, please feel free to ping me and propose the change to the approach).","accounts_in_message":[],"_revision_number":3},{"id":"c09c218aa73d15a375210eade57f699b8300942a","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2017-12-15 12:21:05.000000000","message":"Patch Set 3:\n\n\u003e Hi Bhagyashri,\n \u003e \n \u003e The functionality looks like a good start. There is couple of\n \u003e things we need to address about the implementation \u0027though before\n \u003e this meets the design points that were discussed when we agreed on\n \u003e the feature.\n \u003e \n \u003e First this is tightly coupled to the api_image_impor taskflow and\n \u003e will execute on every import call regardless the configuration. it\n \u003e does not implement the hooks where these tasks can be\n \u003e plugged/unplugged to the flow as discussed during the design phase.\n \u003e What we really need to come up with is entrypoint for these new\n \u003e tasks to be loaded dynamically into the flow, not hardcoded into\n \u003e one specific taskflow. We need to have something that deployer can\n \u003e literally go and delete (or drop a new one and enable it) the file\n \u003e containing the task and glance will still run. There is no way with\n \u003e this implementation that the operator could prevent the property\n \u003e injection task being executed as part of the flow. This also means\n \u003e that the module needs to be self sufficient cotaining the config\n \u003e options within etc. instead of having all that tightly integrated\n \u003e to the rest of the glance codebase.\n \u003e \n \u003e We need to be able to do this on all of the image import methods\n \u003e (that will be their own taskflows) from single source. I will try\n \u003e to find time by next week to come up with some kind of boiler plate\n \u003e how to do this (or if this really makes sense to you and you have\n \u003e an idea in mind how to do it, please feel free to ping me and\n \u003e propose the change to the approach).\n\nHi Erno, Thanks for review,\n\nAs per implementation the task \"InjectImageMetadata\" will only be execute when some property is set in glance-image-import.conf. If the property is empty (or property is not set in glance-image-import.conf inject option) then this task will not be execute. So this is not tightly coupled to the api_image_import taskflow and will not execute on every import call.","accounts_in_message":[],"_revision_number":3},{"id":"494ab3fc04fcbc3699813f92c03cce60885b4c3e","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2017-12-20 10:38:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1007fb4bb358855fd7b549bf13218bd4afd2de04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-20 10:45:47.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"041464cc0fc29ee21f869414aa591352b5b43d53","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2017-12-21 03:53:27.000000000","message":"Patch Set 4: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d2cf9c520d8e85db14a567f4920251e446d3d8cb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2017-12-21 04:00:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6558f7499d8cefae48463a065d46d5bcb1a0078","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2017-12-21 12:19:33.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\nThanks Bhagyashri,\n\nThis start to look more like it. Abhishek had some very valid comments there and added couple myself.\n\nwe\u0027re definitely going towards right direction with this.","accounts_in_message":[],"_revision_number":4},{"id":"40f0b2662fccfc39f7ab038c5e8d7237f667580d","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2017-12-22 12:26:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4a03f1e912dab0bf75ff633d2f37de32ca9aa053","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-22 14:02:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/5/check/openstack-tox-py35/1761e20/ : SUCCESS in 8m 24s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/5/check/openstack-tox-pep8/e11d9b3/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/5/check/openstack-tox-py27/4b947ff/ : SUCCESS in 7m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/5/check/build-openstack-sphinx-docs/a06d25f/html/ : SUCCESS in 4m 55s\n- tempest-full http://logs.openstack.org/35/527635/5/check/tempest-full/37f9926/ : SUCCESS in 1h 30m 47s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/35/527635/5/check/legacy-grenade-dsvm-neutron/bb0253f/ : FAILURE in 1h 14m 47s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/5/check/legacy-tempest-dsvm-py35/4a3772b/ : SUCCESS in 1h 11m 37s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/5/check/build-openstack-releasenotes/7042da1/html/ : SUCCESS in 4m 27s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/5/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/d13d4ac/ : SUCCESS in 1h 08m 15s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/5/check/legacy-rally-dsvm-glance/3640061/ : SUCCESS in 43m 07s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/5/check/legacy-rally-dsvm-py35-glance/4728ccb/ : FAILURE in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4955b7516742e45327c415fe6843f1fd8bb5e9a7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2017-12-26 05:21:16.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)\n\nAdded some comments, this looks to be promising now.","accounts_in_message":[],"_revision_number":5},{"id":"3eb0e6cbab6a023528c6610704c89c43b7f700f5","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2017-12-26 09:09:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f310333e6d154765974a5da648e71cf9eade0523","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2017-12-26 09:14:41.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\none minor change, else looks good to me","accounts_in_message":[],"_revision_number":6},{"id":"4d90f4163bdbd4239ad666858b57a5d3789ee49c","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2017-12-26 09:23:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f59e550a2cd3c606116196bbd8c02758b446d7b3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2017-12-26 09:32:43.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nsorry bhagyashri somehow I missed this comment in earlier patch.","accounts_in_message":[],"_revision_number":7},{"id":"27c3c751af5fce8f8c9802ba0753c9dea7f1bb19","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2017-12-26 09:50:07.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f7bf46206abda57d7508547e451415aa26eea12f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2017-12-26 09:56:11.000000000","message":"Patch Set 8: Code-Review+1\n\nWill +2 once zuul passes","accounts_in_message":[],"_revision_number":8},{"id":"c21f22460f2c08b27ada896238c7c1a8bc1fd792","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-26 11:38:59.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/8/check/openstack-tox-py35/0972e90/ : SUCCESS in 7m 21s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/8/check/openstack-tox-pep8/ec9fc43/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/8/check/openstack-tox-py27/d7f92a3/ : SUCCESS in 7m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/8/check/build-openstack-sphinx-docs/1044fb6/html/ : SUCCESS in 5m 01s\n- tempest-full http://logs.openstack.org/35/527635/8/check/tempest-full/541984c/ : SUCCESS in 1h 20m 02s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/35/527635/8/check/legacy-grenade-dsvm-neutron/4b80140/ : FAILURE in 43m 54s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/8/check/legacy-tempest-dsvm-py35/6cc1993/ : SUCCESS in 1h 46m 41s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/8/check/build-openstack-releasenotes/568b953/html/ : SUCCESS in 5m 42s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/8/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/c8d2828/ : SUCCESS in 1h 18m 38s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/8/check/legacy-rally-dsvm-glance/0d7db0b/ : SUCCESS in 52m 41s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/8/check/legacy-rally-dsvm-py35-glance/0711110/ : FAILURE in 8m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"850610588c694fbe1182c3e3322db21b4fe3e9d7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2018-01-10 03:27:22.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"a11e80abc552915000f43245514e054e82967559","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-10 03:50:36.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/8/check-not-vote/ext-glance-zuul/99c69ea : FAILURE in 22m 31s","accounts_in_message":[],"_revision_number":8},{"id":"953e8bcbb683b3f92c1566d5640eb36a642b4d8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-10 05:10:31.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/8/check/openstack-tox-py35/d9961da/ : FAILURE in 3m 13s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/8/check/openstack-tox-pep8/d644a1d/ : FAILURE in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/8/check/openstack-tox-py27/bf972c1/ : FAILURE in 9m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/8/check/build-openstack-sphinx-docs/b897ee9/html/ : SUCCESS in 7m 13s\n- tempest-full http://logs.openstack.org/35/527635/8/check/tempest-full/2379843/ : SUCCESS in 1h 19m 31s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/35/527635/8/check/legacy-grenade-dsvm-neutron/b39e228/ : FAILURE in 1h 00m 19s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/8/check/legacy-tempest-dsvm-py35/8bbc890/ : SUCCESS in 1h 22m 41s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/8/check/build-openstack-releasenotes/0107835/html/ : SUCCESS in 4m 47s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/8/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/21f1f1b/ : SUCCESS in 1h 11m 37s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/8/check/legacy-rally-dsvm-glance/d6c0395/ : SUCCESS in 1h 10m 00s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/8/check/legacy-rally-dsvm-py35-glance/45fa131/ : FAILURE in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a419856450936a8c05e4b8046be93bb3a02f7ec9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-01-10 08:46:51.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"2ec5f1eb85ea26c02b45c38d378eed821f57c329","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-10 09:05:32.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/8/check-not-vote/ext-glance-zuul/90e36c9 : FAILURE in 17m 49s","accounts_in_message":[],"_revision_number":8},{"id":"6a0f28695c586279d97d071eab06d2a2ab498fcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-10 10:48:31.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/8/check/openstack-tox-py35/bf5e847/ : FAILURE in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/8/check/openstack-tox-pep8/493b503/ : FAILURE in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/8/check/openstack-tox-py27/1b46f98/ : FAILURE in 4m 36s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/8/check/build-openstack-sphinx-docs/140e526/html/ : SUCCESS in 5m 10s\n- tempest-full http://logs.openstack.org/35/527635/8/check/tempest-full/cb38d70/ : FAILURE in 1h 10m 41s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/35/527635/8/check/legacy-grenade-dsvm-neutron/0e7ef47/ : FAILURE in 50m 47s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/8/check/legacy-tempest-dsvm-py35/d0277db/ : SUCCESS in 1h 25m 31s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/8/check/build-openstack-releasenotes/ba65b42/html/ : SUCCESS in 4m 50s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/8/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/dfc489c/ : SUCCESS in 1h 56m 14s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/8/check/legacy-rally-dsvm-glance/caf879e/ : SUCCESS in 46m 16s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/8/check/legacy-rally-dsvm-py35-glance/e9508c5/ : FAILURE in 7m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ba8e663e0a6769be29d72b8bd4b334c9d0d6edff","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2018-01-10 13:52:54.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"499ec45e92928845458c66a80e0541041fbdeafa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-10 15:34:55.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/8/check-not-vote/ext-glance-zuul/08849a9 : FAILURE in 26m 23s","accounts_in_message":[],"_revision_number":8},{"id":"541307a04654e89118a9fa4a852c177741ee34d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-10 16:35:13.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/8/check/openstack-tox-py35/784749a/ : SUCCESS in 8m 05s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/8/check/openstack-tox-pep8/c62e854/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/8/check/openstack-tox-py27/d5fa453/ : SUCCESS in 8m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/8/check/build-openstack-sphinx-docs/9201d3d/html/ : SUCCESS in 4m 58s\n- tempest-full http://logs.openstack.org/35/527635/8/check/tempest-full/7236d68/ : SUCCESS in 1h 16m 34s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/35/527635/8/check/legacy-grenade-dsvm-neutron/f1dddf5/ : FAILURE in 49m 06s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/8/check/legacy-tempest-dsvm-py35/81c52b2/ : SUCCESS in 1h 10m 01s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/8/check/build-openstack-releasenotes/2b11709/ : TIMED_OUT in 39m 16s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/8/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/012cf32/ : SUCCESS in 1h 09m 34s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/8/check/legacy-rally-dsvm-glance/d8ddba8/ : SUCCESS in 49m 50s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/8/check/legacy-rally-dsvm-py35-glance/d65fe7a/ : FAILURE in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"be466181c76a096ff35c727a377e690e5a04dd93","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2018-01-16 11:01:02.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3f28849dc0acbfe048f9b6c719fe966348e2ca0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-16 12:57:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/9/check/openstack-tox-py35/b1a9d43/ : SUCCESS in 11m 40s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/9/check/openstack-tox-pep8/fe66c11/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/9/check/openstack-tox-py27/8b13b1f/ : SUCCESS in 7m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/9/check/build-openstack-sphinx-docs/b215755/html/ : SUCCESS in 6m 16s\n- tempest-full http://logs.openstack.org/35/527635/9/check/tempest-full/0631991/ : SUCCESS in 1h 33m 45s\n- neutron-grenade http://logs.openstack.org/35/527635/9/check/neutron-grenade/2bf3b47/ : FAILURE in 47m 15s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/9/check/legacy-tempest-dsvm-py35/85a578c/ : SUCCESS in 1h 12m 08s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/9/check/build-openstack-releasenotes/864ddf6/html/ : SUCCESS in 5m 19s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/9/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/a58a3af/ : SUCCESS in 1h 11m 03s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/9/check/legacy-rally-dsvm-glance/d1f4366/ : SUCCESS in 49m 58s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/9/check/legacy-rally-dsvm-py35-glance/27ecad2/ : FAILURE in 7m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7e4da9405d7e2b4e430a6086c34255dccd945766","author":{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},"date":"2018-01-17 02:28:31.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"f96f6b920d7e5aa52f3f9eeb3ea90bea7352f171","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-17 02:52:43.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/9/check-not-vote/ext-glance-zuul/ccd0565 : SUCCESS in 23m 30s","accounts_in_message":[],"_revision_number":9},{"id":"06c29c71e0da473fb49e10534cf794260ee83122","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-17 04:09:59.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/9/check/openstack-tox-py35/55839a0/ : SUCCESS in 9m 35s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/9/check/openstack-tox-pep8/bd1cb1b/ : SUCCESS in 6m 23s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/9/check/openstack-tox-py27/32617df/ : SUCCESS in 9m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/9/check/build-openstack-sphinx-docs/b801be9/html/ : SUCCESS in 6m 54s\n- tempest-full http://logs.openstack.org/35/527635/9/check/tempest-full/f613eef/ : FAILURE in 1h 15m 19s\n- neutron-grenade http://logs.openstack.org/35/527635/9/check/neutron-grenade/f5208a1/ : FAILURE in 51m 12s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/9/check/legacy-tempest-dsvm-py35/edec53d/ : SUCCESS in 1h 17m 50s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/9/check/build-openstack-releasenotes/2c92ed8/html/ : SUCCESS in 5m 48s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/9/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/e26265b/ : SUCCESS in 1h 06m 09s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/9/check/legacy-rally-dsvm-glance/14477fd/ : SUCCESS in 1h 31m 09s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/9/check/legacy-rally-dsvm-py35-glance/3cb18ea/ : FAILURE in 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9802dc45b57b159e85a9a52f93acf2c7578dcf14","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2018-01-17 05:05:44.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"73f5ffb28f06fffa75fe265b4043074cd5765d0b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-17 05:33:50.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/9/check-not-vote/ext-glance-zuul/6a554a5 : SUCCESS in 27m 42s","accounts_in_message":[],"_revision_number":9},{"id":"8bb4a44e3418c6d6c3898192f3ba751b9a34e5e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-17 06:40:41.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/9/check/openstack-tox-py35/2859508/ : SUCCESS in 7m 53s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/9/check/openstack-tox-pep8/b58cc5d/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/9/check/openstack-tox-py27/7e90820/ : SUCCESS in 9m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/9/check/build-openstack-sphinx-docs/c044592/html/ : SUCCESS in 5m 12s\n- tempest-full http://logs.openstack.org/35/527635/9/check/tempest-full/3dc1fda/ : SUCCESS in 1h 15m 03s\n- neutron-grenade http://logs.openstack.org/35/527635/9/check/neutron-grenade/844a25d/ : FAILURE in 43m 47s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/9/check/legacy-tempest-dsvm-py35/b8fb954/ : SUCCESS in 1h 27m 13s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/9/check/build-openstack-releasenotes/65cde1e/html/ : SUCCESS in 6m 34s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/9/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/753b5e4/ : SUCCESS in 1h 14m 10s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/9/check/legacy-rally-dsvm-glance/bc19de3/ : SUCCESS in 46m 41s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/9/check/legacy-rally-dsvm-py35-glance/6aa1d2d/ : FAILURE in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2585d03ad859c0d8973a6b3c0cd2028deaa2759d","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2018-01-18 08:42:16.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"91535e03b84b77df4daa87008607588b00252ed5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-01-18 09:35:21.000000000","message":"Patch Set 10: Code-Review-1\n\n(3 comments)\n\nSome minor comments inside, also you should add DocImpact flag in the commit message.","accounts_in_message":[],"_revision_number":10},{"id":"ed5c09364ebc317a0578d5affc35f1c888536d97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-18 10:39:44.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/10/check/openstack-tox-py35/fd4427f/ : SUCCESS in 8m 08s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/10/check/openstack-tox-pep8/ebb624f/ : SUCCESS in 6m 13s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/10/check/openstack-tox-py27/2d84aec/ : SUCCESS in 8m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/10/check/build-openstack-sphinx-docs/8d6f058/html/ : SUCCESS in 5m 30s\n- tempest-full http://logs.openstack.org/35/527635/10/check/tempest-full/0d7ef0c/ : SUCCESS in 1h 44m 04s\n- neutron-grenade http://logs.openstack.org/35/527635/10/check/neutron-grenade/f9e9b87/ : SUCCESS in 53m 10s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/10/check/legacy-tempest-dsvm-py35/f698b60/ : SUCCESS in 1h 09m 49s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/10/check/build-openstack-releasenotes/33c57ea/html/ : SUCCESS in 5m 19s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/10/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/eea9972/ : SUCCESS in 1h 05m 38s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/10/check/legacy-rally-dsvm-glance/284ad1c/ : SUCCESS in 57m 01s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/10/check/legacy-rally-dsvm-py35-glance/0da33f8/ : FAILURE in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e51493694b5059ba862c77afcb8caf87a97dd6d3","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2018-01-18 11:24:17.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"fea070013966c046d3068d9b7a09e99802afc692","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2018-01-18 11:59:00.000000000","message":"Patch Set 10:\n\n(3 comments)\n\nThanks for review :)","accounts_in_message":[],"_revision_number":10},{"id":"ca872cf9086c74f6b4368d0125566acd822b744a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-18 12:53:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/11/check/openstack-tox-py35/c33de58/ : SUCCESS in 8m 14s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/11/check/openstack-tox-pep8/c835b75/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/11/check/openstack-tox-py27/714eff6/ : SUCCESS in 9m 08s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/11/check/build-openstack-sphinx-docs/0ca6ae2/html/ : SUCCESS in 5m 16s\n- tempest-full http://logs.openstack.org/35/527635/11/check/tempest-full/b304229/ : SUCCESS in 1h 20m 06s\n- neutron-grenade http://logs.openstack.org/35/527635/11/check/neutron-grenade/1161136/ : SUCCESS in 52m 19s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/11/check/legacy-tempest-dsvm-py35/27fce55/ : SUCCESS in 1h 12m 13s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/11/check/build-openstack-releasenotes/757220c/html/ : SUCCESS in 5m 42s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/11/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/720a5bc/ : SUCCESS in 1h 00m 00s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/11/check/legacy-rally-dsvm-glance/9544354/ : SUCCESS in 51m 20s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/11/check/legacy-rally-dsvm-py35-glance/9dc1b48/ : FAILURE in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2ad19ec05da291a6e35c2c88528b3f2b535b733a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-18 13:24:50.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/11/check-not-vote/ext-glance-zuul/cd7c45c : SUCCESS in 30m 40s","accounts_in_message":[],"_revision_number":11},{"id":"f49b92ee13f724e7b4038f79a41df87514c31a5b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-01-18 14:02:48.000000000","message":"Patch Set 11: Code-Review+2\n\nThanks bhagyashri!","accounts_in_message":[],"_revision_number":11},{"id":"077a1f139890629bd7bd33447a7a58f55b72ce93","author":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"date":"2018-01-18 19:51:40.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)\n\noverall, looks good","accounts_in_message":[],"_revision_number":11},{"id":"aaebc633fbd6627728b68bcac3c5485388ae2c0d","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2018-01-19 14:31:18.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"e5d86137f8dd03c1792547e4545dc820879187c8","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2018-01-19 14:44:21.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nThanks for review :)","accounts_in_message":[],"_revision_number":11},{"id":"c9b25d14ab51f23d3d357adc74a3a8c9fb9b8413","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-19 16:22:13.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/12/check/openstack-tox-py35/7e8028f/ : SUCCESS in 8m 17s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/12/check/openstack-tox-pep8/5f88c3f/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/12/check/openstack-tox-py27/e143008/ : SUCCESS in 12m 00s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/12/check/build-openstack-sphinx-docs/0b32ab9/html/ : SUCCESS in 5m 11s\n- tempest-full http://logs.openstack.org/35/527635/12/check/tempest-full/b0cce45/ : SUCCESS in 1h 18m 24s\n- neutron-grenade http://logs.openstack.org/35/527635/12/check/neutron-grenade/df923e5/ : SUCCESS in 57m 27s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-py35/82a1b68/ : SUCCESS in 1h 41m 32s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/12/check/build-openstack-releasenotes/ecc9f90/html/ : SUCCESS in 5m 43s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/6761495/ : SUCCESS in 1h 12m 59s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-glance/529bc9f/ : SUCCESS in 48m 10s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-py35-glance/a8d9fb3/ : FAILURE in 10m 33s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"7aa9c055c1ca7d02bb4649073bcca1bf7758a40e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-19 16:50:15.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/12/check-not-vote/ext-glance-zuul/34d7765 : SUCCESS in 27m 41s","accounts_in_message":[],"_revision_number":12},{"id":"5c419303a68ab315c18456c56a397f9218bc741c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-01-19 17:40:06.000000000","message":"Patch Set 12: Code-Review+2\n\nThank you Bhagyashree!","accounts_in_message":[],"_revision_number":12},{"id":"7cae6bf04026cd06bb5395ee01e00feb8df3ae3d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2018-01-20 18:01:10.000000000","message":"Patch Set 12: Workflow+1 Code-Review+2\n\n(5 comments)\n\nVery nice work, Bhagyashri!\n\nA few notes for follow-up are inline.","accounts_in_message":[],"_revision_number":12},{"id":"5a0ecabc3fd41aa676b6214bcc6b2419ca8b1486","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-20 18:03:44.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"cbbf5089982a2fe795ccbe0400c23bf013ca0346","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-20 20:13:36.000000000","message":"Patch Set 12:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"198c4ba3c01bd381597596eff28401becc1e59e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-21 00:06:06.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/12/gate/openstack-tox-py35/754e8ef/ : SUCCESS in 9m 03s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/12/gate/openstack-tox-pep8/d498faf/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/12/gate/openstack-tox-py27/8e8af04/ : TIMED_OUT in 32m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/12/gate/build-openstack-sphinx-docs/1a28b24/html/ : SUCCESS in 5m 06s\n- tempest-full http://logs.openstack.org/35/527635/12/gate/tempest-full/842a5d0/ : SUCCESS in 1h 14m 25s\n- neutron-grenade http://logs.openstack.org/35/527635/12/gate/neutron-grenade/0e81808/ : SUCCESS in 54m 41s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/12/gate/legacy-tempest-dsvm-py35/29ad6eb/ : SUCCESS in 1h 24m 18s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/12/gate/build-openstack-releasenotes/b1ae850/html/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":12},{"id":"d3fc39a707d4fd1579576e21ca97a1de53385892","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2018-01-21 08:11:28.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"bc1f31ecf79044a710c51627e9bcf0a45d55a317","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-21 08:54:22.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/12/check-not-vote/ext-glance-zuul/597c0aa : FAILURE in 42m 32s","accounts_in_message":[],"_revision_number":12},{"id":"993836d49b5a070e48660b255de55a2bbfe84b05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-21 09:51:27.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/12/check/openstack-tox-py35/065f457/ : SUCCESS in 7m 50s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/12/check/openstack-tox-pep8/ba4e8fa/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/12/check/openstack-tox-py27/0514736/ : TIMED_OUT in 32m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/12/check/build-openstack-sphinx-docs/4a8be90/html/ : SUCCESS in 6m 32s\n- tempest-full http://logs.openstack.org/35/527635/12/check/tempest-full/a52141f/ : SUCCESS in 1h 14m 53s\n- neutron-grenade http://logs.openstack.org/35/527635/12/check/neutron-grenade/708d893/ : SUCCESS in 53m 13s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-py35/79fcb1a/ : SUCCESS in 1h 31m 06s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/12/check/build-openstack-releasenotes/699b4e7/html/ : SUCCESS in 4m 15s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/95018b3/ : SUCCESS in 1h 02m 32s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-glance/454e21b/ : SUCCESS in 47m 39s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-py35-glance/586bf59/ : FAILURE in 9m 11s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0800c8180a3adf3c0985e603d33f6d55046a57fc","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2018-01-21 13:47:33.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"f824f0f8c6b815628b436f3af59c49c8f1a392fb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-21 14:15:18.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/12/check-not-vote/ext-glance-zuul/a978d9f : SUCCESS in 27m 20s","accounts_in_message":[],"_revision_number":12},{"id":"19a03dec8e5c262bc7d3ba1474433dfc9f155bb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-21 16:57:25.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/12/check/openstack-tox-py35/2215292/ : SUCCESS in 12m 04s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/12/check/openstack-tox-pep8/5de8cc6/ : SUCCESS in 4m 50s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/12/check/openstack-tox-py27/7f5059a/ : TIMED_OUT in 32m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/12/check/build-openstack-sphinx-docs/78b60a7/html/ : SUCCESS in 4m 37s\n- tempest-full http://logs.openstack.org/35/527635/12/check/tempest-full/cfceb05/ : SUCCESS in 1h 32m 21s\n- neutron-grenade http://logs.openstack.org/35/527635/12/check/neutron-grenade/0b6497f/ : SUCCESS in 48m 23s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-py35/c9984be/ : SUCCESS in 1h 12m 46s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/12/check/build-openstack-releasenotes/7067b06/html/ : SUCCESS in 4m 12s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/d0e7e5b/ : FAILURE in 3h 03m 00s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-glance/dfc0192/ : SUCCESS in 57m 27s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-py35-glance/9e3080d/ : FAILURE in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"cb450a5fa44f218e6d21c8dfb6be24c2c2dcb291","author":{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},"date":"2018-01-22 00:55:36.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"a8eee2f12da5763179d030ecc94859cfc36efa47","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-22 01:28:09.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/12/check-not-vote/ext-glance-zuul/6ae88ea : SUCCESS in 32m 12s","accounts_in_message":[],"_revision_number":12},{"id":"a46df7625d3df5060bd8c090d81e89149dd4b46f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-22 02:20:34.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/12/check/openstack-tox-py35/6233c15/ : SUCCESS in 8m 38s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/12/check/openstack-tox-pep8/8fb6b17/ : SUCCESS in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/12/check/openstack-tox-py27/cdd226e/ : TIMED_OUT in 32m 27s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/12/check/build-openstack-sphinx-docs/fa50882/html/ : SUCCESS in 5m 01s\n- tempest-full http://logs.openstack.org/35/527635/12/check/tempest-full/ad3a3c9/ : SUCCESS in 1h 18m 15s\n- neutron-grenade http://logs.openstack.org/35/527635/12/check/neutron-grenade/363a856/ : SUCCESS in 55m 42s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-py35/d7db0e4/ : SUCCESS in 1h 22m 42s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/12/check/build-openstack-releasenotes/54a3bef/html/ : SUCCESS in 4m 44s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/b419383/ : SUCCESS in 1h 04m 27s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-glance/d584d6f/ : SUCCESS in 52m 00s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-py35-glance/3a55637/ : FAILURE in 6m 52s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8c6e4faf52235f834373cee748e4840f0a9c36f6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-01-22 04:28:45.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"4b5af0860f76252a65421dde00aecd25fa60ce73","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-01-22 04:35:10.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"230b89b3e8ad2f6d14e392e2c2962524e3391dbc","author":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"date":"2018-01-22 04:35:55.000000000","message":"Patch Set 12:\n\n“openstack-tox-py27” failure is not because of the changes made in this patch.\nThis is failing on several other patches too.\n\nhttp://logs.openstack.org/41/536141/1/check/cross-glance-py27/12d0a3b/job-output.txt.gz\nhttp://logs.openstack.org/64/535564/1/gate/cross-glance-py27/c7c7938/job-output.txt.gz\nhttp://logs.openstack.org/37/520337/18/check/openstack-tox-py27/b54ad49/job-output.txt.gz\n\n\nLooks like an issue with this:\nhttp://logs.openstack.org/35/527635/12/check/openstack-tox-py27/cdd226e/job-output.txt.gz#_2018-01-22_01_32_59_432602","accounts_in_message":[],"_revision_number":12},{"id":"4e7890454aecd59b6ea1c5e0046d55e98ce4bc13","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-01-22 04:53:09.000000000","message":"Patch Set 12:\n\nIt\u0027s stuck while executing functional tests and then time outs.\nSome functional tests are failing with error \u0027AssertionError: False is not true : Unexpected server launch status for: api,\u0027. Certainly issue is with the gate as all functional + unit tests passes locally.","accounts_in_message":[],"_revision_number":12},{"id":"8349d7ac3f60c230540b74844196acd8755777e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-22 05:57:07.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/12/check/openstack-tox-py35/0a1990d/ : SUCCESS in 9m 04s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/12/check/openstack-tox-pep8/d3d8029/ : SUCCESS in 4m 50s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/12/check/openstack-tox-py27/f90785a/ : TIMED_OUT in 33m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/12/check/build-openstack-sphinx-docs/4eadf8d/html/ : SUCCESS in 5m 50s\n- tempest-full http://logs.openstack.org/35/527635/12/check/tempest-full/0782ad8/ : SUCCESS in 1h 22m 06s\n- neutron-grenade http://logs.openstack.org/35/527635/12/check/neutron-grenade/b07ad71/ : SUCCESS in 51m 54s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-py35/55bc37e/ : SUCCESS in 1h 12m 32s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/12/check/build-openstack-releasenotes/8d88557/html/ : SUCCESS in 5m 28s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/7a6eb41/ : SUCCESS in 1h 24m 40s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-glance/0a5e1bb/ : SUCCESS in 50m 32s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-py35-glance/69a0504/ : FAILURE in 7m 35s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e1c63024b9cf859e675e6d940b0cfb9c5348c7bf","author":{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},"date":"2018-01-23 01:57:57.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"8317f30b2fa0c415a07c94163a6cb48c8d1ee731","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-23 03:20:02.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/12/check/openstack-tox-py35/f35c464/ : SUCCESS in 7m 28s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/12/check/openstack-tox-pep8/8cabd33/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/12/check/openstack-tox-py27/6b47c2a/ : TIMED_OUT in 32m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/12/check/build-openstack-sphinx-docs/bdece98/html/ : SUCCESS in 5m 15s\n- tempest-full http://logs.openstack.org/35/527635/12/check/tempest-full/a6211b1/ : SUCCESS in 1h 17m 50s\n- neutron-grenade http://logs.openstack.org/35/527635/12/check/neutron-grenade/42f58dd/ : SUCCESS in 51m 23s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-py35/9338385/ : SUCCESS in 1h 14m 27s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/12/check/build-openstack-releasenotes/b80ed6c/html/ : SUCCESS in 4m 34s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/00e7a7f/ : SUCCESS in 1h 05m 24s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-glance/22f8fe5/ : SUCCESS in 44m 58s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-py35-glance/dde15ed/ : FAILURE in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"809138bc3abbb508ba13a83edb02740c57a1d029","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-23 04:57:07.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/12/check-not-vote/ext-glance-zuul/4cce071 : SUCCESS in 33m 08s","accounts_in_message":[],"_revision_number":12},{"id":"7781abdef2765b630296594e84ec92d177cbad70","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2018-01-23 17:58:39.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"26014f019a8ac386f14ce3258c6844c38676cb79","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-23 18:30:51.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/12/check-not-vote/ext-glance-zuul/c41439b : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":12},{"id":"40f1e8f4d8bfa6cb4aa210b2e9bf987cfaedcd0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-23 21:41:34.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/12/check/openstack-tox-py35/e8578c5/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/12/check/openstack-tox-pep8/75390ee/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/12/check/openstack-tox-py27/82a7570/ : SUCCESS in 4m 59s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/12/check/build-openstack-sphinx-docs/78eb92f/html/ : SUCCESS in 8m 55s\n- tempest-full http://logs.openstack.org/35/527635/12/check/tempest-full/28cf42f/ : SUCCESS in 1h 49m 07s\n- neutron-grenade http://logs.openstack.org/35/527635/12/check/neutron-grenade/0af0790/ : SUCCESS in 1h 36m 43s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-py35/46eb30c/ : SUCCESS in 1h 13m 48s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/12/check/build-openstack-releasenotes/e2a4574/html/ : SUCCESS in 5m 47s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/35/527635/12/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/1b0a692/ : SUCCESS in 1h 20m 41s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-glance/392e614/ : SUCCESS in 48m 34s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/35/527635/12/check/legacy-rally-dsvm-py35-glance/2408d2a/ : FAILURE in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"485a05bbbee75afd5fe8363caf1c9a86c9db5a9d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-23 22:02:12.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-glance http://207.189.188.190/logs/35/527635/12/check-not-vote/ext-glance-zuul/ce54fc2 : FAILURE in 20m 10s","accounts_in_message":[],"_revision_number":12},{"id":"05e154b5fe0b4ea0b22b42d165d830587c714565","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-23 22:04:19.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"74b4656bbb73d965acff3f79adf34a267c4127a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-24 09:39:58.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/35/527635/12/gate/openstack-tox-py35/4a918b1/ : SUCCESS in 9m 01s\n- openstack-tox-pep8 http://logs.openstack.org/35/527635/12/gate/openstack-tox-pep8/0ed1b1c/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/35/527635/12/gate/openstack-tox-py27/d15b494/ : SUCCESS in 9m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/527635/12/gate/build-openstack-sphinx-docs/52ccd87/html/ : SUCCESS in 8m 06s\n- tempest-full http://logs.openstack.org/35/527635/12/gate/tempest-full/b08aba5/ : SUCCESS in 1h 29m 03s\n- neutron-grenade http://logs.openstack.org/35/527635/12/gate/neutron-grenade/ee86166/ : SUCCESS in 50m 36s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/35/527635/12/gate/legacy-tempest-dsvm-py35/a02d0aa/ : SUCCESS in 1h 18m 40s\n- build-openstack-releasenotes http://logs.openstack.org/35/527635/12/gate/build-openstack-releasenotes/20893d4/html/ : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":12},{"id":"1f39b9ef1ca37406de0010431438f9d14a23e537","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-24 09:40:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"f5ae04b9f392fb9a9f5c5932b4c5920de98efffc","revisions":{"9ca0c59c82ef7351de27f76e38b1542d8241e70f":{"kind":"REWORK","_number":1,"created":"2017-12-13 08:59:32.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/35/527635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/527635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/527635/1"}}},"commit":{"parents":[{"commit":"830526c3ed9e49e558ac3ca17579923cc89a8abd","subject":"[import-tests] Enhance image import tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/830526c3ed9e49e558ac3ca17579923cc89a8abd"}]}],"author":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-10-24 10:42:06.000000000","tz":330},"committer":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-12-13 08:57:55.000000000","tz":330},"subject":"Implementation of Inject metadata properties","message":"Implementation of Inject metadata properties\n\nImplements the spec Inject metadata properties automatically\nto non-admin images\n\nThis commit adds new task \u0027_InjectImageMetadata\u0027 to inject\nthe metadata properties automatically to non-admin images\nat the time of creation of images using newly introduce\n\u0027image-import\u0027 api in v2.\n\nImplements: I6a7ed31d5fae677cbbc9a6f6053f79d3e9326561\nChange-Id: I98be97c42f23b60a72d520aad5f6078a96372c59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ca0c59c82ef7351de27f76e38b1542d8241e70f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ca0c59c82ef7351de27f76e38b1542d8241e70f"}]},"branch":"refs/heads/master"},"8d9fd30a6a2c955f8def53e200d3215cd0326a78":{"kind":"REWORK","_number":2,"created":"2017-12-14 12:39:13.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/35/527635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/527635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/527635/2"}}},"commit":{"parents":[{"commit":"1e34163eb238e399bef22baf8fc35e6b8bc3b9c2","subject":"Create glance-image-import.conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e34163eb238e399bef22baf8fc35e6b8bc3b9c2"}]}],"author":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-12-14 10:16:21.000000000","tz":330},"committer":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-12-14 11:55:59.000000000","tz":330},"subject":"Implementation of Inject metadata properties","message":"Implementation of Inject metadata properties\n\nImplements the spec Inject metadata properties automatically\nto non-admin images\n\nThis commit adds new task \u0027_InjectImageMetadata\u0027 to inject\nthe metadata properties automatically to non-admin images\nat the time of creation of images using newly introduce\n\u0027image-import\u0027 api in v2.\n\nDepends-On: I665507db1838a50e344d3be909d7490f1f52040c\nImplements: I6a7ed31d5fae677cbbc9a6f6053f79d3e9326561\nChange-Id: I98be97c42f23b60a72d520aad5f6078a96372c59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d9fd30a6a2c955f8def53e200d3215cd0326a78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d9fd30a6a2c955f8def53e200d3215cd0326a78"}]},"branch":"refs/heads/master"},"e11d18da79c0656f6cf7c646fd7c27db1d9c5376":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2017-12-14 12:49:10.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/35/527635/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/527635/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/527635/3"}}},"commit":{"parents":[{"commit":"160a5e683082433084e2925457adc3fcb37e7240","subject":"Create glance-image-import.conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/160a5e683082433084e2925457adc3fcb37e7240"}]}],"author":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-12-14 10:16:21.000000000","tz":330},"committer":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-12-14 12:47:22.000000000","tz":330},"subject":"Implementation of Inject metadata properties","message":"Implementation of Inject metadata properties\n\nImplements the spec Inject metadata properties automatically\nto non-admin images\n\nThis commit adds new task \u0027_InjectImageMetadata\u0027 to inject\nthe metadata properties automatically to non-admin images\nat the time of creation of images using newly introduce\n\u0027image-import\u0027 api in v2.\n\nDepends-On: I665507db1838a50e344d3be909d7490f1f52040c\nImplements: I6a7ed31d5fae677cbbc9a6f6053f79d3e9326561\nChange-Id: I98be97c42f23b60a72d520aad5f6078a96372c59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e11d18da79c0656f6cf7c646fd7c27db1d9c5376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e11d18da79c0656f6cf7c646fd7c27db1d9c5376"}]},"branch":"refs/heads/master"},"66b38037b7ed298b6c6f917e86ce83ab32312811":{"kind":"REWORK","_number":4,"created":"2017-12-20 10:38:00.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/35/527635/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/527635/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/527635/4"}}},"commit":{"parents":[{"commit":"5e6d56e1dbf3cc3de2d4bc97501169d7920c4972","subject":"Add hooks for Image Import plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e6d56e1dbf3cc3de2d4bc97501169d7920c4972"}]}],"author":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-10-24 10:42:06.000000000","tz":330},"committer":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-12-20 10:11:46.000000000","tz":330},"subject":"Implementation of Inject metadata properties","message":"Implementation of Inject metadata properties\n\nImplements the spec Inject metadata properties automatically\nto non-admin images\n\nThis commit adds new task \u0027_InjectImageMetadata\u0027 to inject\nthe metadata properties automatically to non-admin images\nat the time of creation of images using newly introduce\n\u0027image-import\u0027 api in v2.\n\nDepends-On: I665507db1838a50e344d3be909d7490f1f52040c\nImplements: I6a7ed31d5fae677cbbc9a6f6053f79d3e9326561\nChange-Id: I98be97c42f23b60a72d520aad5f6078a96372c59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66b38037b7ed298b6c6f917e86ce83ab32312811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66b38037b7ed298b6c6f917e86ce83ab32312811"}]},"branch":"refs/heads/master"},"3e11d19abbf9a3a8ab5172b9d251a6e184ef6eea":{"kind":"REWORK","_number":5,"created":"2017-12-22 12:26:32.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/35/527635/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/527635/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/527635/5"}}},"commit":{"parents":[{"commit":"e36ecf9f9ba5ef9392780fcc15b9c184ec8275fd","subject":"Add hooks for Image Import plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e36ecf9f9ba5ef9392780fcc15b9c184ec8275fd"}]}],"author":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-10-24 10:42:06.000000000","tz":330},"committer":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-12-22 12:22:33.000000000","tz":330},"subject":"Implementation of Inject metadata properties","message":"Implementation of Inject metadata properties\n\nImplements the spec Inject metadata properties automatically\nto non-admin images\n\nThis commit adds new task \u0027_InjectImageMetadata\u0027 to inject\nthe metadata properties automatically to non-admin images\nat the time of creation of images using newly introduce\n\u0027image-import\u0027 api in v2.\n\nDepends-On: I665507db1838a50e344d3be909d7490f1f52040c\nImplements: I6a7ed31d5fae677cbbc9a6f6053f79d3e9326561\nChange-Id: I98be97c42f23b60a72d520aad5f6078a96372c59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e11d19abbf9a3a8ab5172b9d251a6e184ef6eea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e11d19abbf9a3a8ab5172b9d251a6e184ef6eea"}]},"branch":"refs/heads/master"},"3915bf3dc6f7d8df3b829073dbc02aadd7f3c17f":{"kind":"REWORK","_number":6,"created":"2017-12-26 09:09:09.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/35/527635/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/527635/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/527635/6"}}},"commit":{"parents":[{"commit":"e36ecf9f9ba5ef9392780fcc15b9c184ec8275fd","subject":"Add hooks for Image Import plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e36ecf9f9ba5ef9392780fcc15b9c184ec8275fd"}]}],"author":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-10-24 10:42:06.000000000","tz":330},"committer":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-12-26 09:01:40.000000000","tz":330},"subject":"Implementation of Inject metadata properties","message":"Implementation of Inject metadata properties\n\nImplements the spec Inject metadata properties automatically\nto non-admin images\n\nThis commit adds new task \u0027_InjectImageMetadata\u0027 to inject\nthe metadata properties automatically to non-admin images\nat the time of creation of images using newly introduce\n\u0027image-import\u0027 api in v2.\n\nDepends-On: I665507db1838a50e344d3be909d7490f1f52040c\nImplements: I6a7ed31d5fae677cbbc9a6f6053f79d3e9326561\nChange-Id: I98be97c42f23b60a72d520aad5f6078a96372c59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3915bf3dc6f7d8df3b829073dbc02aadd7f3c17f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3915bf3dc6f7d8df3b829073dbc02aadd7f3c17f"}]},"branch":"refs/heads/master"},"fcbc63a6bd5bea754af9bc6df55e595081e785eb":{"kind":"REWORK","_number":7,"created":"2017-12-26 09:23:31.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/35/527635/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/527635/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/527635/7"}}},"commit":{"parents":[{"commit":"e36ecf9f9ba5ef9392780fcc15b9c184ec8275fd","subject":"Add hooks for Image Import plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e36ecf9f9ba5ef9392780fcc15b9c184ec8275fd"}]}],"author":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-10-24 10:42:06.000000000","tz":330},"committer":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-12-26 09:22:51.000000000","tz":330},"subject":"Implementation of Inject metadata properties","message":"Implementation of Inject metadata properties\n\nImplements the spec Inject metadata properties automatically\nto non-admin images\n\nThis commit adds new task \u0027_InjectImageMetadata\u0027 to inject\nthe metadata properties automatically to non-admin images\nat the time of creation of images using newly introduce\n\u0027image-import\u0027 api in v2.\n\nDepends-On: I665507db1838a50e344d3be909d7490f1f52040c\nImplements: I6a7ed31d5fae677cbbc9a6f6053f79d3e9326561\nChange-Id: I98be97c42f23b60a72d520aad5f6078a96372c59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fcbc63a6bd5bea754af9bc6df55e595081e785eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fcbc63a6bd5bea754af9bc6df55e595081e785eb"}]},"branch":"refs/heads/master"},"cc4a68a63e1a567bdf9b9487685948b69f2cc301":{"kind":"REWORK","_number":8,"created":"2017-12-26 09:50:07.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/35/527635/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/527635/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/527635/8"}}},"commit":{"parents":[{"commit":"e36ecf9f9ba5ef9392780fcc15b9c184ec8275fd","subject":"Add hooks for Image Import plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e36ecf9f9ba5ef9392780fcc15b9c184ec8275fd"}]}],"author":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-10-24 10:42:06.000000000","tz":330},"committer":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-12-26 09:47:59.000000000","tz":330},"subject":"Implementation of Inject metadata properties","message":"Implementation of Inject metadata properties\n\nImplements the spec Inject metadata properties automatically\nto non-admin images\n\nThis commit adds new task \u0027_InjectImageMetadata\u0027 to inject\nthe metadata properties automatically to non-admin images\nat the time of creation of images using newly introduce\n\u0027image-import\u0027 api in v2.\n\nDepends-On: I665507db1838a50e344d3be909d7490f1f52040c\nImplements: I6a7ed31d5fae677cbbc9a6f6053f79d3e9326561\nChange-Id: I98be97c42f23b60a72d520aad5f6078a96372c59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc4a68a63e1a567bdf9b9487685948b69f2cc301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc4a68a63e1a567bdf9b9487685948b69f2cc301"}]},"branch":"refs/heads/master"},"5879e448d31fe7ea5df0e867682bb453aa1b324b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":9,"created":"2018-01-16 11:01:02.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/35/527635/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/527635/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/527635/9"}}},"commit":{"parents":[{"commit":"fe21b1d19fc666163fa10f9a1f8c4954b640bb6f","subject":"Update scrubber documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fe21b1d19fc666163fa10f9a1f8c4954b640bb6f"}]}],"author":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-10-24 10:42:06.000000000","tz":330},"committer":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2018-01-16 11:00:16.000000000","tz":330},"subject":"Implementation of Inject metadata properties","message":"Implementation of Inject metadata properties\n\nImplements the spec Inject metadata properties automatically\nto non-admin images\n\nThis commit adds new task \u0027_InjectImageMetadata\u0027 to inject\nthe metadata properties automatically to non-admin images\nat the time of creation of images using newly introduce\n\u0027image-import\u0027 api in v2.\n\nImplements: I6a7ed31d5fae677cbbc9a6f6053f79d3e9326561\nChange-Id: I98be97c42f23b60a72d520aad5f6078a96372c59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5879e448d31fe7ea5df0e867682bb453aa1b324b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5879e448d31fe7ea5df0e867682bb453aa1b324b"}]},"branch":"refs/heads/master"},"17eb5204175fa4e5c8986755b750efb7b8298ff8":{"kind":"REWORK","_number":10,"created":"2018-01-18 08:42:16.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/35/527635/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/527635/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/527635/10"}}},"commit":{"parents":[{"commit":"08e28cd921920cf8f5401c839376b65c19a97880","subject":"Merge \"Add documentation for image import plugins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08e28cd921920cf8f5401c839376b65c19a97880"}]}],"author":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-10-24 10:42:06.000000000","tz":330},"committer":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2018-01-18 08:38:35.000000000","tz":330},"subject":"Implementation of Inject metadata properties","message":"Implementation of Inject metadata properties\n\nImplements the spec Inject metadata properties automatically\nto non-admin images\n\nThis commit adds new task \u0027_InjectImageMetadata\u0027 to inject\nthe metadata properties automatically to non-admin images\nat the time of creation of images using newly introduce\n\u0027image-import\u0027 api in v2.\n\nImplements: I6a7ed31d5fae677cbbc9a6f6053f79d3e9326561\nChange-Id: I98be97c42f23b60a72d520aad5f6078a96372c59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17eb5204175fa4e5c8986755b750efb7b8298ff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17eb5204175fa4e5c8986755b750efb7b8298ff8"}]},"branch":"refs/heads/master"},"ef70a5fa5db199887103fd7686f814f222a2a766":{"kind":"REWORK","_number":11,"created":"2018-01-18 11:24:17.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/35/527635/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/527635/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/527635/11"}}},"commit":{"parents":[{"commit":"08e28cd921920cf8f5401c839376b65c19a97880","subject":"Merge \"Add documentation for image import plugins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08e28cd921920cf8f5401c839376b65c19a97880"}]}],"author":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-10-24 10:42:06.000000000","tz":330},"committer":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2018-01-18 10:36:25.000000000","tz":330},"subject":"Implementation of Inject metadata properties","message":"Implementation of Inject metadata properties\n\nImplements the spec Inject metadata properties automatically\nto non-admin images\n\nThis commit adds new task \u0027_InjectImageMetadata\u0027 to inject\nthe metadata properties automatically to non-admin images\nat the time of creation of images using newly introduce\n\u0027image-import\u0027 api in v2.\n\nDocImpact\n\nImplements: I6a7ed31d5fae677cbbc9a6f6053f79d3e9326561\nChange-Id: I98be97c42f23b60a72d520aad5f6078a96372c59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef70a5fa5db199887103fd7686f814f222a2a766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef70a5fa5db199887103fd7686f814f222a2a766"}]},"branch":"refs/heads/master"},"f5ae04b9f392fb9a9f5c5932b4c5920de98efffc":{"kind":"REWORK","_number":12,"created":"2018-01-19 14:31:18.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/35/527635/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/527635/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/527635/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/527635/12"}}},"commit":{"parents":[{"commit":"631add1ba849eac67ecb36ea4c90b57a300a96fa","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/631add1ba849eac67ecb36ea4c90b57a300a96fa"}]}],"author":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2017-10-24 10:42:06.000000000","tz":330},"committer":{"name":"bhagyashris","email":"bhagyashri.shewale@nttdata.com","date":"2018-01-19 06:41:31.000000000","tz":330},"subject":"Implementation of Inject metadata properties","message":"Implementation of Inject metadata properties\n\nImplements the spec Inject metadata properties automatically\nto non-admin images\n\nThis commit adds new task \u0027_InjectImageMetadata\u0027 to inject\nthe metadata properties automatically to non-admin images\nat the time of creation of images using newly introduce\n\u0027image-import\u0027 api in v2.\n\nDocImpact\n\nImplements: I6a7ed31d5fae677cbbc9a6f6053f79d3e9326561\nChange-Id: I98be97c42f23b60a72d520aad5f6078a96372c59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f5ae04b9f392fb9a9f5c5932b4c5920de98efffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f5ae04b9f392fb9a9f5c5932b4c5920de98efffc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
