)]}'
{"id":"openstack%2Fglance~531498","triplet_id":"openstack%2Fglance~master~I0758590e2286205c4f6dc441241dc32c0d840dd5","project":"openstack/glance","branch":"master","topic":"bug/1712463","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-06-28 16:09:29.000000000","reason":"Change was abandoned"},"5196":{"account":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"last_update":"2023-06-28 16:09:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0758590e2286205c4f6dc441241dc32c0d840dd5","subject":"Specify a python executable for prlimits when using uwsgi","status":"ABANDONED","created":"2018-01-05 20:47:18.000000000","updated":"2023-06-28 16:09:29.000000000","total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"711cdea65fbaed3a576019a90d9835e4dcd54d75","_number":531498,"virtual_id_number":531498,"owner":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14676,"name":"Kairat Kushaev","email":"kkushaev@mirantis.com","username":"kairat_kushaev"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-05 00:52:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14676,"name":"Kairat Kushaev","email":"kkushaev@mirantis.com","username":"kairat_kushaev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14676,"name":"Kairat Kushaev","email":"kkushaev@mirantis.com","username":"kairat_kushaev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14676,"name":"Kairat Kushaev","email":"kkushaev@mirantis.com","username":"kairat_kushaev"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14676,"name":"Kairat Kushaev","email":"kkushaev@mirantis.com","username":"kairat_kushaev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-30 09:07:07.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2018-03-02 08:14:19.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2018-03-05 08:47:32.000000000","updated_by":{"_account_id":14676,"name":"Kairat Kushaev","email":"kkushaev@mirantis.com","username":"kairat_kushaev"},"reviewer":{"_account_id":14676,"name":"Kairat Kushaev","email":"kkushaev@mirantis.com","username":"kairat_kushaev"},"state":"REVIEWER"},{"updated":"2018-03-07 17:53:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-03-08 02:27:48.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-03-11 17:56:48.000000000","updated_by":{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},"reviewer":{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},"state":"REVIEWER"},{"updated":"2018-05-08 18:58:11.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2018-05-14 15:37:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2018-08-28 12:07:11.000000000","updated_by":{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},"reviewer":{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},"state":"REVIEWER"}],"messages":[{"id":"1b54087a4385b9a9782b6b512ba03aaf97170544","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-01-05 20:47:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20ab7166bfadbf6198a1c38736bfe4620ca01d41","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-01-05 20:50:18.000000000","message":"Patch Set 1:\n\nThis is going to be blocked until the corresponding oslo.concurrency patch is in the next release. The depends on won\u0027t actually enforce that.","accounts_in_message":[],"_revision_number":1},{"id":"f6f17cc795457180d5e565103607174a33df1943","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-05 22:19:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/98/531498/1/check/openstack-tox-py35/2ad1ce7/ : SUCCESS in 9m 07s\n- openstack-tox-pep8 http://logs.openstack.org/98/531498/1/check/openstack-tox-pep8/2950879/ : SUCCESS in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/98/531498/1/check/openstack-tox-py27/98d38ac/ : SUCCESS in 8m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/531498/1/check/build-openstack-sphinx-docs/9ef25ed/ : FAILURE in 5m 56s\n- tempest-full http://logs.openstack.org/98/531498/1/check/tempest-full/cd44f52/ : SUCCESS in 1h 23m 47s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/98/531498/1/check/legacy-grenade-dsvm-neutron/9c6a970/ : SUCCESS in 45m 44s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/98/531498/1/check/legacy-tempest-dsvm-py35/1d68d8d/ : SUCCESS in 1h 21m 24s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/98/531498/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/c218cc2/ : SUCCESS in 1h 03m 57s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/98/531498/1/check/legacy-rally-dsvm-glance/d3f4a5a/ : SUCCESS in 1h 23m 29s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/98/531498/1/check/legacy-rally-dsvm-py35-glance/15e9aca/ : FAILURE in 7m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8680939d4097f8453ce0a9c295a11f193acf2fa0","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-01-06 02:07:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a59c6e62096b92d6a0e129a246b8b897d406ce4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-06 03:37:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/98/531498/2/check/openstack-tox-py35/387245f/ : SUCCESS in 6m 30s\n- openstack-tox-pep8 http://logs.openstack.org/98/531498/2/check/openstack-tox-pep8/1b8e92d/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/98/531498/2/check/openstack-tox-py27/9f661db/ : SUCCESS in 7m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/531498/2/check/build-openstack-sphinx-docs/faa386f/html/ : SUCCESS in 6m 46s\n- tempest-full http://logs.openstack.org/98/531498/2/check/tempest-full/746f595/ : SUCCESS in 1h 16m 16s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/98/531498/2/check/legacy-grenade-dsvm-neutron/e33193d/ : SUCCESS in 47m 12s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/98/531498/2/check/legacy-tempest-dsvm-py35/5413a03/ : SUCCESS in 1h 25m 56s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/98/531498/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/a0b4488/ : SUCCESS in 1h 00m 02s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/98/531498/2/check/legacy-rally-dsvm-glance/df528c3/ : SUCCESS in 46m 40s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/98/531498/2/check/legacy-rally-dsvm-py35-glance/3a0e213/ : FAILURE in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"701dc9388d0d7b491daf8aa731f8cf96efee2887","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-06 04:08:15.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/98/531498/2/check-not-vote/ext-glance-zuul/e1be840 : SUCCESS in 30m 10s","accounts_in_message":[],"_revision_number":2},{"id":"60aeb9c38aaeefd89266de895c5c256ed2fa4890","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-01-11 20:30:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f0b739d0f01f4dd0e05436d645d4b27ed2539958","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-11 21:57:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/98/531498/3/check/openstack-tox-py35/2de5d43/ : SUCCESS in 7m 03s\n- openstack-tox-pep8 http://logs.openstack.org/98/531498/3/check/openstack-tox-pep8/ce62b35/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/98/531498/3/check/openstack-tox-py27/b5dc022/ : SUCCESS in 7m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/531498/3/check/build-openstack-sphinx-docs/a68368b/html/ : SUCCESS in 5m 05s\n- tempest-full http://logs.openstack.org/98/531498/3/check/tempest-full/75cb190/ : SUCCESS in 1h 22m 34s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/98/531498/3/check/legacy-grenade-dsvm-neutron/18a1a6e/ : SUCCESS in 50m 20s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/98/531498/3/check/legacy-tempest-dsvm-py35/f69aeee/ : SUCCESS in 1h 09m 16s\n- build-openstack-releasenotes http://logs.openstack.org/98/531498/3/check/build-openstack-releasenotes/b28a725/html/ : SUCCESS in 4m 52s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/98/531498/3/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/a493ba2/ : SUCCESS in 1h 05m 38s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/98/531498/3/check/legacy-rally-dsvm-glance/3826fdb/ : SUCCESS in 43m 25s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/98/531498/3/check/legacy-rally-dsvm-py35-glance/029e99c/ : FAILURE in 9m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5cbae1b430810d09994b46cffcf4e62070025a09","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-11 22:32:22.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/98/531498/3/check-not-vote/ext-glance-zuul/4c62152 : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":3},{"id":"0e07b1b6ef4215a62a8c326d50aae8395e2b5385","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-01-15 21:01:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3962d5542a9a7528734c3f6f064a5ebddad4f151","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-01-15 21:02:14.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"92216797336a92cd58b374db445ed1dc1758e382","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-15 23:26:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/98/531498/5/check/openstack-tox-py35/5771e2d/ : SUCCESS in 8m 41s\n- openstack-tox-pep8 http://logs.openstack.org/98/531498/5/check/openstack-tox-pep8/3caf20c/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/98/531498/5/check/openstack-tox-py27/8bdf1f9/ : SUCCESS in 8m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/531498/5/check/build-openstack-sphinx-docs/8eb1cd2/html/ : SUCCESS in 6m 34s\n- requirements-check http://logs.openstack.org/98/531498/5/check/requirements-check/eb044ce/ : SUCCESS in 2m 19s\n- tempest-full http://logs.openstack.org/98/531498/5/check/tempest-full/f875562/ : SUCCESS in 1h 32m 42s\n- neutron-grenade http://logs.openstack.org/98/531498/5/check/neutron-grenade/1773037/ : SUCCESS in 45m 23s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/98/531498/5/check/legacy-tempest-dsvm-py35/537f36a/ : SUCCESS in 1h 09m 47s\n- build-openstack-releasenotes http://logs.openstack.org/98/531498/5/check/build-openstack-releasenotes/cd2a518/html/ : SUCCESS in 5m 21s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/98/531498/5/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/3488826/ : SUCCESS in 1h 03m 08s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/98/531498/5/check/legacy-rally-dsvm-glance/102753c/ : SUCCESS in 48m 53s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/98/531498/5/check/legacy-rally-dsvm-py35-glance/824b0c4/ : FAILURE in 9m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"357c9a9ebda3222ce5d69324f36e60546514fa69","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-01-16 09:28:53.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)\n\nTypo with a suggestion inline.\nThanks matt for working on this.","accounts_in_message":[],"_revision_number":5},{"id":"96cdeb2b6ca3236370a270b2f7c7889b04238e8e","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-01-16 15:14:33.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"82a1566705a6e28f4c2b1be5017c50f938f1d1f4","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-01-16 15:14:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8da46e336ee4370f77db6a6802a254385c16002e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-16 18:02:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/98/531498/6/check/openstack-tox-py35/06cff29/ : SUCCESS in 8m 11s\n- openstack-tox-pep8 http://logs.openstack.org/98/531498/6/check/openstack-tox-pep8/549f995/ : SUCCESS in 5m 41s\n- openstack-tox-py27 http://logs.openstack.org/98/531498/6/check/openstack-tox-py27/d2bc601/ : SUCCESS in 7m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/531498/6/check/build-openstack-sphinx-docs/114386d/html/ : SUCCESS in 5m 25s\n- requirements-check http://logs.openstack.org/98/531498/6/check/requirements-check/416e830/ : SUCCESS in 2m 35s\n- tempest-full http://logs.openstack.org/98/531498/6/check/tempest-full/a940736/ : SUCCESS in 1h 16m 43s\n- neutron-grenade http://logs.openstack.org/98/531498/6/check/neutron-grenade/df1f566/ : SUCCESS in 1h 08m 15s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/98/531498/6/check/legacy-tempest-dsvm-py35/c202cf5/ : SUCCESS in 1h 09m 47s\n- build-openstack-releasenotes http://logs.openstack.org/98/531498/6/check/build-openstack-releasenotes/af534f2/html/ : SUCCESS in 4m 53s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/98/531498/6/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/232937f/ : SUCCESS in 1h 07m 21s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/98/531498/6/check/legacy-rally-dsvm-glance/9629e68/ : SUCCESS in 47m 34s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/98/531498/6/check/legacy-rally-dsvm-py35-glance/f5aaf9b/ : FAILURE in 7m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b636f25fe99b33f9a72eafb8d7c9b635ca170cd2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-01-19 05:29:40.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c9f3fecb97f20f870ad8fec7210ac08870e3e495","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-01-22 20:24:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4d6dbf7a118d75bc6517e742812f34fb25ce594e","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-01-22 20:29:47.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"b5a3f08c81cb84dc5636602b0bf6e00499d6e876","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-01-22 20:35:34.000000000","message":"Patch Set 7:\n\nAbishek you\u0027ll have to remove your -1 because it was just a rebase (as I indicated earlier) to remove that line like you wanted. Clean rebases persist votes, because they aren\u0027t functional revisions to the patch...","accounts_in_message":[],"_revision_number":7},{"id":"25293c712046a4e70da11c2cc868f4a6e818a1ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-22 22:21:18.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/98/531498/7/check/openstack-tox-py35/a534af2/ : SUCCESS in 9m 04s\n- openstack-tox-pep8 http://logs.openstack.org/98/531498/7/check/openstack-tox-pep8/7025603/ : SUCCESS in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/98/531498/7/check/openstack-tox-py27/7c785d4/ : TIMED_OUT in 33m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/531498/7/check/build-openstack-sphinx-docs/e313291/html/ : SUCCESS in 5m 11s\n- tempest-full http://logs.openstack.org/98/531498/7/check/tempest-full/e1f3c1c/ : SUCCESS in 1h 15m 22s\n- neutron-grenade http://logs.openstack.org/98/531498/7/check/neutron-grenade/ad6e822/ : SUCCESS in 51m 42s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/98/531498/7/check/legacy-tempest-dsvm-py35/a1645bd/ : SUCCESS in 1h 15m 23s\n- build-openstack-releasenotes http://logs.openstack.org/98/531498/7/check/build-openstack-releasenotes/ae64d2b/html/ : SUCCESS in 4m 19s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/98/531498/7/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/edf8a2e/ : SUCCESS in 1h 11m 14s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/98/531498/7/check/legacy-rally-dsvm-glance/febd5f5/ : SUCCESS in 1h 00m 21s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/98/531498/7/check/legacy-rally-dsvm-py35-glance/c142216/ : FAILURE in 8m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3b98301187a216e4bfac783dd84692eb3a6912ef","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-01-30 09:07:07.000000000","message":"Patch Set 7: -Code-Review","accounts_in_message":[],"_revision_number":7},{"id":"fd5feaccc0d787481e84f2cd7405c55c31c60b81","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-02-07 16:57:35.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9c7ca6bf0938f964b476c24be30c189ee99ae1dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-07 18:36:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/98/531498/8/check/openstack-tox-py35/7632558/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/98/531498/8/check/openstack-tox-pep8/78e3e6e/ : FAILURE in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/98/531498/8/check/openstack-tox-py27/ed071dd/ : SUCCESS in 4m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/531498/8/check/build-openstack-sphinx-docs/fbb0625/html/ : SUCCESS in 5m 04s\n- tempest-full http://logs.openstack.org/98/531498/8/check/tempest-full/2d79c44/ : SUCCESS in 1h 22m 23s\n- neutron-grenade http://logs.openstack.org/98/531498/8/check/neutron-grenade/1f3f476/ : SUCCESS in 53m 07s\n- tempest-full-py3 http://logs.openstack.org/98/531498/8/check/tempest-full-py3/dec28d5/ : SUCCESS in 1h 10m 07s\n- build-openstack-releasenotes http://logs.openstack.org/98/531498/8/check/build-openstack-releasenotes/78561f2/html/ : SUCCESS in 4m 52s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/98/531498/8/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/b9ae4b2/ : SUCCESS in 1h 01m 10s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/98/531498/8/check/legacy-rally-dsvm-glance/e22e82c/ : SUCCESS in 44m 42s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/98/531498/8/check/legacy-rally-dsvm-py35-glance/00dc642/ : FAILURE in 5m 50s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/98/531498/8/check/openstack-tox-functional/5432cdf/ : SUCCESS in 6m 58s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/531498/8/check/openstack-tox-functional-py35/53bd22c/ : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":8},{"id":"85337fb29637873e4f7e80e8a10f8f6bf875114c","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-02-07 21:10:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9dece0f123d68be799c98b8ca4c7ef314c9eb61f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-07 23:09:59.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/98/531498/9/check/openstack-tox-py35/817980c/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/98/531498/9/check/openstack-tox-pep8/3c0ee54/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/98/531498/9/check/openstack-tox-py27/78cc675/ : SUCCESS in 4m 34s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/531498/9/check/build-openstack-sphinx-docs/a54ed67/html/ : SUCCESS in 5m 05s\n- tempest-full http://logs.openstack.org/98/531498/9/check/tempest-full/20e6e57/ : SUCCESS in 1h 34m 53s\n- neutron-grenade http://logs.openstack.org/98/531498/9/check/neutron-grenade/d49a64b/ : SUCCESS in 56m 11s\n- tempest-full-py3 http://logs.openstack.org/98/531498/9/check/tempest-full-py3/119ee11/ : SUCCESS in 1h 12m 16s\n- build-openstack-releasenotes http://logs.openstack.org/98/531498/9/check/build-openstack-releasenotes/ad2684f/html/ : SUCCESS in 4m 56s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/98/531498/9/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/58e601f/ : SUCCESS in 1h 01m 26s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/98/531498/9/check/legacy-rally-dsvm-glance/413b14e/ : SUCCESS in 50m 48s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/98/531498/9/check/legacy-rally-dsvm-py35-glance/08dac51/ : FAILURE in 7m 18s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/98/531498/9/check/openstack-tox-functional/66ae7a5/ : SUCCESS in 9m 14s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/531498/9/check/openstack-tox-functional-py35/2b1e601/ : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":9},{"id":"f9190db0ed60a0bea2c210f1b5ebfceaae8d535d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-02-08 00:05:28.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/98/531498/9/check-not-vote/ext-glance-zuul/f519ec5 : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":9},{"id":"565f52d815bfec29df1829ad4570b5a9f5e3a1e4","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-02-09 16:49:20.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"0640dd326b3a3785bf9a58336ef608660aceeedd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-09 19:07:47.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/98/531498/10/check/openstack-tox-py35/3161cc4/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/98/531498/10/check/openstack-tox-pep8/75b0e01/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/98/531498/10/check/openstack-tox-py27/7b5e9b6/ : SUCCESS in 4m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/531498/10/check/build-openstack-sphinx-docs/28f959f/html/ : SUCCESS in 5m 14s\n- tempest-full http://logs.openstack.org/98/531498/10/check/tempest-full/b5f4f12/ : SUCCESS in 1h 44m 31s\n- neutron-grenade http://logs.openstack.org/98/531498/10/check/neutron-grenade/3b4ebd0/ : SUCCESS in 53m 50s\n- tempest-full-py3 http://logs.openstack.org/98/531498/10/check/tempest-full-py3/f1fdc5e/ : SUCCESS in 1h 13m 52s\n- build-openstack-releasenotes http://logs.openstack.org/98/531498/10/check/build-openstack-releasenotes/c3a1cbc/html/ : SUCCESS in 6m 48s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/98/531498/10/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/60534a5/ : SUCCESS in 1h 03m 36s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/98/531498/10/check/legacy-rally-dsvm-glance/661dea5/ : SUCCESS in 51m 11s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/98/531498/10/check/legacy-rally-dsvm-py35-glance/b0b5c5d/ : FAILURE in 10m 32s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/98/531498/10/check/openstack-tox-functional/2221d40/ : SUCCESS in 8m 34s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/531498/10/check/openstack-tox-functional-py35/6995c0b/ : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":10},{"id":"22b982c72e43816c6ec5adb19fefbc234b764963","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-02-09 19:40:28.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/98/531498/10/check-not-vote/ext-glance-zuul/9ec9260 : SUCCESS in 32m 19s","accounts_in_message":[],"_revision_number":10},{"id":"74e4a6644a7eb2b5dec3784f95865345704d573c","author":{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},"date":"2018-02-10 16:15:23.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)\n\nI was just wondering what if, we could search the python binary in the system instead of fixing it to /usr/bin/\u003c\u003e","accounts_in_message":[],"_revision_number":10},{"id":"2bf6b1920791d565f30d3b3f93c4cb2b2626726c","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-02-13 17:34:34.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nThe python path isn\u0027t known to the running python code because of what is mentioned in the commit message. You could attempt to figure it out by investigating $PATH etc. Except that would not be a guarantee on all environments and more likely to get things wrong. Trying to be more clever here is more effort than it\u0027s worth. \u0027/usr/bin/python*\u0027 is just a default, and frankly one that will likely work for the majority of users. If it doesn\u0027t you just set it to what your environment demands.","accounts_in_message":[],"_revision_number":10},{"id":"ca52ce42b1411307d2e17ce508add0ca8c21a5cd","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-02-13 17:34:45.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5cb740a982657c81e03a12dade5c0064ca639111","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-13 19:16:22.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/98/531498/11/check/openstack-tox-py35/bf4c9c5/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/98/531498/11/check/openstack-tox-pep8/ad461b5/ : SUCCESS in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/98/531498/11/check/openstack-tox-py27/3646e9f/ : SUCCESS in 4m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/531498/11/check/build-openstack-sphinx-docs/5519c1d/html/ : SUCCESS in 4m 57s\n- tempest-full http://logs.openstack.org/98/531498/11/check/tempest-full/1bddba1/ : SUCCESS in 1h 17m 35s\n- neutron-grenade http://logs.openstack.org/98/531498/11/check/neutron-grenade/6eed175/ : SUCCESS in 50m 22s\n- tempest-full-py3 http://logs.openstack.org/98/531498/11/check/tempest-full-py3/fd8edb2/ : SUCCESS in 1h 22m 12s\n- build-openstack-releasenotes http://logs.openstack.org/98/531498/11/check/build-openstack-releasenotes/d83bfc7/html/ : SUCCESS in 7m 27s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/98/531498/11/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/7d6517f/ : SUCCESS in 1h 02m 21s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/98/531498/11/check/legacy-rally-dsvm-glance/1ef49c5/ : SUCCESS in 44m 08s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/98/531498/11/check/legacy-rally-dsvm-py35-glance/98d7d9c/ : FAILURE in 6m 09s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/98/531498/11/check/openstack-tox-functional/aa22b6c/ : SUCCESS in 8m 24s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/531498/11/check/openstack-tox-functional-py35/36b89e5/ : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":11},{"id":"b14ee1a0e5bb845de7634c0d3cd307824f6e1c76","author":{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},"date":"2018-02-15 10:37:25.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"bf218626345e997000efaa8e07d5df55acc02c38","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2018-02-19 13:54:03.000000000","message":"Patch Set 11:\n\nWe\u0027re going to hold off on this until rocky-1 after we\u0027ve had a chance to talk at the PTG.  I\u0027m in agreement with the general strategy on this patch.  The sticking point is that the Glance community at the 15 Feb meeting feels that merging this may give the wrong impression about the current state of Glance uwsgi support.  This is a key patch to get one of the Glance tasks to work correctly under uwsgi, but currently Glance tasks don\u0027t run under uwsgi.","accounts_in_message":[],"_revision_number":11},{"id":"b9ae6078d5df9d2d898dc69873c11ff04e27296d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2018-03-02 08:14:19.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)\n\nTwo quick changes.  They\u0027re nits, but the releasenotes and config help text are user-facing, so it would be good to correct them.","accounts_in_message":[],"_revision_number":11},{"id":"4f915c92e755c0aa56e24e50728a780a2d6729b6","author":{"_account_id":14676,"name":"Kairat Kushaev","email":"kkushaev@mirantis.com","username":"kairat_kushaev"},"date":"2018-03-05 08:47:32.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"36bc24290029fe5f4d9ee694ee768ce5e9eee9f9","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-03-07 15:39:40.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"f6e2d79a078a4ae2d4c6b06d0fff3d1f7d44faf8","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2018-03-07 15:40:01.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5171cabc867d2fd8098459923152ca8040fb30c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-07 17:53:12.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/98/531498/12/check/openstack-tox-py35/a7ebec1/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/98/531498/12/check/openstack-tox-pep8/69abcd0/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/98/531498/12/check/openstack-tox-py27/5152d6d/ : SUCCESS in 4m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/531498/12/check/build-openstack-sphinx-docs/15a50f5/html/ : SUCCESS in 4m 58s\n- tempest-full http://logs.openstack.org/98/531498/12/check/tempest-full/6f7b3c9/ : SUCCESS in 1h 38m 00s\n- neutron-grenade http://logs.openstack.org/98/531498/12/check/neutron-grenade/fe9ba93/ : SUCCESS in 44m 00s\n- tempest-full-py3 http://logs.openstack.org/98/531498/12/check/tempest-full-py3/f4ddb14/ : SUCCESS in 1h 25m 12s\n- build-openstack-releasenotes http://logs.openstack.org/98/531498/12/check/build-openstack-releasenotes/723adcd/html/ : SUCCESS in 5m 32s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/98/531498/12/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/5434274/ : SUCCESS in 1h 17m 30s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/98/531498/12/check/openstack-tox-functional/be16118/ : SUCCESS in 7m 42s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/531498/12/check/openstack-tox-functional-py35/a240a1d/ : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":12},{"id":"7cd1d2d4728d51fed6dcfb1ccc28b295ecd79ee2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-03-08 02:27:48.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/98/531498/12/check-not-vote/ext-glance-zuul/b590504 : SUCCESS in 28m 18s","accounts_in_message":[],"_revision_number":12},{"id":"55ac756267274343e57a17c74a16bd7ec78a20fe","author":{"_account_id":27248,"name":"Prateek Goel","email":"prateek.goel2@tatacommunications.com","username":"prateek.goel"},"date":"2018-03-11 17:56:48.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"733def7948ceaf78687aa1a46892b6500e76af39","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2018-05-08 18:58:11.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"45f68999da61291c680d9427f76ed0e5d300cd53","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-05-14 15:37:15.000000000","message":"Patch Set 12: Code-Review+2\n\nZane has a good suggestion, but I think that could be done in a follow up. This is definite improvement.","accounts_in_message":[],"_revision_number":12},{"id":"ebcb4a5d59f1f16bfb5ab7e9f16e3e59f045cef5","author":{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},"date":"2018-08-28 12:07:11.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"214fb3c6b4f133a7a68aa4e0c9c1cc5990ce6a63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-04 21:06:46.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d8f20cf61de0b1a774390ca34eb3a86189584d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 23:01:28.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c55415c98bc40c78a5a5726c7fd58db\n\n- grenade https://zuul.opendev.org/t/openstack/build/ca341d2ce59e47528e1411870ee8de8c : SUCCESS in 58m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6f3fb7494f764d9da84eea6319c9a5c6 : SUCCESS in 1h 18m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63b9b4c9d29045bbb20b2be34bd3dc70 : SUCCESS in 55m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9665c1da2754471b88c991334d17fed : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5775d54d1219402fa4200e476cbdcc39 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4390cd8569c4e528520568b99888ffc : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04da1aa8e81c42a8bcf892a7c11368ae : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6284a1c2f204dfabbaeb048b69ba09a : FAILURE in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bffd53b34bda4d12bb5de6d89059801e : SUCCESS in 5m 11s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2a2c2a37220d421999d4bdbbfb2f066b : SUCCESS in 13m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/79a6c336910b4dc09e44a71889b0febe : SUCCESS in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f5f51b9a3b544cf3b4d0b6289fd06cb1 : SUCCESS in 8m 46s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/034820b7b90b4e02a096ffe1d52527ca : SUCCESS in 4m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0b1008b9fafa42f7bf4aec2289e3a4ee : SUCCESS in 1h 32m 00s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/42632e7e7bd84c5880ae4d1ad05a8780 : SUCCESS in 1h 38m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9b200ebecc9e41889ec33d40cc5f2b9d : SUCCESS in 1h 15m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/96dc3ec2f4584adaa2c2d287fcd31276 : SUCCESS in 1h 48m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/823e078115fb486486a3c887eedab2dd : SUCCESS in 1h 10m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cebcedb392b1472cb6b7619a9bd9382c : SUCCESS in 1h 29m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/813c5d9c075b4b859cfe08c674c36b47 : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":13},{"id":"3805d4164351098ad8d40b1cd43a40903c50a0a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-04 23:25:15.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"5c2547db87872f072460ae7b715ce859119dee33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 00:52:56.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c14e6df715f4e898b0e5399d4096763\n\n- grenade https://zuul.opendev.org/t/openstack/build/9c0966de11f249139534e205630317dc : SUCCESS in 52m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b44c7cf8953479583539371bd8a0175 : SUCCESS in 1h 20m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb711f1729fb4a3891834d3a3dffa68e : SUCCESS in 58m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c2518820d3347a7a0752439aab593cd : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcd227e5b8df4bf8aa70660eb5b4f6d2 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdcb36a50afd4b5592fbd511318ef23a : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e8083411356476f84323b2ce4c62d46 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a799b003bbb481a96c753a8b0dac4fd : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9a56e0d28f54818bec4455e444d2c61 : SUCCESS in 4m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ef2887bbd6ef44e5bf1d5c088d58fe8f : SUCCESS in 10m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f9c80124ea614b3886a9af09fc031971 : SUCCESS in 7m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/318066f0491e431ea3c5cad184fe2180 : SUCCESS in 8m 03s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/77b492b91b7e4985b733383d76fc56b0 : SUCCESS in 3m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/76e68aeb92374abd88feee02bf6f318d : SUCCESS in 1h 19m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8682564a0f6f4224bccdf9af05626405 : SUCCESS in 1h 12m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/242c4226bc054172a32f985219a08017 : SUCCESS in 1h 10m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/95e57773fe7742bba56b89e7da40e573 : SUCCESS in 1h 23m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e509427e688402990a2ae9ae56ac2d4 : SUCCESS in 47m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0b6ed00788624cf08c1f5f69f401c58d : SUCCESS in 58m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/130ca90affd3407bb42289ea05731ebe : SUCCESS in 33m 04s","accounts_in_message":[],"_revision_number":14},{"id":"711cdea65fbaed3a576019a90d9835e4dcd54d75","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-28 16:09:29.000000000","message":"Abandoned\n\nAlready fixed with I7cb2e135d6ea2cb21de55060df3f7bf40b3e64b6","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"30c615892bb31ae2b93b117c39944bbdc1f42875","revisions":{"9adc8bf850c599b8ba2a1e93bdeb9317dd3b6352":{"kind":"REWORK","_number":1,"created":"2018-01-05 20:47:18.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/98/531498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/1"}}},"commit":{"parents":[{"commit":"05e9bdb656d9c120ed3cd6ffc8ae7dbf5614b5e4","subject":"Fix member create to handle unicode characters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05e9bdb656d9c120ed3cd6ffc8ae7dbf5614b5e4"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:47:17.000000000","tz":-300},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nPartial-Bug: #1712463\n\nDepends-On: I0ae60f0b4cc3700c783f6018e837358f0e053a09\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9adc8bf850c599b8ba2a1e93bdeb9317dd3b6352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9adc8bf850c599b8ba2a1e93bdeb9317dd3b6352"}]},"branch":"refs/heads/master"},"7c3fc1c3587f76f07434b3f5b637b65822dc8a88":{"kind":"REWORK","_number":2,"created":"2018-01-06 02:07:49.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/98/531498/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/2"}}},"commit":{"parents":[{"commit":"05e9bdb656d9c120ed3cd6ffc8ae7dbf5614b5e4","subject":"Fix member create to handle unicode characters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05e9bdb656d9c120ed3cd6ffc8ae7dbf5614b5e4"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-06 02:03:50.000000000","tz":-300},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nPartial-Bug: #1712463\n\nDepends-On: I0ae60f0b4cc3700c783f6018e837358f0e053a09\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7c3fc1c3587f76f07434b3f5b637b65822dc8a88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7c3fc1c3587f76f07434b3f5b637b65822dc8a88"}]},"branch":"refs/heads/master"},"540f2915b6a990235b05647d1e04c61441ad6d2f":{"kind":"REWORK","_number":3,"created":"2018-01-11 20:30:00.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/98/531498/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/3"}}},"commit":{"parents":[{"commit":"05e9bdb656d9c120ed3cd6ffc8ae7dbf5614b5e4","subject":"Fix member create to handle unicode characters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05e9bdb656d9c120ed3cd6ffc8ae7dbf5614b5e4"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-11 20:29:20.000000000","tz":-300},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nCloses-Bug: #1712463\n\nDepends-On: I0ae60f0b4cc3700c783f6018e837358f0e053a09\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/540f2915b6a990235b05647d1e04c61441ad6d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/540f2915b6a990235b05647d1e04c61441ad6d2f"}]},"branch":"refs/heads/master"},"ca583f62b0b0175db867105c56717c919de8db8e":{"kind":"REWORK","_number":4,"created":"2018-01-15 21:01:09.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/98/531498/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/4"}}},"commit":{"parents":[{"commit":"05e9bdb656d9c120ed3cd6ffc8ae7dbf5614b5e4","subject":"Fix member create to handle unicode characters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05e9bdb656d9c120ed3cd6ffc8ae7dbf5614b5e4"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-15 21:01:01.000000000","tz":-300},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nCloses-Bug: #1712463\n\nDepends-On: I0ae60f0b4cc3700c783f6018e837358f0e053a09\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ca583f62b0b0175db867105c56717c919de8db8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ca583f62b0b0175db867105c56717c919de8db8e"}]},"branch":"refs/heads/master"},"e3955037bc86bd7da3a9ae22c2c3147f78a7334b":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2018-01-15 21:02:14.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/98/531498/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/5"}}},"commit":{"parents":[{"commit":"05e9bdb656d9c120ed3cd6ffc8ae7dbf5614b5e4","subject":"Fix member create to handle unicode characters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05e9bdb656d9c120ed3cd6ffc8ae7dbf5614b5e4"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-15 21:01:34.000000000","tz":-300},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nCloses-Bug: #1712463\n\nDepends-On: I2b2b0d865224413254051581b382ff206aae0180\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3955037bc86bd7da3a9ae22c2c3147f78a7334b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3955037bc86bd7da3a9ae22c2c3147f78a7334b"}]},"branch":"refs/heads/master"},"66bc7c31dd9724c92860e76fbbc7359b5401e1ac":{"kind":"REWORK","_number":6,"created":"2018-01-16 15:14:40.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/98/531498/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/6"}}},"commit":{"parents":[{"commit":"05e9bdb656d9c120ed3cd6ffc8ae7dbf5614b5e4","subject":"Fix member create to handle unicode characters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05e9bdb656d9c120ed3cd6ffc8ae7dbf5614b5e4"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-16 15:14:24.000000000","tz":-300},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nCloses-Bug: #1712463\n\nDepends-On: I2b2b0d865224413254051581b382ff206aae0180\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66bc7c31dd9724c92860e76fbbc7359b5401e1ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66bc7c31dd9724c92860e76fbbc7359b5401e1ac"}]},"branch":"refs/heads/master"},"ab3dec749c6e0f36d6f88a4e44ec7a41769fee6c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2018-01-22 20:29:47.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/98/531498/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/7"}}},"commit":{"parents":[{"commit":"631add1ba849eac67ecb36ea4c90b57a300a96fa","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/631add1ba849eac67ecb36ea4c90b57a300a96fa"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-22 20:29:10.000000000","tz":-300},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nCloses-Bug: #1712463\n\nDepends-On: I2b2b0d865224413254051581b382ff206aae0180\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab3dec749c6e0f36d6f88a4e44ec7a41769fee6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab3dec749c6e0f36d6f88a4e44ec7a41769fee6c"}]},"branch":"refs/heads/master"},"e95155645a6fc0676d8b65e6ec35a019e9e7f7b5":{"kind":"REWORK","_number":8,"created":"2018-02-07 16:57:35.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/98/531498/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/8"}}},"commit":{"parents":[{"commit":"8682876a8b5d18cd678be061ad8bb64aaeb3dcff","subject":"Exiting with user friendly message and SystemExit()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8682876a8b5d18cd678be061ad8bb64aaeb3dcff"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-02-07 16:57:06.000000000","tz":-300},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nCloses-Bug: #1712463\n\nDepends-On: I2b2b0d865224413254051581b382ff206aae0180\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e95155645a6fc0676d8b65e6ec35a019e9e7f7b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e95155645a6fc0676d8b65e6ec35a019e9e7f7b5"}]},"branch":"refs/heads/master"},"ae79b3fc12e0f4f3559a293e4ccfa0f853902a1f":{"kind":"REWORK","_number":9,"created":"2018-02-07 21:10:31.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/98/531498/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/9"}}},"commit":{"parents":[{"commit":"8682876a8b5d18cd678be061ad8bb64aaeb3dcff","subject":"Exiting with user friendly message and SystemExit()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8682876a8b5d18cd678be061ad8bb64aaeb3dcff"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-02-07 21:10:26.000000000","tz":-300},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nCloses-Bug: #1712463\n\nDepends-On: I2b2b0d865224413254051581b382ff206aae0180\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae79b3fc12e0f4f3559a293e4ccfa0f853902a1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae79b3fc12e0f4f3559a293e4ccfa0f853902a1f"}]},"branch":"refs/heads/master"},"a9b696abbd73657e9b77a97c29fc8e029db38485":{"kind":"REWORK","_number":10,"created":"2018-02-09 16:49:20.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/98/531498/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/10"}}},"commit":{"parents":[{"commit":"8fdb9537a706da276749d8744b2b08c7360ad715","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fdb9537a706da276749d8744b2b08c7360ad715"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-02-09 16:46:28.000000000","tz":-300},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nCloses-Bug: #1712463\n\nDepends-On: I2b2b0d865224413254051581b382ff206aae0180\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a9b696abbd73657e9b77a97c29fc8e029db38485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a9b696abbd73657e9b77a97c29fc8e029db38485"}]},"branch":"refs/heads/master"},"6f778c44ca24ba4d733790264fd5dc4201af36c6":{"kind":"REWORK","_number":11,"created":"2018-02-13 17:34:45.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/98/531498/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/11"}}},"commit":{"parents":[{"commit":"8fdb9537a706da276749d8744b2b08c7360ad715","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fdb9537a706da276749d8744b2b08c7360ad715"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-02-13 17:30:22.000000000","tz":-300},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nCloses-Bug: #1712463\n\nDepends-On: I2b2b0d865224413254051581b382ff206aae0180\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6f778c44ca24ba4d733790264fd5dc4201af36c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6f778c44ca24ba4d733790264fd5dc4201af36c6"}]},"branch":"refs/heads/master"},"b3ba6b8ba2639b7148f881d9fd009e6be8ce998f":{"kind":"REWORK","_number":12,"created":"2018-03-07 15:40:01.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/98/531498/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/12"}}},"commit":{"parents":[{"commit":"c53d7f19437bc700f81dc37b8aae914b14896d02","subject":"api-ref: correct typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c53d7f19437bc700f81dc37b8aae914b14896d02"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-03-07 15:39:45.000000000","tz":-300},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nCloses-Bug: #1712463\n\nDepends-On: I2b2b0d865224413254051581b382ff206aae0180\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3ba6b8ba2639b7148f881d9fd009e6be8ce998f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3ba6b8ba2639b7148f881d9fd009e6be8ce998f"}]},"branch":"refs/heads/master"},"aea6a2f1c166a0fc8e655ea9948d97fe2a4abeeb":{"kind":"REWORK","_number":13,"created":"2022-07-04 21:06:46.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/98/531498/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/13"}}},"commit":{"parents":[{"commit":"a71f7ba44e57a7945addb82afd2a19962d5c04cc","subject":"[APIImpact] Correct API response code for DELETE cache APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a71f7ba44e57a7945addb82afd2a19962d5c04cc"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-04 21:02:49.000000000","tz":120},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nCloses-Bug: #1712463\n\nDepends-On: I2b2b0d865224413254051581b382ff206aae0180\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aea6a2f1c166a0fc8e655ea9948d97fe2a4abeeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aea6a2f1c166a0fc8e655ea9948d97fe2a4abeeb"}]},"branch":"refs/heads/master"},"30c615892bb31ae2b93b117c39944bbdc1f42875":{"kind":"REWORK","_number":14,"created":"2022-07-04 23:25:15.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/98/531498/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/531498/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/531498/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/531498/14"}}},"commit":{"parents":[{"commit":"a71f7ba44e57a7945addb82afd2a19962d5c04cc","subject":"[APIImpact] Correct API response code for DELETE cache APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a71f7ba44e57a7945addb82afd2a19962d5c04cc"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2018-01-05 20:42:03.000000000","tz":-300},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-04 23:20:52.000000000","tz":120},"subject":"Specify a python executable for prlimits when using uwsgi","message":"Specify a python executable for prlimits when using uwsgi\n\nThis commit address the prlimit issue when running taskflow executors\nunder uwsgi. The uwsgi server links libpython.so and the path to the\nuwsgi binary becomes the value for sys.executable so processutils can\u0027t\nfigure out whch python executable is actually being used when it needs\nto launch python for prlimits enforcement. An option to workaround this\nby specifing an executable was added to oslo.concurrency and this commit\nleverages it. By default glance will use /usr/bin/python{2,3} (based on\npython version) but a user can optionally specify a specific path to an\nexecutable to use for this.\n\nCloses-Bug: #1712463\n\nDepends-On: I2b2b0d865224413254051581b382ff206aae0180\nChange-Id: I0758590e2286205c4f6dc441241dc32c0d840dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30c615892bb31ae2b93b117c39944bbdc1f42875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30c615892bb31ae2b93b117c39944bbdc1f42875"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
