)]}'
{"id":"openstack%2Fglance~547948","triplet_id":"openstack%2Fglance~master~I8115aee691c2913915a481c6e0fb010016fd07ec","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-05-06 16:56:19.000000000","reason":"Change was submitted"},"27577":{"account":{"_account_id":27577,"name":"caishan","email":"caishan1993@foxmail.com","username":"caishan"},"last_update":"2026-05-06 16:56:19.000000000","reason":"Change was submitted"},"21690":{"account":{"_account_id":21690,"name":"chenpengzi","email":"1523688226@qq.com","username":"chenpengzi"},"last_update":"2026-05-06 16:56:19.000000000","reason":"Change was submitted"},"27882":{"account":{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},"last_update":"2026-05-06 16:56:19.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-05-04 15:03:15.000000000","reason":"removed on reply"},"23517":{"account":{"_account_id":23517,"name":"chenyb4","email":"cybing4@gmail.com","username":"Yuanbin.Chen"},"last_update":"2026-05-06 16:56:19.000000000","reason":"Change was submitted"},"25167":{"account":{"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"},"last_update":"2026-05-06 16:56:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8115aee691c2913915a481c6e0fb010016fd07ec","subject":"Remove not used variable value","status":"MERGED","created":"2018-02-26 09:11:41.000000000","updated":"2026-05-06 16:57:26.000000000","submitted":"2026-05-06 16:56:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"547948","meta_rev_id":"38e2a0f83fccb072f72effccb201e7a8fc9aece2","_number":547948,"virtual_id_number":547948,"owner":{"_account_id":21690,"name":"chenpengzi","email":"1523688226@qq.com","username":"chenpengzi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27577,"name":"caishan","email":"caishan1993@foxmail.com","username":"caishan"},{"value":0,"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-06 16:56:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23517,"name":"chenyb4","email":"cybing4@gmail.com","username":"Yuanbin.Chen"},{"value":0,"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27577,"name":"caishan","email":"caishan1993@foxmail.com","username":"caishan"},{"value":0,"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},{"value":2,"date":"2026-05-06 14:29:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23517,"name":"chenyb4","email":"cybing4@gmail.com","username":"Yuanbin.Chen"},{"value":0,"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27577,"name":"caishan","email":"caishan1993@foxmail.com","username":"caishan"},{"value":0,"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},{"value":1,"date":"2026-05-06 14:29:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23517,"name":"chenyb4","email":"cybing4@gmail.com","username":"Yuanbin.Chen"},{"value":0,"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27577,"name":"caishan","email":"caishan1993@foxmail.com","username":"caishan"},{"value":0,"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23517,"name":"chenyb4","email":"cybing4@gmail.com","username":"Yuanbin.Chen"},{"value":0,"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23517,"name":"chenyb4","email":"cybing4@gmail.com","username":"Yuanbin.Chen"},{"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"},{"_account_id":27577,"name":"caishan","email":"caishan1993@foxmail.com","username":"caishan"},{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-02-26 10:58:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-26 11:24:48.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-03-06 03:16:39.000000000","updated_by":{"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"},"reviewer":{"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"},"state":"REVIEWER"},{"updated":"2018-03-06 04:17:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2018-03-19 08:59:26.000000000","updated_by":{"_account_id":27577,"name":"caishan","email":"caishan1993@foxmail.com","username":"caishan"},"reviewer":{"_account_id":27577,"name":"caishan","email":"caishan1993@foxmail.com","username":"caishan"},"state":"REVIEWER"},{"updated":"2018-03-30 02:03:39.000000000","updated_by":{"_account_id":23517,"name":"chenyb4","email":"cybing4@gmail.com","username":"Yuanbin.Chen"},"reviewer":{"_account_id":23517,"name":"chenyb4","email":"cybing4@gmail.com","username":"Yuanbin.Chen"},"state":"REVIEWER"},{"updated":"2018-08-28 12:07:59.000000000","updated_by":{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},"reviewer":{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},"state":"REVIEWER"},{"updated":"2026-04-28 15:04:09.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"d080b3bd0ad665eb3312b3c979318939ea65db70","author":{"_account_id":21690,"name":"chenpengzi","email":"1523688226@qq.com","username":"chenpengzi"},"date":"2018-02-26 09:11:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"058f633316d373f633e1926f0afb550f81bd4576","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-02-26 09:18:05.000000000","message":"Patch Set 1: Code-Review-1\n\nThis is used and needed. If the if block on the following line does not evaluate to true, then when this variable is used later on it will fail because it is not defined.","accounts_in_message":[],"_revision_number":1},{"id":"db4a4e222f6492703df9a8a7a3d783e275d9899d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-26 10:58:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/48/547948/1/check/openstack-tox-py35/c893743/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/48/547948/1/check/openstack-tox-pep8/5560dbb/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/48/547948/1/check/openstack-tox-py27/1bc1cd1/ : SUCCESS in 4m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/48/547948/1/check/build-openstack-sphinx-docs/7b98221/html/ : SUCCESS in 4m 34s\n- tempest-full http://logs.openstack.org/48/547948/1/check/tempest-full/e7a1c0a/ : SUCCESS in 1h 41m 15s\n- neutron-grenade http://logs.openstack.org/48/547948/1/check/neutron-grenade/a4e3689/ : SUCCESS in 55m 36s\n- tempest-full-py3 http://logs.openstack.org/48/547948/1/check/tempest-full-py3/30e4b65/ : SUCCESS in 1h 25m 33s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/48/547948/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/1fb1c0f/ : SUCCESS in 1h 00m 58s (non-voting)\n- legacy-rally-dsvm-glance http://logs.openstack.org/48/547948/1/check/legacy-rally-dsvm-glance/c8101fa/ : SUCCESS in 44m 26s (non-voting)\n- legacy-rally-dsvm-py35-glance http://logs.openstack.org/48/547948/1/check/legacy-rally-dsvm-py35-glance/28ee2be/ : FAILURE in 8m 40s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/48/547948/1/check/openstack-tox-functional/ab5529c/ : SUCCESS in 7m 07s\n- openstack-tox-functional-py35 http://logs.openstack.org/48/547948/1/check/openstack-tox-functional-py35/eaf4efe/ : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ecb6523d4e040efd5661d64cac472f283535d93a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-02-26 11:24:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-glance http://207.189.188.190/logs/48/547948/1/check-not-vote/ext-glance-zuul/2bf25d8 : FAILURE in 26m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f60e8ed19d074bc65ccf6ae2dc999287e148f168","author":{"_account_id":27577,"name":"caishan","email":"caishan1993@foxmail.com","username":"caishan"},"date":"2018-02-27 02:05:01.000000000","message":"Patch Set 1:\n\n\u003e This is used and needed. If the if block on the following line does\n \u003e not evaluate to true, then when this variable is used later on it\n \u003e will fail because it is not defined.\n\nhi, if the IF block does not evaluate to true, then the ELSE block will works, and sys.exit(). \nThis is fine.","accounts_in_message":[],"_revision_number":1},{"id":"f91ee6cdd03d4187873e2d3154ee92f43f76c66c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-02-27 07:52:32.000000000","message":"Patch Set 1: -Code-Review\n\nOh, true. We don\u0027t have a pylint job running, but I think that might still complain about this - even though you are right, code logic wise it should always end up defined if it reaches where it is used.","accounts_in_message":[],"_revision_number":1},{"id":"4306acd1cca9a840a28e729c57b494347a3460ea","author":{"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"},"date":"2018-03-06 03:16:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2359d73053b2b13558118374383e750b049ac4fc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-03-06 04:17:01.000000000","message":"Patch Set 1: Code-Review+2\n\nOK, I manually ran pylint and it did not complain about these. Thanks for bearing with me.","accounts_in_message":[],"_revision_number":1},{"id":"a8c0e8565e156a40221f148e2f8493502a113252","author":{"_account_id":27577,"name":"caishan","email":"caishan1993@foxmail.com","username":"caishan"},"date":"2018-03-19 08:59:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1c8cf024ff0fd9861d52636965f16f5e39bb2819","author":{"_account_id":23517,"name":"chenyb4","email":"cybing4@gmail.com","username":"Yuanbin.Chen"},"date":"2018-03-30 02:03:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"489c537960250f73d514c6a9e136d27900e25c3e","author":{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},"date":"2018-08-28 12:07:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a62a35d063190c6741b7ac98e6650a6404fd795b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-04-28 15:04:09.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b18ee0ba144fba06bacfc359f926c04e6a7f6a37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 17:53:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e87e08d457f9484c8ecde8ec4812824d\n\n- grenade https://zuul.opendev.org/t/openstack/build/88d5e29009d44c66bfeca9cab84e68a4 : SUCCESS in 1h 08m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dd1d2670f6a842faa106034672cee609 : SUCCESS in 1h 04m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a4d3164c9234481afed8e3ba07ec887 : SUCCESS in 1h 42m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ef609ec05ab4b40a429bae72d470b44 : SUCCESS in 1h 03m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4d0fff1a4584bd1a7b7e2d2ad5e23cc : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d94543e6294474ebed8f854a395571b : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/156dcb88f4244a7ca1fd5d1a01ab047c : SUCCESS in 5m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/52d26b845b6042898abc5b4233892d6e : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c140d80d44d04f8d82666cd6893aa98c : SUCCESS in 6m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9dd5b2a9352448b8a361c64b033d55c8 : SUCCESS in 6m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/68c676f628f245a0a5f1fdabc13d9142 : SUCCESS in 6m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a33193eac83c49e6a530bcc708feb506 : SUCCESS in 1h 53m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d011f3a3ea86461689cecfdad7ae93bf : SUCCESS in 2h 10m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f626f56468ca48d68b7862837c031adb : FAILURE in 2h 05m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/895a39ffd4e849da9a87df6852a10cee : SUCCESS in 2h 43m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d9ce4d2317c49a0ab8967c60216d77f : SUCCESS in 1h 04m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/34aeb3492bbf467e8cbbf4cc91c30e49 : SUCCESS in 1h 24m 18s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ff4dd8cad1da429491b8cdeea3fa8bdd : FAILURE in 7m 33s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/19a634d86f6244c59caf82dc1af8219e : SUCCESS in 1h 55m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b6c760a581404ca19353989f707752fb : SUCCESS in 40m 26s","accounts_in_message":[],"_revision_number":2},{"id":"4daa3e719499dc1ce595be6b3f546974b996f991","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 18:48:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b707a69c9bf4c17bbaa2cdbbbc55583\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e9e82b4688d4d8fb0f3f5da26174aeb : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6803b6eaea1f4a4091849de84ba9df98 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8996e7586eac4be0a335f7f0e1920f5d : SUCCESS in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0d8123a4c16d31bbcc5f118172058ce5f116b334","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-04-29 16:04:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4f6336c49ef0daf36c84b40b85c08bbad7555447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 20:28:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9598478087e24263997be6a61ae2aa0b\n\n- grenade https://zuul.opendev.org/t/openstack/build/5d8d36c59a4f4c8c9278933e3d8c28a5 : SUCCESS in 1h 08m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ce6e8ab07d447359d51cf617eacf488 : SUCCESS in 1h 05m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce07846664c54b3e93aa8145ea7ef49f : SUCCESS in 1h 25m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c5f1b4833ff41d29a51d6831311a93b : SUCCESS in 1h 13m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3fb59170bda42d79a0cd2087f00768a : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/daf4e1388fb64800a3ae87e6ab4016dd : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31efa049358344e4a0fc8ce413bd6aab : SUCCESS in 4m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/48acc220f56a46daae03cacb17d96b95 : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53c9acd18d7e49e09dc9321953bffdff : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b389042edb3143b4aa6a0f55ef40f918 : SUCCESS in 7m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/76e52520bb4346009ea7628aa67d92e4 : SUCCESS in 7m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dfaa71416a874a2891d5d6cd8c9ff700 : SUCCESS in 1h 53m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/36303668a15845d4b252340920c35f0d : SUCCESS in 2h 09m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b94d3ef94f0e4e0eb04b931f6a8d6c19 : SUCCESS in 2h 10m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9a6e0beba26e4debb4c7c207dbc7b03b : SUCCESS in 2h 15m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/004dc750f6a14ea88be1f052be509bc5 : SUCCESS in 29m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/649edb8228384b24b09960811380937e : FAILURE in 22m 48s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7940135b519e46db996a2693604ded24 : FAILURE in 9m 02s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/6e2f7cd70ea946adb57bad7f742f1f13 : SUCCESS in 1h 51m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7f91a0be33f94cb1ad5bb65c48a44364 : SUCCESS in 35m 35s","accounts_in_message":[],"_revision_number":2},{"id":"63ce94345c4e62c3f2100fd75aed807815f5cde6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 21:17:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4913c03adf94407b9d0f485c5ac4aae1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2596cbad3bfb471f93cc59fea342cb5d : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c627bc585c57437b962633db1d1f90a1 : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ff07cff4a7494c65bebf35b5b5e47785 : SUCCESS in 15m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e37723868f1819f026aaa62e433a65b184677694","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-05-04 15:03:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"98e8ccfb297372dc372b020a80fd6719e3cf50d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 15:50:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/863a3979961a42bb9b2d27aa2d9d5521\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03ea1ddc18524f009261c3d3dc9cee4c : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e5244f2863c84f858db341cca4412da0 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/56a226186dbf4ef786e9e40c3d0250ca : SUCCESS in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b681ca55ae1022271644ffe23a7af28453ac16d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 17:38:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5556c772ef9a475a85867fc1e67c69d0\n\n- grenade https://zuul.opendev.org/t/openstack/build/3ae8129d28374c2ba94b5d61d1c0a588 : SUCCESS in 1h 08m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/285a1c91c8cb44f891af5675cfc1c6c1 : SUCCESS in 1h 03m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5298003c31c40a3809d9dac5dbef312 : SUCCESS in 1h 37m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47e7a4107f1d48bc9a1e58a36480617e : SUCCESS in 43m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f68130b58c48ba97a012febe7fffaf : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b48d0cda47b54d2d98a87fb3cc0cac03 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1f8f449ffdf4b3e82b09ec26a1899d8 : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0ab1c13df6f94930ac10395a9dfe8d95 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a344a8d441d54d93850c440d096920f3 : SUCCESS in 5m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d5773076098b4b3989eeb0e2ba09936a : SUCCESS in 6m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0c4d7033497f4b4a856dffb404996ad9 : SUCCESS in 5m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/daeaa454c42a445eb3a74f6e349a00c3 : SUCCESS in 1h 18m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2ebd1e8bfc0f43998509bc4c8542ab31 : SUCCESS in 1h 40m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6a3539fe66ae4268a854bcbfe7716a9e : SUCCESS in 2h 18m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/629221ad0d5b43b2a9c5829934f3486b : SUCCESS in 2h 27m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cddf3c9809ed41a3a301030cd3ef66ab : SUCCESS in 59m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b90b9e047ca8436a9deb72b1986205bd : SUCCESS in 2h 07m 30s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a138e24ede114cd098396c15970af54b : FAILURE in 11m 16s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/dab24e6214b3451a81a30a25740e2f3a : FAILURE in 28m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cfab4d178d5e46d69a0c4c00023d5dc4 : SUCCESS in 17m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e56aabb32d9b0b6636f537d44a3d4854a617f8c3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-05-06 14:29:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5b58b149000b3903f63029f7380c3319d1eff51c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 14:30:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b47c8e4083f4c6dca7c0f1d9382c1b88269fd2b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 16:56:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97f4c9c9db6949a98da658b915c83023\n\n- grenade https://zuul.opendev.org/t/openstack/build/12bc07c3ed7c4d419494f184324b53f4 : SUCCESS in 1h 04m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d02b78908f73403a8ab59f81c468e278 : SUCCESS in 55m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6fbd15baa555459a9301f4f0fae4e582 : SUCCESS in 1h 16m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1b573847a6e444183d255a8c1dce48e : SUCCESS in 1h 07m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/523bc1e33c0a444dad4a7201c5dc7b24 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46ff84dd416d45b88748e059b164f430 : SUCCESS in 5m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bca072ae41d1442f9c0c78d58580c91e : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/019826109f014c57bdb5cf313a156004 : SUCCESS in 6m 05s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6129a64a4fb145f2b7072f4b076a4c76 : SUCCESS in 6m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/89ba5691e0894b5692a4cf0f7f5de57a : SUCCESS in 4m 38s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0091c02e5b1549a19a7945ca0c7179f1 : SUCCESS in 2h 19m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/711e9f85ea804d01a9858553ac819cef : SUCCESS in 2h 21m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4a31e847a4647f0bb8268e36bbede2f : SUCCESS in 31m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4646d243474e4154abf53413ada13605 : SUCCESS in 1h 49m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3e9dd8ccc84945a79256bce6fee67a87 : SUCCESS in 38m 22s","accounts_in_message":[],"_revision_number":2},{"id":"2b7b0539ab75307a086b6aa558748cf973c1d1eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 16:56:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"38e2a0f83fccb072f72effccb201e7a8fc9aece2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 16:57:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3142e5dff67c4ca69253a3c77faafd19\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ee344d9cb3e4555a46b5e7b32a0e773 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ab085c6ec29a577fae4a365c7c440463216c2e7e","revisions":{"ba14efb81fb97c22cf9e8c517cc6b766c74bed59":{"kind":"REWORK","_number":1,"created":"2018-02-26 09:11:41.000000000","uploader":{"_account_id":21690,"name":"chenpengzi","email":"1523688226@qq.com","username":"chenpengzi"},"ref":"refs/changes/48/547948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/547948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/547948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/547948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/547948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/547948/1"}}},"commit":{"parents":[{"commit":"b00df9cdb6ceb420bbedd479ec75a2103c856017","subject":"Hide Pike note on Queens Release Notes page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b00df9cdb6ceb420bbedd479ec75a2103c856017"}]}],"author":{"name":"chenpengzi","email":"1523688226@qq.com","date":"2018-02-26 09:11:14.000000000","tz":480},"committer":{"name":"chenpengzi","email":"1523688226@qq.com","date":"2018-02-26 09:11:34.000000000","tz":480},"subject":"Remove not used variable value","message":"Remove not used variable value\n\nChange-Id: I8115aee691c2913915a481c6e0fb010016fd07ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba14efb81fb97c22cf9e8c517cc6b766c74bed59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba14efb81fb97c22cf9e8c517cc6b766c74bed59"}]},"branch":"refs/heads/master"},"ab085c6ec29a577fae4a365c7c440463216c2e7e":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2026-04-28 15:04:09.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/48/547948/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/547948/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/547948/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/547948/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/547948/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/547948/2"}}},"commit":{"parents":[{"commit":"728ab7743977d484facb8ed8151707e7b3b10c56","subject":"Merge \"Add tools/db-migration to update the DB migration constant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/728ab7743977d484facb8ed8151707e7b3b10c56"}]}],"author":{"name":"chenpengzi","email":"1523688226@qq.com","date":"2018-02-26 09:11:14.000000000","tz":480},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-04-28 15:03:44.000000000","tz":120},"subject":"Remove not used variable value","message":"Remove not used variable value\n\nChange-Id: I8115aee691c2913915a481c6e0fb010016fd07ec\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab085c6ec29a577fae4a365c7c440463216c2e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab085c6ec29a577fae4a365c7c440463216c2e7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
