)]}'
{"id":"openstack%2Fglance~578529","triplet_id":"openstack%2Fglance~master~I3c5d4e068ee5f2919a4061e8f7b78f724cd61724","project":"openstack/glance","branch":"master","topic":"578529","hashtags":[],"change_id":"I3c5d4e068ee5f2919a4061e8f7b78f724cd61724","subject":"Switch to using stestr","status":"MERGED","created":"2018-06-27 19:04:04.000000000","updated":"2019-04-23 16:01:41.000000000","submitted":"2019-04-23 16:01:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"578529-1556035301482-776347d6","meta_rev_id":"ea04151123a025acc7cf35c6265ad7d87fe4d18b","_number":578529,"virtual_id_number":578529,"owner":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-15 15:58:35.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-04-15 21:24:03.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-04-23 16:01:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-16 13:05:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-04-23 11:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-04-23 11:45:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-04-23 16:01:41.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-07-14 23:08:44.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-03-07 02:34:55.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-15 15:58:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-15 21:24:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-16 13:05:22.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-04-23 11:45:11.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-04-23 16:01:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c343c19707cdea33db639afe5fdc0edd3f2ae13e","author":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"date":"2018-06-27 19:04:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9e78102ebfec50f4b42f59e4d917863a40d73e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-27 21:12:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/29/578529/1/check/openstack-tox-py35/a9f53d0/ : SUCCESS in 5m 55s\n- openstack-tox-pep8 http://logs.openstack.org/29/578529/1/check/openstack-tox-pep8/67edbc5/ : SUCCESS in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/29/578529/1/check/openstack-tox-py27/6f9ee9a/ : SUCCESS in 6m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/29/578529/1/check/build-openstack-sphinx-docs/ce80434/html/ : SUCCESS in 6m 30s\n- requirements-check http://logs.openstack.org/29/578529/1/check/requirements-check/99033b6/ : FAILURE in 3m 47s\n- tempest-full-py3 http://logs.openstack.org/29/578529/1/check/tempest-full-py3/c3be20c/ : SUCCESS in 1h 15m 23s\n- build-openstack-releasenotes http://logs.openstack.org/29/578529/1/check/build-openstack-releasenotes/16589d9/html/ : SUCCESS in 7m 26s\n- build-openstack-api-ref http://logs.openstack.org/29/578529/1/check/build-openstack-api-ref/1527817/html/ : SUCCESS in 5m 43s\n- openstack-tox-functional http://logs.openstack.org/29/578529/1/check/openstack-tox-functional/587589b/ : SUCCESS in 9m 45s\n- openstack-tox-functional-py35 http://logs.openstack.org/29/578529/1/check/openstack-tox-functional-py35/8b17824/ : SUCCESS in 9m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/578529/1/check/openstack-tox-lower-constraints/be83ee2/ : SUCCESS in 6m 07s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/29/578529/1/check/glance-eventlet-ssl-handshake-broken-py35/41ca2de/ : FAILURE in 6m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87c7e0bff62eb1f7bd140288f6a6d0a99a3eabd7","author":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"date":"2018-06-28 13:03:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"88e3474f14b480d3045efcb608f752360c3d87b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-28 14:43:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/29/578529/2/check/openstack-tox-py35/24660a0/ : SUCCESS in 6m 31s\n- openstack-tox-pep8 http://logs.openstack.org/29/578529/2/check/openstack-tox-pep8/74c90ce/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/29/578529/2/check/openstack-tox-py27/cb6daec/ : SUCCESS in 5m 14s\n- build-openstack-sphinx-docs http://logs.openstack.org/29/578529/2/check/build-openstack-sphinx-docs/3ab8489/html/ : SUCCESS in 6m 54s\n- requirements-check http://logs.openstack.org/29/578529/2/check/requirements-check/fc3e283/ : SUCCESS in 3m 31s\n- tempest-full http://logs.openstack.org/29/578529/2/check/tempest-full/d404f87/ : SUCCESS in 1h 29m 06s\n- neutron-grenade http://logs.openstack.org/29/578529/2/check/neutron-grenade/4b6a2b2/ : SUCCESS in 1h 01m 27s\n- tempest-full-py3 http://logs.openstack.org/29/578529/2/check/tempest-full-py3/d0325b8/ : SUCCESS in 1h 18m 11s\n- build-openstack-releasenotes http://logs.openstack.org/29/578529/2/check/build-openstack-releasenotes/1bfda2d/html/ : SUCCESS in 7m 09s\n- build-openstack-api-ref http://logs.openstack.org/29/578529/2/check/build-openstack-api-ref/0297795/html/ : SUCCESS in 5m 18s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/29/578529/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/63f64c4/ : SUCCESS in 1h 21m 48s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/29/578529/2/check/openstack-tox-functional/5ae7c1b/ : SUCCESS in 9m 24s\n- openstack-tox-functional-py35 http://logs.openstack.org/29/578529/2/check/openstack-tox-functional-py35/df9910e/ : SUCCESS in 10m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/578529/2/check/openstack-tox-lower-constraints/97db18e/ : SUCCESS in 7m 00s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/29/578529/2/check/glance-eventlet-ssl-handshake-broken-py35/e38a6d9/ : FAILURE in 5m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a79e1ea5a10cab669aa6beb7c0b94bb8cea213fb","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-07-14 23:02:41.000000000","message":"Patch Set 2: Code-Review-1\n\nThis is wrong.  It misses important config changes and set the wrong version of stestr","accounts_in_message":[],"_revision_number":2},{"id":"9deff14c8842911b49948759070b86dd8973c393","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-07-14 23:06:42.000000000","message":"Patch Set 2:\n\n\u003e This is wrong.  It misses important config changes and set the\n \u003e wrong version of stestr\n\nActually it seems the config changes were already there sorry.","accounts_in_message":[],"_revision_number":2},{"id":"e87f424fa6144d4128dcd5667118121dcd9a5527","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-07-14 23:08:44.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"6bd433850edf81a0f9080138411678214387cec4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-03-06 21:05:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0b029e56452b2eac80aa385f4349ee4be526ca70","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-03-06 21:06:28.000000000","message":"Patch Set 3: Code-Review+2\n\nUpdated to pick up new py37 environment.\n\nThis is needed to address failures in https://review.openstack.org/#/c/641207/2","accounts_in_message":[],"_revision_number":3},{"id":"b3817316814d974d1ca636f49e3adbeabc9c01bc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-06 21:44:32.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/29/578529/3/check-not-vote/ext-glance-zuul/8a0956d : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":3},{"id":"2eb6ff2611d2cf6891ea61282e805f985aec0ec8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 00:55:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/29/578529/3/check/build-openstack-api-ref/2e8d9e5/html/ : SUCCESS in 4m 33s\n- requirements-check http://logs.openstack.org/29/578529/3/check/requirements-check/6192987/ : FAILURE in 5m 21s\n- tempest-full http://logs.openstack.org/29/578529/3/check/tempest-full/f22305e/ : SUCCESS in 1h 51m 49s\n- neutron-grenade http://logs.openstack.org/29/578529/3/check/neutron-grenade/127146c/ : SUCCESS in 1h 10m 10s\n- grenade-py3 http://logs.openstack.org/29/578529/3/check/grenade-py3/4180474/ : SUCCESS in 58m 47s\n- tempest-full-py3 http://logs.openstack.org/29/578529/3/check/tempest-full-py3/0e2808b/ : SUCCESS in 1h 31m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/578529/3/check/openstack-tox-lower-constraints/190386c/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/29/578529/3/check/openstack-tox-pep8/129d23d/ : SUCCESS in 6m 28s\n- openstack-tox-py27 http://logs.openstack.org/29/578529/3/check/openstack-tox-py27/2e9f308/ : SUCCESS in 5m 28s\n- openstack-tox-py35 http://logs.openstack.org/29/578529/3/check/openstack-tox-py35/2e334e2/ : SUCCESS in 7m 35s\n- openstack-tox-py36 http://logs.openstack.org/29/578529/3/check/openstack-tox-py36/6414c20/ : SUCCESS in 5m 14s\n- openstack-tox-py37 http://logs.openstack.org/29/578529/3/check/openstack-tox-py37/29e2ca9/ : SUCCESS in 6m 13s\n- openstack-tox-docs http://logs.openstack.org/29/578529/3/check/openstack-tox-docs/b33dff7/html/ : SUCCESS in 6m 12s\n- build-openstack-releasenotes http://logs.openstack.org/29/578529/3/check/build-openstack-releasenotes/98af9e4/html/ : SUCCESS in 8m 39s\n- openstack-tox-functional http://logs.openstack.org/29/578529/3/check/openstack-tox-functional/577099d/ : SUCCESS in 9m 25s\n- openstack-tox-functional-py35 http://logs.openstack.org/29/578529/3/check/openstack-tox-functional-py35/90a4872/ : SUCCESS in 9m 22s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/29/578529/3/check/glance-eventlet-ssl-handshake-broken-py35/b0c0afe/ : FAILURE in 4m 23s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/29/578529/3/check/devstack-plugin-ceph-tempest/a85fc9e/ : SUCCESS in 1h 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0bd39e2fe5cd58ba229ff14e2211e8dca4bb174d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-03-07 01:56:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2b0330c34ae4bd2ad36e2443cb144c76fcac14bc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-03-07 01:56:43.000000000","message":"Patch Set 4: Review-Priority+1\n\nRebase errors in the last revision.","accounts_in_message":[],"_revision_number":4},{"id":"34f9997f870822fb0e89de2eaa345887266aa9b9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-07 02:34:55.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/29/578529/4/check-not-vote/ext-glance-zuul/319bdf1 : SUCCESS in 38m 24s","accounts_in_message":[],"_revision_number":4},{"id":"7c4c424f35cd34eb284e77f80a5afadadce6bc67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 03:13:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/29/578529/4/check/build-openstack-api-ref/10dedce/html/ : SUCCESS in 4m 24s\n- requirements-check http://logs.openstack.org/29/578529/4/check/requirements-check/a21a9af/ : FAILURE in 4m 06s\n- grenade-py3 http://logs.openstack.org/29/578529/4/check/grenade-py3/f3b586c/ : SUCCESS in 58m 34s\n- tempest-full-py3 http://logs.openstack.org/29/578529/4/check/tempest-full-py3/c2e97e2/ : SUCCESS in 1h 15m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/578529/4/check/openstack-tox-lower-constraints/fafd589/ : SUCCESS in 5m 12s\n- openstack-tox-pep8 http://logs.openstack.org/29/578529/4/check/openstack-tox-pep8/ca3a116/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/29/578529/4/check/openstack-tox-py27/c6fd78f/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/29/578529/4/check/openstack-tox-py35/7ef36da/ : FAILURE in 6m 22s\n- openstack-tox-py36 http://logs.openstack.org/29/578529/4/check/openstack-tox-py36/b59ef44/ : SUCCESS in 3m 49s\n- openstack-tox-py37 http://logs.openstack.org/29/578529/4/check/openstack-tox-py37/9fb7e31/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/29/578529/4/check/openstack-tox-docs/af96104/html/ : SUCCESS in 5m 06s\n- build-openstack-releasenotes http://logs.openstack.org/29/578529/4/check/build-openstack-releasenotes/edeb90f/html/ : SUCCESS in 6m 51s\n- openstack-tox-functional http://logs.openstack.org/29/578529/4/check/openstack-tox-functional/42c79d7/ : SUCCESS in 9m 18s\n- openstack-tox-functional-py35 http://logs.openstack.org/29/578529/4/check/openstack-tox-functional-py35/b0c4968/ : SUCCESS in 8m 17s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/29/578529/4/check/glance-eventlet-ssl-handshake-broken-py35/1a0d44b/ : FAILURE in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8fd5d31f0ca3bb9ed45c15618e30eb426ed67eee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-15 15:30:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ae5945ba030fd7c1e1ef4271f9cb0fcd750d3660","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 17:01:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/29/578529/5/check/build-openstack-api-ref/7aca45a/html/ : SUCCESS in 5m 20s\n- requirements-check http://logs.openstack.org/29/578529/5/check/requirements-check/2d5e9ad/ : FAILURE in 4m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/578529/5/check/openstack-tox-lower-constraints/ac2e3d1/ : SUCCESS in 6m 12s\n- openstack-tox-pep8 http://logs.openstack.org/29/578529/5/check/openstack-tox-pep8/a761de6/ : SUCCESS in 6m 39s\n- openstack-tox-py27 http://logs.openstack.org/29/578529/5/check/openstack-tox-py27/00cd46e/ : SUCCESS in 6m 07s\n- openstack-tox-py35 http://logs.openstack.org/29/578529/5/check/openstack-tox-py35/1878c42/ : SUCCESS in 5m 53s\n- openstack-tox-py36 http://logs.openstack.org/29/578529/5/check/openstack-tox-py36/a2002c3/ : SUCCESS in 5m 55s\n- openstack-tox-py37 http://logs.openstack.org/29/578529/5/check/openstack-tox-py37/4514129/ : SUCCESS in 6m 51s\n- openstack-tox-docs http://logs.openstack.org/29/578529/5/check/openstack-tox-docs/b4b183b/html/ : SUCCESS in 6m 46s\n- build-openstack-releasenotes http://logs.openstack.org/29/578529/5/check/build-openstack-releasenotes/a41099e/html/ : SUCCESS in 9m 10s\n- openstack-tox-functional http://logs.openstack.org/29/578529/5/check/openstack-tox-functional/94a01eb/ : SUCCESS in 10m 21s\n- openstack-tox-functional-py35 http://logs.openstack.org/29/578529/5/check/openstack-tox-functional-py35/747d1f2/ : TIMED_OUT in 31m 02s","accounts_in_message":[],"_revision_number":5},{"id":"10fb6ae0b7cb9995cb10b4c32de7e949d5e21d5f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-04-15 18:49:28.000000000","message":"Patch Set 5:\n\nNeed to wait for https://review.openstack.org/#/c/650550/ to merge before rechecking.","accounts_in_message":[],"_revision_number":5},{"id":"049ef4f106c0e2eba67e2a5480782a123c213bb7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-15 21:24:03.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"aa22eefd0b7acbc76d6055b69aa6da37cc1e1b83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 22:04:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/29/578529/5/check/build-openstack-api-ref/d605e6c/html/ : SUCCESS in 4m 43s\n- requirements-check http://logs.openstack.org/29/578529/5/check/requirements-check/5cd3cf1/ : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/578529/5/check/openstack-tox-lower-constraints/3f92068/ : SUCCESS in 6m 20s\n- openstack-tox-pep8 http://logs.openstack.org/29/578529/5/check/openstack-tox-pep8/472bbb6/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/29/578529/5/check/openstack-tox-py27/6ce5bf1/ : SUCCESS in 5m 33s\n- openstack-tox-py35 http://logs.openstack.org/29/578529/5/check/openstack-tox-py35/99ee83f/ : SUCCESS in 6m 26s\n- openstack-tox-py36 http://logs.openstack.org/29/578529/5/check/openstack-tox-py36/a076a88/ : SUCCESS in 5m 42s\n- openstack-tox-py37 http://logs.openstack.org/29/578529/5/check/openstack-tox-py37/c64e58c/ : SUCCESS in 5m 19s\n- openstack-tox-docs http://logs.openstack.org/29/578529/5/check/openstack-tox-docs/f822041/html/ : SUCCESS in 5m 43s\n- build-openstack-releasenotes http://logs.openstack.org/29/578529/5/check/build-openstack-releasenotes/db1c87b/html/ : SUCCESS in 9m 41s\n- openstack-tox-functional http://logs.openstack.org/29/578529/5/check/openstack-tox-functional/0344342/ : SUCCESS in 10m 15s\n- openstack-tox-functional-py35 http://logs.openstack.org/29/578529/5/check/openstack-tox-functional-py35/f4f9a29/ : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":5},{"id":"3a3f493bcfcc47d4f01f3bc866b45a073f6169b0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-04-16 13:05:22.000000000","message":"Patch Set 5: Code-Review+2\n\nThis is kind of overdue.","accounts_in_message":[],"_revision_number":5},{"id":"a5991d73355b84071c09ef3f43cdaff606300124","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-04-23 11:45:11.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nI did not realize this was still hanging for the glance service repo","accounts_in_message":[],"_revision_number":5},{"id":"85fdcc4a1cc82712fb18eea7ab6fc37c8aae800b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 11:45:29.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"42c60dd3557fdd3be0b138bc638fc77fd0a96c9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 16:01:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"0b633dda52c2b1c3e24ee4d5e92c24051ce72444","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 16:01:41.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/29/578529/5/gate/build-openstack-api-ref/f12b9d2/html/ : SUCCESS in 6m 11s\n- requirements-check http://logs.openstack.org/29/578529/5/gate/requirements-check/5f02266/ : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/578529/5/gate/openstack-tox-lower-constraints/13f26b4/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/29/578529/5/gate/openstack-tox-pep8/62803a5/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/29/578529/5/gate/openstack-tox-py27/94d182c/ : SUCCESS in 5m 40s\n- openstack-tox-py35 http://logs.openstack.org/29/578529/5/gate/openstack-tox-py35/4058228/ : SUCCESS in 6m 00s\n- openstack-tox-py36 http://logs.openstack.org/29/578529/5/gate/openstack-tox-py36/825499b/ : SUCCESS in 5m 49s\n- openstack-tox-py37 http://logs.openstack.org/29/578529/5/gate/openstack-tox-py37/d703b1e/ : SUCCESS in 6m 40s\n- openstack-tox-docs http://logs.openstack.org/29/578529/5/gate/openstack-tox-docs/0476660/html/ : SUCCESS in 5m 46s\n- build-openstack-releasenotes http://logs.openstack.org/29/578529/5/gate/build-openstack-releasenotes/cb9770c/html/ : SUCCESS in 8m 57s\n- openstack-tox-functional http://logs.openstack.org/29/578529/5/gate/openstack-tox-functional/8d588e6/ : SUCCESS in 8m 56s\n- openstack-tox-functional-py35 http://logs.openstack.org/29/578529/5/gate/openstack-tox-functional-py35/929175a/ : SUCCESS in 10m 07s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0c9a419d8f8d4123c339ff26aec448acd2de62b0","revisions":{"4d3aed944d853717e83adf043577f286a72ba1e7":{"kind":"REWORK","_number":1,"created":"2018-06-27 19:04:04.000000000","uploader":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"ref":"refs/changes/29/578529/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/578529/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/578529/1"}}},"commit":{"parents":[{"commit":"54329c6a21b0d3f845b09e79f710fc795976a175","subject":"Merge \"Replace Chinese punctuation with English punctuation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54329c6a21b0d3f845b09e79f710fc795976a175"}]}],"author":{"name":"Chuck Short","email":"chucks@redhat.com","date":"2018-06-27 19:02:58.000000000","tz":-240},"committer":{"name":"Chuck Short","email":"chucks@redhat.com","date":"2018-06-27 19:04:03.000000000","tz":-240},"subject":"Switch to using stestr","message":"Switch to using stestr\n\nAccording to Openstack summit session [1] stestr is maintained\nproject to which all Openstack projects should migrate.\nLet\u0027s switch it then.\n\n[1] https://etherpad.openstack.org/p/YVR-python-pti\n\nChange-Id: I3c5d4e068ee5f2919a4061e8f7b78f724cd61724\nSigned-off-by: Chuck Short \u003cchucks@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4d3aed944d853717e83adf043577f286a72ba1e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4d3aed944d853717e83adf043577f286a72ba1e7"}]},"branch":"refs/heads/master"},"811d3f0e19ef9d535240dafe3fa988c41d9bdafd":{"kind":"REWORK","_number":2,"created":"2018-06-28 13:03:26.000000000","uploader":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"ref":"refs/changes/29/578529/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/578529/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/578529/2"}}},"commit":{"parents":[{"commit":"54329c6a21b0d3f845b09e79f710fc795976a175","subject":"Merge \"Replace Chinese punctuation with English punctuation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54329c6a21b0d3f845b09e79f710fc795976a175"}]}],"author":{"name":"Chuck Short","email":"chucks@redhat.com","date":"2018-06-27 19:02:58.000000000","tz":-240},"committer":{"name":"Chuck Short","email":"chucks@redhat.com","date":"2018-06-28 13:03:21.000000000","tz":-240},"subject":"Switch to using stestr","message":"Switch to using stestr\n\nAccording to Openstack summit session [1] stestr is maintained\nproject to which all Openstack projects should migrate.\nLet\u0027s switch it then.\n\n[1] https://etherpad.openstack.org/p/YVR-python-pti\n\nChange-Id: I3c5d4e068ee5f2919a4061e8f7b78f724cd61724\nSigned-off-by: Chuck Short \u003cchucks@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/811d3f0e19ef9d535240dafe3fa988c41d9bdafd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/811d3f0e19ef9d535240dafe3fa988c41d9bdafd"}]},"branch":"refs/heads/master"},"bc6048be3f85d773d7b13ed835e5e79c6f81daba":{"kind":"REWORK","_number":3,"created":"2019-03-06 21:05:16.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/29/578529/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/578529/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/578529/3"}}},"commit":{"parents":[{"commit":"2cf78f339d6eb2de45fda0aeb3f67a083fca8b05","subject":"Merge \"Document fileystem drv chunk size option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2cf78f339d6eb2de45fda0aeb3f67a083fca8b05"}]}],"author":{"name":"Chuck Short","email":"chucks@redhat.com","date":"2018-06-27 19:02:58.000000000","tz":-240},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-03-06 21:05:09.000000000","tz":-360},"subject":"Switch to using stestr","message":"Switch to using stestr\n\nAccording to Openstack summit session [1] stestr is maintained\nproject to which all Openstack projects should migrate.\nLet\u0027s switch it then.\n\n[1] https://etherpad.openstack.org/p/YVR-python-pti\n\nChange-Id: I3c5d4e068ee5f2919a4061e8f7b78f724cd61724\nSigned-off-by: Chuck Short \u003cchucks@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc6048be3f85d773d7b13ed835e5e79c6f81daba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc6048be3f85d773d7b13ed835e5e79c6f81daba"}]},"branch":"refs/heads/master"},"152fbee7d527dedad1e2386edbb9b1dd906d4365":{"kind":"REWORK","_number":4,"created":"2019-03-07 01:56:12.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/29/578529/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/578529/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/578529/4"}}},"commit":{"parents":[{"commit":"2cf78f339d6eb2de45fda0aeb3f67a083fca8b05","subject":"Merge \"Document fileystem drv chunk size option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2cf78f339d6eb2de45fda0aeb3f67a083fca8b05"}]}],"author":{"name":"Chuck Short","email":"chucks@redhat.com","date":"2018-06-27 19:02:58.000000000","tz":-240},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-03-07 01:56:03.000000000","tz":-360},"subject":"Switch to using stestr","message":"Switch to using stestr\n\nAccording to Openstack summit session [1] stestr is maintained\nproject to which all Openstack projects should migrate.\nLet\u0027s switch it then.\n\n[1] https://etherpad.openstack.org/p/YVR-python-pti\n\nChange-Id: I3c5d4e068ee5f2919a4061e8f7b78f724cd61724\nSigned-off-by: Chuck Short \u003cchucks@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/152fbee7d527dedad1e2386edbb9b1dd906d4365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/152fbee7d527dedad1e2386edbb9b1dd906d4365"}]},"branch":"refs/heads/master"},"0c9a419d8f8d4123c339ff26aec448acd2de62b0":{"kind":"REWORK","_number":5,"created":"2019-04-15 15:30:38.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/29/578529/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/578529/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/578529/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/578529/5"}}},"commit":{"parents":[{"commit":"6cd76cb5917019109555ca3feb4be22894212c1b","subject":"Merge \"Separate out configuration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6cd76cb5917019109555ca3feb4be22894212c1b"}]}],"author":{"name":"Chuck Short","email":"chucks@redhat.com","date":"2018-06-27 19:02:58.000000000","tz":-240},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-15 15:30:30.000000000","tz":-300},"subject":"Switch to using stestr","message":"Switch to using stestr\n\nAccording to Openstack summit session [1] stestr is maintained\nproject to which all Openstack projects should migrate.\nLet\u0027s switch it then.\n\n[1] https://etherpad.openstack.org/p/YVR-python-pti\n\nChange-Id: I3c5d4e068ee5f2919a4061e8f7b78f724cd61724\nSigned-off-by: Chuck Short \u003cchucks@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c9a419d8f8d4123c339ff26aec448acd2de62b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c9a419d8f8d4123c339ff26aec448acd2de62b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
