)]}'
{"id":"openstack%2Fglance~597368","triplet_id":"openstack%2Fglance~master~I1308fb1dec21002a9777bd0c77e9c02e59527551","project":"openstack/glance","branch":"master","topic":"locations-with-checksum","hashtags":[],"change_id":"I1308fb1dec21002a9777bd0c77e9c02e59527551","subject":"Embed validation data in locations","status":"MERGED","created":"2018-08-29 03:54:32.000000000","updated":"2018-11-12 06:50:23.000000000","submitted":"2018-11-12 06:50:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"597368-1542005423507-18b0b557","meta_rev_id":"4b79ced2dd63141709259b67a49af67c538d54c6","_number":597368,"virtual_id_number":597368,"owner":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-11-09 21:11:48.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},{"value":2,"date":"2018-11-12 06:50:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":2,"date":"2018-11-12 05:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":1,"date":"2018-11-12 05:15:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-29 05:53:26.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2018-08-29 05:53:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2018-08-29 05:54:53.000000000","updated_by":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"reviewer":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"state":"REVIEWER"},{"updated":"2018-09-10 05:53:55.000000000","updated_by":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"reviewer":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"state":"REVIEWER"},{"updated":"2018-09-17 07:14:13.000000000","updated_by":{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},"reviewer":{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},"state":"REVIEWER"},{"updated":"2018-11-09 19:42:19.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2018-11-09 21:11:48.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-11-12 05:15:38.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2018-11-12 06:50:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"909e659c4017bbb38957aebd0028bc23aad65385","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-08-29 03:54:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b7ed1279df3f1271567c2dc656385ede2976c6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-29 05:27:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/68/597368/1/check/openstack-tox-py35/621bf05/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/1/check/openstack-tox-pep8/e8ac788/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/1/check/openstack-tox-py27/61bb4c3/ : SUCCESS in 4m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/68/597368/1/check/build-openstack-sphinx-docs/dc7ec0b/html/ : SUCCESS in 4m 51s\n- tempest-full http://logs.openstack.org/68/597368/1/check/tempest-full/941e28d/ : SUCCESS in 1h 31m 14s\n- neutron-grenade http://logs.openstack.org/68/597368/1/check/neutron-grenade/228b445/ : SUCCESS in 47m 18s\n- tempest-full-py3 http://logs.openstack.org/68/597368/1/check/tempest-full-py3/ade10f4/ : SUCCESS in 1h 16m 02s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/68/597368/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/c9b2562/ : SUCCESS in 1h 09m 21s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/68/597368/1/check/openstack-tox-functional/e6f2f42/ : SUCCESS in 8m 37s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/1/check/openstack-tox-functional-py35/ced6ef2/ : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/1/check/openstack-tox-lower-constraints/463651f/ : SUCCESS in 3m 39s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/1/check/glance-eventlet-ssl-handshake-broken-py35/3a5317e/ : FAILURE in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"abe8967e3c492caf2b4a50ee112f7922f2927e66","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-29 06:01:06.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/68/597368/1/check-not-vote/ext-glance-zuul/a13b0e8 : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f703b1e8fd23edd6c03a18f3a847a43884a01a56","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-08-29 20:06:06.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"96cdea1b8a305f6c01fd73531e69d8ad41f10d85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-29 21:53:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/68/597368/2/check/openstack-tox-py35/d29f6ef/ : SUCCESS in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/2/check/openstack-tox-pep8/3506fd6/ : SUCCESS in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/2/check/openstack-tox-py27/9ec2dfb/ : SUCCESS in 4m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/68/597368/2/check/build-openstack-sphinx-docs/5ad8514/html/ : SUCCESS in 5m 32s\n- tempest-full http://logs.openstack.org/68/597368/2/check/tempest-full/4382c30/ : SUCCESS in 1h 46m 35s\n- neutron-grenade http://logs.openstack.org/68/597368/2/check/neutron-grenade/8eba230/ : SUCCESS in 54m 12s\n- tempest-full-py3 http://logs.openstack.org/68/597368/2/check/tempest-full-py3/b5c4321/ : SUCCESS in 1h 43m 36s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/68/597368/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/a400dbe/ : SUCCESS in 1h 22m 34s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/68/597368/2/check/openstack-tox-functional/fec29ad/ : SUCCESS in 8m 19s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/2/check/openstack-tox-functional-py35/65f3b6c/ : SUCCESS in 8m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/2/check/openstack-tox-lower-constraints/7e5163a/ : SUCCESS in 3m 38s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/2/check/glance-eventlet-ssl-handshake-broken-py35/6ba92d2/ : FAILURE in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"756c91445af647845b7fafc097d0bc4049704882","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-09-01 00:11:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bcb81fa00a72df7c93573bef33eef3b09788b10e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-01 02:03:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/68/597368/3/check/openstack-tox-py35/fd0b143/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/3/check/openstack-tox-pep8/42c8bf8/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/3/check/openstack-tox-py27/d7de62a/ : SUCCESS in 3m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/68/597368/3/check/build-openstack-sphinx-docs/97150e5/html/ : SUCCESS in 4m 40s\n- tempest-full http://logs.openstack.org/68/597368/3/check/tempest-full/7b5cb2b/ : SUCCESS in 1h 50m 38s\n- neutron-grenade http://logs.openstack.org/68/597368/3/check/neutron-grenade/7f32328/ : SUCCESS in 50m 15s\n- tempest-full-py3 http://logs.openstack.org/68/597368/3/check/tempest-full-py3/8cc70e4/ : SUCCESS in 1h 44m 25s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/68/597368/3/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/b9b3211/ : SUCCESS in 1h 09m 56s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/68/597368/3/check/openstack-tox-functional/125f63e/ : SUCCESS in 7m 21s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/3/check/openstack-tox-functional-py35/3dd6d9a/ : SUCCESS in 7m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/3/check/openstack-tox-lower-constraints/bb0fd61/ : SUCCESS in 3m 35s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/3/check/glance-eventlet-ssl-handshake-broken-py35/b37de71/ : FAILURE in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"db35d3722009378ece4f3483e1d5eb5d67d69104","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-09-02 22:46:00.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/68/597368/3/check-not-vote/ext-glance-zuul/a7f6f01 : FAILURE in 35m 43s","accounts_in_message":[],"_revision_number":3},{"id":"4ab7952c95c85f3ae3959e6aad03fbc5ada770a4","author":{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},"date":"2018-09-03 07:35:23.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nLooks good, but I left 2 comments for you ;)","accounts_in_message":[],"_revision_number":3},{"id":"d7d71fe3800d52f27ff2b7dedcb621d877f00c15","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-09-04 21:30:53.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6176dc00cd9d6003aff76ac75bb9e2efa297702a","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-09-04 22:58:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1e877d5cd13089ebceb7e2fc6cff13caad65d735","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-09-04 23:41:54.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/68/597368/4/check-not-vote/ext-glance-zuul/046f5b3 : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":4},{"id":"b9cd003a7e5e5ec9ffb8c0d9de777b7a781804e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-05 01:14:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/68/597368/4/check/openstack-tox-py35/e7a3e21/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/4/check/openstack-tox-pep8/2a0f8e9/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/4/check/openstack-tox-py27/644583b/ : SUCCESS in 4m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/68/597368/4/check/build-openstack-sphinx-docs/735e3a7/html/ : SUCCESS in 4m 47s\n- tempest-full http://logs.openstack.org/68/597368/4/check/tempest-full/f94495b/ : SUCCESS in 1h 47m 02s\n- neutron-grenade http://logs.openstack.org/68/597368/4/check/neutron-grenade/d69334e/ : SUCCESS in 47m 05s\n- tempest-full-py3 http://logs.openstack.org/68/597368/4/check/tempest-full-py3/264534c/ : SUCCESS in 1h 43m 47s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/68/597368/4/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/787e1fa/ : SUCCESS in 1h 22m 08s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/68/597368/4/check/openstack-tox-functional/9152ade/ : SUCCESS in 8m 04s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/4/check/openstack-tox-functional-py35/372ee89/ : SUCCESS in 9m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/4/check/openstack-tox-lower-constraints/35c4c2e/ : SUCCESS in 4m 12s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/4/check/glance-eventlet-ssl-handshake-broken-py35/4695557/ : FAILURE in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0f784a99617aac15d00e971bacfc7ed2ea71d029","author":{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},"date":"2018-09-05 06:20:53.000000000","message":"Patch Set 4: Code-Review+1\n\nNice, thanks for your replies.","accounts_in_message":[],"_revision_number":4},{"id":"8225bb6f64d50c7838eddf72dd3aac6d6ad0af4f","author":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"date":"2018-09-10 05:53:55.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5b786a545a25af79240e0a6948c8c121de8ce605","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-09-14 19:07:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d020c766cb5f9ebad7b3736f033a4261b0d64007","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-09-14 19:45:03.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/68/597368/5/check-not-vote/ext-glance-zuul/17f9c0f : SUCCESS in 36m 24s","accounts_in_message":[],"_revision_number":5},{"id":"63e94114b0c060862f6613148e6221508674675a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-14 21:10:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/68/597368/5/check/openstack-tox-py35/9055e3a/ : SUCCESS in 8m 22s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/5/check/openstack-tox-pep8/be95e65/ : SUCCESS in 5m 40s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/5/check/openstack-tox-py27/71eb0e8/ : SUCCESS in 5m 47s\n- build-openstack-sphinx-docs http://logs.openstack.org/68/597368/5/check/build-openstack-sphinx-docs/76ddef7/html/ : SUCCESS in 5m 54s\n- tempest-full http://logs.openstack.org/68/597368/5/check/tempest-full/41a07ba/ : SUCCESS in 1h 46m 59s\n- neutron-grenade http://logs.openstack.org/68/597368/5/check/neutron-grenade/5528d44/ : SUCCESS in 50m 58s\n- tempest-full-py3 http://logs.openstack.org/68/597368/5/check/tempest-full-py3/76c0d8e/ : SUCCESS in 1h 35m 25s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/68/597368/5/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/79b7301/ : SUCCESS in 1h 28m 32s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/68/597368/5/check/openstack-tox-functional/3bc5b46/ : FAILURE in 10m 46s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/5/check/openstack-tox-functional-py35/7433cb9/ : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/5/check/openstack-tox-lower-constraints/f5cf590/ : SUCCESS in 3m 45s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/5/check/glance-eventlet-ssl-handshake-broken-py35/0e05fe8/ : FAILURE in 8m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7baadb8c9580a29ead53d63702d99a48036e8bda","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-09-15 00:52:46.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"79678129bacbf958cc1e3867add03bb8612c4340","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-09-15 01:33:18.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/68/597368/5/check-not-vote/ext-glance-zuul/ba6bd1a : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":5},{"id":"1de518291c3f8c753e2a939ea9f97dd677dfc474","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-15 02:14:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/68/597368/5/check/openstack-tox-py35/abedde3/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/5/check/openstack-tox-pep8/14edab2/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/5/check/openstack-tox-py27/275f075/ : SUCCESS in 3m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/68/597368/5/check/build-openstack-sphinx-docs/ac02138/html/ : SUCCESS in 5m 06s\n- tempest-full http://logs.openstack.org/68/597368/5/check/tempest-full/a7b2b4f/ : SUCCESS in 1h 21m 12s\n- neutron-grenade http://logs.openstack.org/68/597368/5/check/neutron-grenade/6dbf466/ : SUCCESS in 51m 55s\n- tempest-full-py3 http://logs.openstack.org/68/597368/5/check/tempest-full-py3/705284b/ : SUCCESS in 1h 19m 16s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/68/597368/5/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/fd2997b/ : SUCCESS in 1h 11m 33s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/68/597368/5/check/openstack-tox-functional/f18fea2/ : SUCCESS in 7m 53s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/5/check/openstack-tox-functional-py35/3150225/ : SUCCESS in 7m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/5/check/openstack-tox-lower-constraints/0bf3a65/ : SUCCESS in 3m 52s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/5/check/glance-eventlet-ssl-handshake-broken-py35/62fcd5e/ : FAILURE in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"494c32f85717d2b391edb892a23d64b6f10634d0","author":{"_account_id":27882,"name":"Piotr Bielak","email":"piotrbielak2@gmail.com","username":"pbielak"},"date":"2018-09-17 07:14:13.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"7ba801fb662a12b58d5ee4fc448cf63de8b8fc41","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-10-05 22:36:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d9836556343e793babc5514aa94e0be7c4d96d5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-06 00:15:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/68/597368/6/check/tempest-full/3772815/ : SUCCESS in 1h 28m 47s\n- neutron-grenade http://logs.openstack.org/68/597368/6/check/neutron-grenade/5877162/ : SUCCESS in 1h 00m 25s\n- tempest-full-py3 http://logs.openstack.org/68/597368/6/check/tempest-full-py3/5004257/ : SUCCESS in 1h 09m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/6/check/openstack-tox-lower-constraints/d0ec928/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/6/check/openstack-tox-pep8/7da3543/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/6/check/openstack-tox-py27/c5f1db5/ : SUCCESS in 5m 26s\n- openstack-tox-py35 http://logs.openstack.org/68/597368/6/check/openstack-tox-py35/797c1c5/ : FAILURE in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/68/597368/6/check/openstack-tox-py36/e30e66c/ : SUCCESS in 4m 04s\n- openstack-tox-docs http://logs.openstack.org/68/597368/6/check/openstack-tox-docs/4123f2e/html/ : SUCCESS in 5m 30s\n- openstack-tox-functional http://logs.openstack.org/68/597368/6/check/openstack-tox-functional/3f5df99/ : SUCCESS in 8m 01s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/6/check/openstack-tox-functional-py35/586bb21/ : SUCCESS in 8m 23s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/6/check/glance-eventlet-ssl-handshake-broken-py35/306457b/ : FAILURE in 4m 28s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/68/597368/6/check/devstack-plugin-ceph-tempest/4ab4857/ : SUCCESS in 1h 35m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"98c53bb8bc010589bdcf29b984b14d6c5d27cf12","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-10-06 00:32:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ba0c4cbcbbd699301f3fe421bc991ee6baa9d42e","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-10-06 01:10:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e050f09e5cc54461d72f4f737bcfb2fa7bdd15a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-06 03:01:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/68/597368/7/check/tempest-full/200ab3b/ : SUCCESS in 1h 35m 05s\n- neutron-grenade http://logs.openstack.org/68/597368/7/check/neutron-grenade/dc1e313/ : SUCCESS in 45m 42s\n- tempest-full-py3 http://logs.openstack.org/68/597368/7/check/tempest-full-py3/046d5fa/ : SUCCESS in 1h 30m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/7/check/openstack-tox-lower-constraints/3e31db6/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/7/check/openstack-tox-pep8/f2aa0e6/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/7/check/openstack-tox-py27/30c2975/ : SUCCESS in 4m 08s\n- openstack-tox-py35 http://logs.openstack.org/68/597368/7/check/openstack-tox-py35/22c9ade/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/68/597368/7/check/openstack-tox-py36/28384e9/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/68/597368/7/check/openstack-tox-docs/54ba03a/html/ : SUCCESS in 4m 00s\n- openstack-tox-functional http://logs.openstack.org/68/597368/7/check/openstack-tox-functional/c0f2ded/ : SUCCESS in 8m 03s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/7/check/openstack-tox-functional-py35/cd3979f/ : SUCCESS in 7m 19s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/7/check/glance-eventlet-ssl-handshake-broken-py35/32bed5c/ : FAILURE in 3m 59s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/68/597368/7/check/devstack-plugin-ceph-tempest/df0fb6e/ : SUCCESS in 1h 49m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0297cd556ccd0b291f4e745e4be237c8654cdf42","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-10-11 18:02:35.000000000","message":"Patch Set 7: Workflow-1\n\nCosmetic fix required... pending spec-lite approval.","accounts_in_message":[],"_revision_number":7},{"id":"706d7fad1d467ee581669c4280e0212695d8cdf3","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-10-19 23:45:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f53fa0dc36a0cd939adac1925ce3433e6e11d12d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-10-20 00:32:29.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/68/597368/8/check-not-vote/ext-glance-zuul/05bb989 : SUCCESS in 46m 48s","accounts_in_message":[],"_revision_number":8},{"id":"7fd5387612af39380b8a5647f1aacc4461c88b9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-20 02:01:05.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/68/597368/8/check/tempest-full/082b16e/ : SUCCESS in 1h 41m 40s\n- neutron-grenade http://logs.openstack.org/68/597368/8/check/neutron-grenade/11ddd33/ : SUCCESS in 45m 14s\n- tempest-full-py3 http://logs.openstack.org/68/597368/8/check/tempest-full-py3/d663c6c/ : TIMED_OUT in 2h 08m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/8/check/openstack-tox-lower-constraints/8bb1ef6/ : SUCCESS in 4m 05s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/8/check/openstack-tox-pep8/e37d91f/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/8/check/openstack-tox-py27/de5d20f/ : SUCCESS in 4m 22s\n- openstack-tox-py35 http://logs.openstack.org/68/597368/8/check/openstack-tox-py35/5427a15/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/68/597368/8/check/openstack-tox-py36/ccc9669/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/68/597368/8/check/openstack-tox-docs/531d73d/html/ : SUCCESS in 5m 33s\n- openstack-tox-functional http://logs.openstack.org/68/597368/8/check/openstack-tox-functional/61bf324/ : SUCCESS in 7m 37s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/8/check/openstack-tox-functional-py35/aea3251/ : SUCCESS in 11m 34s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/8/check/glance-eventlet-ssl-handshake-broken-py35/f145e26/ : FAILURE in 4m 08s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/68/597368/8/check/devstack-plugin-ceph-tempest/3351e31/ : SUCCESS in 1h 35m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"679d0e91189326846c6d8c1064e822dfb0992473","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-10-20 04:33:53.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"147d02582f9301994bdf18ff94586986135f7f19","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-10-20 05:10:38.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/68/597368/8/check-not-vote/ext-glance-zuul/c68ed96 : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":8},{"id":"ebe80fd83df90b37b4dafb6db4f88325d4ffd63e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-20 06:26:22.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/68/597368/8/check/tempest-full/29f71fc/ : SUCCESS in 1h 26m 37s\n- neutron-grenade http://logs.openstack.org/68/597368/8/check/neutron-grenade/6dee518/ : SUCCESS in 48m 04s\n- tempest-full-py3 http://logs.openstack.org/68/597368/8/check/tempest-full-py3/80cd0b6/ : SUCCESS in 1h 28m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/8/check/openstack-tox-lower-constraints/8b08e53/ : SUCCESS in 3m 53s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/8/check/openstack-tox-pep8/fc57b7a/ : SUCCESS in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/8/check/openstack-tox-py27/a5dd9a5/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/68/597368/8/check/openstack-tox-py35/d014408/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/68/597368/8/check/openstack-tox-py36/daeabb0/ : SUCCESS in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/68/597368/8/check/openstack-tox-docs/e14d395/html/ : SUCCESS in 5m 03s\n- openstack-tox-functional http://logs.openstack.org/68/597368/8/check/openstack-tox-functional/acbd2f5/ : SUCCESS in 7m 47s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/8/check/openstack-tox-functional-py35/f49cfc0/ : SUCCESS in 7m 14s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/8/check/glance-eventlet-ssl-handshake-broken-py35/a3c0df6/ : FAILURE in 4m 04s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/68/597368/8/check/devstack-plugin-ceph-tempest/0cc4529/ : SUCCESS in 1h 48m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"751aaedc0946393363064d941786a23fa4e922f4","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-10-24 22:21:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f83d08b883458c21d62ee9f8782723552664bcdd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-10-24 23:00:44.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/68/597368/9/check-not-vote/ext-glance-zuul/234348a : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":9},{"id":"228e00324e1ea6d889164d4f2f496e3a56cf059f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-25 02:29:45.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/68/597368/9/check/tempest-full/88949a6/ : SUCCESS in 1h 31m 38s\n- neutron-grenade http://logs.openstack.org/68/597368/9/check/neutron-grenade/23fa0c6/ : SUCCESS in 1h 11m 05s\n- tempest-full-py3 http://logs.openstack.org/68/597368/9/check/tempest-full-py3/b157df5/ : SUCCESS in 1h 24m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/9/check/openstack-tox-lower-constraints/4b1b93d/ : SUCCESS in 4m 24s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/9/check/openstack-tox-pep8/1548079/ : SUCCESS in 6m 44s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/9/check/openstack-tox-py27/27936be/ : SUCCESS in 5m 21s\n- openstack-tox-py35 http://logs.openstack.org/68/597368/9/check/openstack-tox-py35/bc3e744/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/68/597368/9/check/openstack-tox-py36/2b2c7e3/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/68/597368/9/check/openstack-tox-docs/1771473/html/ : SUCCESS in 7m 50s\n- openstack-tox-functional http://logs.openstack.org/68/597368/9/check/openstack-tox-functional/aa69046/ : SUCCESS in 8m 31s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/9/check/openstack-tox-functional-py35/a586347/ : TIMED_OUT in 31m 22s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/9/check/glance-eventlet-ssl-handshake-broken-py35/3bc276a/ : FAILURE in 4m 57s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/68/597368/9/check/devstack-plugin-ceph-tempest/475298c/ : SUCCESS in 1h 46m 46s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"88ae249e355f571ce48632b8a8821db188142e3a","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-10-25 04:04:02.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"1f4d4c2101f8850a8b5a4c04e70718ffe9b18d78","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-10-25 04:43:21.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/68/597368/9/check-not-vote/ext-glance-zuul/6f5a620 : SUCCESS in 39m 00s","accounts_in_message":[],"_revision_number":9},{"id":"d15d643af845e74be5cb7b83a8e636325e76b191","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-25 06:04:44.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/68/597368/9/check/tempest-full/7b67c80/ : SUCCESS in 1h 33m 30s\n- neutron-grenade http://logs.openstack.org/68/597368/9/check/neutron-grenade/4754073/ : SUCCESS in 56m 45s\n- tempest-full-py3 http://logs.openstack.org/68/597368/9/check/tempest-full-py3/58dfb06/ : SUCCESS in 1h 20m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/9/check/openstack-tox-lower-constraints/77bf7f6/ : SUCCESS in 5m 33s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/9/check/openstack-tox-pep8/beb20e4/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/9/check/openstack-tox-py27/76d3e90/ : SUCCESS in 3m 38s\n- openstack-tox-py35 http://logs.openstack.org/68/597368/9/check/openstack-tox-py35/4a3b684/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/68/597368/9/check/openstack-tox-py36/d7cb6d6/ : SUCCESS in 5m 20s\n- openstack-tox-docs http://logs.openstack.org/68/597368/9/check/openstack-tox-docs/2367943/html/ : SUCCESS in 5m 31s\n- openstack-tox-functional http://logs.openstack.org/68/597368/9/check/openstack-tox-functional/f1c5a82/ : SUCCESS in 7m 15s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/9/check/openstack-tox-functional-py35/13598ab/ : SUCCESS in 9m 16s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/9/check/glance-eventlet-ssl-handshake-broken-py35/70a9d34/ : FAILURE in 3m 58s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/68/597368/9/check/devstack-plugin-ceph-tempest/8fd2ee1/ : SUCCESS in 1h 59m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ad07f39f70be0cebfe66c90bce9cca61f57c40e9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-10-26 06:58:10.000000000","message":"Patch Set 9: Code-Review-1\n\n(6 comments)\n\nOverall looks good, some suggestions inline.","accounts_in_message":[],"_revision_number":9},{"id":"b2379e5880752d54367bdd08af991767d2b2a3a6","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-10-26 15:57:23.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"1050eae0f31858c00f3a122946215972c53fa903","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-10-26 20:23:44.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f6e21dd74cbd8f7d1170c7f01ca46b282fa3eaf4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-10-26 20:59:03.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/68/597368/10/check-not-vote/ext-glance-zuul/bd2074e : SUCCESS in 34m 58s","accounts_in_message":[],"_revision_number":10},{"id":"7c005b45d2b080815b7a979dbffcd0feb3264900","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-27 03:04:40.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/68/597368/10/check/tempest-full/032ae17/ : SUCCESS in 1h 40m 13s\n- neutron-grenade http://logs.openstack.org/68/597368/10/check/neutron-grenade/4c7ba44/ : SUCCESS in 1h 00m 26s\n- tempest-full-py3 http://logs.openstack.org/68/597368/10/check/tempest-full-py3/7b0a7fc/ : SUCCESS in 1h 27m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/10/check/openstack-tox-lower-constraints/a5cca74/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/10/check/openstack-tox-pep8/f0651d4/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/10/check/openstack-tox-py27/e4f1547/ : SUCCESS in 4m 20s\n- openstack-tox-py35 http://logs.openstack.org/68/597368/10/check/openstack-tox-py35/ffe2a6a/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/68/597368/10/check/openstack-tox-py36/e6c6500/ : SUCCESS in 3m 44s\n- openstack-tox-docs http://logs.openstack.org/68/597368/10/check/openstack-tox-docs/dfc1490/html/ : SUCCESS in 4m 40s\n- openstack-tox-functional http://logs.openstack.org/68/597368/10/check/openstack-tox-functional/f5cdc72/ : FAILURE in 9m 37s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/10/check/openstack-tox-functional-py35/ddc7dc2/ : SUCCESS in 8m 46s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/10/check/glance-eventlet-ssl-handshake-broken-py35/e41b8df/ : FAILURE in 6m 23s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/68/597368/10/check/devstack-plugin-ceph-tempest/ddc3495/ : SUCCESS in 1h 50m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ff4c2513a2d13624204d0473ac63eb4bd30a5d43","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-10-27 06:26:58.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"fcb47fa91a1d50dfa5d449a0e3f0682406c5e99d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-10-27 07:00:22.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/68/597368/10/check-not-vote/ext-glance-zuul/5c3899c : SUCCESS in 33m 03s","accounts_in_message":[],"_revision_number":10},{"id":"8d0f4dc89ff0716a4562e0262e4d26b611d99dcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-27 08:41:50.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/68/597368/10/check/tempest-full/82d7a75/ : TIMED_OUT in 2h 07m 59s\n- neutron-grenade http://logs.openstack.org/68/597368/10/check/neutron-grenade/0dc5c27/ : SUCCESS in 43m 23s\n- tempest-full-py3 http://logs.openstack.org/68/597368/10/check/tempest-full-py3/e8ad597/ : SUCCESS in 1h 24m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/10/check/openstack-tox-lower-constraints/c6d4029/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/10/check/openstack-tox-pep8/415b6af/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/10/check/openstack-tox-py27/8e3ca97/ : SUCCESS in 4m 37s\n- openstack-tox-py35 http://logs.openstack.org/68/597368/10/check/openstack-tox-py35/1ed19d7/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/68/597368/10/check/openstack-tox-py36/046ea69/ : SUCCESS in 4m 05s\n- openstack-tox-docs http://logs.openstack.org/68/597368/10/check/openstack-tox-docs/50a47b9/html/ : SUCCESS in 5m 47s\n- openstack-tox-functional http://logs.openstack.org/68/597368/10/check/openstack-tox-functional/e1b6b56/ : FAILURE in 7m 25s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/10/check/openstack-tox-functional-py35/afb413f/ : TIMED_OUT in 31m 10s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/10/check/glance-eventlet-ssl-handshake-broken-py35/40caf12/ : FAILURE in 3m 48s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/68/597368/10/check/devstack-plugin-ceph-tempest/a2e2c1e/ : SUCCESS in 1h 31m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e3642998f89acf6b8cd4b695563b8412a17cb43b","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-10-27 17:20:52.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"4a050f5bf080282745073cada53bdede541f994e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-10-27 17:53:53.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/68/597368/10/check-not-vote/ext-glance-zuul/1e8b3f4 : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":10},{"id":"6cae9ea0299633fde07e654977d7b78f83e8c3da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-27 19:10:13.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/68/597368/10/check/tempest-full/8dfa1db/ : SUCCESS in 1h 43m 17s\n- neutron-grenade http://logs.openstack.org/68/597368/10/check/neutron-grenade/8fe30e8/ : SUCCESS in 57m 17s\n- tempest-full-py3 http://logs.openstack.org/68/597368/10/check/tempest-full-py3/6011edf/ : SUCCESS in 1h 16m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/10/check/openstack-tox-lower-constraints/d5773f9/ : SUCCESS in 3m 30s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/10/check/openstack-tox-pep8/b284199/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/10/check/openstack-tox-py27/8c54f8f/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/68/597368/10/check/openstack-tox-py35/5ec7d1e/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/68/597368/10/check/openstack-tox-py36/9e9730f/ : SUCCESS in 4m 06s\n- openstack-tox-docs http://logs.openstack.org/68/597368/10/check/openstack-tox-docs/4067df6/html/ : SUCCESS in 4m 06s\n- openstack-tox-functional http://logs.openstack.org/68/597368/10/check/openstack-tox-functional/039664e/ : SUCCESS in 7m 43s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/10/check/openstack-tox-functional-py35/a52f505/ : SUCCESS in 7m 11s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/10/check/glance-eventlet-ssl-handshake-broken-py35/886fc55/ : FAILURE in 4m 19s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/68/597368/10/check/devstack-plugin-ceph-tempest/96fc0a6/ : SUCCESS in 1h 43m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"01d24916a1ca49bcbc5f869789b7ada2a907e9f6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-10-30 05:31:17.000000000","message":"Patch Set 10: Code-Review+2\n\nthanks imacdonn!!","accounts_in_message":[],"_revision_number":10},{"id":"f779e38635e0e56367a5a84cdf10ce55b26d860a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2018-11-09 19:42:19.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)\n\nNot approving because there\u0027s one really minor change I\u0027d like you to make.  I am +2ing so that Abhishek or Sean can carry the +2 to the next patch set and ninja-approve the updated patch without waiting for me.  BTW, I tested this out in a devstack with multiple backends configured, and it works as advertised.","accounts_in_message":[],"_revision_number":10},{"id":"6c5d033ef0198a69f5f37ae748aa8e754555c0c8","author":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"date":"2018-11-09 20:35:56.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"8399d1bf8979a03f3d30244708e2685d7b9cca17","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-11-09 21:11:48.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/68/597368/11/check-not-vote/ext-glance-zuul/68fd32e : SUCCESS in 35m 27s","accounts_in_message":[],"_revision_number":11},{"id":"36c2864d7d3dc4adc535d5155722411eb8a4051b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 23:48:31.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/68/597368/11/check/tempest-full/cdc61d8/ : SUCCESS in 1h 51m 45s\n- neutron-grenade http://logs.openstack.org/68/597368/11/check/neutron-grenade/3e82a17/ : SUCCESS in 59m 20s\n- tempest-full-py3 http://logs.openstack.org/68/597368/11/check/tempest-full-py3/bde3a80/ : SUCCESS in 1h 35m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/11/check/openstack-tox-lower-constraints/33342dd/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/11/check/openstack-tox-pep8/e21d85a/ : SUCCESS in 6m 15s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/11/check/openstack-tox-py27/24c6dd8/ : SUCCESS in 5m 37s\n- openstack-tox-py35 http://logs.openstack.org/68/597368/11/check/openstack-tox-py35/eda19d7/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/68/597368/11/check/openstack-tox-py36/254f2ae/ : SUCCESS in 5m 06s\n- openstack-tox-docs http://logs.openstack.org/68/597368/11/check/openstack-tox-docs/60ce69d/html/ : SUCCESS in 5m 43s\n- openstack-tox-functional http://logs.openstack.org/68/597368/11/check/openstack-tox-functional/4f1425c/ : SUCCESS in 9m 29s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/11/check/openstack-tox-functional-py35/bc3d581/ : SUCCESS in 9m 02s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/68/597368/11/check/glance-eventlet-ssl-handshake-broken-py35/e873c29/ : FAILURE in 4m 17s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/68/597368/11/check/devstack-plugin-ceph-tempest/38c3f5a/ : FAILURE in 19m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"902d27d0fa4dbae9dc27a294d7efdb1f5ca4d0aa","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-11-12 05:15:38.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\nNinja approving (Considering Brian\u0027s +2 on previous PS).\nThan you Imacdonn!!!","accounts_in_message":[],"_revision_number":11},{"id":"c6be33ae236cd891787b50cc622da7b3fb3093b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 05:15:50.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"17f84cf292572215d7d87dd1117f53499b8d231e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 06:50:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"2fc4f917def33348b45298aaf95ff1838444df0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 06:50:23.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/68/597368/11/gate/tempest-full/a2079ae/ : SUCCESS in 1h 15m 02s\n- neutron-grenade http://logs.openstack.org/68/597368/11/gate/neutron-grenade/642521f/ : SUCCESS in 46m 02s\n- tempest-full-py3 http://logs.openstack.org/68/597368/11/gate/tempest-full-py3/3a1d7d8/ : SUCCESS in 1h 29m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/597368/11/gate/openstack-tox-lower-constraints/35c6328/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/68/597368/11/gate/openstack-tox-pep8/162414d/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/68/597368/11/gate/openstack-tox-py27/227ec60/ : SUCCESS in 4m 10s\n- openstack-tox-py35 http://logs.openstack.org/68/597368/11/gate/openstack-tox-py35/4bdedb7/ : SUCCESS in 3m 42s\n- openstack-tox-py36 http://logs.openstack.org/68/597368/11/gate/openstack-tox-py36/572a37d/ : SUCCESS in 3m 16s\n- openstack-tox-docs http://logs.openstack.org/68/597368/11/gate/openstack-tox-docs/d36476f/html/ : SUCCESS in 4m 39s\n- openstack-tox-functional http://logs.openstack.org/68/597368/11/gate/openstack-tox-functional/c82b038/ : SUCCESS in 7m 14s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/597368/11/gate/openstack-tox-functional-py35/b283139/ : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"b25a80fcd43d81c63fe88218e1c7819a73d86a8b","revisions":{"2dabf58ca9412407c3cea56905c8ba8f2998c0e1":{"kind":"REWORK","_number":1,"created":"2018-08-29 03:54:32.000000000","uploader":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"ref":"refs/changes/68/597368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/597368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/597368/1"}}},"commit":{"parents":[{"commit":"3023edd70ae58c0f45ecebbe4048c95d3d4ebe8d","subject":"Revert \"Add a default rootwrap.conf file.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3023edd70ae58c0f45ecebbe4048c95d3d4ebe8d"}]}],"author":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-08-29 03:52:05.000000000","tz":0},"committer":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-08-29 03:54:14.000000000","tz":0},"subject":"Checksum/hash in metadata when replacing locations","message":"Checksum/hash in metadata when replacing locations\n\nAllow embedding of values for checksum, os_hash_algo and\nos_hash_value in the \"metadata\" for the new location(s) if\n(and only if) the image is in \"queued\" state, and those attributes\ndo not already have values. Values may be included with one or\nmore of the \"locations\", but must not have different values. These\nitems will be removed (popped) from the metadata and stored only in\nthe appropriate image properties.\n\nspec-lite dependency to be added when specs repo ready for stein\n\nChange-Id: I1308fb1dec21002a9777bd0c77e9c02e59527551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2dabf58ca9412407c3cea56905c8ba8f2998c0e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2dabf58ca9412407c3cea56905c8ba8f2998c0e1"}]},"branch":"refs/heads/master"},"c29eb1cee6298115cc85eab78f87a6d86b293613":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-08-29 20:06:06.000000000","uploader":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"ref":"refs/changes/68/597368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/597368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/597368/2"}}},"commit":{"parents":[{"commit":"3023edd70ae58c0f45ecebbe4048c95d3d4ebe8d","subject":"Revert \"Add a default rootwrap.conf file.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3023edd70ae58c0f45ecebbe4048c95d3d4ebe8d"}]}],"author":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-08-29 03:52:05.000000000","tz":0},"committer":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-08-29 20:04:42.000000000","tz":0},"subject":"Checksum/hash in metadata when replacing locations","message":"Checksum/hash in metadata when replacing locations\n\nAllow embedding of values for checksum, os_hash_algo and\nos_hash_value in the \"metadata\" for the new location(s) if\n(and only if) the image is in \"queued\" state, and those attributes\ndo not already have values. Values may be included with one or\nmore of the \"locations\", but must not have different values. These\nitems will be removed (popped) from the metadata and stored only in\nthe appropriate image properties.\n\nDepends-On: https://review.openstack.org/597648\nChange-Id: I1308fb1dec21002a9777bd0c77e9c02e59527551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c29eb1cee6298115cc85eab78f87a6d86b293613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c29eb1cee6298115cc85eab78f87a6d86b293613"}]},"branch":"refs/heads/master"},"08047a6040df643f72533886b1adeb189c093b32":{"kind":"REWORK","_number":3,"created":"2018-09-01 00:11:08.000000000","uploader":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"ref":"refs/changes/68/597368/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/597368/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/597368/3"}}},"commit":{"parents":[{"commit":"3023edd70ae58c0f45ecebbe4048c95d3d4ebe8d","subject":"Revert \"Add a default rootwrap.conf file.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3023edd70ae58c0f45ecebbe4048c95d3d4ebe8d"}]}],"author":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-08-29 03:52:05.000000000","tz":0},"committer":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-09-01 00:11:05.000000000","tz":0},"subject":"Checksum/hash in metadata when setting locations","message":"Checksum/hash in metadata when setting locations\n\nAllow embedding of values for checksum, os_hash_algo and\nos_hash_value in the \"metadata\" for the new location(s) if\n(and only if) the image is in \"queued\" state, and those attributes\ndo not already have values. Values may be included with one or\nmore of the \"locations\", but must not have different values. These\nitems will be removed (popped) from the metadata and stored only in\nthe appropriate image properties.\n\nDepends-On: https://review.openstack.org/597648\nChange-Id: I1308fb1dec21002a9777bd0c77e9c02e59527551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08047a6040df643f72533886b1adeb189c093b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08047a6040df643f72533886b1adeb189c093b32"}]},"branch":"refs/heads/master"},"69a6ba4ca31f0fcfb1f6fc144db94785d319bd73":{"kind":"REWORK","_number":4,"created":"2018-09-04 22:58:59.000000000","uploader":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"ref":"refs/changes/68/597368/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/597368/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/597368/4"}}},"commit":{"parents":[{"commit":"3023edd70ae58c0f45ecebbe4048c95d3d4ebe8d","subject":"Revert \"Add a default rootwrap.conf file.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3023edd70ae58c0f45ecebbe4048c95d3d4ebe8d"}]}],"author":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-08-29 03:52:05.000000000","tz":0},"committer":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-09-04 22:58:57.000000000","tz":0},"subject":"Checksum/hash in metadata when setting locations","message":"Checksum/hash in metadata when setting locations\n\nAllow embedding of new values for checksum, os_hash_algo and os_hash_value in\nmetadata when adding to or replacing the locations for an image, if (and only\nif) the image is in \"queued\" status. Values may be included with one or more of\nthe \"locations\", but must not have different values. These items will be\nremoved (popped) from the metadata and stored only in the appropriate image\nproperties.\n\nDepends-On: https://review.openstack.org/597648\nChange-Id: I1308fb1dec21002a9777bd0c77e9c02e59527551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69a6ba4ca31f0fcfb1f6fc144db94785d319bd73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69a6ba4ca31f0fcfb1f6fc144db94785d319bd73"}]},"branch":"refs/heads/master"},"c1cd1f31fd7b913a236c6eed059d95246211ac36":{"kind":"REWORK","_number":5,"created":"2018-09-14 19:07:30.000000000","uploader":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"ref":"refs/changes/68/597368/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/597368/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/597368/5"}}},"commit":{"parents":[{"commit":"3023edd70ae58c0f45ecebbe4048c95d3d4ebe8d","subject":"Revert \"Add a default rootwrap.conf file.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3023edd70ae58c0f45ecebbe4048c95d3d4ebe8d"}]}],"author":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-08-29 03:52:05.000000000","tz":0},"committer":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-09-14 19:07:26.000000000","tz":0},"subject":"Embed validation data when setting locations","message":"Embed validation data when setting locations\n\nAllow embedding of checksum, os_hash_algo and os_hash_value when updating\nlocations for an image.\n\nDepends-On: https://review.openstack.org/597648\nChange-Id: I1308fb1dec21002a9777bd0c77e9c02e59527551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1cd1f31fd7b913a236c6eed059d95246211ac36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1cd1f31fd7b913a236c6eed059d95246211ac36"}]},"branch":"refs/heads/master"},"ecb4345950da0c3ed68238f4ca2756b40c0b4c26":{"kind":"REWORK","_number":6,"created":"2018-10-05 22:36:57.000000000","uploader":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"ref":"refs/changes/68/597368/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/597368/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/597368/6"}}},"commit":{"parents":[{"commit":"2f402ff9eab381b14bd357b882be7db73f32bec9","subject":"fix \"it\u0027s\" typos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f402ff9eab381b14bd357b882be7db73f32bec9"}]}],"author":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-10-05 22:33:54.000000000","tz":0},"committer":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-10-05 22:36:47.000000000","tz":0},"subject":"Embed validation data when setting locations","message":"Embed validation data when setting locations\n\nAllow embedding of checksum, os_hash_algo and os_hash_value in a\nspecial writeOnly json-patch object when updating locations for an image.\n\nDepends-On: https://review.openstack.org/597648\nChange-Id: I1308fb1dec21002a9777bd0c77e9c02e59527551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ecb4345950da0c3ed68238f4ca2756b40c0b4c26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ecb4345950da0c3ed68238f4ca2756b40c0b4c26"}]},"branch":"refs/heads/master"},"3b52e4ded53dc4666f78834ee9226b80048a5d00":{"kind":"REWORK","_number":7,"created":"2018-10-06 01:10:04.000000000","uploader":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"ref":"refs/changes/68/597368/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/597368/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/597368/7"}}},"commit":{"parents":[{"commit":"2f402ff9eab381b14bd357b882be7db73f32bec9","subject":"fix \"it\u0027s\" typos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f402ff9eab381b14bd357b882be7db73f32bec9"}]}],"author":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-10-05 22:33:54.000000000","tz":0},"committer":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-10-06 01:09:47.000000000","tz":0},"subject":"Embed validation data when setting locations","message":"Embed validation data when setting locations\n\nAllow embedding of checksum, os_hash_algo and os_hash_value in a\nspecial writeOnly json-patch object when updating locations for an image.\n\nDepends-On: https://review.openstack.org/597648\nChange-Id: I1308fb1dec21002a9777bd0c77e9c02e59527551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b52e4ded53dc4666f78834ee9226b80048a5d00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b52e4ded53dc4666f78834ee9226b80048a5d00"}]},"branch":"refs/heads/master"},"7b9220b1d943ed9ad0b9e1fbe8237fb21930ed1a":{"kind":"REWORK","_number":8,"created":"2018-10-19 23:45:20.000000000","uploader":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"ref":"refs/changes/68/597368/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/597368/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/597368/8"}}},"commit":{"parents":[{"commit":"2f402ff9eab381b14bd357b882be7db73f32bec9","subject":"fix \"it\u0027s\" typos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f402ff9eab381b14bd357b882be7db73f32bec9"}]}],"author":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-10-05 22:33:54.000000000","tz":0},"committer":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-10-19 23:44:56.000000000","tz":0},"subject":"Embed validation data when setting locations","message":"Embed validation data when setting locations\n\nAllow embedding of checksum, os_hash_algo and os_hash_value in a\nspecial writeOnly json-patch object when updating locations for an image.\n\nDepends-On: https://review.openstack.org/597648\nChange-Id: I1308fb1dec21002a9777bd0c77e9c02e59527551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b9220b1d943ed9ad0b9e1fbe8237fb21930ed1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b9220b1d943ed9ad0b9e1fbe8237fb21930ed1a"}]},"branch":"refs/heads/master"},"b5474c09f9b58d2073686464f3196f74691c04be":{"kind":"REWORK","_number":9,"created":"2018-10-24 22:21:31.000000000","uploader":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"ref":"refs/changes/68/597368/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/597368/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/597368/9"}}},"commit":{"parents":[{"commit":"2f402ff9eab381b14bd357b882be7db73f32bec9","subject":"fix \"it\u0027s\" typos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f402ff9eab381b14bd357b882be7db73f32bec9"}]}],"author":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-10-05 22:33:54.000000000","tz":0},"committer":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-10-24 22:20:53.000000000","tz":0},"subject":"Embed validation data in locations","message":"Embed validation data in locations\n\nAllow embedding of checksum, os_hash_algo and os_hash_value in a\nspecial writeOnly json-patch object when updating locations for an image.\n\nDepends-On: https://review.openstack.org/597648\nChange-Id: I1308fb1dec21002a9777bd0c77e9c02e59527551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5474c09f9b58d2073686464f3196f74691c04be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5474c09f9b58d2073686464f3196f74691c04be"}]},"branch":"refs/heads/master"},"48c244fafa46e1f4eb1ea8098fa7f9bed00e25bc":{"kind":"REWORK","_number":10,"created":"2018-10-26 20:23:44.000000000","uploader":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"ref":"refs/changes/68/597368/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/597368/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/597368/10"}}},"commit":{"parents":[{"commit":"2f402ff9eab381b14bd357b882be7db73f32bec9","subject":"fix \"it\u0027s\" typos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f402ff9eab381b14bd357b882be7db73f32bec9"}]}],"author":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-10-05 22:33:54.000000000","tz":0},"committer":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-10-26 20:23:21.000000000","tz":0},"subject":"Embed validation data in locations","message":"Embed validation data in locations\n\nAllow embedding of checksum, os_hash_algo and os_hash_value in a\nspecial writeOnly json-patch object when updating locations for an image.\n\nDepends-On: https://review.openstack.org/597648\nChange-Id: I1308fb1dec21002a9777bd0c77e9c02e59527551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/48c244fafa46e1f4eb1ea8098fa7f9bed00e25bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/48c244fafa46e1f4eb1ea8098fa7f9bed00e25bc"}]},"branch":"refs/heads/master"},"b25a80fcd43d81c63fe88218e1c7819a73d86a8b":{"kind":"REWORK","_number":11,"created":"2018-11-09 20:35:56.000000000","uploader":{"_account_id":23561,"name":"iain MacDonnell","email":"iain.macdonnell@oracle.com","username":"imacdonn"},"ref":"refs/changes/68/597368/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/597368/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/597368/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/597368/11"}}},"commit":{"parents":[{"commit":"2f402ff9eab381b14bd357b882be7db73f32bec9","subject":"fix \"it\u0027s\" typos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f402ff9eab381b14bd357b882be7db73f32bec9"}]}],"author":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-10-05 22:33:54.000000000","tz":0},"committer":{"name":"imacdonn","email":"iain.macdonnell@oracle.com","date":"2018-11-09 20:35:23.000000000","tz":0},"subject":"Embed validation data in locations","message":"Embed validation data in locations\n\nAllow embedding of checksum, os_hash_algo and os_hash_value in a\nspecial writeOnly json-patch object when updating locations for an image.\n\nDepends-On: https://review.openstack.org/597648\nChange-Id: I1308fb1dec21002a9777bd0c77e9c02e59527551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b25a80fcd43d81c63fe88218e1c7819a73d86a8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b25a80fcd43d81c63fe88218e1c7819a73d86a8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
