)]}'
{"id":"openstack%2Fglance~611253","triplet_id":"openstack%2Fglance~master~I86ec513c5fc653dbb97b79d953d8430f014e684f","project":"openstack/glance","branch":"master","topic":"bp/rethinking-filesystem-access","hashtags":[],"change_id":"I86ec513c5fc653dbb97b79d953d8430f014e684f","subject":"Rethinking filesystem access","status":"MERGED","created":"2018-10-17 07:31:38.000000000","updated":"2019-10-01 19:13:36.000000000","submitted":"2019-10-01 19:12:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"611253-1569957131297-a241494d","meta_rev_id":"7db5e5d3ac206956ccdcc3a2ca0b2734eb745fa8","_number":611253,"virtual_id_number":611253,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-01 10:38:13.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 19:12:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-01 12:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-01 12:09:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-10-01 19:13:36.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 15:58:49.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-09-12 14:44:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-30 20:51:12.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-10-01 10:38:13.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-01 12:09:32.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-10-01 19:12:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6b8d62a50bf8fb92cde7fabcda8113e6ec01dca","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-10-17 07:31:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3283fb112fb178c1c3ac5a84747140c0c0d1dd9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-10-17 07:38:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"08bc7858470a89f80a20594d588d342de851d30f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-17 09:52:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/53/611253/2/check/tempest-full/33ae30f/ : SUCCESS in 1h 27m 34s\n- neutron-grenade http://logs.openstack.org/53/611253/2/check/neutron-grenade/bbdf203/ : SUCCESS in 57m 36s\n- tempest-full-py3 http://logs.openstack.org/53/611253/2/check/tempest-full-py3/3c1e772/ : SUCCESS in 1h 50m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/2/check/openstack-tox-lower-constraints/e83b0bb/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/2/check/openstack-tox-pep8/d209ecf/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/2/check/openstack-tox-py27/4185f46/ : SUCCESS in 5m 51s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/2/check/openstack-tox-py35/37411b5/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/2/check/openstack-tox-py36/33ad93b/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/53/611253/2/check/openstack-tox-docs/fa4b46e/html/ : SUCCESS in 5m 19s\n- openstack-tox-functional http://logs.openstack.org/53/611253/2/check/openstack-tox-functional/b88aeac/ : SUCCESS in 9m 22s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/2/check/openstack-tox-functional-py35/d6883cc/ : SUCCESS in 8m 17s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/53/611253/2/check/glance-eventlet-ssl-handshake-broken-py35/f02ea99/ : FAILURE in 4m 07s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/2/check/devstack-plugin-ceph-tempest/acddafb/ : SUCCESS in 1h 57m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf4ea5608ae6b3540588d82766eeb42b2d6c39f2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-10-31 04:48:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa9be68dea4792554141b07b1a64847e1947a23c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-10-31 05:26:17.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/3/check-not-vote/ext-glance-zuul/f87c0db : SUCCESS in 37m 36s","accounts_in_message":[],"_revision_number":3},{"id":"11f9ea9b91879fe1a25bf2613de8019b300d7aa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-31 06:36:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/53/611253/3/check/tempest-full/67c0ab0/ : SUCCESS in 1h 41m 54s\n- neutron-grenade http://logs.openstack.org/53/611253/3/check/neutron-grenade/053e30c/ : SUCCESS in 56m 57s\n- tempest-full-py3 http://logs.openstack.org/53/611253/3/check/tempest-full-py3/b53f269/ : SUCCESS in 1h 09m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/3/check/openstack-tox-lower-constraints/7681dba/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/3/check/openstack-tox-pep8/26db54b/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/3/check/openstack-tox-py27/d2c0900/ : SUCCESS in 3m 58s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/3/check/openstack-tox-py35/4173c92/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/3/check/openstack-tox-py36/3b9423a/ : SUCCESS in 4m 00s\n- openstack-tox-docs http://logs.openstack.org/53/611253/3/check/openstack-tox-docs/078ac32/html/ : SUCCESS in 5m 48s\n- openstack-tox-functional http://logs.openstack.org/53/611253/3/check/openstack-tox-functional/84bf030/ : SUCCESS in 8m 32s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/3/check/openstack-tox-functional-py35/cc370aa/ : SUCCESS in 8m 18s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/53/611253/3/check/glance-eventlet-ssl-handshake-broken-py35/240166b/ : FAILURE in 5m 30s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/3/check/devstack-plugin-ceph-tempest/dfff269/ : FAILURE in 23m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"765c05171555ea7567d3c085c95b832f650eaa45","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-01-21 05:18:29.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"a03ff38879eb4e886f24c52c0244fa4de7b1a4de","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-15 06:28:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2c80f77725a30ebe56b9ced4bddff01c35660bf5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-15 07:06:02.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/4/check-not-vote/ext-glance-zuul/ad60491 : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":4},{"id":"bed070c2a8532c7ae6e74b334fe18339fbb79bfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-15 08:37:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/4/check/tempest-full/3c45ba0/ : SUCCESS in 1h 56m 52s\n- neutron-grenade http://logs.openstack.org/53/611253/4/check/neutron-grenade/5a22a33/ : SUCCESS in 1h 01m 17s\n- grenade-py3 http://logs.openstack.org/53/611253/4/check/grenade-py3/7e64d64/ : SUCCESS in 1h 03m 46s\n- tempest-full-py3 http://logs.openstack.org/53/611253/4/check/tempest-full-py3/d0cfaf1/ : SUCCESS in 1h 37m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/4/check/openstack-tox-lower-constraints/6d4dc9d/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/4/check/openstack-tox-pep8/f0e6bfe/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/4/check/openstack-tox-py27/8c9b154/ : SUCCESS in 4m 13s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/4/check/openstack-tox-py35/8a565f8/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/4/check/openstack-tox-py36/101f18d/ : SUCCESS in 3m 54s\n- openstack-tox-docs http://logs.openstack.org/53/611253/4/check/openstack-tox-docs/b03e615/html/ : SUCCESS in 4m 38s\n- openstack-tox-functional http://logs.openstack.org/53/611253/4/check/openstack-tox-functional/d523d0e/ : SUCCESS in 8m 14s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/4/check/openstack-tox-functional-py35/45218be/ : TIMED_OUT in 30m 49s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/53/611253/4/check/glance-eventlet-ssl-handshake-broken-py35/fe7f1cc/ : FAILURE in 4m 06s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/4/check/devstack-plugin-ceph-tempest/e808ba1/ : SUCCESS in 1h 52m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7e75bf321bc9341ff77bf6d4820101ff790858c1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-15 08:41:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"395040ce7d787d2a21d7073207bb7b0269699f16","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-15 09:16:02.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/4/check-not-vote/ext-glance-zuul/90bb556 : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":4},{"id":"6bff91016a8e2c1de1bc5590647c0570fa44e3f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-15 10:42:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/53/611253/4/check/tempest-full/d0e210b/ : SUCCESS in 1h 58m 45s\n- neutron-grenade http://logs.openstack.org/53/611253/4/check/neutron-grenade/b0f8e2b/ : SUCCESS in 1h 03m 28s\n- grenade-py3 http://logs.openstack.org/53/611253/4/check/grenade-py3/4b4c5d1/ : SUCCESS in 1h 08m 12s\n- tempest-full-py3 http://logs.openstack.org/53/611253/4/check/tempest-full-py3/1ff3ad5/ : SUCCESS in 1h 47m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/4/check/openstack-tox-lower-constraints/2a504b5/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/4/check/openstack-tox-pep8/d190e27/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/4/check/openstack-tox-py27/6734dd8/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/4/check/openstack-tox-py35/53aa4ea/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/4/check/openstack-tox-py36/8fb5a0a/ : SUCCESS in 3m 23s\n- openstack-tox-docs http://logs.openstack.org/53/611253/4/check/openstack-tox-docs/7836d61/html/ : SUCCESS in 4m 43s\n- openstack-tox-functional http://logs.openstack.org/53/611253/4/check/openstack-tox-functional/6ad49fd/ : SUCCESS in 7m 53s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/4/check/openstack-tox-functional-py35/6b3e81f/ : SUCCESS in 8m 10s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/53/611253/4/check/glance-eventlet-ssl-handshake-broken-py35/4507a92/ : FAILURE in 5m 30s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/4/check/devstack-plugin-ceph-tempest/9b20349/ : SUCCESS in 1h 50m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7c101a6bd12f3973b94c8e8fef1cd083eb59fd3b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-22 06:00:55.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"89cd575d63ce6ee559568ea136233c3ed0696abf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-27 18:12:47.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"09f851b5f7fdbe71cd0d37ae653da4cfd832412a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-27 18:17:05.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"dad6767542140cb602091c81a2644fd648ec8826","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-27 18:17:14.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"1b5e33046b358ba4568e5f15a4bc21902aa83292","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-27 20:33:30.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/6/check-not-vote/ext-glance-zuul/eb0f09c : SUCCESS in 39m 32s","accounts_in_message":[],"_revision_number":6},{"id":"6ea520ecc34e8f0b424cb5224ac91df3bd0c90a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 23:39:36.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/6/check/tempest-full/4df3451/ : SUCCESS in 2h 05m 03s\n- neutron-grenade http://logs.openstack.org/53/611253/6/check/neutron-grenade/3a78e96/ : SUCCESS in 1h 01m 25s\n- grenade-py3 http://logs.openstack.org/53/611253/6/check/grenade-py3/2d326e4/ : SUCCESS in 1h 02m 16s\n- tempest-full-py3 http://logs.openstack.org/53/611253/6/check/tempest-full-py3/cbec4fc/ : FAILURE in 1h 15m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/6/check/openstack-tox-lower-constraints/10818a5/ : SUCCESS in 6m 15s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/6/check/openstack-tox-pep8/df374bb/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/6/check/openstack-tox-py27/dc12687/ : SUCCESS in 6m 03s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/6/check/openstack-tox-py35/d4b7e0c/ : SUCCESS in 5m 04s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/6/check/openstack-tox-py36/f02d170/ : SUCCESS in 4m 18s\n- openstack-tox-docs http://logs.openstack.org/53/611253/6/check/openstack-tox-docs/07355c2/html/ : SUCCESS in 6m 09s\n- openstack-tox-functional http://logs.openstack.org/53/611253/6/check/openstack-tox-functional/a541418/ : TIMED_OUT in 31m 11s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/6/check/openstack-tox-functional-py35/4a7a88e/ : TIMED_OUT in 31m 19s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/53/611253/6/check/glance-eventlet-ssl-handshake-broken-py35/3eec57e/ : FAILURE in 6m 27s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/6/check/devstack-plugin-ceph-tempest/c3a708c/ : FAILURE in 1h 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a682c6bbf5402d4968edf0006da295b876a3e957","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-28 05:10:44.000000000","message":"Patch Set 6:\n\nrechek","accounts_in_message":[],"_revision_number":6},{"id":"0be46c1c543adee9182613775a26616cc81c79dd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-28 06:32:27.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"505ba194ff80167fc634630f0f5988de859ea154","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-28 07:08:39.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/7/check-not-vote/ext-glance-zuul/5f88dff : SUCCESS in 35m 52s","accounts_in_message":[],"_revision_number":7},{"id":"ea0625001ec1667bfc38b5fab3650bf5b195129e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-28 08:45:44.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/7/check/tempest-full/0255f89/ : TIMED_OUT in 2h 06m 19s\n- neutron-grenade http://logs.openstack.org/53/611253/7/check/neutron-grenade/667e8ed/ : SUCCESS in 1h 04m 06s\n- grenade-py3 http://logs.openstack.org/53/611253/7/check/grenade-py3/7678308/ : SUCCESS in 1h 20m 16s\n- tempest-full-py3 http://logs.openstack.org/53/611253/7/check/tempest-full-py3/3522498/ : SUCCESS in 1h 24m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/7/check/openstack-tox-lower-constraints/0ac64f1/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/7/check/openstack-tox-pep8/b987809/ : SUCCESS in 7m 07s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/7/check/openstack-tox-py27/740f684/ : SUCCESS in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/7/check/openstack-tox-py35/17ea90a/ : SUCCESS in 4m 52s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/7/check/openstack-tox-py36/b197dba/ : SUCCESS in 4m 26s\n- openstack-tox-docs http://logs.openstack.org/53/611253/7/check/openstack-tox-docs/06d5dac/html/ : SUCCESS in 7m 02s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/7/check/build-openstack-releasenotes/869518b/html/ : SUCCESS in 7m 46s\n- openstack-tox-functional http://logs.openstack.org/53/611253/7/check/openstack-tox-functional/947eef4/ : TIMED_OUT in 31m 05s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/7/check/openstack-tox-functional-py35/b63b797/ : TIMED_OUT in 30m 57s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/53/611253/7/check/glance-eventlet-ssl-handshake-broken-py35/c185de4/ : FAILURE in 5m 13s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/7/check/devstack-plugin-ceph-tempest/9668347/ : SUCCESS in 1h 51m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4c01a4b5ae7fa5faee6afca2fe922eb4f11be3ce","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-28 08:51:59.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"fe379d4170c69ed87a27d89fd6c4456d3484ad0b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-28 09:32:24.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/7/check-not-vote/ext-glance-zuul/e72a581 : SUCCESS in 40m 04s","accounts_in_message":[],"_revision_number":7},{"id":"92d804e4f38048fcb3855b43f16af3ee305647ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-28 10:47:46.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/7/check/tempest-full/60346ed/ : SUCCESS in 1h 51m 54s\n- neutron-grenade http://logs.openstack.org/53/611253/7/check/neutron-grenade/e0969d6/ : SUCCESS in 1h 03m 50s\n- grenade-py3 http://logs.openstack.org/53/611253/7/check/grenade-py3/9b77696/ : SUCCESS in 1h 07m 37s\n- tempest-full-py3 http://logs.openstack.org/53/611253/7/check/tempest-full-py3/286aba6/ : SUCCESS in 1h 42m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/7/check/openstack-tox-lower-constraints/fc7c4b7/ : SUCCESS in 7m 52s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/7/check/openstack-tox-pep8/50bc0ed/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/7/check/openstack-tox-py27/2617677/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/7/check/openstack-tox-py35/91e21c1/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/7/check/openstack-tox-py36/19e7c5e/ : SUCCESS in 3m 58s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/7/check/openstack-tox-py37/d238d34/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/53/611253/7/check/openstack-tox-docs/eae29cd/html/ : SUCCESS in 4m 59s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/7/check/build-openstack-releasenotes/2a5eeef/html/ : SUCCESS in 7m 06s\n- openstack-tox-functional http://logs.openstack.org/53/611253/7/check/openstack-tox-functional/b0d8eef/ : TIMED_OUT in 30m 56s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/7/check/openstack-tox-functional-py35/3646e8b/ : TIMED_OUT in 30m 53s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/53/611253/7/check/glance-eventlet-ssl-handshake-broken-py35/273c5ab/ : FAILURE in 10m 58s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/7/check/devstack-plugin-ceph-tempest/db51cf8/ : SUCCESS in 1h 21m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"36336c3e893af3671d9b7aefa9912917901285a4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-28 10:50:02.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"0775ab0f37ee11fee3ee19b3f029c1d39eb0960f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-28 13:06:39.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/7/check/tempest-full/3d788a6/ : SUCCESS in 1h 36m 54s\n- neutron-grenade http://logs.openstack.org/53/611253/7/check/neutron-grenade/d45227f/ : SUCCESS in 57m 55s\n- grenade-py3 http://logs.openstack.org/53/611253/7/check/grenade-py3/10150d1/ : SUCCESS in 1h 05m 43s\n- tempest-full-py3 http://logs.openstack.org/53/611253/7/check/tempest-full-py3/77d5987/ : TIMED_OUT in 2h 03m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/7/check/openstack-tox-lower-constraints/4e59823/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/7/check/openstack-tox-pep8/88270d5/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/7/check/openstack-tox-py27/33ec3fe/ : SUCCESS in 5m 17s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/7/check/openstack-tox-py35/6ec9fa3/ : SUCCESS in 5m 06s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/7/check/openstack-tox-py36/5813ad3/ : SUCCESS in 4m 18s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/7/check/openstack-tox-py37/9b967f5/ : SUCCESS in 5m 32s\n- openstack-tox-docs http://logs.openstack.org/53/611253/7/check/openstack-tox-docs/b084044/html/ : SUCCESS in 6m 01s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/7/check/build-openstack-releasenotes/824db27/html/ : SUCCESS in 9m 47s\n- openstack-tox-functional http://logs.openstack.org/53/611253/7/check/openstack-tox-functional/b56f6ca/ : TIMED_OUT in 31m 24s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/7/check/openstack-tox-functional-py35/d4e6132/ : TIMED_OUT in 31m 50s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/53/611253/7/check/glance-eventlet-ssl-handshake-broken-py35/2319690/ : FAILURE in 4m 51s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/7/check/devstack-plugin-ceph-tempest/a369231/ : SUCCESS in 1h 27m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3f7ef7aa35321430278451239b11ce09c40542e3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-28 13:22:15.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/7/check-not-vote/ext-glance-zuul/817a872 : FAILURE in 14m 04s","accounts_in_message":[],"_revision_number":7},{"id":"dbc1e6b87206901d946eaa4cb621a8dd368fdd7d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-28 13:32:07.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"c90d15c1927a029b54017511d262e5deafe16aed","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-28 14:22:03.000000000","message":"Patch Set 7: Workflow-1\n\nMoved to Train!!!!","accounts_in_message":[],"_revision_number":7},{"id":"774b9a26d6c9212dcf289cfd133dba2ac5e2ff14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-28 15:13:51.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/7/check/tempest-full/b692884/ : SUCCESS in 1h 35m 13s\n- neutron-grenade http://logs.openstack.org/53/611253/7/check/neutron-grenade/2432f24/ : SUCCESS in 57m 28s\n- grenade-py3 http://logs.openstack.org/53/611253/7/check/grenade-py3/8539f07/ : SUCCESS in 58m 55s\n- tempest-full-py3 http://logs.openstack.org/53/611253/7/check/tempest-full-py3/ab5e1d1/ : SUCCESS in 1h 34m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/7/check/openstack-tox-lower-constraints/71fd152/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/7/check/openstack-tox-pep8/be92af6/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/7/check/openstack-tox-py27/aba4581/ : SUCCESS in 4m 29s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/7/check/openstack-tox-py35/9255ce2/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/7/check/openstack-tox-py36/22abafe/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/7/check/openstack-tox-py37/0dcfcb5/ : SUCCESS in 5m 19s\n- openstack-tox-docs http://logs.openstack.org/53/611253/7/check/openstack-tox-docs/1f530d0/html/ : SUCCESS in 4m 59s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/7/check/build-openstack-releasenotes/5007513/html/ : SUCCESS in 7m 37s\n- openstack-tox-functional http://logs.openstack.org/53/611253/7/check/openstack-tox-functional/e4f43db/ : TIMED_OUT in 31m 02s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/7/check/openstack-tox-functional-py35/9739a69/ : TIMED_OUT in 30m 54s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/53/611253/7/check/glance-eventlet-ssl-handshake-broken-py35/64afc6d/ : FAILURE in 4m 41s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/7/check/devstack-plugin-ceph-tempest/1f46dcf/ : SUCCESS in 1h 34m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"852277c080145d1cebbe51c7ec6292bf837c4d5e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-28 15:26:08.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/7/check-not-vote/ext-glance-zuul/175f436 : SUCCESS in 34m 56s","accounts_in_message":[],"_revision_number":7},{"id":"667f32b618d30533b75b572f6034496115370560","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-14 08:41:19.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"75e3fdc81e7a02bc165601c8447ed916e742ae9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 10:31:26.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/8/check/tempest-full/1d5b307/ : SUCCESS in 1h 49m 04s\n- neutron-grenade http://logs.openstack.org/53/611253/8/check/neutron-grenade/a8b0e5a/ : SUCCESS in 50m 22s\n- grenade-py3 http://logs.openstack.org/53/611253/8/check/grenade-py3/6ce97b9/ : SUCCESS in 47m 45s\n- tempest-full-py3 http://logs.openstack.org/53/611253/8/check/tempest-full-py3/b68abac/ : SUCCESS in 1h 36m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/8/check/openstack-tox-lower-constraints/f9f9cd4/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/8/check/openstack-tox-pep8/998fac4/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/8/check/openstack-tox-py27/c6e0536/ : SUCCESS in 5m 09s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/8/check/openstack-tox-py35/6b21ffb/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/8/check/openstack-tox-py36/ffa2448/ : SUCCESS in 5m 12s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/8/check/openstack-tox-py37/7b32fa3/ : SUCCESS in 7m 23s\n- openstack-tox-docs http://logs.openstack.org/53/611253/8/check/openstack-tox-docs/b21ccaa/html/ : SUCCESS in 4m 54s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/8/check/build-openstack-releasenotes/1f155be/html/ : SUCCESS in 7m 14s\n- openstack-tox-functional http://logs.openstack.org/53/611253/8/check/openstack-tox-functional/c9368a2/ : FAILURE in 8m 22s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/8/check/openstack-tox-functional-py35/5c3273e/ : FAILURE in 8m 41s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/53/611253/8/check/glance-eventlet-ssl-handshake-broken-py35/b63579e/ : FAILURE in 2m 27s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/8/check/glance-code-constants-check/d15a227/ : SUCCESS in 4m 03s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/8/check/devstack-plugin-ceph-tempest/124f0be/ : SUCCESS in 1h 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a49daa5f10a2989e63db6f3fa73793b05dde9f8e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-14 10:33:41.000000000","message":"Patch Set 8:\n\nFunctional tests will fail until glance_store patch is merged.","accounts_in_message":[],"_revision_number":8},{"id":"89c24764c2814937e1cf8703eb58a6cda44c1dde","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-14 18:40:43.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/8/check-not-vote/ext-glance-zuul/0699dfd : FAILURE in 23m 03s","accounts_in_message":[],"_revision_number":8},{"id":"c0fb404a8becf2c83fd021d35cdd09fd4bd9b0fe","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-27 06:30:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"66912cfd7014ae2a2c6dbbc4961d2706fd455031","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-27 07:07:03.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/9/check-not-vote/ext-glance-zuul/3f8576a : FAILURE in 36m 35s","accounts_in_message":[],"_revision_number":9},{"id":"9114c23f84c57934b474632382d537ba45a59063","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 08:34:25.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/9/check/tempest-full/f35199f/ : SUCCESS in 1h 47m 26s\n- neutron-grenade http://logs.openstack.org/53/611253/9/check/neutron-grenade/28924df/ : SUCCESS in 55m 06s\n- grenade-py3 http://logs.openstack.org/53/611253/9/check/grenade-py3/32cf297/ : SUCCESS in 1h 01m 31s\n- tempest-full-py3 http://logs.openstack.org/53/611253/9/check/tempest-full-py3/2714dea/ : SUCCESS in 1h 30m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/9/check/openstack-tox-lower-constraints/c8e6e98/ : SUCCESS in 3m 12s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/9/check/openstack-tox-pep8/b0a2888/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/9/check/openstack-tox-py27/bde5d69/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/9/check/openstack-tox-py35/08136bd/ : SUCCESS in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/9/check/openstack-tox-py36/bf57b30/ : SUCCESS in 4m 15s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/9/check/openstack-tox-py37/c3e96c9/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/53/611253/9/check/openstack-tox-docs/5600d6e/html/ : SUCCESS in 5m 08s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/9/check/build-openstack-releasenotes/ed4e569/html/ : SUCCESS in 7m 53s\n- openstack-tox-functional http://logs.openstack.org/53/611253/9/check/openstack-tox-functional/962df83/ : FAILURE in 8m 51s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/9/check/openstack-tox-functional-py35/85305db/ : FAILURE in 8m 03s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/53/611253/9/check/glance-eventlet-ssl-handshake-broken-py35/c22dbce/ : FAILURE in 2m 44s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/9/check/glance-code-constants-check/ee93061/ : SUCCESS in 3m 30s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/9/check/devstack-plugin-ceph-tempest/13fa27f/ : SUCCESS in 1h 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"953ae9a2b379d4b4393e1c57f6904960a726f09a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-04 04:56:55.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"47e7409c389a34d9012ec7629cd4024ec9bb83ef","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-04 05:42:50.000000000","message":"Patch Set 10: Review-Priority+1","accounts_in_message":[],"_revision_number":10},{"id":"0e89c12a9fcac13325e2be89c8cf5eb6be78bb44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 06:52:03.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/10/check/tempest-full/65ede1b/ : SUCCESS in 1h 43m 08s\n- neutron-grenade http://logs.openstack.org/53/611253/10/check/neutron-grenade/c5d4e2d/ : SUCCESS in 53m 16s\n- grenade-py3 http://logs.openstack.org/53/611253/10/check/grenade-py3/2677dfd/ : SUCCESS in 56m 24s\n- tempest-full-py3 http://logs.openstack.org/53/611253/10/check/tempest-full-py3/8592ca3/ : SUCCESS in 1h 37m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/10/check/openstack-tox-lower-constraints/03195a4/ : SUCCESS in 3m 34s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/10/check/openstack-tox-pep8/28268a4/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/10/check/openstack-tox-py27/035b623/ : SUCCESS in 3m 58s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/10/check/openstack-tox-py35/783eeda/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/10/check/openstack-tox-py36/a386994/ : SUCCESS in 3m 58s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/10/check/openstack-tox-py37/b671ed2/ : SUCCESS in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/53/611253/10/check/openstack-tox-docs/cc1caf5/html/ : SUCCESS in 4m 30s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/10/check/build-openstack-releasenotes/55fee44/html/ : SUCCESS in 8m 14s\n- openstack-tox-functional http://logs.openstack.org/53/611253/10/check/openstack-tox-functional/7a12ed2/ : TIMED_OUT in 30m 48s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/10/check/openstack-tox-functional-py35/66c0438/ : FAILURE in 10m 12s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/53/611253/10/check/glance-eventlet-ssl-handshake-broken-py35/7f761c5/ : FAILURE in 2m 38s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/10/check/glance-code-constants-check/19afa95/ : SUCCESS in 3m 39s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/10/check/devstack-plugin-ceph-tempest/8fc4a86/ : SUCCESS in 1h 35m 18s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"67d4e4a820d8749dd7070a71e052567cae620da2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-04 14:21:59.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/10/check-not-vote/ext-glance-zuul/7483f09 : SUCCESS in 41m 41s","accounts_in_message":[],"_revision_number":10},{"id":"7f3e626ba3363522feb48db276a971b108730461","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 11:19:29.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"87681df377e2e687961ca05725974d28f0435d50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:33:10.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/11/check/tempest-full/6e960e1/ : SUCCESS in 1h 42m 39s\n- neutron-grenade http://logs.openstack.org/53/611253/11/check/neutron-grenade/723c034/ : SUCCESS in 54m 37s\n- grenade-py3 http://logs.openstack.org/53/611253/11/check/grenade-py3/bb564dd/ : SUCCESS in 58m 39s\n- tempest-full-py3 http://logs.openstack.org/53/611253/11/check/tempest-full-py3/cb7c1ef/ : FAILURE in 1h 21m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/11/check/openstack-tox-lower-constraints/bb1cb05/ : SUCCESS in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/11/check/openstack-tox-pep8/a5e0e08/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/11/check/openstack-tox-py27/5f135c4/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/11/check/openstack-tox-py35/1b69d39/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/11/check/openstack-tox-py36/49661f3/ : SUCCESS in 3m 43s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/11/check/openstack-tox-py37/1f25844/ : SUCCESS in 5m 03s\n- openstack-tox-docs http://logs.openstack.org/53/611253/11/check/openstack-tox-docs/1457c6a/html/ : SUCCESS in 4m 33s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/11/check/build-openstack-releasenotes/5b89d79/html/ : SUCCESS in 6m 58s\n- openstack-tox-functional http://logs.openstack.org/53/611253/11/check/openstack-tox-functional/b65541d/ : TIMED_OUT in 30m 56s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/11/check/openstack-tox-functional-py35/c529e13/ : FAILURE in 10m 02s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/53/611253/11/check/glance-eventlet-ssl-handshake-broken-py35/9ad4970/ : FAILURE in 4m 05s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/11/check/glance-code-constants-check/e626072/ : SUCCESS in 3m 45s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/11/check/devstack-plugin-ceph-tempest/a7be31c/ : SUCCESS in 1h 16m 10s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a9996c9f2591fa04ce27a452d9191cbe7763e1bd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 13:38:45.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"777fbf574be28c3c9a006265ab570c073f61f767","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-07 14:11:48.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/11/check-not-vote/ext-glance-zuul/34ea24a : SUCCESS in 39m 36s","accounts_in_message":[],"_revision_number":11},{"id":"068bd5fcd2699d173313c9b9f20f2ee273de4c53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 18:46:38.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/11/check/tempest-full/5ba203b/ : SUCCESS in 1h 44m 44s\n- neutron-grenade http://logs.openstack.org/53/611253/11/check/neutron-grenade/1b4f2b9/ : SUCCESS in 47m 38s\n- grenade-py3 http://logs.openstack.org/53/611253/11/check/grenade-py3/23d5fbd/ : SUCCESS in 56m 55s\n- tempest-full-py3 http://logs.openstack.org/53/611253/11/check/tempest-full-py3/c181cb1/ : SUCCESS in 1h 40m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/11/check/openstack-tox-lower-constraints/002debf/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/11/check/openstack-tox-pep8/b763fa3/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/11/check/openstack-tox-py27/e933edb/ : FAILURE in 3m 46s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/11/check/openstack-tox-py35/a9b096d/ : FAILURE in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/11/check/openstack-tox-py36/7de0b38/ : FAILURE in 4m 30s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/11/check/openstack-tox-py37/96fe834/ : FAILURE in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/53/611253/11/check/openstack-tox-docs/84ea073/html/ : SUCCESS in 6m 00s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/11/check/build-openstack-releasenotes/15fb8bb/html/ : SUCCESS in 8m 34s\n- openstack-tox-functional http://logs.openstack.org/53/611253/11/check/openstack-tox-functional/f23898e/ : TIMED_OUT in 30m 56s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/11/check/openstack-tox-functional-py35/a25202a/ : FAILURE in 10m 46s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/53/611253/11/check/glance-eventlet-ssl-handshake-broken-py3/e428dc1/ : FAILURE in 2m 45s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/11/check/glance-code-constants-check/4722d02/ : SUCCESS in 3m 47s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/11/check/devstack-plugin-ceph-tempest/879e8f9/ : SUCCESS in 1h 27m 34s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2fe1229e787bc345f4ee2f85752991fba68c7596","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-18 05:01:15.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"afe3ea5a0c696456504cd33c4fd25b3398964a81","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-18 05:40:12.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"593930927e9d5d44dc9482dd3d1d0bd4b682e653","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-18 07:13:36.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/12/check-not-vote/ext-glance-zuul/6833c71 : SUCCESS in 38m 12s","accounts_in_message":[],"_revision_number":12},{"id":"6d0068f3f7b2ec97b553f05d1d6349c4ed8ce40e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:38:19.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/53/611253/12/check/tempest-full/6bc9089/ : SUCCESS in 1h 42m 22s\n- neutron-grenade http://logs.openstack.org/53/611253/12/check/neutron-grenade/49c7428/ : SUCCESS in 1h 04m 25s\n- grenade-py3 http://logs.openstack.org/53/611253/12/check/grenade-py3/927a431/ : SUCCESS in 1h 03m 29s\n- tempest-full-py3 http://logs.openstack.org/53/611253/12/check/tempest-full-py3/c993955/ : SUCCESS in 1h 36m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/12/check/openstack-tox-lower-constraints/5cdbe13/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/12/check/openstack-tox-pep8/842c570/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/12/check/openstack-tox-py27/373beba/ : SUCCESS in 4m 15s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/12/check/openstack-tox-py35/0ca48e5/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/12/check/openstack-tox-py36/ef6ce07/ : SUCCESS in 4m 02s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/12/check/openstack-tox-py37/69f2b7e/ : SUCCESS in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/53/611253/12/check/openstack-tox-docs/18ea27a/html/ : SUCCESS in 4m 49s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/12/check/build-openstack-releasenotes/68a4eb3/html/ : SUCCESS in 7m 47s\n- openstack-tox-functional http://logs.openstack.org/53/611253/12/check/openstack-tox-functional/ae2ed29/ : SUCCESS in 7m 40s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/12/check/openstack-tox-functional-py35/4a69393/ : SUCCESS in 8m 32s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/53/611253/12/check/glance-eventlet-ssl-handshake-broken-py3/6b7c169/ : FAILURE in 6m 30s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/12/check/glance-code-constants-check/2884cc3/ : SUCCESS in 3m 58s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/12/check/devstack-plugin-ceph-tempest/d767d5a/ : SUCCESS in 1h 50m 18s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5dc17769f3051871009680780f3aba449e3ad97b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-18 08:57:17.000000000","message":"Patch Set 12: Review-Priority+1","accounts_in_message":[],"_revision_number":12},{"id":"52ae8b7b83c9580cc418815e62915bcd3a8a2e2d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-18 23:17:48.000000000","message":"Patch Set 12:\n\n(5 comments)\n\nThis looks good, a few things noted inline.  I\u0027m going to upload a new PS, the only difference will be the release note and the doc page (it was easier to rewrite them than to provide comments).","accounts_in_message":[],"_revision_number":12},{"id":"67423562776ef9252e8ff211aa2eaa0578089f69","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-18 23:18:44.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"53c5b1df9e9ff982bafaebe1eceb0f727c17170b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-18 23:19:22.000000000","message":"Patch Set 13:\n\nOnly the doc and releasenote are different from PS 12.","accounts_in_message":[],"_revision_number":13},{"id":"d8b81cac14af68a0fd90c3e2cef9532d42b17479","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-18 23:35:13.000000000","message":"Patch Set 13:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/13/check-not-vote/ext-glance-zuul/aa69bdc : FAILURE in 16m 09s","accounts_in_message":[],"_revision_number":13},{"id":"43452fb3a7d9c3fb14169a29b1d97c1aed7bbc74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 01:06:04.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/13/check/tempest-full/788c90f/ : SUCCESS in 1h 41m 36s\n- neutron-grenade http://logs.openstack.org/53/611253/13/check/neutron-grenade/728cf92/ : SUCCESS in 56m 34s\n- grenade-py3 http://logs.openstack.org/53/611253/13/check/grenade-py3/e90c36e/ : SUCCESS in 1h 06m 37s\n- tempest-full-py3 http://logs.openstack.org/53/611253/13/check/tempest-full-py3/dbb7861/ : SUCCESS in 1h 22m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/13/check/openstack-tox-lower-constraints/8f63f84/ : SUCCESS in 4m 46s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/13/check/openstack-tox-pep8/db9a43b/ : FAILURE in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/13/check/openstack-tox-py27/9b6b112/ : SUCCESS in 5m 44s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/13/check/openstack-tox-py35/789535f/ : SUCCESS in 6m 28s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/13/check/openstack-tox-py36/1ec2ed5/ : SUCCESS in 6m 11s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/13/check/openstack-tox-py37/52b5613/ : SUCCESS in 6m 29s\n- openstack-tox-docs http://logs.openstack.org/53/611253/13/check/openstack-tox-docs/7309979/html/ : SUCCESS in 5m 52s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/13/check/build-openstack-releasenotes/75578b3/html/ : SUCCESS in 8m 35s\n- openstack-tox-functional http://logs.openstack.org/53/611253/13/check/openstack-tox-functional/53df007/ : SUCCESS in 9m 17s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/13/check/openstack-tox-functional-py35/397235c/ : SUCCESS in 10m 59s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/53/611253/13/check/glance-eventlet-ssl-handshake-broken-py3/6e46e90/ : FAILURE in 3m 59s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/13/check/glance-code-constants-check/c961351/ : SUCCESS in 4m 04s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/13/check/devstack-plugin-ceph-tempest/8d43d29/ : SUCCESS in 1h 09m 42s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"06ce6cd921701e0cb0e186c1d24dad1d4f2c87d6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-19 05:28:04.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"6edfd18586c753789771809c6bde7672b91e3701","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-19 05:28:25.000000000","message":"Patch Set 14: Review-Priority+1","accounts_in_message":[],"_revision_number":14},{"id":"507c47f0a6fc8be1261a535dca0af0d9d04d3e10","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-19 06:18:11.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/14/check-not-vote/ext-glance-zuul/9aa4a46 : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":14},{"id":"4823cca6f9e2a2f9a7bdb718a7cc129c5efcd189","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 07:07:46.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/14/check/tempest-full/b19daa0/ : SUCCESS in 1h 38m 10s\n- neutron-grenade http://logs.openstack.org/53/611253/14/check/neutron-grenade/6997ed8/ : SUCCESS in 51m 54s\n- grenade-py3 http://logs.openstack.org/53/611253/14/check/grenade-py3/7e458f0/ : SUCCESS in 54m 06s\n- tempest-full-py3 http://logs.openstack.org/53/611253/14/check/tempest-full-py3/96b1362/ : SUCCESS in 1h 23m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/14/check/openstack-tox-lower-constraints/d9427d8/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/14/check/openstack-tox-pep8/81a8046/ : SUCCESS in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/14/check/openstack-tox-py27/142a1e7/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/14/check/openstack-tox-py35/5f84a6f/ : SUCCESS in 5m 47s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/14/check/openstack-tox-py36/bd4ff3d/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/14/check/openstack-tox-py37/9e2d3a7/ : SUCCESS in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/53/611253/14/check/openstack-tox-docs/3f97c10/html/ : SUCCESS in 5m 22s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/14/check/build-openstack-releasenotes/b51939e/html/ : SUCCESS in 7m 56s\n- openstack-tox-functional http://logs.openstack.org/53/611253/14/check/openstack-tox-functional/f6cc2da/ : SUCCESS in 7m 34s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/14/check/openstack-tox-functional-py35/38063c5/ : FAILURE in 9m 43s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/53/611253/14/check/glance-eventlet-ssl-handshake-broken-py3/be2380b/ : FAILURE in 3m 59s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/14/check/glance-code-constants-check/b4751b8/ : SUCCESS in 4m 54s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/14/check/devstack-plugin-ceph-tempest/0f3a69c/ : SUCCESS in 1h 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b3bd0c60cc159dfe9a7f0b849306bd4f47cf31d3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-19 07:23:42.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"6d4ed64ac9db6aa46804a7c3afbe849b6c7c3a90","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-19 08:14:59.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/14/check-not-vote/ext-glance-zuul/c40457d : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":14},{"id":"d8ab4b73a50d0d6e8515fc74cf4e6ff0b260600a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 08:48:45.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/53/611253/14/check/tempest-full/2623af5/ : SUCCESS in 1h 20m 38s\n- neutron-grenade http://logs.openstack.org/53/611253/14/check/neutron-grenade/8bf6e66/ : SUCCESS in 58m 36s\n- grenade-py3 http://logs.openstack.org/53/611253/14/check/grenade-py3/39f2c15/ : SUCCESS in 1h 09m 30s\n- tempest-full-py3 http://logs.openstack.org/53/611253/14/check/tempest-full-py3/3aafc83/ : SUCCESS in 1h 12m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/14/check/openstack-tox-lower-constraints/849b6a3/ : SUCCESS in 4m 05s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/14/check/openstack-tox-pep8/f21b67b/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/14/check/openstack-tox-py27/cf00c78/ : SUCCESS in 3m 58s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/14/check/openstack-tox-py35/126b859/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/14/check/openstack-tox-py36/023b029/ : SUCCESS in 3m 47s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/14/check/openstack-tox-py37/c4636e5/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/53/611253/14/check/openstack-tox-docs/f5f9e5c/html/ : SUCCESS in 5m 08s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/14/check/build-openstack-releasenotes/0e69741/html/ : SUCCESS in 8m 18s\n- openstack-tox-functional http://logs.openstack.org/53/611253/14/check/openstack-tox-functional/d8a47c5/ : SUCCESS in 8m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/14/check/openstack-tox-functional-py35/9611a50/ : SUCCESS in 9m 29s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/53/611253/14/check/glance-eventlet-ssl-handshake-broken-py3/a89b01e/ : FAILURE in 3m 49s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/14/check/glance-code-constants-check/ab1d000/ : SUCCESS in 3m 48s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/14/check/devstack-plugin-ceph-tempest/bfa6e0d/ : SUCCESS in 1h 08m 36s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2042c82d7e6aa22d5983e4e88fa40e3fe826399c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-19 12:24:48.000000000","message":"Patch Set 14: Code-Review-1\n\n(6 comments)\n\nJust a few minor points, then I think this is ready to go.","accounts_in_message":[],"_revision_number":14},{"id":"a81c083564cd34810777cdeccf8d3ff7fc69ad18","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-22 06:04:20.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"3c290984fe09f76e98ae8c15caea8a11f2a2aa94","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-22 06:09:25.000000000","message":"Patch Set 15:\n\n(5 comments)\n\nHi Brian, \n\nThank you for review, I have fixed your comment.\nI have made this patch dependent on https://review.opendev.org/670236 to address your comment given on that patch.","accounts_in_message":[],"_revision_number":15},{"id":"bec78bec2291a029c2ca841b592bb532ab383efe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 08:30:57.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/53/611253/15/check/tempest-full/9169e0a/ : SUCCESS in 1h 40m 06s\n- neutron-grenade http://logs.openstack.org/53/611253/15/check/neutron-grenade/534d00e/ : SUCCESS in 55m 17s\n- grenade-py3 http://logs.openstack.org/53/611253/15/check/grenade-py3/e7852f6/ : SUCCESS in 58m 46s\n- tempest-full-py3 http://logs.openstack.org/53/611253/15/check/tempest-full-py3/4a34f3e/ : SUCCESS in 1h 33m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/15/check/openstack-tox-lower-constraints/d851f16/ : SUCCESS in 5m 56s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/15/check/openstack-tox-pep8/ba2a0c2/ : SUCCESS in 5m 52s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/15/check/openstack-tox-py27/07c78ba/ : SUCCESS in 5m 25s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/15/check/openstack-tox-py35/bdb57d6/ : SUCCESS in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/15/check/openstack-tox-py36/230538f/ : SUCCESS in 7m 07s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/15/check/openstack-tox-py37/ea68530/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/53/611253/15/check/openstack-tox-docs/73c4765/html/ : SUCCESS in 5m 48s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/15/check/build-openstack-releasenotes/570dc06/html/ : SUCCESS in 9m 23s\n- openstack-tox-functional http://logs.openstack.org/53/611253/15/check/openstack-tox-functional/64799c6/ : SUCCESS in 9m 07s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/15/check/openstack-tox-functional-py35/4ea333c/ : SUCCESS in 9m 44s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/53/611253/15/check/glance-eventlet-ssl-handshake-broken-py3/5fc5b31/ : FAILURE in 5m 35s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/15/check/glance-code-constants-check/a070b3e/ : SUCCESS in 5m 16s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/15/check/devstack-plugin-ceph-tempest/504ea08/ : SUCCESS in 1h 16m 22s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"8298a0688befa1341913c9f0a67adda6d926d11d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-22 09:59:56.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/15/check-not-vote/ext-glance-zuul/19242f4 : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":15},{"id":"91b77c7a1537810e8401041bfdb2f18ec84157a8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-22 20:44:59.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)\n\nSmall problem noted inline, then this looks ready to go.","accounts_in_message":[],"_revision_number":15},{"id":"b9eee3b4760c85387184803b5222101742ccf61d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-23 05:06:11.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"c105a86fa1debe874cd69c81c90c28cf8920532b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-23 05:06:38.000000000","message":"Patch Set 15:\n\n(1 comment)\n\nHi Brian,\n\nThank you for review!","accounts_in_message":[],"_revision_number":15},{"id":"9919471cc240647a2e03ef3b2a6710bd7ca9d159","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-23 05:46:04.000000000","message":"Patch Set 16:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/16/check-not-vote/ext-glance-zuul/657c3a8 : SUCCESS in 39m 29s","accounts_in_message":[],"_revision_number":16},{"id":"30ddcbe65509466727059a87fa8ab22a92acc95e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 06:50:19.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/16/check/tempest-full/b508d81/ : FAILURE in 1h 42m 44s\n- neutron-grenade http://logs.openstack.org/53/611253/16/check/neutron-grenade/60245eb/ : SUCCESS in 1h 02m 20s\n- grenade-py3 http://logs.openstack.org/53/611253/16/check/grenade-py3/b7c3261/ : SUCCESS in 1h 02m 31s\n- tempest-full-py3 http://logs.openstack.org/53/611253/16/check/tempest-full-py3/121f419/ : SUCCESS in 1h 33m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/16/check/openstack-tox-lower-constraints/c17f644/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/16/check/openstack-tox-pep8/97617fb/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/16/check/openstack-tox-py27/b35bef2/ : SUCCESS in 5m 11s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/16/check/openstack-tox-py35/4ab9f9c/ : SUCCESS in 6m 02s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/16/check/openstack-tox-py36/3ce7825/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/16/check/openstack-tox-py37/541c683/ : SUCCESS in 5m 02s\n- openstack-tox-docs http://logs.openstack.org/53/611253/16/check/openstack-tox-docs/f20c5ad/html/ : SUCCESS in 4m 43s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/16/check/build-openstack-releasenotes/a25d2ed/html/ : SUCCESS in 9m 14s\n- openstack-tox-functional http://logs.openstack.org/53/611253/16/check/openstack-tox-functional/bd6547b/ : SUCCESS in 8m 47s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/16/check/openstack-tox-functional-py35/264e93c/ : SUCCESS in 9m 28s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/53/611253/16/check/glance-eventlet-ssl-handshake-broken-py3/e43b8e7/ : FAILURE in 5m 20s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/16/check/glance-code-constants-check/07b93fd/ : SUCCESS in 7m 03s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/16/check/devstack-plugin-ceph-tempest/fddca04/ : SUCCESS in 1h 17m 43s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"6f46016f84e6ed5f34568bafeb384922a6d594a6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-23 06:52:16.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"d8b2f88e481a33d373dc96ebe63b4c3924ceeb99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 08:47:57.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/53/611253/16/check/tempest-full/44ef6d9/ : SUCCESS in 1h 53m 33s\n- neutron-grenade http://logs.openstack.org/53/611253/16/check/neutron-grenade/09fa632/ : SUCCESS in 1h 02m 10s\n- grenade-py3 http://logs.openstack.org/53/611253/16/check/grenade-py3/7dc05fc/ : SUCCESS in 1h 09m 28s\n- tempest-full-py3 http://logs.openstack.org/53/611253/16/check/tempest-full-py3/bf780a4/ : SUCCESS in 1h 34m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/16/check/openstack-tox-lower-constraints/fe797d9/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/16/check/openstack-tox-pep8/3057127/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/16/check/openstack-tox-py27/6e946c4/ : SUCCESS in 5m 17s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/16/check/openstack-tox-py35/ad4c151/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/16/check/openstack-tox-py36/ff06102/ : SUCCESS in 5m 06s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/16/check/openstack-tox-py37/5651b1b/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/53/611253/16/check/openstack-tox-docs/347393c/html/ : SUCCESS in 6m 16s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/16/check/build-openstack-releasenotes/8221087/html/ : SUCCESS in 7m 30s\n- openstack-tox-functional http://logs.openstack.org/53/611253/16/check/openstack-tox-functional/7355590/ : SUCCESS in 8m 50s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/16/check/openstack-tox-functional-py35/fe57946/ : SUCCESS in 10m 33s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/53/611253/16/check/glance-eventlet-ssl-handshake-broken-py3/716d4c4/ : FAILURE in 4m 04s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/16/check/glance-code-constants-check/5d34cb4/ : SUCCESS in 4m 19s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/16/check/devstack-plugin-ceph-tempest/8afc991/ : FAILURE in 1h 23m 28s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ecf2496dae2f48ea5989f9fc6b945a0d5d4e45d1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-23 08:51:08.000000000","message":"Patch Set 16:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/16/check-not-vote/ext-glance-zuul/4c0136e : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":16},{"id":"d4600d3989a13d9a8bf1c38149bab458ec4aeba0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-23 13:20:58.000000000","message":"Patch Set 16: Code-Review+2\n\nLooks good.","accounts_in_message":[],"_revision_number":16},{"id":"a3900c06e3746542376516db99c8c5c0c52f62eb","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-07-25 11:55:13.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)\n\nWe should flip 2.8 as current with this patch as it states the multi-store being supported rather than experimental.\n\nOne more concern inline.","accounts_in_message":[],"_revision_number":16},{"id":"d4c729bd0d2c24c83d724d8c8983a0db16428f5b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-25 12:45:45.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"d45c6451c625aac498baed9046026da086c510d6","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-07-25 13:11:04.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"8b6a49cc922cdd531ddefeb94b6419b0f0ac2e56","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-25 15:58:37.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"06288a7d9f0597780b002604dc20130b2ab28e76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 19:25:40.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/17/check/tempest-full/d2626b5/ : SUCCESS in 1h 29m 21s\n- neutron-grenade http://logs.openstack.org/53/611253/17/check/neutron-grenade/d2d1354/ : SUCCESS in 1h 04m 29s\n- grenade-py3 http://logs.openstack.org/53/611253/17/check/grenade-py3/c381778/ : SUCCESS in 1h 08m 33s\n- tempest-full-py3 http://logs.openstack.org/53/611253/17/check/tempest-full-py3/7703a36/ : FAILURE in 1h 33m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/17/check/openstack-tox-lower-constraints/fe97ebc/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/17/check/openstack-tox-pep8/bbbe514/ : SUCCESS in 6m 37s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/17/check/openstack-tox-py27/dc75ded/ : SUCCESS in 5m 33s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/17/check/openstack-tox-py35/8cd9373/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/17/check/openstack-tox-py36/3dd7236/ : SUCCESS in 5m 36s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/17/check/openstack-tox-py37/9381e59/ : SUCCESS in 6m 41s\n- openstack-tox-docs http://logs.openstack.org/53/611253/17/check/openstack-tox-docs/4085f0a/html/ : SUCCESS in 6m 17s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/17/check/build-openstack-releasenotes/3883818/html/ : SUCCESS in 8m 31s\n- openstack-tox-functional http://logs.openstack.org/53/611253/17/check/openstack-tox-functional/c719fe2/ : SUCCESS in 9m 21s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/17/check/openstack-tox-functional-py35/53d4dfb/ : SUCCESS in 10m 44s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/53/611253/17/check/glance-eventlet-ssl-handshake-broken-py3/7736bd1/ : FAILURE in 4m 27s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/17/check/glance-code-constants-check/3530bc7/ : SUCCESS in 6m 05s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/17/check/devstack-plugin-ceph-tempest/6c6355e/ : SUCCESS in 1h 13m 14s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"fa0d622563ce57ce001fc5c8f3cbbb5c808419f7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-25 19:35:43.000000000","message":"Patch Set 17: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"720fe540ad4420295839515d7652287334b040c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 21:01:36.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/53/611253/17/check/tempest-full/72cd364/ : FAILURE in 1h 20m 26s\n- neutron-grenade http://logs.openstack.org/53/611253/17/check/neutron-grenade/0e8cd65/ : SUCCESS in 56m 43s\n- grenade-py3 http://logs.openstack.org/53/611253/17/check/grenade-py3/7bb1c0d/ : SUCCESS in 1h 01m 58s\n- tempest-full-py3 http://logs.openstack.org/53/611253/17/check/tempest-full-py3/4c6b977/ : SUCCESS in 1h 19m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/17/check/openstack-tox-lower-constraints/62626ea/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/17/check/openstack-tox-pep8/7bb7cac/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/17/check/openstack-tox-py27/5448f62/ : SUCCESS in 4m 11s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/17/check/openstack-tox-py35/f39af86/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/17/check/openstack-tox-py36/b78a0bc/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/17/check/openstack-tox-py37/5b837fe/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/53/611253/17/check/openstack-tox-docs/b7b778a/html/ : SUCCESS in 5m 04s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/17/check/build-openstack-releasenotes/e82eca4/html/ : SUCCESS in 8m 28s\n- openstack-tox-functional http://logs.openstack.org/53/611253/17/check/openstack-tox-functional/8809e4e/ : SUCCESS in 9m 20s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/17/check/openstack-tox-functional-py35/51c3ab8/ : SUCCESS in 11m 12s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/53/611253/17/check/glance-eventlet-ssl-handshake-broken-py3/5874dc2/ : FAILURE in 4m 05s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/17/check/glance-code-constants-check/c02a06d/ : SUCCESS in 4m 23s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/17/check/devstack-plugin-ceph-tempest/64ea3c7/ : SUCCESS in 1h 05m 40s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"80b4fe13761cfb2bc411e2b06a9e7d5865b5ecf6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-25 22:13:46.000000000","message":"Patch Set 17:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/17/check-not-vote/ext-glance-zuul/5cc910f : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":17},{"id":"7b3d961e297183d6a16e1980cb2d97a882cf71e7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-26 04:46:37.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"b2d3d3185b23a5ed22437db5aa013d749099302b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-26 05:23:17.000000000","message":"Patch Set 17:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/17/check-not-vote/ext-glance-zuul/8721952 : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":17},{"id":"d6c08851254c4aab59d4e6a26961221ef57c5153","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 06:23:50.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/53/611253/17/check/tempest-full/3b5d56f/ : SUCCESS in 1h 31m 51s\n- neutron-grenade http://logs.openstack.org/53/611253/17/check/neutron-grenade/da27c96/ : SUCCESS in 58m 35s\n- grenade-py3 http://logs.openstack.org/53/611253/17/check/grenade-py3/0dcc145/ : SUCCESS in 58m 56s\n- tempest-full-py3 http://logs.openstack.org/53/611253/17/check/tempest-full-py3/3f45db6/ : SUCCESS in 1h 21m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/17/check/openstack-tox-lower-constraints/d53f017/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/17/check/openstack-tox-pep8/0a5019e/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/17/check/openstack-tox-py27/64ffb28/ : SUCCESS in 3m 40s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/17/check/openstack-tox-py35/8d643ee/ : SUCCESS in 3m 51s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/17/check/openstack-tox-py36/911e63d/ : SUCCESS in 3m 42s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/17/check/openstack-tox-py37/dbd5f2b/ : SUCCESS in 4m 31s\n- openstack-tox-docs http://logs.openstack.org/53/611253/17/check/openstack-tox-docs/0e60fc1/html/ : SUCCESS in 4m 24s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/17/check/build-openstack-releasenotes/6ebf1ae/html/ : SUCCESS in 8m 18s\n- openstack-tox-functional http://logs.openstack.org/53/611253/17/check/openstack-tox-functional/c9303fa/ : SUCCESS in 8m 23s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/17/check/openstack-tox-functional-py35/97276c4/ : SUCCESS in 8m 30s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/53/611253/17/check/glance-eventlet-ssl-handshake-broken-py3/8bcdccc/ : FAILURE in 3m 50s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/17/check/glance-code-constants-check/e30b529/ : SUCCESS in 3m 20s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/17/check/devstack-plugin-ceph-tempest/7dcc687/ : SUCCESS in 1h 13m 16s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"b88bc5c86f905c9363e16d23682fba9a8542aca3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-26 13:40:51.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"c6800d7ff6fa0a75bea29ea43889c04865b12a0f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-26 13:41:52.000000000","message":"Patch Set 18:\n\nCaught a typo, but also revised the commit message a bit.  Please review it for accuracy.","accounts_in_message":[],"_revision_number":18},{"id":"51c8bf6b8b5e142f2ccdc6ed757d87b683df731a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-26 14:00:39.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"5c7047e6ad5e15583ffce597cb4a9e028ac3926f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-26 14:20:20.000000000","message":"Patch Set 18:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/18/check-not-vote/ext-glance-zuul/deae0dc : SUCCESS in 39m 09s","accounts_in_message":[],"_revision_number":18},{"id":"6c3aeb73047c501a11fa28c8ea243be0ab1ae883","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 15:30:11.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/53/611253/18/check/tempest-full/ac84729/ : SUCCESS in 1h 43m 14s\n- neutron-grenade http://logs.openstack.org/53/611253/18/check/neutron-grenade/d3fb66a/ : SUCCESS in 1h 08m 51s\n- grenade-py3 http://logs.openstack.org/53/611253/18/check/grenade-py3/1241218/ : SUCCESS in 1h 10m 04s\n- tempest-full-py3 http://logs.openstack.org/53/611253/18/check/tempest-full-py3/943d626/ : SUCCESS in 1h 25m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/611253/18/check/openstack-tox-lower-constraints/426016e/ : SUCCESS in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/53/611253/18/check/openstack-tox-pep8/9c34c32/ : SUCCESS in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/53/611253/18/check/openstack-tox-py27/3e9dc16/ : SUCCESS in 5m 26s\n- openstack-tox-py35 http://logs.openstack.org/53/611253/18/check/openstack-tox-py35/11b44ee/ : SUCCESS in 5m 40s\n- openstack-tox-py36 http://logs.openstack.org/53/611253/18/check/openstack-tox-py36/b67a308/ : SUCCESS in 4m 58s\n- openstack-tox-py37 http://logs.openstack.org/53/611253/18/check/openstack-tox-py37/80ccfe2/ : SUCCESS in 5m 49s\n- openstack-tox-docs http://logs.openstack.org/53/611253/18/check/openstack-tox-docs/7013991/html/ : SUCCESS in 6m 21s\n- build-openstack-releasenotes http://logs.openstack.org/53/611253/18/check/build-openstack-releasenotes/1014254/html/ : SUCCESS in 9m 28s\n- openstack-tox-functional http://logs.openstack.org/53/611253/18/check/openstack-tox-functional/3292477/ : SUCCESS in 8m 32s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/611253/18/check/openstack-tox-functional-py35/99c5fe0/ : SUCCESS in 9m 28s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/53/611253/18/check/glance-eventlet-ssl-handshake-broken-py3/b32582e/ : FAILURE in 4m 16s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/53/611253/18/check/glance-code-constants-check/0fa9021/ : SUCCESS in 4m 12s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/53/611253/18/check/devstack-plugin-ceph-tempest/35f2889/ : SUCCESS in 1h 16m 05s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"f54980e99c9760af0b76897b8579fb2a106fa14d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-01 14:10:42.000000000","message":"Patch Set 18: Code-Review+2\n\nAll my concerns have been addressed.","accounts_in_message":[],"_revision_number":18},{"id":"328a1689421613a26b6e5a83cd63239f78a5474b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-06 08:35:39.000000000","message":"Patch Set 18: Review-Priority+1","accounts_in_message":[],"_revision_number":18},{"id":"e0580b2f92076d6e9f27fb6d3e20dde33fe40326","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-08-13 15:58:49.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"a9bd07174f2b4691c8913431ad1c72db89a8a297","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-12 04:48:37.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"cb07eeaea315ce20b5ee29e8a94427561d4b4fba","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-12 04:49:44.000000000","message":"Patch Set 19: Review-Priority+1\n\nResolved merged conflict in test_wsgi.py\n\nNo changes in code.","accounts_in_message":[],"_revision_number":19},{"id":"28f0eacd0105d3eeb3a42788002899c8a223d9be","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-12 05:06:15.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"a2ba733e7385c11b1555b3cfc7ccc1f55558ea3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 06:33:49.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b7b59f9b575940bc944b720fa7a16798 : FAILURE in 13m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/83b7a291972c42b5bbf1808b062c4834 : FAILURE in 51m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/43cfccf6bcf543b1a6177ca027604c8e : SUCCESS in 1h 03m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/033a8c1b43154b3fbb1b0eab71347585 : SUCCESS in 1h 30m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b98873498de14aa293ef153a20086ea9 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a0bc924e9134748a2786c0fcd4aec69 : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec61a38a349547f3a686e986e4e8a128 : FAILURE in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b6fafacadc644e08e91c8d56ea8dfea : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c27ef806da54e8b81e6bea77d815b38 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75ab2eb941ca4c6d8887741b239f36c2 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b4264eaa07842ad889c77c3a1045d65 : SUCCESS in 9m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5351c6f7e57840619cf67b9121f30197 : FAILURE in 3m 57s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/632a045ce30e4f1dbeb4296ef35e7327 : SUCCESS in 8m 57s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/abfc1eeb8f4c4527941e4b94ce94d54a : FAILURE in 4m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/67c8a0005ea546ecbae37b91a24f57c4 : SUCCESS in 4m 20s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/27bc86fa13a941af86c3e8a42f009296 : FAILURE in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"09ece118f1b96cac57c1ddd49d82b0979182cb9c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-12 06:56:29.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"fe3da955ee3f3f02c92c0ba36cfe165c792b3947","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 11:39:39.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d64c800e79ba40b6a7f27af63b7bd970 : SUCCESS in 1h 38m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8020b3495e8d4a168ea6828cde3c46f5 : SUCCESS in 1h 06m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c4387efeb0524d368b5879f610b7a282 : SUCCESS in 1h 18m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b02d91dc54d43cb9c8084b498bea0e9 : SUCCESS in 1h 22m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c278fff41d564d0c8a333e060b1380ce : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/176c5fdfb0284ee186b2aa77ec0b7e78 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/108750aa6d294a0fb9030938260d0327 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f963c443563148608354a28448c42a5c : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae2d73c1861642b6b014144f17992e4f : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42ac3b7470174201891257328bc30351 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fe00e207231450e85d205afbcbb7085 : SUCCESS in 8m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7efb3bbdf19b45ccb62c795765b21965 : SUCCESS in 11m 39s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/98fd65fc0d7e42c8b2d982c3da5aadc5 : SUCCESS in 9m 53s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/78cc786897c24e6a8339dc0d42d152b4 : FAILURE in 5m 57s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/868ee3cb2fce4732938ad23f1132770f : SUCCESS in 4m 56s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/933a6a8ab1c04ec99dd2e63745d76656 : SUCCESS in 1h 12m 33s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"43408e615e2d78e137d36970b4dfcc9132b2c725","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-12 14:15:01.000000000","message":"Patch Set 19: Code-Review+2\n\nOnly change was merge conflict resolution and all tests are passing, so I renew my previous +2.","accounts_in_message":[],"_revision_number":19},{"id":"2dd69274e2859e41f17f6940f2fe90ad9dbdfa78","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-12 14:44:55.000000000","message":"Patch Set 19: Code-Review+2\n\n(1 comment)\n\nLooks fine to me. Will leave to another core to approve though.","accounts_in_message":[],"_revision_number":19},{"id":"542fb0a44d8197d364f28f3353b0859125aaa78f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-12 18:06:22.000000000","message":"Patch Set 19:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/19/check-not-vote/ext-glance-zuul/8f167a6 : FAILURE in 16m 02s","accounts_in_message":[],"_revision_number":19},{"id":"358d850f0d49a4c675a61377b15a879bb5815544","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-13 02:42:15.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)\n\nSean has a good point, plus the changes in common/config.py were already made (with more detail in the help text and deprecation note) by change I992cf468f9ce156ba51b1dd025459939acd8dce0 (which merged earlier today).\n\nI\u0027m going to rebase this on master, which I suspect will yield a merge conflict in common/config.py, which I\u0027ll resolve in favor of the earlier patch.  If that\u0027s the only conflict, I\u0027ll post a new patch and hopefully it will make it through tests before Abhishek and Erno start work on Friday.","accounts_in_message":[],"_revision_number":19},{"id":"9fd3622750d94968b575523047e2ee356352e6a1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-13 02:56:45.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"a0065dc8aa8a576049dd48cf4be46db68799ff63","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-13 02:58:58.000000000","message":"Patch Set 20:\n\nRebased on master and resolved merge conflict in common/config.py by dropping the changes to that file made on PS 19.","accounts_in_message":[],"_revision_number":20},{"id":"c0040e481adc57af102f2a7860d413079edd1c8d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-13 07:26:46.000000000","message":"Patch Set 20: Code-Review+2\n\nLooks good to me!!\n\nThank you Brian for rebasing!!","accounts_in_message":[],"_revision_number":20},{"id":"49d9fc3abeb9d0239eedf1e9369cd4051fdade92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 10:19:32.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/aacc8de821c24634a9744a2a0d520569 : SUCCESS in 1h 38m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f872021d58f7401aaeea60446df6fea9 : SUCCESS in 59m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bc882653318e4547974af0a6fe7b7536 : SUCCESS in 59m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75f5327b76b34aebbab699cd31101c2a : SUCCESS in 1h 25m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a9f0b594d9542fdb5fb0f389afabc0b : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccfbd69b06a14ad9ba955f8a12a969d5 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd03c3f0a32e427d9d8bc1a273ee2fa9 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/385380d720e74cdba5d87f5c3c525be1 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e9e5f1760ca4a5db004ea23a63051d3 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25849294398048dcbbfaee15e8f60131 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3308d2e0169449b68e91cbe494410ddd : SUCCESS in 10m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5127e8d142b04a9fa92e92494f343300 : TIMED_OUT in 31m 00s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/bb7fcb2ff28f431d8c60510f8a586533 : FAILURE in 9m 05s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/e592940d4dc745a5bd4469fa431f2013 : FAILURE in 3m 49s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6b86547af0c2422d95288d2c4a097b38 : SUCCESS in 6m 19s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/430d7c08964c42d2bfb26f6e3f8d5faa : SUCCESS in 1h 13m 01s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"0c1ed40dfcb011e4788b6534c14c7b09e56b463d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-13 10:41:06.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"7a63c39ebf4fd9180f211215d0501d50ca92b2b8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-13 14:06:04.000000000","message":"Patch Set 20:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/20/check-not-vote/ext-glance-zuul/4e0270b : FAILURE in 16m 11s","accounts_in_message":[],"_revision_number":20},{"id":"09970a72ea0ba1f0917b47b5ae9f610c68805d8b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 21:41:27.000000000","message":"Abandoned\n\nGate failures, fast failing.","accounts_in_message":[],"_revision_number":20},{"id":"d60ec0098e8e666ee41504c1fae54138a0a28574","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 21:41:35.000000000","message":"Restored","accounts_in_message":[],"_revision_number":20},{"id":"653ec2645f0b1456de3b2e1ad1f385dab2cf4c0e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-13 21:50:33.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1\n\nRenewing my +2.","accounts_in_message":[],"_revision_number":20},{"id":"616d69fa15d8f45ab6a6251f0827392b8b71b250","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 04:10:04.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/df3a78c65b27456183beec81fda5f4b4 : SUCCESS in 1h 43m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4d4df6b0b1214fdcb57ff47ad9d6e1ce : SUCCESS in 1h 08m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/052fb08ccf3846c9ab1be39cf6d78e57 : SUCCESS in 1h 01m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe5d531100cf4cfa894ee88fb1fc6e17 : SUCCESS in 1h 32m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e24eb071a0f34714a468c8e93e7378aa : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0e8624e23c84a6ab4be67d862167f62 : SUCCESS in 6m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21f86528d62445ba8166c0d65fbe34ed : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88ac6149b3e84074bda663e8b9bd4f5b : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80acfa61695745f69632d0559dda4280 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46cf2c36e8c944bab2f1d40526cfa5d1 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/313469f8794c42a090418031ac20a9c6 : SUCCESS in 10m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4bc9bcd2b13640b2897e03ff28cd24ed : TIMED_OUT in 31m 19s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e442ab2953ef42e1b1766b3f2b98b492 : FAILURE in 9m 26s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/3c284c4771c3477eb23a81f328b555c3 : FAILURE in 4m 45s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/42157aae836946f493ea8633da89dc44 : SUCCESS in 3m 50s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f994e27c05c9432da57e6ceb49d7fe20 : SUCCESS in 1h 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"1e373c8cb135e302d24806f7aea0200e8629fb1a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-14 08:01:15.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"f7445de310463a5e244f4b34ea92d17f26070068","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 10:41:39.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/bfb73314ee6f465b917e12c320418e4f : SUCCESS in 1h 52m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/78e1b521f3594dde8ca1e9f092dc9962 : SUCCESS in 1h 11m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d74f50c64a7f41b1bacabe1089e1c7bb : SUCCESS in 1h 25m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/999099b623e84418b6b017fd6507816a : SUCCESS in 1h 25m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b1f874402bb49798cd70f9c25bbce48 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5779f5d40ca4cac83b84bd7caef0d64 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a17c415d1c824895a5496ee148e566b9 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0da90923c52b41af915c12edbfadc75e : FAILURE in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/970cdfedae8e4b46ab4f1f91847402b5 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/637134a602a747b5b16e9ca0384dcee1 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/853aabb8ab7e4b92ad5808d78a562151 : SUCCESS in 10m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dade00539145471dbb83ef48d1fc3094 : TIMED_OUT in 30m 51s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/62125215433e49f3b3cd397650624130 : FAILURE in 9m 13s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/4bd0dba74bb948f5839b97707e4a506c : FAILURE in 5m 56s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2adf712afb734231a6a3f64d13954b33 : SUCCESS in 4m 58s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/084fd2b0d934497782956d824e1cf801 : SUCCESS in 1h 18m 23s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"9a2ace28ef856a7538c6888a2ab93f42fefee46d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-14 11:34:48.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"0ae5240dfa0d4c6c288d9b9560358e18863dd819","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 13:13:14.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d8b45c5f5e3b45a988c1f090b626ffee : SUCCESS in 1h 32m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5d24b748819547eab567a92915ee5613 : SUCCESS in 59m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c716c14022194d31984f48ce08f52561 : SUCCESS in 1h 00m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f9c74612c724bcd81875578748d826c : SUCCESS in 1h 23m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efa4b792bae54f63961094409acb238c : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/982f47953b4f407fa06873eb2918d729 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09e3f1f6de9744e2a658ad06cb57d979 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47f4dec97e2a487d81c7397579bb56a3 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75bd7ec124a3472ea1e5f169efa4b3b6 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec146eff95b405396a014012b3c09cb : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59818bb58b944ce6b73e3dffe733286b : SUCCESS in 10m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/775d00103049491f906126b9715c9ec6 : TIMED_OUT in 31m 24s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/5e162f74a39047058231b6e4b9cd4c20 : FAILURE in 9m 13s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/a37470e2fb9f42d28006fc7dc6247291 : FAILURE in 5m 44s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b317038b3bfe4de5a39e7982cf9b564e : SUCCESS in 3m 28s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6f63f3fdcf694be985170d4b535eb57e : SUCCESS in 1h 17m 35s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"438e30cb05b24921cddeeadb690806fbade3c4aa","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-14 13:16:15.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"c2df3880074d3fe880c5f78dbc57945452a794e9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-14 14:06:46.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/20/check-not-vote/ext-glance-zuul/190f90a : SUCCESS in 40m 05s","accounts_in_message":[],"_revision_number":20},{"id":"868d03f6f34f681edeba9f550e085fbd13a866cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 15:13:49.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b28720380c8942a184f03415279c6091 : SUCCESS in 1h 54m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ab9d6a1897d54fb4a7c283fa13b3c7a0 : SUCCESS in 1h 03m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8e679f3335ba4bd8b03011a5e2a49277 : SUCCESS in 59m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ed5b3e0daa542129202bc3f524cafe3 : SUCCESS in 1h 28m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f71208172ec74494a1f46570b0ed8ffe : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eb02c5ad266492fa38c303ab91f9dc3 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f09192f220ea4f0985499ad9af180b90 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f82156222a564704a24c7761013c9826 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a5dc217979340f2b8f3bbc24d487911 : FAILURE in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b45b51cf825e458480c7a6e344330e18 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21284e97d7b54b39a012c0b2fbff5698 : SUCCESS in 9m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/51e8cddd5e004cdbb47641df4fbc45e4 : TIMED_OUT in 30m 49s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/fb2b97b9bad2452c8f0c688699ced79f : FAILURE in 21m 48s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/25124333904d452c904eb580e5e73b1e : FAILURE in 4m 39s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/218741caf3fe4bac81973a1a7c60a593 : SUCCESS in 7m 50s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/362a0f4430374b15ab1831b2289b19ff : SUCCESS in 1h 12m 35s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"4d93af919ed2d6e46bd0e683278a3fb98d0d9fb6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-14 15:51:47.000000000","message":"Patch Set 20:\n\nsubunit parser problem, even for the py37 unit tests","accounts_in_message":[],"_revision_number":20},{"id":"6f50d1dfef5c0a0a6a21e27ae83d0ef78c6f8fca","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-14 15:51:55.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"6d0944b1e5d00590c7c33296a3eef6ecec1ac365","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-14 16:08:16.000000000","message":"Patch Set 20:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/20/check-not-vote/ext-glance-zuul/bcef9a4 : FAILURE in 15m 58s","accounts_in_message":[],"_revision_number":20},{"id":"be3d207d69c1e28c3d08a91b57b93b749a2c2db8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 17:30:32.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/55bb9b2301084dd2b70322cf5c5e4df9 : SUCCESS in 1h 30m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/890d32481c264bc7a45ad0cd3bb2bbc0 : SUCCESS in 53m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e32d179a3f7e42e6ac43130420414f4d : SUCCESS in 54m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b60f037da01d4888b6d41d9e82260f05 : SUCCESS in 1h 18m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fced1afa0b740c08e1e283cf8cc6e2b : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dffa6e3432c48708c1d62088cc84edc : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/220fd84a31fe49c3b017cc1920881e3d : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7768cc5d5bff42a48f6ccecc8a88c92a : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/219ffdc189cc4bbe8fd2f891500d2bfe : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79fc9b42b2e24497873715cc33f3f068 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2406ba64e014905850d53f110e2dce0 : SUCCESS in 8m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9717d08fe0c94173b0db0b247d0391e4 : TIMED_OUT in 30m 49s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/483a8b7e1e9c42449f8683e8efb66925 : FAILURE in 8m 08s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/51ee51db34ea4b1cab6f2b49e16b90a0 : FAILURE in 4m 11s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/eb0c6fa959764167ba548ceab42a9470 : SUCCESS in 4m 30s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3f1c977f2391448fbe03e85062875cac : SUCCESS in 1h 29m 54s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"d1847e9dca7764cf4528cf1a9d6d7b2637ef426e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-15 04:38:39.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"30531d21300f574e0107e509d70d3d3353679786","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-15 05:21:05.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/20/check-not-vote/ext-glance-zuul/dbd4db6 : SUCCESS in 42m 04s","accounts_in_message":[],"_revision_number":20},{"id":"2705dfae7a44e2696e7be897022aad82f4379af1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 06:37:18.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/38a8dda35fcb4769980815a5af6a1a2c : SUCCESS in 1h 53m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8eca2933f2f94eeb9a781c6bcfec1569 : SUCCESS in 57m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a7168c807f924a2c9ce7cf67f1dfc59a : SUCCESS in 56m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23437fff62834fe1b30437c31744752e : SUCCESS in 1h 11m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79cc4b26ef044fd88456028ab666d91b : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3543683b65c0436bb888935f3b4f6393 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a9d6e379821b4912b92d265d8f9ab726 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a8b3c5676a9429189e45108a2a179a5 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f468022c9584ae7abe50e84d348fbd1 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd2e8e53d8d04c51afa4a9d30bda4af1 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/010b4d8e0b2a43abafd29e4e45b0dfba : SUCCESS in 8m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/682df408c7de429db977fb34d8a7a210 : TIMED_OUT in 30m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/78870fe1c71e492f91d5552e6ac7a415 : FAILURE in 7m 44s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/945700748fbb4f24b79a4a50bca77caa : FAILURE in 4m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/85c9239fa5a24d4c9908af1dec520d5e : SUCCESS in 3m 46s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ff43d30130f54c109c257102c06187a7 : SUCCESS in 1h 03m 16s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"58b0054eb9b69990a53c3e87f9b7a198ff210d6b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-15 12:54:02.000000000","message":"Patch Set 20:\n\nI don\u0027t understand the functional job timeout.  I wonder whether it\u0027s also the subunit parser problem, just manifested differently in py27.","accounts_in_message":[],"_revision_number":20},{"id":"47730f0fe47490835794b834f373c387b6e1f817","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-15 12:54:11.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"6773ef1bf64b1fb5752d20d86bf77432d93bc38e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 15:04:20.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/996745dadaad4674826995f6e7548712 : SUCCESS in 2h 05m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6ece499232cd4f3b92a8f7d39ee0ccae : SUCCESS in 1h 01m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/14c093e0c0f843aaabf76a9e14232176 : SUCCESS in 1h 03m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d70d579dbec94389a19073cb1266946a : SUCCESS in 1h 22m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4b17ce518f3472cac9566eef227980f : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06125ecb16ac4f52a6e66522d77791d6 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/77099bb7793349c28aa2214675096cbc : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6918f01d775468a9fb4c8f152f1714c : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b3bfe4ab73e46cfa21ade864418ef3f : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/507298230561447487c63ce0d3c5e9e7 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/313e587cae764a0eb4f04c4bb4ed483d : SUCCESS in 9m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f92c75354254485caf03c07a9b996f13 : TIMED_OUT in 30m 41s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0f58a2f933e544b684c9a18ee3b63270 : FAILURE in 8m 08s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/574edf35b4174dfa8cd8cdc0c387919b : FAILURE in 4m 00s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/31b9c69471e44915bfb9d8a7ecff6ea0 : SUCCESS in 3m 24s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/78a3e9b610d847e1b8a3eda9181db0bb : SUCCESS in 1h 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"19d80da903b1f4dca5fd4a48067b89e887c42a99","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-15 15:10:46.000000000","message":"Patch Set 20:\n\nLooks like some problem with the patch, need to check it locally.","accounts_in_message":[],"_revision_number":20},{"id":"125f29fc80345e762b206d5f7c4439965598c2e7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-15 15:15:07.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/20/check-not-vote/ext-glance-zuul/ea33fd5 : SUCCESS in 46m 25s","accounts_in_message":[],"_revision_number":20},{"id":"3c4031b9d6ee7b567418d6647c18ff6500886ebb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-16 06:17:17.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"ba7f066fc6fa89dcfa44a90d1bf42f9df2689625","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 08:26:28.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6c3814d52ba94ed2b495b5b2cb800ed8 : SUCCESS in 1h 47m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bc7bf4dbe0b64365a89996e19314ae19 : SUCCESS in 1h 07m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1f28ac3f3204457eab1774fe2df3e187 : SUCCESS in 59m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/192a0264f01f4ffc8903d02c9ba32da7 : SUCCESS in 1h 33m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca596cb67b234a0297fdcc7c7ef731dc : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14df952eb2b34e46bcfa7b000c0e39e8 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8890db732376425384973864a1c5ce3c : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8b88ae50f0e4ce28661a10c85b9675a : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bf9ac6143ef45f38c486d9db565cf88 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c13a05ffb724b3c941fc697c525eba1 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91e7ada4fbb846cea7e1154b4848bdc4 : SUCCESS in 11m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1f718e3c4d0542b0aabfcabf90139774 : TIMED_OUT in 31m 02s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/39fc3304a14349c1b57dfae983d936fc : FAILURE in 9m 19s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/80bbd42fa54c455cad47a0f1696c65d8 : FAILURE in 4m 35s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4551b479b9594cc6a6263c8004ed9c5d : SUCCESS in 4m 02s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d8153b9f46264c3d9c322d3676653395 : SUCCESS in 1h 19m 28s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"d3957801bf648828fe46e08fb43d20030b3e4c38","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-16 09:14:56.000000000","message":"Patch Set 20:\n\nFunctional tests are timing out/failing because of this issue.\nhttps://bugs.launchpad.net/glance/+bug/1844108","accounts_in_message":[],"_revision_number":20},{"id":"c6b57cdb3c78d8c453307f433bf542d1df697f59","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-16 09:37:09.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/20/check-not-vote/ext-glance-zuul/82e88bf : SUCCESS in 50m 24s","accounts_in_message":[],"_revision_number":20},{"id":"d849f35a6fe58cf793815c453602e946b05c0944","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-16 09:41:51.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"c1bab57f09bd37f13dc9569e8212948f670b51af","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-16 10:49:46.000000000","message":"Patch Set 21:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/21/check-not-vote/ext-glance-zuul/16a6d2f : FAILURE in 15m 56s","accounts_in_message":[],"_revision_number":21},{"id":"aebd83a9e88fa5484cb75fe53003b14f64f44dd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 13:01:53.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/492a4921e2d14c7e814b3175070ef02d : SUCCESS in 2h 04m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3e156d5837544faba5df15055e809cfb : SUCCESS in 56m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0856e38b4c7b4ec5ab4340686eeb9aea : SUCCESS in 1h 08m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d4238ce1047430b9c7553a26d4d4970 : SUCCESS in 1h 34m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14bce083f9a1458facd6b97507062324 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27de5746f8f54bc4af069ff5b6714fef : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/940d5c74116d4312aa6f6272ebcc307e : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b3dbad01d22455a93039807f829a7e4 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ce928460dfc40f0b55bd3916222253d : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8dded3994564f67b925f68f7de47ce9 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14ccccc880a2483fb1cc4773fcc72649 : SUCCESS in 10m 34s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/db2558d20c96427eaca943cf1c8b607a : SUCCESS in 9m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d3039e9f6bd04640a1ed297e8502266f : SUCCESS in 9m 12s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/4d7e3cc2a2604f279348227a580b3154 : FAILURE in 5m 14s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8a315f18a7c34686a4305488cc1ef32a : SUCCESS in 3m 54s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f6affc18d1814866b34520564a21bf88 : TIMED_OUT in 2h 05m 25s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"ccc9df8da1bd5e643b444596fc10082143f1177f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-16 13:05:16.000000000","message":"Patch Set 21:\n\nThanks for tracking this down, Abhishek.  Glad to see this patch finally passing.  The parent of this patch passed all functional tests but failed py37 unit tests (subunit parser problem).  Can\u0027t be an error in the parent patch because otherwise this one wouldn\u0027t have passed all checks.","accounts_in_message":[],"_revision_number":21},{"id":"4894eae78450f91e7daf2c7ddb26b1e74ed56de3","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-09-30 16:04:24.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1\n\n(2 comments)\n\nNot holding this back due to the docs and code comment nits I have. We\u0027re holding RC1 for this after all.","accounts_in_message":[],"_revision_number":21},{"id":"0fa1e6eeee3d3d7574a45683b078497e9e9806d3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-30 17:36:47.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"c97ebdb2da89a2bfcb72cc700ef43e8dca41f673","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 19:39:30.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c65c5540575f46df8d8a8363baa2dc6c : SUCCESS in 1h 57m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/70a3abb8e4fa43419c4638a7af4084d2 : SUCCESS in 1h 06m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/deb97a5e89054d3cb781934bd862636e : SUCCESS in 51m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69fa6266f5d847149fe3e68a17f24c33 : SUCCESS in 1h 18m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/818692c0b27543979bbc48e9f167682a : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74a85810edc841848e78bb12561b9e64 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/12fcaf1d356a40d38e33624c0f250d97 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fba8895476bd4ddbb0ecf260ec4dde4b : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b7ed52d957a4df3b86126e5f8c795af : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94485c680524f4a8a19d4d6b8172570 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/226ce23329064a23bf3f9dfe49abdb88 : SUCCESS in 8m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e971619f32574347a268cc1a7be0fad0 : SUCCESS in 7m 55s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9f11022a2e1d4ae9afd66241d28a0fdd : SUCCESS in 7m 50s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/42e7eb0142c143ec8ba2f11bd9ea4033 : FAILURE in 4m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/22d3e082d1664ac3b8d1bfecce011310 : SUCCESS in 4m 09s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/76644aebacee4c05b03f132ed7b1de03 : FAILURE in 1h 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"e6d3c8bbbceccbce0f5417092a355f77f260f9d3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-30 20:51:12.000000000","message":"Patch Set 22: Workflow+1\n\nRebased patch passes all tests.","accounts_in_message":[],"_revision_number":22},{"id":"433f6d8642ec36978f4cbaa1f5285641ab5d4994","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 20:51:59.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"72936e101649ad1fb19abbc3973128acaee1c237","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 23:29:03.000000000","message":"Patch Set 22: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/5e4d6517d6494fd4af74bda6e28ecc7d : SUCCESS in 1h 32m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/da833e1d1f7c4f4cb4e7822549db3d7b : SUCCESS in 1h 01m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/87548347249e443fa7b3a9380fa8e5b5 : SUCCESS in 1h 05m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cb0d5c66f4b41c2ae93fb1cf9e28517 : SUCCESS in 1h 26m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c030d1fa9f54357bf558d9d0f842b9b : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d19978b7c0124cac9cab52277c442357 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc1ec6378d57446db6d7b84165e393f5 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5923e26dd22d4edd80f6097c805a586a : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2840a30e25f24462899468a9fe336d81 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63bef02996794c65928622d05aca086c : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4c032742ec148f387cc7b6670632f56 : SUCCESS in 8m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0ea4564312034365a65c672f0318b571 : FAILURE in 9m 34s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/63bb454630134a8cbfa8936f69fd1016 : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":22},{"id":"489ccbccfa43dcb6f71ce694c1d1be565e657667","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-01 00:58:20.000000000","message":"Patch Set 22:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/22/check-not-vote/ext-glance-zuul/8125286 : SUCCESS in 39m 24s","accounts_in_message":[],"_revision_number":22},{"id":"9f3e36f97256753e38c255dadaa9ff2378de17eb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-10-01 02:22:52.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"d79fed05429fda36dde86766293eff218490828c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-01 03:05:31.000000000","message":"Patch Set 22:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/22/check-not-vote/ext-glance-zuul/d4ca362 : SUCCESS in 42m 19s","accounts_in_message":[],"_revision_number":22},{"id":"a865ef2c1db0c624e9d2ace9301f5911640628d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 04:27:11.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1d863e605a8b4911a0e835a258359531 : SUCCESS in 1h 55m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e093501d728f445d937e7f14866165e7 : SUCCESS in 57m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/df342da98d424769a4c22f4eb09f1afc : SUCCESS in 1h 09m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d688302ca484d998376ef24c7896d83 : SUCCESS in 1h 11m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce733a9db4764e4e88b5f484bd959c45 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b5f80d109c04ffcae3dd35634df34b8 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d901cde13c3445c89494152ac4c1fd8c : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad49debf51b848a6912a11248dbb8d01 : FAILURE in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cf7eba9bdda4a4daff8606657ad53c4 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7391cfdde7434c95baae12d4cc4cc4be : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1910f35145284449841f81460bafad37 : SUCCESS in 7m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/be58db63f99146b4b3a1590804758d47 : FAILURE in 10m 03s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c4b64494cca249e4826fa94c7c3bd5ec : SUCCESS in 8m 49s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/2b200986894b47a0965cc68abe1245f6 : FAILURE in 5m 37s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a04be0ebdcdc48609b7ad2e3488236fc : SUCCESS in 4m 03s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2ddac07097be4cfbb80d38ace95975f8 : FAILURE in 1h 31m 34s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"01dec4296e91451d12d67d17a448654cd9837e5e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-10-01 04:33:19.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"864fa788d2a7ee29f4a4b1db8bea0cbe6fe29a6b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-01 05:13:23.000000000","message":"Patch Set 22:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/22/check-not-vote/ext-glance-zuul/8d6dc93 : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":22},{"id":"05f1e337401cb4ef748ef50fc4e71859e754faf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 06:24:45.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/52bb3a7a8d8f43ef90346b6183115dc8 : SUCCESS in 1h 47m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/37ea68cb992c411889f6ac4051641969 : SUCCESS in 57m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c44efe39c44b46e59fc213d3f1fa70d3 : SUCCESS in 54m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cac714c8cbee4888a5a09b9d901c94a5 : SUCCESS in 1h 18m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19d06205d6e64046952fc65f23682f83 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdd56b130b9046118e1aac0bb4c7756f : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b4aa0b535404479bb0902c95083f003 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12127cd4794f4264b16cc73474a5d531 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf11d43492ae447aaba717bd369d1310 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a5937109cc5402c90563c5a0018a701 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae027f977dfa4e95b73e0aefbdea96e3 : SUCCESS in 8m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9e403e4275eb4a8d9d021dd7287eeeed : FAILURE in 9m 02s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/7b658b31a99c407e96a4645503a2c3b0 : SUCCESS in 8m 13s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/75a85a2eb6cd43e0ac2642f7beb928a8 : FAILURE in 4m 13s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a312734afb2b4d77ae57c9294080aebb : SUCCESS in 3m 59s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/05179d918b4b42a7b61dd0bf4b9f6b9f : SUCCESS in 1h 06m 32s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"c0f8e4343704e7ac571cfbcd91d28f4e23aa3673","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-10-01 06:26:00.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"43ee1c65aea35a8bae2bdad5addf99d6f9526032","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-01 07:07:17.000000000","message":"Patch Set 22:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/22/check-not-vote/ext-glance-zuul/3ed3237 : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":22},{"id":"d105ed376fbb4d48ee9eedd2e2a4076c6227e25c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-10-01 07:47:34.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"c65a46c333e3302a13c2cd3086c0dadac55c36c2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-01 08:33:33.000000000","message":"Patch Set 23:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/23/check-not-vote/ext-glance-zuul/c8b61dd : SUCCESS in 45m 38s","accounts_in_message":[],"_revision_number":23},{"id":"54b443ea08dcdf9c4d2280175bb908ad14ed84a1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-10-01 09:23:41.000000000","message":"Patch Set 23:\n\nResolved comments given by erno and failing tests.","accounts_in_message":[],"_revision_number":23},{"id":"11448f1a456941608d5320e4d6401494a60f036f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 09:23:45.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/21bf0215a8a64d1992e70e15048ab74e : SUCCESS in 1h 32m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/605cea5c8d74416a8940851c65b40b81 : SUCCESS in 1h 00m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a61ce560da5f49adad59e6240ee161ce : SUCCESS in 58m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/007415fab5d142df931f36d11187f4c4 : SUCCESS in 1h 23m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb1c6174d6cf4f8eb6d1c72df7212ea3 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05f467d102d84fe295ca27806f1c89d8 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a852c391eacf443b81e98a06ec442431 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/044bea08aa664025a37f82f1f85e2e93 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/623342a8b08a41e3b2156ee4189b5671 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8a03bd47a85438b93529204b6826a2d : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ed392a1bfc5412ebc831bbbd7290884 : SUCCESS in 9m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6b8653bd1ad64cdd9e9917667a9e73bb : SUCCESS in 10m 32s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9d1f84a0dc69418caaff3a013f0c4790 : SUCCESS in 9m 45s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/e8c5642973dd41f586e37232b1b933a7 : FAILURE in 5m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b4c2058c767142dd9e3eb4325a2394c9 : SUCCESS in 4m 12s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9371c1a113504d32840b229ca9167660 : SUCCESS in 1h 18m 26s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"0a8532aeedc1a5840041adc1c9a9f08afb9917d2","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-10-01 09:30:43.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1\n\nVery minor change from previous PS to allow the tests pass. Going to ninja this in so we can tag RC1 asap.","accounts_in_message":[],"_revision_number":23},{"id":"75f557f0c43609d6198ad445991f3d4f0fd0aca8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 09:30:55.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"b8527db902af3ff84907d30d9ca72b0928956efd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-10-01 09:53:58.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"86df18e929f095bfda0ae613be35c18864c0f944","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-01 10:38:13.000000000","message":"Patch Set 24:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/611253/24/check-not-vote/ext-glance-zuul/2ddb2e2 : SUCCESS in 43m 55s","accounts_in_message":[],"_revision_number":24},{"id":"13940ae9c153184614ab62b400960fb751af1c82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 12:05:58.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4634bf63a13d48108bd763d6895ae4d2 : SUCCESS in 1h 56m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/83f0b71d0cc04b6aa46a1638d4723358 : SUCCESS in 56m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8f05ce99af074bd0bcb2d56c32557924 : SUCCESS in 52m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a663684a1f049aebb3e5e91b8391c94 : SUCCESS in 1h 33m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d95bb9aee714e419ed6200f64416aa0 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/565f918c829e478ba0118e5bd22983ac : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/997654cc03fc44568389d4eb60fcc553 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c923570f65f747d786472ed84c4e1cd7 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf061723c68944b8a3f43cd017c71967 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c47762e14a7948f68ddaa2572078f131 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9260eac0d0443f9935e21000d09ad0a : SUCCESS in 9m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b05a2781f3e245f6adebd96c1a82d60b : SUCCESS in 9m 24s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/303eaa9a3a2a4ba19f31174147294bb7 : SUCCESS in 10m 02s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/9d8edbd78c9e439dae86f90bfbe4c5e2 : FAILURE in 5m 04s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d2233ebe35a14802a296a4f83024a336 : SUCCESS in 3m 43s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c473edc172324855b77e6967320bd2f6 : SUCCESS in 1h 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"849c59a48288e29e8ef2c05a1ba31e8d0327b105","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-10-01 12:09:32.000000000","message":"Patch Set 24: Code-Review+2 Workflow+1\n\nCheck passed trying to regate.","accounts_in_message":[],"_revision_number":24},{"id":"fbc1e4755e9fe3244d09d7aac2f6f7f563351f50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 12:09:48.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"d870a90526df7d06ca022a758804143245dc8df2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 14:11:49.000000000","message":"Patch Set 24: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c61b0cd6a9254e6d99c8bbd6de07c460 : SUCCESS in 1h 55m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/920e7b51ccea40e68b671f6391620bcb : SUCCESS in 1h 12m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a317378d56e647279fadf961705ddc75 : SUCCESS in 1h 00m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f80ce41822947a3bd5089e01b501bf1 : SUCCESS in 1h 33m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28d36f3c21164fc0b9fe1d70e6d67919 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72b11313b04f46e7a0fbd78d5c7e4433 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1fc310296e7401da6d586a0fd9fc7c1 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79701aca670b41bd84f77a8b4d1d8efe : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/807ed1680d04460a90e1809b61989814 : FAILURE in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/346e8b0f7c154492bb9c81c805b8f6db : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87e87325577b4597b75634273621ec86 : SUCCESS in 8m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/72ead3a635804330898046278ff88429 : FAILURE in 13m 02s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/07aacf66b0cb40868e459af72669a285 : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":24},{"id":"fb6f077c22b35de6f12a7e1be5e257f752fe23bd","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-10-01 14:20:27.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"2535bf8dce372425b724ede70394744423fadb1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 17:21:29.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/94368f49432a40b1a2a75232485519ea : SUCCESS in 1h 58m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e74d2fb4d08a451d88f0675ddfc1f16c : SUCCESS in 1h 11m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9896d58f981b4d61ab7ff22ecd845af4 : SUCCESS in 1h 04m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7edec55543c442789df4181942602127 : SUCCESS in 1h 20m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8511b356bdaf440d984ce9d53c1097f1 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e668fd1885e04eeaab943105886bf4c7 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/782dab917ede4ca59634414ecca3a8a6 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e06a72083d724cfbaae354b2b2205087 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16d3dd834b6a4295a42d1aafe9272812 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa5b234a6f1d48d696ad7ab5931c329a : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a160538c398489e8fd1a8ebd19a6598 : SUCCESS in 9m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c161e0c204df41029bbc5a58d1c3849f : SUCCESS in 8m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3d75b3d87f534485bb3b652146d76530 : SUCCESS in 8m 18s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/94463be0e4854270812764d8fb2d9fe9 : FAILURE in 5m 08s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6b4544050b1f47cea5924ec40418da39 : SUCCESS in 4m 15s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4cf28059251f4a6c99e351f014d461d0 : SUCCESS in 1h 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"68954bdf65adde28216a0b557f745311a341b0e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 17:22:00.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"6e61b3b17183d1d1193a2eeb2ff3842aeaed4aa4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 19:12:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":24},{"id":"8383fcf6405a2fa3ccd97b19a3e341dcdc786fb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 19:12:11.000000000","message":"Patch Set 24: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/0c586f022ee349d4a9d26cceb4e3d21a : SUCCESS in 1h 47m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d6f54f0a072b4b59ba88c89b6841b961 : SUCCESS in 1h 08m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4d9af05af22d449e8fe03ff5b54a145c : SUCCESS in 1h 04m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9318f599454e4a10be6bb41a4b9289d7 : SUCCESS in 1h 27m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9a3d62e57794d52b28b241490033e96 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8baf343b4a0448fd9664ef54a7c6737f : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5186945783f24413a8197095fbf4fb46 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3e67b60dd6b4349ab22b7a90cf02b73 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1fd4d83a7ce1401bb80379299ec3157b : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f886939ce2024e86ad8189c15b4a0e79 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86ee6fdf9449475fb52f3287d6c43f80 : SUCCESS in 9m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c797bab9fb2f43378faa359ab01ec9cb : SUCCESS in 8m 55s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/019cd59b3fca42c2b4c30b63b75284ac : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":24},{"id":"2fc609bc869f7e7d2caf9b2c929f7da82782b815","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 19:13:36.000000000","message":"Patch Set 24:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3156035c2e0425db894ce4c99108a7a : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e173bc0e756491d995d181e43afbbee : SUCCESS in 53s","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"6dba83ba3a7c047df282f2edb7217548f43d6251","revisions":{"53a27727f9e2615ba6c8b09a3da519b3a30782a6":{"kind":"REWORK","_number":1,"created":"2018-10-17 07:31:38.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/1"}}},"commit":{"parents":[{"commit":"02a405a41012b96970c0a5cd7b5781deaadc5877","subject":"Merge \"Do not use oslo.messaging 9.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02a405a41012b96970c0a5cd7b5781deaadc5877"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:30:56.000000000","tz":0},"subject":"Rethinking filesystem access for tasks and staging API","message":"Rethinking filesystem access for tasks and staging API\n\nUsed multiple backends support to reserve certain filesystem stores\nfor tasks and staging operations. Deperecated \u0027work_dir\u0027 and\n\u0027node_staging_uri\u0027 configuration options.\n\nNOTE:\nOperator needs to reserved \u0027tasks_store\u0027 and \u0027staging_store\u0027 in\nglance-api.conf configuration file as shown below.\n\n[DEFAULT]\nenabled_backends \u003d tasks_store:file,staging_store:file\n\n[tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/605584\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53a27727f9e2615ba6c8b09a3da519b3a30782a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53a27727f9e2615ba6c8b09a3da519b3a30782a6"}]},"branch":"refs/heads/master"},"571efbd30769a3d77d94665607e10f0b4b0f9c82":{"kind":"REWORK","_number":2,"created":"2018-10-17 07:38:53.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/2"}}},"commit":{"parents":[{"commit":"02a405a41012b96970c0a5cd7b5781deaadc5877","subject":"Merge \"Do not use oslo.messaging 9.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02a405a41012b96970c0a5cd7b5781deaadc5877"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:38:46.000000000","tz":0},"subject":"Rethinking filesystem access for tasks and staging API","message":"Rethinking filesystem access for tasks and staging API\n\nUsed multiple backends support to reserve certain filesystem stores\nfor tasks and staging operations. Deperecated \u0027work_dir\u0027 and\n\u0027node_staging_uri\u0027 configuration options.\n\nNOTE:\nOperator needs to reserved \u0027tasks_store\u0027 and \u0027staging_store\u0027 in\nglance-api.conf configuration file as shown below.\n\n[DEFAULT]\nenabled_backends \u003d tasks_store:file,staging_store:file\n\n[tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/605584\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/571efbd30769a3d77d94665607e10f0b4b0f9c82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/571efbd30769a3d77d94665607e10f0b4b0f9c82"}]},"branch":"refs/heads/master"},"beb3947cc5cf527919bac7bd230348735500197b":{"kind":"REWORK","_number":3,"created":"2018-10-31 04:48:20.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/3"}}},"commit":{"parents":[{"commit":"0b13db85bb71de2ebfa1a9962703242b9d02a44e","subject":"Merge \"Replace openSUSE experimental check with newer version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b13db85bb71de2ebfa1a9962703242b9d02a44e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-31 04:48:10.000000000","tz":0},"subject":"Rethinking filesystem access for tasks and staging API","message":"Rethinking filesystem access for tasks and staging API\n\nUsed multiple backends support to reserve certain filesystem stores\nfor tasks and staging operations. Deperecated \u0027work_dir\u0027 and\n\u0027node_staging_uri\u0027 configuration options.\n\nNOTE:\nOperator needs to reserved \u0027tasks_store\u0027 and \u0027staging_store\u0027 in\nglance-api.conf configuration file as shown below.\n\n[DEFAULT]\nenabled_backends \u003d tasks_store:file,staging_store:file\n\n[tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/605584\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/beb3947cc5cf527919bac7bd230348735500197b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/beb3947cc5cf527919bac7bd230348735500197b"}]},"branch":"refs/heads/master"},"3516acad94cfe3cda6467961a4a93c18908697af":{"kind":"REWORK","_number":4,"created":"2019-02-15 06:28:37.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/4"}}},"commit":{"parents":[{"commit":"1d3f9ed234eb3b2c9e9e796818fb122372ec72fb","subject":"Merge \"Open Stein for data migrations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d3f9ed234eb3b2c9e9e796818fb122372ec72fb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-02-15 06:24:23.000000000","tz":0},"subject":"Rethinking filesystem access for tasks and staging API","message":"Rethinking filesystem access for tasks and staging API\n\nUsed multiple backends support to reserve certain filesystem stores\nfor tasks and staging operations. Deperecated \u0027work_dir\u0027 and\n\u0027node_staging_uri\u0027 configuration options.\n\nNOTE:\nOperator needs to reserved \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[DEFAULT]\nenabled_backends \u003d os_glance_tasks_store:file,os_glance_staging_store:file\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/605584\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3516acad94cfe3cda6467961a4a93c18908697af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3516acad94cfe3cda6467961a4a93c18908697af"}]},"branch":"refs/heads/master"},"43810a74a1962db43ed767f858c37ff7b04a1eaa":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-02-27 18:12:47.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/5"}}},"commit":{"parents":[{"commit":"1d3f9ed234eb3b2c9e9e796818fb122372ec72fb","subject":"Merge \"Open Stein for data migrations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d3f9ed234eb3b2c9e9e796818fb122372ec72fb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-02-27 18:12:32.000000000","tz":0},"subject":"Rethinking filesystem access for tasks and staging API","message":"Rethinking filesystem access for tasks and staging API\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options and reserve\ntwo filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese stores will be `hard-coded` in glance for the time being\nand injected to enabled_backends config option to load when\nglance_store will be initialized at the time of glance-api service\nstarts. Operator needs to insure that these stores will not\nbe included in enabled_backends config option in glance-api.conf.\n\nNOTE:\nOperator needs to define \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/639765/\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43810a74a1962db43ed767f858c37ff7b04a1eaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43810a74a1962db43ed767f858c37ff7b04a1eaa"}]},"branch":"refs/heads/master"},"725509a4bb8f93fb4b421c61afe176f324f5f931":{"kind":"REWORK","_number":6,"created":"2019-02-27 18:17:05.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/6"}}},"commit":{"parents":[{"commit":"1d3f9ed234eb3b2c9e9e796818fb122372ec72fb","subject":"Merge \"Open Stein for data migrations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d3f9ed234eb3b2c9e9e796818fb122372ec72fb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-02-27 18:16:23.000000000","tz":0},"subject":"Rethinking filesystem access for tasks and staging API","message":"Rethinking filesystem access for tasks and staging API\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options and reserve\ntwo filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese stores will be `hard-coded` in glance for the time being\nand injected to enabled_backends config option to load when\nglance_store will be initialized at the time of glance-api service\nstarts. Operator needs to insure that these stores will not\nbe included in enabled_backends config option in glance-api.conf.\n\nNOTE:\nOperator needs to define \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/639765/\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/725509a4bb8f93fb4b421c61afe176f324f5f931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/725509a4bb8f93fb4b421c61afe176f324f5f931"}]},"branch":"refs/heads/master"},"f78e146d6fce2a468ce4fedc99def42f49be67ed":{"kind":"REWORK","_number":7,"created":"2019-02-28 06:32:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/7"}}},"commit":{"parents":[{"commit":"4af0bc6712a6ed56c4ddef664a5cad3538467d55","subject":"Add job definition locations in comment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4af0bc6712a6ed56c4ddef664a5cad3538467d55"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-02-28 06:31:24.000000000","tz":0},"subject":"Rethinking filesystem access for tasks and staging API","message":"Rethinking filesystem access for tasks and staging API\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options and reserve\ntwo filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese stores will be `hard-coded` in glance for the time being\nand injected to enabled_backends config option to load when\nglance_store will be initialized at the time of glance-api service\nstarts. Operator needs to insure that these stores will not\nbe included in enabled_backends config option in glance-api.conf.\n\nNOTE:\nOperator needs to define \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/639765/\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f78e146d6fce2a468ce4fedc99def42f49be67ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f78e146d6fce2a468ce4fedc99def42f49be67ed"}]},"branch":"refs/heads/master"},"06f69c88790167950bae5784122728c6604d453c":{"kind":"REWORK","_number":8,"created":"2019-05-14 08:41:19.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/8"}}},"commit":{"parents":[{"commit":"5c81a3f38f78446543f222f8bde9971eac283d0e","subject":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c81a3f38f78446543f222f8bde9971eac283d0e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 08:40:34.000000000","tz":0},"subject":"Rethinking filesystem access for tasks and staging API","message":"Rethinking filesystem access for tasks and staging API\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options and reserve\ntwo filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese stores will be `hard-coded` in glance for the time being\nand injected to enabled_backends config option to load when\nglance_store will be initialized at the time of glance-api service\nstarts. Operator needs to insure that these stores will not\nbe included in enabled_backends config option in glance-api.conf.\n\nNOTE:\nOperator needs to define \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/639765/\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06f69c88790167950bae5784122728c6604d453c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06f69c88790167950bae5784122728c6604d453c"}]},"branch":"refs/heads/master"},"ca1a92c9f5d6b1c95a258b4118ff62bf50e60c42":{"kind":"REWORK","_number":9,"created":"2019-05-27 06:30:04.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/9"}}},"commit":{"parents":[{"commit":"5c81a3f38f78446543f222f8bde9971eac283d0e","subject":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c81a3f38f78446543f222f8bde9971eac283d0e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-27 06:28:27.000000000","tz":0},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese stores will be `hard-coded` in glance for the time being\nand injected to enabled_backends config option to load when\nglance_store will be initialized at the time of glance-api service\nstarts. Operator needs to insure that these stores will not\nbe included in enabled_backends config option in glance-api.conf.\n\nNOTE:\nOperator needs to define \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/639765/\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ca1a92c9f5d6b1c95a258b4118ff62bf50e60c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ca1a92c9f5d6b1c95a258b4118ff62bf50e60c42"}]},"branch":"refs/heads/master"},"0360d0bfab6672f4107e8893b4fc9ab86862e914":{"kind":"REWORK","_number":10,"created":"2019-06-04 04:56:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/10"}}},"commit":{"parents":[{"commit":"8f0180d30c4fcf4f25d6f18e83031b85370fb63f","subject":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8f0180d30c4fcf4f25d6f18e83031b85370fb63f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-04 04:52:27.000000000","tz":0},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese stores will be `hard-coded` in glance for the time being\nand injected to enabled_backends config option to load when\nglance_store will be initialized at the time of glance-api service\nstarts. Operator needs to insure that these stores will not\nbe included in enabled_backends config option in glance-api.conf.\n\nNOTE:\nOperator needs to define \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0360d0bfab6672f4107e8893b4fc9ab86862e914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0360d0bfab6672f4107e8893b4fc9ab86862e914"}]},"branch":"refs/heads/master"},"31efd5996d3f2a766eb9d496c412d00956488c45":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-06-07 11:19:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/11"}}},"commit":{"parents":[{"commit":"be13eb05a0bc127069e743c9a6b14aa81b448b5d","subject":"Remove additional \" from config help","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be13eb05a0bc127069e743c9a6b14aa81b448b5d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-07 11:18:46.000000000","tz":0},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese stores will be `hard-coded` in glance for the time being\nand injected to enabled_backends config option to load when\nglance_store will be initialized at the time of glance-api service\nstarts. Operator needs to insure that these stores will not\nbe included in enabled_backends config option in glance-api.conf.\n\nNOTE:\nOperator needs to define \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/31efd5996d3f2a766eb9d496c412d00956488c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/31efd5996d3f2a766eb9d496c412d00956488c45"}]},"branch":"refs/heads/master"},"ef7ac138bc30ab9ca3c6fc8dfe197f2f5558deab":{"kind":"REWORK","_number":12,"created":"2019-06-18 05:40:12.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/12"}}},"commit":{"parents":[{"commit":"b5b827f9eabd5b85e56ef522f94bb971f5308091","subject":"Merge \"Bring py27 tests back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5b827f9eabd5b85e56ef522f94bb971f5308091"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-18 05:40:05.000000000","tz":0},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese stores will be `hard-coded` in glance for the time being\nand injected to enabled_backends config option to load when\nglance_store will be initialized at the time of glance-api service\nstarts. Operator needs to insure that these stores will not\nbe included in enabled_backends config option in glance-api.conf.\n\nNOTE:\nOperator needs to define \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef7ac138bc30ab9ca3c6fc8dfe197f2f5558deab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef7ac138bc30ab9ca3c6fc8dfe197f2f5558deab"}]},"branch":"refs/heads/master"},"421ecc179488b68de200c39f7b4770d048ee2ece":{"kind":"REWORK","_number":13,"created":"2019-07-18 23:18:44.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/53/611253/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/13"}}},"commit":{"parents":[{"commit":"b5b827f9eabd5b85e56ef522f94bb971f5308091","subject":"Merge \"Bring py27 tests back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5b827f9eabd5b85e56ef522f94bb971f5308091"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-07-18 23:18:10.000000000","tz":-240},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese stores will be `hard-coded` in glance for the time being\nand injected to enabled_backends config option to load when\nglance_store will be initialized at the time of glance-api service\nstarts. Operator needs to insure that these stores will not\nbe included in enabled_backends config option in glance-api.conf.\n\nNOTE:\nOperator needs to define \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/421ecc179488b68de200c39f7b4770d048ee2ece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/421ecc179488b68de200c39f7b4770d048ee2ece"}]},"branch":"refs/heads/master"},"cb8aad5e70f8941768577fda6f0ff5cb032fc025":{"kind":"REWORK","_number":14,"created":"2019-07-19 05:28:04.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/14"}}},"commit":{"parents":[{"commit":"b5b827f9eabd5b85e56ef522f94bb971f5308091","subject":"Merge \"Bring py27 tests back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5b827f9eabd5b85e56ef522f94bb971f5308091"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-19 05:27:49.000000000","tz":0},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese stores will be `hard-coded` in glance for the time being\nand injected to enabled_backends config option to load when\nglance_store will be initialized at the time of glance-api service\nstarts. Operator needs to insure that these stores will not\nbe included in enabled_backends config option in glance-api.conf.\n\nNOTE:\nOperator needs to define \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb8aad5e70f8941768577fda6f0ff5cb032fc025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb8aad5e70f8941768577fda6f0ff5cb032fc025"}]},"branch":"refs/heads/master"},"e1fd1026b8935b9dbf38ecd78245eac73644d8aa":{"kind":"REWORK","_number":15,"created":"2019-07-22 06:04:20.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/15"}}},"commit":{"parents":[{"commit":"7f74a92338894d01feebf706663b1b5a10d99ece","subject":"Image deletion returns 500 if \u0027file\u0027 store is not enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7f74a92338894d01feebf706663b1b5a10d99ece"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-22 06:03:51.000000000","tz":0},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese stores will be `hard-coded` in glance for the time being\nand injected to enabled_backends config option to load when\nglance_store will be initialized at the time of glance-api service\nstarts. Operator needs to insure that these stores will not\nbe included in enabled_backends config option in glance-api.conf.\n\nNOTE:\nOperator needs to define \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1fd1026b8935b9dbf38ecd78245eac73644d8aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1fd1026b8935b9dbf38ecd78245eac73644d8aa"}]},"branch":"refs/heads/master"},"65e94f7a8eebf70d0385d29219019a2cdad688c8":{"kind":"REWORK","_number":16,"created":"2019-07-23 05:06:11.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/16"}}},"commit":{"parents":[{"commit":"7f74a92338894d01feebf706663b1b5a10d99ece","subject":"Image deletion returns 500 if \u0027file\u0027 store is not enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7f74a92338894d01feebf706663b1b5a10d99ece"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-23 05:06:01.000000000","tz":0},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese stores will be `hard-coded` in glance for the time being\nand injected to enabled_backends config option to load when\nglance_store will be initialized at the time of glance-api service\nstarts. Operator needs to insure that these stores will not\nbe included in enabled_backends config option in glance-api.conf.\n\nNOTE:\nOperator needs to define \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65e94f7a8eebf70d0385d29219019a2cdad688c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65e94f7a8eebf70d0385d29219019a2cdad688c8"}]},"branch":"refs/heads/master"},"f6b182d3a93de2e60ef4c9e8b0b62bc1cfc04a1f":{"kind":"REWORK","_number":17,"created":"2019-07-25 15:58:37.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/53/611253/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/17"}}},"commit":{"parents":[{"commit":"7f74a92338894d01feebf706663b1b5a10d99ece","subject":"Image deletion returns 500 if \u0027file\u0027 store is not enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7f74a92338894d01feebf706663b1b5a10d99ece"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-07-25 15:58:26.000000000","tz":-240},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese stores will be `hard-coded` in glance for the time being\nand injected to enabled_backends config option to load when\nglance_store will be initialized at the time of glance-api service\nstarts. Operator needs to insure that these stores will not\nbe included in enabled_backends config option in glance-api.conf.\n\nNOTE:\nOperator needs to define \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027 in glance-api.conf configuration\nfile as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f6b182d3a93de2e60ef4c9e8b0b62bc1cfc04a1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f6b182d3a93de2e60ef4c9e8b0b62bc1cfc04a1f"}]},"branch":"refs/heads/master"},"ed9e73ba7c90a4b278b5f0048664bd7002220835":{"kind":"REWORK","_number":18,"created":"2019-07-26 13:40:51.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/53/611253/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/18"}}},"commit":{"parents":[{"commit":"7f74a92338894d01feebf706663b1b5a10d99ece","subject":"Image deletion returns 500 if \u0027file\u0027 store is not enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7f74a92338894d01feebf706663b1b5a10d99ece"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-07-26 13:30:06.000000000","tz":-240},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese internal stores are considered \"reserved stores\" by Glance.\nFor the time being, these are hard-coded as filesystem stores.  The\nstore prefix \u0027os_glance_\u0027 is reserved for internal Glance use and\nthe glance-api service will refuse to start if a store with this\nprefix is included in the enabled_backends config option in\nglance-api.conf.\n\nNOTE: Because there are no sensible default values for the location\nof the datadir for each of these stores, the operator must define\n\u0027os_glance_tasks_store\u0027 and \u0027os_glance_staging_store\u0027 in\nglance-api.conf configuration file as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nEach filesystem store must have a unique datadir.\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed9e73ba7c90a4b278b5f0048664bd7002220835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed9e73ba7c90a4b278b5f0048664bd7002220835"}]},"branch":"refs/heads/master"},"1c45567ef0279247502ac7ab8136149c2b1e5a88":{"kind":"REWORK","_number":19,"created":"2019-09-12 04:48:37.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/19"}}},"commit":{"parents":[{"commit":"73fefddd969e75e2f2d82866f4b8b9eeca488770","subject":"Add periodic job to prefetch images into cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73fefddd969e75e2f2d82866f4b8b9eeca488770"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-09-12 04:45:16.000000000","tz":0},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese internal stores are considered \"reserved stores\" by Glance.\nFor the time being, these are hard-coded as filesystem stores.  The\nstore prefix \u0027os_glance_\u0027 is reserved for internal Glance use and\nthe glance-api service will refuse to start if a store with this\nprefix is included in the enabled_backends config option in\nglance-api.conf.\n\nNOTE: Because there are no sensible default values for the location\nof the datadir for each of these stores, the operator must define\n\u0027os_glance_tasks_store\u0027 and \u0027os_glance_staging_store\u0027 in\nglance-api.conf configuration file as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nEach filesystem store must have a unique datadir.\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c45567ef0279247502ac7ab8136149c2b1e5a88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c45567ef0279247502ac7ab8136149c2b1e5a88"}]},"branch":"refs/heads/master"},"a376b33bfb716fd20fe8673dd8d99ee91dd231b8":{"kind":"REWORK","_number":20,"created":"2019-09-13 02:56:45.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/53/611253/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/20"}}},"commit":{"parents":[{"commit":"99f3235acf39a9be534d60d3cbcb338fcf15bf07","subject":"Merge \"Multiple backend support for scrubber\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99f3235acf39a9be534d60d3cbcb338fcf15bf07"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-09-13 02:54:36.000000000","tz":-240},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese internal stores are considered \"reserved stores\" by Glance.\nFor the time being, these are hard-coded as filesystem stores.  The\nstore prefix \u0027os_glance_\u0027 is reserved for internal Glance use and\nthe glance-api service will refuse to start if a store with this\nprefix is included in the enabled_backends config option in\nglance-api.conf.\n\nNOTE: Because there are no sensible default values for the location\nof the datadir for each of these stores, the operator must define\n\u0027os_glance_tasks_store\u0027 and \u0027os_glance_staging_store\u0027 in\nglance-api.conf configuration file as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nEach filesystem store must have a unique datadir.\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a376b33bfb716fd20fe8673dd8d99ee91dd231b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a376b33bfb716fd20fe8673dd8d99ee91dd231b8"}]},"branch":"refs/heads/master"},"dc2144d42b3f861b90fa8fe32bb0063dc81b46b0":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2019-09-16 09:41:51.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/21"}}},"commit":{"parents":[{"commit":"60f9c1d01a19645af0e4a114ec8c2327a8b458ce","subject":"Avoid duplication of registering reserved_store options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/60f9c1d01a19645af0e4a114ec8c2327a8b458ce"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-09-16 09:36:23.000000000","tz":0},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese internal stores are considered \"reserved stores\" by Glance.\nFor the time being, these are hard-coded as filesystem stores.  The\nstore prefix \u0027os_glance_\u0027 is reserved for internal Glance use and\nthe glance-api service will refuse to start if a store with this\nprefix is included in the enabled_backends config option in\nglance-api.conf.\n\nNOTE: Because there are no sensible default values for the location\nof the datadir for each of these stores, the operator must define\n\u0027os_glance_tasks_store\u0027 and \u0027os_glance_staging_store\u0027 in\nglance-api.conf configuration file as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nEach filesystem store must have a unique datadir.\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc2144d42b3f861b90fa8fe32bb0063dc81b46b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc2144d42b3f861b90fa8fe32bb0063dc81b46b0"}]},"branch":"refs/heads/master"},"23e6315ba2c701043e9aac0f308a80343075364d":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2019-09-30 17:36:47.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/53/611253/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/22"}}},"commit":{"parents":[{"commit":"feb8ebd75bfcf1296dfa40e000a4a16eb2082971","subject":"Merge \"Add support for oslo.reports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/feb8ebd75bfcf1296dfa40e000a4a16eb2082971"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-09-30 17:36:27.000000000","tz":-240},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese internal stores are considered \"reserved stores\" by Glance.\nFor the time being, these are hard-coded as filesystem stores.  The\nstore prefix \u0027os_glance_\u0027 is reserved for internal Glance use and\nthe glance-api service will refuse to start if a store with this\nprefix is included in the enabled_backends config option in\nglance-api.conf.\n\nNOTE: Because there are no sensible default values for the location\nof the datadir for each of these stores, the operator must define\n\u0027os_glance_tasks_store\u0027 and \u0027os_glance_staging_store\u0027 in\nglance-api.conf configuration file as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nEach filesystem store must have a unique datadir.\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/23e6315ba2c701043e9aac0f308a80343075364d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/23e6315ba2c701043e9aac0f308a80343075364d"}]},"branch":"refs/heads/master"},"74a7e691f93789141d3418fbced57d83ea564042":{"kind":"REWORK","_number":23,"created":"2019-10-01 07:47:34.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/23"}}},"commit":{"parents":[{"commit":"feb8ebd75bfcf1296dfa40e000a4a16eb2082971","subject":"Merge \"Add support for oslo.reports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/feb8ebd75bfcf1296dfa40e000a4a16eb2082971"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-10-01 07:45:50.000000000","tz":0},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese internal stores are considered \"reserved stores\" by Glance.\nFor the time being, these are hard-coded as filesystem stores.  The\nstore prefix \u0027os_glance_\u0027 is reserved for internal Glance use and\nthe glance-api service will refuse to start if a store with this\nprefix is included in the enabled_backends config option in\nglance-api.conf.\n\nNOTE: Because there are no sensible default values for the location\nof the datadir for each of these stores, the operator must define\n\u0027os_glance_tasks_store\u0027 and \u0027os_glance_staging_store\u0027 in\nglance-api.conf configuration file as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nEach filesystem store must have a unique datadir.\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74a7e691f93789141d3418fbced57d83ea564042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74a7e691f93789141d3418fbced57d83ea564042"}]},"branch":"refs/heads/master"},"6dba83ba3a7c047df282f2edb7217548f43d6251":{"kind":"REWORK","_number":24,"created":"2019-10-01 09:53:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/611253/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/611253/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/611253/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/611253/24"}}},"commit":{"parents":[{"commit":"feb8ebd75bfcf1296dfa40e000a4a16eb2082971","subject":"Merge \"Add support for oslo.reports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/feb8ebd75bfcf1296dfa40e000a4a16eb2082971"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-10-17 07:28:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-10-01 09:53:48.000000000","tz":0},"subject":"Rethinking filesystem access","message":"Rethinking filesystem access\n\nIn Rocky multiple backend support is added as experimental feature. In\norder to take advantage of this feature it is decided to deprecate\nwork_dir and node_staging_uri configuration options\nand reserve two filesystem stores \u0027os_glance_tasks_store\u0027 and\n\u0027os_glance_staging_store\u0027, which can be used to get rid of initializing\nstore via internal functions.\n\nThese internal stores are considered \"reserved stores\" by Glance.\nFor the time being, these are hard-coded as filesystem stores.  The\nstore prefix \u0027os_glance_\u0027 is reserved for internal Glance use and\nthe glance-api service will refuse to start if a store with this\nprefix is included in the enabled_backends config option in\nglance-api.conf.\n\nNOTE: Because there are no sensible default values for the location\nof the datadir for each of these stores, the operator must define\n\u0027os_glance_tasks_store\u0027 and \u0027os_glance_staging_store\u0027 in\nglance-api.conf configuration file as shown below.\n\n[os_glance_tasks_store]\nfilesystem_store_datadir \u003d /var/lib/glance/tasks_work_dir/\n\n[os_glance_staging_store]\nfilesystem_store_datadir \u003d /var/lib/glance/staging/\n\nEach filesystem store must have a unique datadir.\n\nDepends-On: https://review.openstack.org/#/c/639765/\nImplements: blueprint rethinking-filesystem-access\nChange-Id: I86ec513c5fc653dbb97b79d953d8430f014e684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6dba83ba3a7c047df282f2edb7217548f43d6251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6dba83ba3a7c047df282f2edb7217548f43d6251"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
