)]}'
{"id":"openstack%2Fglance~617057","triplet_id":"openstack%2Fglance~master~I82333377692f0f8d9076ef12e67ea73da60e9fed","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"27366":{"account":{"_account_id":27366,"name":"guotao","email":"guotao.bj@inspur.com","username":"taoguo"},"last_update":"2022-07-07 15:42:31.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-07-07 15:42:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I82333377692f0f8d9076ef12e67ea73da60e9fed","subject":"Doc: fix URL in API call","status":"MERGED","created":"2018-11-10 01:57:24.000000000","updated":"2022-07-07 15:44:49.000000000","submitted":"2022-07-07 15:42:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"617057","meta_rev_id":"0b359e0031d156d393e977bf0c1e8d250a9845fd","_number":617057,"virtual_id_number":617057,"owner":{"_account_id":27366,"name":"guotao","email":"guotao.bj@inspur.com","username":"taoguo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-07 15:42:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"value":2,"date":"2022-07-07 14:19:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"value":1,"date":"2022-07-07 14:19:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-10 02:07:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-11-11 11:36:36.000000000","updated_by":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"reviewer":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"state":"REVIEWER"},{"updated":"2022-07-05 20:17:55.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2022-07-07 14:19:08.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"b00728820dd74dda7667c018fc25adf6b9712b25","author":{"_account_id":27366,"name":"guotao","email":"guotao.bj@inspur.com","username":"taoguo"},"date":"2018-11-10 01:57:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27bf451697ee1c5626bf6e48bb7e1cd4829e5851","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-10 02:07:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/57/617057/1/check/openstack-tox-pep8/ac158f7/ : SUCCESS in 4m 15s\n- openstack-tox-docs http://logs.openstack.org/57/617057/1/check/openstack-tox-docs/a02f621/html/ : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"45b1a0672121a39c4acd9e9f8840c708b5ae3635","author":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"date":"2018-11-11 11:36:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d548a389d9877b16e11ef4556e66ec587467d2eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-05 20:17:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"62d94f97ad843678efc88edeccb9d6a49752e917","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-05 20:17:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aee5f00ad9216881cae00aa0a2d02c40dfd4d433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 20:29:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10294aeb24c94548ab8846e7b95fda50\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf6aa675ec7a4f2883a7744cded2576b : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc7196ca93345d5816005dba8d64cef : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":2},{"id":"ab5e266fb514e55bc4495f5326dfd1cdcd05bd3c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-07 14:19:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44ee3a5bcd7dcaabbbe8d775f6c09b0abe335a2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 14:19:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a025d2855b71fabc681a706b9b45efde6287cf65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:42:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c7bc8d4497a405f8d795aa17a610a8b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f6f31c676dd4a8cbb42bcaa0258252f : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0153c876dc694240a11d3bcec9e9fe7b : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":2},{"id":"60067339ae3a1231f98502b6d5087d1bab159882","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:42:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0b359e0031d156d393e977bf0c1e8d250a9845fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:44:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b704d18e3c5c4d10ba48d3d43f3c9930\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e23eacea2e0544938fda81048482dabe : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"675cdb92edb6701961efb718d282b9ffe605f08f","revisions":{"01ead2120e89cb4c017e6bfd7d2dcafa8431bf97":{"kind":"REWORK","_number":1,"created":"2018-11-10 01:57:24.000000000","uploader":{"_account_id":27366,"name":"guotao","email":"guotao.bj@inspur.com","username":"taoguo"},"ref":"refs/changes/57/617057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/617057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/617057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/617057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/617057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/617057/1"}}},"commit":{"parents":[{"commit":"97dac0f3800ad9768e5c118656c6bbbf55fc866d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/97dac0f3800ad9768e5c118656c6bbbf55fc866d"}]}],"author":{"name":"taoguo","email":"guotao.bj@inspur.com","date":"2018-11-10 01:51:21.000000000","tz":480},"committer":{"name":"taoguo","email":"guotao.bj@inspur.com","date":"2018-11-10 01:57:01.000000000","tz":480},"subject":"Update the example link","message":"Update the example link\n\nFix the wrong link\n\nChange-Id: I82333377692f0f8d9076ef12e67ea73da60e9fed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/01ead2120e89cb4c017e6bfd7d2dcafa8431bf97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/01ead2120e89cb4c017e6bfd7d2dcafa8431bf97"}]},"branch":"refs/heads/master"},"675cdb92edb6701961efb718d282b9ffe605f08f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2022-07-05 20:17:24.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/57/617057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/617057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/617057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/617057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/617057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/617057/2"}}},"commit":{"parents":[{"commit":"2866713392de2a122145a186a7ed2391e208b4ba","subject":"Merge \"glance-cache-prefetcher: Set up access to multistore backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2866713392de2a122145a186a7ed2391e208b4ba"}]}],"author":{"name":"taoguo","email":"guotao.bj@inspur.com","date":"2018-11-10 01:51:21.000000000","tz":480},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-05 20:16:34.000000000","tz":120},"subject":"Doc: fix URL in API call","message":"Doc: fix URL in API call\n\nChange-Id: I82333377692f0f8d9076ef12e67ea73da60e9fed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/675cdb92edb6701961efb718d282b9ffe605f08f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/675cdb92edb6701961efb718d282b9ffe605f08f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
