)]}'
{"id":"openstack%2Fglance~625702","triplet_id":"openstack%2Fglance~master~I5bf0c8af9dfe87e0d17f7a16d4676f387b4379f6","project":"openstack/glance","branch":"master","topic":"bug/1808375","hashtags":[],"change_id":"I5bf0c8af9dfe87e0d17f7a16d4676f387b4379f6","subject":"Update show_multiple_locations deprecation note","status":"MERGED","created":"2018-12-17 20:01:42.000000000","updated":"2018-12-20 17:03:40.000000000","submitted":"2018-12-20 17:03:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"625702-1545325420316-aa5b6d22","meta_rev_id":"2d4d685b8f66a40311372ba77e09fdfd86d5a191","_number":625702,"virtual_id_number":625702,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-12-18 14:47:15.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2018-12-20 17:03:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2018-12-18 17:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2018-12-20 15:05:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2018-12-20 14:40:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2018-12-20 15:05:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-18 14:47:15.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-12-18 17:10:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2018-12-20 14:40:19.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2018-12-20 15:05:19.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2018-12-20 17:03:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6a363711e3dc03fd555d0a44328e63238bd46a6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2018-12-17 20:01:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48bbf59a48a50964e350370e42993db973616e55","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-17 20:02:02.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"0449c571b895d42044a384c0aadbc37b7faae76a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-12-17 20:11:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nGreat details - good idea to have this clearly explained in the release notes.","accounts_in_message":[],"_revision_number":1},{"id":"032e71bfb7b5b8f27d9501a36132297cac74bbad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-17 21:44:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/02/625702/1/check/tempest-full/28201b8/ : SUCCESS in 1h 14m 29s\n- neutron-grenade http://logs.openstack.org/02/625702/1/check/neutron-grenade/8671704/ : SUCCESS in 1h 15m 31s\n- grenade-py3 http://logs.openstack.org/02/625702/1/check/grenade-py3/ed18639/ : SUCCESS in 1h 06m 41s\n- tempest-full-py3 http://logs.openstack.org/02/625702/1/check/tempest-full-py3/ef624bb/ : SUCCESS in 1h 14m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/625702/1/check/openstack-tox-lower-constraints/27a53b9/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/02/625702/1/check/openstack-tox-pep8/4d02a9c/ : SUCCESS in 4m 34s\n- openstack-tox-py27 http://logs.openstack.org/02/625702/1/check/openstack-tox-py27/11d27b3/ : SUCCESS in 4m 28s\n- openstack-tox-py35 http://logs.openstack.org/02/625702/1/check/openstack-tox-py35/a86eaa4/ : SUCCESS in 4m 44s\n- openstack-tox-py36 http://logs.openstack.org/02/625702/1/check/openstack-tox-py36/b94649c/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/02/625702/1/check/openstack-tox-docs/045379c/html/ : SUCCESS in 5m 24s\n- build-openstack-releasenotes http://logs.openstack.org/02/625702/1/check/build-openstack-releasenotes/ff5a9ac/html/ : SUCCESS in 7m 16s\n- openstack-tox-functional http://logs.openstack.org/02/625702/1/check/openstack-tox-functional/587c3ae/ : SUCCESS in 7m 38s\n- openstack-tox-functional-py35 http://logs.openstack.org/02/625702/1/check/openstack-tox-functional-py35/3ca1942/ : SUCCESS in 8m 11s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/02/625702/1/check/glance-eventlet-ssl-handshake-broken-py35/a062ceb/ : FAILURE in 4m 09s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/02/625702/1/check/devstack-plugin-ceph-tempest/4ce56db/ : SUCCESS in 1h 36m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bb38e1d356316e7fcef2ad0331af438f9fa6716","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2018-12-18 14:01:29.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nNew patch coming up to address Sean\u0027s point.","accounts_in_message":[],"_revision_number":1},{"id":"39e1ac5eafb89929400e7a8b3bcc53c183c3f8f4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2018-12-18 14:05:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"108fc5bfdb69fecacfa6eb6dfba9a3af3c2532e1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-18 14:47:15.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/02/625702/2/check-not-vote/ext-glance-zuul/b26e9d4 : SUCCESS in 41m 09s","accounts_in_message":[],"_revision_number":2},{"id":"d71ddca5743956720b5b226bfeb0a1bd30e7bcbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-18 17:06:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/02/625702/2/check/tempest-full/b7bf3b8/ : SUCCESS in 1h 28m 45s\n- neutron-grenade http://logs.openstack.org/02/625702/2/check/neutron-grenade/9a5d31f/ : SUCCESS in 54m 03s\n- grenade-py3 http://logs.openstack.org/02/625702/2/check/grenade-py3/eb2a14d/ : SUCCESS in 54m 31s\n- tempest-full-py3 http://logs.openstack.org/02/625702/2/check/tempest-full-py3/8e6361d/ : SUCCESS in 1h 35m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/625702/2/check/openstack-tox-lower-constraints/da2c4fc/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/02/625702/2/check/openstack-tox-pep8/a403e7d/ : SUCCESS in 5m 16s\n- openstack-tox-py27 http://logs.openstack.org/02/625702/2/check/openstack-tox-py27/c0c6128/ : SUCCESS in 4m 18s\n- openstack-tox-py35 http://logs.openstack.org/02/625702/2/check/openstack-tox-py35/f866bf6/ : SUCCESS in 4m 04s\n- openstack-tox-py36 http://logs.openstack.org/02/625702/2/check/openstack-tox-py36/4d34360/ : SUCCESS in 3m 28s\n- openstack-tox-docs http://logs.openstack.org/02/625702/2/check/openstack-tox-docs/244abd3/html/ : SUCCESS in 4m 43s\n- build-openstack-releasenotes http://logs.openstack.org/02/625702/2/check/build-openstack-releasenotes/62dd7f1/html/ : SUCCESS in 8m 08s\n- openstack-tox-functional http://logs.openstack.org/02/625702/2/check/openstack-tox-functional/b73eedc/ : SUCCESS in 8m 06s\n- openstack-tox-functional-py35 http://logs.openstack.org/02/625702/2/check/openstack-tox-functional-py35/ab25a5d/ : SUCCESS in 8m 04s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/02/625702/2/check/glance-eventlet-ssl-handshake-broken-py35/c36b61f/ : FAILURE in 3m 55s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/02/625702/2/check/devstack-plugin-ceph-tempest/aaaefd8/ : SUCCESS in 1h 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ed869a6b2046bce76f878e6bf6b9983d451abef","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-12-18 17:10:02.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"ed9d4a088eaebe74552fb4a6f633b10f4436624a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2018-12-20 14:40:19.000000000","message":"Patch Set 2: Code-Review+2\n\nLet jokke have a look as well,\n\nThank you rosmaita!!!","accounts_in_message":[],"_revision_number":2},{"id":"c045569acffcc692530b4d47e512149154b85a61","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2018-12-20 15:05:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThis looks great, thanks Brian!","accounts_in_message":[],"_revision_number":2},{"id":"f1a644a3a36e7d9c1acc259288cc41719fa4a856","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 15:07:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0cb335ff424e92572029fda82db0dca70d344609","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 17:03:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3274db5c931ab717dfb58c289de6e4b3dc6d6656","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 17:03:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/02/625702/2/gate/tempest-full/577758a/ : SUCCESS in 1h 31m 35s\n- neutron-grenade http://logs.openstack.org/02/625702/2/gate/neutron-grenade/be43e58/ : SUCCESS in 55m 09s\n- grenade-py3 http://logs.openstack.org/02/625702/2/gate/grenade-py3/bc2b88f/ : SUCCESS in 55m 02s\n- tempest-full-py3 http://logs.openstack.org/02/625702/2/gate/tempest-full-py3/f6c78f3/ : SUCCESS in 1h 14m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/625702/2/gate/openstack-tox-lower-constraints/c6dcb10/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/02/625702/2/gate/openstack-tox-pep8/6cafd5e/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/02/625702/2/gate/openstack-tox-py27/3e3dd62/ : SUCCESS in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/02/625702/2/gate/openstack-tox-py35/08f2096/ : SUCCESS in 5m 04s\n- openstack-tox-py36 http://logs.openstack.org/02/625702/2/gate/openstack-tox-py36/0cdc3ce/ : SUCCESS in 3m 47s\n- openstack-tox-docs http://logs.openstack.org/02/625702/2/gate/openstack-tox-docs/6686d99/html/ : SUCCESS in 5m 12s\n- build-openstack-releasenotes http://logs.openstack.org/02/625702/2/gate/build-openstack-releasenotes/a9e6974/html/ : SUCCESS in 7m 47s\n- openstack-tox-functional http://logs.openstack.org/02/625702/2/gate/openstack-tox-functional/e68d1d5/ : SUCCESS in 7m 53s\n- openstack-tox-functional-py35 http://logs.openstack.org/02/625702/2/gate/openstack-tox-functional-py35/3c976e8/ : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de72ac4a01be1ccc75825eb941dad64c170d6bea","revisions":{"f867a3da422b9a5d460a9dc705ad6be14ebf455e":{"kind":"REWORK","_number":1,"created":"2018-12-17 20:01:42.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/02/625702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/625702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/625702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/625702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/625702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/625702/1"}}},"commit":{"parents":[{"commit":"8d1beed2a890ddba174b226e7e117068caaf9d56","subject":"Merge \"Implement scaffolding for upgrade checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d1beed2a890ddba174b226e7e117068caaf9d56"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2018-12-14 14:38:07.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2018-12-17 19:55:18.000000000","tz":-300},"subject":"Update show_multiple_locations deprecation note","message":"Update show_multiple_locations deprecation note\n\nUpdate logged text to indicate our intentions with respect\nto the show_multiple_locations configuration option, which\nhas been deprecated since Newton and advertised as subject\nto removal in Pike.  Also correct the misleading impression\nin the deprecation notice that functionality that requires\nshow_multiple_locations \u003d True can be configured via policies\nat the current time.\n\nChange-Id: I5bf0c8af9dfe87e0d17f7a16d4676f387b4379f6\nCloses-bug: #1808375\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f867a3da422b9a5d460a9dc705ad6be14ebf455e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f867a3da422b9a5d460a9dc705ad6be14ebf455e"}]},"branch":"refs/heads/master"},"de72ac4a01be1ccc75825eb941dad64c170d6bea":{"kind":"REWORK","_number":2,"created":"2018-12-18 14:05:39.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/02/625702/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/625702/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/625702/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/625702/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/625702/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/625702/2"}}},"commit":{"parents":[{"commit":"8d1beed2a890ddba174b226e7e117068caaf9d56","subject":"Merge \"Implement scaffolding for upgrade checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d1beed2a890ddba174b226e7e117068caaf9d56"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2018-12-14 14:38:07.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2018-12-18 14:04:21.000000000","tz":-300},"subject":"Update show_multiple_locations deprecation note","message":"Update show_multiple_locations deprecation note\n\nUpdate logged text to indicate our intentions with respect\nto the show_multiple_locations configuration option, which\nhas been deprecated since Newton and advertised as subject\nto removal in Pike.  Also correct the misleading impression\nin the deprecation notice and earlier release notes that\nfunctionality that requires show_multiple_locations \u003d True\ncan be configured via policies at the current time.  Also\nadds a release note to this effect.\n\nChange-Id: I5bf0c8af9dfe87e0d17f7a16d4676f387b4379f6\nCloses-bug: #1808375\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de72ac4a01be1ccc75825eb941dad64c170d6bea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de72ac4a01be1ccc75825eb941dad64c170d6bea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
