)]}'
{"id":"openstack%2Fglance~626008","triplet_id":"openstack%2Fglance~master~I7a191ac44b004f840c986ded61341e43b35f1d49","project":"openstack/glance","branch":"master","topic":"correct-tox-lc","hashtags":[],"change_id":"I7a191ac44b004f840c986ded61341e43b35f1d49","subject":"Add tests to lower-constraints job","status":"MERGED","created":"2018-12-18 21:01:53.000000000","updated":"2020-05-14 19:17:54.000000000","submitted":"2020-05-14 19:16:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"626008-1589483778792-d9d82c02","meta_rev_id":"508ba1127d0d5c66576bcb4048ec48c283488fd0","_number":626008,"virtual_id_number":626008,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-13 13:36:16.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-14 19:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-13 16:19:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-05-14 19:04:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-05-14 19:04:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 13:36:16.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-13 16:19:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-14 19:04:09.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-05-14 19:16:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78f4045713d6ce99d55490c85341120c4abb2bf2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2018-12-18 21:01:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21dcd069739a18f0b4ceac43e53d8683ea130c7d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2018-12-18 21:04:57.000000000","message":"Patch Set 1:\n\nIt looks like this makes a difference:\n\nbefore: http://paste.openstack.org/show/737628/\n\nafter: http://paste.openstack.org/show/737631/","accounts_in_message":[],"_revision_number":1},{"id":"d437cb54bc648c210fdcdf27c358b5a028fcf522","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-18 21:32:19.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/08/626008/1/check-not-vote/ext-glance-zuul/7538725 : FAILURE in 30m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4f2dc947926b830afef9f76c8cf1c26d5e23ef90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-18 23:07:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/08/626008/1/check/grenade-py3/0fc0300/ : SUCCESS in 1h 03m 19s\n- tempest-full-py3 http://logs.openstack.org/08/626008/1/check/tempest-full-py3/fcbf731/ : SUCCESS in 1h 16m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/626008/1/check/openstack-tox-lower-constraints/768069f/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/08/626008/1/check/openstack-tox-pep8/4bc3f00/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/08/626008/1/check/openstack-tox-py27/0117c33/ : SUCCESS in 4m 11s\n- openstack-tox-py35 http://logs.openstack.org/08/626008/1/check/openstack-tox-py35/93ef787/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/08/626008/1/check/openstack-tox-py36/e1922cd/ : SUCCESS in 3m 15s\n- openstack-tox-docs http://logs.openstack.org/08/626008/1/check/openstack-tox-docs/d66786f/html/ : SUCCESS in 4m 44s\n- build-openstack-releasenotes http://logs.openstack.org/08/626008/1/check/build-openstack-releasenotes/81cf82d/html/ : SUCCESS in 8m 26s\n- openstack-tox-functional http://logs.openstack.org/08/626008/1/check/openstack-tox-functional/503a826/ : SUCCESS in 8m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/08/626008/1/check/openstack-tox-functional-py35/0685fa5/ : SUCCESS in 11m 17s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/08/626008/1/check/glance-eventlet-ssl-handshake-broken-py35/49e530f/ : FAILURE in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2fedce218012e612768bb908aa7656df8d7a551c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-07 12:11:41.000000000","message":"Patch Set 1:\n\nIs this still needed Brian?","accounts_in_message":[],"_revision_number":1},{"id":"e48fda325b3ada71f9a16aa6be6170e0090389ac","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-26 15:17:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7bbe2917011a5e665e91fa2741132e81235bd5cc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-11-26 15:48:53.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/08/626008/2/check-not-vote/ext-glance-zuul/b4b925b : FAILURE in 30m 35s","accounts_in_message":[],"_revision_number":2},{"id":"860220d061e8517bec23c197362b69bd015c8807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 16:18:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5f738c491d6f4fbe99e386a9e42d5f0f : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/658a3b52330a4f2cb22c3c5d893adcc9 : FAILURE in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f5f6dab128f409ebbfa8ef6b44db615 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9315248ea1d4e2eac467521b9f17c4e : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d95099e47504538b7c87c6d145623b4 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac40f57d1ad4f679bcb8ca3a733e5e1 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e544e70b6b8436bb1f544dbe2645e33 : SUCCESS in 10m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5deef3b23db54b28b681d10d14a52938 : SUCCESS in 11m 07s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a08a9073227647d2a1bd1842ea163c60 : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":2},{"id":"88cb4f5063d907548a6bd346db8f5202136b61f4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-26 18:04:48.000000000","message":"Patch Set 2:\n\nFailure was the subunit parser problem during functional tests.","accounts_in_message":[],"_revision_number":2},{"id":"301c247ab3ffbf2cc7f1250384f5f52f47685d16","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-26 18:04:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"500ea956e7c240df645db5f84fc2d2071218de2d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-11-26 18:34:20.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/08/626008/2/check-not-vote/ext-glance-zuul/34d2f0a : FAILURE in 28m 55s","accounts_in_message":[],"_revision_number":2},{"id":"f9c13eaeeb180e3040b0cb4b92fd374a80e5005d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 19:04:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/938c486f69614208ae9603f7567c4094 : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49c0a4c9d3d14b3ca02a03351b616b6a : FAILURE in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eecb68794e464f1aa6a169edeb38fc92 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a94b717e42c44a71a629bfc9bb82ae03 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad827bbb15014baeac6bd214fdb146e6 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/143a5c2dfb3645cca0728f44f74e98fb : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2d0e867061f4a54a83f218619290b2d : SUCCESS in 9m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d65607ed40a6428a9bbb4944d58c64f2 : SUCCESS in 8m 10s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/fdf9ffd6b7ca453096cda534d5ccf7b9 : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4f1d70971eccca3fefdfa78393b851ccdd260f99","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-02 20:42:08.000000000","message":"Patch Set 2:\n\nsubunit parser problem again","accounts_in_message":[],"_revision_number":2},{"id":"4f6397c617ff617b3a32700144c8b7849523c1b0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-02 20:42:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f2f3762b522861494e5222a008da005d159cd687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 20:58:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/451a6eca909d455ea7578a1794dc6f7a : SUCCESS in 6m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41447abd42ae4d3f829e7b623c67f14f : FAILURE in 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dbdeac9da894138a525ae33e38f60f9 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46ed75f678554b9e8ea53208afb1b395 : FAILURE in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c525118845b421b96eb37c73b6e4266 : FAILURE in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5dc586d4a7d4868b09c40a9e1bf8e3f : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2eb3019153464031ac2a959d6bd9e80a : SUCCESS in 8m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b5e2d2804f9c4dfca5d824ad47037419 : SUCCESS in 8m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f15809d83b654fa39ea7dbceac88f849 : SUCCESS in 9m 16s","accounts_in_message":[],"_revision_number":2},{"id":"97c8fb992d935712e723dc2566a64bf40bb98847","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-13 21:01:26.000000000","message":"Patch Set 2:\n\nSubunit parser problem again.","accounts_in_message":[],"_revision_number":2},{"id":"9b72015de9778b0758f6cb1d7361e9c867e56609","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-13 21:01:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3f6e3b3207ef03d7034478a49b745e1f370b49bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 21:01:49.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"0a0ac28973d637611432e5ba3456fe82034b91ee","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-13 21:09:44.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"cae2d20b14476699c7492d7ecf2d9fa3b92da265","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-17 21:31:06.000000000","message":"Patch Set 2:\n\nI\u0027m going to wait to rebase \u0026 resolve merge conflicts until after https://review.opendev.org/#/c/695707/ has merged.","accounts_in_message":[],"_revision_number":2},{"id":"d2f28a9dc3336c8115da7c72be53fb4a03bded83","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-18 19:38:24.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"3c05bbf5dee80e0f2b51506004a37ce279f525bd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-18 19:40:08.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/08/626008/3/check-not-vote/ext-glance-zuul/8336cd3 : FAILURE in 1m 24s","accounts_in_message":[],"_revision_number":3},{"id":"84d820cbb1e56f9c021f26c7045385147bed4e55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 19:50:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/03f39a4fa6b6430ab869fc801873df8b : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a875e129a534e858b2d430e4ecdd9a3 : FAILURE in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be4aca6317ad4eea8e8998ed15288f44 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ef053865d24431591f5d8e76a660863 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ee7c66a4a3d425286d437979d949530 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac792e539874ae3a3ae8611a132079f : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d96a48370d5e40daa3bf6f09de7d7356 : SUCCESS in 9m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/afbf2b539d5f4cd19d41f1e6b3f11cc0 : SUCCESS in 10m 25s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3cec6145da434a168486191ca06a6c9b : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":3},{"id":"bb0297b4a41b0f177dd69f127e3b401236bb1a04","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-18 19:55:07.000000000","message":"Patch Set 3:\n\nLooks like this may have actually uncovered some issues in what we have listed as our lower constraints. Can\u0027t track things down right now, but maybe need to at least raise the minimums for some oslo libraries?","accounts_in_message":[],"_revision_number":3},{"id":"cc32686bd25ffbc14fa86f2d87ad19d69c15fd9a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-02-19 13:37:19.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"9481e731c17945a39405ce7f20c959fc845b11d0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-02-19 13:39:04.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/08/626008/4/check-not-vote/ext-glance-zuul/e5c2e59 : FAILURE in 1m 25s","accounts_in_message":[],"_revision_number":4},{"id":"259ffb0db3bafddb3c79bd09d5c8946b05008ca3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-19 13:50:32.000000000","message":"Patch Set 4: Code-Review+2\n\nThis surprises me, since I would have assumed installing requirements.txt ahead of time would have made the development install of the project not need to install anything else. Are we missing something?\n\nAnyway, doesn\u0027t really matter as this seems like a good way to handle this job regardless.","accounts_in_message":[],"_revision_number":4},{"id":"3dceebae4f9f0b620f070cfd0626a7a006ef5676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 16:13:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/88dfc0abb40d4271a84f54ea1b849134 : SUCCESS in 6m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81c4eab055184dd39172ffca46d03c1c : FAILURE in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af157d8659b64aeb8dab6fb671ddccf2 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8f4c3f1627b4eb7872d9c96b6e977bb : SUCCESS in 9m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85dd3ac0c4ac46c8aba9b69846d2e663 : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e8a7ffdb770475ea9989bbb316db27a : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5537f5bdd024dddb05f9c3e7ae8139b : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c366fdd71e4482b8bc8cc3a4e92f0f8 : SUCCESS in 10m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7dadefd190ac4016b5a49c4c72917c8d : SUCCESS in 10m 27s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/5be889cfa10f4f43b88c13030f978521 : SUCCESS in 10m 57s","accounts_in_message":[],"_revision_number":4},{"id":"8b0434981b435ba9393053d1f183d6e1751fd69f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-19 16:32:26.000000000","message":"Patch Set 4: -Code-Review\n\nLooks like this may be uncovering an issue?","accounts_in_message":[],"_revision_number":4},{"id":"8af6b4739217d345afaa45cb64c4204c9a72e34f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-04-14 16:11:51.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"6eac4b40578dc5acd4211fed8585ed9495b36e6b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-14 16:12:11.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"81f1f907b6d11de054468d967554a2838043aab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 22:23:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d38f754f713b4148a6e0af3829c27178 : SUCCESS in 6m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec3f49b87d3e4282827ff409bf683441 : FAILURE in 16m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceae18f24c3f4ef5b8ebc4eed490ba80 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/754c1dca101f4be1aa3921f2b26fe140 : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0408d286ecae4d24884b4ab96c495c73 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dc2c2228a62419aaae0e27c877b6012 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3af5e439ed74529b7bef7a6867d9026 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebcdaf8b07d6426dabb5d30be71b47fa : SUCCESS in 10m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cdef6e66ca5246e18d00adaf9f81df1d : SUCCESS in 7m 33s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ab7788353d974e91a31d7646eafc382d : FAILURE in 8m 15s","accounts_in_message":[],"_revision_number":5},{"id":"7fb5912cd5419d83a60ae31341059ea855c0c42b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-05-07 14:59:14.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"aba3a232fa20e6fd98f086999b89d9644d3b7ed4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-05-07 14:59:37.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"94135fa049b6e85dfa4315256b571ce6b26e14a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:42:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0aac414655664d2fa2655b2dfe29c166 : SUCCESS in 10m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27e5c2eafbe04f6a9ca61964687966a8 : FAILURE in 19m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2137f196604451ea4e6ac0e91e547f1 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f19e0a8bab2849a7a26f2b67b08de7f0 : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64b2a8f2180d491d8a51c0f335ea3390 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43bac0190dae4b5ba6f79c69970dffd4 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8010536d820b43ca949bf00bec7cf899 : SUCCESS in 10m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d95d9be45666477f96402abda00f52be : SUCCESS in 7m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/40af6bb108fe4dbea60ebca60a19274d : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":6},{"id":"8de1344ac6db4b1d77523ae731dac88efd97b4af","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-07 15:52:05.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a9c2951100acd5999aebd5e3d79c3b12046771ae","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-05-13 13:35:40.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"3fef4ffe975d00070fa0da2a9bde37ce69583793","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-05-13 13:35:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f00c65216e7fbab3b97b4875f31f9cd640da0ac5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-05-13 13:36:16.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"5877d9330c8c24323398264dc333e0d09282fc5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 16:14:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aff4d86f403c4372945106974116e8be : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b5f82b413e34570b6de1560a2432c09 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6b1c600abff42e3b5c3fb04a9166d72 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a885bbc3173d40dcb1e4b794eb9e2c6c : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74e314c8c15a48969123283dfe427eec : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e012f11083048c1a4ff3069563cf4a8 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8aebc68dcb64c3fb8e327304cdafb91 : SUCCESS in 9m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/66fb68b2a2a64f17896785be510161ac : SUCCESS in 7m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f63c1e52ae8847b5a3af5719dcd464a9 : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":7},{"id":"7db1333126df6b962560a67b7b7549c15521257f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-13 16:19:28.000000000","message":"Patch Set 7: Code-Review+2\n\nLooks like that works.","accounts_in_message":[],"_revision_number":7},{"id":"8f8cef543ada0fcbf52e76d8354d0a4971250558","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-05-14 19:04:09.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nThank you Brian!!!","accounts_in_message":[],"_revision_number":7},{"id":"f6e87eed4d8a8f897061ed219a938861595abd21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:04:22.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"81376127d53f329c9be2ac82463902953629bae2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:16:17.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4dcdbc56cfde4e318e58ee7d4cf8e71b : SUCCESS in 5m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e7c3b56a5a7435285944a5e0322bcdc : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5372c7b1c164dd0988f219dfb1d1db7 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22d165d4befc4007ac4c3643e6fe1a8e : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83483b6b5ba347af91413ce1054bf535 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/857b193bda7d49a9bc52901a6e511405 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8eeebb149434fceb01d8b1c9b546c08 : SUCCESS in 10m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d174ce29d07d499aa4f52c31eb1f9f30 : SUCCESS in 8m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e7ef6576ae4e45c798aa7b789a2dc70e : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":7},{"id":"d9d7f32531016c0f980bb6d0438d1fd18f3e8675","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:16:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"aca6e63e1ba0639fe54d1e5b00395f2361aeebc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:17:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6a0d207dbdd14090ab4cd87e7f76d7ec : SUCCESS in 1m 08s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fafa18348434fddbf19a4b398f02cd6 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8205297f105349cca7a0baf33a0e1188 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"890b740adc761a2a38657152a726b9450f172b97","revisions":{"4769beaa2c1176766637c9303196683633575fe9":{"kind":"REWORK","_number":1,"created":"2018-12-18 21:01:53.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/08/626008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/626008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/626008/1"}}},"commit":{"parents":[{"commit":"3d52684346f3a7eb56b7d8836b4e0e8efafee647","subject":"Merge \"Fix for FK constraint violation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d52684346f3a7eb56b7d8836b4e0e8efafee647"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2018-12-17 19:23:16.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2018-12-18 21:01:28.000000000","tz":-300},"subject":"WIP: reconfigure lower-constraints job","message":"WIP: reconfigure lower-constraints job\n\nChange-Id: I7a191ac44b004f840c986ded61341e43b35f1d49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4769beaa2c1176766637c9303196683633575fe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4769beaa2c1176766637c9303196683633575fe9"}]},"branch":"refs/heads/master"},"459cabf6e0ed6c63612d3a331022e8529b776ae4":{"kind":"REWORK","_number":2,"created":"2019-11-26 15:17:52.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/08/626008/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/626008/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/626008/2"}}},"commit":{"parents":[{"commit":"3039e71023bca0ce41f7078b158f8f669e89f826","subject":"Remove py2 testing, jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3039e71023bca0ce41f7078b158f8f669e89f826"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2018-12-17 19:23:16.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-26 15:16:06.000000000","tz":-300},"subject":"Reconfigure lower-constraints job","message":"Reconfigure lower-constraints job\n\nChange to usedevelop\u003dFalse (following the discussion on\nChange Id66a28f7ace6fc2adf0e1201d9de5f901234d870) and run\nsome tests to actually check that the code runs using the\ncurrent lower-constraints.txt file.\n\nChange-Id: I7a191ac44b004f840c986ded61341e43b35f1d49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/459cabf6e0ed6c63612d3a331022e8529b776ae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/459cabf6e0ed6c63612d3a331022e8529b776ae4"}]},"branch":"refs/heads/master"},"c217593e6814aff9b44eade95c3c62d4c3c657ab":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-18 19:38:24.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/08/626008/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/626008/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/626008/3"}}},"commit":{"parents":[{"commit":"40bd312410d6ff23ea5d8b6d6263f40bd456fffd","subject":"Remove py2 testing, jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40bd312410d6ff23ea5d8b6d6263f40bd456fffd"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2018-12-17 19:23:16.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-12-18 19:33:37.000000000","tz":-300},"subject":"Reconfigure lower-constraints job","message":"Reconfigure lower-constraints job\n\nChange to usedevelop\u003dFalse (following the discussion on\nChange Id66a28f7ace6fc2adf0e1201d9de5f901234d870) and run\nsome tests to actually check that the code runs using the\ncurrent lower-constraints.txt file.\n\nChange-Id: I7a191ac44b004f840c986ded61341e43b35f1d49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c217593e6814aff9b44eade95c3c62d4c3c657ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c217593e6814aff9b44eade95c3c62d4c3c657ab"}]},"branch":"refs/heads/master"},"341c827c01fc08f1b009dc9e43aec1f281805d2e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-02-19 13:37:19.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/08/626008/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/626008/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/626008/4"}}},"commit":{"parents":[{"commit":"0035ab253819246491f11ffeecdab99e78d75629","subject":"Revert \"Remove all example configs for deprecated registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0035ab253819246491f11ffeecdab99e78d75629"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2018-12-17 19:23:16.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-02-19 13:37:19.000000000","tz":0},"subject":"Reconfigure lower-constraints job","message":"Reconfigure lower-constraints job\n\nChange to usedevelop\u003dFalse (following the discussion on\nChange Id66a28f7ace6fc2adf0e1201d9de5f901234d870) and run\nsome tests to actually check that the code runs using the\ncurrent lower-constraints.txt file.\n\nChange-Id: I7a191ac44b004f840c986ded61341e43b35f1d49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/341c827c01fc08f1b009dc9e43aec1f281805d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/341c827c01fc08f1b009dc9e43aec1f281805d2e"}]},"branch":"refs/heads/master"},"6a1259383d322b1f2d2e4042637ac781d9e6bf6a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-04-14 16:11:51.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/08/626008/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/626008/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/626008/5"}}},"commit":{"parents":[{"commit":"fa18f745df3a5200637e661fd04f751b5e8bb595","subject":"Merge \"Revise admin interoperable image import docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa18f745df3a5200637e661fd04f751b5e8bb595"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2018-12-17 19:23:16.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-04-14 16:11:51.000000000","tz":0},"subject":"Reconfigure lower-constraints job","message":"Reconfigure lower-constraints job\n\nChange to usedevelop\u003dFalse (following the discussion on\nChange Id66a28f7ace6fc2adf0e1201d9de5f901234d870) and run\nsome tests to actually check that the code runs using the\ncurrent lower-constraints.txt file.\n\nChange-Id: I7a191ac44b004f840c986ded61341e43b35f1d49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a1259383d322b1f2d2e4042637ac781d9e6bf6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a1259383d322b1f2d2e4042637ac781d9e6bf6a"}]},"branch":"refs/heads/master"},"436e22503b394317d80da48a4b5e15ca0bb64036":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-05-07 14:59:14.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/08/626008/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/626008/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/626008/6"}}},"commit":{"parents":[{"commit":"3e2e51a5d001ad5c5d62739b35580c8c11ab4282","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e2e51a5d001ad5c5d62739b35580c8c11ab4282"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2018-12-17 19:23:16.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-05-07 14:59:14.000000000","tz":0},"subject":"Reconfigure lower-constraints job","message":"Reconfigure lower-constraints job\n\nChange to usedevelop\u003dFalse (following the discussion on\nChange Id66a28f7ace6fc2adf0e1201d9de5f901234d870) and run\nsome tests to actually check that the code runs using the\ncurrent lower-constraints.txt file.\n\nChange-Id: I7a191ac44b004f840c986ded61341e43b35f1d49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/436e22503b394317d80da48a4b5e15ca0bb64036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/436e22503b394317d80da48a4b5e15ca0bb64036"}]},"branch":"refs/heads/master"},"890b740adc761a2a38657152a726b9450f172b97":{"kind":"REWORK","_number":7,"created":"2020-05-13 13:35:52.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/08/626008/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/626008/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/626008/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/626008/7"}}},"commit":{"parents":[{"commit":"d114ed5a7e4e46421c460db706443a2c341f5ba7","subject":"Merge \"Use py38 instead of py37 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d114ed5a7e4e46421c460db706443a2c341f5ba7"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2018-12-17 19:23:16.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-05-13 13:29:06.000000000","tz":-240},"subject":"Add tests to lower-constraints job","message":"Add tests to lower-constraints job\n\nThe current job simply installs requirements constrained by\nlower-constraints.txt.  While it is nice to know that the\nrequirements files are compatible, it would be even better to\nrun some tests to actually check that the code runs using the\ncurrent lower-constraints.txt file.\n\nChange-Id: I7a191ac44b004f840c986ded61341e43b35f1d49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/890b740adc761a2a38657152a726b9450f172b97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/890b740adc761a2a38657152a726b9450f172b97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
