)]}'
{"id":"openstack%2Fglance~628506","triplet_id":"openstack%2Fglance~master~Ia89ecb755783854791613061fcd3bef269893de7","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"Ia89ecb755783854791613061fcd3bef269893de7","subject":"Update hacking version","status":"ABANDONED","created":"2019-01-04 15:59:20.000000000","updated":"2019-08-13 15:54:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"74fc438d3b10066a9d9b28426ca543e23d9d5208","_number":628506,"virtual_id_number":628506,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-04 07:27:31.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":-1,"date":"2019-07-04 09:33:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-13 14:42:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-11 08:42:20.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-07-04 07:27:31.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-04 09:33:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-13 14:42:33.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"61e61167f8778967bdc705969be2e92a0ca301d8","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-01-04 15:59:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cfe4f2b702ceea33771e188a08cf27c32990c41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-04 18:19:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/06/628506/1/check/build-openstack-api-ref/e0a649c/html/ : SUCCESS in 5m 11s\n- requirements-check http://logs.openstack.org/06/628506/1/check/requirements-check/e80e072/ : SUCCESS in 2m 32s\n- grenade-py3 http://logs.openstack.org/06/628506/1/check/grenade-py3/a0c8774/ : FAILURE in 1h 10m 53s\n- tempest-full-py3 http://logs.openstack.org/06/628506/1/check/tempest-full-py3/8045842/ : SUCCESS in 1h 06m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/628506/1/check/openstack-tox-lower-constraints/7190416/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/06/628506/1/check/openstack-tox-pep8/f068b6b/ : FAILURE in 8m 19s\n- openstack-tox-py27 http://logs.openstack.org/06/628506/1/check/openstack-tox-py27/e4d99a1/ : SUCCESS in 5m 53s\n- openstack-tox-py35 http://logs.openstack.org/06/628506/1/check/openstack-tox-py35/6ef518e/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/06/628506/1/check/openstack-tox-py36/e195df0/ : SUCCESS in 3m 53s\n- openstack-tox-docs http://logs.openstack.org/06/628506/1/check/openstack-tox-docs/6c5da03/html/ : SUCCESS in 4m 55s\n- build-openstack-releasenotes http://logs.openstack.org/06/628506/1/check/build-openstack-releasenotes/47a3a1d/html/ : SUCCESS in 8m 11s\n- openstack-tox-functional http://logs.openstack.org/06/628506/1/check/openstack-tox-functional/070f9dc/ : SUCCESS in 8m 37s\n- openstack-tox-functional-py35 http://logs.openstack.org/06/628506/1/check/openstack-tox-functional-py35/14366b6/ : SUCCESS in 8m 43s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/06/628506/1/check/glance-eventlet-ssl-handshake-broken-py35/80d0c3f/ : FAILURE in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5c5d04fc884c25d4fc3ddb32c0acf25aab4240c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2019-03-14 15:23:43.000000000","message":"Patch Set 1: Code-Review-1\n\ngate failed","accounts_in_message":[],"_revision_number":1},{"id":"1d56206c3d0ec3b883affa2a8f534474342af238","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-11 08:42:20.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"ede80f2ba9eea2340a7e6c085ef8e36682b3ce55","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-11 09:06:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3365dfe7bec7d1bfb43bb79c9488b37179332dae","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-11 09:08:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c838e91c8706318d57d67f551bb6f4577f527711","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 11:00:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/06/628506/3/check/build-openstack-api-ref/b59e189/html/ : SUCCESS in 3m 55s\n- requirements-check http://logs.openstack.org/06/628506/3/check/requirements-check/5a9b255/ : FAILURE in 4m 59s\n- tempest-full http://logs.openstack.org/06/628506/3/check/tempest-full/10d71ef/ : SUCCESS in 1h 50m 10s\n- neutron-grenade http://logs.openstack.org/06/628506/3/check/neutron-grenade/d8d79af/ : FAILURE in 10m 05s (non-voting)\n- grenade-py3 http://logs.openstack.org/06/628506/3/check/grenade-py3/ca87b60/ : SUCCESS in 1h 04m 49s\n- tempest-full-py3 http://logs.openstack.org/06/628506/3/check/tempest-full-py3/783f2b1/ : SUCCESS in 1h 45m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/628506/3/check/openstack-tox-lower-constraints/a266285/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/06/628506/3/check/openstack-tox-pep8/a081128/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/06/628506/3/check/openstack-tox-py27/eb30b12/ : SUCCESS in 5m 04s\n- openstack-tox-py35 http://logs.openstack.org/06/628506/3/check/openstack-tox-py35/4bc5db0/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/06/628506/3/check/openstack-tox-py36/50175ce/ : SUCCESS in 5m 00s\n- openstack-tox-py37 http://logs.openstack.org/06/628506/3/check/openstack-tox-py37/996cd20/ : SUCCESS in 4m 55s\n- openstack-tox-docs http://logs.openstack.org/06/628506/3/check/openstack-tox-docs/d8452f6/html/ : SUCCESS in 5m 13s\n- build-openstack-releasenotes http://logs.openstack.org/06/628506/3/check/build-openstack-releasenotes/682b3b5/html/ : SUCCESS in 8m 42s\n- openstack-tox-functional http://logs.openstack.org/06/628506/3/check/openstack-tox-functional/d5a69a6/ : FAILURE in 8m 32s\n- openstack-tox-functional-py35 http://logs.openstack.org/06/628506/3/check/openstack-tox-functional-py35/e0a29a4/ : FAILURE in 8m 24s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/06/628506/3/check/glance-eventlet-ssl-handshake-broken-py35/bf2facb/ : FAILURE in 3m 02s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/06/628506/3/check/devstack-plugin-ceph-tempest/5b47f33/ : SUCCESS in 1h 23m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2b3451823ce71ceaf9334e6613b18260e7d38f71","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-11 11:41:45.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/06/628506/3/check-not-vote/ext-glance-zuul/e1f633f : FAILURE in 56m 09s","accounts_in_message":[],"_revision_number":3},{"id":"8a6522e8ea6bdfa64234a8b094a018c217ed26c3","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-07-04 07:27:10.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"82723744297220e7643e02d7ffb87cbefc58bca0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-04 07:27:31.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"a206c298cc531bdcf1eb498ec19373ea335f071d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 09:33:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/06/628506/4/check/build-openstack-api-ref/3707477/html/ : SUCCESS in 5m 54s\n- requirements-check http://logs.openstack.org/06/628506/4/check/requirements-check/5bfc33a/ : SUCCESS in 4m 06s\n- tempest-full http://logs.openstack.org/06/628506/4/check/tempest-full/8b68393/ : SUCCESS in 1h 46m 02s\n- neutron-grenade http://logs.openstack.org/06/628506/4/check/neutron-grenade/73e4bc7/ : SUCCESS in 1h 01m 38s\n- grenade-py3 http://logs.openstack.org/06/628506/4/check/grenade-py3/3ab7a5d/ : SUCCESS in 58m 54s\n- tempest-full-py3 http://logs.openstack.org/06/628506/4/check/tempest-full-py3/c6a5e20/ : SUCCESS in 1h 35m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/628506/4/check/openstack-tox-lower-constraints/d5a902e/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/06/628506/4/check/openstack-tox-pep8/78fe3ca/ : SUCCESS in 6m 15s\n- openstack-tox-py27 http://logs.openstack.org/06/628506/4/check/openstack-tox-py27/5c9daec/ : SUCCESS in 5m 54s\n- openstack-tox-py35 http://logs.openstack.org/06/628506/4/check/openstack-tox-py35/3285d4b/ : SUCCESS in 9m 30s\n- openstack-tox-py36 http://logs.openstack.org/06/628506/4/check/openstack-tox-py36/9c5e62b/ : SUCCESS in 6m 10s\n- openstack-tox-py37 http://logs.openstack.org/06/628506/4/check/openstack-tox-py37/cf0360b/ : SUCCESS in 4m 52s\n- openstack-tox-docs http://logs.openstack.org/06/628506/4/check/openstack-tox-docs/de70bef/html/ : SUCCESS in 5m 04s\n- build-openstack-releasenotes http://logs.openstack.org/06/628506/4/check/build-openstack-releasenotes/32cf0d7/html/ : SUCCESS in 7m 08s\n- openstack-tox-functional http://logs.openstack.org/06/628506/4/check/openstack-tox-functional/85c47fc/ : FAILURE in 8m 38s\n- openstack-tox-functional-py35 http://logs.openstack.org/06/628506/4/check/openstack-tox-functional-py35/22ff9ad/ : FAILURE in 8m 21s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/06/628506/4/check/glance-eventlet-ssl-handshake-broken-py3/18408ef/ : FAILURE in 4m 04s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/06/628506/4/check/glance-code-constants-check/fc5f968/ : SUCCESS in 3m 51s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/06/628506/4/check/devstack-plugin-ceph-tempest/19080fe/ : SUCCESS in 1h 24m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"97ce83fff77212c4e4e3cacf300612b7cae6c85d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2019-08-13 14:42:33.000000000","message":"Patch Set 4: Code-Review-1\n\ntake care of zuul jobs","accounts_in_message":[],"_revision_number":4},{"id":"74fc438d3b10066a9d9b28426ca543e23d9d5208","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-08-13 15:54:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"892b178e23606f304f6d4931113dbeef2b9c16e3","revisions":{"6602b0512b246e7437a15e7e620cca9d060bf95f":{"kind":"REWORK","_number":1,"created":"2019-01-04 15:59:20.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/06/628506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/628506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/628506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/628506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/628506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/628506/1"}}},"commit":{"parents":[{"commit":"7c6b851d66d355a3e15847af3e63bb2433b9e968","subject":"Merge \"Correct typo in config option choices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7c6b851d66d355a3e15847af3e63bb2433b9e968"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 15:59:00.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 15:59:16.000000000","tz":480},"subject":"Update hacking version to latest","message":"Update hacking version to latest\n\nChange-Id: Ia89ecb755783854791613061fcd3bef269893de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6602b0512b246e7437a15e7e620cca9d060bf95f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6602b0512b246e7437a15e7e620cca9d060bf95f"}]},"branch":"refs/heads/master"},"d5f76eabc39610278bae163b731ece09a286d74a":{"kind":"REWORK","_number":2,"created":"2019-04-11 09:06:16.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/06/628506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/628506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/628506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/628506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/628506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/628506/2"}}},"commit":{"parents":[{"commit":"d91d886101321404371b5f0ac06ebfd011c5257b","subject":"Merge \"add libpq-dev to build psycopg2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d91d886101321404371b5f0ac06ebfd011c5257b"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 15:59:00.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-11 09:06:12.000000000","tz":480},"subject":"Update hacking version to latest","message":"Update hacking version to latest\n\nChange-Id: Ia89ecb755783854791613061fcd3bef269893de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5f76eabc39610278bae163b731ece09a286d74a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5f76eabc39610278bae163b731ece09a286d74a"}]},"branch":"refs/heads/master"},"121ea4e129f4f7a7184d9cc51f136e1ffaf433de":{"kind":"REWORK","_number":3,"created":"2019-04-11 09:08:30.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/06/628506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/628506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/628506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/628506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/628506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/628506/3"}}},"commit":{"parents":[{"commit":"d91d886101321404371b5f0ac06ebfd011c5257b","subject":"Merge \"add libpq-dev to build psycopg2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d91d886101321404371b5f0ac06ebfd011c5257b"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 15:59:00.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-11 09:08:17.000000000","tz":480},"subject":"Update hacking version","message":"Update hacking version\n\nUse latest release 1.1.0 and compatible changes w.r.t pep8\n\nChange-Id: Ia89ecb755783854791613061fcd3bef269893de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/121ea4e129f4f7a7184d9cc51f136e1ffaf433de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/121ea4e129f4f7a7184d9cc51f136e1ffaf433de"}]},"branch":"refs/heads/master"},"892b178e23606f304f6d4931113dbeef2b9c16e3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-04 07:27:10.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/06/628506/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/628506/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/628506/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/628506/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/628506/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/628506/4"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 15:59:00.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-07-04 07:27:10.000000000","tz":0},"subject":"Update hacking version","message":"Update hacking version\n\nUse latest release 1.1.0 and compatible changes w.r.t pep8\n\nChange-Id: Ia89ecb755783854791613061fcd3bef269893de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/892b178e23606f304f6d4931113dbeef2b9c16e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/892b178e23606f304f6d4931113dbeef2b9c16e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
