)]}'
{"id":"openstack%2Fglance~634228","triplet_id":"openstack%2Fglance~stable%2Fqueens~Ie6641659b54543ed9f96c393d664e52a26bfaf6a","project":"openstack/glance","branch":"stable/queens","topic":"bp/mitigate-ossn-0075-stable/queens","hashtags":[],"change_id":"Ie6641659b54543ed9f96c393d664e52a26bfaf6a","subject":"Mitigate OSSN-0075","status":"MERGED","created":"2019-01-31 13:34:32.000000000","updated":"2019-09-10 15:36:25.000000000","submitted":"2019-09-10 15:36:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"634228-1568129785507-9806e538","meta_rev_id":"765a2896af5e5dd32571cd03c86a8397d3bac4e7","_number":634228,"virtual_id_number":634228,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2019-09-10 15:36:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2019-09-04 11:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2019-09-10 13:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2019-09-10 13:54:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2019-07-26 06:32:10.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 11:44:48.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-09-10 13:54:54.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-09-10 15:36:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26297d782c50aff625c70e61c47f03e88e47dbb6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-01-31 13:34:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"53aec20c9508ccd943b2ec8a8d76ba0b6d284868","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-01-31 13:34:51.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a77dc19c1631a0f547a4551d4d8c179d0c83c8aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-31 15:11:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/28/634228/1/check/openstack-tox-py35/8be29c5/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/28/634228/1/check/openstack-tox-pep8/42eb832/ : SUCCESS in 6m 26s\n- openstack-tox-py27 http://logs.openstack.org/28/634228/1/check/openstack-tox-py27/3103e90/ : SUCCESS in 6m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/28/634228/1/check/build-openstack-sphinx-docs/7dcadd0/html/ : SUCCESS in 6m 33s\n- tempest-full http://logs.openstack.org/28/634228/1/check/tempest-full/da2ca93/ : FAILURE in 1h 17m 04s\n- neutron-grenade http://logs.openstack.org/28/634228/1/check/neutron-grenade/a9f83b9/ : SUCCESS in 58m 53s\n- tempest-full-py3 http://logs.openstack.org/28/634228/1/check/tempest-full-py3/3a49e59/ : SUCCESS in 1h 14m 21s\n- openstack-tox-functional http://logs.openstack.org/28/634228/1/check/openstack-tox-functional/7019ac6/ : SUCCESS in 8m 31s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634228/1/check/openstack-tox-functional-py35/5c5724d/ : SUCCESS in 10m 56s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/28/634228/1/check/devstack-plugin-ceph-tempest/30e5c51/ : SUCCESS in 1h 33m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17777a2a47f0aee89c48a34287bbb8c415250169","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-24 04:13:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"445fa2babc0d201b512328001b65137bf09a77f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 05:43:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/28/634228/1/check/openstack-tox-py35/a435208/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/28/634228/1/check/openstack-tox-pep8/06be94d/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/28/634228/1/check/openstack-tox-py27/43948a2/ : SUCCESS in 4m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/28/634228/1/check/build-openstack-sphinx-docs/7cf0fbc/html/ : SUCCESS in 6m 00s\n- tempest-full http://logs.openstack.org/28/634228/1/check/tempest-full/7d374b6/ : SUCCESS in 1h 17m 04s\n- neutron-grenade http://logs.openstack.org/28/634228/1/check/neutron-grenade/b8b534e/ : SUCCESS in 53m 22s\n- tempest-full-py3 http://logs.openstack.org/28/634228/1/check/tempest-full-py3/b5ab7ae/ : SUCCESS in 1h 18m 28s\n- openstack-tox-functional http://logs.openstack.org/28/634228/1/check/openstack-tox-functional/5b05c3e/ : SUCCESS in 9m 14s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634228/1/check/openstack-tox-functional-py35/d4b650d/ : SUCCESS in 7m 48s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/28/634228/1/check/devstack-plugin-ceph-tempest/ee6f5d8/ : SUCCESS in 56m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac5bc05e1223fc3743d4b884a02f9752ea0b6406","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-26 06:32:10.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"e7f23bb36e7f0477ad136cafebc66db30b280651","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-09-04 11:44:48.000000000","message":"Patch Set 1: Code-Review+2\n\nAs a security patch this indeed is valid backport and should be done.","accounts_in_message":[],"_revision_number":1},{"id":"08cb6d2eed89788df387bb46353a5ec4818388f9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-10 13:54:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nAgree that this is an appropriate backport.","accounts_in_message":[],"_revision_number":1},{"id":"61ece5b7a0292520230423700c15c31ae12ba84e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 13:55:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea9bd616de1e650f634d1dd3c34e8491c8b506df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 15:36:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a5ca92a4e2fc417981c50f669930be74 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b588dfc0b245452cab4025edf501aca9 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e25f1d95e2b4870bf979871911ea1bb : SUCCESS in 5m 45s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/69bf7bde60ab4bd4a10f715c3162df90 : SUCCESS in 6m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/83440e3623f84a8cbb2c64f505253d39 : SUCCESS in 1h 20m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/19520ea2c3f94e139e14fd8a4e82f290 : SUCCESS in 1h 21m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59d8a066a22f47538f75f7a306216dcb : SUCCESS in 1h 27m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fcaf26ce3dcd4688a3f5d41cfc206317 : SUCCESS in 8m 51s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/94f8ae6bb0504d7abcc133b004ef8bf9 : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e8b550fcced39fa53883e49deae71e214af6486d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 15:36:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fdabcabf31b90e703bf9a68fa16c75731685a21","revisions":{"9fdabcabf31b90e703bf9a68fa16c75731685a21":{"kind":"REWORK","_number":1,"created":"2019-01-31 13:34:32.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/28/634228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/634228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/634228/1"}}},"commit":{"parents":[{"commit":"36e8be5136c87b5ea7b652007f210a07d5df80b0","subject":"Update show_multiple_locations deprecation note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/36e8be5136c87b5ea7b652007f210a07d5df80b0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2018-07-02 10:03:48.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-01-31 13:34:32.000000000","tz":0},"subject":"Mitigate OSSN-0075","message":"Mitigate OSSN-0075\n\nModified the current ``glance-manage db purge`` command to eliminate images\ntable from purging the records. Added new command\n``glance-manage db purge_images_table`` to purge the records from images\ntable.\n\nDocImpact\nSecurityImpact\n\nChange-Id: Ie6641659b54543ed9f96c393d664e52a26bfaf6a\nImplements: blueprint mitigate-ossn-0075\n(cherry picked from commit 5cc9d999352c21d3f4b5c39d3ea9d4378ca86544)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9fdabcabf31b90e703bf9a68fa16c75731685a21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9fdabcabf31b90e703bf9a68fa16c75731685a21"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
