)]}'
{"id":"openstack%2Fglance~634628","triplet_id":"openstack%2Fglance~master~I5f3bc734c63749ac1cdc34404d9243e2651e6936","project":"openstack/glance","branch":"master","topic":"pareser-error","hashtags":[],"change_id":"I5f3bc734c63749ac1cdc34404d9243e2651e6936","subject":"DNM: Fix subunit.parser error","status":"ABANDONED","created":"2019-02-04 06:43:10.000000000","updated":"2019-10-01 09:08:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9d7633bb0530e814659485e715214243767b0842","_number":634628,"virtual_id_number":634628,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-03 08:22:51.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"date":"2019-09-03 08:22:31.000000000","_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":-1,"date":"2019-09-03 08:43:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-04 06:49:01.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-02-04 06:49:06.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-02-04 06:49:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-02-04 06:50:15.000000000","updated_by":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"reviewer":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"state":"REVIEWER"},{"updated":"2019-09-03 08:22:51.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-03 08:43:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06b016307310da24d7dfe576429aa9b9d9205152","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 06:43:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f22c034caa2d382d4b7bf1ea57876d96e1edd52","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 06:47:54.000000000","message":"Patch Set 1: Workflow-1\n\nSome how I am able to reproduce this failure for single test in my environment for both functional and functional-py35 jobs. I have simply changed the response code [1] to trigger the failure and then ran single test using both functional and functional-py35 environments where I was able to see subunit.parser error.\n\nTo run single test:\n\ntox -e functional-py35 -- -n glance.tests.functional.v2.test_images.TestImages.test_owning_tenant_can_delete_image\n\ntox -e functional -- -n glance.tests.functional.v2.test_images.TestImages.test_owning_tenant_can_delete_image\n\n[1] https://github.com/openstack/glance/blob/master/glance/tests/functional/v2/test_images.py#L1974","accounts_in_message":[],"_revision_number":1},{"id":"d301f9771883b7c709b97284c6f2875bb02139a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 06:55:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/894791e/ : SUCCESS in 9m 49s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/f60bcdb/ : SUCCESS in 10m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5dd4cacdfc2768cdc2037eb33add18fff18606e3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 06:56:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"a0da48838f4aff2dcfa1b55ef8dccc40b949912e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 06:56:52.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"323aa0fdb46956f92239ab10a43dc192a0bfc3c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 07:30:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/e6325eb/ : SUCCESS in 8m 41s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/b86a88b/ : TIMED_OUT in 30m 53s","accounts_in_message":[],"_revision_number":1},{"id":"06a28716ba4675b131bd69f9cd1d7d19ce5f81db","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 07:32:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6424626ac47ff16ae68f304ee943c6deb324d5d9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 07:34:25.000000000","message":"Patch Set 1:\n\nFailure is timed_out, not a clue in the log. Compared result with functional jobs, functional job has around 743 tests and in case of timed_out 650 tests has been covered, not a clue why remaining 93 tests are not executed.","accounts_in_message":[],"_revision_number":1},{"id":"807a9cf28e54d199863e62edcb5a364069523ad3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 07:42:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/3291bf5/ : SUCCESS in 8m 03s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/ba7dddc/ : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"3603849a37df46b9074c21d0fb53cb9fe2d7fc28","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 07:42:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9f5f5119a8332f76d09231629075156c46a7d4cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 07:53:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/8fca640/ : SUCCESS in 7m 43s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/efa44de/ : SUCCESS in 9m 38s","accounts_in_message":[],"_revision_number":1},{"id":"657b54b7e78d4348b3ed1e0abdc071eeb25297b2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 07:55:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1188350357fa4df2ef29f4adb384f665c1bf5ef1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-04 08:09:13.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/28/634628/1/check-not-vote/ext-glance-zuul/c81d3ea : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e01d766153a4885a7370faeac81c211594bf4c78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 08:09:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/5fea26c/ : SUCCESS in 12m 01s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/02b0ff8/ : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2a31ffac1a487d400e5314754d2cc3a2039311ba","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 08:09:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f9b95fdcd06058fadfa0540ef016aa4d70d7edda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 08:41:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/165106c/ : SUCCESS in 7m 31s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/5e3a82b/ : TIMED_OUT in 31m 00s","accounts_in_message":[],"_revision_number":1},{"id":"032df483cb536b0336e52d7d9fc80e5319faf866","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-04 08:45:40.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/28/634628/1/check-not-vote/ext-glance-zuul/30bda8a : SUCCESS in 35m 38s","accounts_in_message":[],"_revision_number":1},{"id":"527642d2ca98fca2f32c80ad233da69b38347454","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 08:49:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c216b3c3410806d0da22febace078263005a153","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 09:22:00.000000000","message":"Patch Set 1:\n\nrechek","accounts_in_message":[],"_revision_number":1},{"id":"c48ba5fc00166a20de6507f4dcbdae9feb2ed9f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 09:22:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/77070ab/ : SUCCESS in 8m 06s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/5933fce/ : TIMED_OUT in 30m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6403b8d2ff0134818d62856229661d8463878160","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 09:23:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fe7ed215067bc84f4c315720286e53cc43beacdc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-04 09:25:28.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/28/634628/1/check-not-vote/ext-glance-zuul/b6a5357 : SUCCESS in 35m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b4b64edc4027eb6fb4db99114bad07b81b220603","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 09:55:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/5c75632/ : SUCCESS in 8m 42s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/c5109e9/ : TIMED_OUT in 31m 12s","accounts_in_message":[],"_revision_number":1},{"id":"364ca59adaad39caf63121b1f105724e601f45a8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 10:04:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"672c450396fae23cb0f92ce39c4cbbe97ffa5b60","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 10:05:02.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"edd8d952f8255560ceedea7a90be33877c7a10fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 10:15:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/74d5383/ : SUCCESS in 8m 57s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/6b7bcc3/ : SUCCESS in 8m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6ebc1c98dbc2c6b70acd655e8805fa915ebd9766","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 11:18:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"fa86f423887e50da477fcbeea066661d5cb01e50","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-04 11:18:50.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"b991c0a2982a0c891dd3998cd3f0051684e853dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 11:51:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/0d88ff2/ : SUCCESS in 10m 53s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/d364fa1/ : TIMED_OUT in 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7c46144e6f5545b2d94daab527acc86e4d31eda9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-05 07:06:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"3009ce7b162a576606e0420225662d6625532a0f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-05 07:06:12.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"9aa66d64df573fa9f3a025ced7c393eb9e29cee2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-05 07:14:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/207e925/ : SUCCESS in 7m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/5904d16/ : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":1},{"id":"681b5be7cbd35f20b1a896b1a5605732c803db51","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-05 07:15:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"92ddb3fa8b478b160c07dffd26f22b4b44b0038e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-05 07:16:02.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"ff093e03e4556ed9fb2670c19aefbfda25622386","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-05 07:52:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/3cf1629/ : SUCCESS in 7m 50s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/34d8595/ : TIMED_OUT in 30m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8cda9b6a886c25d57aed557d71f306ce0b9f3899","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-05 07:53:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"60126b83d907cebb6333fb048b30d09ba8a9dcb0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-05 07:53:49.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"671df0831d530f0c4c9d8f60397845075e728dab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-05 08:25:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/2b61d49/ : SUCCESS in 9m 38s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/c80b933/ : TIMED_OUT in 30m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6e77c08c3be504c1c35b32d77cd673d5e9d970da","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-05 08:30:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"14f978656074482018437c10063ecf0d3c79b84b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-05 08:30:18.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"b87792b24de5224c3f2b697d3048821116d81482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-05 08:43:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/7df1bcc/ : SUCCESS in 11m 25s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/f9cd1a8/ : SUCCESS in 9m 49s","accounts_in_message":[],"_revision_number":1},{"id":"06757e0a81aeeab8eee7faf9632fe54aa427407e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-05 08:44:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"a08660c31c698358ca07e85ee2759165a5e41a32","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-05 08:44:59.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"ecd3c173c71ea3d94fd2e3979047a849f90e82b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-05 08:57:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/9076bcc/ : SUCCESS in 7m 10s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/38a766a/ : FAILURE in 10m 23s","accounts_in_message":[],"_revision_number":1},{"id":"2549237ce55c4d0973f52b1f4be579bec6f5a144","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-05 08:59:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"88f82670233e0471f45f802597b9ccd778a01ce1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-05 08:59:47.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"9e6769f470037ce8b90cb116604f619014b7a088","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-05 09:32:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/8a04800/ : SUCCESS in 8m 47s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/21c68b0/ : TIMED_OUT in 30m 57s","accounts_in_message":[],"_revision_number":1},{"id":"346f2a7860175f42350088e2818aeed697af05a6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-06 08:11:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"8faefbf2cd828692daf2fc344c1e2a292ef1e0d1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-06 08:11:29.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"47ac746e2b62888db007b7053cf052eeb864421a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-06 08:29:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/fb49c2d/ : SUCCESS in 9m 28s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/68f030b/ : FAILURE in 8m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f207c5589e8463f3286fc019f488381c27dbdfeb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-06 08:30:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"8f46f54dc45b8a5c648060f18bfc07642bba403a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-06 08:30:42.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"79f355f4db11948c28c701e01019b3081ab8e095","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-06 08:46:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/8c55b07/ : SUCCESS in 9m 17s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/0ebe10b/ : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"6f46e8b7440bd92bb922a729c9b4463837b9b669","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-06 08:49:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"cf98e1ae4affb127236dbdb49f1d8afebbafe015","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-06 08:49:36.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"a1f11ab473aa5a30f4d81cc7b25bc8afda69f05c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-06 09:21:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/2f3a090/ : SUCCESS in 7m 58s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/4d87009/ : TIMED_OUT in 31m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d3c53758799cd69f526c945c7a712b1dad0ab9f0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-06 09:25:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"999e71c646ca35334361e5b50f39f79a89a862d6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-06 09:25:58.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"ac6350629625be95dd15c058bf1b0f63cb0ae099","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-06 09:35:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/1/check/openstack-tox-functional/bdfc001/ : SUCCESS in 7m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/1/check/openstack-tox-functional-py35/389308e/ : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":1},{"id":"832bdf353bb0db44116b0c9031ea041ecb59568e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-21 08:51:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9362944ac3550f80a2bd85e9dc18a4b9fd3be661","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-21 09:02:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/2/check/openstack-tox-functional/468ee72/ : SUCCESS in 9m 08s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/2/check/openstack-tox-functional-py35/66f7f83/ : FAILURE in 9m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f35b02d9c2156bd2a9fc53d119243b573a5d0c36","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-21 09:09:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e612fc45381fafcbb25d47fa173293cd2bc6b134","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-21 09:21:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/2/check/openstack-tox-functional/36c4320/ : SUCCESS in 7m 24s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/2/check/openstack-tox-functional-py35/fb973eb/ : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":2},{"id":"18db7c1e7a62b4f622e999efbb35620ab625f6ed","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-21 09:22:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"29d90806bccb6a545c9a26f7642705568193c48c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-21 09:28:22.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/28/634628/2/check-not-vote/ext-glance-zuul/d6a4ae5 : SUCCESS in 36m 14s","accounts_in_message":[],"_revision_number":2},{"id":"ce7200068f4b84b0a2bd9f4d6d5c3b0af21bdc20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-21 09:54:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/2/check/openstack-tox-functional/c59d3e1/ : SUCCESS in 8m 54s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/2/check/openstack-tox-functional-py35/d75181b/ : TIMED_OUT in 31m 01s","accounts_in_message":[],"_revision_number":2},{"id":"ec235177c2db5b38bd2b3cc1a6bb4424328a4543","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-03-29 07:37:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"869ff9e355ac5d7f797de447f09a5cbfab7fdf72","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 07:22:57.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"9f2e7c1429b6ff7801f247b73cc84666780eff48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 07:23:10.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"ae96d04cf294de5c8a155e86827398d83b91a4bd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 07:39:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"06c3881d7c22d977899ef20fb32bcd33d6097422","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 07:40:05.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  found undefined alias \u0027tempest-irrelevant-files\u0027\n  in \"openstack/glance/.zuul.yaml@master\", line 229, column 31","accounts_in_message":[],"_revision_number":3},{"id":"d8425c571db9a1ff1b0fe46ac9383e69e433614e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 07:58:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b9026c91288ff20c0af3a8a9a81b00c0e81815dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 08:23:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/4/check/openstack-tox-functional/170f114/ : SUCCESS in 8m 09s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/4/check/openstack-tox-functional-py35/d7539cb/ : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":4},{"id":"f126e7aa624296be0434c8f2ee54d10ad5a9625a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 08:25:53.000000000","message":"Patch Set 4:\n\nrechek","accounts_in_message":[],"_revision_number":4},{"id":"ae9b44018ea5f4a02e3f4f024e73fffbda37cc9a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 08:45:17.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"8e82cb59d7ae24c6d30036f0691574df25f634f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 09:07:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/4/check/openstack-tox-functional/85ae791/ : SUCCESS in 9m 05s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/4/check/openstack-tox-functional-py35/166bb7a/ : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":4},{"id":"5de0847f31375f3c34ffead63df65f536ffbcbae","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 09:14:10.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d5a198ba91f4a8b4874afbba24aad16ac17ce6d5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-07 09:43:51.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/28/634628/4/check-not-vote/ext-glance-zuul/fd438f5 : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":4},{"id":"5e79c48341b8f4fa113bdbfcb7e16617d973895b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 10:27:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional http://logs.openstack.org/28/634628/4/check/openstack-tox-functional/0992e8f/ : SUCCESS in 8m 02s\n- openstack-tox-functional-py35 http://logs.openstack.org/28/634628/4/check/openstack-tox-functional-py35/3a5324a/ : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":4},{"id":"89db642dbaf392676204e254ce8cd1d26274dc9d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-29 12:58:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e35316fba046172993d63a49af4c49171fa0c59e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 14:30:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4743a1dff9694181aa4cb5e693096fad : SUCCESS in 9m 14s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c1cf55f75eba4f0994611469e87cdd83 : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":5},{"id":"6336127f2561db82c65308d059c8cd3d74574045","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-29 18:33:54.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3e03bb870510901d7a3ad7a73f87fe7ef9810daa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 21:13:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/48cc8aafcd154211b445df1a055f7248 : SUCCESS in 11m 13s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7848f2d0a9b841d4ae5cbc1101603419 : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":5},{"id":"5aa86dff7e2434f82e272679cc5df14f32401560","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-30 06:32:14.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"45a826b0e9e10a445616c5ec062170019a5af3ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 07:34:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/196fe8c29053411bbbd55495a22765ec : SUCCESS in 9m 39s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/836d8450e6224e4a8be1eb5f70a0488b : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":5},{"id":"3d00e6c3f457bb75faf86574105e9ce619506cb4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-30 07:37:48.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"619b831f66aaa5987a17c0d35207eb16afe0620d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 08:20:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6424bb88065941f2a4dd26930717783a : SUCCESS in 12m 54s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/626e7441e2624ac18ea142ced3a0807b : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":5},{"id":"7fc64d4a3950753ec0754ee714559b954270a069","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-30 08:22:29.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"8fe44114da0286918fedc00b2fee66b535b1db2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 08:37:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d09bc49e970847d4b1df8e3054c87079 : SUCCESS in 9m 44s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e3f1723bc87e4b1983742cfedfdb72e8 : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":5},{"id":"e92e6883a8baabe004fcad080b31ac42cd358b3a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-30 08:47:47.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"79a1444552f0f86179c78f89d27da96640a863d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 09:03:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d7b3fc15e9d4449aaceac5228b778123 : SUCCESS in 10m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c359c80321b4469a80052bfc3638ba36 : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":5},{"id":"b8980076950816d3602905b888cd243e7c2558e7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-30 09:22:25.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"9f18c8072f877f2be4bf720449a215a124c084b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 10:44:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3d008beaa783412292466036b4ffeaf7 : SUCCESS in 11m 41s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/249397fc932542f29830a5873ccb6512 : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":5},{"id":"d2625460e1cd9039c5e3b37d6e4b3f61fce07250","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-03 05:14:40.000000000","message":"Patch Set 5:\n\nrecheck 7","accounts_in_message":[],"_revision_number":5},{"id":"a7f45e3911a594541457dbe406f9a56e2650f478","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 05:28:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d311b3643c644c1ebff22aad0c20598c : SUCCESS in 8m 26s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e98842c45ae4422cbc1602228d8969c9 : SUCCESS in 9m 03s","accounts_in_message":[],"_revision_number":5},{"id":"34b2f4130917b45779bc6a2eb825815352589a91","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-03 05:37:44.000000000","message":"Patch Set 5:\n\nrecheck 8","accounts_in_message":[],"_revision_number":5},{"id":"9a9ed78ce0d4f6055baf659e9bfc2bd567883aef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 05:51:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/140f8a2b7a474505ad59f0584002977e : SUCCESS in 8m 32s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/407bd8bbb7ec4e25a977f5a9067a37de : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":5},{"id":"686cd2d768df30dbad16949c729eb1fed88022d9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-03 05:53:41.000000000","message":"Patch Set 5:\n\nrecheck 9","accounts_in_message":[],"_revision_number":5},{"id":"d6cfb84da487f465658f43a078e238f2c23d4a3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 06:05:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/aaec7e98e2af41fdb2f0927becf0ac67 : SUCCESS in 10m 05s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1da5cc0f2aa646b6955bb02315bb8a59 : SUCCESS in 9m 26s","accounts_in_message":[],"_revision_number":5},{"id":"3d4c9ccf56b973c1e287ea1ce716dedfe200b0a4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-03 06:07:29.000000000","message":"Patch Set 5:\n\nrecheck 10","accounts_in_message":[],"_revision_number":5},{"id":"194adbfb63ba54451ab469adfe674c6ed65aafba","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-03 06:07:52.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"5823877b7b0e36156660f3333a203a5706b240c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 07:00:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/864e443ac5a542388f39ae27d909d983 : SUCCESS in 9m 06s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/207d8eef5dd540dab67c2d4d7603fb69 : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":5},{"id":"fae5a7d038507ca90027e15e18cedbfa22f31adb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-03 07:16:46.000000000","message":"Patch Set 5:\n\nrecheck 11","accounts_in_message":[],"_revision_number":5},{"id":"19687a23943412bd04db8ac1bceacc84d2cb7940","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-03 07:17:06.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"41bd0fc55ab114a84ad6358a282964d5474c4a4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 07:55:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/035fb12ff3c8413cba7f90f8ebf1c80b : SUCCESS in 8m 41s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e9e23de625fc46b4bdaf22c432898f75 : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":5},{"id":"06dfe77df09488852bc2dd7f8a5779d99111ebf4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-03 08:05:47.000000000","message":"Patch Set 5:\n\nrecheck 12","accounts_in_message":[],"_revision_number":5},{"id":"49b7c6411e0fc5066aa1dbe22508e60d2eba64a3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-03 08:06:07.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"a40e958e56ffe8f051669b7449f3d58f851e546f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 08:20:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dad46ca4b1944ddfb7878356a7ea5264 : SUCCESS in 9m 07s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/829a6fa39188451f8227d1ab6d066a00 : SUCCESS in 8m 59s","accounts_in_message":[],"_revision_number":5},{"id":"a03197681dff4115836676783a2dfab7b2133ad0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-03 08:22:31.000000000","message":"Patch Set 5:\n\nrecheck 13","accounts_in_message":[],"_revision_number":5},{"id":"ab6649dee18fa08eec7d2040bc101e48a5fafddc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-03 08:22:51.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"383774aa6b26c9f8d67ee7f3438cddef5d692dda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 08:43:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/94154834c72d47dfa11c88078bd5b660 : SUCCESS in 10m 07s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a9f1557c0b56489985db795bdd131ce9 : FAILURE in 14m 25s","accounts_in_message":[],"_revision_number":5},{"id":"d0ea85c2ff1d4746af0c4296a64bfc229dcb1b32","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-17 09:53:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5},{"id":"76748e5084e34760bf36e68f60a5ea8cb5322a0c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-10-01 05:01:00.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"e7264232d0c650880f1118d819bedbdfcb453f76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 05:01:13.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"1bf0534fc8f2e36098ef26e9ba6820679a9aaa59","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-10-01 09:08:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"390f5edb17456910cd63ef0719d941d3904ed69c","revisions":{"158548394fb0bd8d60737aea55be3474ee031653":{"kind":"REWORK","_number":1,"created":"2019-02-04 06:43:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/28/634628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/634628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/634628/1"}}},"commit":{"parents":[{"commit":"1d3f9ed234eb3b2c9e9e796818fb122372ec72fb","subject":"Merge \"Open Stein for data migrations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d3f9ed234eb3b2c9e9e796818fb122372ec72fb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-02-04 06:40:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-02-04 06:40:26.000000000","tz":0},"subject":"DNM: Fix subunit.parser error","message":"DNM: Fix subunit.parser error\n\nThe reason for parsing error is when any functional test fails\nit tries to attach entire api.log file and registry.log file to the\ntest result.\n\nSkipped adding entire logs to the output on failure as just stack\ntrace of failure will be sufficient to know what went wrong.\n\nChange-Id: I5f3bc734c63749ac1cdc34404d9243e2651e6936\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/158548394fb0bd8d60737aea55be3474ee031653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/158548394fb0bd8d60737aea55be3474ee031653"}]},"branch":"refs/heads/master"},"bbc84f77ab9d9d993f1a972a92718569bb260719":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-02-21 08:51:49.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/28/634628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/634628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/634628/2"}}},"commit":{"parents":[{"commit":"88a8ad78232cb53fbb2553247b62fedb8cf4dada","subject":"Fix DeprecationWarnings for RequestContext.tenant/user usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88a8ad78232cb53fbb2553247b62fedb8cf4dada"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-02-04 06:40:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-02-21 08:51:49.000000000","tz":0},"subject":"DNM: Fix subunit.parser error","message":"DNM: Fix subunit.parser error\n\nThe reason for parsing error is when any functional test fails\nit tries to attach entire api.log file and registry.log file to the\ntest result.\n\nSkipped adding entire logs to the output on failure as just stack\ntrace of failure will be sufficient to know what went wrong.\n\nChange-Id: I5f3bc734c63749ac1cdc34404d9243e2651e6936\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bbc84f77ab9d9d993f1a972a92718569bb260719"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bbc84f77ab9d9d993f1a972a92718569bb260719"}]},"branch":"refs/heads/master"},"7a36dfcc998ed57af6341c6154710b592b9aed9a":{"kind":"REWORK","_number":3,"created":"2019-06-07 07:39:51.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/28/634628/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/634628/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/634628/3"}}},"commit":{"parents":[{"commit":"be13eb05a0bc127069e743c9a6b14aa81b448b5d","subject":"Remove additional \" from config help","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be13eb05a0bc127069e743c9a6b14aa81b448b5d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-07 07:39:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-07 07:39:29.000000000","tz":0},"subject":"DNM: Fix subunit.parser error","message":"DNM: Fix subunit.parser error\n\nThe reason for parsing error is when any functional test fails\nit tries to attach entire api.log file and registry.log file to the\ntest result.\n\nSkipped adding entire logs to the output on failure as just stack\ntrace of failure will be sufficient to know what went wrong.\n\nChange-Id: I5f3bc734c63749ac1cdc34404d9243e2651e6936\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a36dfcc998ed57af6341c6154710b592b9aed9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a36dfcc998ed57af6341c6154710b592b9aed9a"}]},"branch":"refs/heads/master"},"be42c02bfcdbcdd0cf96aa4bd449b1d6a48f0beb":{"kind":"REWORK","_number":4,"created":"2019-06-07 07:58:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/28/634628/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/634628/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/634628/4"}}},"commit":{"parents":[{"commit":"be13eb05a0bc127069e743c9a6b14aa81b448b5d","subject":"Remove additional \" from config help","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be13eb05a0bc127069e743c9a6b14aa81b448b5d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-07 07:39:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-07 07:57:56.000000000","tz":0},"subject":"DNM: Fix subunit.parser error","message":"DNM: Fix subunit.parser error\n\nThe reason for parsing error is when any functional test fails\nit tries to attach entire api.log file and registry.log file to the\ntest result.\n\nSkipped adding entire logs to the output on failure as just stack\ntrace of failure will be sufficient to know what went wrong.\n\nChange-Id: I5f3bc734c63749ac1cdc34404d9243e2651e6936\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be42c02bfcdbcdd0cf96aa4bd449b1d6a48f0beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be42c02bfcdbcdd0cf96aa4bd449b1d6a48f0beb"}]},"branch":"refs/heads/master"},"390f5edb17456910cd63ef0719d941d3904ed69c":{"kind":"REWORK","_number":5,"created":"2019-08-29 12:58:24.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/28/634628/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/634628/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/634628/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/634628/5"}}},"commit":{"parents":[{"commit":"e9353d78611b278d3c4a2445d8d010f11abb2684","subject":"Merge \"Add \u0027compressed\u0027 option to container_format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e9353d78611b278d3c4a2445d8d010f11abb2684"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-07 07:39:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-08-29 12:57:43.000000000","tz":0},"subject":"DNM: Fix subunit.parser error","message":"DNM: Fix subunit.parser error\n\nThe reason for parsing error is when any functional test fails\nit tries to attach entire api.log file and registry.log file to the\ntest result.\n\nSkipped adding entire registry logs to the output on failure.\n\nChange-Id: I5f3bc734c63749ac1cdc34404d9243e2651e6936\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/390f5edb17456910cd63ef0719d941d3904ed69c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/390f5edb17456910cd63ef0719d941d3904ed69c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
