)]}'
{"id":"openstack%2Fglance~650879","triplet_id":"openstack%2Fglance~master~If86d805ddc46944e3aa2785aa5d775fe15b8468c","project":"openstack/glance","branch":"master","topic":"bug/1823703","hashtags":[],"change_id":"If86d805ddc46944e3aa2785aa5d775fe15b8468c","subject":"Use application_url in API version document","status":"MERGED","created":"2019-04-08 12:49:48.000000000","updated":"2020-04-06 11:34:52.000000000","submitted":"2020-04-06 07:08:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"650879-1586156904022-dc9227fa","meta_rev_id":"a714aabfac53a42a7f9d1e6abf5028937c71cf6d","_number":650879,"virtual_id_number":650879,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":31321,"name":"Andrey Bolgov","email":"bolgov@selectel.ru","username":"a-bolgov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-06 07:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-05 15:54:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-02-12 06:15:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":31321,"name":"Andrey Bolgov","email":"bolgov@selectel.ru","username":"a-bolgov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-04-06 05:43:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":31321,"name":"Andrey Bolgov","email":"bolgov@selectel.ru","username":"a-bolgov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":31321,"name":"Andrey Bolgov","email":"bolgov@selectel.ru","username":"a-bolgov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"_account_id":31321,"name":"Andrey Bolgov","email":"bolgov@selectel.ru","username":"a-bolgov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 20:25:34.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-05-13 20:25:56.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-05-14 06:25:32.000000000","updated_by":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"reviewer":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"state":"REVIEWER"},{"updated":"2019-12-17 13:35:30.000000000","updated_by":{"_account_id":31321,"name":"Andrey Bolgov","email":"bolgov@selectel.ru","username":"a-bolgov"},"reviewer":{"_account_id":31321,"name":"Andrey Bolgov","email":"bolgov@selectel.ru","username":"a-bolgov"},"state":"REVIEWER"},{"updated":"2019-12-17 13:35:51.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-05 15:54:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-06 05:43:07.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-04-06 07:08:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae123c525e17159ff6b9e1c773fc5cc459f41579","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-04-08 12:49:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5569a71744bb847511a5e37fb03cddf909cbf15d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-08 13:35:48.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/79/650879/1/check-not-vote/ext-glance-zuul/d578025 : FAILURE in 42m 59s","accounts_in_message":[],"_revision_number":1},{"id":"76511387d1dcfb538017fc33f289c7cc0d15178e","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-04-08 14:17:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ac5a3b29401623996d38f1b0beecc8dcdb1e39a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-08 15:28:53.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/79/650879/2/check-not-vote/ext-glance-zuul/ba14a08 : FAILURE in 36m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0306593fc0629e4edfb4f1686b1b261f0926dbce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 16:18:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/79/650879/2/check/tempest-full/120b132/ : FAILURE in 1h 34m 49s\n- neutron-grenade http://logs.openstack.org/79/650879/2/check/neutron-grenade/b7c726c/ : SUCCESS in 59m 11s (non-voting)\n- grenade-py3 http://logs.openstack.org/79/650879/2/check/grenade-py3/feeee5e/ : SUCCESS in 1h 06m 23s\n- tempest-full-py3 http://logs.openstack.org/79/650879/2/check/tempest-full-py3/4a48957/ : SUCCESS in 1h 31m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/650879/2/check/openstack-tox-lower-constraints/18643a2/ : SUCCESS in 5m 50s\n- openstack-tox-pep8 http://logs.openstack.org/79/650879/2/check/openstack-tox-pep8/c855d70/ : SUCCESS in 7m 20s\n- openstack-tox-py27 http://logs.openstack.org/79/650879/2/check/openstack-tox-py27/1da2397/ : SUCCESS in 7m 39s\n- openstack-tox-py35 http://logs.openstack.org/79/650879/2/check/openstack-tox-py35/1baceed/ : SUCCESS in 6m 26s\n- openstack-tox-py36 http://logs.openstack.org/79/650879/2/check/openstack-tox-py36/7cc07ed/ : SUCCESS in 6m 32s\n- openstack-tox-py37 http://logs.openstack.org/79/650879/2/check/openstack-tox-py37/ed5ce6c/ : SUCCESS in 8m 02s\n- openstack-tox-docs http://logs.openstack.org/79/650879/2/check/openstack-tox-docs/0d14241/html/ : SUCCESS in 7m 21s\n- openstack-tox-functional http://logs.openstack.org/79/650879/2/check/openstack-tox-functional/61d2899/ : SUCCESS in 11m 33s\n- openstack-tox-functional-py35 http://logs.openstack.org/79/650879/2/check/openstack-tox-functional-py35/6092b99/ : FAILURE in 10m 33s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/79/650879/2/check/glance-eventlet-ssl-handshake-broken-py35/c692c49/ : FAILURE in 5m 38s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/79/650879/2/check/devstack-plugin-ceph-tempest/4e89c3c/ : SUCCESS in 1h 30m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3de546942af3c6d9031b33fcd32bcf67212f6758","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-04-08 16:35:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f75f118779670cbd34897113f27b45d6e54b1fab","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-08 17:41:18.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/79/650879/2/check-not-vote/ext-glance-zuul/ebb3665 : FAILURE in 38m 53s","accounts_in_message":[],"_revision_number":2},{"id":"164978c820ccced086a92b6fce1be3abe7043e4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 19:16:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/79/650879/2/check/tempest-full/b453025/ : SUCCESS in 1h 30m 04s\n- neutron-grenade http://logs.openstack.org/79/650879/2/check/neutron-grenade/62235f4/ : SUCCESS in 51m 33s (non-voting)\n- grenade-py3 http://logs.openstack.org/79/650879/2/check/grenade-py3/843d574/ : SUCCESS in 1h 02m 44s\n- tempest-full-py3 http://logs.openstack.org/79/650879/2/check/tempest-full-py3/9cb9dbf/ : SUCCESS in 1h 38m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/650879/2/check/openstack-tox-lower-constraints/07aa58e/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/79/650879/2/check/openstack-tox-pep8/8928913/ : SUCCESS in 7m 43s\n- openstack-tox-py27 http://logs.openstack.org/79/650879/2/check/openstack-tox-py27/c157175/ : SUCCESS in 7m 37s\n- openstack-tox-py35 http://logs.openstack.org/79/650879/2/check/openstack-tox-py35/056dc19/ : SUCCESS in 5m 44s\n- openstack-tox-py36 http://logs.openstack.org/79/650879/2/check/openstack-tox-py36/237723f/ : SUCCESS in 5m 17s\n- openstack-tox-py37 http://logs.openstack.org/79/650879/2/check/openstack-tox-py37/c4268e5/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/79/650879/2/check/openstack-tox-docs/ffddecd/html/ : SUCCESS in 5m 33s\n- openstack-tox-functional http://logs.openstack.org/79/650879/2/check/openstack-tox-functional/c52a5c2/ : SUCCESS in 8m 57s\n- openstack-tox-functional-py35 http://logs.openstack.org/79/650879/2/check/openstack-tox-functional-py35/c9df2dd/ : SUCCESS in 16m 46s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/79/650879/2/check/glance-eventlet-ssl-handshake-broken-py35/8072100/ : FAILURE in 3m 42s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/79/650879/2/check/devstack-plugin-ceph-tempest/22b12c0/ : SUCCESS in 1h 37m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a819db316e73b2ba991206abc710f80117ade25d","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-05-14 06:25:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8e4c04d6158a4273779ce5f1d68ad4daa5691f25","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-12-16 14:20:59.000000000","message":"Patch Set 2:\n\nPlease, review","accounts_in_message":[],"_revision_number":2},{"id":"5e41f87eb4c2ff7494693dbde7654931366f06ae","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-17 06:14:11.000000000","message":"Patch Set 2:\n\nneed unit test to verify this","accounts_in_message":[],"_revision_number":2},{"id":"e79dd3e144b57af4731f013db3fcf4506d7fb41a","author":{"_account_id":31321,"name":"Andrey Bolgov","email":"bolgov@selectel.ru","username":"a-bolgov"},"date":"2019-12-17 13:35:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8a3117dd7d3c6644eae24d9535208150e0b18a02","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-17 13:35:51.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/79/650879/3/check-not-vote/ext-glance-zuul/6d4b2bd : FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"aca6353fc8270884d44d93213d02f0a7e1dc9801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 15:12:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9d7db74a82554354a9f357a2ad01837d : SUCCESS in 1h 33m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c79da3b3d6a14b05bef4eb0c13365086 : SUCCESS in 51m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a067d4def11e4b288d23e5992c8bedb9 : SUCCESS in 1h 25m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0c8bdeeec3448709d4c4f3a0911b66c : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6135c77aece749679f970cf2e16c3428 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5dc8f92890bc4003b0fbf84d97fac553 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6aacc3f284734f0a9366cd92421eba1e : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9f6399790604855b41a2a565bd6244a : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c1b974af5a400f8544b03776f33f62 : SUCCESS in 8m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3dc77b9727064bcba13764f2b2a5cc8a : SUCCESS in 10m 20s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d552518d9fe04d0ba7a0e9d7c0e49a99 : SUCCESS in 10m 17s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1538f0e5150d4c65a31225ef61e82141 : SUCCESS in 5m 33s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2b4da87f31e94e06b2338c1928e14f9b : SUCCESS in 1h 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a28f606898e9e8a24e574871294da2c7eb35640f","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-12-23 09:34:13.000000000","message":"Patch Set 3:\n\n\u003e need unit test to verify this\n\nThe test has been written.","accounts_in_message":[],"_revision_number":3},{"id":"98dc44366fae147dec241de9a1b51f97da2ad78d","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-02-03 19:22:59.000000000","message":"Topic changed from 650879 to bug/650879","accounts_in_message":[],"_revision_number":3},{"id":"e780a450182bc2435c8f8bda21f4fba9198ba33c","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-02-03 19:23:15.000000000","message":"Topic changed from bug/650879 to bug/1823703","accounts_in_message":[],"_revision_number":3},{"id":"5eede8e3dbfc5e561075558424fc19850038ac4b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-06 08:56:29.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nLooks good, nit in commit message.","accounts_in_message":[],"_revision_number":3},{"id":"068285aff97bf183f056e55a9cf0c1c861436850","author":{"_account_id":31321,"name":"Andrey Bolgov","email":"bolgov@selectel.ru","username":"a-bolgov"},"date":"2020-02-06 09:28:25.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9e21b3566eec9d291f8ce7366e8375406ed89011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 11:13:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a399e8eddf124cd29b65ae847afe4aa0 : FAILURE in 1h 01m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eee2591e2df3425eb7b8991990ef58bb : SUCCESS in 1h 34m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/496d231164e141b4aa1f5b2fe177497f : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8835c26a57c44e779d6e75f4caa5e77b : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f8c2db7ad23411ea80d9269ae88c48c : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af77bf6c23d942f59e6fa3aeca65a966 : FAILURE in 9m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2ea1c936fe947378531bf710ec8ba4e : FAILURE in 9m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/525da49f6d67465495c217b2b187bf79 : SUCCESS in 10m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9ca47222be244529946ab9aeb84aa881 : TIMED_OUT in 31m 39s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/676625d578024ae382839d7aad5684f7 : SUCCESS in 12m 49s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e02b5d1f11f34711b1db7092d0841863 : SUCCESS in 6m 39s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/23f4ab29148d4453ab1587547b05d1cc : SUCCESS in 1h 06m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd92752a86b04595982c12abccb4c678 : SUCCESS in 1h 20m 55s","accounts_in_message":[],"_revision_number":4},{"id":"6ee47750340fbcd7ad5947c4845ec275dee1d29a","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-02-06 12:57:56.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"d3dbcfcce573b8782658f5f817f736a921cad7fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 14:22:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/51d279770aaf4bd497c0228c5a2084b5 : SUCCESS in 54m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f50707bc198b428f86bd219e025132f0 : SUCCESS in 1h 19m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e78c5673a6764ead850c0e701672c5a8 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5daf52f1b3bb446783ee2aad51e740f4 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00bf2b2f7c534b53b9513be326a8a796 : SUCCESS in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fdb3e25e725d4263a729b308b0a29da1 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33f9324c44a54f6cb99d76e08f64ca90 : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/239348889ef24debaadd39dde776ff99 : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2406ace2d3054058b9996d41bcbe480d : SUCCESS in 9m 14s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/7f4826ac6a32430b94f676cf5d1ae5a1 : SUCCESS in 11m 48s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ab102a76785f41d99138f33eb12a2e8b : SUCCESS in 5m 13s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7bd5a78452b044f989d90e56588bcb3d : SUCCESS in 1h 09m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d722859c96a44578f8bc7f4152c0edd : SUCCESS in 1h 10m 22s","accounts_in_message":[],"_revision_number":5},{"id":"b5a8b67f0a07ca85146e1ae46154a09d211ec860","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-12 06:15:54.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"4232ee203bd23c59a5a5a002dd804a88e1441d67","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-05 15:54:00.000000000","message":"Patch Set 5: Code-Review+2\n\nLooks like this is good and our existing test cases provide coverage to make sure this doesn\u0027t break other formats. I will let someone else approve though.","accounts_in_message":[],"_revision_number":5},{"id":"9e2b51d19d5e4fc508cc44b851274ff6eca5285b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-06 05:43:07.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"eb3061d8eea501dde44bc3157eed9a7ff393d5ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 05:43:19.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6949673321877afaf3cfb421c30b1b3ce5dd6a59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 07:08:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2fbf3efec79941eb867ecffd62af1012 : SUCCESS in 1h 19m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9c55892668d7499b885324faa04aac6a : SUCCESS in 1h 22m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8e3766c55b34e1f8ac9decb17af4439 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fa7d9617cf9424ba046fae756d4a906 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fe044b1db364f3ab766c70fcedc30a4 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bbe3a24488194de7a619a5ed26409a6d : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceec20a17f414e239c3cd84fdc2bce28 : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d664f81a54f84387a3de6c610d35db18 : SUCCESS in 12m 09s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3dfbf7a725e64be3a9343f56a5fcd38d : SUCCESS in 12m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9c6dd19727f46a0a16878b98274246c : SUCCESS in 57m 38s\n\nWarning:\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file glance/cmd/api.py","accounts_in_message":[],"_revision_number":5},{"id":"5ed5f8f67431edb4f396e9b1b4861e589b2e6796","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 07:08:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"1ad867efde30e82edf36d5e99ffb6cf228ede7aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 07:10:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb4d23c2dc2144d58930a235dd770385 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":5},{"id":"a714aabfac53a42a7f9d1e6abf5028937c71cf6d","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-04-06 11:34:52.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 6bed540dcdbb8251c6aa4ae0887e668cfc72f2ac","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"33626369e591b03eb75c44b5c45c7969c7cb40e9","revisions":{"cfee80bb30f56b97886e9297b49aff2e3713ef3e":{"kind":"REWORK","_number":1,"created":"2019-04-08 12:49:48.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/79/650879/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/650879/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/650879/1"}}},"commit":{"parents":[{"commit":"218aa1cd3942927df41a3d094929a73de68b6416","subject":"Merge \"Update CLI doc header for Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/218aa1cd3942927df41a3d094929a73de68b6416"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-04-08 12:49:16.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-04-08 12:49:46.000000000","tz":180},"subject":"Use application_url in API version document","message":"Use application_url in API version document\n\nThis handles the possible vhost in Glance API path correctly\n(like https://cloud.example.com/image) contrary to host_url.\n\nChange-Id: If86d805ddc46944e3aa2785aa5d775fe15b8468c\nCloses-Bug: #1823703\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfee80bb30f56b97886e9297b49aff2e3713ef3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfee80bb30f56b97886e9297b49aff2e3713ef3e"}]},"branch":"refs/heads/master"},"b586b461eac019426c6b3ef222fb6f3d8bf2c514":{"kind":"REWORK","_number":2,"created":"2019-04-08 14:17:57.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/79/650879/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/650879/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/650879/2"}}},"commit":{"parents":[{"commit":"218aa1cd3942927df41a3d094929a73de68b6416","subject":"Merge \"Update CLI doc header for Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/218aa1cd3942927df41a3d094929a73de68b6416"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-04-08 12:49:16.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-04-08 14:17:24.000000000","tz":180},"subject":"Use application_url in API version document","message":"Use application_url in API version document\n\nThis handles the possible vhost in Glance API path correctly\n(like https://cloud.example.com/image) contrary to host_url.\n\nChange-Id: If86d805ddc46944e3aa2785aa5d775fe15b8468c\nCloses-Bug: #1823703\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b586b461eac019426c6b3ef222fb6f3d8bf2c514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b586b461eac019426c6b3ef222fb6f3d8bf2c514"}]},"branch":"refs/heads/master"},"7df7313d7c708cda59ab140545f3920f91bdb84b":{"kind":"REWORK","_number":3,"created":"2019-12-17 13:35:30.000000000","uploader":{"_account_id":31321,"name":"Andrey Bolgov","email":"bolgov@selectel.ru","username":"a-bolgov"},"ref":"refs/changes/79/650879/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/650879/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/650879/3"}}},"commit":{"parents":[{"commit":"218aa1cd3942927df41a3d094929a73de68b6416","subject":"Merge \"Update CLI doc header for Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/218aa1cd3942927df41a3d094929a73de68b6416"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-04-08 12:49:16.000000000","tz":180},"committer":{"name":"Bolgov Andrey","email":"bolgov@selectel.ru","date":"2019-12-17 13:34:35.000000000","tz":180},"subject":"Use application_url in API version document","message":"Use application_url in API version document\n\nThis handles the possible vhost in Glance API path correctly\n(like https://cloud.example.com/image) contrary to host_url.\n\nChange-Id: If86d805ddc46944e3aa2785aa5d775fe15b8468c\nCloses-Bug: #1823703\n\nUse application_url in API version document\n\nThis handles the possible vhost in Glance API path correctly\n(like https://cloud.example.com/image) contrary to host_url.\n\nChange-Id: If86d805ddc46944e3aa2785aa5d775fe15b8468c\nCloses-Bug: #1823703\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7df7313d7c708cda59ab140545f3920f91bdb84b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7df7313d7c708cda59ab140545f3920f91bdb84b"}]},"branch":"refs/heads/master"},"ae86008054e31e6fad944ffaa798059421efcd34":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-02-06 09:28:25.000000000","uploader":{"_account_id":31321,"name":"Andrey Bolgov","email":"bolgov@selectel.ru","username":"a-bolgov"},"ref":"refs/changes/79/650879/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/650879/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/650879/4"}}},"commit":{"parents":[{"commit":"218aa1cd3942927df41a3d094929a73de68b6416","subject":"Merge \"Update CLI doc header for Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/218aa1cd3942927df41a3d094929a73de68b6416"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-04-08 12:49:16.000000000","tz":180},"committer":{"name":"Bolgov Andrey","email":"bolgov@selectel.ru","date":"2020-02-06 09:25:06.000000000","tz":180},"subject":"Use application_url in API version document","message":"Use application_url in API version document\n\nThis handles the possible vhost in Glance API path correctly\n(like https://cloud.example.com/image) contrary to host_url.\n\nChange-Id: If86d805ddc46944e3aa2785aa5d775fe15b8468c\nCloses-Bug: #1823703\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae86008054e31e6fad944ffaa798059421efcd34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae86008054e31e6fad944ffaa798059421efcd34"}]},"branch":"refs/heads/master"},"33626369e591b03eb75c44b5c45c7969c7cb40e9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-02-06 12:57:56.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/79/650879/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/650879/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/650879/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/650879/5"}}},"commit":{"parents":[{"commit":"8649fdc2a293009899d2aa918ec360e30adc8e77","subject":"Ensure store ID parsed from URI is properly encoded","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8649fdc2a293009899d2aa918ec360e30adc8e77"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-04-08 12:49:16.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2020-02-06 12:57:56.000000000","tz":0},"subject":"Use application_url in API version document","message":"Use application_url in API version document\n\nThis handles the possible vhost in Glance API path correctly\n(like https://cloud.example.com/image) contrary to host_url.\n\nChange-Id: If86d805ddc46944e3aa2785aa5d775fe15b8468c\nCloses-Bug: #1823703\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/33626369e591b03eb75c44b5c45c7969c7cb40e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/33626369e591b03eb75c44b5c45c7969c7cb40e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
