)]}'
{"id":"openstack%2Fglance~654277","triplet_id":"openstack%2Fglance~master~Ibe48a4ea537915bc6ce142812354b1d3b3ab1cf7","project":"openstack/glance","branch":"master","topic":"opendev","hashtags":[],"change_id":"Ibe48a4ea537915bc6ce142812354b1d3b3ab1cf7","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-22 03:01:13.000000000","updated":"2019-05-03 06:59:50.000000000","submitted":"2019-05-03 06:59:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654277-1556866790122-577aa814","meta_rev_id":"49a22b0e535a13e89d99b54e4d11b0329e4e515c","_number":654277,"virtual_id_number":654277,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-05-03 03:54:22.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"date":"2019-05-03 03:15:24.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-05-03 06:59:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-02 16:00:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-02 13:31:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-02 13:31:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 01:14:21.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-04-23 05:44:35.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-05-02 13:31:08.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-05-02 16:00:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-03 03:54:22.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-03 06:59:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7303218aea1dcfb6e69359c7292dea204a8bae00","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-22 03:01:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c227ea36de907655a60f1c6e5ea7afb34295fc77","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-22 03:01:38.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"a3d1b9d81bcd01285ecfc658de172bf625c92850","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 04:45:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/77/654277/1/check/build-openstack-api-ref/c2c6d36/html/ : SUCCESS in 4m 19s\n- tempest-full http://logs.openstack.org/77/654277/1/check/tempest-full/513d28c/ : SUCCESS in 1h 39m 23s\n- neutron-grenade http://logs.openstack.org/77/654277/1/check/neutron-grenade/93c888c/ : SUCCESS in 57m 27s\n- grenade-py3 http://logs.openstack.org/77/654277/1/check/grenade-py3/c42b4db/ : SUCCESS in 1h 05m 42s\n- tempest-full-py3 http://logs.openstack.org/77/654277/1/check/tempest-full-py3/5e33e6d/ : SUCCESS in 1h 34m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/654277/1/check/openstack-tox-lower-constraints/10d3853/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/77/654277/1/check/openstack-tox-pep8/d3ea677/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/77/654277/1/check/openstack-tox-py27/4066e0e/ : SUCCESS in 4m 31s\n- openstack-tox-py35 http://logs.openstack.org/77/654277/1/check/openstack-tox-py35/cbc0353/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/77/654277/1/check/openstack-tox-py36/4148ce6/ : SUCCESS in 4m 22s\n- openstack-tox-py37 http://logs.openstack.org/77/654277/1/check/openstack-tox-py37/3b02133/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/77/654277/1/check/openstack-tox-docs/6f2457b/html/ : SUCCESS in 4m 25s\n- build-openstack-releasenotes http://logs.openstack.org/77/654277/1/check/build-openstack-releasenotes/e11a984/html/ : SUCCESS in 8m 25s\n- openstack-tox-functional http://logs.openstack.org/77/654277/1/check/openstack-tox-functional/a6ef2ac/ : SUCCESS in 8m 22s\n- openstack-tox-functional-py35 http://logs.openstack.org/77/654277/1/check/openstack-tox-functional-py35/6262806/ : SUCCESS in 7m 50s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/77/654277/1/check/glance-eventlet-ssl-handshake-broken-py35/4d83f3e/ : FAILURE in 2m 40s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/77/654277/1/check/glance-code-constants-check/3642a6b/ : SUCCESS in 4m 00s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/77/654277/1/check/devstack-plugin-ceph-tempest/3e9a80c/ : SUCCESS in 1h 30m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"126c76b2f5189af47f76c7a4e077dfa8d428e976","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 11:16:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c7a87f7c30f08d18223f068927d8612cd0fe269","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 01:14:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5f98ed7f8c7eea0a7312e95e04f6d8d2b21e446c","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-23 05:44:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"52d5b91c6d612475791b7c9ab6f4714c438f6a89","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-24 01:43:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f702c347ae216ef4b1f244036f758d649baa2500","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-24 01:46:28.000000000","message":"Patch Set 2:\n\nI checked all the urls and they are all available. Every place in this project has changed","accounts_in_message":[],"_revision_number":2},{"id":"76275e8f26c361008e86fd6805fd884dc6c2a8fc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-24 03:26:27.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/77/654277/2/check-not-vote/ext-glance-zuul/500d50a : FAILURE in 38m 27s","accounts_in_message":[],"_revision_number":2},{"id":"63f5959d37c4cb8315eaef9834778223d7f6feb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 06:14:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/77/654277/2/check/build-openstack-api-ref/9591fc9/html/ : SUCCESS in 4m 54s\n- tempest-full http://logs.openstack.org/77/654277/2/check/tempest-full/19f3b53/ : SUCCESS in 1h 44m 52s\n- neutron-grenade http://logs.openstack.org/77/654277/2/check/neutron-grenade/32fbe29/ : SUCCESS in 50m 09s\n- grenade-py3 http://logs.openstack.org/77/654277/2/check/grenade-py3/3a7e5ed/ : SUCCESS in 56m 29s\n- tempest-full-py3 http://logs.openstack.org/77/654277/2/check/tempest-full-py3/54da946/ : SUCCESS in 1h 22m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/654277/2/check/openstack-tox-lower-constraints/a941302/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/77/654277/2/check/openstack-tox-pep8/4205321/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/77/654277/2/check/openstack-tox-py27/3866ba8/ : SUCCESS in 4m 04s\n- openstack-tox-py35 http://logs.openstack.org/77/654277/2/check/openstack-tox-py35/64434d3/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/77/654277/2/check/openstack-tox-py36/bcba01f/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/77/654277/2/check/openstack-tox-py37/a1b3a38/ : SUCCESS in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/77/654277/2/check/openstack-tox-docs/9490bde/html/ : SUCCESS in 4m 36s\n- build-openstack-releasenotes http://logs.openstack.org/77/654277/2/check/build-openstack-releasenotes/fd833e8/html/ : SUCCESS in 7m 30s\n- openstack-tox-functional http://logs.openstack.org/77/654277/2/check/openstack-tox-functional/373b872/ : SUCCESS in 7m 21s\n- openstack-tox-functional-py35 http://logs.openstack.org/77/654277/2/check/openstack-tox-functional-py35/75a7991/ : SUCCESS in 8m 21s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/77/654277/2/check/glance-eventlet-ssl-handshake-broken-py35/5cacc12/ : FAILURE in 2m 41s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/77/654277/2/check/glance-code-constants-check/46c0a91/ : SUCCESS in 4m 12s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/77/654277/2/check/devstack-plugin-ceph-tempest/8c75551/ : SUCCESS in 1h 55m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5068319067369dc527f51a6d54a8c78ff0d6f38c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-04-24 12:40:00.000000000","message":"Patch Set 2: Code-Review-1\n\nThanks for picking this up.  There seems to be a problem with the release notes (probably not your fault).  If you look at the generated html for Queens:\nhttp://logs.openstack.org/77/654277/2/check/build-openstack-releasenotes/fd833e8/html/queens.html\n\nThe \u0027draft Policy Refactor spec\u0027 link is to the old location.  Same thing if you scroll down to the metadefs changes (16.0.0 \u003e Upgrade Notes).\n\nI didn\u0027t check any others.  I also didn\u0027t try to generate the pages locally to see if the correct links show up.  I didn\u0027t see anything on the mailing list, either.\n\nIf the HTML pages generate correctly for you locally, it may be a problem with the zuul job configuration, so if that\u0027s the case, you can ask for help in #openstack-infra.  If the links aren\u0027t correct locally, I guess the best place to ask someone is either #openstack-doc or #openstack-dev","accounts_in_message":[],"_revision_number":2},{"id":"b0a1fb946cf5af876b9479a2b5cf323f0f6755e0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-25 02:22:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"836b9e725b2b8ce0b8a4a1fb326ee0b073716aa6","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-25 02:22:19.000000000","message":"Patch Set 2:\n\nThe release note can be updated by only updating the file in that branch, please check [1] for more information.\n\n[1] https://docs.openstack.org/ironic/latest/contributor/faq.html#update-a-release-note","accounts_in_message":[],"_revision_number":2},{"id":"8d7aa1123d5de9b4356e6249082a6cf479eaea87","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-25 02:58:28.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/77/654277/3/check-not-vote/ext-glance-zuul/4666154 : FAILURE in 36m 07s","accounts_in_message":[],"_revision_number":3},{"id":"914a25be88e5a9af1fbffaee10fdfd32ee1fefc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 07:26:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/77/654277/3/check/build-openstack-api-ref/1250fbf/html/ : SUCCESS in 3m 50s\n- tempest-full http://logs.openstack.org/77/654277/3/check/tempest-full/a23499e/ : SUCCESS in 1h 35m 08s\n- neutron-grenade http://logs.openstack.org/77/654277/3/check/neutron-grenade/8030194/ : SUCCESS in 54m 48s\n- grenade-py3 http://logs.openstack.org/77/654277/3/check/grenade-py3/2a96162/ : SUCCESS in 1h 03m 53s\n- tempest-full-py3 http://logs.openstack.org/77/654277/3/check/tempest-full-py3/1ae3e9d/ : SUCCESS in 1h 47m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/654277/3/check/openstack-tox-lower-constraints/5f26712/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/77/654277/3/check/openstack-tox-pep8/c05c9e5/ : SUCCESS in 5m 54s\n- openstack-tox-py27 http://logs.openstack.org/77/654277/3/check/openstack-tox-py27/5d9d481/ : SUCCESS in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/77/654277/3/check/openstack-tox-py35/f32f7e3/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/77/654277/3/check/openstack-tox-py36/b4cc162/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/77/654277/3/check/openstack-tox-py37/110a291/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/77/654277/3/check/openstack-tox-docs/a448637/html/ : SUCCESS in 4m 32s\n- build-openstack-releasenotes http://logs.openstack.org/77/654277/3/check/build-openstack-releasenotes/f1e72d2/html/ : SUCCESS in 8m 02s\n- openstack-tox-functional http://logs.openstack.org/77/654277/3/check/openstack-tox-functional/195203f/ : SUCCESS in 7m 52s\n- openstack-tox-functional-py35 http://logs.openstack.org/77/654277/3/check/openstack-tox-functional-py35/04355e4/ : TIMED_OUT in 30m 54s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/77/654277/3/check/glance-eventlet-ssl-handshake-broken-py35/9217991/ : FAILURE in 3m 12s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/77/654277/3/check/glance-code-constants-check/4a96afe/ : SUCCESS in 4m 14s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/77/654277/3/check/devstack-plugin-ceph-tempest/6d18877/ : SUCCESS in 1h 27m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d202b928f5f5ac5eef08b07008e62728560d6fc9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-25 07:48:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fc13c9e7250b4b0a88a86f164b8f6d99f03c6970","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-25 08:21:30.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/77/654277/3/check-not-vote/ext-glance-zuul/8a6c130 : SUCCESS in 32m 58s","accounts_in_message":[],"_revision_number":3},{"id":"6dfdbc2bba117f48b3f7c6e20e9d7777a1c86b8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 13:08:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/77/654277/3/check/build-openstack-api-ref/9d61b5f/html/ : SUCCESS in 3m 44s\n- tempest-full http://logs.openstack.org/77/654277/3/check/tempest-full/d0964a1/ : SUCCESS in 1h 45m 26s\n- neutron-grenade http://logs.openstack.org/77/654277/3/check/neutron-grenade/7a8b9ab/ : SUCCESS in 1h 02m 28s\n- grenade-py3 http://logs.openstack.org/77/654277/3/check/grenade-py3/0b207a4/ : SUCCESS in 1h 00m 17s\n- tempest-full-py3 http://logs.openstack.org/77/654277/3/check/tempest-full-py3/9a1e125/ : SUCCESS in 1h 34m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/654277/3/check/openstack-tox-lower-constraints/74ddb8b/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/77/654277/3/check/openstack-tox-pep8/10fde60/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/77/654277/3/check/openstack-tox-py27/9502113/ : SUCCESS in 4m 05s\n- openstack-tox-py35 http://logs.openstack.org/77/654277/3/check/openstack-tox-py35/2914f7c/ : SUCCESS in 4m 44s\n- openstack-tox-py36 http://logs.openstack.org/77/654277/3/check/openstack-tox-py36/68202e9/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/77/654277/3/check/openstack-tox-py37/2585e17/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/77/654277/3/check/openstack-tox-docs/cac2881/html/ : SUCCESS in 4m 45s\n- build-openstack-releasenotes http://logs.openstack.org/77/654277/3/check/build-openstack-releasenotes/977d374/html/ : SUCCESS in 7m 38s\n- openstack-tox-functional http://logs.openstack.org/77/654277/3/check/openstack-tox-functional/6850834/ : SUCCESS in 8m 12s\n- openstack-tox-functional-py35 http://logs.openstack.org/77/654277/3/check/openstack-tox-functional-py35/66d659b/ : SUCCESS in 9m 52s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/77/654277/3/check/glance-eventlet-ssl-handshake-broken-py35/d348be0/ : FAILURE in 4m 03s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/77/654277/3/check/glance-code-constants-check/9188d19/ : SUCCESS in 4m 02s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/77/654277/3/check/devstack-plugin-ceph-tempest/d8e02ea/ : SUCCESS in 1h 41m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d7b851ccfebcd6a570805d2a4769692c42a9cacd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-04-26 01:15:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nThanks for tracking down the release notes info.  It makes sense to leave them un-corrected.\n\nTwo small changes noted inline, and then this will be ready to go.","accounts_in_message":[],"_revision_number":3},{"id":"46bfa0942e8fd754837d60f3271e5c4035bfdfc8","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-28 01:38:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0305febcd9c54359d7ce947baafe27dc7f962b91","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-28 02:14:31.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/77/654277/4/check-not-vote/ext-glance-zuul/0898171 : FAILURE in 35m 16s","accounts_in_message":[],"_revision_number":4},{"id":"344a172d377dddd0c8f7d98b55c1ee65dd6a95b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 03:13:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/77/654277/4/check/build-openstack-api-ref/0d17524/html/ : SUCCESS in 4m 06s\n- tempest-full http://logs.openstack.org/77/654277/4/check/tempest-full/335d594/ : FAILURE in 1h 24m 41s\n- neutron-grenade http://logs.openstack.org/77/654277/4/check/neutron-grenade/84efe5a/ : SUCCESS in 50m 39s\n- grenade-py3 http://logs.openstack.org/77/654277/4/check/grenade-py3/7625b31/ : SUCCESS in 1h 01m 16s\n- tempest-full-py3 http://logs.openstack.org/77/654277/4/check/tempest-full-py3/279c7ba/ : SUCCESS in 1h 27m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/654277/4/check/openstack-tox-lower-constraints/8b74d6b/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/77/654277/4/check/openstack-tox-pep8/40c7a36/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/77/654277/4/check/openstack-tox-py27/3c4398c/ : SUCCESS in 4m 27s\n- openstack-tox-py35 http://logs.openstack.org/77/654277/4/check/openstack-tox-py35/f4cdc37/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/77/654277/4/check/openstack-tox-py36/f0bc0d3/ : SUCCESS in 6m 05s\n- openstack-tox-py37 http://logs.openstack.org/77/654277/4/check/openstack-tox-py37/5521c5e/ : SUCCESS in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/77/654277/4/check/openstack-tox-docs/c624fe1/html/ : SUCCESS in 6m 35s\n- build-openstack-releasenotes http://logs.openstack.org/77/654277/4/check/build-openstack-releasenotes/2318bd6/html/ : SUCCESS in 6m 55s\n- openstack-tox-functional http://logs.openstack.org/77/654277/4/check/openstack-tox-functional/31cf43d/ : SUCCESS in 10m 12s\n- openstack-tox-functional-py35 http://logs.openstack.org/77/654277/4/check/openstack-tox-functional-py35/bf34056/ : SUCCESS in 8m 09s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/77/654277/4/check/glance-eventlet-ssl-handshake-broken-py35/afea761/ : FAILURE in 4m 10s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/77/654277/4/check/glance-code-constants-check/1004ad3/ : SUCCESS in 4m 32s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/77/654277/4/check/devstack-plugin-ceph-tempest/688ee8c/ : SUCCESS in 1h 33m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3898e52ab8ab00974b196d2ec11d55c376aa7572","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-28 03:51:50.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b74993c0651a6218f3a8c4cf8b953f7010bdafd0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-28 04:26:58.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/77/654277/4/check-not-vote/ext-glance-zuul/152164a : FAILURE in 34m 48s","accounts_in_message":[],"_revision_number":4},{"id":"1cee9483e19256a6fc416fa4651b365058a1dc78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 06:05:39.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/77/654277/4/check/build-openstack-api-ref/78f4620/html/ : SUCCESS in 4m 06s\n- tempest-full http://logs.openstack.org/77/654277/4/check/tempest-full/567fca6/ : TIMED_OUT in 2h 06m 51s\n- neutron-grenade http://logs.openstack.org/77/654277/4/check/neutron-grenade/de2271c/ : SUCCESS in 55m 28s\n- grenade-py3 http://logs.openstack.org/77/654277/4/check/grenade-py3/3c1eb44/ : SUCCESS in 55m 45s\n- tempest-full-py3 http://logs.openstack.org/77/654277/4/check/tempest-full-py3/1a39fc0/ : SUCCESS in 1h 34m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/654277/4/check/openstack-tox-lower-constraints/e1c7622/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/77/654277/4/check/openstack-tox-pep8/c1648a7/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/77/654277/4/check/openstack-tox-py27/ea2b8c8/ : SUCCESS in 3m 50s\n- openstack-tox-py35 http://logs.openstack.org/77/654277/4/check/openstack-tox-py35/7de0a56/ : SUCCESS in 4m 38s\n- openstack-tox-py36 http://logs.openstack.org/77/654277/4/check/openstack-tox-py36/4c38c17/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/77/654277/4/check/openstack-tox-py37/219c7aa/ : SUCCESS in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/77/654277/4/check/openstack-tox-docs/5269fd3/html/ : SUCCESS in 4m 35s\n- build-openstack-releasenotes http://logs.openstack.org/77/654277/4/check/build-openstack-releasenotes/dc48e12/html/ : SUCCESS in 6m 31s\n- openstack-tox-functional http://logs.openstack.org/77/654277/4/check/openstack-tox-functional/e69c749/ : SUCCESS in 8m 35s\n- openstack-tox-functional-py35 http://logs.openstack.org/77/654277/4/check/openstack-tox-functional-py35/47521b6/ : SUCCESS in 8m 38s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/77/654277/4/check/glance-eventlet-ssl-handshake-broken-py35/ef13f8a/ : FAILURE in 2m 13s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/77/654277/4/check/glance-code-constants-check/fda427b/ : SUCCESS in 4m 55s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/77/654277/4/check/devstack-plugin-ceph-tempest/6c325ce/ : SUCCESS in 1h 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f73a19fc7beac4f7d75a868b2f6b61fc8605f7be","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-29 00:57:04.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e4c900bb1ee99be80f2e89811e8c9e5a9af6d363","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-29 01:38:03.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/77/654277/4/check-not-vote/ext-glance-zuul/b0fe130 : FAILURE in 40m 39s","accounts_in_message":[],"_revision_number":4},{"id":"c253881874d910ebd7546ebe556678349f6f5d5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 02:50:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/77/654277/4/check/build-openstack-api-ref/a99f15e/html/ : SUCCESS in 4m 05s\n- tempest-full http://logs.openstack.org/77/654277/4/check/tempest-full/a73d646/ : SUCCESS in 1h 48m 02s\n- neutron-grenade http://logs.openstack.org/77/654277/4/check/neutron-grenade/86388ad/ : SUCCESS in 58m 28s\n- grenade-py3 http://logs.openstack.org/77/654277/4/check/grenade-py3/118efa4/ : SUCCESS in 1h 01m 30s\n- tempest-full-py3 http://logs.openstack.org/77/654277/4/check/tempest-full-py3/805abad/ : SUCCESS in 1h 35m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/654277/4/check/openstack-tox-lower-constraints/b53a1a5/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/77/654277/4/check/openstack-tox-pep8/79ec560/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/77/654277/4/check/openstack-tox-py27/4f4f61c/ : SUCCESS in 5m 43s\n- openstack-tox-py35 http://logs.openstack.org/77/654277/4/check/openstack-tox-py35/7fb5baf/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/77/654277/4/check/openstack-tox-py36/8e71c91/ : SUCCESS in 4m 52s\n- openstack-tox-py37 http://logs.openstack.org/77/654277/4/check/openstack-tox-py37/01bb485/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/77/654277/4/check/openstack-tox-docs/8da4a31/html/ : SUCCESS in 4m 40s\n- build-openstack-releasenotes http://logs.openstack.org/77/654277/4/check/build-openstack-releasenotes/73f353d/html/ : SUCCESS in 7m 44s\n- openstack-tox-functional http://logs.openstack.org/77/654277/4/check/openstack-tox-functional/e8dcce7/ : SUCCESS in 9m 22s\n- openstack-tox-functional-py35 http://logs.openstack.org/77/654277/4/check/openstack-tox-functional-py35/199203b/ : SUCCESS in 10m 23s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/77/654277/4/check/glance-eventlet-ssl-handshake-broken-py35/26afc39/ : FAILURE in 2m 29s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/77/654277/4/check/glance-code-constants-check/09ba1d9/ : SUCCESS in 4m 21s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/77/654277/4/check/devstack-plugin-ceph-tempest/bdb6207/ : SUCCESS in 1h 33m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5651abf74721fe6a9f2ec37f69055f2973a0dd4e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-05-02 13:31:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nNinja-approving a docs change.","accounts_in_message":[],"_revision_number":4},{"id":"f145078b60a010361814039263d4e942abe71910","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 13:31:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"050658f33a182508596524bf1569d8a3debc2692","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 15:43:54.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/77/654277/4/gate/build-openstack-api-ref/4759aca/html/ : SUCCESS in 3m 49s\n- tempest-full http://logs.openstack.org/77/654277/4/gate/tempest-full/ddc137d/ : TIMED_OUT in 2h 07m 30s\n- neutron-grenade http://logs.openstack.org/77/654277/4/gate/neutron-grenade/adf0df1/ : SUCCESS in 1h 02m 04s\n- grenade-py3 http://logs.openstack.org/77/654277/4/gate/grenade-py3/aab80f8/ : SUCCESS in 1h 01m 04s\n- tempest-full-py3 http://logs.openstack.org/77/654277/4/gate/tempest-full-py3/5b845b5/ : SUCCESS in 1h 22m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/654277/4/gate/openstack-tox-lower-constraints/2f76ca2/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/77/654277/4/gate/openstack-tox-pep8/8d6155f/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/77/654277/4/gate/openstack-tox-py27/0213b78/ : SUCCESS in 4m 05s\n- openstack-tox-py35 http://logs.openstack.org/77/654277/4/gate/openstack-tox-py35/1941aef/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/77/654277/4/gate/openstack-tox-py36/26f3e50/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/77/654277/4/gate/openstack-tox-py37/bf7b183/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/77/654277/4/gate/openstack-tox-docs/e3c3c3c/html/ : SUCCESS in 4m 14s\n- build-openstack-releasenotes http://logs.openstack.org/77/654277/4/gate/build-openstack-releasenotes/7957cd6/html/ : SUCCESS in 8m 11s\n- openstack-tox-functional http://logs.openstack.org/77/654277/4/gate/openstack-tox-functional/0d00401/ : SUCCESS in 8m 13s\n- openstack-tox-functional-py35 http://logs.openstack.org/77/654277/4/gate/openstack-tox-functional-py35/3510df5/ : SUCCESS in 9m 34s","accounts_in_message":[],"_revision_number":4},{"id":"62f3b0b0f1844df0a40ff932b8100d03e01d3e39","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-02 16:00:06.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6ccfd5e56393ba9219720d9ab7b0e83407b8a768","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 19:21:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/77/654277/4/check/build-openstack-api-ref/22a0025/html/ : SUCCESS in 4m 51s\n- tempest-full http://logs.openstack.org/77/654277/4/check/tempest-full/8c17e66/ : TIMED_OUT in 2h 06m 50s\n- neutron-grenade http://logs.openstack.org/77/654277/4/check/neutron-grenade/0e7dc84/ : SUCCESS in 1h 20m 48s\n- grenade-py3 http://logs.openstack.org/77/654277/4/check/grenade-py3/f6add9f/ : SUCCESS in 1h 08m 15s\n- tempest-full-py3 http://logs.openstack.org/77/654277/4/check/tempest-full-py3/62e1e46/ : SUCCESS in 1h 33m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/654277/4/check/openstack-tox-lower-constraints/fa5c8a2/ : SUCCESS in 8m 24s\n- openstack-tox-pep8 http://logs.openstack.org/77/654277/4/check/openstack-tox-pep8/87f3e7c/ : SUCCESS in 5m 07s\n- openstack-tox-py27 http://logs.openstack.org/77/654277/4/check/openstack-tox-py27/86d5667/ : SUCCESS in 5m 57s\n- openstack-tox-py35 http://logs.openstack.org/77/654277/4/check/openstack-tox-py35/f496b39/ : SUCCESS in 4m 35s\n- openstack-tox-py36 http://logs.openstack.org/77/654277/4/check/openstack-tox-py36/c5c2b35/ : SUCCESS in 4m 52s\n- openstack-tox-py37 http://logs.openstack.org/77/654277/4/check/openstack-tox-py37/6abc3df/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/77/654277/4/check/openstack-tox-docs/240e56b/html/ : SUCCESS in 4m 59s\n- build-openstack-releasenotes http://logs.openstack.org/77/654277/4/check/build-openstack-releasenotes/e98689b/html/ : SUCCESS in 7m 48s\n- openstack-tox-functional http://logs.openstack.org/77/654277/4/check/openstack-tox-functional/f58bb96/ : SUCCESS in 8m 22s\n- openstack-tox-functional-py35 http://logs.openstack.org/77/654277/4/check/openstack-tox-functional-py35/e65d662/ : SUCCESS in 8m 17s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/77/654277/4/check/glance-eventlet-ssl-handshake-broken-py35/d0dc9dc/ : FAILURE in 3m 14s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/77/654277/4/check/glance-code-constants-check/94c5c27/ : SUCCESS in 4m 40s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/77/654277/4/check/devstack-plugin-ceph-tempest/db5bfab/ : SUCCESS in 1h 44m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"eee79cb575b8afc3be0d4145522c8aab9442856d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-02 21:57:45.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"998004d6357046c5611bf71f37cb02f6b55c77ac","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-02 23:38:23.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/77/654277/4/check-not-vote/ext-glance-zuul/9bd5148 : FAILURE in 40m 59s","accounts_in_message":[],"_revision_number":4},{"id":"7f3f22f2af71606151d469aec7597126c5208de2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 23:47:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/77/654277/4/check/build-openstack-api-ref/a1db62c/html/ : SUCCESS in 4m 29s\n- tempest-full http://logs.openstack.org/77/654277/4/check/tempest-full/9e0a2a9/ : SUCCESS in 1h 48m 10s\n- neutron-grenade http://logs.openstack.org/77/654277/4/check/neutron-grenade/0364bcc/ : SUCCESS in 49m 24s\n- grenade-py3 http://logs.openstack.org/77/654277/4/check/grenade-py3/623421f/ : SUCCESS in 55m 00s\n- tempest-full-py3 http://logs.openstack.org/77/654277/4/check/tempest-full-py3/fec0c42/ : SUCCESS in 1h 37m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/654277/4/check/openstack-tox-lower-constraints/c2dbf2b/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/77/654277/4/check/openstack-tox-pep8/87d1892/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/77/654277/4/check/openstack-tox-py27/fde65bd/ : SUCCESS in 6m 07s\n- openstack-tox-py35 http://logs.openstack.org/77/654277/4/check/openstack-tox-py35/51bbdfe/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/77/654277/4/check/openstack-tox-py36/bb27090/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/77/654277/4/check/openstack-tox-py37/fff505a/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/77/654277/4/check/openstack-tox-docs/781c01b/html/ : SUCCESS in 4m 56s\n- build-openstack-releasenotes http://logs.openstack.org/77/654277/4/check/build-openstack-releasenotes/3a0c43c/html/ : SUCCESS in 8m 42s\n- openstack-tox-functional http://logs.openstack.org/77/654277/4/check/openstack-tox-functional/9776b6a/ : SUCCESS in 9m 44s\n- openstack-tox-functional-py35 http://logs.openstack.org/77/654277/4/check/openstack-tox-functional-py35/adc522d/ : SUCCESS in 8m 06s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/77/654277/4/check/glance-eventlet-ssl-handshake-broken-py35/e118445/ : FAILURE in 3m 13s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/77/654277/4/check/glance-code-constants-check/a699a9f/ : SUCCESS in 3m 47s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/77/654277/4/check/devstack-plugin-ceph-tempest/2487d7f/ : SUCCESS in 1h 32m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cf2a677df2fb357f59b60adf17e96da47ccca710","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 23:47:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"192fe7da16403749b701fd93921c0f0c58b3d60c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 01:55:47.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/77/654277/4/gate/build-openstack-api-ref/86512bc/html/ : SUCCESS in 3m 51s\n- tempest-full http://logs.openstack.org/77/654277/4/gate/tempest-full/106338b/ : FAILURE in 1h 51m 21s\n- neutron-grenade http://logs.openstack.org/77/654277/4/gate/neutron-grenade/63d526c/ : SUCCESS in 54m 59s\n- grenade-py3 http://logs.openstack.org/77/654277/4/gate/grenade-py3/79e174c/ : SUCCESS in 59m 19s\n- tempest-full-py3 http://logs.openstack.org/77/654277/4/gate/tempest-full-py3/eb881ed/ : SUCCESS in 1h 51m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/654277/4/gate/openstack-tox-lower-constraints/ae90b5a/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/77/654277/4/gate/openstack-tox-pep8/43b28f8/ : SUCCESS in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/77/654277/4/gate/openstack-tox-py27/85cb1cc/ : SUCCESS in 6m 13s\n- openstack-tox-py35 http://logs.openstack.org/77/654277/4/gate/openstack-tox-py35/de564ef/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/77/654277/4/gate/openstack-tox-py36/187e67f/ : SUCCESS in 4m 16s\n- openstack-tox-py37 http://logs.openstack.org/77/654277/4/gate/openstack-tox-py37/19845ff/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/77/654277/4/gate/openstack-tox-docs/50a3371/html/ : SUCCESS in 4m 31s\n- build-openstack-releasenotes http://logs.openstack.org/77/654277/4/gate/build-openstack-releasenotes/59da5e6/html/ : SUCCESS in 7m 52s\n- openstack-tox-functional http://logs.openstack.org/77/654277/4/gate/openstack-tox-functional/c24e6af/ : SUCCESS in 9m 41s\n- openstack-tox-functional-py35 http://logs.openstack.org/77/654277/4/gate/openstack-tox-functional-py35/9ffc966/ : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":4},{"id":"cf2a3d2ff92b2fbe0d3fdd598c6afc094079bad2","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-03 03:15:24.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"8ef8efb94125b2be0bdac0481590ebc7b8d88264","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-03 03:54:22.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/77/654277/4/check-not-vote/ext-glance-zuul/5abeb04 : FAILURE in 38m 38s","accounts_in_message":[],"_revision_number":4},{"id":"2941c75ab5d1a46996503ea15a9a640263d362e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 05:09:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/77/654277/4/check/build-openstack-api-ref/5643148/html/ : SUCCESS in 3m 55s\n- tempest-full http://logs.openstack.org/77/654277/4/check/tempest-full/2e50136/ : SUCCESS in 1h 52m 12s\n- neutron-grenade http://logs.openstack.org/77/654277/4/check/neutron-grenade/eaad048/ : SUCCESS in 49m 17s\n- grenade-py3 http://logs.openstack.org/77/654277/4/check/grenade-py3/fc1d9e5/ : SUCCESS in 49m 09s\n- tempest-full-py3 http://logs.openstack.org/77/654277/4/check/tempest-full-py3/5af6a7f/ : SUCCESS in 1h 20m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/654277/4/check/openstack-tox-lower-constraints/473cd1b/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/77/654277/4/check/openstack-tox-pep8/ce8c72d/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/77/654277/4/check/openstack-tox-py27/43ea424/ : SUCCESS in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/77/654277/4/check/openstack-tox-py35/251ed9c/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/77/654277/4/check/openstack-tox-py36/1f61050/ : SUCCESS in 4m 04s\n- openstack-tox-py37 http://logs.openstack.org/77/654277/4/check/openstack-tox-py37/bff6af7/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/77/654277/4/check/openstack-tox-docs/69877c3/html/ : SUCCESS in 4m 33s\n- build-openstack-releasenotes http://logs.openstack.org/77/654277/4/check/build-openstack-releasenotes/2cd9626/html/ : SUCCESS in 7m 55s\n- openstack-tox-functional http://logs.openstack.org/77/654277/4/check/openstack-tox-functional/c2344da/ : SUCCESS in 8m 49s\n- openstack-tox-functional-py35 http://logs.openstack.org/77/654277/4/check/openstack-tox-functional-py35/3dec172/ : SUCCESS in 8m 42s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/77/654277/4/check/glance-eventlet-ssl-handshake-broken-py35/b2aac06/ : FAILURE in 2m 35s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/77/654277/4/check/glance-code-constants-check/a08a259/ : SUCCESS in 3m 47s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/77/654277/4/check/devstack-plugin-ceph-tempest/bd7bba3/ : SUCCESS in 1h 32m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"20315232d6e93f9dd3f79dd547b7300cdf9fd5ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 05:09:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1a698fdd9199f41e9d2fa240e8e6833660522a92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 06:59:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/77/654277/4/gate/build-openstack-api-ref/ef16858/html/ : SUCCESS in 4m 53s\n- tempest-full http://logs.openstack.org/77/654277/4/gate/tempest-full/7190f3b/ : SUCCESS in 1h 45m 31s\n- neutron-grenade http://logs.openstack.org/77/654277/4/gate/neutron-grenade/f470463/ : SUCCESS in 45m 29s\n- grenade-py3 http://logs.openstack.org/77/654277/4/gate/grenade-py3/e8835ef/ : SUCCESS in 57m 24s\n- tempest-full-py3 http://logs.openstack.org/77/654277/4/gate/tempest-full-py3/d960ee2/ : SUCCESS in 1h 29m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/654277/4/gate/openstack-tox-lower-constraints/77763dc/ : SUCCESS in 4m 31s\n- openstack-tox-pep8 http://logs.openstack.org/77/654277/4/gate/openstack-tox-pep8/334746c/ : SUCCESS in 5m 05s\n- openstack-tox-py27 http://logs.openstack.org/77/654277/4/gate/openstack-tox-py27/d5da8a2/ : SUCCESS in 4m 15s\n- openstack-tox-py35 http://logs.openstack.org/77/654277/4/gate/openstack-tox-py35/35baa3c/ : SUCCESS in 4m 04s\n- openstack-tox-py36 http://logs.openstack.org/77/654277/4/gate/openstack-tox-py36/59fa9fa/ : SUCCESS in 5m 21s\n- openstack-tox-py37 http://logs.openstack.org/77/654277/4/gate/openstack-tox-py37/dea70ac/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/77/654277/4/gate/openstack-tox-docs/771cf47/html/ : SUCCESS in 5m 23s\n- build-openstack-releasenotes http://logs.openstack.org/77/654277/4/gate/build-openstack-releasenotes/5667bcb/html/ : SUCCESS in 6m 47s\n- openstack-tox-functional http://logs.openstack.org/77/654277/4/gate/openstack-tox-functional/ae07bf8/ : SUCCESS in 8m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/77/654277/4/gate/openstack-tox-functional-py35/27aa743/ : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":4},{"id":"aa44712b804a1e9df462117d816d8f4fd6d23b5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 06:59:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8e16e3a857dcfc1d0db612dbc69e24947f0aeebc","revisions":{"37e320546c2abbe2a8f83881bfa348b3a2b99f5f":{"kind":"REWORK","_number":1,"created":"2019-04-22 03:01:13.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/77/654277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/654277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/654277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/654277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/654277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/654277/1"}}},"commit":{"parents":[{"commit":"305efc5cb2555a23a6bb137eb13e7961f5acc858","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/305efc5cb2555a23a6bb137eb13e7961f5acc858"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-22 02:59:18.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-22 02:59:18.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org\ncounterparts.\n\nChange-Id: Ibe48a4ea537915bc6ce142812354b1d3b3ab1cf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/37e320546c2abbe2a8f83881bfa348b3a2b99f5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/37e320546c2abbe2a8f83881bfa348b3a2b99f5f"}]},"branch":"refs/heads/master"},"e85985263860b4e83f3bc797a84f5db254b50754":{"kind":"REWORK","_number":2,"created":"2019-04-24 01:43:44.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/77/654277/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/654277/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/654277/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/654277/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/654277/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/654277/2"}}},"commit":{"parents":[{"commit":"fa13302a58e8c3f913e569d38522bdc9288c94d5","subject":"Merge \"Switch to using stestr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa13302a58e8c3f913e569d38522bdc9288c94d5"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-22 02:59:18.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-24 01:43:36.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org\ncounterparts.\n\nChange-Id: Ibe48a4ea537915bc6ce142812354b1d3b3ab1cf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e85985263860b4e83f3bc797a84f5db254b50754"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e85985263860b4e83f3bc797a84f5db254b50754"}]},"branch":"refs/heads/master"},"eca8e10112f9d4558257d3922343ad8b57e78d30":{"kind":"REWORK","_number":3,"created":"2019-04-25 02:22:02.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/77/654277/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/654277/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/654277/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/654277/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/654277/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/654277/3"}}},"commit":{"parents":[{"commit":"fa13302a58e8c3f913e569d38522bdc9288c94d5","subject":"Merge \"Switch to using stestr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa13302a58e8c3f913e569d38522bdc9288c94d5"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-22 02:59:18.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-25 02:20:14.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org\ncounterparts.\n\nChange-Id: Ibe48a4ea537915bc6ce142812354b1d3b3ab1cf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eca8e10112f9d4558257d3922343ad8b57e78d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eca8e10112f9d4558257d3922343ad8b57e78d30"}]},"branch":"refs/heads/master"},"8e16e3a857dcfc1d0db612dbc69e24947f0aeebc":{"kind":"REWORK","_number":4,"created":"2019-04-28 01:38:50.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/77/654277/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/654277/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/654277/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/654277/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/654277/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/654277/4"}}},"commit":{"parents":[{"commit":"fa13302a58e8c3f913e569d38522bdc9288c94d5","subject":"Merge \"Switch to using stestr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa13302a58e8c3f913e569d38522bdc9288c94d5"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-22 02:59:18.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 01:26:48.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org\ncounterparts.\n\nChange-Id: Ibe48a4ea537915bc6ce142812354b1d3b3ab1cf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e16e3a857dcfc1d0db612dbc69e24947f0aeebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e16e3a857dcfc1d0db612dbc69e24947f0aeebc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
