)]}'
{"id":"openstack%2Fglance~654826","triplet_id":"openstack%2Fglance~master~I7afd43c2e0581d30a78d5ed4e30f8eb2af1cf444","project":"openstack/glance","branch":"master","topic":"bug/1826136","hashtags":[],"change_id":"I7afd43c2e0581d30a78d5ed4e30f8eb2af1cf444","subject":"Added powervm as hypervisor type for image metadata prefiltering.","status":"MERGED","created":"2019-04-23 08:15:16.000000000","updated":"2019-05-02 00:46:55.000000000","submitted":"2019-05-02 00:46:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654826-1556758015002-af911147","meta_rev_id":"775ccbd138d758a0a208d42f65d5f7830309157e","_number":654826,"virtual_id_number":654826,"owner":{"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-29 07:08:14.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-05-02 00:46:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-29 06:37:22.000000000","_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":2,"date":"2019-05-01 20:48:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":1,"date":"2019-04-30 18:04:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":1,"date":"2019-05-01 20:48:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 08:15:27.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-04-23 09:42:53.000000000","updated_by":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"reviewer":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"state":"REVIEWER"},{"updated":"2019-04-24 06:58:50.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-04-24 06:59:09.000000000","updated_by":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"reviewer":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"state":"REVIEWER"},{"updated":"2019-04-25 00:18:41.000000000","updated_by":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"reviewer":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"state":"REVIEWER"},{"updated":"2019-04-29 07:08:14.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 18:04:05.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-05-01 20:48:16.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-05-02 00:46:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd4ab922bcbcd488b3f0a35a601976816dd70570","author":{"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"},"date":"2019-04-23 08:15:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf38318205d851aaaa992f74607b4a2699602184","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-04-23 08:15:27.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"2ed36b7d9ce4c53b2f35aa909cc732e66b6eaccb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-23 08:52:08.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/26/654826/1/check-not-vote/ext-glance-zuul/28e72ad : FAILURE in 36m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7c9735eb224b3b9523c75a0eade8076b4f785786","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:41:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/26/654826/1/check/openstack-tox-lower-constraints/0097655/ : SUCCESS in 4m 13s\n- openstack-tox-pep8 http://logs.openstack.org/26/654826/1/check/openstack-tox-pep8/22164a1/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/26/654826/1/check/openstack-tox-py27/2e2eb57/ : SUCCESS in 4m 42s\n- openstack-tox-py35 http://logs.openstack.org/26/654826/1/check/openstack-tox-py35/c310583/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/26/654826/1/check/openstack-tox-py36/277fba0/ : SUCCESS in 4m 37s\n- openstack-tox-py37 http://logs.openstack.org/26/654826/1/check/openstack-tox-py37/ec7854b/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/26/654826/1/check/openstack-tox-docs/128041c/html/ : SUCCESS in 6m 18s\n- openstack-tox-functional http://logs.openstack.org/26/654826/1/check/openstack-tox-functional/6f96dd5/ : SUCCESS in 11m 42s\n- openstack-tox-functional-py35 http://logs.openstack.org/26/654826/1/check/openstack-tox-functional-py35/ec8f7be/ : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":1},{"id":"356d80a8659b01b6f97c19f5b7fd5114094d161e","author":{"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"},"date":"2019-04-24 08:59:46.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"10be8be9c93bab68dee3ba04f9d417c76d8d0490","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-24 09:30:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"000e0b19ac93d627292b215f72b62a3059e3b780","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-24 09:40:38.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/26/654826/2/check-not-vote/ext-glance-zuul/5bb1e66 : FAILURE in 40m 33s","accounts_in_message":[],"_revision_number":2},{"id":"274976a8bba3f45592639788b38848e4086ceaac","author":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"date":"2019-04-24 10:10:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"31b9400af6ff5633869fdc0cf617dd7e1c2ffac3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 11:46:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/26/654826/2/check/openstack-tox-lower-constraints/50cff93/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/26/654826/2/check/openstack-tox-pep8/15c9569/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/26/654826/2/check/openstack-tox-py27/694d8b9/ : SUCCESS in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/26/654826/2/check/openstack-tox-py35/efd7c3e/ : SUCCESS in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/26/654826/2/check/openstack-tox-py36/34aaa52/ : SUCCESS in 4m 58s\n- openstack-tox-py37 http://logs.openstack.org/26/654826/2/check/openstack-tox-py37/8b7a8ef/ : SUCCESS in 5m 45s\n- openstack-tox-docs http://logs.openstack.org/26/654826/2/check/openstack-tox-docs/b8b20e4/html/ : SUCCESS in 4m 55s\n- openstack-tox-functional http://logs.openstack.org/26/654826/2/check/openstack-tox-functional/2b15769/ : SUCCESS in 8m 08s\n- openstack-tox-functional-py35 http://logs.openstack.org/26/654826/2/check/openstack-tox-functional-py35/c0513db/ : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":2},{"id":"f7b76e23f077121ce97fa449242a353f541a54ca","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-04-24 20:07:50.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nI think it\u0027s fine to add this even if the spec you cite in the bug hasn\u0027t been implemented yet -- it\u0027s a legitimate hypervisor identifier.","accounts_in_message":[],"_revision_number":2},{"id":"1efdd150d57ad6f5b8e5e11473b89eb85b14ca27","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-04-24 20:13:49.000000000","message":"Patch Set 2: Code-Review-1\n\nIn addition to the two inline comments above, it looks like you\u0027re proposing the first metadefs change for Train, so you have the honor of beginning the Train metadefs release note.  You can use releasenotes/notes/rocky-metadefs-changes-cb00c006ff51b541.yaml as a model.","accounts_in_message":[],"_revision_number":2},{"id":"f39bd51a3d7cf28b0f3a81dd47a548f503421143","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-24 20:35:38.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"47950929f9b10936d22e746a4ce9edd06fa59670","author":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"date":"2019-04-25 00:18:41.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"b2bc037ada7e6e4bf288a75c2a7d2c29fc0a97d6","author":{"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"},"date":"2019-04-26 07:29:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6cbfc27cfaed9dde631df15b16d721e97ae71eaf","author":{"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"},"date":"2019-04-26 07:42:43.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"56580d4e0b913a9f3183104181f7167a091885cd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-26 08:10:18.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/26/654826/3/check-not-vote/ext-glance-zuul/d21ca00 : FAILURE in 40m 17s","accounts_in_message":[],"_revision_number":3},{"id":"218862f583e0d0462d2270b1bd71afed822e6f97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 10:06:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/26/654826/3/check/openstack-tox-lower-constraints/effa3fd/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/26/654826/3/check/openstack-tox-pep8/f3511c9/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/26/654826/3/check/openstack-tox-py27/6a5123d/ : SUCCESS in 4m 59s\n- openstack-tox-py35 http://logs.openstack.org/26/654826/3/check/openstack-tox-py35/35a9eea/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/26/654826/3/check/openstack-tox-py36/69df738/ : SUCCESS in 4m 21s\n- openstack-tox-py37 http://logs.openstack.org/26/654826/3/check/openstack-tox-py37/7c659ce/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/26/654826/3/check/openstack-tox-docs/4a4b01e/html/ : SUCCESS in 4m 58s\n- build-openstack-releasenotes http://logs.openstack.org/26/654826/3/check/build-openstack-releasenotes/10c3c63/html/ : SUCCESS in 8m 22s\n- openstack-tox-functional http://logs.openstack.org/26/654826/3/check/openstack-tox-functional/9fcb290/ : SUCCESS in 8m 21s\n- openstack-tox-functional-py35 http://logs.openstack.org/26/654826/3/check/openstack-tox-functional-py35/629c4be/ : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":3},{"id":"8638a9c230dc210570dd5f987d2b64ee8b3642d8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-04-26 12:46:10.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nI verified that the json file content is valid UTF-8 (in case anyone is worried about the (R) symbols).\n\nI\u0027d like you to revise the release note; see my comments inline.\n\nSorry I missed you in IRC, here are the Glance contributor guidelines about release notes: https://docs.openstack.org/glance/latest/contributor/release-notes.html\n\nI just noticed that the page linked above doesn\u0027t mention how the Glance team likes to handle release notes -- it would be great if you could submit a patch to update the docs (don\u0027t do it on this review, though, it should be a separate review).","accounts_in_message":[],"_revision_number":3},{"id":"77666c14fee0ae4f02ca7ccb61201391e37d79a6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-04-26 12:46:33.000000000","message":"Patch Set 3: Code-Review-1\n\nForgot to record my vote.  One more change requested.","accounts_in_message":[],"_revision_number":3},{"id":"7a6275ec4adfd936b0bc418ce69164b14ab8b189","author":{"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"},"date":"2019-04-29 06:32:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a741fa616e6c3303f985d7914bbc01a31a31da13","author":{"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"},"date":"2019-04-29 06:37:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3967b790d640a16edbeb21df9bd3a29959c41f69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 06:52:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/26/654826/4/check/openstack-tox-lower-constraints/0a254c8/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/26/654826/4/check/openstack-tox-pep8/6f4774c/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/26/654826/4/check/openstack-tox-py27/b66c4a6/ : SUCCESS in 4m 28s\n- openstack-tox-py35 http://logs.openstack.org/26/654826/4/check/openstack-tox-py35/008034a/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/26/654826/4/check/openstack-tox-py36/3c14830/ : SUCCESS in 4m 44s\n- openstack-tox-py37 http://logs.openstack.org/26/654826/4/check/openstack-tox-py37/7205a91/ : SUCCESS in 4m 26s\n- openstack-tox-docs http://logs.openstack.org/26/654826/4/check/openstack-tox-docs/e2917d4/html/ : SUCCESS in 5m 09s\n- build-openstack-releasenotes http://logs.openstack.org/26/654826/4/check/build-openstack-releasenotes/2a6a8f9/html/ : SUCCESS in 7m 35s\n- openstack-tox-functional http://logs.openstack.org/26/654826/4/check/openstack-tox-functional/83e247d/ : SUCCESS in 8m 28s\n- openstack-tox-functional-py35 http://logs.openstack.org/26/654826/4/check/openstack-tox-functional-py35/7214e34/ : SUCCESS in 8m 25s","accounts_in_message":[],"_revision_number":4},{"id":"28d1d0ec90e9d8aff96ad16400b5ab9d3a377d63","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-29 07:08:14.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/26/654826/4/check-not-vote/ext-glance-zuul/c0345ee : FAILURE in 35m 47s","accounts_in_message":[],"_revision_number":4},{"id":"ee498fec6bb9f8caa7465b3871e92b958f70608c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-30 18:04:05.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9bf1dcbc97e71efeed30b00c873978781f8bb15b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-05-01 20:48:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nNinja-approving as basically a docs change.","accounts_in_message":[],"_revision_number":4},{"id":"46ffc96fb74de909fb587a210c1077a406edcdc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 20:48:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b6070c83b19ba8aaa8cd5a9491d544b3a396c7f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 00:46:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/26/654826/4/gate/openstack-tox-lower-constraints/27a7646/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/26/654826/4/gate/openstack-tox-pep8/31ac37b/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/26/654826/4/gate/openstack-tox-py27/c554c0d/ : SUCCESS in 5m 01s\n- openstack-tox-py35 http://logs.openstack.org/26/654826/4/gate/openstack-tox-py35/c3f4334/ : SUCCESS in 5m 22s\n- openstack-tox-py36 http://logs.openstack.org/26/654826/4/gate/openstack-tox-py36/7a73bcb/ : SUCCESS in 4m 31s\n- openstack-tox-py37 http://logs.openstack.org/26/654826/4/gate/openstack-tox-py37/d44dc54/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/26/654826/4/gate/openstack-tox-docs/f0d2e24/html/ : SUCCESS in 5m 15s\n- build-openstack-releasenotes http://logs.openstack.org/26/654826/4/gate/build-openstack-releasenotes/a3fe62c/html/ : SUCCESS in 8m 43s\n- openstack-tox-functional http://logs.openstack.org/26/654826/4/gate/openstack-tox-functional/c711a2f/ : SUCCESS in 9m 07s\n- openstack-tox-functional-py35 http://logs.openstack.org/26/654826/4/gate/openstack-tox-functional-py35/bfdfea6/ : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":4},{"id":"7df31cc4236a9374f267f1fe7585ed60583f05c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 00:46:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ceba6db664661cfb51a520b0634789fbcc2eb051","revisions":{"fcd6b4eb65bdf50d524fb236861a1863799bbb35":{"kind":"REWORK","_number":1,"created":"2019-04-23 08:15:16.000000000","uploader":{"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"},"ref":"refs/changes/26/654826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/654826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/654826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/654826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/654826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/654826/1"}}},"commit":{"parents":[{"commit":"305efc5cb2555a23a6bb137eb13e7961f5acc858","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/305efc5cb2555a23a6bb137eb13e7961f5acc858"}]}],"author":{"name":"Ankit Arora","email":"aarora06@in.ibm.com","date":"2019-04-23 08:05:45.000000000","tz":-300},"committer":{"name":"Ankit Arora","email":"aarora06@in.ibm.com","date":"2019-04-23 08:05:45.000000000","tz":-300},"subject":"Added powervm as hypervisor type for image metadata prefiltering.","message":"Added powervm as hypervisor type for image metadata prefiltering.\n\nChange-Id: I7afd43c2e0581d30a78d5ed4e30f8eb2af1cf444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fcd6b4eb65bdf50d524fb236861a1863799bbb35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fcd6b4eb65bdf50d524fb236861a1863799bbb35"}]},"branch":"refs/heads/master"},"a7eb76610ca8687e85b12e3dbaa34e39471739b6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-24 08:59:46.000000000","uploader":{"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"},"ref":"refs/changes/26/654826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/654826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/654826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/654826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/654826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/654826/2"}}},"commit":{"parents":[{"commit":"305efc5cb2555a23a6bb137eb13e7961f5acc858","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/305efc5cb2555a23a6bb137eb13e7961f5acc858"}]}],"author":{"name":"Ankit Arora","email":"aarora06@in.ibm.com","date":"2019-04-23 08:05:45.000000000","tz":-300},"committer":{"name":"Ankit Arora","email":"aarora06@in.ibm.com","date":"2019-04-24 08:54:38.000000000","tz":-300},"subject":"Added powervm as hypervisor type for image metadata prefiltering.","message":"Added powervm as hypervisor type for image metadata prefiltering.\n\n\"powervm\" as hypervisor type is required for image metadata\nprefiltering in glance.\n\nCloses-Bug: #1826136\n\nChange-Id: I7afd43c2e0581d30a78d5ed4e30f8eb2af1cf444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7eb76610ca8687e85b12e3dbaa34e39471739b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7eb76610ca8687e85b12e3dbaa34e39471739b6"}]},"branch":"refs/heads/master"},"58621bc1f1ddf81cd9b7151dec8b34ae05ffb226":{"kind":"REWORK","_number":3,"created":"2019-04-26 07:29:42.000000000","uploader":{"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"},"ref":"refs/changes/26/654826/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/654826/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/654826/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/654826/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/654826/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/654826/3"}}},"commit":{"parents":[{"commit":"305efc5cb2555a23a6bb137eb13e7961f5acc858","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/305efc5cb2555a23a6bb137eb13e7961f5acc858"}]}],"author":{"name":"Ankit Arora","email":"aarora06@in.ibm.com","date":"2019-04-23 08:05:45.000000000","tz":-300},"committer":{"name":"Ankit Arora","email":"aarora06@in.ibm.com","date":"2019-04-26 07:22:45.000000000","tz":-300},"subject":"Added powervm as hypervisor type for image metadata prefiltering.","message":"Added powervm as hypervisor type for image metadata prefiltering.\n\n\"powervm\" as hypervisor type is required for image metadata\nprefiltering in glance.\n\nCloses-Bug: #1826136\n\nChange-Id: I7afd43c2e0581d30a78d5ed4e30f8eb2af1cf444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/58621bc1f1ddf81cd9b7151dec8b34ae05ffb226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/58621bc1f1ddf81cd9b7151dec8b34ae05ffb226"}]},"branch":"refs/heads/master"},"ceba6db664661cfb51a520b0634789fbcc2eb051":{"kind":"REWORK","_number":4,"created":"2019-04-29 06:32:07.000000000","uploader":{"_account_id":26575,"name":"Ankit Arora","email":"aarora06@in.ibm.com","username":"aarora06"},"ref":"refs/changes/26/654826/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/654826/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/654826/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/654826/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/654826/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/654826/4"}}},"commit":{"parents":[{"commit":"305efc5cb2555a23a6bb137eb13e7961f5acc858","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/305efc5cb2555a23a6bb137eb13e7961f5acc858"}]}],"author":{"name":"Ankit Arora","email":"aarora06@in.ibm.com","date":"2019-04-23 08:05:45.000000000","tz":-300},"committer":{"name":"Ankit Arora","email":"aarora06@in.ibm.com","date":"2019-04-29 06:31:07.000000000","tz":-300},"subject":"Added powervm as hypervisor type for image metadata prefiltering.","message":"Added powervm as hypervisor type for image metadata prefiltering.\n\n\"powervm\" as hypervisor type is required for image metadata\nprefiltering in glance.\n\nCloses-Bug: #1826136\n\nChange-Id: I7afd43c2e0581d30a78d5ed4e30f8eb2af1cf444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ceba6db664661cfb51a520b0634789fbcc2eb051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ceba6db664661cfb51a520b0634789fbcc2eb051"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
