)]}'
{"id":"openstack%2Fglance~656472","triplet_id":"openstack%2Fglance~master~I407503ee2f78c744c699bd7759c44f3b73d343ce","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"21284":{"account":{"_account_id":21284,"name":"Pavel Gluschak (scsnow)","email":"pglushchak@virtuozzo.com","username":"scsnow"},"last_update":"2022-07-12 15:30:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I407503ee2f78c744c699bd7759c44f3b73d343ce","subject":"Fixed upload_image policy doesn\u0027t support params","status":"ABANDONED","created":"2019-04-30 08:55:10.000000000","updated":"2022-07-12 15:30:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8d4b18f820348342bf23dbfba8f16ccd07de5524","_number":656472,"virtual_id_number":656472,"owner":{"_account_id":21284,"name":"Pavel Gluschak (scsnow)","email":"pglushchak@virtuozzo.com","username":"scsnow"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-04-30 09:31:45.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2019-05-09 09:46:00.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":1,"date":"2019-04-30 10:52:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2019-05-08 12:59:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":1,"date":"2019-05-11 15:25:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":1,"date":"2019-05-09 02:37:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-30 09:31:45.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 10:52:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-08 12:59:40.000000000","updated_by":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"reviewer":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"state":"REVIEWER"},{"updated":"2019-05-09 02:37:41.000000000","updated_by":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"reviewer":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"state":"REVIEWER"},{"updated":"2019-05-09 09:46:00.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-05-11 15:25:52.000000000","updated_by":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"reviewer":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"state":"REVIEWER"},{"updated":"2022-07-12 15:30:30.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"}],"messages":[{"id":"fca270683e5ea5de4d4ae1028b3d65b769809d11","author":{"_account_id":21284,"name":"Pavel Gluschak (scsnow)","email":"pglushchak@virtuozzo.com","username":"scsnow"},"date":"2019-04-30 08:55:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"425964f15687207a594bbe032dc0cb85b0a27e63","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-30 09:31:45.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/72/656472/1/check-not-vote/ext-glance-zuul/e0b098f : FAILURE in 36m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2b18b06afabef489f15249abab260464aa7e9675","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 10:52:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/72/656472/1/check/tempest-full/d93c9d7/ : SUCCESS in 1h 55m 47s\n- neutron-grenade http://logs.openstack.org/72/656472/1/check/neutron-grenade/960685b/ : SUCCESS in 1h 03m 01s\n- grenade-py3 http://logs.openstack.org/72/656472/1/check/grenade-py3/c9c710d/ : SUCCESS in 1h 10m 11s\n- tempest-full-py3 http://logs.openstack.org/72/656472/1/check/tempest-full-py3/a9a89be/ : SUCCESS in 1h 30m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/656472/1/check/openstack-tox-lower-constraints/ce49b1c/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/72/656472/1/check/openstack-tox-pep8/866626d/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/72/656472/1/check/openstack-tox-py27/5f19820/ : SUCCESS in 4m 13s\n- openstack-tox-py35 http://logs.openstack.org/72/656472/1/check/openstack-tox-py35/5fecefe/ : SUCCESS in 4m 38s\n- openstack-tox-py36 http://logs.openstack.org/72/656472/1/check/openstack-tox-py36/e6a9a0f/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/72/656472/1/check/openstack-tox-py37/38bfb45/ : SUCCESS in 5m 11s\n- openstack-tox-docs http://logs.openstack.org/72/656472/1/check/openstack-tox-docs/fd8fe99/html/ : SUCCESS in 4m 27s\n- openstack-tox-functional http://logs.openstack.org/72/656472/1/check/openstack-tox-functional/a9ffa80/ : SUCCESS in 8m 05s\n- openstack-tox-functional-py35 http://logs.openstack.org/72/656472/1/check/openstack-tox-functional-py35/9fa06c1/ : SUCCESS in 12m 52s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/72/656472/1/check/glance-eventlet-ssl-handshake-broken-py35/219eea5/ : FAILURE in 2m 24s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/72/656472/1/check/glance-code-constants-check/750cd99/ : SUCCESS in 4m 33s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/72/656472/1/check/devstack-plugin-ceph-tempest/2f5d539/ : SUCCESS in 1h 38m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7de701382b7e0903409b3667c46de62f977348f","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2019-05-08 12:59:40.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"c6acb0fb2da817f84437cc43c70b780690d7b5da","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-05-09 02:37:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9f924ce381a6c0a0ec44110eec13f52e8842d98a","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-05-09 09:46:00.000000000","message":"Patch Set 1:\n\nI really don\u0027t understand why one would want to make such policy. Glance already allows only the owner to upload the data to the queued image.\n\nCould you please open a bug for this to explain the context bit better and if it really needs fixing, we can then target it to the stable branches as well.","accounts_in_message":[],"_revision_number":1},{"id":"48912654413e2862deb69cc330a12ccb20390394","author":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"date":"2019-05-11 15:25:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8d4b18f820348342bf23dbfba8f16ccd07de5524","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-12 15:30:30.000000000","message":"Abandoned\n\nA similar change was applied in 31414b9f61ac4c923ede30a3c7e32662e9c14e1a and the code has changed a lot since.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"639f0003d4d25623a1e8de6588961dea0c8b371c","revisions":{"639f0003d4d25623a1e8de6588961dea0c8b371c":{"kind":"REWORK","_number":1,"created":"2019-04-30 08:55:10.000000000","uploader":{"_account_id":21284,"name":"Pavel Gluschak (scsnow)","email":"pglushchak@virtuozzo.com","username":"scsnow"},"ref":"refs/changes/72/656472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/656472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/656472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/656472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/656472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/656472/1"}}},"commit":{"parents":[{"commit":"fa13302a58e8c3f913e569d38522bdc9288c94d5","subject":"Merge \"Switch to using stestr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa13302a58e8c3f913e569d38522bdc9288c94d5"}]}],"author":{"name":"Pavel Glushchak","email":"pglushchak@virtuozzo.com","date":"2019-04-30 08:36:41.000000000","tz":180},"committer":{"name":"Pavel Glushchak","email":"pglushchak@virtuozzo.com","date":"2019-04-30 08:53:08.000000000","tz":180},"subject":"Fixed upload_image policy doesn\u0027t support params","message":"Fixed upload_image policy doesn\u0027t support params\n\nOtherwise following policy won\u0027t work:\n\"upload_image\": \"rule:context_is_admin or tenant:%(owner)s\"\n\nChange-Id: I407503ee2f78c744c699bd7759c44f3b73d343ce\nSigned-off-by: Pavel Glushchak \u003cpglushchak@virtuozzo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/639f0003d4d25623a1e8de6588961dea0c8b371c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/639f0003d4d25623a1e8de6588961dea0c8b371c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
