)]}'
{"id":"openstack%2Fglance~657966","triplet_id":"openstack%2Fglance~master~I37d881315e801926bd5a7ae46402038fd69e060b","project":"openstack/glance","branch":"master","topic":"py3-ssl","hashtags":[],"change_id":"I37d881315e801926bd5a7ae46402038fd69e060b","subject":"Don\u0027t target broken ssl tests to specific py3.minor","status":"MERGED","created":"2019-05-09 09:57:18.000000000","updated":"2019-06-07 17:24:45.000000000","submitted":"2019-06-07 17:24:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657966-1559928285600-2d20edbd","meta_rev_id":"bd3b11c0a72bd7a26aac7318ea530cf2ede295a2","_number":657966,"virtual_id_number":657966,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-07 16:10:29.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2019-05-16 11:20:46.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":2,"date":"2019-06-07 17:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-03 18:25:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2019-05-30 06:09:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":2,"date":"2019-05-30 04:34:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-03 18:25:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 03:03:01.000000000","updated_by":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"reviewer":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"state":"REVIEWER"},{"updated":"2019-05-11 15:25:42.000000000","updated_by":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"reviewer":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"state":"REVIEWER"},{"updated":"2019-05-30 04:34:31.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-05-30 06:09:22.000000000","updated_by":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"reviewer":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"state":"REVIEWER"},{"updated":"2019-06-03 18:25:27.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-06-07 16:10:29.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 17:24:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f80d9bef4177ce4ca1207550abd8fe10cf0f089f","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-05-09 09:57:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"554bc448843bf9678eb3d1a60e2d2cca32215585","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-09 10:34:00.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/66/657966/1/check-not-vote/ext-glance-zuul/0d490c1 : FAILURE in 36m 20s","accounts_in_message":[],"_revision_number":1},{"id":"319356e80e7ef429ac2a17ef8c45602c0c2e0310","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 12:29:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/66/657966/1/check/tempest-full/9116b86/ : TIMED_OUT in 2h 06m 08s\n- neutron-grenade http://logs.openstack.org/66/657966/1/check/neutron-grenade/2dfe07c/ : SUCCESS in 1h 04m 43s\n- grenade-py3 http://logs.openstack.org/66/657966/1/check/grenade-py3/424c1dc/ : SUCCESS in 54m 48s\n- tempest-full-py3 http://logs.openstack.org/66/657966/1/check/tempest-full-py3/6a3a28c/ : SUCCESS in 1h 34m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/657966/1/check/openstack-tox-lower-constraints/0edaf4b/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/66/657966/1/check/openstack-tox-pep8/9c77544/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/66/657966/1/check/openstack-tox-py27/dc1485e/ : SUCCESS in 5m 53s\n- openstack-tox-py35 http://logs.openstack.org/66/657966/1/check/openstack-tox-py35/a54e06b/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/66/657966/1/check/openstack-tox-py36/6220cb5/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/66/657966/1/check/openstack-tox-py37/351acb2/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/66/657966/1/check/openstack-tox-docs/85027b7/html/ : SUCCESS in 4m 44s\n- build-openstack-releasenotes http://logs.openstack.org/66/657966/1/check/build-openstack-releasenotes/51f676c/html/ : SUCCESS in 7m 57s\n- openstack-tox-functional http://logs.openstack.org/66/657966/1/check/openstack-tox-functional/c09baf1/ : SUCCESS in 8m 03s\n- openstack-tox-functional-py35 http://logs.openstack.org/66/657966/1/check/openstack-tox-functional-py35/77964cf/ : SUCCESS in 9m 56s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/66/657966/1/check/glance-eventlet-ssl-handshake-broken-py3/f8b9f1d/ : FAILURE in 3m 52s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/66/657966/1/check/glance-code-constants-check/146b0bf/ : SUCCESS in 4m 05s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/66/657966/1/check/devstack-plugin-ceph-tempest/db50df2/ : SUCCESS in 1h 31m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6fdd18510349bf6c898eadf5e9724ea9bdf4f77c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-05-10 01:09:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"040fb798f8f3ff8e22e3e7856cd5c6fba9f52129","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 02:59:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/66/657966/1/check/tempest-full/d385b25/ : SUCCESS in 1h 44m 03s\n- neutron-grenade http://logs.openstack.org/66/657966/1/check/neutron-grenade/c18225c/ : SUCCESS in 55m 32s\n- grenade-py3 http://logs.openstack.org/66/657966/1/check/grenade-py3/3e49d61/ : SUCCESS in 58m 15s\n- tempest-full-py3 http://logs.openstack.org/66/657966/1/check/tempest-full-py3/16af480/ : SUCCESS in 1h 25m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/657966/1/check/openstack-tox-lower-constraints/cbe9424/ : SUCCESS in 3m 53s\n- openstack-tox-pep8 http://logs.openstack.org/66/657966/1/check/openstack-tox-pep8/81c2e69/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/66/657966/1/check/openstack-tox-py27/925cc8b/ : SUCCESS in 3m 58s\n- openstack-tox-py35 http://logs.openstack.org/66/657966/1/check/openstack-tox-py35/b0c9fc5/ : SUCCESS in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/66/657966/1/check/openstack-tox-py36/8f12e8c/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/66/657966/1/check/openstack-tox-py37/2787e4e/ : SUCCESS in 6m 06s\n- openstack-tox-docs http://logs.openstack.org/66/657966/1/check/openstack-tox-docs/7eff199/html/ : SUCCESS in 4m 27s\n- build-openstack-releasenotes http://logs.openstack.org/66/657966/1/check/build-openstack-releasenotes/e42a9d0/html/ : SUCCESS in 7m 59s\n- openstack-tox-functional http://logs.openstack.org/66/657966/1/check/openstack-tox-functional/1147583/ : SUCCESS in 8m 06s\n- openstack-tox-functional-py35 http://logs.openstack.org/66/657966/1/check/openstack-tox-functional-py35/18181d1/ : SUCCESS in 8m 42s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/66/657966/1/check/glance-eventlet-ssl-handshake-broken-py3/cdd99f4/ : FAILURE in 4m 00s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/66/657966/1/check/glance-code-constants-check/2df3dea/ : SUCCESS in 3m 45s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/66/657966/1/check/devstack-plugin-ceph-tempest/c3c406e/ : SUCCESS in 1h 37m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db1f5c77815a2bd7580c10a8377820553762286c","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-05-10 03:03:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f5f89ab5910ae5f352b2892f75b9de0ffa631b75","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-05-10 12:14:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nMakes sense to me.","accounts_in_message":[],"_revision_number":1},{"id":"32e97f6bc0fb304ef51fb4ccae18098451a50fef","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-10 15:12:19.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/66/657966/1/check-not-vote/ext-glance-zuul/f49dba5 : FAILURE in 36m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a4d4047c4397d99a40eb92ceadcc877f61ea8639","author":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"date":"2019-05-11 15:25:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d7614da25f68b72f2c45604edf5f25eb6e4d0e73","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-05-14 11:34:13.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2b7e06e2ab0ac4124324f844fb0762863b4e1d3a","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-05-14 11:34:28.000000000","message":"Patch Set 2:\n\nchanged the comments for clarity.","accounts_in_message":[],"_revision_number":2},{"id":"56ade7cdd1c45451c554cd8dbeaa382bca97f15f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 13:20:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/66/657966/2/check/tempest-full/63c9284/ : SUCCESS in 1h 28m 17s\n- neutron-grenade http://logs.openstack.org/66/657966/2/check/neutron-grenade/6b733af/ : FAILURE in 25m 31s\n- grenade-py3 http://logs.openstack.org/66/657966/2/check/grenade-py3/6801a3d/ : FAILURE in 25m 47s\n- tempest-full-py3 http://logs.openstack.org/66/657966/2/check/tempest-full-py3/1b39a36/ : SUCCESS in 1h 26m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/657966/2/check/openstack-tox-lower-constraints/8000f17/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/66/657966/2/check/openstack-tox-pep8/f6eb6c7/ : SUCCESS in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/66/657966/2/check/openstack-tox-py27/9f10273/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/66/657966/2/check/openstack-tox-py35/5c1e7f1/ : SUCCESS in 4m 04s\n- openstack-tox-py36 http://logs.openstack.org/66/657966/2/check/openstack-tox-py36/fd4ee6a/ : SUCCESS in 4m 16s\n- openstack-tox-py37 http://logs.openstack.org/66/657966/2/check/openstack-tox-py37/1d13f82/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/66/657966/2/check/openstack-tox-docs/cccc119/html/ : SUCCESS in 4m 47s\n- build-openstack-releasenotes http://logs.openstack.org/66/657966/2/check/build-openstack-releasenotes/8f2e337/html/ : SUCCESS in 8m 21s\n- openstack-tox-functional http://logs.openstack.org/66/657966/2/check/openstack-tox-functional/be14e42/ : SUCCESS in 8m 01s\n- openstack-tox-functional-py35 http://logs.openstack.org/66/657966/2/check/openstack-tox-functional-py35/eca7bba/ : SUCCESS in 8m 26s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/66/657966/2/check/glance-eventlet-ssl-handshake-broken-py3/7966871/ : FAILURE in 3m 57s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/66/657966/2/check/glance-code-constants-check/8d9be74/ : SUCCESS in 3m 31s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/66/657966/2/check/devstack-plugin-ceph-tempest/4d9ef48/ : SUCCESS in 1h 40m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8fd7e6a6c1aa09499439fc12c060a68d8d2e6d84","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-14 20:49:49.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/66/657966/2/check-not-vote/ext-glance-zuul/d35c46d : FAILURE in 22m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1a830eb11bf857fb1e3d6d8067ed3a6fb01d103e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-05-16 11:20:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e14ae6e8506ab0184ae345423f977cda496f092f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 13:07:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/66/657966/2/check/tempest-full/c3ae20c/ : SUCCESS in 1h 44m 24s\n- neutron-grenade http://logs.openstack.org/66/657966/2/check/neutron-grenade/dbf1efd/ : SUCCESS in 51m 05s\n- grenade-py3 http://logs.openstack.org/66/657966/2/check/grenade-py3/eaf5395/ : SUCCESS in 59m 29s\n- tempest-full-py3 http://logs.openstack.org/66/657966/2/check/tempest-full-py3/ed51417/ : SUCCESS in 1h 15m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/657966/2/check/openstack-tox-lower-constraints/4e687c0/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/66/657966/2/check/openstack-tox-pep8/40382ec/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/66/657966/2/check/openstack-tox-py27/6bdd6d3/ : SUCCESS in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/66/657966/2/check/openstack-tox-py35/c0f4622/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/66/657966/2/check/openstack-tox-py36/15ded68/ : SUCCESS in 3m 59s\n- openstack-tox-py37 http://logs.openstack.org/66/657966/2/check/openstack-tox-py37/7543d8e/ : SUCCESS in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/66/657966/2/check/openstack-tox-docs/baa494e/html/ : SUCCESS in 4m 52s\n- build-openstack-releasenotes http://logs.openstack.org/66/657966/2/check/build-openstack-releasenotes/d321a66/html/ : SUCCESS in 6m 56s\n- openstack-tox-functional http://logs.openstack.org/66/657966/2/check/openstack-tox-functional/8617b0c/ : SUCCESS in 8m 02s\n- openstack-tox-functional-py35 http://logs.openstack.org/66/657966/2/check/openstack-tox-functional-py35/43338eb/ : SUCCESS in 7m 43s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/66/657966/2/check/glance-eventlet-ssl-handshake-broken-py3/33571b6/ : FAILURE in 3m 33s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/66/657966/2/check/glance-code-constants-check/a0dea9b/ : SUCCESS in 4m 15s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/66/657966/2/check/devstack-plugin-ceph-tempest/dfeb53a/ : FAILURE in 1h 44m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b6d51358ab4bfb248b7a508d697e3cb11c7de534","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-16 21:09:38.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/66/657966/2/check-not-vote/ext-glance-zuul/b7d5502 : FAILURE in 24m 07s","accounts_in_message":[],"_revision_number":2},{"id":"76bfcec70147b40f8eedd7af373931b68c5fd47d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-30 04:34:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7997a6b28b125e28b49f410478c602a381bf720a","author":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"date":"2019-05-30 06:09:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9ca5acc3acd0e8d3ee5378cc09a4caefe5cb478f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-06-03 18:25:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLooks good.","accounts_in_message":[],"_revision_number":2},{"id":"a237bd5d17ecf3778aaf26e64905450849335ea8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 18:25:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"25ed40cf3010b60c52551c539720c3444ecb4729","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 20:35:37.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/66/657966/2/gate/build-openstack-api-ref/c722a5d/html/ : SUCCESS in 3m 23s\n- tempest-full http://logs.openstack.org/66/657966/2/gate/tempest-full/3a4c11a/ : TIMED_OUT in 2h 07m 44s\n- neutron-grenade http://logs.openstack.org/66/657966/2/gate/neutron-grenade/ab6d291/ : SUCCESS in 58m 45s\n- grenade-py3 http://logs.openstack.org/66/657966/2/gate/grenade-py3/3be181c/ : SUCCESS in 58m 20s\n- tempest-full-py3 http://logs.openstack.org/66/657966/2/gate/tempest-full-py3/c0c8309/ : SUCCESS in 1h 33m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/657966/2/gate/openstack-tox-lower-constraints/b686e70/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/66/657966/2/gate/openstack-tox-pep8/e5e7b3d/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/66/657966/2/gate/openstack-tox-py27/fcb04eb/ : SUCCESS in 4m 06s\n- openstack-tox-py35 http://logs.openstack.org/66/657966/2/gate/openstack-tox-py35/9caa829/ : SUCCESS in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/66/657966/2/gate/openstack-tox-py36/fa3aa09/ : SUCCESS in 4m 02s\n- openstack-tox-py37 http://logs.openstack.org/66/657966/2/gate/openstack-tox-py37/e7494e4/ : SUCCESS in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/66/657966/2/gate/openstack-tox-docs/cf8ac9b/html/ : SUCCESS in 4m 21s\n- build-openstack-releasenotes http://logs.openstack.org/66/657966/2/gate/build-openstack-releasenotes/dab21d0/html/ : SUCCESS in 7m 59s\n- openstack-tox-functional http://logs.openstack.org/66/657966/2/gate/openstack-tox-functional/0db3c54/ : SUCCESS in 13m 49s\n- openstack-tox-functional-py35 http://logs.openstack.org/66/657966/2/gate/openstack-tox-functional-py35/3be2ce3/ : TIMED_OUT in 30m 57s","accounts_in_message":[],"_revision_number":2},{"id":"b63a0085ba717857d74d3a574cfaea437b06e311","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-06-07 12:29:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0ab8133896d1d09d95bc89c8b1ab153e9b86bdeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 14:17:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/66/657966/2/check/build-openstack-api-ref/70201ae/html/ : SUCCESS in 3m 29s\n- tempest-full http://logs.openstack.org/66/657966/2/check/tempest-full/24f1dd8/ : SUCCESS in 1h 40m 40s\n- neutron-grenade http://logs.openstack.org/66/657966/2/check/neutron-grenade/35c5038/ : SUCCESS in 58m 46s\n- grenade-py3 http://logs.openstack.org/66/657966/2/check/grenade-py3/e293780/ : SUCCESS in 59m 11s\n- tempest-full-py3 http://logs.openstack.org/66/657966/2/check/tempest-full-py3/08206e6/ : SUCCESS in 1h 40m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/657966/2/check/openstack-tox-lower-constraints/b4af4ee/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/66/657966/2/check/openstack-tox-pep8/3452f65/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/66/657966/2/check/openstack-tox-py27/f5acd11/ : SUCCESS in 4m 19s\n- openstack-tox-py35 http://logs.openstack.org/66/657966/2/check/openstack-tox-py35/fc8104a/ : SUCCESS in 4m 17s\n- openstack-tox-py36 http://logs.openstack.org/66/657966/2/check/openstack-tox-py36/5df5a7d/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/66/657966/2/check/openstack-tox-py37/47992ee/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/66/657966/2/check/openstack-tox-docs/83613eb/html/ : SUCCESS in 4m 50s\n- build-openstack-releasenotes http://logs.openstack.org/66/657966/2/check/build-openstack-releasenotes/5eed51f/html/ : SUCCESS in 8m 18s\n- openstack-tox-functional http://logs.openstack.org/66/657966/2/check/openstack-tox-functional/f96db4a/ : SUCCESS in 11m 34s\n- openstack-tox-functional-py35 http://logs.openstack.org/66/657966/2/check/openstack-tox-functional-py35/0c8c719/ : SUCCESS in 9m 58s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/66/657966/2/check/glance-eventlet-ssl-handshake-broken-py3/c696763/ : FAILURE in 4m 26s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/66/657966/2/check/glance-code-constants-check/03f75d9/ : SUCCESS in 4m 23s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/66/657966/2/check/devstack-plugin-ceph-tempest/6047f6b/ : SUCCESS in 1h 24m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e18d88f529d09fd7c8a982b8f256faea83d57c26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 14:17:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"efb64cd6ba909b1d9183739faf784a760a7c6ce2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-07 16:10:29.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/66/657966/2/check-not-vote/ext-glance-zuul/7b94bba : SUCCESS in 38m 12s","accounts_in_message":[],"_revision_number":2},{"id":"aaacab40fae4ee598d3d9dfb484b56fa41fd4ca4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 17:24:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/66/657966/2/gate/build-openstack-api-ref/cd08ca2/html/ : SUCCESS in 3m 37s\n- tempest-full http://logs.openstack.org/66/657966/2/gate/tempest-full/e5f0c87/ : SUCCESS in 1h 43m 53s\n- neutron-grenade http://logs.openstack.org/66/657966/2/gate/neutron-grenade/cb85ee5/ : SUCCESS in 57m 37s\n- grenade-py3 http://logs.openstack.org/66/657966/2/gate/grenade-py3/494cf14/ : SUCCESS in 1h 09m 37s\n- tempest-full-py3 http://logs.openstack.org/66/657966/2/gate/tempest-full-py3/f31e209/ : SUCCESS in 1h 24m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/657966/2/gate/openstack-tox-lower-constraints/ca4eb6a/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/66/657966/2/gate/openstack-tox-pep8/e1a77bf/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/66/657966/2/gate/openstack-tox-py27/135af5a/ : SUCCESS in 5m 08s\n- openstack-tox-py35 http://logs.openstack.org/66/657966/2/gate/openstack-tox-py35/1951583/ : SUCCESS in 5m 37s\n- openstack-tox-py36 http://logs.openstack.org/66/657966/2/gate/openstack-tox-py36/b8dbce8/ : SUCCESS in 4m 48s\n- openstack-tox-py37 http://logs.openstack.org/66/657966/2/gate/openstack-tox-py37/bc44aa5/ : SUCCESS in 6m 27s\n- openstack-tox-docs http://logs.openstack.org/66/657966/2/gate/openstack-tox-docs/0ce7071/html/ : SUCCESS in 5m 14s\n- build-openstack-releasenotes http://logs.openstack.org/66/657966/2/gate/build-openstack-releasenotes/78358ce/html/ : SUCCESS in 8m 26s\n- openstack-tox-functional http://logs.openstack.org/66/657966/2/gate/openstack-tox-functional/893207f/ : SUCCESS in 8m 28s\n- openstack-tox-functional-py35 http://logs.openstack.org/66/657966/2/gate/openstack-tox-functional-py35/4b147fe/ : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":2},{"id":"d1a52b8fab8d1ae75c7f4410f91643eb26c7d381","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 17:24:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8aea36403d80fc5b278e5398bf539e0cee2bda6a","revisions":{"e0792bbe108e6971830ff57ee28876c3dcbfd6b1":{"kind":"REWORK","_number":1,"created":"2019-05-09 09:57:18.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/66/657966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/657966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/657966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/657966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/657966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/657966/1"}}},"commit":{"parents":[{"commit":"61298a47e4abd903fe38957ffbbbcf036e79bb3e","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61298a47e4abd903fe38957ffbbbcf036e79bb3e"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-05-09 09:56:53.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-05-09 09:56:53.000000000","tz":60},"subject":"Don\u0027t target broken ssl tests to specific py3.minor","message":"Don\u0027t target broken ssl tests to specific py3.minor\n\nChange-Id: I37d881315e801926bd5a7ae46402038fd69e060b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0792bbe108e6971830ff57ee28876c3dcbfd6b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0792bbe108e6971830ff57ee28876c3dcbfd6b1"}]},"branch":"refs/heads/master"},"8aea36403d80fc5b278e5398bf539e0cee2bda6a":{"kind":"REWORK","_number":2,"created":"2019-05-14 11:34:13.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/66/657966/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/657966/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/657966/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/657966/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/657966/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/657966/2"}}},"commit":{"parents":[{"commit":"61298a47e4abd903fe38957ffbbbcf036e79bb3e","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61298a47e4abd903fe38957ffbbbcf036e79bb3e"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-05-09 09:56:53.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-05-14 11:34:04.000000000","tz":0},"subject":"Don\u0027t target broken ssl tests to specific py3.minor","message":"Don\u0027t target broken ssl tests to specific py3.minor\n\nChange-Id: I37d881315e801926bd5a7ae46402038fd69e060b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8aea36403d80fc5b278e5398bf539e0cee2bda6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8aea36403d80fc5b278e5398bf539e0cee2bda6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
