)]}'
{"id":"openstack%2Fglance~658714","triplet_id":"openstack%2Fglance~master~I75ac27ea78bde86651e1b3c0920b194b5c5a510f","project":"openstack/glance","branch":"master","topic":"broken_url","hashtags":[],"change_id":"I75ac27ea78bde86651e1b3c0920b194b5c5a510f","subject":"Change some URLs to point to more accurate targets","status":"ABANDONED","created":"2019-05-13 02:13:24.000000000","updated":"2022-07-05 19:05:37.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6b5289b21bd528857b3810a7e73a22b3264538e7","_number":658714,"virtual_id_number":658714,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":1,"date":"2019-06-12 10:39:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-08-09 02:17:35.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-07 12:21:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 03:11:35.000000000","updated_by":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"reviewer":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"state":"REVIEWER"},{"updated":"2019-05-14 15:07:57.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-12 10:39:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-07 12:21:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"76f15d446db941e71094740eeead71798304fae5","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-05-13 02:13:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be42f4b1652ae9ce4cc43353c0f784fe221aca20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 02:27:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/14/658714/1/check/openstack-tox-lower-constraints/9b3f4d5/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/14/658714/1/check/openstack-tox-pep8/c14523d/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/14/658714/1/check/openstack-tox-py27/4eedec9/ : SUCCESS in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/14/658714/1/check/openstack-tox-py35/392dae3/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/14/658714/1/check/openstack-tox-py36/a97967b/ : SUCCESS in 3m 53s\n- openstack-tox-py37 http://logs.openstack.org/14/658714/1/check/openstack-tox-py37/b74895b/ : SUCCESS in 4m 06s\n- openstack-tox-docs http://logs.openstack.org/14/658714/1/check/openstack-tox-docs/6ad163d/html/ : SUCCESS in 6m 18s\n- openstack-tox-functional http://logs.openstack.org/14/658714/1/check/openstack-tox-functional/e3c75d8/ : SUCCESS in 8m 18s\n- openstack-tox-functional-py35 http://logs.openstack.org/14/658714/1/check/openstack-tox-functional-py35/c0d254c/ : SUCCESS in 8m 45s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/14/658714/1/check/glance-eventlet-ssl-handshake-broken-py35/0b931da/ : FAILURE in 2m 36s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/14/658714/1/check/glance-code-constants-check/eebbb77/ : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a7ca1545fdfa1d5902e6aab514efe927d4f50e33","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-13 02:45:30.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/14/658714/1/check-not-vote/ext-glance-zuul/fe2d429 : FAILURE in 31m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e8f9a590e0ee45fdd72916a29971b92351b615e0","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-05-13 03:11:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"68af2563fe4f749e4ffded62631d199ea8d8f25c","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-05-14 03:15:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61fe53f02caf071baab52e3fe9906d2770084ded","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 03:30:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/14/658714/2/check/openstack-tox-lower-constraints/0646510/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/14/658714/2/check/openstack-tox-pep8/e504c25/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/14/658714/2/check/openstack-tox-py27/2e7cac9/ : SUCCESS in 4m 17s\n- openstack-tox-py35 http://logs.openstack.org/14/658714/2/check/openstack-tox-py35/fefec2d/ : SUCCESS in 4m 27s\n- openstack-tox-py36 http://logs.openstack.org/14/658714/2/check/openstack-tox-py36/e6532b6/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/14/658714/2/check/openstack-tox-py37/5f3c7f1/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/14/658714/2/check/openstack-tox-docs/be8d182/html/ : SUCCESS in 4m 48s\n- openstack-tox-functional http://logs.openstack.org/14/658714/2/check/openstack-tox-functional/9b9ba5d/ : SUCCESS in 8m 57s\n- openstack-tox-functional-py35 http://logs.openstack.org/14/658714/2/check/openstack-tox-functional-py35/a3a9e17/ : SUCCESS in 9m 35s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/14/658714/2/check/glance-eventlet-ssl-handshake-broken-py35/1996bca/ : FAILURE in 3m 05s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/14/658714/2/check/glance-code-constants-check/3eb1323/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"64ac25067e6d7cf10b9416ba82f430147148b892","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-14 15:07:57.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/14/658714/2/check-not-vote/ext-glance-zuul/949d97e : FAILURE in 27m 22s","accounts_in_message":[],"_revision_number":2},{"id":"dd0b47b1ce818ca4fc4d63fb86364d60ffc6249c","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-06-12 01:28:26.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e33482775aa92d39abd568c31e42f074e2b8518b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 10:39:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/14/658714/3/check/openstack-tox-pep8/62db0a1/ : SUCCESS in 4m 06s\n- openstack-tox-docs http://logs.openstack.org/14/658714/3/check/openstack-tox-docs/ab65762/html/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":3},{"id":"7069508ee64e1279d3ebc3abcdcde53798db4c24","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-06-24 09:17:40.000000000","message":"Patch Set 3:\n\nany opinion?","accounts_in_message":[],"_revision_number":3},{"id":"e526fc81fe6c25786212b380bc8c062160aee1f9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-07 12:21:10.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nLooks fine to me, but it would be great if you could expand this to cover all of the little updates.","accounts_in_message":[],"_revision_number":3},{"id":"26bd0436a0893fdf48391f3caa23ad4f8b668022","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-08-09 02:17:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6b5289b21bd528857b3810a7e73a22b3264538e7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-05 19:05:37.000000000","message":"Abandoned\n\nIn favor of https://review.opendev.org/c/openstack/glance/+/848771","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"51db77b5f2d41e34ae7bf0cb62fd8ec5790957b9","revisions":{"ad4d179b7b3204b636174c75ad6acfbbf48b0b92":{"kind":"REWORK","_number":1,"created":"2019-05-13 02:13:24.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/14/658714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/658714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/658714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/658714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/658714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/658714/1"}}},"commit":{"parents":[{"commit":"61298a47e4abd903fe38957ffbbbcf036e79bb3e","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61298a47e4abd903fe38957ffbbbcf036e79bb3e"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-05-13 02:09:11.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-05-13 02:09:11.000000000","tz":480},"subject":"Change some URLs to point to more accurate targets","message":"Change some URLs to point to more accurate targets\n\nThis is to change some URLs to point to more accurate targets.\n\nChange-Id: I75ac27ea78bde86651e1b3c0920b194b5c5a510f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad4d179b7b3204b636174c75ad6acfbbf48b0b92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad4d179b7b3204b636174c75ad6acfbbf48b0b92"}]},"branch":"refs/heads/master"},"5dbe9cc7a8a76d041534dda04d9064f4519fa384":{"kind":"REWORK","_number":2,"created":"2019-05-14 03:15:22.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/14/658714/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/658714/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/658714/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/658714/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/658714/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/658714/2"}}},"commit":{"parents":[{"commit":"61298a47e4abd903fe38957ffbbbcf036e79bb3e","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61298a47e4abd903fe38957ffbbbcf036e79bb3e"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-05-13 02:09:11.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-05-14 03:12:51.000000000","tz":480},"subject":"Change some URLs to point to more accurate targets","message":"Change some URLs to point to more accurate targets\n\nThis is to change some URLs to point to more accurate targets.\n\nChange-Id: I75ac27ea78bde86651e1b3c0920b194b5c5a510f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5dbe9cc7a8a76d041534dda04d9064f4519fa384"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5dbe9cc7a8a76d041534dda04d9064f4519fa384"}]},"branch":"refs/heads/master"},"51db77b5f2d41e34ae7bf0cb62fd8ec5790957b9":{"kind":"REWORK","_number":3,"created":"2019-06-12 01:28:26.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/14/658714/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/658714/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/658714/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/658714/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/658714/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/658714/3"}}},"commit":{"parents":[{"commit":"61298a47e4abd903fe38957ffbbbcf036e79bb3e","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61298a47e4abd903fe38957ffbbbcf036e79bb3e"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-05-13 02:09:11.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-06-12 01:28:18.000000000","tz":0},"subject":"Change some URLs to point to more accurate targets","message":"Change some URLs to point to more accurate targets\n\nThis is to change some URLs to point to more accurate targets.\n\nChange-Id: I75ac27ea78bde86651e1b3c0920b194b5c5a510f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51db77b5f2d41e34ae7bf0cb62fd8ec5790957b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51db77b5f2d41e34ae7bf0cb62fd8ec5790957b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
