)]}'
{"id":"openstack%2Fglance~658962","triplet_id":"openstack%2Fglance~master~Iae3503cba6be362b372e1fc3e75c2ddb1e99b763","project":"openstack/glance","branch":"master","topic":"bug/1802587","hashtags":[],"change_id":"Iae3503cba6be362b372e1fc3e75c2ddb1e99b763","subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","status":"MERGED","created":"2019-05-14 05:31:23.000000000","updated":"2019-08-20 17:08:46.000000000","submitted":"2019-08-20 17:05:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658962-1566320726608-3b288848","meta_rev_id":"2d3ccf2b5dcb0549abbf858cd0f2f061610ae82b","_number":658962,"virtual_id_number":658962,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-09 02:58:25.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2019-08-20 17:05:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-20 14:52:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-08-15 09:33:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2019-08-13 15:58:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-20 14:52:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":2,"date":"2019-08-08 04:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 10:59:05.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2019-08-02 15:01:03.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2019-08-05 13:11:08.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-08-09 02:58:25.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-13 15:58:09.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-08-15 09:33:10.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-08-20 14:52:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-20 17:05:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ce8f2330ce61af3f736ca469f4ba613f2a25551","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-14 05:31:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4cc9174d0b7cfbec72aca266737e0563cd4d663","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-14 06:10:37.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5b58cc36b50e8e836e5a113af2fa62f76515b6a7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-14 07:01:17.000000000","message":"Patch Set 2:\n\nUnit/functional tests will fail until glance_store patches gets merged and new version of glance_store is released.","accounts_in_message":[],"_revision_number":2},{"id":"272cfd5ae5a7f696ee97bc6b8f820f8a75d58e6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 08:05:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/62/658962/2/check/tempest-full/d1bfc63/ : SUCCESS in 1h 46m 41s\n- neutron-grenade http://logs.openstack.org/62/658962/2/check/neutron-grenade/0e75161/ : SUCCESS in 57m 16s\n- grenade-py3 http://logs.openstack.org/62/658962/2/check/grenade-py3/2e82c7d/ : SUCCESS in 1h 01m 48s\n- tempest-full-py3 http://logs.openstack.org/62/658962/2/check/tempest-full-py3/529dd6b/ : SUCCESS in 1h 34m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/658962/2/check/openstack-tox-lower-constraints/96f7aa0/ : SUCCESS in 4m 41s\n- openstack-tox-pep8 http://logs.openstack.org/62/658962/2/check/openstack-tox-pep8/65ccbd9/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/62/658962/2/check/openstack-tox-py27/254de1d/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/62/658962/2/check/openstack-tox-py35/c73a792/ : SUCCESS in 4m 27s\n- openstack-tox-py36 http://logs.openstack.org/62/658962/2/check/openstack-tox-py36/7a51a57/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/62/658962/2/check/openstack-tox-py37/13b7d7c/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/62/658962/2/check/openstack-tox-docs/51da7f8/html/ : SUCCESS in 4m 35s\n- openstack-tox-functional http://logs.openstack.org/62/658962/2/check/openstack-tox-functional/4472f0b/ : FAILURE in 10m 43s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/658962/2/check/openstack-tox-functional-py35/a6e5cba/ : FAILURE in 9m 32s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/62/658962/2/check/glance-eventlet-ssl-handshake-broken-py35/30108d2/ : FAILURE in 2m 56s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/62/658962/2/check/glance-code-constants-check/7ee4e92/ : SUCCESS in 4m 46s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/62/658962/2/check/devstack-plugin-ceph-tempest/e3545d1/ : SUCCESS in 1h 27m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b73b64f34a023bd4d5c9290616022134c587d1c","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-05-14 11:29:27.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f541a37f22f4a689776a79ceb875268d49639e80","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-14 16:43:51.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/62/658962/2/check-not-vote/ext-glance-zuul/5e347d8 : FAILURE in 23m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f7645f9f396b9062c3cc0c0611a5661681fb3536","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-15 05:14:39.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b94a2a7ea6960911c2aeb948a285e6ecf3c3713a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 05:19:38.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"f691bd0b27eee9132c53da8bf53c635edb25c654","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-15 05:39:37.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/62/658962/3/check-not-vote/ext-glance-zuul/ea068e7 : FAILURE in 24m 37s","accounts_in_message":[],"_revision_number":3},{"id":"bda2c853c892994472d498de6144d1a2837a90be","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-04 04:56:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d509029194028338bc0a5401397c14498866a699","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-04 05:29:59.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/62/658962/4/check-not-vote/ext-glance-zuul/82a769e : FAILURE in 32m 37s","accounts_in_message":[],"_revision_number":4},{"id":"78c7144b70f1742935593554beb13da3d7618b82","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-04 05:43:28.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"1045e00c200ba609b2f5070fd4a57f2dac019cf6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 06:55:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/62/658962/4/check/tempest-full/816cd65/ : SUCCESS in 1h 50m 35s\n- neutron-grenade http://logs.openstack.org/62/658962/4/check/neutron-grenade/bab66fc/ : SUCCESS in 52m 13s\n- grenade-py3 http://logs.openstack.org/62/658962/4/check/grenade-py3/071be8b/ : SUCCESS in 47m 00s\n- tempest-full-py3 http://logs.openstack.org/62/658962/4/check/tempest-full-py3/2c03e88/ : SUCCESS in 1h 34m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/658962/4/check/openstack-tox-lower-constraints/e4f42dc/ : SUCCESS in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/62/658962/4/check/openstack-tox-pep8/dc6f0c2/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/62/658962/4/check/openstack-tox-py27/530bcfb/ : SUCCESS in 4m 14s\n- openstack-tox-py35 http://logs.openstack.org/62/658962/4/check/openstack-tox-py35/672846f/ : SUCCESS in 4m 51s\n- openstack-tox-py36 http://logs.openstack.org/62/658962/4/check/openstack-tox-py36/b24f263/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/62/658962/4/check/openstack-tox-py37/f1991ed/ : SUCCESS in 5m 03s\n- openstack-tox-docs http://logs.openstack.org/62/658962/4/check/openstack-tox-docs/7ff7c28/html/ : SUCCESS in 4m 44s\n- openstack-tox-functional http://logs.openstack.org/62/658962/4/check/openstack-tox-functional/2065fad/ : FAILURE in 8m 04s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/658962/4/check/openstack-tox-functional-py35/54be110/ : FAILURE in 9m 34s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/62/658962/4/check/glance-eventlet-ssl-handshake-broken-py35/491aef1/ : FAILURE in 2m 47s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/62/658962/4/check/glance-code-constants-check/6777cab/ : SUCCESS in 3m 37s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/62/658962/4/check/devstack-plugin-ceph-tempest/5655644/ : SUCCESS in 1h 16m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cc2cf8ea9b0fd40276862422e3fddf7407233bb2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 05:19:37.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"3e793e1944fc07eadc4b1fcfdac7be44028b8321","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-07 05:54:57.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/62/658962/4/check-not-vote/ext-glance-zuul/6acbdbe : SUCCESS in 34m 55s","accounts_in_message":[],"_revision_number":4},{"id":"c40ac4573f5415e7d685dbea372dad27670a99ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 07:10:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/62/658962/4/check/tempest-full/68264bf/ : SUCCESS in 1h 45m 33s\n- neutron-grenade http://logs.openstack.org/62/658962/4/check/neutron-grenade/ace4cf9/ : SUCCESS in 54m 53s\n- grenade-py3 http://logs.openstack.org/62/658962/4/check/grenade-py3/c3a78ae/ : SUCCESS in 57m 51s\n- tempest-full-py3 http://logs.openstack.org/62/658962/4/check/tempest-full-py3/6d25faa/ : SUCCESS in 1h 33m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/658962/4/check/openstack-tox-lower-constraints/4ade964/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/62/658962/4/check/openstack-tox-pep8/ddb465d/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/62/658962/4/check/openstack-tox-py27/e4f4270/ : SUCCESS in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/62/658962/4/check/openstack-tox-py35/728bae7/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/62/658962/4/check/openstack-tox-py36/7cc6635/ : SUCCESS in 5m 11s\n- openstack-tox-py37 http://logs.openstack.org/62/658962/4/check/openstack-tox-py37/9ecbd77/ : SUCCESS in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/62/658962/4/check/openstack-tox-docs/ab74cd8/html/ : SUCCESS in 4m 22s\n- openstack-tox-functional http://logs.openstack.org/62/658962/4/check/openstack-tox-functional/a8130bf/ : FAILURE in 8m 20s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/658962/4/check/openstack-tox-functional-py35/d15a152/ : FAILURE in 7m 55s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/62/658962/4/check/glance-eventlet-ssl-handshake-broken-py35/61c7ba3/ : FAILURE in 2m 30s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/62/658962/4/check/glance-code-constants-check/7053281/ : SUCCESS in 3m 43s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/62/658962/4/check/devstack-plugin-ceph-tempest/127b884/ : SUCCESS in 1h 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"298194a5958124bcf46cf2f7e64d1a86eae98f16","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-10 06:29:16.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"ed3f5e8c924ea7705415b490e41f6a10d974f643","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-10 07:11:57.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/62/658962/5/check-not-vote/ext-glance-zuul/642cb90 : SUCCESS in 42m 20s","accounts_in_message":[],"_revision_number":5},{"id":"31d46fd3d72e5da6b3f0ef295bcb69b870fcb51a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 08:18:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/62/658962/5/check/tempest-full/bfcc361/ : SUCCESS in 1h 43m 48s\n- neutron-grenade http://logs.openstack.org/62/658962/5/check/neutron-grenade/8608bf1/ : SUCCESS in 56m 14s\n- grenade-py3 http://logs.openstack.org/62/658962/5/check/grenade-py3/3e1bae4/ : SUCCESS in 1h 06m 50s\n- tempest-full-py3 http://logs.openstack.org/62/658962/5/check/tempest-full-py3/27878b5/ : SUCCESS in 1h 30m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/658962/5/check/openstack-tox-lower-constraints/e25c9f3/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/62/658962/5/check/openstack-tox-pep8/ecfcf5e/ : SUCCESS in 8m 00s\n- openstack-tox-py27 http://logs.openstack.org/62/658962/5/check/openstack-tox-py27/0454aca/ : SUCCESS in 4m 15s\n- openstack-tox-py35 http://logs.openstack.org/62/658962/5/check/openstack-tox-py35/a5c6742/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/62/658962/5/check/openstack-tox-py36/092b863/ : SUCCESS in 3m 54s\n- openstack-tox-py37 http://logs.openstack.org/62/658962/5/check/openstack-tox-py37/965a6c7/ : SUCCESS in 7m 27s\n- openstack-tox-docs http://logs.openstack.org/62/658962/5/check/openstack-tox-docs/063bd9f/html/ : SUCCESS in 4m 58s\n- openstack-tox-functional http://logs.openstack.org/62/658962/5/check/openstack-tox-functional/0a7cb83/ : FAILURE in 11m 05s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/658962/5/check/openstack-tox-functional-py35/d13ef17/ : FAILURE in 8m 54s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/62/658962/5/check/glance-eventlet-ssl-handshake-broken-py3/0989a42/ : FAILURE in 4m 10s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/62/658962/5/check/glance-code-constants-check/bf386b3/ : SUCCESS in 4m 10s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/62/658962/5/check/devstack-plugin-ceph-tempest/8b5cf78/ : FAILURE in 1h 37m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4f0b36fd6ac5376d9efa04551ab7ce57a5ce3080","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-22 09:58:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bd30792954c72c0c4d38e13afa23363f30ce6f85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 12:06:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/62/658962/6/check/tempest-full/f2499b7/ : SUCCESS in 1h 45m 26s\n- neutron-grenade http://logs.openstack.org/62/658962/6/check/neutron-grenade/f522b0a/ : SUCCESS in 1h 09m 47s\n- grenade-py3 http://logs.openstack.org/62/658962/6/check/grenade-py3/c5f2922/ : SUCCESS in 1h 00m 26s\n- tempest-full-py3 http://logs.openstack.org/62/658962/6/check/tempest-full-py3/13ee636/ : SUCCESS in 1h 32m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/658962/6/check/openstack-tox-lower-constraints/a490793/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/62/658962/6/check/openstack-tox-pep8/b63f275/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/62/658962/6/check/openstack-tox-py27/7a9db02/ : SUCCESS in 5m 11s\n- openstack-tox-py35 http://logs.openstack.org/62/658962/6/check/openstack-tox-py35/c3df19a/ : SUCCESS in 4m 56s\n- openstack-tox-py36 http://logs.openstack.org/62/658962/6/check/openstack-tox-py36/862bd19/ : SUCCESS in 6m 22s\n- openstack-tox-py37 http://logs.openstack.org/62/658962/6/check/openstack-tox-py37/00fa9fd/ : SUCCESS in 7m 43s\n- openstack-tox-docs http://logs.openstack.org/62/658962/6/check/openstack-tox-docs/3c7c31a/html/ : SUCCESS in 5m 17s\n- openstack-tox-functional http://logs.openstack.org/62/658962/6/check/openstack-tox-functional/4ef7e24/ : FAILURE in 9m 29s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/658962/6/check/openstack-tox-functional-py35/08b3328/ : FAILURE in 9m 35s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/62/658962/6/check/glance-eventlet-ssl-handshake-broken-py3/271c2ed/ : FAILURE in 5m 34s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/62/658962/6/check/glance-code-constants-check/720483f/ : SUCCESS in 4m 38s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/62/658962/6/check/devstack-plugin-ceph-tempest/f310af6/ : SUCCESS in 1h 09m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7c30d884663a804e43265d5557d2ab509ecb06e1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-22 14:52:42.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/62/658962/6/check-not-vote/ext-glance-zuul/08873a2 : SUCCESS in 43m 36s","accounts_in_message":[],"_revision_number":6},{"id":"fd7e31e0d6260dc2e9ede0d08ebcf1cc1170ba9f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-26 11:06:08.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nOnce upper-constraints includes glance_store 1.0.0, this patch should include a change to requirements.txt to set the minimum version of glance-store to 1.0.0, and also modify lower-constraints.txt to set glance-store\u003d\u003d1.0.0.","accounts_in_message":[],"_revision_number":6},{"id":"3613ae7a854648f92275e64867f44acc0a599208","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-26 17:03:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0b0ef6fa2f254d784ddaabc7d875d1b31f64f48c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-26 17:04:11.000000000","message":"Patch Set 6:\n\n(2 comments)\n\n\u003e (2 comments)\n \u003e \n \u003e Once upper-constraints includes glance_store 1.0.0, this patch\n \u003e should include a change to requirements.txt to set the minimum\n \u003e version of glance-store to 1.0.0, and also modify lower-constraints.txt\n \u003e to set glance-store\u003d\u003d1.0.0.\n\nHi Brian,\n\nThank you for review, made changes as per your suggestion.","accounts_in_message":[],"_revision_number":6},{"id":"b25eca325654e3d0ce4e6e49a8b2655fd10110b2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-26 17:29:33.000000000","message":"Patch Set 7: Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"776f8fcffc913159a4575e9e052d0cf27bdaf401","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-26 17:43:52.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/62/658962/7/check-not-vote/ext-glance-zuul/e566fc5 : SUCCESS in 40m 14s","accounts_in_message":[],"_revision_number":7},{"id":"360aca074149ed0543cfd3c2b1018c2cf055bfc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 18:56:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/62/658962/7/check/requirements-check/3fac445/ : SUCCESS in 4m 36s\n- tempest-full http://logs.openstack.org/62/658962/7/check/tempest-full/1ee3cfd/ : SUCCESS in 1h 51m 16s\n- neutron-grenade http://logs.openstack.org/62/658962/7/check/neutron-grenade/0ac29e3/ : SUCCESS in 57m 39s\n- grenade-py3 http://logs.openstack.org/62/658962/7/check/grenade-py3/6e54e1c/ : SUCCESS in 1h 03m 15s\n- tempest-full-py3 http://logs.openstack.org/62/658962/7/check/tempest-full-py3/3b6e3ad/ : SUCCESS in 1h 35m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/658962/7/check/openstack-tox-lower-constraints/0e04b99/ : SUCCESS in 5m 33s\n- openstack-tox-pep8 http://logs.openstack.org/62/658962/7/check/openstack-tox-pep8/5db163a/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/62/658962/7/check/openstack-tox-py27/6580f4a/ : SUCCESS in 4m 11s\n- openstack-tox-py35 http://logs.openstack.org/62/658962/7/check/openstack-tox-py35/7109071/ : SUCCESS in 5m 49s\n- openstack-tox-py36 http://logs.openstack.org/62/658962/7/check/openstack-tox-py36/79ea9e3/ : SUCCESS in 6m 50s\n- openstack-tox-py37 http://logs.openstack.org/62/658962/7/check/openstack-tox-py37/5cc838a/ : SUCCESS in 6m 10s\n- openstack-tox-docs http://logs.openstack.org/62/658962/7/check/openstack-tox-docs/c104458/html/ : SUCCESS in 5m 14s\n- openstack-tox-functional http://logs.openstack.org/62/658962/7/check/openstack-tox-functional/df4fa87/ : SUCCESS in 7m 38s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/658962/7/check/openstack-tox-functional-py35/2ce421c/ : SUCCESS in 9m 54s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/62/658962/7/check/glance-eventlet-ssl-handshake-broken-py3/35b1be8/ : FAILURE in 5m 20s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/62/658962/7/check/glance-code-constants-check/749a958/ : SUCCESS in 5m 28s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/62/658962/7/check/devstack-plugin-ceph-tempest/7a61fc1/ : SUCCESS in 1h 20m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"32b708307f230b350f512a5549445e123b231fce","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-27 14:15:06.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nSorry that I didn\u0027t notice the schema change earlier.  Please see my comment inline.","accounts_in_message":[],"_revision_number":7},{"id":"1d637a914b10149b17733ac65245b401d12e3c1c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-29 05:19:35.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cdbd4649ef012fa55bd5560362881e9a9279ecda","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-29 05:22:29.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nThank you Brian for review, made changes as per your suggestion.","accounts_in_message":[],"_revision_number":7},{"id":"93a457d3cd6b05d0f5365e495e9932a9c7bc4798","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-29 06:00:55.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/62/658962/8/check-not-vote/ext-glance-zuul/5ea5422 : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":8},{"id":"139dc2b664e69a8e5bd5f712df27208c2e873508","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 07:09:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/62/658962/8/check/requirements-check/d35e605/ : SUCCESS in 2m 27s\n- tempest-full http://logs.openstack.org/62/658962/8/check/tempest-full/c4c79d4/ : SUCCESS in 1h 44m 07s\n- neutron-grenade http://logs.openstack.org/62/658962/8/check/neutron-grenade/31e1d5e/ : SUCCESS in 58m 07s\n- grenade-py3 http://logs.openstack.org/62/658962/8/check/grenade-py3/b424c2e/ : SUCCESS in 1h 00m 43s\n- tempest-full-py3 http://logs.openstack.org/62/658962/8/check/tempest-full-py3/345daa3/ : SUCCESS in 1h 26m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/658962/8/check/openstack-tox-lower-constraints/bed467a/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/62/658962/8/check/openstack-tox-pep8/8e7ed7a/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/62/658962/8/check/openstack-tox-py27/ed9ca8d/ : SUCCESS in 5m 16s\n- openstack-tox-py35 http://logs.openstack.org/62/658962/8/check/openstack-tox-py35/4efe1d0/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/62/658962/8/check/openstack-tox-py36/c4190b6/ : SUCCESS in 3m 45s\n- openstack-tox-py37 http://logs.openstack.org/62/658962/8/check/openstack-tox-py37/9d65339/ : SUCCESS in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/62/658962/8/check/openstack-tox-docs/d034a1a/html/ : SUCCESS in 4m 28s\n- openstack-tox-functional http://logs.openstack.org/62/658962/8/check/openstack-tox-functional/ac81837/ : SUCCESS in 7m 54s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/658962/8/check/openstack-tox-functional-py35/d47f748/ : SUCCESS in 8m 01s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/62/658962/8/check/glance-eventlet-ssl-handshake-broken-py3/8f64fda/ : FAILURE in 4m 28s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/62/658962/8/check/glance-code-constants-check/8ebd055/ : SUCCESS in 3m 16s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/62/658962/8/check/devstack-plugin-ceph-tempest/16c6c68/ : SUCCESS in 1h 22m 32s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"01af4c1df30ed06f91fbbcdacc6afd19a91f2c6e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-30 04:40:52.000000000","message":"Patch Set 8: Review-Priority+1","accounts_in_message":[],"_revision_number":8},{"id":"608177083d7ef8b1cad9e2a1c2df6d713488e48b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-30 06:17:59.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"24a568af65a5ef7b9f6deb7f49f7b51331d4f7c7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-30 06:57:39.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/62/658962/9/check-not-vote/ext-glance-zuul/dbc2b78 : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":9},{"id":"0634a59ed48818544e282fe0ac4571b4d99144e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 09:15:53.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/62/658962/9/check/requirements-check/4b4cf2d/ : SUCCESS in 3m 18s\n- tempest-full http://logs.openstack.org/62/658962/9/check/tempest-full/8e2f2f0/ : SUCCESS in 1h 50m 08s\n- neutron-grenade http://logs.openstack.org/62/658962/9/check/neutron-grenade/80b956b/ : SUCCESS in 58m 10s\n- grenade-py3 http://logs.openstack.org/62/658962/9/check/grenade-py3/00dc931/ : SUCCESS in 1h 04m 24s\n- tempest-full-py3 http://logs.openstack.org/62/658962/9/check/tempest-full-py3/af0e045/ : SUCCESS in 1h 24m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/658962/9/check/openstack-tox-lower-constraints/23096ee/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/62/658962/9/check/openstack-tox-pep8/f63cf54/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/62/658962/9/check/openstack-tox-py27/99484ae/ : SUCCESS in 4m 50s\n- openstack-tox-py35 http://logs.openstack.org/62/658962/9/check/openstack-tox-py35/65b7d2a/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/62/658962/9/check/openstack-tox-py36/cc3705a/ : SUCCESS in 5m 47s\n- openstack-tox-py37 http://logs.openstack.org/62/658962/9/check/openstack-tox-py37/b4a8c7e/ : SUCCESS in 5m 38s\n- openstack-tox-docs http://logs.openstack.org/62/658962/9/check/openstack-tox-docs/97ad8e4/html/ : SUCCESS in 5m 39s\n- openstack-tox-functional http://logs.openstack.org/62/658962/9/check/openstack-tox-functional/a4e123d/ : SUCCESS in 10m 20s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/658962/9/check/openstack-tox-functional-py35/0cc26a6/ : SUCCESS in 11m 14s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/62/658962/9/check/glance-eventlet-ssl-handshake-broken-py3/8f091fe/ : FAILURE in 5m 33s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/62/658962/9/check/glance-code-constants-check/f5cc03b/ : SUCCESS in 4m 53s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/62/658962/9/check/devstack-plugin-ceph-tempest/a0a9464/ : SUCCESS in 1h 09m 55s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"49376bccca647780e208c8ffea42b86859e856aa","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-30 12:38:12.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nOne more change on the \u0027stores\u0027 description in the image schema.\n\nDo you think we need a release note with this, or is your plan to have a more general vocabulary correction release note?","accounts_in_message":[],"_revision_number":9},{"id":"e773f77de7b903aa10b8e6275292c6c499b8536c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-30 12:38:34.000000000","message":"Patch Set 9: Code-Review-1\n\nForgot to vote with my last comment.","accounts_in_message":[],"_revision_number":9},{"id":"f9542d327c2bdb5e81ccad4fd87b33912890e18d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-30 15:43:06.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nNew PS coming soon!!","accounts_in_message":[],"_revision_number":9},{"id":"3d72ef10dfb7369eed2840cbbe0dc4bad1afcf1b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-30 17:00:29.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3a98691b8acedbd9c05238d363472faa1e59021f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-30 17:24:30.000000000","message":"Patch Set 10: Review-Priority+1","accounts_in_message":[],"_revision_number":10},{"id":"530793d7c0a08fc1e0e503c28ecf59492497e52b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-30 17:33:41.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"395b1284ff61c28ed58e44b10a093ac8272b53c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 22:43:00.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/62/658962/11/check/requirements-check/649f23a/ : SUCCESS in 3m 56s\n- tempest-full http://logs.openstack.org/62/658962/11/check/tempest-full/f9efd8b/ : SUCCESS in 1h 42m 47s\n- neutron-grenade http://logs.openstack.org/62/658962/11/check/neutron-grenade/b39011d/ : SUCCESS in 59m 27s\n- grenade-py3 http://logs.openstack.org/62/658962/11/check/grenade-py3/236b747/ : SUCCESS in 1h 04m 56s\n- tempest-full-py3 http://logs.openstack.org/62/658962/11/check/tempest-full-py3/3daffb3/ : SUCCESS in 1h 28m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/658962/11/check/openstack-tox-lower-constraints/4b566b6/ : SUCCESS in 6m 04s\n- openstack-tox-pep8 http://logs.openstack.org/62/658962/11/check/openstack-tox-pep8/31d4233/ : SUCCESS in 5m 49s\n- openstack-tox-py27 http://logs.openstack.org/62/658962/11/check/openstack-tox-py27/67862e5/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/62/658962/11/check/openstack-tox-py35/49cba93/ : SUCCESS in 5m 47s\n- openstack-tox-py36 http://logs.openstack.org/62/658962/11/check/openstack-tox-py36/48157e0/ : SUCCESS in 5m 38s\n- openstack-tox-py37 http://logs.openstack.org/62/658962/11/check/openstack-tox-py37/8601836/ : SUCCESS in 8m 01s\n- openstack-tox-docs http://logs.openstack.org/62/658962/11/check/openstack-tox-docs/b23a67c/html/ : SUCCESS in 6m 15s\n- openstack-tox-functional http://logs.openstack.org/62/658962/11/check/openstack-tox-functional/16a573b/ : SUCCESS in 10m 13s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/658962/11/check/openstack-tox-functional-py35/8273fdf/ : SUCCESS in 11m 12s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/62/658962/11/check/glance-eventlet-ssl-handshake-broken-py3/9559125/ : FAILURE in 5m 28s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/62/658962/11/check/glance-code-constants-check/883f15a/ : SUCCESS in 5m 09s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/62/658962/11/check/devstack-plugin-ceph-tempest/7602806/ : SUCCESS in 1h 29m 49s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"67dbb3c5ccda3a417645e37fa4f714db4e7f38e9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-30 23:22:08.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/62/658962/11/check-not-vote/ext-glance-zuul/fb6ce0b : SUCCESS in 41m 29s","accounts_in_message":[],"_revision_number":11},{"id":"b1eb7ff32a8fb76706f1d344099bbcb5c41a57e9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-31 06:56:23.000000000","message":"Patch Set 11: Review-Priority+2\n\nNeed to merge as with glance_store v1.0 and latest master user will not be able to create the image.\n\nhttps://bugs.launchpad.net/glance/+bug/1802587/comments/8","accounts_in_message":[],"_revision_number":11},{"id":"66e94bd6217dddb48933bec16fad2b6340130e6d","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-08-02 15:01:03.000000000","message":"Patch Set 11: Code-Review+1\n\ntested this myself and it fixed the problem I was seeing.","accounts_in_message":[],"_revision_number":11},{"id":"a658ba3cccc946cf045dc09c5a3aa99bb36fe61a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-05 13:11:08.000000000","message":"Patch Set 11: Code-Review+2\n\nAll my concerns have been addressed.\n\n@hemna: thanks for giving this a real-life workout!","accounts_in_message":[],"_revision_number":11},{"id":"d1e4a3089ee9292d951b04f5f0b64e2ec298ed46","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-07 12:23:37.000000000","message":"Patch Set 11:\n\nLooks good, but unfortunately needs a rebase now.","accounts_in_message":[],"_revision_number":11},{"id":"f7cd7aa4dfede431b124e3f6270c773b23f4e955","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-08 04:54:53.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"eb6ff123ba60e64c64903b5b7919bf942779c519","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 06:58:46.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/62/658962/12/check/requirements-check/2678cc1/ : SUCCESS in 3m 54s\n- tempest-full https://logs.opendev.org/62/658962/12/check/tempest-full/a2d65dd/ : SUCCESS in 1h 50m 26s\n- neutron-grenade https://logs.opendev.org/62/658962/12/check/neutron-grenade/c1da31f/ : SUCCESS in 58m 39s\n- grenade-py3 https://logs.opendev.org/62/658962/12/check/grenade-py3/9fca092/ : SUCCESS in 1h 00m 51s\n- tempest-full-py3 https://logs.opendev.org/62/658962/12/check/tempest-full-py3/6213a75/ : SUCCESS in 1h 21m 58s\n- openstack-tox-lower-constraints https://logs.opendev.org/62/658962/12/check/openstack-tox-lower-constraints/c2fd82f/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://logs.opendev.org/62/658962/12/check/openstack-tox-pep8/3452b86/ : SUCCESS in 5m 18s\n- openstack-tox-py27 https://logs.opendev.org/62/658962/12/check/openstack-tox-py27/469ee04/ : SUCCESS in 5m 07s\n- openstack-tox-py35 https://logs.opendev.org/62/658962/12/check/openstack-tox-py35/2936564/ : SUCCESS in 5m 23s\n- openstack-tox-py36 https://logs.opendev.org/62/658962/12/check/openstack-tox-py36/7e5a37e/ : SUCCESS in 5m 21s\n- openstack-tox-py37 https://logs.opendev.org/62/658962/12/check/openstack-tox-py37/8cbc233/ : SUCCESS in 5m 16s\n- openstack-tox-docs https://logs.opendev.org/62/658962/12/check/openstack-tox-docs/aad73bb/html/ : SUCCESS in 5m 45s\n- openstack-tox-functional https://logs.opendev.org/62/658962/12/check/openstack-tox-functional/4110fa1/ : SUCCESS in 10m 44s\n- openstack-tox-functional-py35 https://logs.opendev.org/62/658962/12/check/openstack-tox-functional-py35/23110e9/ : FAILURE in 9m 36s\n- glance-eventlet-ssl-handshake-broken-py3 https://logs.opendev.org/62/658962/12/check/glance-eventlet-ssl-handshake-broken-py3/5ec1bab/ : FAILURE in 4m 07s (non-voting)\n- glance-code-constants-check https://logs.opendev.org/62/658962/12/check/glance-code-constants-check/22da861/ : SUCCESS in 4m 26s\n- devstack-plugin-ceph-tempest https://logs.opendev.org/62/658962/12/check/devstack-plugin-ceph-tempest/cce6a16/ : SUCCESS in 1h 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2a046e5c79684926704a28a4a840a1a179679948","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-08 06:59:51.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"c148acc4ca6abd2823b76e2484ed860160cad5e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 09:24:59.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://logs.opendev.org/62/658962/12/check/requirements-check/daab2fa/ : SUCCESS in 2m 43s\n- tempest-full https://logs.opendev.org/62/658962/12/check/tempest-full/00e3e69/ : SUCCESS in 1h 41m 35s\n- neutron-grenade https://logs.opendev.org/62/658962/12/check/neutron-grenade/a88fd48/ : SUCCESS in 1h 02m 34s\n- grenade-py3 https://logs.opendev.org/62/658962/12/check/grenade-py3/36d122d/ : SUCCESS in 57m 34s\n- tempest-full-py3 https://logs.opendev.org/62/658962/12/check/tempest-full-py3/f7f318f/ : SUCCESS in 1h 35m 51s\n- openstack-tox-lower-constraints https://logs.opendev.org/62/658962/12/check/openstack-tox-lower-constraints/5c1a5b1/ : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://logs.opendev.org/62/658962/12/check/openstack-tox-pep8/c844bef/ : SUCCESS in 4m 10s\n- openstack-tox-py27 https://logs.opendev.org/62/658962/12/check/openstack-tox-py27/d4d7e5f/ : SUCCESS in 5m 30s\n- openstack-tox-py35 https://logs.opendev.org/62/658962/12/check/openstack-tox-py35/24d01fc/ : SUCCESS in 8m 10s\n- openstack-tox-py36 https://logs.opendev.org/62/658962/12/check/openstack-tox-py36/0f19afe/ : SUCCESS in 8m 22s\n- openstack-tox-py37 https://logs.opendev.org/62/658962/12/check/openstack-tox-py37/4ec596b/ : SUCCESS in 5m 47s\n- openstack-tox-docs https://logs.opendev.org/62/658962/12/check/openstack-tox-docs/4bd2747/html/ : SUCCESS in 6m 49s\n- openstack-tox-functional https://logs.opendev.org/62/658962/12/check/openstack-tox-functional/faad510/ : SUCCESS in 9m 31s\n- openstack-tox-functional-py35 https://logs.opendev.org/62/658962/12/check/openstack-tox-functional-py35/d05bfe1/ : SUCCESS in 20m 48s\n- glance-eventlet-ssl-handshake-broken-py3 https://logs.opendev.org/62/658962/12/check/glance-eventlet-ssl-handshake-broken-py3/9868c13/ : FAILURE in 4m 19s (non-voting)\n- glance-code-constants-check https://logs.opendev.org/62/658962/12/check/glance-code-constants-check/df9445b/ : SUCCESS in 4m 37s\n- devstack-plugin-ceph-tempest https://logs.opendev.org/62/658962/12/check/devstack-plugin-ceph-tempest/6c90d05/ : SUCCESS in 1h 20m 21s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0af91bd3bf1bb3d43b1426df5283531849550a66","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-09 02:58:25.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/62/658962/12/check-not-vote/ext-glance-zuul/97edda8 : FAILURE in 16m 02s","accounts_in_message":[],"_revision_number":12},{"id":"bf516c09e3674c6a16b1da92a5a6f93d62ea2bf4","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-08-13 15:58:09.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"6cd4da620e38851601f070dcc6c8ebd03781d6b3","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-08-15 09:33:10.000000000","message":"Patch Set 12: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":12},{"id":"e8e768e866b77fdfdd487b70172018ba430adbe0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-20 14:52:07.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\nAll cases I could find have been covered. This looks good to me.\n\nThanks Abhishek!","accounts_in_message":[],"_revision_number":12},{"id":"b4f963214317a79868465fdbbdcccf92a4db6bbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 14:52:29.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"a74de238bdc3790fd31bb0c238dc526b973a9723","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:05:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"2737dea9d97fdb137baffb8d3769878f01811ad9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:05:26.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b33edd5244d84cf38140374ecb1d8d52 : SUCCESS in 5m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b658ce61422947e7befb4b3795b355f1 : SUCCESS in 2h 05m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/293292e0978d48c89fe7ca1352556662 : SUCCESS in 1h 12m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4486c6fd86a14c55892a43ce8f3df134 : SUCCESS in 1h 15m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e71b314664c741eeb26af89ab750a6bd : SUCCESS in 1h 31m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f10c4a3ef8044379afff1b1b84637c1f : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eae1a08825c4442887633f57af0f687a : SUCCESS in 7m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9f4fdc2952a4edc92124a6607b6e8f1 : SUCCESS in 8m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c6b6d75f00fc41cbbd99ae0d9e571195 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f87e93cf4cc4767934c3b1a8461ff65 : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61595624be8c448bbaa339d868dac9bc : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f02d0a05e27f47dbb2266ae6802bd817 : SUCCESS in 8m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/534da137c778458f8447350ae0f48f72 : SUCCESS in 10m 54s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/080af2df36214c62b5ef495ca5f2e4f8 : SUCCESS in 12m 07s","accounts_in_message":[],"_revision_number":12},{"id":"c236a2df1ebf54bc6e76816f9ba936a0acae4af9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:08:46.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3bf56314fd445c3b962e57e59bbb2f9 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"0e55ad71a808b9c6f151ebdd6eb0d8f5f52217ac","revisions":{"5c81a3f38f78446543f222f8bde9971eac283d0e":{"kind":"REWORK","_number":1,"created":"2019-05-14 05:31:23.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/62/658962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/658962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/658962/1"}}},"commit":{"parents":[{"commit":"61298a47e4abd903fe38957ffbbbcf036e79bb3e","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61298a47e4abd903fe38957ffbbbcf036e79bb3e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 05:29:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 05:29:26.000000000","tz":0},"subject":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027","message":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027\n\nAs a part of vocabulory correction, changed the location\nmetadata key name from \u0027backend\u0027 to \u0027store\u0027. Modified\ncorresponding tests as well.\n\nNOTE: Removed \u0027backend\u0027 attribute from images schema as it\nwas not used anywhere.\n\nbp:multi-store-vocabulary-correction\n\nChange-Id: Iae3503cba6be362b372e1fc3e75c2ddb1e99b763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c81a3f38f78446543f222f8bde9971eac283d0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c81a3f38f78446543f222f8bde9971eac283d0e"}]},"branch":"refs/heads/master"},"5994c5a3c8197c248b831275f798a4837fec9306":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-14 06:10:37.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/62/658962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/658962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/658962/2"}}},"commit":{"parents":[{"commit":"61298a47e4abd903fe38957ffbbbcf036e79bb3e","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61298a47e4abd903fe38957ffbbbcf036e79bb3e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 05:29:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 06:10:15.000000000","tz":0},"subject":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027","message":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027\n\nAs a part of vocabulory correction, changed the location\nmetadata key name from \u0027backend\u0027 to \u0027store\u0027. Modified\ncorresponding tests as well.\n\nNOTE: Removed \u0027backend\u0027 attribute from images schema as it\nwas not used anywhere.\n\nDepends-On: https://review.opendev.org/#/c/658960/\nbp:multi-store-vocabulary-correction\n\nChange-Id: Iae3503cba6be362b372e1fc3e75c2ddb1e99b763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5994c5a3c8197c248b831275f798a4837fec9306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5994c5a3c8197c248b831275f798a4837fec9306"}]},"branch":"refs/heads/master"},"e23203cacb5932c9262ee5792884c65b94599eed":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-15 05:14:39.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/62/658962/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/658962/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/658962/3"}}},"commit":{"parents":[{"commit":"61298a47e4abd903fe38957ffbbbcf036e79bb3e","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61298a47e4abd903fe38957ffbbbcf036e79bb3e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 05:29:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-15 05:14:04.000000000","tz":0},"subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","message":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027\n\nAs a part of vocabulary correction, changed the location\nmetadata key name from \u0027backend\u0027 to \u0027store\u0027. Modified\ncorresponding tests as well.\n\nNOTE: Removed \u0027backend\u0027 attribute from images schema as it\nwas not used anywhere.\n\nDepends-On: https://review.opendev.org/#/c/658960/\nbp:multi-store-vocabulary-correction\n\nChange-Id: Iae3503cba6be362b372e1fc3e75c2ddb1e99b763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e23203cacb5932c9262ee5792884c65b94599eed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e23203cacb5932c9262ee5792884c65b94599eed"}]},"branch":"refs/heads/master"},"8f0180d30c4fcf4f25d6f18e83031b85370fb63f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2019-06-04 04:56:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/62/658962/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/658962/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/658962/4"}}},"commit":{"parents":[{"commit":"d99cea0cc76e981debece97b3d9fd0e5ef6434d2","subject":"Merge \"Mark http store read-only in discovery call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d99cea0cc76e981debece97b3d9fd0e5ef6434d2"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 05:29:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-04 03:51:08.000000000","tz":0},"subject":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027","message":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027\n\nAs a part of vocabulory correction, changed the location\nmetadata key name from \u0027backend\u0027 to \u0027store\u0027. Modified\ncorresponding tests as well.\n\nNOTE: Removed \u0027backend\u0027 attribute from images schema as it\nwas not used anywhere.\n\nbp:multi-store-vocabulary-correction\n\nChange-Id: Iae3503cba6be362b372e1fc3e75c2ddb1e99b763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8f0180d30c4fcf4f25d6f18e83031b85370fb63f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8f0180d30c4fcf4f25d6f18e83031b85370fb63f"}]},"branch":"refs/heads/master"},"bc84af70e4d8c836be169f159a25c5611ade435f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-10 06:29:16.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/62/658962/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/658962/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/658962/5"}}},"commit":{"parents":[{"commit":"b8a34d31284227d29c467ffe33de1873e21a4633","subject":"Merge \"Don\u0027t target broken ssl tests to specific py3.minor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8a34d31284227d29c467ffe33de1873e21a4633"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 05:29:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-10 06:20:42.000000000","tz":0},"subject":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027","message":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027\n\nAs a part of vocabulory correction, changed the location\nmetadata key name from \u0027backend\u0027 to \u0027store\u0027. Modified\ncorresponding tests as well.\n\nNOTE: Removed \u0027backend\u0027 attribute from images schema as it\nwas not used anywhere.\n\nbp:multi-store-vocabulary-correction\n\nChange-Id: Iae3503cba6be362b372e1fc3e75c2ddb1e99b763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc84af70e4d8c836be169f159a25c5611ade435f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc84af70e4d8c836be169f159a25c5611ade435f"}]},"branch":"refs/heads/master"},"5a3131e5c98106a7102eaaf0da539e57832a445b":{"kind":"REWORK","_number":6,"created":"2019-07-22 09:58:20.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/62/658962/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/658962/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/658962/6"}}},"commit":{"parents":[{"commit":"9bb432c30d8a4be7ce82f5c6d79f85f9c4355a3a","subject":"Merge \"MultiStore: Avoid potential KeyError in functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9bb432c30d8a4be7ce82f5c6d79f85f9c4355a3a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 05:29:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-22 09:58:01.000000000","tz":0},"subject":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027","message":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027\n\nAs a part of vocabulory correction, changed the location\nmetadata key name from \u0027backend\u0027 to \u0027store\u0027. Modified\ncorresponding tests as well.\n\nNOTE: Removed \u0027backend\u0027 attribute from images schema as it\nwas not used anywhere.\n\nbp:multi-store-vocabulary-correction\n\nChange-Id: Iae3503cba6be362b372e1fc3e75c2ddb1e99b763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5a3131e5c98106a7102eaaf0da539e57832a445b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5a3131e5c98106a7102eaaf0da539e57832a445b"}]},"branch":"refs/heads/master"},"cc2cb8a438693095ff78d5bd624fe65ede6d78b7":{"kind":"REWORK","_number":7,"created":"2019-07-26 17:03:16.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/62/658962/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/658962/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/658962/7"}}},"commit":{"parents":[{"commit":"62da67070e5ef8aaf4dd9dcd0ac06a3b62c4f0c6","subject":"Merge \"Skip tests to avoid trouble releasing store 1.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62da67070e5ef8aaf4dd9dcd0ac06a3b62c4f0c6"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 05:29:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-26 17:02:35.000000000","tz":0},"subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","message":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027\n\nAs a part of vocabulary correction, changed the location\nmetadata key name from \u0027backend\u0027 to \u0027store\u0027. Modified\ncorresponding tests as well.\n\nNOTE: Removed \u0027backend\u0027 attribute from images schema as it\nwas not used anywhere.\n\nbp:multi-store-vocabulary-correction\n\nChange-Id: Iae3503cba6be362b372e1fc3e75c2ddb1e99b763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc2cb8a438693095ff78d5bd624fe65ede6d78b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc2cb8a438693095ff78d5bd624fe65ede6d78b7"}]},"branch":"refs/heads/master"},"f37459268d9aa1d2ba8d6a8e521f66459f4b3cc0":{"kind":"REWORK","_number":8,"created":"2019-07-29 05:19:35.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/62/658962/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/658962/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/658962/8"}}},"commit":{"parents":[{"commit":"62da67070e5ef8aaf4dd9dcd0ac06a3b62c4f0c6","subject":"Merge \"Skip tests to avoid trouble releasing store 1.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62da67070e5ef8aaf4dd9dcd0ac06a3b62c4f0c6"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 05:29:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-29 05:17:23.000000000","tz":0},"subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","message":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027\n\nAs a part of vocabulary correction, changed the location\nmetadata key name from \u0027backend\u0027 to \u0027store\u0027. Modified\ncorresponding tests as well.\n\nbp:multi-store-vocabulary-correction\nChange-Id: Iae3503cba6be362b372e1fc3e75c2ddb1e99b763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f37459268d9aa1d2ba8d6a8e521f66459f4b3cc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f37459268d9aa1d2ba8d6a8e521f66459f4b3cc0"}]},"branch":"refs/heads/master"},"f06b6fa739a775f84812e9fe04d319972afb6a17":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-07-30 06:17:59.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/62/658962/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/658962/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/658962/9"}}},"commit":{"parents":[{"commit":"733095a89947718e54b610ad528f6243e6ece440","subject":"Merge \"Update api-ref location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/733095a89947718e54b610ad528f6243e6ece440"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 05:29:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-30 04:41:23.000000000","tz":0},"subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","message":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027\n\nAs a part of vocabulary correction, changed the location\nmetadata key name from \u0027backend\u0027 to \u0027store\u0027. Modified\ncorresponding tests as well.\n\nbp:multi-store-vocabulary-correction\nChange-Id: Iae3503cba6be362b372e1fc3e75c2ddb1e99b763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f06b6fa739a775f84812e9fe04d319972afb6a17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f06b6fa739a775f84812e9fe04d319972afb6a17"}]},"branch":"refs/heads/master"},"1477192b41751bd2a752a2c0714f6789b69d5cc0":{"kind":"REWORK","_number":10,"created":"2019-07-30 17:00:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/62/658962/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/658962/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/658962/10"}}},"commit":{"parents":[{"commit":"733095a89947718e54b610ad528f6243e6ece440","subject":"Merge \"Update api-ref location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/733095a89947718e54b610ad528f6243e6ece440"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 05:29:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-30 15:40:58.000000000","tz":0},"subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","message":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027\n\nAs a part of vocabulary correction, changed the location\nmetadata key name from \u0027backend\u0027 to \u0027store\u0027. Modified\ncorresponding tests as well.\n\nbp:multi-store-vocabulary-correction\nChange-Id: Iae3503cba6be362b372e1fc3e75c2ddb1e99b763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1477192b41751bd2a752a2c0714f6789b69d5cc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1477192b41751bd2a752a2c0714f6789b69d5cc0"}]},"branch":"refs/heads/master"},"77f03bbf1b9626dd6d9d21e4821b3d530db96af8":{"kind":"NO_CHANGE","_number":11,"created":"2019-07-30 17:33:41.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/62/658962/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/658962/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/658962/11"}}},"commit":{"parents":[{"commit":"733095a89947718e54b610ad528f6243e6ece440","subject":"Merge \"Update api-ref location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/733095a89947718e54b610ad528f6243e6ece440"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 05:29:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-30 17:29:00.000000000","tz":0},"subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","message":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027\n\nAs a part of vocabulary correction, changed the location\nmetadata key name from \u0027backend\u0027 to \u0027store\u0027. Modified\ncorresponding tests as well.\n\nbp:multi-store-vocabulary-correction\nChange-Id: Iae3503cba6be362b372e1fc3e75c2ddb1e99b763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77f03bbf1b9626dd6d9d21e4821b3d530db96af8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77f03bbf1b9626dd6d9d21e4821b3d530db96af8"}]},"branch":"refs/heads/master"},"0e55ad71a808b9c6f151ebdd6eb0d8f5f52217ac":{"kind":"REWORK","_number":12,"created":"2019-08-08 04:54:53.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/62/658962/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/658962/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/658962/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/658962/12"}}},"commit":{"parents":[{"commit":"4915fb8793a89374fb5a881fa0d13388cbb0587b","subject":"Merge \"Do not use glance_store 0.29.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4915fb8793a89374fb5a881fa0d13388cbb0587b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-14 05:29:26.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-08-08 04:52:51.000000000","tz":0},"subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","message":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027\n\nAs a part of vocabulary correction, changed the location\nmetadata key name from \u0027backend\u0027 to \u0027store\u0027. Modified\ncorresponding tests as well.\n\nbp:multi-store-vocabulary-correction\nChange-Id: Iae3503cba6be362b372e1fc3e75c2ddb1e99b763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e55ad71a808b9c6f151ebdd6eb0d8f5f52217ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e55ad71a808b9c6f151ebdd6eb0d8f5f52217ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
