)]}'
{"id":"openstack%2Fglance~660645","triplet_id":"openstack%2Fglance~master~I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d","project":"openstack/glance","branch":"master","topic":"bug/1802587","hashtags":[],"change_id":"I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d","subject":"Lazy update stores information","status":"MERGED","created":"2019-05-22 10:30:33.000000000","updated":"2019-08-20 17:10:49.000000000","submitted":"2019-08-20 17:05:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660645-1566320728389-dee321bd","meta_rev_id":"6a9003e6535d09f5a2baf56df230ff48d4d40ca7","_number":660645,"virtual_id_number":660645,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-09 04:08:29.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2019-08-20 17:05:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-08 04:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-08-15 09:28:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":1,"date":"2019-08-13 15:56:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-08-15 09:28:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-08-08 04:54:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 11:39:13.000000000","updated_by":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"reviewer":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"state":"REVIEWER"},{"updated":"2019-08-01 16:51:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-06 15:17:58.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-08-09 04:08:29.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-13 15:56:06.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-08-15 09:28:10.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-08-20 17:05:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c1e3561af4ac321b7a7a11746efbed0b4d05f5f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-22 10:30:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7cdb7ad767ac5ccf2f01581f2120e7c8177aef8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-22 11:09:01.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/1/check-not-vote/ext-glance-zuul/e26ada4 : FAILURE in 38m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e6f5607efe4d6ee77545043da32ac15575ab8f8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 14:53:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/45/660645/1/check/tempest-full/5c2ded9/ : SUCCESS in 2h 01m 23s\n- neutron-grenade http://logs.openstack.org/45/660645/1/check/neutron-grenade/b5ae3b2/ : SUCCESS in 1h 29m 35s\n- grenade-py3 http://logs.openstack.org/45/660645/1/check/grenade-py3/3f60139/ : SUCCESS in 1h 29m 18s\n- tempest-full-py3 http://logs.openstack.org/45/660645/1/check/tempest-full-py3/a9845b8/ : TIMED_OUT in 2h 09m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/660645/1/check/openstack-tox-lower-constraints/87efb27/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/45/660645/1/check/openstack-tox-pep8/6359a63/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/45/660645/1/check/openstack-tox-py27/2013dbc/ : FAILURE in 6m 00s\n- openstack-tox-py35 http://logs.openstack.org/45/660645/1/check/openstack-tox-py35/3150144/ : FAILURE in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/45/660645/1/check/openstack-tox-py36/f6c0b45/ : FAILURE in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/45/660645/1/check/openstack-tox-py37/1c871d2/ : FAILURE in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/45/660645/1/check/openstack-tox-docs/b4d2c88/html/ : SUCCESS in 5m 05s\n- openstack-tox-functional http://logs.openstack.org/45/660645/1/check/openstack-tox-functional/be08441/ : FAILURE in 8m 34s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/660645/1/check/openstack-tox-functional-py35/1a0daa4/ : FAILURE in 8m 22s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/45/660645/1/check/glance-eventlet-ssl-handshake-broken-py35/8e0130f/ : FAILURE in 2m 55s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/45/660645/1/check/glance-code-constants-check/a860d0b/ : SUCCESS in 4m 39s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/660645/1/check/devstack-plugin-ceph-tempest/7050d6b/ : SUCCESS in 1h 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc2e82ef2ffe6783a2e41b6d1de6566f9db3f617","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-24 08:46:52.000000000","message":"Patch Set 1: Review-Priority+1\n\nTests will fail unless dependent glance-store patch is merged and new version is released.","accounts_in_message":[],"_revision_number":1},{"id":"711245c411c15890471a5c7360ae8e5aeacd1aad","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-04 05:42:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd358d657b440a777eb296425084411e3b404a78","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-04 05:43:50.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"343572ac4878315a4dd810db7aa70b56ae846ca5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 07:31:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/45/660645/2/check/tempest-full/c19ddb9/ : SUCCESS in 1h 36m 35s\n- neutron-grenade http://logs.openstack.org/45/660645/2/check/neutron-grenade/49f6e08/ : SUCCESS in 1h 06m 25s\n- grenade-py3 http://logs.openstack.org/45/660645/2/check/grenade-py3/2f65819/ : SUCCESS in 1h 18m 26s\n- tempest-full-py3 http://logs.openstack.org/45/660645/2/check/tempest-full-py3/20f6051/ : SUCCESS in 1h 20m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/660645/2/check/openstack-tox-lower-constraints/33d8c6b/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/45/660645/2/check/openstack-tox-pep8/89b7027/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/45/660645/2/check/openstack-tox-py27/e615f80/ : FAILURE in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/45/660645/2/check/openstack-tox-py35/0e9f24b/ : FAILURE in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/45/660645/2/check/openstack-tox-py36/e808305/ : FAILURE in 4m 31s\n- openstack-tox-py37 http://logs.openstack.org/45/660645/2/check/openstack-tox-py37/1becf48/ : FAILURE in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/45/660645/2/check/openstack-tox-docs/44936a4/html/ : SUCCESS in 5m 20s\n- openstack-tox-functional http://logs.openstack.org/45/660645/2/check/openstack-tox-functional/29b8a37/ : FAILURE in 16m 23s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/660645/2/check/openstack-tox-functional-py35/a8e3dda/ : FAILURE in 10m 35s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/45/660645/2/check/glance-eventlet-ssl-handshake-broken-py35/a73f02c/ : FAILURE in 3m 35s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/45/660645/2/check/glance-code-constants-check/9269d52/ : SUCCESS in 4m 25s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/660645/2/check/devstack-plugin-ceph-tempest/b096e04/ : SUCCESS in 1h 47m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e54909b40d0fa93ee93a9ff9d380aa9e42c5587a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-04 16:10:03.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/2/check-not-vote/ext-glance-zuul/a7517dc : SUCCESS in 36m 28s","accounts_in_message":[],"_revision_number":2},{"id":"b9e3402e030ec6119ee5f2abee0dc1ffd2be3424","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-10 06:29:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3482e67babdbaa923e32e9ef29d22d35cff16660","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 08:33:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/45/660645/3/check/tempest-full/3eae66d/ : SUCCESS in 1h 58m 56s\n- neutron-grenade http://logs.openstack.org/45/660645/3/check/neutron-grenade/393faf9/ : SUCCESS in 58m 54s\n- grenade-py3 http://logs.openstack.org/45/660645/3/check/grenade-py3/d151c5c/ : SUCCESS in 1h 02m 17s\n- tempest-full-py3 http://logs.openstack.org/45/660645/3/check/tempest-full-py3/376c3ad/ : SUCCESS in 1h 32m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/660645/3/check/openstack-tox-lower-constraints/0e889c6/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/45/660645/3/check/openstack-tox-pep8/fccdf76/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/45/660645/3/check/openstack-tox-py27/06c98c8/ : FAILURE in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/45/660645/3/check/openstack-tox-py35/0ac3016/ : FAILURE in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/45/660645/3/check/openstack-tox-py36/b5379b1/ : FAILURE in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/45/660645/3/check/openstack-tox-py37/311165b/ : FAILURE in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/45/660645/3/check/openstack-tox-docs/27b59a0/html/ : SUCCESS in 4m 54s\n- openstack-tox-functional http://logs.openstack.org/45/660645/3/check/openstack-tox-functional/f438321/ : FAILURE in 8m 34s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/660645/3/check/openstack-tox-functional-py35/5adcb3d/ : FAILURE in 9m 04s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/45/660645/3/check/glance-eventlet-ssl-handshake-broken-py3/81596f3/ : FAILURE in 4m 20s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/45/660645/3/check/glance-code-constants-check/37df514/ : SUCCESS in 4m 32s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/660645/3/check/devstack-plugin-ceph-tempest/f98bb96/ : SUCCESS in 1h 26m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9ce2fdbefaf1753f1096a8abcf500152aa4e5d66","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-10 08:48:53.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/3/check-not-vote/ext-glance-zuul/5ba0418 : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":3},{"id":"7f3a428f5e51712a5b58308ede239ccc4a458267","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-18 04:55:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"173465941b733afe05fc96a00942a7eb219e947e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-18 05:23:08.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"3c387ba4bebccd76e4a3f9da3a5db268b8b72f93","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-18 05:58:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ff46e34daef230b2f3a1850258db1a7ca56a12c0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-18 08:53:39.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/5/check-not-vote/ext-glance-zuul/7af1bb7 : SUCCESS in 40m 53s","accounts_in_message":[],"_revision_number":5},{"id":"73b38ccb7bb2fb4d68c41ceb8224b90d61ea189c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 08:55:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/45/660645/5/check/tempest-full/d8d02b2/ : SUCCESS in 2h 02m 29s\n- neutron-grenade http://logs.openstack.org/45/660645/5/check/neutron-grenade/2c5863f/ : SUCCESS in 1h 13m 16s\n- grenade-py3 http://logs.openstack.org/45/660645/5/check/grenade-py3/f66c2f9/ : SUCCESS in 1h 06m 46s\n- tempest-full-py3 http://logs.openstack.org/45/660645/5/check/tempest-full-py3/69f00ca/ : SUCCESS in 1h 38m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/660645/5/check/openstack-tox-lower-constraints/01cea77/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/45/660645/5/check/openstack-tox-pep8/a549934/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/45/660645/5/check/openstack-tox-py27/8d86bc4/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/45/660645/5/check/openstack-tox-py35/e7228dd/ : SUCCESS in 5m 09s\n- openstack-tox-py36 http://logs.openstack.org/45/660645/5/check/openstack-tox-py36/22a9b21/ : SUCCESS in 4m 57s\n- openstack-tox-py37 http://logs.openstack.org/45/660645/5/check/openstack-tox-py37/ae67f0a/ : SUCCESS in 4m 57s\n- openstack-tox-docs http://logs.openstack.org/45/660645/5/check/openstack-tox-docs/ee1b432/html/ : SUCCESS in 4m 42s\n- openstack-tox-functional http://logs.openstack.org/45/660645/5/check/openstack-tox-functional/3139623/ : SUCCESS in 9m 06s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/660645/5/check/openstack-tox-functional-py35/56ced30/ : SUCCESS in 8m 50s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/45/660645/5/check/glance-eventlet-ssl-handshake-broken-py3/8b06d87/ : FAILURE in 4m 45s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/45/660645/5/check/glance-code-constants-check/e5f4a9b/ : SUCCESS in 4m 32s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/660645/5/check/devstack-plugin-ceph-tempest/e7b367f/ : FAILURE in 1h 44m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"79501c487e3bbecf7c2632f400a0f4fe747bdb4f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-18 08:57:04.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"949043f4a9d4abc08222d1f1600e429516e439fa","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2019-07-11 06:41:42.000000000","message":"Patch Set 5: Code-Review-1\n\nget_store_id_from_uri function will only verify the scheme of the backend, not if the file is really present, this would lead to an error if you use multiple backend of the same type (ex: 2 rbd backend). if the file is on the second backend, so the first one will be returned by your function, and glance will still trigger a 400 Invalid location when trying to effectively check the file later in the process.\n\nI recommend to use :\n\"get_size_from_uri_and_backend(url, backend)\" from glance_store instead, as it would really check the size of the file in the backend, and trigger an error is not found.\n\nFor me we need to look more at https://review.opendev.org/#/c/618932/2","accounts_in_message":[],"_revision_number":5},{"id":"57637b4656e671b26c3333cc2cb3236841c1b56f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-11 07:00:13.000000000","message":"Patch Set 5:\n\n\u003e get_store_id_from_uri function will only verify the scheme of the\n \u003e backend, not if the file is really present, this would lead to an\n \u003e error if you use multiple backend of the same type (ex: 2 rbd\n \u003e backend). if the file is on the second backend, so the first one\n \u003e will be returned by your function, and glance will still trigger a\n \u003e 400 Invalid location when trying to effectively check the file\n \u003e later in the process.\n \u003e \n \u003e I recommend to use :\n \u003e \"get_size_from_uri_and_backend(url, backend)\" from glance_store\n \u003e instead, as it would really check the size of the file in the\n \u003e backend, and trigger an error is not found.\n \u003e \n \u003e For me we need to look more at https://review.opendev.org/#/c/618932/2\n\nHi Victor,\n\nThis get_store_id_from_uri is not just matching with scheme it is actually matching entire url with the \u0027_url_prefix \u0027 of each store, have a look at [1] https://review.opendev.org/#/c/660340","accounts_in_message":[],"_revision_number":5},{"id":"b9e661e892c530b6bfc910b6d969ab3b5a4cb0d3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-11 07:14:12.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"f0c2839c68fa1f781c9616844a04c055307e0a9b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-11 07:14:32.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"c640c8a262504034aaa6dcdb920f5d78e8b23cd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 09:29:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/45/660645/6/check/tempest-full/194662e/ : TIMED_OUT in 2h 06m 50s\n- neutron-grenade http://logs.openstack.org/45/660645/6/check/neutron-grenade/9307f8a/ : SUCCESS in 1h 00m 15s\n- grenade-py3 http://logs.openstack.org/45/660645/6/check/grenade-py3/9aa990c/ : SUCCESS in 1h 04m 52s\n- tempest-full-py3 http://logs.openstack.org/45/660645/6/check/tempest-full-py3/84ace7d/ : SUCCESS in 1h 36m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/660645/6/check/openstack-tox-lower-constraints/eb34b83/ : SUCCESS in 5m 19s\n- openstack-tox-pep8 http://logs.openstack.org/45/660645/6/check/openstack-tox-pep8/a365f77/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/45/660645/6/check/openstack-tox-py27/a8a199b/ : SUCCESS in 3m 51s\n- openstack-tox-py35 http://logs.openstack.org/45/660645/6/check/openstack-tox-py35/0c670bb/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/45/660645/6/check/openstack-tox-py36/c63e2ce/ : SUCCESS in 3m 55s\n- openstack-tox-py37 http://logs.openstack.org/45/660645/6/check/openstack-tox-py37/b95cf41/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/45/660645/6/check/openstack-tox-docs/40d08ec/html/ : SUCCESS in 4m 38s\n- openstack-tox-functional http://logs.openstack.org/45/660645/6/check/openstack-tox-functional/ea769b3/ : SUCCESS in 8m 08s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/660645/6/check/openstack-tox-functional-py35/50c8693/ : SUCCESS in 8m 48s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/45/660645/6/check/glance-eventlet-ssl-handshake-broken-py3/1404f12/ : FAILURE in 4m 05s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/45/660645/6/check/glance-code-constants-check/3afd047/ : SUCCESS in 3m 44s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/660645/6/check/devstack-plugin-ceph-tempest/9e7e5e1/ : SUCCESS in 1h 31m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c20fa90890ec496efaa2d84aeef30fc49cc0060b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-11 09:40:47.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"9821fb2bb7350d5b131085ee532b923589e6d639","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-11 09:41:07.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"34762876d5a4738f7b20f75e5bd0a0fa928cf216","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 11:43:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/45/660645/6/check/tempest-full/28a49d6/ : SUCCESS in 1h 54m 21s\n- neutron-grenade http://logs.openstack.org/45/660645/6/check/neutron-grenade/00ee072/ : SUCCESS in 59m 59s\n- grenade-py3 http://logs.openstack.org/45/660645/6/check/grenade-py3/162d2b7/ : SUCCESS in 1h 00m 34s\n- tempest-full-py3 http://logs.openstack.org/45/660645/6/check/tempest-full-py3/2f4f4eb/ : SUCCESS in 1h 55m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/660645/6/check/openstack-tox-lower-constraints/ee362eb/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/45/660645/6/check/openstack-tox-pep8/d30bf4e/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/45/660645/6/check/openstack-tox-py27/80cd652/ : SUCCESS in 4m 09s\n- openstack-tox-py35 http://logs.openstack.org/45/660645/6/check/openstack-tox-py35/1f5027c/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/45/660645/6/check/openstack-tox-py36/60ca7b9/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/45/660645/6/check/openstack-tox-py37/a83e1e4/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/45/660645/6/check/openstack-tox-docs/a7891bb/html/ : SUCCESS in 6m 22s\n- openstack-tox-functional http://logs.openstack.org/45/660645/6/check/openstack-tox-functional/fee588e/ : SUCCESS in 7m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/660645/6/check/openstack-tox-functional-py35/093acec/ : SUCCESS in 8m 51s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/45/660645/6/check/glance-eventlet-ssl-handshake-broken-py3/f8ed4b0/ : FAILURE in 4m 10s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/45/660645/6/check/glance-code-constants-check/190b60c/ : SUCCESS in 5m 41s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/660645/6/check/devstack-plugin-ceph-tempest/eddfdf9/ : FAILURE in 1h 46m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"162fed402c08cb53fa5b18c7580e1438412c8104","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-19 13:34:00.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nTwo things:\n\n(1) something to look at noted inline\n\n(2) I was reading over Victor\u0027s comment, and while I think you are correct about the rbd case, from looking at https://review.opendev.org/#/c/660340/, it looks like Victor would be right about at least the http store and the cinder store (I didn\u0027t look through all of them).  Maybe configuring multiple http stores is a weird case (though I\u0027m not convinced of that), but I can see someone using multiple cinder stores.  I think this lazy update needs to be more robust.","accounts_in_message":[],"_revision_number":6},{"id":"7a0578efc860ea7ff04c691789e6b3f8941af249","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2019-07-19 15:28:58.000000000","message":"Patch Set 6:\n\nToo more things to say (sorry for that):\n- Can\u0027t we put the lazy_update_store_info() in a utils function ? Because it is exactly the same code as the _update_store_in_location() function in https://review.opendev.org/#/c/660645 (I will move my test then)\n- I agree that the RBD store should be ok with the url_prefix comparaison, but in case of ceph cluster not available at glance boot time, so you will return only \"rbd://\" as url_prefix, and you will have get the issue I explain previously, even though the ceph cluster is back online since. Does this code executed only at glance boot time or at every request ?","accounts_in_message":[],"_revision_number":6},{"id":"f3e74d08e6716390ff9782bbb3689716013b7cc2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-22 05:03:43.000000000","message":"Patch Set 6:\n\n\u003e Too more things to say (sorry for that):\n \u003e - Can\u0027t we put the lazy_update_store_info() in a utils function ?\n \u003e Because it is exactly the same code as the _update_store_in_location()\n \u003e function in https://review.opendev.org/#/c/660645 (I will move my\n \u003e test then)\n \u003e - I agree that the RBD store should be ok with the url_prefix\n \u003e comparaison, but in case of ceph cluster not available at glance\n \u003e boot time, so you will return only \"rbd://\" as url_prefix, and you\n \u003e will have get the issue I explain previously, even though the ceph\n \u003e cluster is back online since. Does this code executed only at\n \u003e glance boot time or at every request ?\nHi Victor,\n\nThe _url_prefix will be set while initializing the stores. Stores are initialized at the time of boot as well as post, update, delete  requests as well. I have tested this with 2 ceph nodes and never encountered the issue you have mentioned.","accounts_in_message":[],"_revision_number":6},{"id":"4f46bc5058fd637ee14d8dff160613f2327f0c2e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-22 05:07:05.000000000","message":"Patch Set 6:\n\n\u003e (1 comment)\n \u003e \n \u003e Two things:\n \u003e \n \u003e (1) something to look at noted inline\n \u003e \n \u003e (2) I was reading over Victor\u0027s comment, and while I think you are\n \u003e correct about the rbd case, from looking at https://review.opendev.org/#/c/660340/,\n \u003e it looks like Victor would be right about at least the http store\n \u003e and the cinder store (I didn\u0027t look through all of them).  Maybe\n \u003e configuring multiple http stores is a weird case (though I\u0027m not\n \u003e convinced of that), but I can see someone using multiple cinder\n \u003e stores.  I think this lazy update needs to be more robust.\n\nHi Brian,\n\nYou are right, for cinder and http stores this can be problematic only in case of locations operation. This lazy update function is robust at the moment. In order to support cinder and http multiple stores for add/replace location case we need to change the driver implementation and append a uuid to each store instead of just creating the url prefix like cinder:// and http://. Also we need to change the consumer side like cinder or nova as well.","accounts_in_message":[],"_revision_number":6},{"id":"fd8c5273dc2a356997121d4f82b713adb9182d6b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-22 05:11:25.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"9a0f16c65db5dd84a008a5314c45adbc16fb90c8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-22 05:13:04.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nHi Brian and Victor,\n\nThank you for review.\n\n@Victor, I can put a patch for refactoring the same later.\n@Brian, Added comments for your query.","accounts_in_message":[],"_revision_number":7},{"id":"931c90410ffc748b2212b2451aef5e03ab0fc8fa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-22 05:46:02.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/7/check-not-vote/ext-glance-zuul/016d154 : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":7},{"id":"2c07b8ce2b42e252fa1ffa3abbdea4dcdd72d3a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 07:12:24.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/45/660645/7/check/tempest-full/57ea79d/ : SUCCESS in 1h 59m 31s\n- neutron-grenade http://logs.openstack.org/45/660645/7/check/neutron-grenade/69a7849/ : SUCCESS in 55m 45s\n- grenade-py3 http://logs.openstack.org/45/660645/7/check/grenade-py3/4d1e81a/ : SUCCESS in 53m 08s\n- tempest-full-py3 http://logs.openstack.org/45/660645/7/check/tempest-full-py3/abc86dc/ : SUCCESS in 1h 16m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/660645/7/check/openstack-tox-lower-constraints/c33fa9f/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 finger://ze03.openstack.org/6696588b17a5489299ba3ba72d133420 : RETRY_LIMIT in 6m 50s\n- openstack-tox-py27 http://logs.openstack.org/45/660645/7/check/openstack-tox-py27/9f69fb1/ : SUCCESS in 4m 53s\n- openstack-tox-py35 http://logs.openstack.org/45/660645/7/check/openstack-tox-py35/a5be823/ : SUCCESS in 5m 48s\n- openstack-tox-py36 finger://ze01.openstack.org/8ee34c08f40b441ead77a429fe9288e9 : RETRY_LIMIT in 6m 22s\n- openstack-tox-py37 http://logs.openstack.org/45/660645/7/check/openstack-tox-py37/9c5abe1/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/45/660645/7/check/openstack-tox-docs/6e4f812/html/ : SUCCESS in 5m 26s\n- openstack-tox-functional finger://ze02.openstack.org/7c9fe11ad40f4dee857d6d4085a28849 : RETRY_LIMIT in 11m 41s\n- openstack-tox-functional-py35 finger://ze03.openstack.org/a03dc37268724ada89c35bfdec238da8 : RETRY_LIMIT in 15m 19s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/45/660645/7/check/glance-eventlet-ssl-handshake-broken-py3/6190792/ : FAILURE in 4m 55s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/45/660645/7/check/glance-code-constants-check/32ff798/ : SUCCESS in 5m 05s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/660645/7/check/devstack-plugin-ceph-tempest/2fd035b/ : SUCCESS in 1h 14m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"31c99427299a8c1e313d1c3feaec6b387c2c7a7e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-22 07:41:16.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"f915c04e6161cf5ef2929eb13939d73b3814764a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 09:28:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/45/660645/7/check/tempest-full/d8d4e28/ : SUCCESS in 1h 45m 21s\n- neutron-grenade http://logs.openstack.org/45/660645/7/check/neutron-grenade/7588039/ : SUCCESS in 1h 02m 25s\n- grenade-py3 http://logs.openstack.org/45/660645/7/check/grenade-py3/46082f8/ : SUCCESS in 1h 04m 14s\n- tempest-full-py3 http://logs.openstack.org/45/660645/7/check/tempest-full-py3/11723bb/ : SUCCESS in 1h 24m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/660645/7/check/openstack-tox-lower-constraints/9f97ae5/ : SUCCESS in 4m 31s\n- openstack-tox-pep8 http://logs.openstack.org/45/660645/7/check/openstack-tox-pep8/4bb2773/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/45/660645/7/check/openstack-tox-py27/4babbeb/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/45/660645/7/check/openstack-tox-py35/5c2340c/ : SUCCESS in 6m 00s\n- openstack-tox-py36 http://logs.openstack.org/45/660645/7/check/openstack-tox-py36/187b16b/ : SUCCESS in 6m 42s\n- openstack-tox-py37 http://logs.openstack.org/45/660645/7/check/openstack-tox-py37/0d32349/ : SUCCESS in 8m 13s\n- openstack-tox-docs http://logs.openstack.org/45/660645/7/check/openstack-tox-docs/55dd1fd/html/ : SUCCESS in 5m 36s\n- openstack-tox-functional http://logs.openstack.org/45/660645/7/check/openstack-tox-functional/45ea615/ : SUCCESS in 9m 48s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/660645/7/check/openstack-tox-functional-py35/a46a2c5/ : SUCCESS in 9m 11s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/45/660645/7/check/glance-eventlet-ssl-handshake-broken-py3/47a2307/ : FAILURE in 6m 36s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/45/660645/7/check/glance-code-constants-check/30e165e/ : SUCCESS in 5m 09s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/660645/7/check/devstack-plugin-ceph-tempest/8e29589/ : SUCCESS in 1h 22m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"be4b810fdd674acfc2602de42107516e886e8e84","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2019-07-22 11:39:13.000000000","message":"Patch Set 7: Code-Review+1\n\nIt\u0027s ok for my two point then, thanks for your answers !\nWe need to do refacto in https://review.opendev.org/#/c/660645 then.","accounts_in_message":[],"_revision_number":7},{"id":"07c6dbedd2a4c41818f239a07ea7e5d8499874fb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-22 12:07:15.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/7/check-not-vote/ext-glance-zuul/9e7331c : SUCCESS in 41m 34s","accounts_in_message":[],"_revision_number":7},{"id":"d49cf2872d1c206f43c74529c3fd764d126667f2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-24 16:59:29.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nI have a question about this, please see inline.  (-1 just to get your attention, I\u0027m not sure this is really a problem.)","accounts_in_message":[],"_revision_number":7},{"id":"a72598dc9a22388f945868e3fa9a973c6e62af28","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-24 18:05:03.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nHi Brian,\n\nPlease find the inline answer for your doubt.","accounts_in_message":[],"_revision_number":7},{"id":"bba63a15d1b419e76617c66905c6cb037325ddc7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-24 18:15:10.000000000","message":"Patch Set 7: -Code-Review\n\nThanks for the explanation, I missed that the path was included in the url_prefix.  Makes sense now.","accounts_in_message":[],"_revision_number":7},{"id":"e094bf1b97167b4c5e4f74c1efc621fd9c51fa40","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-24 18:16:01.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nOne last thing inline.","accounts_in_message":[],"_revision_number":7},{"id":"50393e046dd8954155890ab435d288cd1c09cc66","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-24 18:27:21.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"30be1359209c73cf45706622e9102be7e098913e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-24 18:27:42.000000000","message":"Patch Set 8:\n\nCorrected a typo in the unit test.","accounts_in_message":[],"_revision_number":8},{"id":"f9976f0a0e2da7c908453cdd91b1ce6fd42e55f7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-24 20:21:03.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/8/check-not-vote/ext-glance-zuul/c1b0d2f : SUCCESS in 43m 31s","accounts_in_message":[],"_revision_number":8},{"id":"88eb39c90e145a6c9ec331a3800cb45fd42309de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 21:01:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/45/660645/8/check/tempest-full/dd4dff1/ : SUCCESS in 1h 58m 13s\n- neutron-grenade http://logs.openstack.org/45/660645/8/check/neutron-grenade/56497ca/ : SUCCESS in 57m 20s\n- grenade-py3 http://logs.openstack.org/45/660645/8/check/grenade-py3/1720c46/ : SUCCESS in 1h 01m 05s\n- tempest-full-py3 http://logs.openstack.org/45/660645/8/check/tempest-full-py3/7eaab89/ : SUCCESS in 1h 26m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/660645/8/check/openstack-tox-lower-constraints/3a948ef/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/45/660645/8/check/openstack-tox-pep8/d60aea7/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/45/660645/8/check/openstack-tox-py27/7907689/ : SUCCESS in 4m 44s\n- openstack-tox-py35 http://logs.openstack.org/45/660645/8/check/openstack-tox-py35/06075cd/ : SUCCESS in 6m 26s\n- openstack-tox-py36 http://logs.openstack.org/45/660645/8/check/openstack-tox-py36/a39d09f/ : SUCCESS in 5m 03s\n- openstack-tox-py37 http://logs.openstack.org/45/660645/8/check/openstack-tox-py37/a17dc38/ : SUCCESS in 6m 36s\n- openstack-tox-docs http://logs.openstack.org/45/660645/8/check/openstack-tox-docs/b83a827/html/ : SUCCESS in 5m 59s\n- openstack-tox-functional http://logs.openstack.org/45/660645/8/check/openstack-tox-functional/1b547e1/ : SUCCESS in 10m 31s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/660645/8/check/openstack-tox-functional-py35/59e8de8/ : SUCCESS in 10m 18s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/45/660645/8/check/glance-eventlet-ssl-handshake-broken-py3/8b80170/ : FAILURE in 5m 41s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/45/660645/8/check/glance-code-constants-check/7cc1ec3/ : SUCCESS in 4m 47s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/660645/8/check/devstack-plugin-ceph-tempest/c419f05/ : SUCCESS in 1h 16m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"11c63fe384946babc38efb57ddf3c73e7ef521b9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-24 21:03:32.000000000","message":"Patch Set 8: Code-Review+2\n\nI only fixed a typo, so I feel OK giving this a +2.","accounts_in_message":[],"_revision_number":8},{"id":"537b2cab5a5ffe59934d297f59a046f8f87b4399","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-07-25 11:38:44.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)\n\nI have concern about the logging level here as well. Other than that it looks good to me.","accounts_in_message":[],"_revision_number":8},{"id":"91084cc7f61ce959743b6e95f510e1c28296eecf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-25 12:44:39.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"f6ac78c71785629188de6afa14d991fca0188440","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-25 14:03:22.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"77a5ae9f57fb9accf0d40762f6165fddfe1fb788","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-25 15:48:27.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/10/check-not-vote/ext-glance-zuul/da0cc63 : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":10},{"id":"7179f0003905502a9f10e9eee4d78e84340f0154","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 16:53:34.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/45/660645/10/check/tempest-full/be9e32f/ : SUCCESS in 1h 44m 08s\n- neutron-grenade http://logs.openstack.org/45/660645/10/check/neutron-grenade/2fc13ff/ : SUCCESS in 58m 49s\n- grenade-py3 http://logs.openstack.org/45/660645/10/check/grenade-py3/ecbc796/ : SUCCESS in 56m 36s\n- tempest-full-py3 http://logs.openstack.org/45/660645/10/check/tempest-full-py3/59441d3/ : SUCCESS in 1h 33m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/660645/10/check/openstack-tox-lower-constraints/76f89ed/ : SUCCESS in 5m 14s\n- openstack-tox-pep8 http://logs.openstack.org/45/660645/10/check/openstack-tox-pep8/24b319c/ : SUCCESS in 5m 58s\n- openstack-tox-py27 http://logs.openstack.org/45/660645/10/check/openstack-tox-py27/234144d/ : SUCCESS in 5m 23s\n- openstack-tox-py35 http://logs.openstack.org/45/660645/10/check/openstack-tox-py35/efd070c/ : SUCCESS in 7m 46s\n- openstack-tox-py36 http://logs.openstack.org/45/660645/10/check/openstack-tox-py36/8458c5d/ : SUCCESS in 5m 27s\n- openstack-tox-py37 http://logs.openstack.org/45/660645/10/check/openstack-tox-py37/b0822a5/ : SUCCESS in 8m 44s\n- openstack-tox-docs http://logs.openstack.org/45/660645/10/check/openstack-tox-docs/7a5404c/html/ : SUCCESS in 5m 39s\n- openstack-tox-functional http://logs.openstack.org/45/660645/10/check/openstack-tox-functional/bdda7f8/ : SUCCESS in 10m 38s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/660645/10/check/openstack-tox-functional-py35/451fa8a/ : SUCCESS in 10m 50s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/45/660645/10/check/glance-eventlet-ssl-handshake-broken-py3/9cae753/ : FAILURE in 5m 27s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/45/660645/10/check/glance-code-constants-check/ce0ab4b/ : SUCCESS in 8m 06s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/660645/10/check/devstack-plugin-ceph-tempest/aec4091/ : SUCCESS in 1h 23m 24s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6521b342708141110a4d813837d22c425f26d56d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-30 06:17:59.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"a1f181781430690282c0c5bd397b03dce2a4171c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-30 06:19:01.000000000","message":"Patch Set 11:\n\nMade it dependent on https://review.opendev.org/#/c/658962/","accounts_in_message":[],"_revision_number":11},{"id":"38482c02712a4ea0b987b80f9fc0fd6e417777e2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-30 06:19:12.000000000","message":"Patch Set 11: Review-Priority+1","accounts_in_message":[],"_revision_number":11},{"id":"1373a9fa6acae6b5af703e0b918522765d64c83d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-30 08:55:23.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/11/check-not-vote/ext-glance-zuul/b7c8a4f : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":11},{"id":"00e8c447367cedf86251319fa2bbfa4dace4a633","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 09:15:56.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/45/660645/11/check/tempest-full/31406e2/ : SUCCESS in 1h 42m 36s\n- neutron-grenade http://logs.openstack.org/45/660645/11/check/neutron-grenade/aa80d29/ : SUCCESS in 1h 03m 43s\n- grenade-py3 http://logs.openstack.org/45/660645/11/check/grenade-py3/c46b1a8/ : SUCCESS in 1h 02m 07s\n- tempest-full-py3 http://logs.openstack.org/45/660645/11/check/tempest-full-py3/4b84737/ : SUCCESS in 1h 25m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/660645/11/check/openstack-tox-lower-constraints/d0556bc/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/45/660645/11/check/openstack-tox-pep8/59474b2/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/45/660645/11/check/openstack-tox-py27/3136f3f/ : SUCCESS in 4m 14s\n- openstack-tox-py35 http://logs.openstack.org/45/660645/11/check/openstack-tox-py35/f3ee5ef/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/45/660645/11/check/openstack-tox-py36/411ab7b/ : SUCCESS in 4m 56s\n- openstack-tox-py37 http://logs.openstack.org/45/660645/11/check/openstack-tox-py37/28b2442/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/45/660645/11/check/openstack-tox-docs/1cf32e5/html/ : SUCCESS in 7m 00s\n- openstack-tox-functional http://logs.openstack.org/45/660645/11/check/openstack-tox-functional/afdffbc/ : SUCCESS in 11m 20s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/660645/11/check/openstack-tox-functional-py35/f47e42e/ : SUCCESS in 9m 02s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/45/660645/11/check/glance-eventlet-ssl-handshake-broken-py3/30c9c17/ : FAILURE in 4m 51s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/45/660645/11/check/glance-code-constants-check/c9a3cfe/ : SUCCESS in 6m 40s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/660645/11/check/devstack-plugin-ceph-tempest/6541f74/ : SUCCESS in 1h 19m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7186b1f590fd57bbdad350071a26ec79e873e6d2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-30 13:28:16.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nQuestion about the logging inline.  Otherwise, this looks good.","accounts_in_message":[],"_revision_number":11},{"id":"a526636b32708a778f907d8447d6a938d6d6c197","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-30 17:00:29.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"061cd9b34a008e508fa933180775ab1c086867e3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-30 17:23:12.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"7ed5cb944230aa1f74d24c5556b1887781887cfc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-30 17:24:04.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nThank you Brian for review. Made changes as per the suggestion.","accounts_in_message":[],"_revision_number":11},{"id":"7b290f2856b42aa7614ddb1f2086acaadaad764f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-30 17:24:20.000000000","message":"Patch Set 13: Review-Priority+1","accounts_in_message":[],"_revision_number":13},{"id":"a874c47044b4b07e435020ac2087bc892d532abd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-30 19:23:13.000000000","message":"Patch Set 13:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/13/check-not-vote/ext-glance-zuul/d468c8c : SUCCESS in 41m 56s","accounts_in_message":[],"_revision_number":13},{"id":"a96b090b47bbaaa110fe76e062387a5029042710","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 21:58:12.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/45/660645/13/check/tempest-full/fe28f67/ : SUCCESS in 1h 28m 44s\n- neutron-grenade http://logs.openstack.org/45/660645/13/check/neutron-grenade/c697443/ : SUCCESS in 58m 47s\n- grenade-py3 http://logs.openstack.org/45/660645/13/check/grenade-py3/d8de0d0/ : SUCCESS in 59m 52s\n- tempest-full-py3 http://logs.openstack.org/45/660645/13/check/tempest-full-py3/b1714cb/ : SUCCESS in 1h 24m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/660645/13/check/openstack-tox-lower-constraints/76107cd/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/45/660645/13/check/openstack-tox-pep8/a92a108/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/45/660645/13/check/openstack-tox-py27/06388df/ : SUCCESS in 5m 21s\n- openstack-tox-py35 http://logs.openstack.org/45/660645/13/check/openstack-tox-py35/f769927/ : SUCCESS in 7m 37s\n- openstack-tox-py36 http://logs.openstack.org/45/660645/13/check/openstack-tox-py36/e7a9151/ : SUCCESS in 5m 07s\n- openstack-tox-py37 http://logs.openstack.org/45/660645/13/check/openstack-tox-py37/b807ed9/ : SUCCESS in 4m 48s\n- openstack-tox-docs http://logs.openstack.org/45/660645/13/check/openstack-tox-docs/24fc9f0/html/ : SUCCESS in 6m 03s\n- openstack-tox-functional http://logs.openstack.org/45/660645/13/check/openstack-tox-functional/ef81008/ : SUCCESS in 9m 20s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/660645/13/check/openstack-tox-functional-py35/5839cb9/ : FAILURE in 9m 21s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/45/660645/13/check/glance-eventlet-ssl-handshake-broken-py3/48ec86a/ : FAILURE in 6m 15s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/45/660645/13/check/glance-code-constants-check/622c1c3/ : SUCCESS in 4m 08s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/660645/13/check/devstack-plugin-ceph-tempest/084229b/ : SUCCESS in 1h 13m 35s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"db603dde049a75f35ad8eae8fb6cceb605d2a980","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-31 04:24:36.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"cd6d4070be53b05fb557d55f71785b77d751a1c7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-31 05:12:35.000000000","message":"Patch Set 13:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/13/check-not-vote/ext-glance-zuul/d1be3b3 : SUCCESS in 47m 33s","accounts_in_message":[],"_revision_number":13},{"id":"227dabd1494664bad0ce11c8a1902d4f493e6c3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 06:24:37.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/45/660645/13/check/tempest-full/9a68db1/ : SUCCESS in 1h 57m 10s\n- neutron-grenade http://logs.openstack.org/45/660645/13/check/neutron-grenade/87d31d1/ : SUCCESS in 1h 14m 34s\n- grenade-py3 http://logs.openstack.org/45/660645/13/check/grenade-py3/effed9d/ : SUCCESS in 1h 15m 14s\n- tempest-full-py3 http://logs.openstack.org/45/660645/13/check/tempest-full-py3/9e6eb08/ : SUCCESS in 1h 51m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/660645/13/check/openstack-tox-lower-constraints/22725f1/ : SUCCESS in 5m 15s\n- openstack-tox-pep8 http://logs.openstack.org/45/660645/13/check/openstack-tox-pep8/5d098e2/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/45/660645/13/check/openstack-tox-py27/a22649d/ : SUCCESS in 6m 22s\n- openstack-tox-py35 http://logs.openstack.org/45/660645/13/check/openstack-tox-py35/8efbaa3/ : SUCCESS in 6m 09s\n- openstack-tox-py36 http://logs.openstack.org/45/660645/13/check/openstack-tox-py36/059dd0f/ : SUCCESS in 4m 44s\n- openstack-tox-py37 http://logs.openstack.org/45/660645/13/check/openstack-tox-py37/d388666/ : SUCCESS in 7m 00s\n- openstack-tox-docs http://logs.openstack.org/45/660645/13/check/openstack-tox-docs/822cdea/html/ : SUCCESS in 5m 14s\n- openstack-tox-functional http://logs.openstack.org/45/660645/13/check/openstack-tox-functional/d1c0507/ : SUCCESS in 9m 56s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/660645/13/check/openstack-tox-functional-py35/91e7582/ : SUCCESS in 10m 24s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/45/660645/13/check/glance-eventlet-ssl-handshake-broken-py3/26d817c/ : FAILURE in 4m 10s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/45/660645/13/check/glance-code-constants-check/de22627/ : SUCCESS in 5m 40s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/660645/13/check/devstack-plugin-ceph-tempest/630536e/ : SUCCESS in 1h 07m 10s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"aeb6b70cc4afc3cfc8d7271527bda3803bcb62ab","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-01 16:39:57.000000000","message":"Patch Set 13:\n\n(2 comments)\n\nSomething to look at inline.  Not sure if it\u0027s a real problem or not.","accounts_in_message":[],"_revision_number":13},{"id":"85319c8e45a1bc8a06a11cd9e37720dac69d5e7a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-01 16:51:20.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"d0015a677ec039622e80429da21851529c5a4ef8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-01 17:57:38.000000000","message":"Patch Set 13:\n\n(2 comments)\n\nBrian, Sean,\n\nThank you for reviews, made changes as per your suggestions.","accounts_in_message":[],"_revision_number":13},{"id":"9ea042778695eea3ec2ce5cff014a70378479354","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-01 18:00:14.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"b2ec87d4c817871f38fdf0142e2c32c45e992554","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-01 18:45:33.000000000","message":"Patch Set 14:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/14/check-not-vote/ext-glance-zuul/19273eb : FAILURE in 44m 52s","accounts_in_message":[],"_revision_number":14},{"id":"be36ac36e54e0c57a6bc6eec69aac1577d1e5075","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 23:22:10.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://logs.opendev.org/45/660645/14/check/tempest-full/740fcb6/ : SUCCESS in 1h 40m 31s\n- neutron-grenade https://logs.opendev.org/45/660645/14/check/neutron-grenade/20b681a/ : SUCCESS in 1h 03m 20s\n- grenade-py3 https://logs.opendev.org/45/660645/14/check/grenade-py3/359e200/ : SUCCESS in 59m 47s\n- tempest-full-py3 https://logs.opendev.org/45/660645/14/check/tempest-full-py3/da344fc/ : SUCCESS in 1h 23m 28s\n- openstack-tox-lower-constraints https://logs.opendev.org/45/660645/14/check/openstack-tox-lower-constraints/9b84035/ : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://logs.opendev.org/45/660645/14/check/openstack-tox-pep8/547e14b/ : SUCCESS in 4m 15s\n- openstack-tox-py27 https://logs.opendev.org/45/660645/14/check/openstack-tox-py27/5c615a7/ : SUCCESS in 4m 24s\n- openstack-tox-py35 https://logs.opendev.org/45/660645/14/check/openstack-tox-py35/a294cf9/ : SUCCESS in 4m 25s\n- openstack-tox-py36 https://logs.opendev.org/45/660645/14/check/openstack-tox-py36/b91c4ce/ : SUCCESS in 5m 19s\n- openstack-tox-py37 https://logs.opendev.org/45/660645/14/check/openstack-tox-py37/e3658a7/ : SUCCESS in 7m 18s\n- openstack-tox-docs https://logs.opendev.org/45/660645/14/check/openstack-tox-docs/003d4d7/html/ : SUCCESS in 6m 04s\n- openstack-tox-functional https://logs.opendev.org/45/660645/14/check/openstack-tox-functional/3939655/ : SUCCESS in 10m 08s\n- openstack-tox-functional-py35 https://logs.opendev.org/45/660645/14/check/openstack-tox-functional-py35/910d2ba/ : SUCCESS in 8m 45s\n- glance-eventlet-ssl-handshake-broken-py3 https://logs.opendev.org/45/660645/14/check/glance-eventlet-ssl-handshake-broken-py3/2f2e273/ : FAILURE in 4m 38s (non-voting)\n- glance-code-constants-check https://logs.opendev.org/45/660645/14/check/glance-code-constants-check/e13f5c5/ : SUCCESS in 6m 07s\n- devstack-plugin-ceph-tempest https://logs.opendev.org/45/660645/14/check/devstack-plugin-ceph-tempest/06f4bf6/ : SUCCESS in 1h 56m 42s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"578e12433492323592bdb0f90ffa8599883cb9ba","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-05 12:30:20.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)\n\nReally minor thing noted inline.","accounts_in_message":[],"_revision_number":14},{"id":"4460c7a4cd78c10620c4f5c108f5204a51cfdc42","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-06 05:42:04.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"75155b71f13dadd5043bb48da098d6644e502014","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-06 05:42:45.000000000","message":"Patch Set 14:\n\n(1 comment)\n\nBrian,\nThank you for review!!!","accounts_in_message":[],"_revision_number":14},{"id":"4236c4fb5130b5f1e332f5453d0dcc5b74072d60","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-06 05:52:09.000000000","message":"Patch Set 15:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/15/check-not-vote/ext-glance-zuul/4df7a23 : FAILURE in 1m 30s","accounts_in_message":[],"_revision_number":15},{"id":"60429072963714aab3c611221ef4678dc10b1216","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 08:11:01.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://logs.opendev.org/45/660645/15/check/tempest-full/f525475/ : SUCCESS in 1h 39m 18s\n- neutron-grenade https://logs.opendev.org/45/660645/15/check/neutron-grenade/7fd5061/ : SUCCESS in 57m 54s\n- grenade-py3 https://logs.opendev.org/45/660645/15/check/grenade-py3/d4ec0c1/ : SUCCESS in 1h 05m 16s\n- tempest-full-py3 https://logs.opendev.org/45/660645/15/check/tempest-full-py3/7b09bd9/ : SUCCESS in 1h 42m 10s\n- openstack-tox-lower-constraints https://logs.opendev.org/45/660645/15/check/openstack-tox-lower-constraints/dcfa8af/ : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://logs.opendev.org/45/660645/15/check/openstack-tox-pep8/3c911ef/ : SUCCESS in 11m 00s\n- openstack-tox-py27 https://logs.opendev.org/45/660645/15/check/openstack-tox-py27/f12c74c/ : SUCCESS in 16m 14s\n- openstack-tox-py35 https://logs.opendev.org/45/660645/15/check/openstack-tox-py35/ea37591/ : SUCCESS in 5m 24s\n- openstack-tox-py36 https://logs.opendev.org/45/660645/15/check/openstack-tox-py36/24f490c/ : SUCCESS in 5m 41s\n- openstack-tox-py37 https://logs.opendev.org/45/660645/15/check/openstack-tox-py37/183f64f/ : SUCCESS in 4m 55s\n- openstack-tox-docs https://logs.opendev.org/45/660645/15/check/openstack-tox-docs/053092d/html/ : SUCCESS in 5m 43s\n- openstack-tox-functional https://logs.opendev.org/45/660645/15/check/openstack-tox-functional/f43092d/ : SUCCESS in 10m 43s\n- openstack-tox-functional-py35 https://logs.opendev.org/45/660645/15/check/openstack-tox-functional-py35/f02dfeb/ : SUCCESS in 10m 56s\n- glance-eventlet-ssl-handshake-broken-py3 https://logs.opendev.org/45/660645/15/check/glance-eventlet-ssl-handshake-broken-py3/5fb65f8/ : FAILURE in 5m 24s (non-voting)\n- glance-code-constants-check https://logs.opendev.org/45/660645/15/check/glance-code-constants-check/d41c270/ : SUCCESS in 6m 40s\n- devstack-plugin-ceph-tempest https://logs.opendev.org/45/660645/15/check/devstack-plugin-ceph-tempest/608ff8a/ : SUCCESS in 1h 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"87660d1ac73636b5d58e1fb060512995259737eb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-06 08:25:09.000000000","message":"Patch Set 15: Review-Priority+1","accounts_in_message":[],"_revision_number":15},{"id":"c6e5d566feaad6d74be990d5cbee854c7628a79f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-06 15:17:58.000000000","message":"Patch Set 15: Code-Review+2\n\nAll my concerns have been addressed.","accounts_in_message":[],"_revision_number":15},{"id":"efcd2a77ebd2b55d93098e74ce88f5fc4a59f783","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-08 04:54:53.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"5f67a0fd89e0842c6bb6d95b21a60e7e60194a41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 06:31:16.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://logs.opendev.org/45/660645/16/check/tempest-full/42da860/ : SUCCESS in 1h 33m 35s\n- neutron-grenade https://logs.opendev.org/45/660645/16/check/neutron-grenade/83a736b/ : SUCCESS in 54m 41s\n- grenade-py3 https://logs.opendev.org/45/660645/16/check/grenade-py3/6d932da/ : SUCCESS in 55m 30s\n- tempest-full-py3 https://logs.opendev.org/45/660645/16/check/tempest-full-py3/3d215cf/ : SUCCESS in 1h 26m 35s\n- openstack-tox-lower-constraints https://logs.opendev.org/45/660645/16/check/openstack-tox-lower-constraints/06be3e8/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://logs.opendev.org/45/660645/16/check/openstack-tox-pep8/8d394a4/ : SUCCESS in 6m 42s\n- openstack-tox-py27 https://logs.opendev.org/45/660645/16/check/openstack-tox-py27/c24ca70/ : SUCCESS in 5m 08s\n- openstack-tox-py35 https://logs.opendev.org/45/660645/16/check/openstack-tox-py35/16f8ec7/ : SUCCESS in 4m 47s\n- openstack-tox-py36 https://logs.opendev.org/45/660645/16/check/openstack-tox-py36/d0ee088/ : SUCCESS in 4m 02s\n- openstack-tox-py37 https://logs.opendev.org/45/660645/16/check/openstack-tox-py37/71293d0/ : SUCCESS in 5m 48s\n- openstack-tox-docs https://logs.opendev.org/45/660645/16/check/openstack-tox-docs/9918fdb/html/ : SUCCESS in 4m 57s\n- openstack-tox-functional https://logs.opendev.org/45/660645/16/check/openstack-tox-functional/084d266/ : SUCCESS in 10m 05s\n- openstack-tox-functional-py35 https://logs.opendev.org/45/660645/16/check/openstack-tox-functional-py35/2baf820/ : SUCCESS in 9m 26s\n- glance-eventlet-ssl-handshake-broken-py3 https://logs.opendev.org/45/660645/16/check/glance-eventlet-ssl-handshake-broken-py3/1d8a0e1/ : FAILURE in 6m 22s (non-voting)\n- glance-code-constants-check https://logs.opendev.org/45/660645/16/check/glance-code-constants-check/b086a6e/ : SUCCESS in 5m 16s\n- devstack-plugin-ceph-tempest https://logs.opendev.org/45/660645/16/check/devstack-plugin-ceph-tempest/5d7e46b/ : SUCCESS in 1h 07m 27s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"3d5b2644f5b92b4852ccbd46c1c5e0c42552e5f6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-09 04:08:29.000000000","message":"Patch Set 16:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/45/660645/16/check-not-vote/ext-glance-zuul/1ce948f : FAILURE in 1m 28s","accounts_in_message":[],"_revision_number":16},{"id":"7ac358994c4fc532def95673c8bef6f99c29aa8d","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-08-13 15:56:06.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"929f48d68a92e33a0828f8bc1252b645bde1e742","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-08-15 09:28:10.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\nLGTM, thanks Abhishek!","accounts_in_message":[],"_revision_number":16},{"id":"8115a49e74cc9806cce24c8d9f4bffc08e6c2923","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 14:52:48.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"eac6ac6207d70b83aa8c8098941c8fc0ad1e0b84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:05:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"5b328be616e330830ab8a443707d981c041344ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:05:28.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f0765df240c14e40b3cce83e3cec2ba5 : SUCCESS in 1h 56m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7c16448b828a4f7597b657e52e708e25 : SUCCESS in 1h 08m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/368df19d622842d3a82362030198d5cc : SUCCESS in 1h 04m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17927e97a85641f0a59d18222abb9101 : SUCCESS in 1h 37m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/600d9946024646179ace7d215620edc3 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74269aaa9da74e1b92fc35c1e08983a3 : SUCCESS in 7m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16c918e6eb2d40a8961451456c51640d : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d1e7a0e3457044feb24d526ce1b51bff : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eac0991bdbe402d810efd40d522e1fd : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7176cee76729480b92ceb79fb2d46bd3 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58183519294c43419bb416416301a4d0 : SUCCESS in 8m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/13be68c310244ec2a493abebee1d5509 : SUCCESS in 10m 11s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b6a57f9b5e564d4e8b48ecd8a13144d5 : SUCCESS in 10m 11s","accounts_in_message":[],"_revision_number":16},{"id":"8da3100aa9e0ef4305b343cd5502d44e49ca2b1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:10:49.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5564d685eaf2468c8c5ef4f049b2d279 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"0e5984edc707a980028e1aa9b0d48d57ed3908d5","revisions":{"55d2a2ab6b51a59de3625852585d2d40664d8409":{"kind":"REWORK","_number":1,"created":"2019-05-22 10:30:33.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/1"}}},"commit":{"parents":[{"commit":"61298a47e4abd903fe38957ffbbbcf036e79bb3e","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61298a47e4abd903fe38957ffbbbcf036e79bb3e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-22 10:26:49.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55d2a2ab6b51a59de3625852585d2d40664d8409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55d2a2ab6b51a59de3625852585d2d40664d8409"}]},"branch":"refs/heads/master"},"1945344de5308fd713409efd8b589c084e7c2084":{"kind":"REWORK","_number":2,"created":"2019-06-04 05:42:35.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/2"}}},"commit":{"parents":[{"commit":"d99cea0cc76e981debece97b3d9fd0e5ef6434d2","subject":"Merge \"Mark http store read-only in discovery call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d99cea0cc76e981debece97b3d9fd0e5ef6434d2"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-04 05:29:27.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1945344de5308fd713409efd8b589c084e7c2084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1945344de5308fd713409efd8b589c084e7c2084"}]},"branch":"refs/heads/master"},"af8cdd90504ec80cff3c776f951870b5b5be7989":{"kind":"REWORK","_number":3,"created":"2019-06-10 06:29:16.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/3"}}},"commit":{"parents":[{"commit":"bc84af70e4d8c836be169f159a25c5611ade435f","subject":"Change lacation metadata key \u0027backend\u0027 to \u0027store\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc84af70e4d8c836be169f159a25c5611ade435f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-10 06:21:19.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/af8cdd90504ec80cff3c776f951870b5b5be7989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/af8cdd90504ec80cff3c776f951870b5b5be7989"}]},"branch":"refs/heads/master"},"7128e8295edf46a91804bb81a357080277579016":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-06-18 05:23:08.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/4"}}},"commit":{"parents":[{"commit":"b5b827f9eabd5b85e56ef522f94bb971f5308091","subject":"Merge \"Bring py27 tests back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5b827f9eabd5b85e56ef522f94bb971f5308091"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-18 05:14:11.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7128e8295edf46a91804bb81a357080277579016"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7128e8295edf46a91804bb81a357080277579016"}]},"branch":"refs/heads/master"},"69bef0982f330b68170c6d77b6996b35dcfad9f5":{"kind":"REWORK","_number":5,"created":"2019-06-18 05:58:23.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/5"}}},"commit":{"parents":[{"commit":"b5b827f9eabd5b85e56ef522f94bb971f5308091","subject":"Merge \"Bring py27 tests back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5b827f9eabd5b85e56ef522f94bb971f5308091"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-18 05:53:35.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69bef0982f330b68170c6d77b6996b35dcfad9f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69bef0982f330b68170c6d77b6996b35dcfad9f5"}]},"branch":"refs/heads/master"},"32598c624e760ef66c130f668a9e729207efb7ac":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-07-11 07:14:12.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/6"}}},"commit":{"parents":[{"commit":"d319acbd649720bc459e929eaa7ee04c3acf0a86","subject":"Update the glance installation page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d319acbd649720bc459e929eaa7ee04c3acf0a86"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-11 06:19:04.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/32598c624e760ef66c130f668a9e729207efb7ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/32598c624e760ef66c130f668a9e729207efb7ac"}]},"branch":"refs/heads/master"},"0244da4fd1afe3fb2d4fae966a622a407a06fc1d":{"kind":"REWORK","_number":7,"created":"2019-07-22 05:11:25.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/7"}}},"commit":{"parents":[{"commit":"d319acbd649720bc459e929eaa7ee04c3acf0a86","subject":"Update the glance installation page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d319acbd649720bc459e929eaa7ee04c3acf0a86"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-22 05:11:19.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0244da4fd1afe3fb2d4fae966a622a407a06fc1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0244da4fd1afe3fb2d4fae966a622a407a06fc1d"}]},"branch":"refs/heads/master"},"f2785ab8b15a7f960aafe538ea50c4bdd91ac7b9":{"kind":"REWORK","_number":8,"created":"2019-07-24 18:27:21.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/45/660645/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/8"}}},"commit":{"parents":[{"commit":"d319acbd649720bc459e929eaa7ee04c3acf0a86","subject":"Update the glance installation page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d319acbd649720bc459e929eaa7ee04c3acf0a86"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-07-24 18:27:04.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2785ab8b15a7f960aafe538ea50c4bdd91ac7b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2785ab8b15a7f960aafe538ea50c4bdd91ac7b9"}]},"branch":"refs/heads/master"},"3a6cbab613b769352b64e877b7c5b87eaa4a8c1f":{"kind":"REWORK","_number":9,"created":"2019-07-25 12:44:39.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/9"}}},"commit":{"parents":[{"commit":"d319acbd649720bc459e929eaa7ee04c3acf0a86","subject":"Update the glance installation page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d319acbd649720bc459e929eaa7ee04c3acf0a86"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-25 12:44:31.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a6cbab613b769352b64e877b7c5b87eaa4a8c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a6cbab613b769352b64e877b7c5b87eaa4a8c1f"}]},"branch":"refs/heads/master"},"d1e85a4c05152d1ceecea39e7ff1efe6b9f7db13":{"kind":"REWORK","_number":10,"created":"2019-07-25 14:03:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/10"}}},"commit":{"parents":[{"commit":"9bb432c30d8a4be7ce82f5c6d79f85f9c4355a3a","subject":"Merge \"MultiStore: Avoid potential KeyError in functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9bb432c30d8a4be7ce82f5c6d79f85f9c4355a3a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-25 13:57:48.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1e85a4c05152d1ceecea39e7ff1efe6b9f7db13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1e85a4c05152d1ceecea39e7ff1efe6b9f7db13"}]},"branch":"refs/heads/master"},"5cab654f6e5c110e8917e819e53d8f33e5d6fc34":{"kind":"REWORK","_number":11,"created":"2019-07-30 06:17:59.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/11"}}},"commit":{"parents":[{"commit":"f06b6fa739a775f84812e9fe04d319972afb6a17","subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f06b6fa739a775f84812e9fe04d319972afb6a17"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-30 05:04:41.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5cab654f6e5c110e8917e819e53d8f33e5d6fc34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5cab654f6e5c110e8917e819e53d8f33e5d6fc34"}]},"branch":"refs/heads/master"},"397f8b5bb93fcbefe5071fc9ebd2897db8d515a7":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-07-30 17:00:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/12"}}},"commit":{"parents":[{"commit":"1477192b41751bd2a752a2c0714f6789b69d5cc0","subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1477192b41751bd2a752a2c0714f6789b69d5cc0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-30 15:44:57.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/397f8b5bb93fcbefe5071fc9ebd2897db8d515a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/397f8b5bb93fcbefe5071fc9ebd2897db8d515a7"}]},"branch":"refs/heads/master"},"ffb277d97751d334ae4ae5665854495239200861":{"kind":"REWORK","_number":13,"created":"2019-07-30 17:23:12.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/13"}}},"commit":{"parents":[{"commit":"1477192b41751bd2a752a2c0714f6789b69d5cc0","subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1477192b41751bd2a752a2c0714f6789b69d5cc0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-07-30 17:20:53.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ffb277d97751d334ae4ae5665854495239200861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ffb277d97751d334ae4ae5665854495239200861"}]},"branch":"refs/heads/master"},"859b6541ee0a49e569074b18c708985fdc930b5b":{"kind":"REWORK","_number":14,"created":"2019-08-01 18:00:14.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/14"}}},"commit":{"parents":[{"commit":"77f03bbf1b9626dd6d9d21e4821b3d530db96af8","subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77f03bbf1b9626dd6d9d21e4821b3d530db96af8"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-08-01 17:54:07.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/859b6541ee0a49e569074b18c708985fdc930b5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/859b6541ee0a49e569074b18c708985fdc930b5b"}]},"branch":"refs/heads/master"},"9cf96d44dc06a8248c4fd12dc57db73307721235":{"kind":"REWORK","_number":15,"created":"2019-08-06 05:42:04.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/15"}}},"commit":{"parents":[{"commit":"77f03bbf1b9626dd6d9d21e4821b3d530db96af8","subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77f03bbf1b9626dd6d9d21e4821b3d530db96af8"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-08-06 05:27:33.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9cf96d44dc06a8248c4fd12dc57db73307721235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9cf96d44dc06a8248c4fd12dc57db73307721235"}]},"branch":"refs/heads/master"},"0e5984edc707a980028e1aa9b0d48d57ed3908d5":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-08-08 04:54:53.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/660645/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/660645/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/660645/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/660645/16"}}},"commit":{"parents":[{"commit":"0e55ad71a808b9c6f151ebdd6eb0d8f5f52217ac","subject":"Change location metadata key \u0027backend\u0027 to \u0027store\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e55ad71a808b9c6f151ebdd6eb0d8f5f52217ac"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-13 05:16:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-08-08 04:53:36.000000000","tz":0},"subject":"Lazy update stores information","message":"Lazy update stores information\n\nAdded new decorator around get image call which will add\nor update store inforamtion to location metadata if it\nis not available.\n\nImplements: blueprint lazy-update-stores\nDepends-On: https://review.opendev.org/#/c/660340/\nChange-Id: I51b6afad56cfa5ccf6c6584bb33e83ce34b4387d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e5984edc707a980028e1aa9b0d48d57ed3908d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e5984edc707a980028e1aa9b0d48d57ed3908d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
