)]}'
{"id":"openstack%2Fglance~661304","triplet_id":"openstack%2Fglance~master~I540e2cfb67fae91822936aa565213e1693c836b6","project":"openstack/glance","branch":"master","topic":"bug/1830026","hashtags":[],"change_id":"I540e2cfb67fae91822936aa565213e1693c836b6","subject":"Glance replicator: move from v1 to v2","status":"ABANDONED","created":"2019-05-24 14:37:05.000000000","updated":"2022-07-14 14:53:02.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4a27dfe6d9c2f038babb81f4c14d34ab493eac36","_number":661304,"virtual_id_number":661304,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-04 07:00:16.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-03 23:50:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":-2,"date":"2019-12-03 22:28:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 14:59:41.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-06-06 16:06:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-12-03 23:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 07:00:16.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"601b8214101eb4e60a2d45e39607ba52ac554b78","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2019-05-24 14:37:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"739e2dee78f60983685a0776be5966bd6b3da345","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-05-24 14:59:41.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)\n\nFixing the replicator is OK with me, problem noted inline.","accounts_in_message":[],"_revision_number":1},{"id":"08b1e34bddfa86b00d12844eabc87a0b7525c185","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2019-05-24 15:09:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1916d1df487bb99bba06dc2db95b805f067faa03","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-24 15:17:58.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/04/661304/1/check-not-vote/ext-glance-zuul/8735076 : FAILURE in 39m 55s","accounts_in_message":[],"_revision_number":1},{"id":"22948a56963c015a381d03e91e254d6b1ecdef23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 18:28:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/04/661304/1/check/tempest-full/2fc8345/ : SUCCESS in 1h 37m 03s\n- neutron-grenade http://logs.openstack.org/04/661304/1/check/neutron-grenade/96ba04c/ : SUCCESS in 1h 14m 42s\n- grenade-py3 http://logs.openstack.org/04/661304/1/check/grenade-py3/7337835/ : SUCCESS in 1h 17m 48s\n- tempest-full-py3 http://logs.openstack.org/04/661304/1/check/tempest-full-py3/3b78a51/ : SUCCESS in 1h 41m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/661304/1/check/openstack-tox-lower-constraints/d0cf09d/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/04/661304/1/check/openstack-tox-pep8/f7faaad/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/04/661304/1/check/openstack-tox-py27/9f08158/ : SUCCESS in 3m 45s\n- openstack-tox-py35 http://logs.openstack.org/04/661304/1/check/openstack-tox-py35/3b4bedb/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/04/661304/1/check/openstack-tox-py36/57295e0/ : SUCCESS in 3m 43s\n- openstack-tox-py37 http://logs.openstack.org/04/661304/1/check/openstack-tox-py37/9b80eaa/ : SUCCESS in 4m 21s\n- openstack-tox-docs http://logs.openstack.org/04/661304/1/check/openstack-tox-docs/d365dbb/html/ : SUCCESS in 5m 27s\n- openstack-tox-functional http://logs.openstack.org/04/661304/1/check/openstack-tox-functional/c94df99/ : SUCCESS in 8m 06s\n- openstack-tox-functional-py35 http://logs.openstack.org/04/661304/1/check/openstack-tox-functional-py35/dee3c00/ : SUCCESS in 9m 15s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/04/661304/1/check/glance-eventlet-ssl-handshake-broken-py35/a79e2c8/ : FAILURE in 2m 37s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/04/661304/1/check/glance-code-constants-check/d1f15d5/ : SUCCESS in 4m 13s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/04/661304/1/check/devstack-plugin-ceph-tempest/6d9b298/ : SUCCESS in 1h 38m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2358e9744ee0a2558dcc60342a0ce3ff1c7a4c80","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2019-05-24 22:22:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9bcedf9c1840787f6564823cf0611d42742eeb51","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-24 22:59:30.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/04/661304/2/check-not-vote/ext-glance-zuul/812e76b : FAILURE in 36m 36s","accounts_in_message":[],"_revision_number":2},{"id":"a894397fed75967082fdd89820a52b649ab8a5a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 00:10:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/04/661304/2/check/tempest-full/ecdfd38/ : SUCCESS in 1h 47m 03s\n- neutron-grenade http://logs.openstack.org/04/661304/2/check/neutron-grenade/a8d6858/ : SUCCESS in 55m 31s\n- grenade-py3 http://logs.openstack.org/04/661304/2/check/grenade-py3/240e8fe/ : SUCCESS in 58m 26s\n- tempest-full-py3 http://logs.openstack.org/04/661304/2/check/tempest-full-py3/d5fb513/ : SUCCESS in 1h 36m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/661304/2/check/openstack-tox-lower-constraints/6cc26c0/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/04/661304/2/check/openstack-tox-pep8/865dca5/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/04/661304/2/check/openstack-tox-py27/c34d08a/ : SUCCESS in 6m 45s\n- openstack-tox-py35 http://logs.openstack.org/04/661304/2/check/openstack-tox-py35/3a51e37/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/04/661304/2/check/openstack-tox-py36/c9a6387/ : SUCCESS in 5m 21s\n- openstack-tox-py37 http://logs.openstack.org/04/661304/2/check/openstack-tox-py37/e0290fa/ : SUCCESS in 5m 36s\n- openstack-tox-docs http://logs.openstack.org/04/661304/2/check/openstack-tox-docs/8e439a7/html/ : SUCCESS in 4m 59s\n- openstack-tox-functional http://logs.openstack.org/04/661304/2/check/openstack-tox-functional/22a8dc2/ : SUCCESS in 9m 33s\n- openstack-tox-functional-py35 http://logs.openstack.org/04/661304/2/check/openstack-tox-functional-py35/b5633fc/ : SUCCESS in 12m 20s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/04/661304/2/check/glance-eventlet-ssl-handshake-broken-py35/d91724b/ : FAILURE in 3m 03s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/04/661304/2/check/glance-code-constants-check/a2dca1f/ : SUCCESS in 4m 46s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/04/661304/2/check/devstack-plugin-ceph-tempest/f04a324/ : SUCCESS in 1h 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c29929931b5d69e568d29ae954067b48ae15827a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-06 16:06:01.000000000","message":"Patch Set 2:\n\n(1 comment)\n\npatch looks ok to me, I need to test it functionally though. Just added one suggestion inside.","accounts_in_message":[],"_revision_number":2},{"id":"d5e92bb5a2f920585be04ed2f529e6d877332f6f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2019-12-03 22:28:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"671ee54cc7e983a8e6b690b25f2b06b2dac6033f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 23:50:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8aa34fad57ba440ebe2ef25fa8d0b0e8 : SUCCESS in 1h 20m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f4b28c523fd644b290de0301b422c02a : SUCCESS in 54m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/81c05e49f72b4142a10b3bf0172b777e : SUCCESS in 54m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35e4ca2df0174d259df9d414d82dcc23 : SUCCESS in 1h 15m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35ef548b30794a0caa52361a41b79ca4 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f092aa06ec1548c184a614dcbfa4ed7c : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a3fdd9c0d1c4b70bc9e63cb2940de75 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7936d9847f434e6a93af0cc54e07ec1b : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af53eb38d606420792c3db44415157d2 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f8822921e74ae9bd41666573eca765 : SUCCESS in 5m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2da07d7f9ee94f6aac7a81b8753ae82c : SUCCESS in 9m 07s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9c665143985e401bb650de65017e17e8 : SUCCESS in 9m 46s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/b30df357b7264743ae755f37437aa5be : FAILURE in 5m 48s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/94bef6022d754c529abde01549f1a1f3 : SUCCESS in 5m 15s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fbbb8d2200a44ae3a297db676295e898 : SUCCESS in 1h 10m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"89b444c5f3ed6eccce9d23b6242b6df1c0ad97d7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-04 07:00:16.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/04/661304/3/check-not-vote/ext-glance-zuul/cf94bcd : FAILURE in 30m 12s","accounts_in_message":[],"_revision_number":3},{"id":"4a27dfe6d9c2f038babb81f4c14d34ab493eac36","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-14 14:53:02.000000000","message":"Abandoned\n\nGlance replicator is broken since long and in the past (approx 4 years ago) it is decided to remove this tool.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8a09f9279e8e5f7f5624d92cc40176a14a0b4811","revisions":{"e87eb4110e75fff441469a7b00ec20545e6012dd":{"kind":"REWORK","_number":1,"created":"2019-05-24 14:37:05.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/04/661304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/04/661304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/661304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/661304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/661304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/04/661304/1"}}},"commit":{"parents":[{"commit":"61298a47e4abd903fe38957ffbbbcf036e79bb3e","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61298a47e4abd903fe38957ffbbbcf036e79bb3e"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2019-05-24 14:31:52.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2019-05-24 14:31:52.000000000","tz":120},"subject":"Glance replicator: move from v1 to v2","message":"Glance replicator: move from v1 to v2\n\nThe first version of the API has been deprecated a while ago, but we\nnever moved glance-replicator to v2.\n\nChange-Id: I540e2cfb67fae91822936aa565213e1693c836b6\nCloses-Bug: #1830026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e87eb4110e75fff441469a7b00ec20545e6012dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e87eb4110e75fff441469a7b00ec20545e6012dd"}]},"branch":"refs/heads/master"},"14cafd76508c837d85ab4454028675cf9208a249":{"kind":"REWORK","_number":2,"created":"2019-05-24 22:22:33.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/04/661304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/04/661304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/661304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/661304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/661304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/04/661304/2"}}},"commit":{"parents":[{"commit":"61298a47e4abd903fe38957ffbbbcf036e79bb3e","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61298a47e4abd903fe38957ffbbbcf036e79bb3e"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2019-05-24 14:31:52.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2019-05-24 15:25:16.000000000","tz":120},"subject":"Glance replicator: move from v1 to v2","message":"Glance replicator: move from v1 to v2\n\nThe first version of the API has been deprecated a while ago, but we\nnever moved glance-replicator to v2.\n\nChange-Id: I540e2cfb67fae91822936aa565213e1693c836b6\nCloses-Bug: #1830026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14cafd76508c837d85ab4454028675cf9208a249"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14cafd76508c837d85ab4454028675cf9208a249"}]},"branch":"refs/heads/master"},"8a09f9279e8e5f7f5624d92cc40176a14a0b4811":{"kind":"REWORK","_number":3,"created":"2019-12-03 22:28:00.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/04/661304/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/04/661304/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/661304/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/661304/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/661304/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/04/661304/3"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2019-05-24 14:31:52.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2019-12-03 22:27:28.000000000","tz":60},"subject":"Glance replicator: move from v1 to v2","message":"Glance replicator: move from v1 to v2\n\nThe first version of the API has been deprecated a while ago, but we\nnever moved glance-replicator to v2. We also remove two related options\nthat have been deprecated since\n\nChange-Id: I540e2cfb67fae91822936aa565213e1693c836b6\nCloses-Bug: #1830026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a09f9279e8e5f7f5624d92cc40176a14a0b4811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a09f9279e8e5f7f5624d92cc40176a14a0b4811"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
