)]}'
{"id":"openstack%2Fglance~661687","triplet_id":"openstack%2Fglance~master~Ibb821fabd479af05bd4daf3d68cac16fc5e1b9c4","project":"openstack/glance","branch":"master","topic":"Bug#sphinx","hashtags":[],"change_id":"Ibb821fabd479af05bd4daf3d68cac16fc5e1b9c4","subject":"Update sphinx dependency","status":"MERGED","created":"2019-05-28 07:34:51.000000000","updated":"2019-05-29 09:27:35.000000000","submitted":"2019-05-29 09:27:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661687-1559122055407-452223fe","meta_rev_id":"21e1177d87fb1120b1088b1960a2680da06c499a","_number":661687,"virtual_id_number":661687,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":2,"date":"2019-05-29 09:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-28 21:29:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-05-29 09:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-05-29 09:07:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-28 21:29:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-05-29 01:10:10.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"date":"2019-05-29 01:10:10.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2019-05-29 01:10:10.000000000","_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-28 21:29:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-29 01:10:10.000000000","updated_by":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"reviewer":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"state":"REVIEWER"},{"updated":"2019-05-29 01:10:10.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-05-29 01:10:10.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-05-29 09:07:18.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-05-29 09:27:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25fefbea20d4fd8152650ed40b97bdc3a3df810b","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-05-28 07:34:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"072cfefde426617189c8ecc159b8d10d4416fc8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 07:47:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/87/661687/1/check/build-openstack-api-ref/4364728/html/ : SUCCESS in 4m 50s\n- requirements-check http://logs.openstack.org/87/661687/1/check/requirements-check/3520cc5/ : FAILURE in 2m 58s\n- openstack-tox-pep8 http://logs.openstack.org/87/661687/1/check/openstack-tox-pep8/c45fc80/ : SUCCESS in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/87/661687/1/check/openstack-tox-docs/48708da/html/ : SUCCESS in 4m 22s\n- build-openstack-releasenotes http://logs.openstack.org/87/661687/1/check/build-openstack-releasenotes/c8ff1e3/html/ : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":1},{"id":"25b9d2df3c271a83962e3bdb2c2523219782bdc7","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-05-28 07:58:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64b0f5d185a09e8a02d938f4128a2d86cc04389c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 08:15:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/87/661687/2/check/build-openstack-api-ref/7aaadbc/html/ : SUCCESS in 4m 56s\n- requirements-check http://logs.openstack.org/87/661687/2/check/requirements-check/4b1e4ab/ : SUCCESS in 2m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/661687/2/check/openstack-tox-lower-constraints/6c9b009/ : SUCCESS in 5m 29s\n- openstack-tox-pep8 http://logs.openstack.org/87/661687/2/check/openstack-tox-pep8/0210faf/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/87/661687/2/check/openstack-tox-py27/4fb0ef6/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/87/661687/2/check/openstack-tox-py35/8c470d0/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/87/661687/2/check/openstack-tox-py36/f6775b8/ : SUCCESS in 5m 49s\n- openstack-tox-py37 http://logs.openstack.org/87/661687/2/check/openstack-tox-py37/e47e057/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/87/661687/2/check/openstack-tox-docs/193fffe/html/ : SUCCESS in 4m 55s\n- build-openstack-releasenotes http://logs.openstack.org/87/661687/2/check/build-openstack-releasenotes/8142896/html/ : SUCCESS in 7m 45s\n- openstack-tox-functional http://logs.openstack.org/87/661687/2/check/openstack-tox-functional/34295b8/ : SUCCESS in 8m 26s\n- openstack-tox-functional-py35 http://logs.openstack.org/87/661687/2/check/openstack-tox-functional-py35/cceeb10/ : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":2},{"id":"a9106ee577a552a718e6460ff5bdc02b9ee49662","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-28 21:29:36.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"aefb9b365e3dd6794d004227387212d8fa2c6c7e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-05-29 09:07:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"85f6ef2511fff3005f11768adcf0bac492a4980a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 09:07:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e91fcd641014ecba1d562254bbd9113080c1af4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 09:27:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"41ccb71560ccd1aa06d9e324106e9cbb3b74e155","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 09:27:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/87/661687/2/gate/build-openstack-api-ref/132c383/html/ : SUCCESS in 3m 49s\n- requirements-check http://logs.openstack.org/87/661687/2/gate/requirements-check/49fa2b5/ : SUCCESS in 3m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/661687/2/gate/openstack-tox-lower-constraints/a4068b3/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/87/661687/2/gate/openstack-tox-pep8/cd87e99/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/87/661687/2/gate/openstack-tox-py27/6e7e930/ : SUCCESS in 4m 05s\n- openstack-tox-py35 http://logs.openstack.org/87/661687/2/gate/openstack-tox-py35/3fa2a35/ : SUCCESS in 4m 17s\n- openstack-tox-py36 http://logs.openstack.org/87/661687/2/gate/openstack-tox-py36/9c6a8c4/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/87/661687/2/gate/openstack-tox-py37/376b21e/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/87/661687/2/gate/openstack-tox-docs/1817c4b/html/ : SUCCESS in 4m 15s\n- build-openstack-releasenotes http://logs.openstack.org/87/661687/2/gate/build-openstack-releasenotes/40663cc/html/ : SUCCESS in 8m 07s\n- openstack-tox-functional http://logs.openstack.org/87/661687/2/gate/openstack-tox-functional/410316e/ : SUCCESS in 8m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/87/661687/2/gate/openstack-tox-functional-py35/fa7eebd/ : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7dea25d8c4cb1cd96777fab43a07138b3b621e2d","revisions":{"b3d8e4ac3d355b2ac8276f95c63eaaf65e33fe1e":{"kind":"REWORK","_number":1,"created":"2019-05-28 07:34:51.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/87/661687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/661687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/661687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/661687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/661687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/661687/1"}}},"commit":{"parents":[{"commit":"61298a47e4abd903fe38957ffbbbcf036e79bb3e","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61298a47e4abd903fe38957ffbbbcf036e79bb3e"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-28 07:02:32.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-28 07:16:17.000000000","tz":480},"subject":"Update sphinx dependency","message":"Update sphinx dependency\n\nSphinx 2.0 no longer works on python 2.7,\nso we need to start capping it there as well\ndepend on https://review.opendev.org/#/c/657890/\n\nChange-Id: Ibb821fabd479af05bd4daf3d68cac16fc5e1b9c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3d8e4ac3d355b2ac8276f95c63eaaf65e33fe1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3d8e4ac3d355b2ac8276f95c63eaaf65e33fe1e"}]},"branch":"refs/heads/master"},"7dea25d8c4cb1cd96777fab43a07138b3b621e2d":{"kind":"REWORK","_number":2,"created":"2019-05-28 07:58:19.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/87/661687/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/661687/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/661687/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/661687/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/661687/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/661687/2"}}},"commit":{"parents":[{"commit":"61298a47e4abd903fe38957ffbbbcf036e79bb3e","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61298a47e4abd903fe38957ffbbbcf036e79bb3e"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-28 07:02:32.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-28 07:39:33.000000000","tz":480},"subject":"Update sphinx dependency","message":"Update sphinx dependency\n\nSphinx 2.0 no longer works on python 2.7,\nso we need to start capping it there as well\ndepend on https://review.opendev.org/#/c/657890/\n\nChange-Id: Ibb821fabd479af05bd4daf3d68cac16fc5e1b9c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7dea25d8c4cb1cd96777fab43a07138b3b621e2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7dea25d8c4cb1cd96777fab43a07138b3b621e2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
