)]}'
{"id":"openstack%2Fglance~663224","triplet_id":"openstack%2Fglance~master~Ica74f2982a75e3faf7a5244390f2299381eb5b89","project":"openstack/glance","branch":"master","topic":"bug#openstackdocstheme","hashtags":[],"change_id":"Ica74f2982a75e3faf7a5244390f2299381eb5b89","subject":"Bump openstackdocstheme to 1.20.0","status":"MERGED","created":"2019-06-05 02:12:25.000000000","updated":"2019-07-17 17:11:28.000000000","submitted":"2019-07-17 17:11:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663224-1563383488475-e47bf6dd","meta_rev_id":"7c322be11f2514f6dbcda940584e832a82445c47","_number":663224,"virtual_id_number":663224,"owner":{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":2,"date":"2019-07-17 17:11:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-17 13:01:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-07-08 14:35:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-17 13:01:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 15:33:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-28 01:38:06.000000000","updated_by":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"reviewer":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"state":"REVIEWER"},{"updated":"2019-06-28 01:38:06.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-06-28 01:38:18.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2019-07-01 07:38:01.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-07-08 14:35:44.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-07-17 13:01:08.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-07-17 17:11:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3703bcf83019a19314dffb97a9cc011549f22ba","author":{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},"date":"2019-06-05 02:12:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eae69877ffeeb3c83ac661071b3759c64dd4f966","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:34:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/24/663224/1/check/build-openstack-api-ref/de0be4f/html/ : SUCCESS in 3m 44s\n- requirements-check http://logs.openstack.org/24/663224/1/check/requirements-check/f8770be/ : SUCCESS in 2m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/24/663224/1/check/openstack-tox-lower-constraints/61248ff/ : SUCCESS in 3m 53s\n- openstack-tox-pep8 http://logs.openstack.org/24/663224/1/check/openstack-tox-pep8/6fed900/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/24/663224/1/check/openstack-tox-py27/708c29b/ : SUCCESS in 4m 55s\n- openstack-tox-py35 http://logs.openstack.org/24/663224/1/check/openstack-tox-py35/cabcbc3/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/24/663224/1/check/openstack-tox-py36/d1a9c37/ : SUCCESS in 4m 31s\n- openstack-tox-py37 http://logs.openstack.org/24/663224/1/check/openstack-tox-py37/6f8c59f/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/24/663224/1/check/openstack-tox-docs/2b8360e/html/ : SUCCESS in 4m 38s\n- build-openstack-releasenotes http://logs.openstack.org/24/663224/1/check/build-openstack-releasenotes/994547b/html/ : SUCCESS in 8m 48s\n- openstack-tox-functional http://logs.openstack.org/24/663224/1/check/openstack-tox-functional/30aa456/ : SUCCESS in 8m 32s\n- openstack-tox-functional-py35 http://logs.openstack.org/24/663224/1/check/openstack-tox-functional-py35/2e2aaac/ : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6ec07176f472ebb08894f13aa3032aa60625e6af","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-05 15:33:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"468feb8ea5d40da2fdca6398f12920b3b0eebbd1","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-28 01:38:18.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"1db42b7d9669f5bbddf2c7f6406103ccb6fbd3ef","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-07-01 07:38:01.000000000","message":"Patch Set 1:\n\nSuch a big bump is not really needed; see also the discussion that happened here: https://review.opendev.org/#/c/668081/","accounts_in_message":[],"_revision_number":1},{"id":"5d4fabe344ba92bf213d6d4b00287cb0adf9f3c0","author":{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},"date":"2019-07-02 01:53:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d15c0f5b7360fed3a5944602314de1cbd648aa33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 02:07:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/24/663224/2/check/build-openstack-api-ref/4c77d67/html/ : SUCCESS in 5m 03s\n- requirements-check http://logs.openstack.org/24/663224/2/check/requirements-check/d0741a8/ : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/24/663224/2/check/openstack-tox-lower-constraints/e4e28a6/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/24/663224/2/check/openstack-tox-pep8/f8bbb4d/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/24/663224/2/check/openstack-tox-py27/b031ea1/ : SUCCESS in 4m 14s\n- openstack-tox-py35 http://logs.openstack.org/24/663224/2/check/openstack-tox-py35/16ed7be/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/24/663224/2/check/openstack-tox-py36/4c96d7f/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/24/663224/2/check/openstack-tox-py37/f1075ea/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/24/663224/2/check/openstack-tox-docs/db94853/html/ : SUCCESS in 5m 51s\n- build-openstack-releasenotes http://logs.openstack.org/24/663224/2/check/build-openstack-releasenotes/b6dd60c/html/ : SUCCESS in 8m 42s\n- openstack-tox-functional http://logs.openstack.org/24/663224/2/check/openstack-tox-functional/1b73fa8/ : SUCCESS in 8m 08s\n- openstack-tox-functional-py35 http://logs.openstack.org/24/663224/2/check/openstack-tox-functional-py35/62d6949/ : SUCCESS in 9m 53s","accounts_in_message":[],"_revision_number":2},{"id":"3abf82a2f23abd2537b5933f499733060772d047","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-07-08 14:35:44.000000000","message":"Patch Set 2: Code-Review+2\n\nI really wouldn\u0027t have minded to bump it right away to the current. Makes no difference apart from us needing to bump it again sooner. Either way fine by me.","accounts_in_message":[],"_revision_number":2},{"id":"27be1e73bfd813bb508b26e47e4f7a089e4f3b0b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-17 13:01:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nGenerated HTML docs look OK.  We\u0027ll probably have to update again (I imagine they\u0027ll find some things that need to be addressed for the community generate-PDF goal), but I\u0027m OK with this bump for now.","accounts_in_message":[],"_revision_number":2},{"id":"0f079b50154cf6ab4eea6e082ac6e1e0d6892926","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 13:01:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"08b13d3012a88e1a603fec58188194c5fbb841da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 17:11:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/24/663224/2/gate/build-openstack-api-ref/3acd169/html/ : SUCCESS in 5m 11s\n- requirements-check http://logs.openstack.org/24/663224/2/gate/requirements-check/48da8e2/ : SUCCESS in 2m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/24/663224/2/gate/openstack-tox-lower-constraints/332fc0e/ : SUCCESS in 4m 38s\n- openstack-tox-pep8 http://logs.openstack.org/24/663224/2/gate/openstack-tox-pep8/c3385e0/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/24/663224/2/gate/openstack-tox-py27/2ebd4c4/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/24/663224/2/gate/openstack-tox-py35/31a3a47/ : SUCCESS in 5m 04s\n- openstack-tox-py36 http://logs.openstack.org/24/663224/2/gate/openstack-tox-py36/bd2b636/ : SUCCESS in 4m 36s\n- openstack-tox-py37 http://logs.openstack.org/24/663224/2/gate/openstack-tox-py37/42cca6e/ : SUCCESS in 6m 46s\n- openstack-tox-docs http://logs.openstack.org/24/663224/2/gate/openstack-tox-docs/9a1852b/html/ : SUCCESS in 5m 52s\n- build-openstack-releasenotes http://logs.openstack.org/24/663224/2/gate/build-openstack-releasenotes/3ff6824/html/ : SUCCESS in 8m 58s\n- openstack-tox-functional http://logs.openstack.org/24/663224/2/gate/openstack-tox-functional/c1e63a1/ : SUCCESS in 9m 40s\n- openstack-tox-functional-py35 http://logs.openstack.org/24/663224/2/gate/openstack-tox-functional-py35/a6bf2a0/ : SUCCESS in 10m 15s","accounts_in_message":[],"_revision_number":2},{"id":"237338d48e1f931ad272c6f1ee03d6038ffef5f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 17:11:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d4d045f31a9b9b402a3614526a25f1e5505b2bc2","revisions":{"e1f284155ba989ac912d5c1c9b95afc1e9e776ea":{"kind":"REWORK","_number":1,"created":"2019-06-05 02:12:25.000000000","uploader":{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},"ref":"refs/changes/24/663224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/663224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/663224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/663224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/663224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/663224/1"}}},"commit":{"parents":[{"commit":"d99cea0cc76e981debece97b3d9fd0e5ef6434d2","subject":"Merge \"Mark http store read-only in discovery call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d99cea0cc76e981debece97b3d9fd0e5ef6434d2"}]}],"author":{"name":"翟小君","email":"zhaixiaojun@gohighsec.com","date":"2019-06-05 02:10:05.000000000","tz":480},"committer":{"name":"翟小君","email":"zhaixiaojun@gohighsec.com","date":"2019-06-05 02:11:23.000000000","tz":480},"subject":"Bump openstackdocstheme to 1.30.0","message":"Bump openstackdocstheme to 1.30.0\n\n...to pick up many improvements, including the return of table borders.\n\nChange-Id: Ica74f2982a75e3faf7a5244390f2299381eb5b89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1f284155ba989ac912d5c1c9b95afc1e9e776ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1f284155ba989ac912d5c1c9b95afc1e9e776ea"}]},"branch":"refs/heads/master"},"d4d045f31a9b9b402a3614526a25f1e5505b2bc2":{"kind":"REWORK","_number":2,"created":"2019-07-02 01:53:17.000000000","uploader":{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},"ref":"refs/changes/24/663224/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/663224/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/663224/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/663224/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/663224/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/663224/2"}}},"commit":{"parents":[{"commit":"d99cea0cc76e981debece97b3d9fd0e5ef6434d2","subject":"Merge \"Mark http store read-only in discovery call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d99cea0cc76e981debece97b3d9fd0e5ef6434d2"}]}],"author":{"name":"翟小君","email":"zhaixiaojun@gohighsec.com","date":"2019-06-05 02:10:05.000000000","tz":480},"committer":{"name":"翟小君","email":"zhaixiaojun@gohighsec.com","date":"2019-07-02 01:52:19.000000000","tz":480},"subject":"Bump openstackdocstheme to 1.20.0","message":"Bump openstackdocstheme to 1.20.0\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Ica74f2982a75e3faf7a5244390f2299381eb5b89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4d045f31a9b9b402a3614526a25f1e5505b2bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4d045f31a9b9b402a3614526a25f1e5505b2bc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
