)]}'
{"id":"openstack%2Fglance~663840","triplet_id":"openstack%2Fglance~master~I18add352238e93c5d4ade8b3e8d967b2f639cdad","project":"openstack/glance","branch":"master","topic":"bug/1831963","hashtags":[],"change_id":"I18add352238e93c5d4ade8b3e8d967b2f639cdad","subject":"Unit tests fails with new glance_store version 0.29.0","status":"MERGED","created":"2019-06-07 06:25:48.000000000","updated":"2019-06-07 13:36:07.000000000","submitted":"2019-06-07 13:36:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663840-1559914567961-bc982933","meta_rev_id":"26eef40fccd872883c119d79e4f0f7ebb65b7b5d","_number":663840,"virtual_id_number":663840,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-07 07:51:57.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2019-06-07 13:36:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-07 11:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-07 12:13:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"date":"2019-06-07 12:58:37.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-06-07 12:13:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-06-07 07:03:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 07:51:57.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 11:22:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-07 12:13:16.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-06-07 12:58:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-06-07 13:36:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"33017fa08ce8dfb63239b1dcdb5455d5c5c837b1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 06:25:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab6b66edcf1e7a19a64fc7ad6eaf7f5449bdfb1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 06:36:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/40/663840/1/check/openstack-tox-lower-constraints/fec8a9e/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/40/663840/1/check/openstack-tox-pep8/8fe7bda/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/40/663840/1/check/openstack-tox-py27/59730f2/ : SUCCESS in 5m 10s\n- openstack-tox-py35 http://logs.openstack.org/40/663840/1/check/openstack-tox-py35/3c3f2df/ : SUCCESS in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/40/663840/1/check/openstack-tox-py36/07d437c/ : SUCCESS in 4m 02s\n- openstack-tox-py37 http://logs.openstack.org/40/663840/1/check/openstack-tox-py37/3a487fd/ : SUCCESS in 6m 09s\n- openstack-tox-docs http://logs.openstack.org/40/663840/1/check/openstack-tox-docs/7324959/html/ : SUCCESS in 6m 01s\n- openstack-tox-functional http://logs.openstack.org/40/663840/1/check/openstack-tox-functional/c84ee2d/ : SUCCESS in 9m 03s\n- openstack-tox-functional-py35 http://logs.openstack.org/40/663840/1/check/openstack-tox-functional-py35/7490dc2/ : FAILURE in 10m 30s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/40/663840/1/check/glance-eventlet-ssl-handshake-broken-py35/0a14149/ : FAILURE in 2m 54s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/40/663840/1/check/glance-code-constants-check/017e15e/ : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b8622cb8caac0b3de5adb6dae37fd98034051719","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 06:37:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"94eb168541ef9be508dc8d00e892c11a74bdd4d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 07:03:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/40/663840/1/check/openstack-tox-lower-constraints/f6050a4/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/40/663840/1/check/openstack-tox-pep8/b1dc526/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/40/663840/1/check/openstack-tox-py27/248166e/ : SUCCESS in 4m 05s\n- openstack-tox-py35 http://logs.openstack.org/40/663840/1/check/openstack-tox-py35/90e6749/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/40/663840/1/check/openstack-tox-py36/c7730d2/ : SUCCESS in 3m 55s\n- openstack-tox-py37 http://logs.openstack.org/40/663840/1/check/openstack-tox-py37/2efe104/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/40/663840/1/check/openstack-tox-docs/0d3435c/html/ : SUCCESS in 4m 43s\n- openstack-tox-functional http://logs.openstack.org/40/663840/1/check/openstack-tox-functional/8d0dcec/ : SUCCESS in 11m 18s\n- openstack-tox-functional-py35 http://logs.openstack.org/40/663840/1/check/openstack-tox-functional-py35/c1610b3/ : SUCCESS in 8m 20s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/40/663840/1/check/glance-eventlet-ssl-handshake-broken-py35/d1d87ac/ : FAILURE in 2m 32s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/40/663840/1/check/glance-code-constants-check/9902987/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"eaf98af56406cbe9335be0e88d66758c9aa23d47","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 07:03:41.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"56568bf4b01aa9cb031ef7ac35ecad6348957bed","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-07 07:51:57.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/40/663840/1/check-not-vote/ext-glance-zuul/30092d0 : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e385126e63ce61a03996a5773102b90b97c1b093","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-07 11:22:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"42b33341132c0fcaf8ce4f815a8fcef5b120a197","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-06-07 11:42:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nThis makes sense, just want to make sure we did not miss configuring those stores anywhere (see my comment inline).","accounts_in_message":[],"_revision_number":1},{"id":"fcc5952a86d3b4f96b61663cdee7846978d96339","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 11:50:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ceb7818a2d32b0dcdcd03eb1258d555b68ad35c1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-06-07 11:54:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nComment inline.  Not -2\u0027ing because I don\u0027t want to hold this up.  But I think a follow-up patch may be in order.","accounts_in_message":[],"_revision_number":1},{"id":"f38827056849f44902bd4bb424c82ac1498847df","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-06-07 12:13:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nBrian: While I totally agree that the unit tests should be mocking external dependencies, I think it\u0027s irrelevat for this specific patch (like you mentioned with follow up as well) unblocking our gate. Lets get this in, gate going and look into refactoring the unit test on follow up. Nice catch btw, I was kind of mixed up as I saw some functional test failures from earlier today and totally did not think that indeed we are dealing with unit tests here.\n\nApproving this to get the gate moving again.","accounts_in_message":[],"_revision_number":1},{"id":"fc97000a84ba61044a3a8553028d26a41b7ef8a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 12:13:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0707f7e8b1fbe117897ebe39558aa3c4a0e75ff8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-06-07 12:58:37.000000000","message":"Patch Set 1: -Code-Review\n\nWithdrawing my objection since Erno has filed https://bugs.launchpad.net/glance/+bug/1831997","accounts_in_message":[],"_revision_number":1},{"id":"a07753991a963e7a2fb3ab6c93a27db7a0f7b0ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:36:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/40/663840/1/gate/openstack-tox-lower-constraints/2c6ceba/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/40/663840/1/gate/openstack-tox-pep8/73323cb/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/40/663840/1/gate/openstack-tox-py27/906c2ea/ : SUCCESS in 4m 48s\n- openstack-tox-py35 http://logs.openstack.org/40/663840/1/gate/openstack-tox-py35/66a6559/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/40/663840/1/gate/openstack-tox-py36/776cd8e/ : SUCCESS in 4m 31s\n- openstack-tox-py37 http://logs.openstack.org/40/663840/1/gate/openstack-tox-py37/e7dac7e/ : SUCCESS in 4m 33s\n- openstack-tox-docs http://logs.openstack.org/40/663840/1/gate/openstack-tox-docs/d57e41a/html/ : SUCCESS in 5m 33s\n- openstack-tox-functional http://logs.openstack.org/40/663840/1/gate/openstack-tox-functional/959ac02/ : SUCCESS in 8m 11s\n- openstack-tox-functional-py35 http://logs.openstack.org/40/663840/1/gate/openstack-tox-functional-py35/069466a/ : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":1},{"id":"bff4804a504856482969161eddf7576c07b01846","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:36:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0182868be5a3d07755a5039d77ff53ff3d770147","revisions":{"0182868be5a3d07755a5039d77ff53ff3d770147":{"kind":"REWORK","_number":1,"created":"2019-06-07 06:25:48.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/40/663840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/663840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/663840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/663840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/663840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/663840/1"}}},"commit":{"parents":[{"commit":"be13eb05a0bc127069e743c9a6b14aa81b448b5d","subject":"Remove additional \" from config help","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be13eb05a0bc127069e743c9a6b14aa81b448b5d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-07 06:24:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-07 06:24:36.000000000","tz":0},"subject":"Unit tests fails with new glance_store version 0.29.0","message":"Unit tests fails with new glance_store version 0.29.0\n\nAround 11 unit tests are failing [1] with a recent patch [2] merged in\nglance_store. The reason behind the failure is glance unit tests are\nloading rbd store which is not supported in gate. Earlier these tests\nwere passing because they were not trying to connect to rados while\nloading the store, but after merging of recent patch [1] it tries to\nconnect to rados via configure_add method of rbd driver.\n\nTo avoid these kind of failures in near future we should drop using\nrbd store in tests as we know it is not supported in gate.\n\nNOTE:\nAlso corrected the naming conventions of store identifiers, instead\nof using file1 etc., used appropriate names.\n\n[1] http://logs.openstack.org/40/663740/1/check/cross-glance-py27/50a9a9c/\n[2] https://review.opendev.org/660340\n\nChange-Id: I18add352238e93c5d4ade8b3e8d967b2f639cdad\nCloses-Bug: #1831963\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0182868be5a3d07755a5039d77ff53ff3d770147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0182868be5a3d07755a5039d77ff53ff3d770147"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
