)]}'
{"id":"openstack%2Fglance~663878","triplet_id":"openstack%2Fglance~master~Iebebdd5898293b1900abc05d92d70fd154a12fd1","project":"openstack/glance","branch":"master","topic":"bp/barbican-secret-deletion-support","hashtags":[],"change_id":"Iebebdd5898293b1900abc05d92d70fd154a12fd1","subject":"PoC for \u0027barbican secret deletion\u0027","status":"ABANDONED","created":"2019-06-07 09:55:13.000000000","updated":"2019-06-11 04:33:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b036aa3496b1efc97a3a6d8bbcb500fa288a0616","_number":663878,"virtual_id_number":663878,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-06-07 12:11:45.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2019-06-07 12:49:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-06-07 09:56:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 12:11:45.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 12:49:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b16bc1ec29944f7e1a0ac6eace778a65ac5b6dcf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 09:55:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa860bb428965c546002f641edd846b42734c3e9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 09:56:02.000000000","message":"Patch Set 1: Workflow-1\n\nSpec is not approved yet and this is just a PoC to get the initial idea.","accounts_in_message":[],"_revision_number":1},{"id":"0186c8ec4f5d06e286ae20308f690500a1ceab91","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-07 12:11:45.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/78/663878/1/check-not-vote/ext-glance-zuul/f77cf2d : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":1},{"id":"94bfffd93ac4425184cf91687675d5db16176faa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 12:49:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/78/663878/1/check/tempest-full/1225cc8/ : SUCCESS in 1h 57m 23s\n- neutron-grenade http://logs.openstack.org/78/663878/1/check/neutron-grenade/c1a3262/ : SUCCESS in 1h 00m 41s\n- grenade-py3 http://logs.openstack.org/78/663878/1/check/grenade-py3/eea9838/ : SUCCESS in 1h 03m 14s\n- tempest-full-py3 http://logs.openstack.org/78/663878/1/check/tempest-full-py3/c8cef0e/ : SUCCESS in 1h 39m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/78/663878/1/check/openstack-tox-lower-constraints/626cb6e/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/78/663878/1/check/openstack-tox-pep8/8be7051/ : SUCCESS in 4m 50s\n- openstack-tox-py27 http://logs.openstack.org/78/663878/1/check/openstack-tox-py27/617e4bc/ : SUCCESS in 4m 15s\n- openstack-tox-py35 http://logs.openstack.org/78/663878/1/check/openstack-tox-py35/d61c4e3/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/78/663878/1/check/openstack-tox-py36/8bbf197/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/78/663878/1/check/openstack-tox-py37/11bc0a2/ : SUCCESS in 5m 49s\n- openstack-tox-docs http://logs.openstack.org/78/663878/1/check/openstack-tox-docs/cb579ee/html/ : SUCCESS in 5m 21s\n- openstack-tox-functional http://logs.openstack.org/78/663878/1/check/openstack-tox-functional/9aec1b4/ : SUCCESS in 10m 45s\n- openstack-tox-functional-py35 http://logs.openstack.org/78/663878/1/check/openstack-tox-functional-py35/2f7bb80/ : SUCCESS in 8m 51s\n- glance-eventlet-ssl-handshake-broken-py35 http://logs.openstack.org/78/663878/1/check/glance-eventlet-ssl-handshake-broken-py35/1a47ffd/ : FAILURE in 3m 13s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/78/663878/1/check/glance-code-constants-check/b954402/ : SUCCESS in 4m 30s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/78/663878/1/check/devstack-plugin-ceph-tempest/e128d77/ : FAILURE in 1h 36m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f39175fe7a4ba8bf543d8ca6e8991c771c254066","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-11 04:33:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"279ca2132df2fb4f96f6a6921f4124216e4cdab6","revisions":{"279ca2132df2fb4f96f6a6921f4124216e4cdab6":{"kind":"REWORK","_number":1,"created":"2019-06-07 09:55:13.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/78/663878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/663878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/663878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/663878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/663878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/663878/1"}}},"commit":{"parents":[{"commit":"be13eb05a0bc127069e743c9a6b14aa81b448b5d","subject":"Remove additional \" from config help","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be13eb05a0bc127069e743c9a6b14aa81b448b5d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-07 09:54:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-07 09:54:14.000000000","tz":0},"subject":"PoC for \u0027barbican secret deletion\u0027","message":"PoC for \u0027barbican secret deletion\u0027\n\nChange-Id: Iebebdd5898293b1900abc05d92d70fd154a12fd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/279ca2132df2fb4f96f6a6921f4124216e4cdab6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/279ca2132df2fb4f96f6a6921f4124216e4cdab6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
