)]}'
{"id":"openstack%2Fglance~664475","triplet_id":"openstack%2Fglance~master~I2dee048b71e3758208fae438423cceae661c8f00","project":"openstack/glance","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I2dee048b71e3758208fae438423cceae661c8f00","subject":"Dropping the py35 testing","status":"MERGED","created":"2019-06-11 01:28:37.000000000","updated":"2019-09-05 17:54:54.000000000","submitted":"2019-09-05 17:49:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664475-1567705788577-cdc6229e","meta_rev_id":"48464d1b4615f389349b3f4f01f096323b3c1a37","_number":664475,"virtual_id_number":664475,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-07-15 01:39:26.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":2,"date":"2019-09-05 17:49:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-21 13:18:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2019-09-05 13:45:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-09-05 15:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-09-05 15:05:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-09-05 17:54:54.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 06:36:30.000000000","updated_by":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"reviewer":{"_account_id":2537,"name":"Nikhil Komawar","email":"nik.komawar@gmail.com","username":"nikhil-komawar"},"state":"REVIEWER"},{"updated":"2019-06-13 06:36:30.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-07-12 20:53:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-15 01:39:26.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 13:18:18.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-09-05 13:45:07.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-09-05 15:05:04.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-09-05 17:49:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3917b5925ef841755613be9304fb72ee29a61fa6","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-11 01:28:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54587e8ad3fa1aa7e779f6c684b06e6942cf4492","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 05:30:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/75/664475/1/check/build-openstack-api-ref/989050e/html/ : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/664475/1/check/openstack-tox-lower-constraints/f9e9b23/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/75/664475/1/check/openstack-tox-pep8/28f7786/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/75/664475/1/check/openstack-tox-py27/d1f92ba/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/75/664475/1/check/openstack-tox-py36/b768f99/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/75/664475/1/check/openstack-tox-py37/38383e8/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/75/664475/1/check/openstack-tox-docs/5ab1ee1/html/ : SUCCESS in 5m 12s\n- build-openstack-releasenotes http://logs.openstack.org/75/664475/1/check/build-openstack-releasenotes/9dd049e/html/ : SUCCESS in 8m 43s\n- openstack-tox-functional http://logs.openstack.org/75/664475/1/check/openstack-tox-functional/45ce423/ : SUCCESS in 8m 45s\n- openstack-tox-functional-py35 http://logs.openstack.org/75/664475/1/check/openstack-tox-functional-py35/82d56f6/ : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d877c6f078bfe95d05863f5b12d237d389a2aa46","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-06-11 11:25:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nI think this change may be a bit out of date, see my comments inline.\n\nWe can discuss the question of what the tox defaults should be at the glance weekly meeting.  I\u0027ll put it on the agenda.","accounts_in_message":[],"_revision_number":1},{"id":"e57ca8f5433c33fceaf1afd6bc7fdb1a879ff067","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-06-11 11:35:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6147286e00bc4a9352efe7ea77c0b25f39f8bec","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-11 16:44:35.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/75/664475/1/check-not-vote/ext-glance-zuul/3279f18 : SUCCESS in 38m 51s","accounts_in_message":[],"_revision_number":1},{"id":"51d3a7ef8d87fd3e3869f11be7e6ab2d4018dad3","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-12 02:13:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3e020756a1e935ca6eceb0f30b973bbe4035372f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-12 02:50:37.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/75/664475/2/check-not-vote/ext-glance-zuul/1d86add : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":2},{"id":"c4b94ffb4359e482d19b27d5f47933b27907f6b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 11:12:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/75/664475/2/check/build-openstack-api-ref/88dd9bb/html/ : SUCCESS in 4m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/664475/2/check/openstack-tox-lower-constraints/36f362b/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/75/664475/2/check/openstack-tox-pep8/617941a/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/75/664475/2/check/openstack-tox-py27/33b080f/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/75/664475/2/check/openstack-tox-py36/c1ae6a3/ : SUCCESS in 4m 38s\n- openstack-tox-py37 http://logs.openstack.org/75/664475/2/check/openstack-tox-py37/29a0de2/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/75/664475/2/check/openstack-tox-docs/f87a92f/html/ : SUCCESS in 4m 39s\n- build-openstack-releasenotes http://logs.openstack.org/75/664475/2/check/build-openstack-releasenotes/1af7340/html/ : SUCCESS in 8m 23s\n- openstack-tox-functional http://logs.openstack.org/75/664475/2/check/openstack-tox-functional/08215ad/ : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":2},{"id":"659c159268e2c8420b6d6a3a854ff980d741c6ca","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-13 14:38:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9b1eb4f0f82bdd26986bfd3d61b79eb4b25b99b","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-06-13 14:39:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ed0391dde594cf24ac144781e93a033c292ff7b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-13 14:39:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9e5d9ffa5d5151cd288394eb6d6b75f0057e9db","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-14 01:17:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5a990c1dab40487ff82deb834c8a0fe5384d7823","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-14 01:54:09.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/75/664475/3/check-not-vote/ext-glance-zuul/9a45123 : SUCCESS in 36m 46s","accounts_in_message":[],"_revision_number":3},{"id":"34e66b15b2400654fe7c60085ef7318309a49462","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 05:54:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/75/664475/3/check/build-openstack-api-ref/0b3b7b1/ : POST_FAILURE in 3m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/664475/3/check/openstack-tox-lower-constraints/147115e/ : FAILURE in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/75/664475/3/check/openstack-tox-pep8/6010957/ : FAILURE in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/75/664475/3/check/openstack-tox-py27/8c567b7/ : FAILURE in 2m 58s\n- openstack-tox-py36 http://logs.openstack.org/75/664475/3/check/openstack-tox-py36/7ccf0b2/ : FAILURE in 2m 26s\n- openstack-tox-py37 http://logs.openstack.org/75/664475/3/check/openstack-tox-py37/5b6c87b/ : FAILURE in 3m 05s\n- openstack-tox-docs http://logs.openstack.org/75/664475/3/check/openstack-tox-docs/2fc4250/ : POST_FAILURE in 3m 22s\n- build-openstack-releasenotes http://logs.openstack.org/75/664475/3/check/build-openstack-releasenotes/015dc8b/html/ : SUCCESS in 8m 00s\n- openstack-tox-functional http://logs.openstack.org/75/664475/3/check/openstack-tox-functional/fd9f9ae/ : FAILURE in 3m 09s","accounts_in_message":[],"_revision_number":3},{"id":"0a2ffa0bfbbd69b06d7879d69b0912298b96caa7","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-14 06:39:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9c4b164e4fcbaccae217136818547fa9dc447e07","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-14 06:51:12.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/75/664475/4/check-not-vote/ext-glance-zuul/e40afe8 : FAILURE in 11m 09s","accounts_in_message":[],"_revision_number":4},{"id":"64aa5571cd5f6f716da40ba781589816ab206557","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 10:45:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/75/664475/4/check/build-openstack-api-ref/cdadc89/html/ : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/664475/4/check/openstack-tox-lower-constraints/eba4cc9/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/75/664475/4/check/openstack-tox-pep8/d791701/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/75/664475/4/check/openstack-tox-py27/b6b1c71/ : SUCCESS in 5m 21s\n- openstack-tox-py36 http://logs.openstack.org/75/664475/4/check/openstack-tox-py36/9d20a31/ : SUCCESS in 5m 58s\n- openstack-tox-py37 http://logs.openstack.org/75/664475/4/check/openstack-tox-py37/9402312/ : SUCCESS in 6m 34s\n- openstack-tox-docs http://logs.openstack.org/75/664475/4/check/openstack-tox-docs/c38dc61/html/ : SUCCESS in 5m 29s\n- build-openstack-releasenotes http://logs.openstack.org/75/664475/4/check/build-openstack-releasenotes/6894983/html/ : SUCCESS in 7m 22s\n- openstack-tox-functional http://logs.openstack.org/75/664475/4/check/openstack-tox-functional/24b60a0/ : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":4},{"id":"2614cc74d2609a31d9a6b83ecf2152174b2e0cad","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-14 15:49:53.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nSorry, noticed one more thing.","accounts_in_message":[],"_revision_number":4},{"id":"60215a638386e8cb15808737cf3ae86c30440dde","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-17 01:37:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"beb19ca9d6dcc160a06a394317a361b29d5e8f30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 01:37:58.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  Job openstack-tox-functional-py37 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - integrated-gate\n        - integrated-gate-py3\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python3-train-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-functional\n          - openstack-tox-functional-py37\n          - glance-eventlet-ssl-handshake-broken-py3:\n              voting: false\n          - glance-code-constants-check\n          - devstack-plugin-ceph-tempest:\n              voting: false\n              irrelevant-files: \u0026tempest-irrelevant-files\n                - ^(test-|)requirements.txt$\n                - ^lower-constraints.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^glance/hacking/.*$\n                - ^glance/locale/.*$\n                - ^glance/tests/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n          - tempest-full:\n              irrelevant-files: *tempest-irrelevant-files\n          - neutron-grenade:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-full-py3:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade-py3:\n              irrelevant-files: *tempest-irrelevant-files\n  \n      gate:\n        jobs:\n          - openstack-tox-functional\n          - openstack-tox-functional-py37\n          - tempest-full:\n              irrelevant-files: *tempest-irrelevant-files\n          - neutron-grenade:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-full-py3:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade-py3:\n              irrelevant-files: *tempest-irrelevant-files\n      experimental:\n        jobs:\n          - glance-tox-py27-glance_store-tips\n          - glance-tox-py36-glance_store-tips\n          - glance-tox-functional-glance_store-tips\n          - glance-tox-functional-py36-glance_store-tips\n          - barbican-simple-crypto-devstack-tempest\n          - glance-dsvm-grenade-multinode\n          - devstack-plugin-ceph-tempest-py3:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-pg-full:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-full-py3-opensuse150:\n              irrelevant-files: *tempest-irrelevant-files\n      periodic:\n        jobs:\n          # NOTE(rosmaita): we only want the \"tips\" jobs to be run against\n          # master, hence the \u0027branches\u0027 qualifiers below.  Without them, when\n          # a stable branch is cut, the tests would be run against the stable\n          # branch as well, which is pointless because these libraries are\n          # frozen (more or less) in the stable branches.\n          #\n          # The \"tips\" jobs can be removed from the stable branch .zuul.yaml\n          # files if someone is so inclined, but that would require manual\n          # maintenance, so we do not do it by default.  Another option is\n          # to define these jobs in the openstack/project-config repo.\n          # That would make us less agile in adjusting these tests, so we\n          # aren\u0027t doing that either.\n          - glance-tox-functional-oslo-tips:\n              branches: master\n          - glance-tox-functional-py36-oslo-tips:\n              branches: master\n          - glance-tox-py27-keystone-tips:\n              branches: master\n          - glance-tox-py36-keystone-tips:\n              branches: master\n          - glance-tox-functional-keystone-tips:\n              branches: master\n          - glance-tox-functional-py36-keystone-tips:\n              branches: master\n          - glance-tox-py27-glance_store-tips:\n              branches: master\n          - glance-tox-py36-glance_store-tips:\n              branches: master\n          - glance-tox-functional-glance_store-tips:\n              branches: master\n          - glance-tox-functional-py36-glance_store-tips:\n              branches: master\n          - glance-tox-py27-cursive-tips:\n              branches: master\n          - glance-tox-py36-cursive-tips:\n              branches: master\n          - glance-tox-functional-cursive-tips:\n              branches: master\n          - glance-tox-functional-py36-cursive-tips:\n              branches: master\n\n  in \"openstack/glance/.zuul.yaml@master\", line 219, column 3","accounts_in_message":[],"_revision_number":5},{"id":"41d992a10970963c125d644321521f5c5e4e0253","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-17 02:15:03.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/75/664475/5/check-not-vote/ext-glance-zuul/6d6f4d1 : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":5},{"id":"0b460f0bda510a01d75c786167c394fc5ef46c8b","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-18 07:38:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bf157bafcaf890174cbfdefa941787e3fadbed75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:38:16.000000000","message":"Patch Set 6: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  found duplicate anchor \u0027tempest-irrelevant-files\u0027; first occurrence\n  in \"openstack/glance/.zuul.yaml@master\", line 234, column 31\nsecond occurrence\n  in \"openstack/glance/.zuul.yaml@master\", line 248, column 31","accounts_in_message":[],"_revision_number":6},{"id":"907a1f544c2824b80e474d728da5e350c9005288","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-18 07:44:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3ec378d14cecb50b81de38fe7b2093ccc9893558","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:45:04.000000000","message":"Patch Set 7: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  found undefined alias \u0027tempest-irrelevant-files\u0027\n  in \"openstack/glance/.zuul.yaml@master\", line 234, column 31","accounts_in_message":[],"_revision_number":7},{"id":"6925c78d50aedeea2ab94e00e48c90bfdcf8771f","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-18 07:46:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3912d20f3355100a8d665a219dbbba30907173c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 10:48:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/75/664475/8/check/build-openstack-api-ref/6b256cf/html/ : SUCCESS in 4m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/664475/8/check/openstack-tox-lower-constraints/2d92f05/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/75/664475/8/check/openstack-tox-pep8/88c719f/ : SUCCESS in 5m 49s\n- openstack-tox-py27 http://logs.openstack.org/75/664475/8/check/openstack-tox-py27/2f0dc1d/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/75/664475/8/check/openstack-tox-py36/9d77a8d/ : SUCCESS in 4m 08s\n- openstack-tox-py37 http://logs.openstack.org/75/664475/8/check/openstack-tox-py37/4b1290e/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/75/664475/8/check/openstack-tox-docs/8327f46/html/ : SUCCESS in 5m 58s\n- build-openstack-releasenotes http://logs.openstack.org/75/664475/8/check/build-openstack-releasenotes/1b42251/html/ : SUCCESS in 7m 26s\n- openstack-tox-functional http://logs.openstack.org/75/664475/8/check/openstack-tox-functional/4e1fb67/ : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":8},{"id":"b3b9738e8ea6eb5dd4176306a7dbc28e78e7d511","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-18 12:50:22.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/75/664475/8/check-not-vote/ext-glance-zuul/be43387 : SUCCESS in 37m 51s","accounts_in_message":[],"_revision_number":8},{"id":"fbaa185256fd150d2b681122246e52df854f2a0b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-24 14:30:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"96af6a20d7945e2e268f618189540dd909947276","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:30:52.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  Job openstack-tox-functional-py37 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - integrated-gate\n        - integrated-gate-py3\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python3-train-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-functional\n          - openstack-tox-functional-py37\n          - glance-eventlet-ssl-handshake-broken-py3:\n              voting: false\n          - glance-code-constants-check\n          - devstack-plugin-ceph-tempest:\n              voting: false\n              irrelevant-files: \u0026tempest-irrelevant-files\n                - ^(test-|)requirements.txt$\n                - ^lower-constraints.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^glance/hacking/.*$\n                - ^glance/locale/.*$\n                - ^glance/tests/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n          - tempest-full:\n              irrelevant-files: *tempest-irrelevant-files\n          - neutron-grenade:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-full-py3:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade-py3:\n              irrelevant-files: *tempest-irrelevant-files\n  \n      gate:\n        jobs:\n          - openstack-tox-functional\n          - openstack-tox-functional-py37\n          - tempest-full:\n              irrelevant-files: *tempest-irrelevant-files\n          - neutron-grenade:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-full-py3:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade-py3:\n              irrelevant-files: *tempest-irrelevant-files\n      experimental:\n        jobs:\n          - glance-tox-py27-glance_store-tips\n          - glance-tox-py36-glance_store-tips\n          - glance-tox-functional-glance_store-tips\n          - glance-tox-functional-py36-glance_store-tips\n          - barbican-simple-crypto-devstack-tempest\n          - glance-dsvm-grenade-multinode\n          - devstack-plugin-ceph-tempest-py3:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-pg-full:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-full-py3-opensuse150:\n              irrelevant-files: *tempest-irrelevant-files\n      periodic:\n        jobs:\n          # NOTE(rosmaita): we only want the \"tips\" jobs to be run against\n          # master, hence the \u0027branches\u0027 qualifiers below.  Without them, when\n          # a stable branch is cut, the tests would be run against the stable\n          # branch as well, which is pointless because these libraries are\n          # frozen (more or less) in the stable branches.\n          #\n          # The \"tips\" jobs can be removed from the stable branch .zuul.yaml\n          # files if someone is so inclined, but that would require manual\n          # maintenance, so we do not do it by default.  Another option is\n          # to define these jobs in the openstack/project-config repo.\n          # That would make us less agile in adjusting these tests, so we\n          # aren\u0027t doing that either.\n          - glance-tox-functional-oslo-tips:\n              branches: master\n          - glance-tox-functional-py36-oslo-tips:\n              branches: master\n          - glance-tox-py27-keystone-tips:\n              branches: master\n          - glance-tox-py36-keystone-tips:\n              branches: master\n          - glance-tox-functional-keystone-tips:\n              branches: master\n          - glance-tox-functional-py36-keystone-tips:\n              branches: master\n          - glance-tox-py27-glance_store-tips:\n              branches: master\n          - glance-tox-py36-glance_store-tips:\n              branches: master\n          - glance-tox-functional-glance_store-tips:\n              branches: master\n          - glance-tox-functional-py36-glance_store-tips:\n              branches: master\n          - glance-tox-py27-cursive-tips:\n              branches: master\n          - glance-tox-py36-cursive-tips:\n              branches: master\n          - glance-tox-functional-cursive-tips:\n              branches: master\n          - glance-tox-functional-py36-cursive-tips:\n              branches: master\n\n  in \"openstack/glance/.zuul.yaml@master\", line 219, column 3","accounts_in_message":[],"_revision_number":9},{"id":"45167e881e5cf4ddb7bab999e219bb79969ff92a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-24 14:33:40.000000000","message":"Topic changed from Bug#2 to python3-train","accounts_in_message":[],"_revision_number":9},{"id":"6b4969c541d5fd243b21caa5a76f4dcc85025e4d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-24 14:37:37.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"dd2bc0f1cc566fabe96c51d678a6c1fe69d9691e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-24 14:38:14.000000000","message":"Topic changed from Bug#2 to python3-train","accounts_in_message":[],"_revision_number":10},{"id":"b3c1e68a644a3ee28763c1b6c856585db25e0287","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 15:28:22.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/75/664475/10/check/test-release-openstack/3787cf3/ : SUCCESS in 3m 10s\n- build-openstack-api-ref http://logs.openstack.org/75/664475/10/check/build-openstack-api-ref/9aeb801/html/ : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/664475/10/check/openstack-tox-lower-constraints/a018a01/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/75/664475/10/check/openstack-tox-pep8/81c9769/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/75/664475/10/check/openstack-tox-py27/410752e/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/75/664475/10/check/openstack-tox-py36/12f1bce/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/75/664475/10/check/openstack-tox-py37/c30ed4c/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/75/664475/10/check/openstack-tox-docs/d0339cf/html/ : SUCCESS in 4m 50s\n- build-openstack-releasenotes http://logs.openstack.org/75/664475/10/check/build-openstack-releasenotes/7956613/html/ : SUCCESS in 8m 33s\n- openstack-tox-functional http://logs.openstack.org/75/664475/10/check/openstack-tox-functional/3ad99bc/ : SUCCESS in 9m 03s\n- openstack-tox-functional-py37 http://logs.openstack.org/75/664475/10/check/openstack-tox-functional-py37/2fbe946/ : FAILURE in 3m 58s","accounts_in_message":[],"_revision_number":10},{"id":"42757734ef7a1134624ba4b4a6cfb34facc859b3","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-03 04:20:27.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"23e4d7c6cfc2f1bf7c0dfd481e2709a9757f5eb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 04:31:03.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/75/664475/11/check/test-release-openstack/6fcfaf8/ : SUCCESS in 3m 27s\n- build-openstack-api-ref http://logs.openstack.org/75/664475/11/check/build-openstack-api-ref/aa4f94b/html/ : SUCCESS in 3m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/664475/11/check/openstack-tox-lower-constraints/9a7b2cc/ : SUCCESS in 4m 56s\n- openstack-tox-pep8 http://logs.openstack.org/75/664475/11/check/openstack-tox-pep8/a96a9a7/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/75/664475/11/check/openstack-tox-py27/e9d1619/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/75/664475/11/check/openstack-tox-py36/93fd79b/ : SUCCESS in 4m 21s\n- openstack-tox-py37 http://logs.openstack.org/75/664475/11/check/openstack-tox-py37/449e692/ : SUCCESS in 6m 07s\n- openstack-tox-docs http://logs.openstack.org/75/664475/11/check/openstack-tox-docs/e196988/html/ : SUCCESS in 4m 37s\n- build-openstack-releasenotes http://logs.openstack.org/75/664475/11/check/build-openstack-releasenotes/2f52a0c/html/ : SUCCESS in 7m 29s\n- openstack-tox-functional http://logs.openstack.org/75/664475/11/check/openstack-tox-functional/86f44c0/ : SUCCESS in 7m 44s\n- openstack-tox-functional-py37 http://logs.openstack.org/75/664475/11/check/openstack-tox-functional-py37/e90df2d/ : FAILURE in 2m 41s","accounts_in_message":[],"_revision_number":11},{"id":"97dd5cdfa7412d29e4d4e550dba227ac6f7578d5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-03 11:59:03.000000000","message":"Patch Set 11:\n\nThis also needs https://review.opendev.org/#/c/668466/ to land.","accounts_in_message":[],"_revision_number":11},{"id":"9ae008f34e38a165a0e846088389232fc4bc8ac2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-03 12:16:46.000000000","message":"Patch Set 11:\n\nrecheck\n\nInterpreter patch has now merged.","accounts_in_message":[],"_revision_number":11},{"id":"d9d5977b6c2db2688f4426a04be59d522a6976f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 14:28:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/75/664475/11/check/test-release-openstack/0d23b87/ : SUCCESS in 4m 04s\n- build-openstack-api-ref http://logs.openstack.org/75/664475/11/check/build-openstack-api-ref/18d744f/html/ : SUCCESS in 3m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/664475/11/check/openstack-tox-lower-constraints/0ba4462/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/75/664475/11/check/openstack-tox-pep8/1cd238b/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/75/664475/11/check/openstack-tox-py27/5a1f175/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/75/664475/11/check/openstack-tox-py36/a78bb8c/ : SUCCESS in 4m 16s\n- openstack-tox-py37 http://logs.openstack.org/75/664475/11/check/openstack-tox-py37/ae6499b/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/75/664475/11/check/openstack-tox-docs/07aa125/html/ : SUCCESS in 5m 06s\n- build-openstack-releasenotes http://logs.openstack.org/75/664475/11/check/build-openstack-releasenotes/8385484/html/ : SUCCESS in 8m 50s\n- openstack-tox-functional http://logs.openstack.org/75/664475/11/check/openstack-tox-functional/876d0d6/ : SUCCESS in 8m 01s\n- openstack-tox-functional-py37 http://logs.openstack.org/75/664475/11/check/openstack-tox-functional-py37/8361c17/ : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":11},{"id":"498453ed0288e48dff681e475115929fe3a7183d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-03 15:26:10.000000000","message":"Patch Set 11: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":11},{"id":"bdf3d54348237cc128aa732dd0b6cea0b0a0c064","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-04 13:22:06.000000000","message":"Patch Set 11: Code-Review-2\n\n(1 comment)\n\nPlease remove the change to setup.cfg","accounts_in_message":[],"_revision_number":11},{"id":"eb81975620d742054d65bea2530a2f5810b0ee8c","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-05 01:14:15.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"7ad5a695a74bcd34c5c8ae4387eb0676692033b9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-05 01:14:35.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":12},{"id":"2252dd40515cd72aa47a222e2778c21cfce31657","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 01:27:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/75/664475/12/check/build-openstack-api-ref/0cab73f/html/ : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/664475/12/check/openstack-tox-lower-constraints/71e9098/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/75/664475/12/check/openstack-tox-pep8/9467a30/ : SUCCESS in 6m 10s\n- openstack-tox-py27 http://logs.openstack.org/75/664475/12/check/openstack-tox-py27/d8ea120/ : SUCCESS in 5m 43s\n- openstack-tox-py36 http://logs.openstack.org/75/664475/12/check/openstack-tox-py36/f7240e2/ : SUCCESS in 5m 30s\n- openstack-tox-py37 http://logs.openstack.org/75/664475/12/check/openstack-tox-py37/cdc3a5c/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/75/664475/12/check/openstack-tox-docs/cb6ab4b/html/ : SUCCESS in 5m 33s\n- build-openstack-releasenotes http://logs.openstack.org/75/664475/12/check/build-openstack-releasenotes/757a280/html/ : SUCCESS in 7m 39s\n- openstack-tox-functional http://logs.openstack.org/75/664475/12/check/openstack-tox-functional/9b8c9f6/ : SUCCESS in 7m 36s\n- openstack-tox-functional-py37 http://logs.openstack.org/75/664475/12/check/openstack-tox-functional-py37/3817566/ : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":12},{"id":"daf13aa654a18097adc9e218dad4e01cb9abade0","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-05 02:33:28.000000000","message":"Patch Set 12:\n\n\u003e (1 comment)\n \u003e \n \u003e Please remove the change to setup.cfg\n\nDone","accounts_in_message":[],"_revision_number":12},{"id":"8a43373a78b4ddaa71d67658e923790fb4b214f6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-12 20:53:55.000000000","message":"Patch Set 12:\n\n(1 comment)\n\nBrian, are the updates sufficient to address you -2?","accounts_in_message":[],"_revision_number":12},{"id":"75276269188cd416d208d47a37d303020a616bad","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-15 01:39:05.000000000","message":"Uploaded patch set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"781fe0305b27355f599795e1e078d0d407e81770","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-15 01:39:26.000000000","message":"Patch Set 13:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":13},{"id":"e385be3e270b8372e2470e2ed925ee61b2614298","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 01:53:29.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/75/664475/13/check/build-openstack-api-ref/df210ea/html/ : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/664475/13/check/openstack-tox-lower-constraints/07ddfff/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/75/664475/13/check/openstack-tox-pep8/631c80f/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/75/664475/13/check/openstack-tox-py27/20d9cd8/ : SUCCESS in 5m 01s\n- openstack-tox-py36 http://logs.openstack.org/75/664475/13/check/openstack-tox-py36/01992be/ : SUCCESS in 3m 38s\n- openstack-tox-py37 http://logs.openstack.org/75/664475/13/check/openstack-tox-py37/280a0da/ : SUCCESS in 5m 34s\n- openstack-tox-docs http://logs.openstack.org/75/664475/13/check/openstack-tox-docs/b166939/html/ : SUCCESS in 5m 53s\n- build-openstack-releasenotes http://logs.openstack.org/75/664475/13/check/build-openstack-releasenotes/de9dcc7/html/ : SUCCESS in 8m 06s\n- openstack-tox-functional http://logs.openstack.org/75/664475/13/check/openstack-tox-functional/477942f/ : SUCCESS in 9m 19s\n- openstack-tox-functional-py37 http://logs.openstack.org/75/664475/13/check/openstack-tox-functional-py37/482cf6c/ : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":13},{"id":"471ca95e8ce477e88a17f6eb7a8e0c2be0bf027d","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-08-21 13:18:18.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"b7d118612f2e9e4f64a8e8ca5fc3b3059a0f4837","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-05 13:45:07.000000000","message":"Patch Set 13: Code-Review+2\n\nAll my concerns have been addressed. Please accept my apologies for forgetting to follow up on this sooner!","accounts_in_message":[],"_revision_number":13},{"id":"ddda5d9471bc4e01aae5097b32f9ad315e1ab5a0","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-09-05 15:05:04.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"e0ccc13aa71a0569de06b0adfac664cb60ccc45d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 15:05:30.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"e4b83e42c8b94d171042d17310f76fb720fc809c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 17:49:48.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ccee885c19cb4f329ada3d1d35389254 : SUCCESS in 7m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32ebe643d652449bab697c2ee449e4fb : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ea90f6d72734ed2826537596065594a : SUCCESS in 7m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5ed8449be6549f2b8f52767ce2b541d : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1c8d5db222848f9b1d8798784761efd : SUCCESS in 8m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa5b8bd67ba34cebbbd63b0ac06baf4b : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8097020ec5f441deb78e7afdfbf934c0 : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/483602bc41524592aa54e2f171bbac70 : SUCCESS in 12m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/48d905282d1c44e8aa622014e0da1a12 : SUCCESS in 11m 33s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/028f8b64efb74f64a195a5614e40a08c : SUCCESS in 13m 37s","accounts_in_message":[],"_revision_number":13},{"id":"544f490c16506596315270c429e28ce9b5a2a5f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 17:49:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"c3ef84744da525bb7a6639a989a76420c2b2551e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 17:54:54.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/719a25d401654b8a94fecf80922b25ef : SUCCESS in 1m 18s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ec83e9b08f48a68ccc3e58125f5692 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"b7e9db70465f19d4a4b008a93147a0649ca36535","revisions":{"806d5c319c1f72d468a69fd5dd54794a870ae812":{"kind":"REWORK","_number":1,"created":"2019-06-11 01:28:37.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/75/664475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/664475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/664475/1"}}},"commit":{"parents":[{"commit":"b8a34d31284227d29c467ffe33de1873e21a4633","subject":"Merge \"Don\u0027t target broken ssl tests to specific py3.minor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8a34d31284227d29c467ffe33de1873e21a4633"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:09:01.000000000","tz":480},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/806d5c319c1f72d468a69fd5dd54794a870ae812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/806d5c319c1f72d468a69fd5dd54794a870ae812"}]},"branch":"refs/heads/master"},"13465df69efe513673070a9e711920748f3b254c":{"kind":"REWORK","_number":2,"created":"2019-06-12 02:13:58.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/75/664475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/664475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/664475/2"}}},"commit":{"parents":[{"commit":"b8a34d31284227d29c467ffe33de1873e21a4633","subject":"Merge \"Don\u0027t target broken ssl tests to specific py3.minor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8a34d31284227d29c467ffe33de1873e21a4633"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-12 01:54:25.000000000","tz":480},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13465df69efe513673070a9e711920748f3b254c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13465df69efe513673070a9e711920748f3b254c"}]},"branch":"refs/heads/master"},"15858af818f964d378824a3769e83c2c91b15024":{"kind":"REWORK","_number":3,"created":"2019-06-14 01:17:01.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/75/664475/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/664475/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/664475/3"}}},"commit":{"parents":[{"commit":"b8a34d31284227d29c467ffe33de1873e21a4633","subject":"Merge \"Don\u0027t target broken ssl tests to specific py3.minor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8a34d31284227d29c467ffe33de1873e21a4633"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-14 00:58:16.000000000","tz":480},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/15858af818f964d378824a3769e83c2c91b15024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/15858af818f964d378824a3769e83c2c91b15024"}]},"branch":"refs/heads/master"},"7e041b1024ffa76378249c05f4c6f745502709f1":{"kind":"REWORK","_number":4,"created":"2019-06-14 06:39:42.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/75/664475/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/664475/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/664475/4"}}},"commit":{"parents":[{"commit":"b8a34d31284227d29c467ffe33de1873e21a4633","subject":"Merge \"Don\u0027t target broken ssl tests to specific py3.minor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8a34d31284227d29c467ffe33de1873e21a4633"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-14 06:19:15.000000000","tz":480},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e041b1024ffa76378249c05f4c6f745502709f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e041b1024ffa76378249c05f4c6f745502709f1"}]},"branch":"refs/heads/master"},"56606cd876200dd448677f4fa5471966db513df5":{"kind":"REWORK","_number":5,"created":"2019-06-17 01:37:44.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/75/664475/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/664475/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/664475/5"}}},"commit":{"parents":[{"commit":"b8a34d31284227d29c467ffe33de1873e21a4633","subject":"Merge \"Don\u0027t target broken ssl tests to specific py3.minor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8a34d31284227d29c467ffe33de1873e21a4633"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-17 01:17:27.000000000","tz":480},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56606cd876200dd448677f4fa5471966db513df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56606cd876200dd448677f4fa5471966db513df5"}]},"branch":"refs/heads/master"},"f1d0135517c1333f8c02c6427e4f396e61ac4d3a":{"kind":"REWORK","_number":6,"created":"2019-06-18 07:38:01.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/75/664475/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/664475/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/664475/6"}}},"commit":{"parents":[{"commit":"b8a34d31284227d29c467ffe33de1873e21a4633","subject":"Merge \"Don\u0027t target broken ssl tests to specific py3.minor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8a34d31284227d29c467ffe33de1873e21a4633"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-18 07:19:08.000000000","tz":480},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1d0135517c1333f8c02c6427e4f396e61ac4d3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1d0135517c1333f8c02c6427e4f396e61ac4d3a"}]},"branch":"refs/heads/master"},"14b0f1bb4b0827c419b5079bd750719ac31d1431":{"kind":"REWORK","_number":7,"created":"2019-06-18 07:44:49.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/75/664475/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/664475/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/664475/7"}}},"commit":{"parents":[{"commit":"b8a34d31284227d29c467ffe33de1873e21a4633","subject":"Merge \"Don\u0027t target broken ssl tests to specific py3.minor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8a34d31284227d29c467ffe33de1873e21a4633"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-18 07:25:58.000000000","tz":480},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14b0f1bb4b0827c419b5079bd750719ac31d1431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14b0f1bb4b0827c419b5079bd750719ac31d1431"}]},"branch":"refs/heads/master"},"437d714dcff8231de0deefdf853b1731528d5b1f":{"kind":"REWORK","_number":8,"created":"2019-06-18 07:46:46.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/75/664475/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/664475/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/664475/8"}}},"commit":{"parents":[{"commit":"b8a34d31284227d29c467ffe33de1873e21a4633","subject":"Merge \"Don\u0027t target broken ssl tests to specific py3.minor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8a34d31284227d29c467ffe33de1873e21a4633"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-18 07:27:51.000000000","tz":480},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/437d714dcff8231de0deefdf853b1731528d5b1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/437d714dcff8231de0deefdf853b1731528d5b1f"}]},"branch":"refs/heads/master"},"b5c8e433371692c2bb31f447d26e3c6cc2b85c61":{"kind":"REWORK","_number":9,"created":"2019-06-24 14:30:35.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/75/664475/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/664475/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/664475/9"}}},"commit":{"parents":[{"commit":"b8a34d31284227d29c467ffe33de1873e21a4633","subject":"Merge \"Don\u0027t target broken ssl tests to specific py3.minor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8a34d31284227d29c467ffe33de1873e21a4633"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-06-24 14:29:36.000000000","tz":-300},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5c8e433371692c2bb31f447d26e3c6cc2b85c61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5c8e433371692c2bb31f447d26e3c6cc2b85c61"}]},"branch":"refs/heads/master"},"473905c4df683045600a7762e2077de09217b84b":{"kind":"REWORK","_number":10,"created":"2019-06-24 14:37:37.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/75/664475/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/664475/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/664475/10"}}},"commit":{"parents":[{"commit":"b8a34d31284227d29c467ffe33de1873e21a4633","subject":"Merge \"Don\u0027t target broken ssl tests to specific py3.minor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8a34d31284227d29c467ffe33de1873e21a4633"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-06-24 14:37:23.000000000","tz":-300},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nDepends-on: https://review.opendev.org/667119\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/473905c4df683045600a7762e2077de09217b84b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/473905c4df683045600a7762e2077de09217b84b"}]},"branch":"refs/heads/master"},"d67c011e1c8c1f978afcc63698675d51cf3bfee0":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-07-03 04:20:27.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/75/664475/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/664475/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/664475/11"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-03 04:20:27.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nDepends-on: https://review.opendev.org/667119\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d67c011e1c8c1f978afcc63698675d51cf3bfee0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d67c011e1c8c1f978afcc63698675d51cf3bfee0"}]},"branch":"refs/heads/master"},"0753c1cbbdb4a107e429a7f957188cd76e1d60eb":{"kind":"REWORK","_number":12,"created":"2019-07-05 01:14:15.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/75/664475/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/664475/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/664475/12"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-05 01:14:02.000000000","tz":480},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nDepends-on: https://review.opendev.org/667119\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0753c1cbbdb4a107e429a7f957188cd76e1d60eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0753c1cbbdb4a107e429a7f957188cd76e1d60eb"}]},"branch":"refs/heads/master"},"b7e9db70465f19d4a4b008a93147a0649ca36535":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2019-07-15 01:39:05.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/75/664475/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/664475/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/664475/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/664475/13"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-15 01:37:43.000000000","tz":480},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nStory: #2005924\nTask: #34208\n\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b7e9db70465f19d4a4b008a93147a0649ca36535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b7e9db70465f19d4a4b008a93147a0649ca36535"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
