)]}'
{"id":"openstack%2Fglance~665153","triplet_id":"openstack%2Fglance~master~I4d97c0577460949e8636e9f26b1d4b60b109f5d1","project":"openstack/glance","branch":"master","topic":"bring-py27-tests-back","hashtags":[],"change_id":"I4d97c0577460949e8636e9f26b1d4b60b109f5d1","subject":"Bring py27 tests back","status":"MERGED","created":"2019-06-13 14:50:56.000000000","updated":"2019-06-14 07:45:35.000000000","submitted":"2019-06-14 07:45:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665153-1560498335944-d90c8b36","meta_rev_id":"0e221f8459b9abffe536fe467ce308b872b7c30c","_number":665153,"virtual_id_number":665153,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-06-13 15:34:30.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2019-06-14 07:45:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-13 14:52:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-14 06:14:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-14 01:14:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-14 06:14:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-13 14:52:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 14:52:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-13 15:34:30.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-14 01:14:36.000000000","updated_by":{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"},"reviewer":{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"},"state":"REVIEWER"},{"updated":"2019-06-14 06:14:18.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-06-14 07:45:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc22baafecc6d8077002b83eeb0b19633ebb350d","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-06-13 14:50:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da3c25d3314f8c940ccc63c7743d18960bc4c498","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-13 14:52:28.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2\n\nLooks correct to me.","accounts_in_message":[],"_revision_number":1},{"id":"3eb8b27fc5224c1ce6412d6a383b612ce46722c7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-13 15:34:30.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/53/665153/1/check-not-vote/ext-glance-zuul/41c26c8 : SUCCESS in 43m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0f8209e49911d9b55a54459f1337d463af4f62e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 23:43:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/53/665153/1/check/build-openstack-api-ref/e817c47/html/ : SUCCESS in 3m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/665153/1/check/openstack-tox-lower-constraints/b155dc1/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/53/665153/1/check/openstack-tox-pep8/975bd7e/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/53/665153/1/check/openstack-tox-py27/ad282cf/ : SUCCESS in 4m 04s\n- openstack-tox-py35 http://logs.openstack.org/53/665153/1/check/openstack-tox-py35/bb49dee/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/53/665153/1/check/openstack-tox-py36/905bbfd/ : SUCCESS in 5m 45s\n- openstack-tox-py37 http://logs.openstack.org/53/665153/1/check/openstack-tox-py37/8c700bd/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/53/665153/1/check/openstack-tox-docs/05f8cf9/html/ : SUCCESS in 5m 05s\n- build-openstack-releasenotes http://logs.openstack.org/53/665153/1/check/build-openstack-releasenotes/9b49d71/html/ : SUCCESS in 8m 33s\n- openstack-tox-functional http://logs.openstack.org/53/665153/1/check/openstack-tox-functional/23b30d7/ : SUCCESS in 8m 07s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/665153/1/check/openstack-tox-functional-py35/8d6a4c6/ : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c4567bc6553e42fc105f619655254989583e8bed","author":{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"},"date":"2019-06-14 01:14:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cf5d4b79d55c1a4e35a2cad9290c81392c49072f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-14 06:14:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nConfirmed that tests are running under python2.7 environment after this chanage.","accounts_in_message":[],"_revision_number":1},{"id":"ac23db975e78cf3774df2499311fa95a309838cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 06:14:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"871c877f1d96d6cfe1bd3a2cb815b7f23128db68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 07:45:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"63e5d3ecc3ddd7774420ba433aa40c22501fd6ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 07:45:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/53/665153/1/gate/build-openstack-api-ref/937b444/html/ : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/665153/1/gate/openstack-tox-lower-constraints/ad65eb1/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/53/665153/1/gate/openstack-tox-pep8/349ceae/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/53/665153/1/gate/openstack-tox-py27/8c2af11/ : SUCCESS in 4m 45s\n- openstack-tox-py35 http://logs.openstack.org/53/665153/1/gate/openstack-tox-py35/c48ea99/ : SUCCESS in 5m 16s\n- openstack-tox-py36 http://logs.openstack.org/53/665153/1/gate/openstack-tox-py36/4f79c25/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/53/665153/1/gate/openstack-tox-py37/b361c76/ : SUCCESS in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/53/665153/1/gate/openstack-tox-docs/c543743/html/ : SUCCESS in 5m 25s\n- build-openstack-releasenotes http://logs.openstack.org/53/665153/1/gate/build-openstack-releasenotes/94f9793/html/ : SUCCESS in 8m 35s\n- openstack-tox-functional http://logs.openstack.org/53/665153/1/gate/openstack-tox-functional/f38357a/ : SUCCESS in 9m 10s\n- openstack-tox-functional-py35 http://logs.openstack.org/53/665153/1/gate/openstack-tox-functional-py35/101c761/ : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c57b78653bc0fdc91ab29e42e72aeffda07c4da","revisions":{"3c57b78653bc0fdc91ab29e42e72aeffda07c4da":{"kind":"REWORK","_number":1,"created":"2019-06-13 14:50:56.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/53/665153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/665153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/665153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/665153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/665153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/665153/1"}}},"commit":{"parents":[{"commit":"b8a34d31284227d29c467ffe33de1873e21a4633","subject":"Merge \"Don\u0027t target broken ssl tests to specific py3.minor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8a34d31284227d29c467ffe33de1873e21a4633"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-06-13 14:49:37.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-06-13 14:49:37.000000000","tz":60},"subject":"Bring py27 tests back","message":"Bring py27 tests back\n\nSince defaulting to py3 our 2.7 jobs have actually ran against python 3\ndue to basepython not declared on our 27 evironments. This patch will\nchange that.\n\nChange-Id: I4d97c0577460949e8636e9f26b1d4b60b109f5d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c57b78653bc0fdc91ab29e42e72aeffda07c4da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c57b78653bc0fdc91ab29e42e72aeffda07c4da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
