)]}'
{"id":"openstack%2Fglance~665606","triplet_id":"openstack%2Fglance~master~I1386c535bc8ff4519e6b0bb879026b05c930b791","project":"openstack/glance","branch":"master","topic":"backend-to-store-migration","hashtags":[],"change_id":"I1386c535bc8ff4519e6b0bb879026b05c930b791","subject":"Add migration script to change backend to store","status":"MERGED","created":"2019-06-17 08:19:33.000000000","updated":"2019-08-21 10:42:37.000000000","submitted":"2019-08-15 11:26:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665606-1565868411225-d4deb22e","meta_rev_id":"3ee16ddcf41230f44dbc58ec93aeaadb6af35b89","_number":665606,"virtual_id_number":665606,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-06 08:41:47.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-14 15:23:10.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-08-15 11:26:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-06 16:31:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-08-15 09:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2019-08-13 15:58:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-08-15 09:43:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-08-06 08:23:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2019-08-21 09:23:01.000000000","post_submit":true,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 08:41:47.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-06 16:31:19.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-08-13 15:58:39.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-08-14 15:23:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-15 09:43:15.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-08-15 11:26:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 09:23:01.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"3d09925f433f2b6e7098f6a37d8264557631e92d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-17 08:19:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92c3dcdc20467b9b50d202fe4e0a119f3c304459","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-17 08:54:26.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/06/665606/1/check-not-vote/ext-glance-zuul/aeffdf0 : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":1},{"id":"0c514969c7a52e9e467dde1e07a82693db19dfdf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-17 09:49:17.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4e8af9b227f1e3d2834ffcff81aaf24285ef2cdc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-17 10:02:12.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"fbe393fa62a12da48c1c1a4250440bbfc1a81d3f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-17 10:13:06.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"eb34c1bddef31aa5542aae36f74ea1bb6feaf340","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-17 10:57:58.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/06/665606/4/check-not-vote/ext-glance-zuul/02d91d6 : SUCCESS in 42m 48s","accounts_in_message":[],"_revision_number":4},{"id":"661baf6b2e2d24915c6e815a729acc58ba0c6865","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 12:39:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/06/665606/4/check/tempest-full/2d06a33/ : SUCCESS in 2h 05m 11s\n- neutron-grenade http://logs.openstack.org/06/665606/4/check/neutron-grenade/42956e9/ : SUCCESS in 52m 50s\n- grenade-py3 http://logs.openstack.org/06/665606/4/check/grenade-py3/4eb21ac/ : SUCCESS in 1h 05m 15s\n- tempest-full-py3 http://logs.openstack.org/06/665606/4/check/tempest-full-py3/099ee2e/ : SUCCESS in 1h 42m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/665606/4/check/openstack-tox-lower-constraints/fc56c61/ : SUCCESS in 6m 02s\n- openstack-tox-pep8 http://logs.openstack.org/06/665606/4/check/openstack-tox-pep8/1814074/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/06/665606/4/check/openstack-tox-py27/d75467b/ : SUCCESS in 4m 33s\n- openstack-tox-py35 http://logs.openstack.org/06/665606/4/check/openstack-tox-py35/d2653e6/ : SUCCESS in 6m 35s\n- openstack-tox-py36 http://logs.openstack.org/06/665606/4/check/openstack-tox-py36/fbb6e87/ : FAILURE in 5m 13s\n- openstack-tox-py37 http://logs.openstack.org/06/665606/4/check/openstack-tox-py37/38252dc/ : FAILURE in 5m 53s\n- openstack-tox-docs http://logs.openstack.org/06/665606/4/check/openstack-tox-docs/4afa0c8/html/ : SUCCESS in 5m 49s\n- openstack-tox-functional http://logs.openstack.org/06/665606/4/check/openstack-tox-functional/d5dc3f5/ : SUCCESS in 10m 41s\n- openstack-tox-functional-py35 http://logs.openstack.org/06/665606/4/check/openstack-tox-functional-py35/08ae782/ : SUCCESS in 11m 07s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/06/665606/4/check/glance-eventlet-ssl-handshake-broken-py3/bde2334/ : FAILURE in 4m 58s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/06/665606/4/check/glance-code-constants-check/5dff07d/ : SUCCESS in 4m 49s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/06/665606/4/check/devstack-plugin-ceph-tempest/be7810a/ : TIMED_OUT in 2h 04m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7c583e9c54012a5dfe23bc0069cfd73a6a4e488e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-17 12:43:08.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"092f82ddd891530462a8e6856faf3a5146094c06","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-17 13:19:40.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/06/665606/4/check-not-vote/ext-glance-zuul/405369d : SUCCESS in 36m 12s","accounts_in_message":[],"_revision_number":4},{"id":"bb93c3e58c46b4d2d6c56dccb7118aa7613115e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 16:48:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/06/665606/4/check/tempest-full/e6ee670/ : FAILURE in 18m 36s\n- neutron-grenade http://logs.openstack.org/06/665606/4/check/neutron-grenade/ff17aa4/ : SUCCESS in 54m 17s\n- grenade-py3 http://logs.openstack.org/06/665606/4/check/grenade-py3/d8cff46/ : SUCCESS in 59m 43s\n- tempest-full-py3 http://logs.openstack.org/06/665606/4/check/tempest-full-py3/c7937ab/ : SUCCESS in 1h 28m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/665606/4/check/openstack-tox-lower-constraints/168ed16/ : SUCCESS in 4m 43s\n- openstack-tox-pep8 http://logs.openstack.org/06/665606/4/check/openstack-tox-pep8/a22360d/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/06/665606/4/check/openstack-tox-py27/12b1a06/ : SUCCESS in 4m 58s\n- openstack-tox-py35 http://logs.openstack.org/06/665606/4/check/openstack-tox-py35/5bd7c9a/ : SUCCESS in 4m 44s\n- openstack-tox-py36 http://logs.openstack.org/06/665606/4/check/openstack-tox-py36/183b84f/ : SUCCESS in 4m 34s\n- openstack-tox-py37 http://logs.openstack.org/06/665606/4/check/openstack-tox-py37/503fe1a/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/06/665606/4/check/openstack-tox-docs/812d6e4/html/ : SUCCESS in 5m 36s\n- openstack-tox-functional http://logs.openstack.org/06/665606/4/check/openstack-tox-functional/a64251e/ : SUCCESS in 9m 45s\n- openstack-tox-functional-py35 http://logs.openstack.org/06/665606/4/check/openstack-tox-functional-py35/b0605da/ : FAILURE in 8m 40s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/06/665606/4/check/glance-eventlet-ssl-handshake-broken-py3/4d87b79/ : FAILURE in 4m 41s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/06/665606/4/check/glance-code-constants-check/6c93009/ : SUCCESS in 4m 33s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/06/665606/4/check/devstack-plugin-ceph-tempest/2193e3f/ : SUCCESS in 1h 29m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"218f5f1b3b2ee77c462535fb38c2489a32a5fcd9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-17 16:51:35.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"51f245bab5c17a09bb094387f792ee16745f7097","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-17 17:35:22.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/06/665606/4/check-not-vote/ext-glance-zuul/be7b6be : SUCCESS in 43m 26s","accounts_in_message":[],"_revision_number":4},{"id":"ac5e72b235584aeeb44d75e971d89dec1c7d792d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 21:47:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/06/665606/4/check/tempest-full/018c860/ : SUCCESS in 1h 56m 01s\n- neutron-grenade http://logs.openstack.org/06/665606/4/check/neutron-grenade/bfc1ec0/ : SUCCESS in 56m 10s\n- grenade-py3 http://logs.openstack.org/06/665606/4/check/grenade-py3/212a520/ : SUCCESS in 57m 52s\n- tempest-full-py3 http://logs.openstack.org/06/665606/4/check/tempest-full-py3/0b4d7ab/ : SUCCESS in 1h 31m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/665606/4/check/openstack-tox-lower-constraints/4f15d06/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/06/665606/4/check/openstack-tox-pep8/157d06b/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/06/665606/4/check/openstack-tox-py27/b5d4f91/ : SUCCESS in 4m 42s\n- openstack-tox-py35 http://logs.openstack.org/06/665606/4/check/openstack-tox-py35/6a31192/ : SUCCESS in 4m 37s\n- openstack-tox-py36 http://logs.openstack.org/06/665606/4/check/openstack-tox-py36/9e4f98f/ : SUCCESS in 4m 25s\n- openstack-tox-py37 http://logs.openstack.org/06/665606/4/check/openstack-tox-py37/c8f4326/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/06/665606/4/check/openstack-tox-docs/9757fb2/html/ : SUCCESS in 5m 23s\n- openstack-tox-functional http://logs.openstack.org/06/665606/4/check/openstack-tox-functional/fd0ad87/ : SUCCESS in 8m 44s\n- openstack-tox-functional-py35 http://logs.openstack.org/06/665606/4/check/openstack-tox-functional-py35/8a44562/ : SUCCESS in 9m 17s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/06/665606/4/check/glance-eventlet-ssl-handshake-broken-py3/969806d/ : FAILURE in 5m 13s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/06/665606/4/check/glance-code-constants-check/57023df/ : SUCCESS in 4m 37s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/06/665606/4/check/devstack-plugin-ceph-tempest/75a9e4b/ : SUCCESS in 1h 26m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5229b8a7be0ed3c590c92cc97a1411cc8121390a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-18 04:22:21.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"12bc78908e5d4cb341d20b01ab742f382807cf2f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-05 13:50:54.000000000","message":"Patch Set 4: Code-Review+1\n\n(3 comments)\n\nQuestion about the migration script inline (and two nits).\n\nThe structure of the files, etc., looks correct to me.","accounts_in_message":[],"_revision_number":4},{"id":"37b14c7015c22737735067b6d46a89460e5d71d7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-06 05:49:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"854e165505521ab084380a6e02c2864f53cd5c71","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-06 05:50:02.000000000","message":"Patch Set 4:\n\n(3 comments)\n\nHi Brian,\n\nThank you for review!!!","accounts_in_message":[],"_revision_number":4},{"id":"31c4dc8f7e4aa159219b02e070be92859138de90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 08:20:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://logs.opendev.org/06/665606/5/check/tempest-full/6a8d536/ : SUCCESS in 1h 41m 38s\n- neutron-grenade https://logs.opendev.org/06/665606/5/check/neutron-grenade/6afab06/ : SUCCESS in 1h 00m 54s\n- grenade-py3 https://logs.opendev.org/06/665606/5/check/grenade-py3/cd144b1/ : SUCCESS in 1h 08m 41s\n- tempest-full-py3 https://logs.opendev.org/06/665606/5/check/tempest-full-py3/2131aca/ : SUCCESS in 1h 31m 14s\n- openstack-tox-lower-constraints https://logs.opendev.org/06/665606/5/check/openstack-tox-lower-constraints/0199dc3/ : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://logs.opendev.org/06/665606/5/check/openstack-tox-pep8/ab2e373/ : SUCCESS in 6m 11s\n- openstack-tox-py27 https://logs.opendev.org/06/665606/5/check/openstack-tox-py27/863e38c/ : SUCCESS in 5m 50s\n- openstack-tox-py35 https://logs.opendev.org/06/665606/5/check/openstack-tox-py35/d0aae85/ : SUCCESS in 7m 24s\n- openstack-tox-py36 https://logs.opendev.org/06/665606/5/check/openstack-tox-py36/40e6be2/ : SUCCESS in 5m 47s\n- openstack-tox-py37 https://logs.opendev.org/06/665606/5/check/openstack-tox-py37/f4a85d2/ : SUCCESS in 6m 20s\n- openstack-tox-docs https://logs.opendev.org/06/665606/5/check/openstack-tox-docs/1363325/html/ : SUCCESS in 6m 00s\n- openstack-tox-functional https://logs.opendev.org/06/665606/5/check/openstack-tox-functional/e891888/ : SUCCESS in 8m 52s\n- openstack-tox-functional-py35 https://logs.opendev.org/06/665606/5/check/openstack-tox-functional-py35/1579300/ : FAILURE in 11m 18s\n- glance-eventlet-ssl-handshake-broken-py3 https://logs.opendev.org/06/665606/5/check/glance-eventlet-ssl-handshake-broken-py3/eb566d6/ : FAILURE in 5m 44s (non-voting)\n- glance-code-constants-check https://logs.opendev.org/06/665606/5/check/glance-code-constants-check/c380eac/ : SUCCESS in 5m 17s\n- devstack-plugin-ceph-tempest https://logs.opendev.org/06/665606/5/check/devstack-plugin-ceph-tempest/67531e2/ : SUCCESS in 1h 09m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"df97f89f12a58b38f912103a34ec4706351c575d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-06 08:23:50.000000000","message":"Patch Set 5: Review-Priority+1\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b826d88cb73b4b4ca9c033204ff35c3c205ebd06","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-06 08:41:47.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/06/665606/5/check-not-vote/ext-glance-zuul/4a75441 : FAILURE in 1m 39s","accounts_in_message":[],"_revision_number":5},{"id":"5f5f1e0daf6b2cd0311e20049af5ec3449d8d1cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 12:08:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://logs.opendev.org/06/665606/5/check/tempest-full/1aaad69/ : SUCCESS in 1h 40m 44s\n- neutron-grenade https://logs.opendev.org/06/665606/5/check/neutron-grenade/e713ff3/ : SUCCESS in 1h 07m 29s\n- grenade-py3 https://logs.opendev.org/06/665606/5/check/grenade-py3/bf9de6a/ : SUCCESS in 59m 18s\n- tempest-full-py3 https://logs.opendev.org/06/665606/5/check/tempest-full-py3/eee74ae/ : SUCCESS in 1h 22m 29s\n- openstack-tox-lower-constraints https://logs.opendev.org/06/665606/5/check/openstack-tox-lower-constraints/6e510a0/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://logs.opendev.org/06/665606/5/check/openstack-tox-pep8/386efee/ : SUCCESS in 6m 18s\n- openstack-tox-py27 https://logs.opendev.org/06/665606/5/check/openstack-tox-py27/fc970eb/ : SUCCESS in 4m 59s\n- openstack-tox-py35 https://logs.opendev.org/06/665606/5/check/openstack-tox-py35/4b16713/ : SUCCESS in 6m 09s\n- openstack-tox-py36 https://logs.opendev.org/06/665606/5/check/openstack-tox-py36/037bf22/ : SUCCESS in 5m 23s\n- openstack-tox-py37 https://logs.opendev.org/06/665606/5/check/openstack-tox-py37/7eaa69d/ : SUCCESS in 5m 57s\n- openstack-tox-docs https://logs.opendev.org/06/665606/5/check/openstack-tox-docs/96eed5d/html/ : SUCCESS in 6m 02s\n- openstack-tox-functional https://logs.opendev.org/06/665606/5/check/openstack-tox-functional/3208fa9/ : SUCCESS in 11m 07s\n- openstack-tox-functional-py35 https://logs.opendev.org/06/665606/5/check/openstack-tox-functional-py35/170c9a6/ : SUCCESS in 11m 16s\n- glance-eventlet-ssl-handshake-broken-py3 https://logs.opendev.org/06/665606/5/check/glance-eventlet-ssl-handshake-broken-py3/634ce2c/ : FAILURE in 6m 22s (non-voting)\n- glance-code-constants-check https://logs.opendev.org/06/665606/5/check/glance-code-constants-check/d1babec/ : SUCCESS in 6m 20s\n- devstack-plugin-ceph-tempest https://logs.opendev.org/06/665606/5/check/devstack-plugin-ceph-tempest/691e7b0/ : SUCCESS in 1h 20m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c07cbbb1719fd61bedb20b5d3eeb830c96dce55c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-06 16:31:19.000000000","message":"Patch Set 5: Code-Review+2\n\nAll my concerns have been addressed, and I was able to successfully apply this to a stein DB (version: rocky_contract02) and upgrade to version train_contract01.  Existing locations were updated correctly.","accounts_in_message":[],"_revision_number":5},{"id":"0a0074ae9252116b6d7b04b6d9f2598beb7ed054","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-08-13 15:58:39.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"278ba989d7d21ece098cf86f28bd37f7cdd48734","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-08-14 15:15:44.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nQuestion about commit message, otwerwise looks good.","accounts_in_message":[],"_revision_number":5},{"id":"21150aad80ab43ba5568b45ea87178d85b2c1013","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-14 15:22:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b530077890ab263b9567ed8a121daee5232652d9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-14 15:23:10.000000000","message":"Patch Set 5:\n\nHere is a slightly more recent reference: https://docs.openstack.org/pbr/3.0.0/#version","accounts_in_message":[],"_revision_number":5},{"id":"bb6b0813f1f904917dcde03cb5caaba51e2b7353","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-08-15 09:43:15.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"075ce1ca0e3bb2a316b87ed8482258dc65a719ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 09:43:34.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a2ae53c48dcc8b56d6f1d9e49f384f9a5f33a95f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 11:26:50.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c93941585d214ec1ad61bf67ccce084b : SUCCESS in 1h 41m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cc2ca92baa344d7b8968488cb75b4f1a : SUCCESS in 57m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/04e418f7b6a64938b5bb588f29d68fcf : SUCCESS in 57m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fc577b92051441cba56ab5ee2a869f1 : SUCCESS in 1h 21m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/517b49b03aab4b1e9774b2e97e97757c : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73db2f2c59c24f7a8451a24f5058c971 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8746dbbd9a2488b8a128a2d271a1138 : SUCCESS in 5m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f3a83e5142da43429c6898913efe8027 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60a18edb5ff54b8793517f384024a835 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c239e09f1e384c0f87a976eb777ad6a6 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/268fddc7cb4e48d49cdcaa660722d6f4 : SUCCESS in 5m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/56b875d884774a3598b51c4a38754716 : SUCCESS in 9m 25s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5511b9abd54948e9a63462b589414773 : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":5},{"id":"69e9840be7dc973c5191c73f78afc9ee2ac7a256","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 11:26:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"fdb66c1ae913d49e18ca2c463dc5354ea89ea4da","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-21 09:19:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f8272cc7779d15003defd63f9657d3df56f3974a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-21 10:42:37.000000000","message":"Patch Set 5:\n\nHi Zhurong,\n\nPlease verify https://review.opendev.org/677694","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b4e3cb65ad2e04f0f5b94dcaf1fa951d01bfb025","revisions":{"b38ca58cb6e03e4af5658d27497e91a15eaa3e28":{"kind":"REWORK","_number":1,"created":"2019-06-17 08:19:33.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/06/665606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/665606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/665606/1"}}},"commit":{"parents":[{"commit":"b5b827f9eabd5b85e56ef522f94bb971f5308091","subject":"Merge \"Bring py27 tests back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5b827f9eabd5b85e56ef522f94bb971f5308091"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-17 08:06:15.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-17 08:12:16.000000000","tz":0},"subject":"Add migration script to change backend to store","message":"Add migration script to change backend to store\n\nAs glance is changing metadata value of location from \u0027backend\u0027 to\n\u0027store\u0027, adding migration script which will update old image locations\nto use \u0027store\u0027 as a metadata value.\n\nNOTE:\nBump master (train development) to 19\n\nChange-Id: I1386c535bc8ff4519e6b0bb879026b05c930b791\nSem-Ver: api-break\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b38ca58cb6e03e4af5658d27497e91a15eaa3e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b38ca58cb6e03e4af5658d27497e91a15eaa3e28"}]},"branch":"refs/heads/master"},"c15ac5b6c006841cb6c850b5b5c92af5bab731a0":{"kind":"REWORK","_number":2,"created":"2019-06-17 09:49:17.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/06/665606/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/665606/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/665606/2"}}},"commit":{"parents":[{"commit":"b5b827f9eabd5b85e56ef522f94bb971f5308091","subject":"Merge \"Bring py27 tests back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5b827f9eabd5b85e56ef522f94bb971f5308091"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-17 08:06:15.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-17 09:48:38.000000000","tz":0},"subject":"Add migration script to change backend to store","message":"Add migration script to change backend to store\n\nAs glance is changing metadata value of location from \u0027backend\u0027 to\n\u0027store\u0027, adding migration script which will update old image locations\nto use \u0027store\u0027 as a metadata value.\n\nNOTE:\nBump master (train development) to 19\n\nChange-Id: I1386c535bc8ff4519e6b0bb879026b05c930b791\nSem-Ver: api-break\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c15ac5b6c006841cb6c850b5b5c92af5bab731a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c15ac5b6c006841cb6c850b5b5c92af5bab731a0"}]},"branch":"refs/heads/master"},"6923773572a1f9ecce0eb9af6506791660b412b4":{"kind":"REWORK","_number":3,"created":"2019-06-17 10:02:12.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/06/665606/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/665606/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/665606/3"}}},"commit":{"parents":[{"commit":"b5b827f9eabd5b85e56ef522f94bb971f5308091","subject":"Merge \"Bring py27 tests back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5b827f9eabd5b85e56ef522f94bb971f5308091"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-17 08:06:15.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-17 09:59:37.000000000","tz":0},"subject":"Add migration script to change backend to store","message":"Add migration script to change backend to store\n\nAs glance is changing metadata value of location from \u0027backend\u0027 to\n\u0027store\u0027, adding migration script which will update old image locations\nto use \u0027store\u0027 as a metadata value.\n\nNOTE:\nBump master (train development) to 19\n\nChange-Id: I1386c535bc8ff4519e6b0bb879026b05c930b791\nSem-Ver: api-break\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6923773572a1f9ecce0eb9af6506791660b412b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6923773572a1f9ecce0eb9af6506791660b412b4"}]},"branch":"refs/heads/master"},"5361257dd429d44607870a99c1555b3c24a27d80":{"kind":"REWORK","_number":4,"created":"2019-06-17 10:13:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/06/665606/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/665606/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/665606/4"}}},"commit":{"parents":[{"commit":"b5b827f9eabd5b85e56ef522f94bb971f5308091","subject":"Merge \"Bring py27 tests back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5b827f9eabd5b85e56ef522f94bb971f5308091"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-17 08:06:15.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-17 10:12:59.000000000","tz":0},"subject":"Add migration script to change backend to store","message":"Add migration script to change backend to store\n\nAs glance is changing metadata value of location from \u0027backend\u0027 to\n\u0027store\u0027, adding migration script which will update old image locations\nto use \u0027store\u0027 as a metadata value.\n\nNOTE:\nBump master (train development) to 19\n\nChange-Id: I1386c535bc8ff4519e6b0bb879026b05c930b791\nSem-Ver: api-break\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5361257dd429d44607870a99c1555b3c24a27d80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5361257dd429d44607870a99c1555b3c24a27d80"}]},"branch":"refs/heads/master"},"b4e3cb65ad2e04f0f5b94dcaf1fa951d01bfb025":{"kind":"REWORK","_number":5,"created":"2019-08-06 05:49:23.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/06/665606/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/665606/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/665606/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/665606/5"}}},"commit":{"parents":[{"commit":"733095a89947718e54b610ad528f6243e6ece440","subject":"Merge \"Update api-ref location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/733095a89947718e54b610ad528f6243e6ece440"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-17 08:06:15.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-08-06 05:48:50.000000000","tz":0},"subject":"Add migration script to change backend to store","message":"Add migration script to change backend to store\n\nAs glance is changing metadata value of location from \u0027backend\u0027 to\n\u0027store\u0027, adding migration script which will update old image locations\nto use \u0027store\u0027 as a metadata value.\n\nNOTE:\nBump master (train development) to 19\n\nChange-Id: I1386c535bc8ff4519e6b0bb879026b05c930b791\nSem-Ver: api-break\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b4e3cb65ad2e04f0f5b94dcaf1fa951d01bfb025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b4e3cb65ad2e04f0f5b94dcaf1fa951d01bfb025"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
