)]}'
{"id":"openstack%2Fglance~667132","triplet_id":"openstack%2Fglance~master~Id3ac19488c0a693d7042be4a3c83f3b9f12313d0","project":"openstack/glance","branch":"master","topic":"bp/import-multi-stores","hashtags":[],"change_id":"Id3ac19488c0a693d7042be4a3c83f3b9f12313d0","subject":"Add ability to import image into multi-stores","status":"MERGED","created":"2019-06-24 15:23:56.000000000","updated":"2020-02-10 13:51:32.000000000","submitted":"2020-02-10 13:41:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":43,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667132-1581342061902-71032318","meta_rev_id":"37f244a2242928096df369eb4232e0a13e861db2","_number":667132,"virtual_id_number":667132,"owner":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 13:41:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-10 10:54:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},{"value":2,"date":"2020-02-10 11:10:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},{"value":1,"date":"2020-02-10 11:10:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-10 10:54:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},{"value":1,"date":"2020-02-10 13:51:32.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 15:24:09.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-08-01 10:58:20.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2019-12-05 13:40:20.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2019-12-13 21:56:30.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-01-29 07:25:01.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-08 18:22:10.000000000","updated_by":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"reviewer":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"state":"REVIEWER"},{"updated":"2020-02-10 10:54:10.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-02-10 11:10:52.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-02-10 13:41:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3685022b83940eaad46f38b3cee2f21f5a20e241","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-06-24 15:23:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a08dd739810fdfc47deb522569c1d7ff589aa131","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-06-24 15:24:09.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"622969a905a0fd98fb65d199f2949948ab5196f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 18:15:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/32/667132/1/check/tempest-full/b626fb9/ : SUCCESS in 1h 55m 35s\n- neutron-grenade http://logs.openstack.org/32/667132/1/check/neutron-grenade/2b59eae/ : SUCCESS in 57m 37s\n- grenade-py3 http://logs.openstack.org/32/667132/1/check/grenade-py3/06f1f08/ : SUCCESS in 59m 28s\n- tempest-full-py3 http://logs.openstack.org/32/667132/1/check/tempest-full-py3/4ddc571/ : SUCCESS in 1h 30m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/667132/1/check/openstack-tox-lower-constraints/b37c621/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/32/667132/1/check/openstack-tox-pep8/9861105/ : FAILURE in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/32/667132/1/check/openstack-tox-py27/39acaf9/ : FAILURE in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/32/667132/1/check/openstack-tox-py35/4964617/ : FAILURE in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/32/667132/1/check/openstack-tox-py36/6b01a1d/ : FAILURE in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/32/667132/1/check/openstack-tox-py37/687bf4a/ : FAILURE in 5m 06s\n- openstack-tox-docs http://logs.openstack.org/32/667132/1/check/openstack-tox-docs/6861201/html/ : SUCCESS in 5m 51s\n- openstack-tox-functional http://logs.openstack.org/32/667132/1/check/openstack-tox-functional/944b956/ : FAILURE in 9m 04s\n- openstack-tox-functional-py35 http://logs.openstack.org/32/667132/1/check/openstack-tox-functional-py35/6fd734a/ : FAILURE in 8m 44s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/32/667132/1/check/glance-eventlet-ssl-handshake-broken-py3/736f5d4/ : FAILURE in 5m 25s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/32/667132/1/check/glance-code-constants-check/47abc29/ : SUCCESS in 5m 48s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/32/667132/1/check/devstack-plugin-ceph-tempest/08f20cf/ : SUCCESS in 1h 37m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31c7b9d7440222b6540db6a2cb4e4f2c40afd8dd","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-06-25 12:31:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe9af63a38cef74869a2a5b32843a9435f0532d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 01:24:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/32/667132/2/check/tempest-full/56d5803/ : SUCCESS in 1h 37m 12s\n- neutron-grenade http://logs.openstack.org/32/667132/2/check/neutron-grenade/912e01d/ : SUCCESS in 1h 01m 36s\n- grenade-py3 http://logs.openstack.org/32/667132/2/check/grenade-py3/ea6836a/ : SUCCESS in 1h 00m 13s\n- tempest-full-py3 http://logs.openstack.org/32/667132/2/check/tempest-full-py3/ae4a980/ : SUCCESS in 1h 38m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/667132/2/check/openstack-tox-lower-constraints/1931e80/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/32/667132/2/check/openstack-tox-pep8/fd128cd/ : SUCCESS in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/32/667132/2/check/openstack-tox-py27/daad296/ : SUCCESS in 5m 19s\n- openstack-tox-py35 http://logs.openstack.org/32/667132/2/check/openstack-tox-py35/938d3f3/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/32/667132/2/check/openstack-tox-py36/a668fac/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/32/667132/2/check/openstack-tox-py37/2c7da8f/ : SUCCESS in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/32/667132/2/check/openstack-tox-docs/a9e1b02/html/ : SUCCESS in 4m 54s\n- openstack-tox-functional http://logs.openstack.org/32/667132/2/check/openstack-tox-functional/885336f/ : FAILURE in 8m 09s\n- openstack-tox-functional-py35 http://logs.openstack.org/32/667132/2/check/openstack-tox-functional-py35/2c4cb81/ : FAILURE in 8m 35s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/32/667132/2/check/glance-eventlet-ssl-handshake-broken-py3/a7dd140/ : FAILURE in 3m 48s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/32/667132/2/check/glance-code-constants-check/cb74582/ : SUCCESS in 4m 03s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/32/667132/2/check/devstack-plugin-ceph-tempest/6e08f9e/ : SUCCESS in 1h 16m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a477ad352f4391a94cbd6e40101f46e2b376d4af","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-06-28 07:42:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2c3f3ce2b18d144b590a56e11e0dd68aac279df4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 10:27:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/32/667132/3/check/tempest-full/8e388fb/ : SUCCESS in 2h 01m 59s\n- neutron-grenade http://logs.openstack.org/32/667132/3/check/neutron-grenade/226418b/ : SUCCESS in 57m 03s\n- grenade-py3 http://logs.openstack.org/32/667132/3/check/grenade-py3/e8a3279/ : SUCCESS in 57m 36s\n- tempest-full-py3 http://logs.openstack.org/32/667132/3/check/tempest-full-py3/9cea474/ : SUCCESS in 1h 33m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/667132/3/check/openstack-tox-lower-constraints/fddce13/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/32/667132/3/check/openstack-tox-pep8/585a736/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/32/667132/3/check/openstack-tox-py27/937cfd7/ : SUCCESS in 4m 05s\n- openstack-tox-py35 http://logs.openstack.org/32/667132/3/check/openstack-tox-py35/1b6d774/ : SUCCESS in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/32/667132/3/check/openstack-tox-py36/29a8ac5/ : SUCCESS in 3m 51s\n- openstack-tox-py37 http://logs.openstack.org/32/667132/3/check/openstack-tox-py37/5a57f56/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/32/667132/3/check/openstack-tox-docs/6d2a5e6/html/ : SUCCESS in 4m 46s\n- openstack-tox-functional http://logs.openstack.org/32/667132/3/check/openstack-tox-functional/08bed13/ : SUCCESS in 9m 37s\n- openstack-tox-functional-py35 http://logs.openstack.org/32/667132/3/check/openstack-tox-functional-py35/6868805/ : SUCCESS in 10m 11s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/32/667132/3/check/glance-eventlet-ssl-handshake-broken-py3/777d08f/ : FAILURE in 4m 08s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/32/667132/3/check/glance-code-constants-check/aae6aae/ : SUCCESS in 4m 25s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/32/667132/3/check/devstack-plugin-ceph-tempest/c26b2e7/ : SUCCESS in 1h 20m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ba53bfbcc0f2050b56c5f878296741b30c8b6605","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2019-06-28 11:11:41.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM, thanks for this feature","accounts_in_message":[],"_revision_number":3},{"id":"09cdc4e6276f830c4348ae40043e4d1a30f35d03","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-07-04 15:26:29.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"bd850cbcddf5d2e2908472c25e2740d8ef88fb6a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-04 15:26:48.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"1ea96c276db66915f5aea2849b31a1634a92c59f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 22:03:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/32/667132/4/check/tempest-full/159b392/ : SUCCESS in 1h 44m 18s\n- neutron-grenade http://logs.openstack.org/32/667132/4/check/neutron-grenade/a630927/ : SUCCESS in 58m 26s\n- grenade-py3 http://logs.openstack.org/32/667132/4/check/grenade-py3/65362e4/ : SUCCESS in 55m 11s\n- tempest-full-py3 http://logs.openstack.org/32/667132/4/check/tempest-full-py3/2cf7ee4/ : SUCCESS in 1h 52m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/667132/4/check/openstack-tox-lower-constraints/a1fdb8b/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/32/667132/4/check/openstack-tox-pep8/0dcea0e/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/32/667132/4/check/openstack-tox-py27/1aa6278/ : SUCCESS in 4m 50s\n- openstack-tox-py35 http://logs.openstack.org/32/667132/4/check/openstack-tox-py35/3bc5cbc/ : SUCCESS in 4m 10s\n- openstack-tox-py36 http://logs.openstack.org/32/667132/4/check/openstack-tox-py36/b247077/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/32/667132/4/check/openstack-tox-py37/0f7ae52/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/32/667132/4/check/openstack-tox-docs/adfc5bf/html/ : SUCCESS in 4m 56s\n- openstack-tox-functional http://logs.openstack.org/32/667132/4/check/openstack-tox-functional/3083efa/ : SUCCESS in 10m 36s\n- openstack-tox-functional-py35 http://logs.openstack.org/32/667132/4/check/openstack-tox-functional-py35/f1102ef/ : SUCCESS in 8m 46s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/32/667132/4/check/glance-eventlet-ssl-handshake-broken-py3/2fc5381/ : FAILURE in 4m 51s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/32/667132/4/check/glance-code-constants-check/b318e0d/ : SUCCESS in 4m 32s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/32/667132/4/check/devstack-plugin-ceph-tempest/22a04ba/ : SUCCESS in 1h 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c3359c99788e2193c9b9c393777111445ed4c3d7","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2019-07-06 11:15:23.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"89c96ebaf313dc4c6a25303cc1092aeeb7ba5450","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2019-07-06 11:17:24.000000000","message":"Patch Set 4:\n\nCan you add a releasenotes and edit the documentation according to your change ?","accounts_in_message":[],"_revision_number":4},{"id":"f2929ccda0b5d32420442ef58153c3b0a5a98d53","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-28 07:06:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"83591758ee0df6cc5094aa29249e692aff2a6ada","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-11-28 08:10:17.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/5/check-not-vote/ext-glance-zuul/2bfac47 : FAILURE in 39m 27s","accounts_in_message":[],"_revision_number":5},{"id":"bd367887f0fefad6e43d3abdb1bf6801a022161f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 08:30:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ef86cce3442e4a3b94c944c2d6452f4f : SUCCESS in 1h 22m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a0339f7a525041ca93ddc229d1aacc9c : SUCCESS in 51m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e1f0e89773304d5ca10dfc7ac7f831c6 : SUCCESS in 56m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/569bd75191ce482fabca337c0c780fcb : SUCCESS in 1h 17m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6bfa3deaaee43d189de2232482a1c36 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3265cb13bb6e4394873b87b36492f4e8 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8695ba198d494e16a323baff5583ba0a : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/142b15fc023e4be9a72cee6c02b63ed4 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/477938807a75481bbc5f913e74f3cb2e : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f31a2d24a764a19a44c7ca95c79b1d9 : SUCCESS in 6m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/764cbb5b70564bb7a68b9ec81cfa4435 : SUCCESS in 12m 40s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/cc8797a6c92b4beca6e948cd927fbf90 : SUCCESS in 12m 02s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/6acc3be128874eeda595078b7638fac0 : FAILURE in 4m 27s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ca16a0baf2204501880567a28e70899f : SUCCESS in 4m 20s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6dda5fc4c3694f299100690fd6dfc0ef : SUCCESS in 1h 03m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c7b772470b0672337905fdc83c95c36a937fbe93","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-11-28 10:00:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"268749935fa766bb74ceec86de757421ce2f98ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 11:44:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/7bd82ba5faab46ecae1d241d316d0b55 : SUCCESS in 1h 43m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1b0946d6c21f4b95994d76c7c0dc9108 : SUCCESS in 1h 11m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d6d63268a3c84ea785994d81b48ba62b : SUCCESS in 1h 07m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3722c9c6792f48ba9457a8abe742e713 : SUCCESS in 1h 26m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ceb57da1ca7f4a13859f4cea9c0bdbb9 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3c605bf97394648993f86f82ef8c1ae : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1db88f32e534ad9b8be6b1e8f46a211 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67e89e2afaec43df8e39b5188e276a21 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbfc6dca23b6420bb2f48ac22af74be9 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0f7945084e04682a224f7c6cb471c43 : SUCCESS in 6m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f4e01e0c119140cb89bbfbab93e2fbb6 : FAILURE in 8m 26s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/dfa519e9e7144063993c490b31bfc3de : FAILURE in 8m 40s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/fbe954b01d8e433194f657e27033f4b7 : FAILURE in 4m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e593ac1296444c7e9da798b848049d5e : SUCCESS in 3m 38s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d4852dc846944f11b0f3b1f38d62bf7d : SUCCESS in 1h 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"96df28c3988c645b5dc6ca9b7f67c17ec4ccdc6e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-11-28 12:49:02.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/6/check-not-vote/ext-glance-zuul/4addad4 : FAILURE in 27m 24s","accounts_in_message":[],"_revision_number":6},{"id":"d40268eb37c4ef202e88987f0bbe1ad167d803d0","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-11-28 13:09:29.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"c735829136ebdc6f402527c3c8af52afee263612","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-11-28 13:28:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"47464702a4088999a0efcf33db35a88d8fc67bc6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-28 14:25:18.000000000","message":"Patch Set 7:\n\nIMO you also need to check if image is present or not before each import as discussed and mentioned in the specs https://review.opendev.org/#/c/669201/13/specs/ussuri/approved/glance/import-multi-stores.rst@76","accounts_in_message":[],"_revision_number":7},{"id":"b1d0447ab3b9d335cb60dd2d0eaf2c5b4da078f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 15:57:02.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ad9e52b07a1240a3ac072793556cebd1 : SUCCESS in 1h 27m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/046a562c629443ee9ba67a1f6d60657c : SUCCESS in 1h 14m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/16da3329162e422192cec9f2927ea768 : SUCCESS in 1h 05m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b0816eb8b7e419085ea5171e9cf730b : SUCCESS in 1h 27m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cc6de0e6f434579bbbab1aa9fda13e3 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06abde2d59ac4859b84bb24efee1dfd7 : SUCCESS in 9m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32bfc9205d5f4f188517d8e38e1a9fec : SUCCESS in 10m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46ad7e5619194a27b816fc79d47b8fcc : FAILURE in 7m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26fa19040958495489fbaca7a9a34de7 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a21564c40c4dff9b606b5a0c6ea1e8 : SUCCESS in 11m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6a4000bbd60448fa87c06ceffd0f1a87 : SUCCESS in 13m 09s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/69457ccd7fde4ce687a00642a7f840a4 : SUCCESS in 8m 28s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/7a0a2584681c407fae43282f18d4d66b : FAILURE in 4m 20s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6a45ad2f94a6484882c317a90d3f754f : SUCCESS in 6m 05s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ed94647ec5fc4904a65fc8a8340a40f4 : SUCCESS in 1h 19m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"86502a5ec11eac1e1edc8afa1086eb43750d7427","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-11-28 17:06:25.000000000","message":"Patch Set 7:\n\n\u003e IMO you also need to check if image is present or not before each\n \u003e import as discussed and mentioned in the specs https://review.opendev.org/#/c/669201/13/specs/ussuri/approved/glance/import-multi-stores.rst@76\n\nYes sure I will add it. I just wanted to add the new \"stores\" parameter in this patch to allow you to work with this field on your POC. I couldn\u0027t afford to attend today glance meeting to discuss about this.","accounts_in_message":[],"_revision_number":7},{"id":"16ff88f46a843c7c365eeccc38afc70586d515c2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-11-29 03:27:57.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/7/check-not-vote/ext-glance-zuul/f6cd306 : FAILURE in 27m 39s","accounts_in_message":[],"_revision_number":7},{"id":"086ced12b4cf62e8e525ff2727a7795e2bf6da62","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-29 05:28:14.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)\n\nHi Gregoire,\n\nI know this is in WIP, still given some nits.\nAlso allow_failure is missing (or still in progress).","accounts_in_message":[],"_revision_number":7},{"id":"0d5c11b14cf9bf4c4c953a545706070ae14c4181","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-11-29 08:34:44.000000000","message":"Patch Set 7:\n\n(2 comments)\n\nHi Abhishek,\n\nThanks for the review, I\u0027ll continue the implementation with your comments in mind.","accounts_in_message":[],"_revision_number":7},{"id":"60f14e2f04b151cfb6ca9a6a374bd81ba680a82b","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-12-01 12:37:51.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"94aee4128c15e1ee112a77600ab3ba30189dd102","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-12-01 13:18:11.000000000","message":"Patch Set 7: Code-Review-1\n\nHi Greg,\n\nThanks for the work on this, I truly appreciate your help.\n\nI\u0027m not convinced \u0027though that the set_data is actually the right place to handle the multiple uploads. I really would prefer doing this in taskflow so we have better control and understanding what\u0027s going on. With taskflow at least we have the tasks API to debug and monitor the node. If we put all this multiple stores logic into the set_data, we quickly end up with black box running hours and no-one having any idea what it is doing.\n\nFortunately it would not be wasting much of the work you have done, we\u0027d just move the logic into the construct of that taskflow and add multiple upload to store tasks based on the store list. This also gives us better control of the error scenarios and how we want to recover or fail from them.\n\nI also added an idea of utilizing some reserved properties to indicate the progress into Abhishek\u0027s copy spec comments. Let me know what do you think.","accounts_in_message":[],"_revision_number":7},{"id":"4c3647e41b2f93c3c2a66828715f972fc2cd5aae","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-02 08:59:37.000000000","message":"Patch Set 7:\n\nHello Erno,\n\nThis seems like a good idea, It\u0027ll be easier to handle allow_failure parameter in this case using the revert method of _ImportToStore task.\nI\u0027ve made some refactoring as asked by Abhishek, I\u0027ll upload it with this new implementation.","accounts_in_message":[],"_revision_number":7},{"id":"2fc76e33af02d0af4a3787f051f97575a3ac2c60","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-02 16:43:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c424b3c1124ae2ed5f04c6042791e1458f47dcca","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-02 16:46:59.000000000","message":"Patch Set 8:\n\nFirst version with \"allow_failure\" and taskflow modified.\nEarly push to have feedback to ensure I\u0027m going in right direction.\nStill need to add some tests and make extensive tests on a lab.","accounts_in_message":[],"_revision_number":8},{"id":"23dc20874c1e35370d390ef8425327499d1e1cea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 18:27:42.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/47c24e6fca9a484183ec2808b7d452ce : SUCCESS in 1h 27m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1b20170d13ec48ad9efd501d7544b23c : SUCCESS in 1h 03m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bcb6eafb40de4dba87f60ad5d7f14504 : SUCCESS in 1h 18m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1bad37792774bd587b54f6bb6fe1813 : SUCCESS in 1h 16m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd5b5dd0fc154cbaa7c0ecbf9c072f96 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a67a6492eaf451f9fa67a691da66298 : FAILURE in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7135ffcb9c0843b4aa531434fa5176df : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f89fc3d954243618ecc5da76dead22e : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/003c62522d6f4399afd9dadbf31ecdf5 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b06d306ad140444e967950376a950ea5 : SUCCESS in 6m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/119d0bcfc31a4dd1ac6a987383c7de59 : FAILURE in 11m 08s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3c5c177fcc7b450a8561806bdca616d1 : FAILURE in 11m 46s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/894f7004dcd444f2a7cb6893e7c6b75d : FAILURE in 4m 32s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0dad0e7979154e8da7293d53109cd61c : SUCCESS in 4m 29s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/be104cf8baeb4facb6357e051bcca75c : SUCCESS in 1h 01m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"944157966db306638367169d7c6d9e94f15b9a15","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-03 05:36:00.000000000","message":"Patch Set 8:\n\n(2 comments)\n\nHi Gregoire,\n\nLooks like on right track, added some nits. I guess you need to implement os_glance_failed_imports and os_glance_successful_imports logic along with deletion check before each import starts.\n\nAlso need to send notification as well.\n\nThank you for your work!!!","accounts_in_message":[],"_revision_number":8},{"id":"67bff7cb858bce87ef325f632f1d761660cc9baf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-03 05:58:46.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"caa6af437b7fa79f790ae880cfb989e80e5d7580","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-03 07:17:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b8d962c9b7d842337bd3ba005c44364067c7cdf6","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-03 13:32:06.000000000","message":"Patch Set 8:\n\n(4 comments)\n\nHi Abhishek,\n\nThanks for your review.\nI have to work too on the way I pass the status of the image to active.","accounts_in_message":[],"_revision_number":8},{"id":"ef92cb1463e6fbf1fb8c3a6f54b81be92e9d2484","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-03 16:29:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"bac0b73b5db48a03de3470814cc482958e816680","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-03 17:02:47.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/9/check-not-vote/ext-glance-zuul/e87b501 : FAILURE in 28m 34s","accounts_in_message":[],"_revision_number":9},{"id":"151d29ca0bbfebd9889e8261086d6f820dc26514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 18:28:48.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/0f96068ad9f4499b8d95e1f513fffddd : FAILURE in 29m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f3f45a8586e7455eb72f6871002945b7 : SUCCESS in 54m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9ef156ffb7cf439697b17d804d9f634c : SUCCESS in 57m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e779d7f4c44415bb5c21351fe14c575 : FAILURE in 36m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18b5d5935c464c31a14b3a813c19e149 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf542d87ee3f4ecbb419ac0526f2838f : FAILURE in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/475c98853ade4fdb82cb20e2ac3ebf8a : FAILURE in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8db3bec4c5b84983ad119394ca23cafc : FAILURE in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fcf84a3ec9a94971b25596cf7c623143 : FAILURE in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66e0dca8bc08449c8783c8e0eae7ffda : SUCCESS in 5m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/988e5be21cee476bb0ae4b93380740ce : FAILURE in 8m 16s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/03db9884411d46ceb92d9205c95221de : FAILURE in 9m 52s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/1aa7899ee4364ddb97921676f3a88827 : FAILURE in 4m 18s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/09170e43ced14180879898f0259545b9 : SUCCESS in 4m 16s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8df793e1a81143d0a8682a6ccb3ece81 : FAILURE in 34m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f655c89aad9f485be3dda75a081ab08ee9002b6b","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-12-03 21:10:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"61405cbca4b57a1770a279aed87d5f140645f8c3","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-03 21:13:01.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"69a9f3b57a2d5fb9b9322b4db9cd7b8f0dca4684","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-03 21:16:54.000000000","message":"Patch Set 10:\n\n\u003e (1 comment)\n\nHello, the spec was changed and you can\u0027t anymore specify a comma delimited list in the header \u0027x-image-meta-store\u0027. You have to use the new parameter \"stores\" in the payload request.\nPatches 4/5 work with this old behavior.","accounts_in_message":[],"_revision_number":10},{"id":"1a936270f2e70caf70d88fd7c77b71de822bd8dd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-03 21:20:34.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d33c73be23931aa13790b2fef4db216a3725f2a4","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-03 21:33:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"af44664c32e7bb3b5fcd7c00aefe360966dd21f2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-03 21:40:26.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"92b113469543a41a466b0ed374074e19dff1e2cf","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-03 21:51:51.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"82b73a0739e1bf2b6313a3c827c43967aa5973c4","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-03 21:57:59.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6330322b0c87c888a21745c4b9bb5c4077529e39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 23:29:24.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9aba2f9af24a46eb9200776d061ddbcd : SUCCESS in 1h 30m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8e5a2045708b47069f1855226034cf70 : FAILURE in 35m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7f39bc26d08940139f4c292aa3bdf7a3 : FAILURE in 1h 06m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dceaffb0325f4b7eb4111a717becf91a : SUCCESS in 1h 35m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c940a903c3a245379703c0d1ceb11b2b : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c4a0c66ed8c448b81d53d61dbe4520c : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4804ed7821d24e85b29793545e40f4de : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e581e72a7e4042f38f68f2a8c1fccf84 : FAILURE in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4b17ebb716846fc8d02dc45d5ab7141 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/992f720d47294869b89bb0a045d3f7e7 : SUCCESS in 5m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d970caf4d5f244828089cc163a6dfc61 : FAILURE in 14m 13s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/13d586e24f724eadbb40e4516f4e4810 : SUCCESS in 9m 22s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/56c167b62cc54b3c8d049e03c16f934b : FAILURE in 6m 52s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b57d829d3185486aaf50a1040178f60b : SUCCESS in 4m 39s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5ca6214c76d142179cd114a34488fac9 : SUCCESS in 1h 06m 30s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9f21a3679b37e9ad09d816948a3247e275c4201d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-04 03:11:53.000000000","message":"Patch Set 11:\n\n\u003e (1 comment)\n\nYou are right, it was working perfectly in my copying image case, I haven\u0027t tested imports with it. Yes there is no other way in this case.","accounts_in_message":[],"_revision_number":11},{"id":"c3a34b3e16a950937bc76cb060738df8337051b0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-04 03:12:42.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"0acee23d7005dba046e986df606af0ec73a84584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 04:52:14.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2bebe901954f4490bc3c12f81c5f3be9 : SUCCESS in 1h 38m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/83683589ba8f43cd812bfd2e7ec0f332 : SUCCESS in 55m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8a0d99e666f7467984e11e1d1ee4fef5 : SUCCESS in 59m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f680a1249dfb4359ab12a3bb3ebc91c4 : SUCCESS in 1h 30m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef1cc899c00344a1af333688a1ac8645 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28642a858204b44b0164bb63230dc36 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf4664b52d84464a8291e6a2c22b1778 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39732f4232004cc486a64349c2658329 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/436cb1b79c4a429c86cd75051c4dfb06 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1acc4e45eea648329d458845b013c107 : SUCCESS in 7m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/215c207d244c4f0f926402a4c30d30f6 : SUCCESS in 9m 40s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/357de0fc754d4e4ba6b38e62f272f864 : SUCCESS in 10m 00s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/21345b4632de4468b661a8537e5329ad : FAILURE in 4m 40s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/429525a1b8b646958f6b75143f0c3c36 : SUCCESS in 4m 07s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f194c39a29cd442ea79659b8445c9a45 : SUCCESS in 1h 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c7ef4846888fad0e77a64b72d341f85637ece1ce","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-04 05:29:37.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)\n\nPending work:\n\n1. Check image is deleted or not in each import task \n2. Notification\n3. os_glance_failed_imports, os_glance_successful_imports","accounts_in_message":[],"_revision_number":11},{"id":"ae1ce317480003c327b97a1519bc95beccbc4b47","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-04 06:06:10.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/11/check-not-vote/ext-glance-zuul/cd7cee9 : FAILURE in 34m 30s","accounts_in_message":[],"_revision_number":11},{"id":"df3fc14a5a3b828331f38a3825ebeee1da2ebecf","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-04 18:08:13.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"b450b117388fbc145b792884c394cc84597710d2","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-04 18:08:49.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"936547bf839a6a88c2efa33c549c36e6346b53ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 19:58:11.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e17d113388744ffe9361b519824c35f3 : SUCCESS in 1h 20m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0779fdbd9cdd4e20b334f922350f6d25 : SUCCESS in 1h 01m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/68c770b0c5924a78b6f21ca1604e75f5 : SUCCESS in 1h 00m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92e9547ba9c54c41a17c2d4dd4c6b3a7 : SUCCESS in 1h 14m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b57dd6b3df4d43c8af91267f8e440478 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/922452ffda4e4dd5a3e31ee7909d13be : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b8556fdfa0a4e38a129f872463d2d2f : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f9c3fd57ed84151a5287350290a7ea3 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1713e6e235f542bb83a2770a65c84038 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a25cfe7bb9e4d2289f8eae062e14ec6 : SUCCESS in 5m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c8dea3e02b9543bfa72a5bf97f550189 : SUCCESS in 10m 00s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e350e8a5595745c687ac9dcb39081b69 : SUCCESS in 10m 45s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/80ecf5cd09f54f68814bbb75cb965910 : FAILURE in 5m 41s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ed1e77eb2eec40c3af6697aee3975d37 : SUCCESS in 5m 38s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/62903164b43948009c36cfad3cd40cd2 : SUCCESS in 1h 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3fad1526be245eab98f3abd310b130d771234375","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-04 23:55:37.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/12/check-not-vote/ext-glance-zuul/cfa7100 : FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"63a90535e21596326b1d397943bb5fca95168883","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-12-05 13:40:20.000000000","message":"Patch Set 12: Code-Review+1\n\nFWIW: this patch worked for me using two ceph RBD backends. Thanks!","accounts_in_message":[],"_revision_number":12},{"id":"5ce72dcb6fd81ed9b05a80c97c7eca3f66f171ce","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-10 16:45:52.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e4552ef07b4966c0c22c3ba9ce96c58349b9bb12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 23:12:22.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6ef615e5f8e9459fb024602e4e39a5f3 : SUCCESS in 1h 36m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5279466b2d544f3680cf7fdee115596f : FAILURE in 55m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/23a577376d694032967767a1e0dcf10f : SUCCESS in 1h 00m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/009c6f6032c54a5ba66baf68151e02bf : SUCCESS in 1h 14m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47700b3cd6fc4b7ca6c7d4f798b0646b : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa24a220bcc444be951a2e7457cd6c33 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/abb98dd676a2418a880b77d3d3463133 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28286ef9d3df48548ec2fc6a92141f2a : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7a1820f9c2a49458ffa15fa42082021 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d948e2ecbbe4cf3978fd196cc1b91ae : SUCCESS in 6m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/97be0f3b5c69468680562be0a60cbe83 : FAILURE in 10m 11s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ff3464fe5ad940e1acdcda61fe9a5aa4 : SUCCESS in 10m 00s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/61c5762499c04fee8665e5662511b171 : FAILURE in 6m 11s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b29acc381e314c089749cdd870785736 : SUCCESS in 4m 39s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/01b7ca2f27ea45d0aa94ba277d250f30 : SUCCESS in 1h 07m 37s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"52f4c4a6d707b2bc53c48a7cce75f43f97d0c4e2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-11 06:00:18.000000000","message":"Patch Set 13:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/13/check-not-vote/ext-glance-zuul/2d238c9 : FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"dc93791c66f8e6ea703e0b6dfde9c45bdbc380e5","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-12-11 17:32:40.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"263457530e47f52377b1ca0c97c9aa56b85dcf2c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-11 17:32:59.000000000","message":"Patch Set 13:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/13/check-not-vote/ext-glance-zuul/ad50062 : FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"95f22e524018d1b4d67865ddf10b183d63da58b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 19:20:11.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d7263948957a41dcadb32ae10bb822cb : SUCCESS in 1h 31m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f14a3a1e983c4b47980102c2ec7955d0 : SUCCESS in 1h 07m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44c4bd390c554eb08e0106eabda800e4 : SUCCESS in 1h 19m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a08927f84b6f434ba5f6761f2cbf211b : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/908e6df9248d4c94bae0e6de35b2d02a : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41e3dd38b4b14b0eb507d5faad4f9e52 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/217b0757475d41908816d533708e29d6 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54fe5b4b07b540c78365c7a4d4b0d889 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50368c3e65714b4caf7fd5d669852699 : SUCCESS in 8m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/02278329d6114149920e4ca1019184fc : SUCCESS in 8m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/617fd0df43004a1c935976e06c9745d2 : SUCCESS in 9m 57s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/b284d21dbcb24d9c848aeedf75fd4881 : TIMED_OUT in 31m 17s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/68abb476f1e8451294294b31f920be72 : SUCCESS in 5m 27s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/108ae60ecaca4e10ba2c6733f11868c7 : SUCCESS in 1h 03m 15s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b1cdd2544bd69085705a666b4dacaf251a632a01","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-11 22:59:03.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"921fb6e8c01cfedf69c328de0f5bdf9e65164d7d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-11 23:33:49.000000000","message":"Patch Set 14:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/14/check-not-vote/ext-glance-zuul/26e11e8 : FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"48e670382614b6c3c6171f3ba92c35ee0464334c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 01:05:24.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4e07e1abcb56402f81e8eb5c7448f918 : SUCCESS in 2h 05m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/298f031b041942088d5a218f2c2154d4 : SUCCESS in 1h 14m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/10caf9607078460b91ac358e02cbf772 : SUCCESS in 1h 36m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13a03aa11bf349289c3a758adbd9cfcf : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94ec4adfbc024817b529d594d53323a7 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df76964e08b5454998acaff74282c0a3 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9542f54e54584d6ab89725ea8789d739 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c6f021454744b12accc71d25b9c2f5b : FAILURE in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0255feab9e464c8b85b7da68d896ed67 : SUCCESS in 5m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/346bd280c79b48f0b1e45ec2a82b519b : SUCCESS in 7m 42s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/45a084b7f8cd44bd8c1355c03d2da12c : SUCCESS in 10m 40s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/7824e0b57f0444b68b420f52536d4175 : FAILURE in 4m 49s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/05602379647946109230612d65f9e8f6 : SUCCESS in 4m 28s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e8e3068588b04a728f28218007e6bcbf : SUCCESS in 56m 24s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e8c8f24c25aa5dbcd5881ee57e195cb5c54e6de5","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-12 20:51:02.000000000","message":"Patch Set 14:\n\nI have a question regarding the implementation of os_glance_importing_to_stores and os_glance_failed_import.\nShouldn\u0027t the database model be modified to add these 2 new properties?\nMoreover, as both are list, I think I must add 2 new tables containing the values.\nIs that correct or am I missing something?","accounts_in_message":[],"_revision_number":14},{"id":"1587184250451e3d5887ef9f9ded41eb5f83dcfb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-13 06:51:48.000000000","message":"Patch Set 14:\n\n(2 comments)\n\n\u003e I have a question regarding the implementation of os_glance_importing_to_stores\n \u003e and os_glance_failed_import.\n \u003e Shouldn\u0027t the database model be modified to add these 2 new\n \u003e properties?\n \u003e Moreover, as both are list, I think I must add 2 new tables\n \u003e containing the values.\n \u003e Is that correct or am I missing something?\n\nIMO You can use extra_properties to set these properties","accounts_in_message":[],"_revision_number":14},{"id":"f98c77a6ab910f719d648be5dc1eec75c5795d08","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2019-12-13 20:57:58.000000000","message":"Patch Set 14:\n\n(2 comments)\n\nThanks for the hint on extra_properties, this seems like a great idea.","accounts_in_message":[],"_revision_number":14},{"id":"f370b93286b8d6f0d5902e919080e2778d617eb4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-13 21:56:30.000000000","message":"Patch Set 14:\n\n(1 comment)\n\nI was looking this over in connection with reviewing your spec and noticed a typo.","accounts_in_message":[],"_revision_number":14},{"id":"f9f1b71cc9fd8968d6548a08375e6722ffd39cfb","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-01-05 15:43:09.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"1cba763d0191019bd5338ce9a5c3e468498b1cd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-05 16:57:28.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a440c71b898b46c19c2de6c6e23dc957 : SUCCESS in 52m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9b16ded6ec9242d38e9551f40ddea107 : SUCCESS in 1h 13m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c9a26901dc5462faabdc05b0dc5422e : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466afb65a68e469e926f39bd80d9ea4b : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65ae71d742d84fc58835393cd1fa5ac6 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2165e58972cf4a34917ef7febcc50082 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/163252ff873a42e9aedec271bca74e12 : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bf193a13c1a243efa4cab89acba5fce2 : SUCCESS in 9m 09s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b3bd494e3bbe457ba7ff93f8e2aa2b12 : SUCCESS in 8m 21s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b9ee548786cd4feab23745246802fd89 : SUCCESS in 4m 14s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5770025b58ac4655b5e0db4208103962 : SUCCESS in 1h 04m 14s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5b12bf2485916502ca7e75acb84a35e2c9a9afcb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-06 10:01:18.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"bf925eaecff56600dabd731878449eb1e2cdc22d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-06 10:03:34.000000000","message":"Patch Set 16:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/16/check-not-vote/ext-glance-zuul/7bbb424 : FAILURE in 1m 55s","accounts_in_message":[],"_revision_number":16},{"id":"6e8584f00ec3640261f6e5a4906b246202a36da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 12:13:10.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2947f70e2a8a4d5e8f197b0323638411 : SUCCESS in 1h 05m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e701241c36d849688d638590148ba54e : SUCCESS in 1h 18m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f73a08f790e4bf4b0c8fdb7152fbf46 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6fd3986f90d4b1ab7d418b9e99acfe3 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35a63c53893d4e6698c02f63500f948a : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c17bf1d088a4a7da6786ef1427328a8 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51d152108b1414fa86778ad9ddd9dab : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf39b18c661f44798bce6f64c4d88b43 : SUCCESS in 10m 47s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/bbcd10a4e94440fca33176f106fcd104 : SUCCESS in 12m 20s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c525cf9ab4a44b55af1dd1b012bb0615 : SUCCESS in 5m 27s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9cfc3525eb58433aad3c5799a64d4c6c : SUCCESS in 1h 09m 46s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"99e8342ec3587c4d8b0b8bd747c63d29c8e7feb4","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-01-19 18:43:28.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"75572a13f00be26b9dc76ec5183792a0255fec8a","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-01-19 18:49:22.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"2ac7d3bb6ed2904cf4a5eded8ff1a2da0fa65d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 20:11:16.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d34f7efdda64484799ab7c24f5434f3b : SUCCESS in 1h 01m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d6b03aaf5b8240bda63b312439c6a3d4 : SUCCESS in 1h 16m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54e8b86e9a044ffbbe792c479470ecc5 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2417c83bbc0a475a9df038206db1e514 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e8ed584933b432d9a18a758ade4fdad : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e38f3f61531e46cea7ee7001064e5162 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/729f61ea53bc49538e7314e25b350c4e : SUCCESS in 5m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/20df7c8fc8f34a9aa4639305485f54f5 : SUCCESS in 8m 09s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3139bbb3c6bb4b12839a198ec377f143 : SUCCESS in 8m 56s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e7a06dc39a5049df8f2c783921410127 : SUCCESS in 4m 02s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/69514d3878a4484abbe511e737da69a2 : SUCCESS in 1h 06m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08088dc3266540ed9c6e98ad4649344f : SUCCESS in 1h 03m 52s","accounts_in_message":[],"_revision_number":18},{"id":"e159fc84dcb849bdfece68a26668b3ebb85bfd38","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-19 20:34:52.000000000","message":"Patch Set 18:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/18/check-not-vote/ext-glance-zuul/cf04dd4 : FAILURE in 1m 22s","accounts_in_message":[],"_revision_number":18},{"id":"13057950c4b7cc7727e2b21361f185e521ed8b95","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-01-20 10:21:36.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"6bb0a4303f157d49244b27d53eee28190bffcd56","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-20 11:18:42.000000000","message":"Patch Set 19:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/19/check-not-vote/ext-glance-zuul/53f3ac6 : FAILURE in 1m 31s","accounts_in_message":[],"_revision_number":19},{"id":"615b06f0a0ba9510da48d692d5236504b1380127","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 12:37:39.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8aa3e61e67ee492e8f2b614be298db72 : SUCCESS in 1h 07m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f07543d76b6b4b0fabd04be08464079f : SUCCESS in 1h 48m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38073affac4a450b891da21a9b3a0fb8 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf3957788eb4eae9267891de419dfe7 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11eaed679ac84a2dbf2c9b1c0eb9adee : SUCCESS in 7m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/836b289209154130aafe955691e88a50 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f57021b9faa24131ae3dd5dcd9a2522d : SUCCESS in 9m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c8668b372178406ab03330bbffbec8a3 : SUCCESS in 9m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/af2e488f68c645b38a922436ab22aabc : SUCCESS in 10m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5fc75bc54c224cb48df601c2a866d1d7 : SUCCESS in 6m 09s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aa2a881259f748588f6be2ebad3f2ae3 : SUCCESS in 1h 01m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de93853fef504d51a5c4267fe3f933e3 : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":19},{"id":"cb2144a7cfdab73f6b09459ce781a63e64cc5008","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-01-27 13:30:37.000000000","message":"Patch Set 19: Code-Review-1\n\n(4 comments)\n\nJust noticed the possible issues in the location.py code, please let me know what you think and if you need help figuring out the fix for that. (or the reason why I\u0027m totally mad thinking that would be the case)","accounts_in_message":[],"_revision_number":19},{"id":"aca7c09d18f9d7257a02e4c668d327ffe290c60d","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-01-27 20:48:41.000000000","message":"Patch Set 19:\n\n(2 comments)\n\nI agree with your concerns, and also think that they should be adressed.\nYour explanation is pretty clear, I should be able to implement it on my own.","accounts_in_message":[],"_revision_number":19},{"id":"c035f454aa62c5821eebb27bc2580a46a8bedadc","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-01-28 22:59:35.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"4d28893432f7c2f35a446d2a10030ac8da9d835e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-28 23:01:21.000000000","message":"Patch Set 20:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/20/check-not-vote/ext-glance-zuul/2054c38 : FAILURE in 1m 24s","accounts_in_message":[],"_revision_number":20},{"id":"6d617f33aae0bd943aa510a476760311d7cd820f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 00:49:06.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0d6bad26c3174265b5cbbef296b06157 : SUCCESS in 59m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/708b0822d5b54002a1492587c96493fc : SUCCESS in 1h 22m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8a15b2bb8034dc6a2937a76fb938d12 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/050d4ff68b274cb8b5f9edf2f7cb5a06 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a25c5b5c09e499a8e1625ec144d4c44 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd271513c8e74fa5aa2903fc1b933b04 : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f5536b57b234d0693998aa7d0ae146c : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc6b194fd0584ff3826d228fcd27e309 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cd31d9b2fd94861bee17965d2c7a6e7 : SUCCESS in 10m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/daba4e48acbb436bbef53c0960a85a2d : SUCCESS in 9m 55s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f75367ea70e44731a934f355753ee354 : SUCCESS in 12m 15s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8b43035e8cab44fbbd46bfbfaeda68b7 : SUCCESS in 5m 21s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ff013b58381f4f609b3635172d71185b : SUCCESS in 1h 07m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/136aa026ad354c5dadc0cbed4ccadd38 : SUCCESS in 1h 05m 22s","accounts_in_message":[],"_revision_number":20},{"id":"1c273198bafe2906293f8265600f78a8d88ebfda","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-29 06:18:27.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"6407a034f5d595a6e78b624be9870951fce7e12d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-29 07:25:01.000000000","message":"Patch Set 21:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/32/667132/21/check-not-vote/ext-glance-zuul/550ae41 : FAILURE in 10m 47s","accounts_in_message":[],"_revision_number":21},{"id":"1f942a7978aca9a8cfffc1387d21154f31ba61a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 08:14:16.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/245aa9c9f7974f518c7e17ea4c1031b9 : SUCCESS in 1h 16m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a235cec879b43e4b73f2391d5d61112 : SUCCESS in 1h 20m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68654b9ca00a415d84d8788f56ad8100 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8fffa7dd8f34fc580b983396d559ddf : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a17bd620e5834eeeaa69f0c2fd49efe6 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad947138181549a4932b2aa60e51bc4f : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dddc4b55cf084c23bcc9f6aeda78bedb : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1bd5e0bf75f40b984cbba8ad78f8ef7 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f45f0823136047d79b6d58de0fa457a0 : SUCCESS in 12m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25fb6e7e52f7408aadbaff48b36a4116 : SUCCESS in 9m 47s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/88a339e485d64b199d7cae71d2e921c3 : SUCCESS in 9m 40s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a912da94e89f42a9bac1a000b61a1971 : SUCCESS in 10m 21s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/05ffa3e03a74426b87cea95a57e46b65 : SUCCESS in 1h 25m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0eb85db159214299a8b06df2344b9aa5 : SUCCESS in 1h 07m 51s","accounts_in_message":[],"_revision_number":21},{"id":"683537cf262f8a91dd99c886e85d82e5e209308e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-04 06:32:22.000000000","message":"Patch Set 21: Code-Review+1 Review-Priority+1\n\nLooks good to me, just putting +1 as doc changes are still pending!!!\nIf we decide to add doc changes in followup patch then I will +2 it.","accounts_in_message":[],"_revision_number":21},{"id":"067a0574a32d9534ec231eb4ebd1ea382d50b724","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-02-04 22:36:38.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"e38a3b9eaa633df8bf3a277d96162217d0731ffb","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-02-04 22:38:36.000000000","message":"Patch Set 22:\n\n\u003e Looks good to me, just putting +1 as doc changes are still\n \u003e pending!!!\n \u003e If we decide to add doc changes in followup patch then I will +2\n \u003e it.\n\nI had some time to work on the doc today, I hope it will be good.","accounts_in_message":[],"_revision_number":22},{"id":"4cd486697a88ad5a4619e0ae5008554a26036943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 00:12:34.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0a3bbfbc43694fb8b965e6d8682c09cf : SUCCESS in 4m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/01f6c034d56e417b946e19272e9323f3 : FAILURE in 54m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ce80b623f384b0bb7a4c8f8c057b746 : SUCCESS in 1h 31m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0208b9041a84bbfa54bb8a3db541255 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ea097d1daed4b06bcc14d7cfb60bfc5 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f953c0b4cad4cdfb4481303a0c99fa9 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/384e4d987317470d81fff6b9ee77fce2 : SUCCESS in 8m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf4a2e1158ab4520a7696d849b5e68c4 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b40a6eb166f403994eedb20d6c40dcf : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/789c5fee56134e3683694f8ebc129d9b : SUCCESS in 10m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3806ca30cff248d18ce676d291878f96 : SUCCESS in 8m 43s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/56a675632ec94f799ca45d712b708978 : SUCCESS in 11m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/aed3877d636646c2bb293a731af394e3 : SUCCESS in 6m 03s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/580215cf15e34fadab27601d77b340c6 : SUCCESS in 1h 06m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12bcd26d20924457a995144133cdce84 : SUCCESS in 1h 01m 21s","accounts_in_message":[],"_revision_number":22},{"id":"1914c3e5aeaf8aa7f4ec851e9eda7eacd003d465","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-05 04:29:22.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"401dc44cc1c4ae1dfd27190aecd6ecc98a15a16a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-05 05:30:05.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"23eecb0680cfbad7afc55d16ee1876dcbdb47da0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 06:48:08.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/34ab8ae3bce845069fed182f3016000f : SUCCESS in 4m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1e839b34ea9741ad9ea2a3110f1a3ac0 : SUCCESS in 59m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da7158e30fae41f99f69c1eeca206646 : SUCCESS in 1h 17m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be875204b5834c1fbc9a266cb28d7e4f : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d27c79cb474147dda8fc66aae21b939c : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/759cb469241d42a5bd77e45e324cba91 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/768c6841aca34c04b5d9907a4d57b91d : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f301245375cc4dc299bdecd7361d792a : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/245fc2a2fef94ce9a4a629275516aeab : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6288898ce4614d44972c6f3e16fb5799 : SUCCESS in 8m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/224d79b022264eb4bf7760f7d4dae5d8 : SUCCESS in 12m 50s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c3a5c49635754d5481197b32bf730be9 : SUCCESS in 12m 00s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6ca5d7aae638421f8c9faa66df62770a : SUCCESS in 3m 56s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7d4874b64fd34892b752d7cb3226f704 : SUCCESS in 1h 04m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4230d5026fa4da0b14287f101011ee2 : SUCCESS in 1h 13m 48s","accounts_in_message":[],"_revision_number":23},{"id":"51c86bafc11dae3c2502f0bc863369163e3cf4b9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-05 09:48:36.000000000","message":"Patch Set 23: Review-Priority+1","accounts_in_message":[],"_revision_number":23},{"id":"66fe21a0a294e11366ffbef7f837b2fa053e7670","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-06 05:47:22.000000000","message":"Patch Set 23: Code-Review+1\n\nAPI related doc changes looks good to me, but you also need to explain about this new behavior here;\n\nhttps://docs.openstack.org/glance/latest/admin/interoperable-image-import.html\n\n(doc/source/admin/interoperable-image-import.rst)\n\nWhere you can explain how this will work, the new properties to track the progress of this task etc.","accounts_in_message":[],"_revision_number":23},{"id":"99fdebc2070cfc67298a9e2d549c8952bd28cb99","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-02-08 15:55:54.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"2d460df91485e041bd7657804c81f3db1cdc3888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 17:10:31.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dc2a43e6e02c42b6bee5a9fa3059d474 : SUCCESS in 4m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3c5de263ae4c4705b68c1dc101ece520 : SUCCESS in 54m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cdb491ee08014e05a2c7b9a15cc71b69 : SUCCESS in 1h 13m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57a3b923f27a44458c26aa8b61bfccba : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/157e911c14a840ce9f24db82c60d3cd8 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdd9c15fafff40ea93aa03fae2f4072b : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00ed6a8d4e0c45f7976b90bc16bec0a8 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94c493e3e4c74553823591dd071f90eb : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ad3cddc37d4263a6f4862e1eaf4fd0 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56a2f39007e14c918ceef6aed2abe5bc : SUCCESS in 10m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4cef6374ba444a399fb00b5f993dd66b : SUCCESS in 8m 50s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c113461bbf53489988fc9ecc68167c01 : TIMED_OUT in 31m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9f6dbaf98600477bbc3d790de5d220af : SUCCESS in 3m 48s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9d55b1f57a194c8c81332bd9dcae86e3 : SUCCESS in 1h 00m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dad2aa74021e421f948b6e360c6cdd5f : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":24},{"id":"aa3fc73391f1d231b44a8e5a8b5c99feb0439e00","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-02-08 18:22:10.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"19b2c9c8dc800528b0b0bb47a511592b756958b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 19:40:19.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aea4c0bf295f466a98e35eb327015d15 : SUCCESS in 4m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b2f5a65b349a418c8a63828ba412f12b : SUCCESS in 58m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ac56d8236af470ea64f4453ea9c570d : SUCCESS in 1h 13m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09f756a2401c44598c8a45e39cb6bac3 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73069f2dfbbe418ba082cfdd6cf0c292 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5257f11185b4089b770f7a39a69bdfd : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/474d6d3d70d743d1af828ec469eaff35 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38d5045f97284755b149dcca0c83062a : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a77159d8eef941219adc681856a58eed : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27e635a7335443159e00c974f92101a4 : SUCCESS in 10m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa4b761090834684962d9f268b60637a : SUCCESS in 7m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9cdcdcb3c62d4370a3db68a29c649c2a : TIMED_OUT in 31m 02s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/755122cbf94c4b0aacde27540f681b49 : SUCCESS in 3m 56s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d6f795e0c0db4b82a485fd181ff08bc5 : SUCCESS in 59m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f7bb8fc5dc54772a6c60620467698b6 : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":24},{"id":"10127a680a3a445afaaaff085b189566ba406db4","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-02-09 02:47:34.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"6f55489f6564611c3463b7b7cc7d68b0c9410bad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 04:35:31.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b5b2c1cd29c64847bc99f83390117a5c : SUCCESS in 4m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5da0a82dfd274d34afff8ca5060c3638 : SUCCESS in 59m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a64850636b8e47dcac6b8a6d5135d303 : SUCCESS in 1h 46m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97e60740f4324e6cafded0dc54cac213 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dbfba1a3c054a24918d3abe0abc71b9 : SUCCESS in 11m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a4830ce5c13469d986d3281ae4abade : SUCCESS in 12m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1441845481fb4af5945b045a3a93fa44 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08ef6b5a65b6474bbb8fe6128de972fa : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e507578e712d49799883344e4499fb9a : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49b8d243b3d74d798832ab940ec2f4e4 : SUCCESS in 8m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a10f256a1bcd4e8da2b19acdaa577e25 : SUCCESS in 11m 14s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b9417d5e7cc2416c9edc5199b9644b40 : SUCCESS in 8m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8dcc7a3af3624a18ac6f633d12541282 : SUCCESS in 4m 11s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ced06d8b88cb49d0beb2d7f9045fdd55 : SUCCESS in 1h 02m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb620edea57047b69bcaec976a7f7b5c : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":24},{"id":"bc1075afc12a36452c7ee4e3e55a4b7d341c3629","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-10 05:28:36.000000000","message":"Patch Set 24: Code-Review+1 Review-Priority+1\n\n(7 comments)\n\nLooks good to me, some nits and one question inline.\n\nThank you for your patience!!!","accounts_in_message":[],"_revision_number":24},{"id":"7fd4d96f987b6cc96497171027f87c9c42bff238","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-02-10 08:42:57.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"70b686532e5480411a319d8bbf7b3cb7467f1a02","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-02-10 08:51:59.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"0675e94b023b2fd41cbbef709b42984239345893","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 10:26:30.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1cf18b35a4114329807438f5fcf1889c : SUCCESS in 4m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/29a44e78145547a48e524e8e3e8e0f38 : SUCCESS in 1h 01m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/794141598a7749b1a39d838c988c0c44 : SUCCESS in 1h 28m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e76a6d42ea6546a0aaf6472993bc14f0 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dec1dfd80314e06a74dec2fff5a821e : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97d13f54e9874463953c75e6a9bc0bad : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c7659fb84744ebebf50ded70f812db4 : SUCCESS in 8m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5209ab8fc3434103812af1e252566260 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/286aa3c9b6584115ae5eb785f1b8f6d2 : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b337b18c1de24f0b98394703d8c59580 : SUCCESS in 11m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0750eb4b61cb431ea5eed2f5f57c9c7b : SUCCESS in 13m 36s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ad5f8bf63c184be998176ca8c77f75f4 : SUCCESS in 11m 21s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b42deb4a47504fdebe5d675345136ffc : SUCCESS in 5m 55s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f4e78ae412e542909761776eae4e3258 : SUCCESS in 1h 31m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3e63b414e1e4ad2b6f0c7d9a3404eca : SUCCESS in 1h 16m 48s","accounts_in_message":[],"_revision_number":25},{"id":"cc3dd8d21c6594e2f7ccc5c8e5fe3534c56f9579","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-02-10 10:54:10.000000000","message":"Patch Set 25: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":25},{"id":"4427ae6953622fda98f1763cbcc469cacf69a415","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-10 11:10:52.000000000","message":"Patch Set 25: Code-Review+2 Workflow+1\n\nLooks good to me Greg!!!\n\nThank you for working on it!!!","accounts_in_message":[],"_revision_number":25},{"id":"efea4969b4484df52a7af0caee1c08cade6e64ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 11:11:51.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"5cd6e1ea5f5cf33966c73079d4bd44518b407038","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:41:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":25},{"id":"827d1933fb61f1040fb5d14c683e847501154fe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:41:01.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/97da823ba23d44e4970738dc90ef1794 : SUCCESS in 6m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6a20c41102e840399518c82bcdc210ae : SUCCESS in 59m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c9a8140bccaa4b6d82d698d588b3c289 : SUCCESS in 1h 18m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28a2af1f6633483b902141ba26f116e6 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25c41110b5f547c1b4f757d07172ea96 : SUCCESS in 8m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b37b2bd0eb55444086ec691c7e2645ea : SUCCESS in 8m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1121409929654850924baab81781f155 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/271de332eb55488c920d7163201c63df : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8371ed618de04c0fa86cd0bafd4c7d54 : SUCCESS in 11m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1e93262dd698411c8ca1e12612e25f4e : SUCCESS in 14m 05s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/72fc8a6960f14e539bd156ac32123f4f : SUCCESS in 12m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5379bc3a8513445992bc5f5206d41878 : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":25},{"id":"00aa36d7562eb80264a56d3217c10fda4544e447","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:43:24.000000000","message":"Patch Set 25:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/538a7fa0077a4386ab35ccbaea5efd39 : SUCCESS in 1m 19s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca6eed89d8824a53a3d59e631cf29692 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97a1ceba061f46108c0c0d4f4118ee75 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":25},{"id":"bbe5291e434820005ad050fd848c1daa2f0c9388","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-02-10 13:51:32.000000000","message":"Patch Set 25:\n\n\u003e Looks good to me Greg!!!\n \u003e \n \u003e Thank you for working on it!!!\n\nGreat!!\nIt was very interesting working on it. Thanks for your advices and follow up!","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"92492cf50461e214b777c707148886a8e87f340d","revisions":{"be0ddecaa2a5b28535af32c78a2ea25a25534e0d":{"kind":"REWORK","_number":1,"created":"2019-06-24 15:23:56.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/1"}}},"commit":{"parents":[{"commit":"e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a","subject":"Merge \"Windows: fix flaky tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"subject":"Add ability to import image into multi-stores","message":"Add ability to import image into multi-stores\n\nThe header \u0027x-image-meta-store\u0027 can now handle a list of backends\nseparated by comma (example: \"ceph_fast,ceph_cheap\").\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be0ddecaa2a5b28535af32c78a2ea25a25534e0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be0ddecaa2a5b28535af32c78a2ea25a25534e0d"}]},"branch":"refs/heads/master"},"199795924fea149df64f678b3f5e5ae4df6e0f33":{"kind":"REWORK","_number":2,"created":"2019-06-25 12:31:55.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/2"}}},"commit":{"parents":[{"commit":"e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a","subject":"Merge \"Windows: fix flaky tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-25 12:30:40.000000000","tz":120},"subject":"WIP Add ability to import image into multi-stores","message":"WIP Add ability to import image into multi-stores\n\nThe header \u0027x-image-meta-store\u0027 can now handle a list of backends\nseparated by comma (example: \"ceph_fast,ceph_cheap\").\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/199795924fea149df64f678b3f5e5ae4df6e0f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/199795924fea149df64f678b3f5e5ae4df6e0f33"}]},"branch":"refs/heads/master"},"4c91f5d828e0d5edca77acb9b131743f0e197371":{"kind":"REWORK","_number":3,"created":"2019-06-28 07:42:16.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/3"}}},"commit":{"parents":[{"commit":"e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a","subject":"Merge \"Windows: fix flaky tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-28 07:41:17.000000000","tz":120},"subject":"WIP Add ability to import image into multi-stores","message":"WIP Add ability to import image into multi-stores\n\nThe header \u0027x-image-meta-store\u0027 can now handle a list of backends\nseparated by comma (example: \"ceph_fast,ceph_cheap\").\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4c91f5d828e0d5edca77acb9b131743f0e197371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4c91f5d828e0d5edca77acb9b131743f0e197371"}]},"branch":"refs/heads/master"},"d784b758dc699fae01e844d2464d725abf712551":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-07-04 15:26:29.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/4"}}},"commit":{"parents":[{"commit":"e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a","subject":"Merge \"Windows: fix flaky tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-07-04 15:25:50.000000000","tz":120},"subject":"Add ability to import image into multi-stores","message":"Add ability to import image into multi-stores\n\nThe header \u0027x-image-meta-store\u0027 can now handle a list of backends\nseparated by comma (example: \"ceph_fast,ceph_cheap\").\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d784b758dc699fae01e844d2464d725abf712551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d784b758dc699fae01e844d2464d725abf712551"}]},"branch":"refs/heads/master"},"17664ec6d92c5e133c8eed98fb4a80fb350b7d2e":{"kind":"REWORK","_number":5,"created":"2019-11-28 07:06:34.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/32/667132/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/5"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-28 04:52:31.000000000","tz":0},"subject":"Add ability to import image into multi-stores","message":"Add ability to import image into multi-stores\n\nThe header \u0027x-image-meta-store\u0027 can now handle a list of backends\nseparated by comma (example: \"ceph_fast,ceph_cheap\").\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17664ec6d92c5e133c8eed98fb4a80fb350b7d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17664ec6d92c5e133c8eed98fb4a80fb350b7d2e"}]},"branch":"refs/heads/master"},"4ab906ff3a69b898eb5b3efdde0e2bdd7f584389":{"kind":"REWORK","_number":6,"created":"2019-11-28 10:00:23.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/6"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-11-28 09:59:44.000000000","tz":60},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ab906ff3a69b898eb5b3efdde0e2bdd7f584389"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ab906ff3a69b898eb5b3efdde0e2bdd7f584389"}]},"branch":"refs/heads/master"},"71535810973f82d2b5cb34474b483588e57d3fcb":{"kind":"REWORK","_number":7,"created":"2019-11-28 13:28:07.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/7"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-11-28 13:27:18.000000000","tz":60},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/71535810973f82d2b5cb34474b483588e57d3fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/71535810973f82d2b5cb34474b483588e57d3fcb"}]},"branch":"refs/heads/master"},"518192ab2aeee031364048fc963677e04becc58e":{"kind":"REWORK","_number":8,"created":"2019-12-02 16:43:11.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/8"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-12-02 16:42:53.000000000","tz":60},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/518192ab2aeee031364048fc963677e04becc58e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/518192ab2aeee031364048fc963677e04becc58e"}]},"branch":"refs/heads/master"},"42310fc6295cf1423cf08fab3810e27a19841b0b":{"kind":"REWORK","_number":9,"created":"2019-12-03 16:29:10.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/9"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-12-03 16:27:57.000000000","tz":60},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/42310fc6295cf1423cf08fab3810e27a19841b0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/42310fc6295cf1423cf08fab3810e27a19841b0b"}]},"branch":"refs/heads/master"},"2e03b1c5f526293dcaeffecc48eeb6613a3f8675":{"kind":"REWORK","_number":10,"created":"2019-12-03 21:13:01.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/10"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-12-03 21:12:45.000000000","tz":60},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e03b1c5f526293dcaeffecc48eeb6613a3f8675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e03b1c5f526293dcaeffecc48eeb6613a3f8675"}]},"branch":"refs/heads/master"},"b95652cbb2de12df6dd42276eb22e342dfa27d38":{"kind":"REWORK","_number":11,"created":"2019-12-03 21:51:51.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/11"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-12-03 21:51:12.000000000","tz":60},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b95652cbb2de12df6dd42276eb22e342dfa27d38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b95652cbb2de12df6dd42276eb22e342dfa27d38"}]},"branch":"refs/heads/master"},"7a8b9b70229a5151a60a7ce6ce85e772efb6872c":{"kind":"REWORK","_number":12,"created":"2019-12-04 18:08:49.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/12"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-12-04 18:07:43.000000000","tz":60},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a8b9b70229a5151a60a7ce6ce85e772efb6872c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a8b9b70229a5151a60a7ce6ce85e772efb6872c"}]},"branch":"refs/heads/master"},"6b797ba47b13949064c046eb04f260d5799aaedb":{"kind":"REWORK","_number":13,"created":"2019-12-10 16:45:52.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/13"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-12-10 16:44:39.000000000","tz":60},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6b797ba47b13949064c046eb04f260d5799aaedb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6b797ba47b13949064c046eb04f260d5799aaedb"}]},"branch":"refs/heads/master"},"f1129c8f42e47a8552cfd43bd50729a63012eb4c":{"kind":"REWORK","_number":14,"created":"2019-12-11 22:59:03.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/14"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-12-11 18:03:34.000000000","tz":60},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1129c8f42e47a8552cfd43bd50729a63012eb4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1129c8f42e47a8552cfd43bd50729a63012eb4c"}]},"branch":"refs/heads/master"},"2b6a9329df17c15693824918c9eec1cd0d647979":{"kind":"REWORK","_number":15,"created":"2020-01-05 15:43:09.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/15"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-01-05 15:42:49.000000000","tz":60},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b6a9329df17c15693824918c9eec1cd0d647979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b6a9329df17c15693824918c9eec1cd0d647979"}]},"branch":"refs/heads/master"},"9e269042ec501c93d8e5fe623b347f67548c013c":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2020-01-06 10:01:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/32/667132/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/16"}}},"commit":{"parents":[{"commit":"d41362517238b9ea3741c4a565e0b73a12c55b92","subject":"Change database migration version to ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d41362517238b9ea3741c4a565e0b73a12c55b92"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-06 05:22:59.000000000","tz":0},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e269042ec501c93d8e5fe623b347f67548c013c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e269042ec501c93d8e5fe623b347f67548c013c"}]},"branch":"refs/heads/master"},"c09a44d80d9873eeaa3cb0d7bfc45137dc4ddfa7":{"kind":"REWORK","_number":17,"created":"2020-01-19 18:43:28.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/17"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-01-19 18:42:59.000000000","tz":60},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c09a44d80d9873eeaa3cb0d7bfc45137dc4ddfa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c09a44d80d9873eeaa3cb0d7bfc45137dc4ddfa7"}]},"branch":"refs/heads/master"},"538e7c5414895c2ebd649b912e3195e287970caa":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2020-01-19 18:49:22.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/18"}}},"commit":{"parents":[{"commit":"2291fe673ec2f875bfc4c49e8ee487c074f1b027","subject":"Merge \"Fix string interpolation to delay by logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2291fe673ec2f875bfc4c49e8ee487c074f1b027"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-01-19 18:47:50.000000000","tz":60},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/538e7c5414895c2ebd649b912e3195e287970caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/538e7c5414895c2ebd649b912e3195e287970caa"}]},"branch":"refs/heads/master"},"4e6bc6a0a37f60093312aae68de5aaf79a75bf7e":{"kind":"REWORK","_number":19,"created":"2020-01-20 10:21:36.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/19"}}},"commit":{"parents":[{"commit":"2291fe673ec2f875bfc4c49e8ee487c074f1b027","subject":"Merge \"Fix string interpolation to delay by logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2291fe673ec2f875bfc4c49e8ee487c074f1b027"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-01-20 10:21:12.000000000","tz":60},"subject":"WIP: Add ability to import image into multi-stores","message":"WIP: Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e6bc6a0a37f60093312aae68de5aaf79a75bf7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e6bc6a0a37f60093312aae68de5aaf79a75bf7e"}]},"branch":"refs/heads/master"},"41210a4547d7fc72f1cae3c671e28e5342537faa":{"kind":"REWORK","_number":20,"created":"2020-01-28 22:59:35.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/20"}}},"commit":{"parents":[{"commit":"2291fe673ec2f875bfc4c49e8ee487c074f1b027","subject":"Merge \"Fix string interpolation to delay by logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2291fe673ec2f875bfc4c49e8ee487c074f1b027"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-01-28 22:55:56.000000000","tz":60},"subject":"Add ability to import image into multi-stores","message":"Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41210a4547d7fc72f1cae3c671e28e5342537faa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41210a4547d7fc72f1cae3c671e28e5342537faa"}]},"branch":"refs/heads/master"},"a6ed4eb233f33a78b0f6b046f3f2cf682ed00843":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2020-01-29 06:18:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/32/667132/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/21"}}},"commit":{"parents":[{"commit":"26c5c8dff670617ecbcc2b319b25e632fa88cb4a","subject":"Merge \"Remove OS::Compute::Trust metadef\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26c5c8dff670617ecbcc2b319b25e632fa88cb4a"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-29 05:42:18.000000000","tz":0},"subject":"Add ability to import image into multi-stores","message":"Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6ed4eb233f33a78b0f6b046f3f2cf682ed00843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6ed4eb233f33a78b0f6b046f3f2cf682ed00843"}]},"branch":"refs/heads/master"},"c37fdaef7d08bb626d2fe72102700647b1f76f76":{"kind":"REWORK","_number":22,"created":"2020-02-04 22:36:38.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/22"}}},"commit":{"parents":[{"commit":"26c5c8dff670617ecbcc2b319b25e632fa88cb4a","subject":"Merge \"Remove OS::Compute::Trust metadef\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26c5c8dff670617ecbcc2b319b25e632fa88cb4a"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-02-04 22:36:19.000000000","tz":60},"subject":"Add ability to import image into multi-stores","message":"Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c37fdaef7d08bb626d2fe72102700647b1f76f76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c37fdaef7d08bb626d2fe72102700647b1f76f76"}]},"branch":"refs/heads/master"},"96ff4031c04aff8e3507ab19256e03a15f0aa0e9":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2020-02-05 05:30:05.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/32/667132/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/23"}}},"commit":{"parents":[{"commit":"8649fdc2a293009899d2aa918ec360e30adc8e77","subject":"Ensure store ID parsed from URI is properly encoded","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8649fdc2a293009899d2aa918ec360e30adc8e77"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-05 04:57:14.000000000","tz":0},"subject":"Add ability to import image into multi-stores","message":"Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96ff4031c04aff8e3507ab19256e03a15f0aa0e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96ff4031c04aff8e3507ab19256e03a15f0aa0e9"}]},"branch":"refs/heads/master"},"8b7610ca1fa4de504c5115b1f8177cdc0d6007ac":{"kind":"REWORK","_number":24,"created":"2020-02-08 15:55:54.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/24"}}},"commit":{"parents":[{"commit":"8649fdc2a293009899d2aa918ec360e30adc8e77","subject":"Ensure store ID parsed from URI is properly encoded","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8649fdc2a293009899d2aa918ec360e30adc8e77"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-02-08 15:49:37.000000000","tz":60},"subject":"Add ability to import image into multi-stores","message":"Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b7610ca1fa4de504c5115b1f8177cdc0d6007ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b7610ca1fa4de504c5115b1f8177cdc0d6007ac"}]},"branch":"refs/heads/master"},"92492cf50461e214b777c707148886a8e87f340d":{"kind":"REWORK","_number":25,"created":"2020-02-10 08:51:59.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/32/667132/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/667132/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/667132/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/667132/25"}}},"commit":{"parents":[{"commit":"8649fdc2a293009899d2aa918ec360e30adc8e77","subject":"Ensure store ID parsed from URI is properly encoded","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8649fdc2a293009899d2aa918ec360e30adc8e77"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2019-06-24 14:39:46.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-02-10 08:39:01.000000000","tz":60},"subject":"Add ability to import image into multi-stores","message":"Add ability to import image into multi-stores\n\nThe import image api now supports a list of stores to import data into.\nThis list can be specified through a new \"stores\" field that has been\nadded to the request body.\nDuring import stage, Glance iterates overs this list and send the data\nto each store one by one.\nIf an invalid backend is requested by the user, an exception is raised.\nIf an errors occurs during verify, already pushed data is removed and\nimage state is unchanged.\n\nChange-Id: Id3ac19488c0a693d7042be4a3c83f3b9f12313d0\nImplements: blueprint import-multi-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/92492cf50461e214b777c707148886a8e87f340d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/92492cf50461e214b777c707148886a8e87f340d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
