)]}'
{"id":"openstack%2Fglance~667581","triplet_id":"openstack%2Fglance~master~I2ac631fa4d6a97003296d1d006c855cf538a785d","project":"openstack/glance","branch":"master","topic":"bug/1281453","hashtags":[],"change_id":"I2ac631fa4d6a97003296d1d006c855cf538a785d","subject":"Bug #1281453","status":"ABANDONED","created":"2019-06-26 11:09:55.000000000","updated":"2019-07-03 12:46:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7e9833c0e2a2b4c088399044d98c46d21da15edc","_number":667581,"virtual_id_number":667581,"owner":{"_account_id":30459,"name":"Artem V. Okhotnikov","email":"okhotnikov.artyom@mail.ru","username":"nominedei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":-1,"date":"2019-06-27 00:02:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-2,"date":"2019-06-26 12:39:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 12:39:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-27 00:02:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"911fc02ded23214dfbf0993238d7ec7d13a2a711","author":{"_account_id":30459,"name":"Artem V. Okhotnikov","email":"okhotnikov.artyom@mail.ru","username":"nominedei"},"date":"2019-06-26 11:09:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50501ae0d287a8ea0de47cb5bff062a3a3fe8ee8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-26 12:39:57.000000000","message":"Patch Set 1: Code-Review-2\n\nMany of these are incorrect and definitely not wanted.\n\nCommit message should also state what is being done, not just link to a bug.\n\nhttps://wiki.openstack.org/wiki/GitCommitMessages#Summary_of_Git_commit_message_structure","accounts_in_message":[],"_revision_number":1},{"id":"d662e33acdf36a875ce4d030accbb6d5705335b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 00:02:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/81/667581/1/check/tempest-full/7d99ab7/ : SUCCESS in 1h 44m 01s\n- neutron-grenade http://logs.openstack.org/81/667581/1/check/neutron-grenade/ff6c638/ : SUCCESS in 49m 16s\n- grenade-py3 http://logs.openstack.org/81/667581/1/check/grenade-py3/e5240e7/ : SUCCESS in 58m 37s\n- tempest-full-py3 http://logs.openstack.org/81/667581/1/check/tempest-full-py3/b856c5d/ : SUCCESS in 1h 32m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/667581/1/check/openstack-tox-lower-constraints/3da5937/ : SUCCESS in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/81/667581/1/check/openstack-tox-pep8/0d78f5c/ : FAILURE in 5m 49s\n- openstack-tox-py27 http://logs.openstack.org/81/667581/1/check/openstack-tox-py27/3d0025a/ : SUCCESS in 4m 24s\n- openstack-tox-py35 http://logs.openstack.org/81/667581/1/check/openstack-tox-py35/90284d6/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/81/667581/1/check/openstack-tox-py36/51bb2bd/ : FAILURE in 5m 29s\n- openstack-tox-py37 http://logs.openstack.org/81/667581/1/check/openstack-tox-py37/ff672e7/ : SUCCESS in 6m 03s\n- openstack-tox-docs http://logs.openstack.org/81/667581/1/check/openstack-tox-docs/8bbe99d/html/ : SUCCESS in 4m 51s\n- openstack-tox-functional http://logs.openstack.org/81/667581/1/check/openstack-tox-functional/e691fed/ : FAILURE in 3m 48s\n- openstack-tox-functional-py35 http://logs.openstack.org/81/667581/1/check/openstack-tox-functional-py35/f0f9673/ : FAILURE in 3m 52s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/81/667581/1/check/glance-eventlet-ssl-handshake-broken-py3/dbb8da9/ : FAILURE in 3m 45s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/81/667581/1/check/glance-code-constants-check/da7fe34/ : SUCCESS in 5m 23s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/667581/1/check/devstack-plugin-ceph-tempest/d7362b3/ : SUCCESS in 1h 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d8a530fa32fb94c9e53751d83201718c1c889ae","author":{"_account_id":30459,"name":"Artem V. Okhotnikov","email":"okhotnikov.artyom@mail.ru","username":"nominedei"},"date":"2019-07-03 12:46:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"130ffbc8a7c1a0428459e6f6a51aaa2be57364c2","revisions":{"130ffbc8a7c1a0428459e6f6a51aaa2be57364c2":{"kind":"REWORK","_number":1,"created":"2019-06-26 11:09:55.000000000","uploader":{"_account_id":30459,"name":"Artem V. Okhotnikov","email":"okhotnikov.artyom@mail.ru","username":"nominedei"},"ref":"refs/changes/81/667581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/667581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/667581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/667581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/667581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/667581/1"}}},"commit":{"parents":[{"commit":"e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a","subject":"Merge \"Windows: fix flaky tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a"}]}],"author":{"name":"Artem Okhotnikov","email":"okhotnikov.artyom@mail.ru","date":"2019-06-26 11:03:07.000000000","tz":180},"committer":{"name":"Artem Okhotnikov","email":"okhotnikov.artyom@mail.ru","date":"2019-06-26 11:09:31.000000000","tz":180},"subject":"Bug #1281453","message":"Bug #1281453\n\nChange-Id: I2ac631fa4d6a97003296d1d006c855cf538a785d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/130ffbc8a7c1a0428459e6f6a51aaa2be57364c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/130ffbc8a7c1a0428459e6f6a51aaa2be57364c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
