)]}'
{"id":"openstack%2Fglance~668475","triplet_id":"openstack%2Fglance~master~Ibe9b862a93d663053d15a672ca766d1c730776f6","project":"openstack/glance","branch":"master","topic":"doc-update","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-07-03 14:39:01.000000000","reason":"Change was abandoned"},"29130":{"account":{"_account_id":29130,"name":"BI SHUN, KE","email":"aionshun@livemail.tw","username":"samuel_ke"},"last_update":"2026-07-03 14:39:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibe9b862a93d663053d15a672ca766d1c730776f6","subject":"Add paste_deploy config_file option for glance-api.conf","status":"ABANDONED","created":"2019-07-01 14:09:13.000000000","updated":"2026-07-03 14:39:01.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"23911ce828d0ba585294bf13ea618daa69825c6e","_number":668475,"virtual_id_number":668475,"owner":{"_account_id":29130,"name":"BI SHUN, KE","email":"aionshun@livemail.tw","username":"samuel_ke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-08 02:45:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":-1,"date":"2022-07-08 12:04:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-01 14:49:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-08 02:35:06.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2022-07-08 12:04:11.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"4486e16348ad8d58ee04f70997a5c6590dbd2ec5","author":{"_account_id":29130,"name":"BI SHUN, KE","email":"aionshun@livemail.tw","username":"samuel_ke"},"date":"2019-07-01 14:09:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79ff1aefae2a4a61ac867136d78e915ffcfdaef1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 14:49:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/75/668475/1/check/openstack-tox-pep8/0fa16ad/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/75/668475/1/check/openstack-tox-docs/fb7ac01/html/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2447d231bc269069dba271cf09baf5ce9baeace2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-08 02:31:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a39b700deba069ea38f0b7fe7dd403f4e00ca501","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-08 02:35:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"def3e739a6da697e9ab658373bec7bf5a1a55398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 02:45:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffbe775577e742d39d66b39992086369\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cd4b180d52d4b68b8c274581f4086f9 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c53978a329e48e6bbf5b060932384d6 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":2},{"id":"1c0677208827a0f7595d123f9873a15a87760e29","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-07-08 12:04:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23911ce828d0ba585294bf13ea618daa69825c6e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-03 14:39:01.000000000","message":"Abandoned\n\nThe commit log was not amended in 4 years. Also I think the goal of this section is not to showcase all possible config options. I\u0027ll abandon this, feel free to reopen if needed.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4862314db8bef47418b4b452d632e2e0c07092a1","revisions":{"96623dcaa3950a77e3a7147ce3586ecb7cc5ed41":{"kind":"REWORK","_number":1,"created":"2019-07-01 14:09:13.000000000","uploader":{"_account_id":29130,"name":"BI SHUN, KE","email":"aionshun@livemail.tw","username":"samuel_ke"},"ref":"refs/changes/75/668475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/668475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/668475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/668475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/668475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/668475/1"}}},"commit":{"parents":[{"commit":"e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a","subject":"Merge \"Windows: fix flaky tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a"}]}],"author":{"name":"Bi Shun Ke","email":"aionshun@livemail.tw","date":"2019-07-01 13:47:14.000000000","tz":480},"committer":{"name":"Bi Shun Ke","email":"aionshun@livemail.tw","date":"2019-07-01 13:55:51.000000000","tz":480},"subject":"Add Paste Deploy Config File Path for glance-api.conf and glance-registry.conf","message":"Add Paste Deploy Config File Path for glance-api.conf and glance-registry.conf\n\nChange-Id: Ibe9b862a93d663053d15a672ca766d1c730776f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96623dcaa3950a77e3a7147ce3586ecb7cc5ed41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96623dcaa3950a77e3a7147ce3586ecb7cc5ed41"}]},"branch":"refs/heads/master"},"4862314db8bef47418b4b452d632e2e0c07092a1":{"kind":"REWORK","_number":2,"created":"2022-07-08 02:31:21.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/75/668475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/668475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/668475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/668475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/668475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/668475/2"}}},"commit":{"parents":[{"commit":"fcdbf8a4a4b28ad63efaea127e51f72d4b578118","subject":"Merge \"Doc: fix URL in API call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fcdbf8a4a4b28ad63efaea127e51f72d4b578118"}]}],"author":{"name":"Bi Shun Ke","email":"aionshun@livemail.tw","date":"2019-07-01 13:47:14.000000000","tz":480},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-08 02:28:11.000000000","tz":120},"subject":"Add paste_deploy config_file option for glance-api.conf","message":"Add paste_deploy config_file option for glance-api.conf\n\nChange-Id: Ibe9b862a93d663053d15a672ca766d1c730776f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4862314db8bef47418b4b452d632e2e0c07092a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4862314db8bef47418b4b452d632e2e0c07092a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
