)]}'
{"id":"openstack%2Fglance~668842","triplet_id":"openstack%2Fglance~master~I72fe9b60f9380810e952e90f48ee226c682fff2f","project":"openstack/glance","branch":"master","topic":"update_install_html","hashtags":[],"change_id":"I72fe9b60f9380810e952e90f48ee226c682fff2f","subject":"Update the glance installation page","status":"MERGED","created":"2019-07-03 09:26:07.000000000","updated":"2019-07-08 16:06:15.000000000","submitted":"2019-07-08 16:06:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668842-1562601975242-5d2650a7","meta_rev_id":"fc731e0bc65854083ee5c58f3cdf71ad2238079b","_number":668842,"virtual_id_number":668842,"owner":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"date":"2019-07-04 01:50:32.000000000","_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":2,"date":"2019-07-08 16:06:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-08 14:54:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-08 14:54:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 11:51:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-08 14:54:28.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-07-08 16:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"804da01a9d225d87dda1b6f57f65021c3d0e2cf6","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-07-03 09:26:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d6f13fecc90b26abe4764ba689ac637b372e239","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 10:57:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/42/668842/1/check/openstack-tox-pep8/f7d9e9d/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/42/668842/1/check/openstack-tox-docs/bf46ee8/html/ : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f436febc26050c37de53105b9e9e0c30f9d51736","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-03 11:51:45.000000000","message":"Patch Set 1: Code-Review+2\n\nThis looks good, though I think I would rather see this whole section removed so there isn\u0027t something that someone needs to remember to do every release cycle.","accounts_in_message":[],"_revision_number":1},{"id":"e916e0f0672ffa1e649080a5a484fe0808cb5fb0","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-07-04 01:46:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"55ec2b508da6892eb5b9bc171203d4250b7d7a4b","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-07-04 01:50:32.000000000","message":"Patch Set 2:\n\n\u003e This looks good, though I think I would rather see this whole\n \u003e section removed so there isn\u0027t something that someone needs to\n \u003e remember to do every release cycle.\n\nThank you for your suggestion. I\u0027ll delete the section \"Supported Releases\" as you suggested.","accounts_in_message":[],"_revision_number":2},{"id":"c33d72a636437bf3cba3b002d837c6b662c4a223","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 02:44:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/42/668842/2/check/openstack-tox-pep8/fea7989/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/42/668842/2/check/openstack-tox-docs/5e965a1/html/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b4079eba108033ef7bf313e37e2cb3deb79e7918","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-07-08 14:54:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"3001984f46396acc43d48458dc857375026eced1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 14:54:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9f605809e0ec88c1e2367fbc7b5a1e6adb1a0aba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:06:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/42/668842/2/gate/openstack-tox-pep8/9d3ff95/ : SUCCESS in 8m 45s\n- openstack-tox-docs http://logs.openstack.org/42/668842/2/gate/openstack-tox-docs/014c1dd/html/ : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d0fca2f620994e233deeceff19cbb53740bddf18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:06:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d319acbd649720bc459e929eaa7ee04c3acf0a86","revisions":{"fdbc191359a8c2540e716351ff75af39120788e4":{"kind":"REWORK","_number":1,"created":"2019-07-03 09:26:07.000000000","uploader":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"ref":"refs/changes/42/668842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/668842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/668842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/668842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/668842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/668842/1"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"zengjia","email":"zengjia@awcloud.com","date":"2019-07-03 09:14:20.000000000","tz":480},"committer":{"name":"zengjia","email":"zengjia@awcloud.com","date":"2019-07-03 09:25:41.000000000","tz":480},"subject":"Update the glance installation page","message":"Update the glance installation page\n\nModifying Version Description Errors in Web Page\n\nChange-Id: I72fe9b60f9380810e952e90f48ee226c682fff2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fdbc191359a8c2540e716351ff75af39120788e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fdbc191359a8c2540e716351ff75af39120788e4"}]},"branch":"refs/heads/master"},"d319acbd649720bc459e929eaa7ee04c3acf0a86":{"kind":"REWORK","_number":2,"created":"2019-07-04 01:46:45.000000000","uploader":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"ref":"refs/changes/42/668842/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/668842/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/668842/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/668842/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/668842/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/668842/2"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"zengjia","email":"zengjia@awcloud.com","date":"2019-07-03 09:14:20.000000000","tz":480},"committer":{"name":"zengjia","email":"zengjia@awcloud.com","date":"2019-07-04 01:46:04.000000000","tz":480},"subject":"Update the glance installation page","message":"Update the glance installation page\n\nModifying Version Description Errors in Web Page\n\nChange-Id: I72fe9b60f9380810e952e90f48ee226c682fff2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d319acbd649720bc459e929eaa7ee04c3acf0a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d319acbd649720bc459e929eaa7ee04c3acf0a86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
